)]}'
{"id":"openstack%2Fneutron~770355","triplet_id":"openstack%2Fneutron~stable%2Ftrain~Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c","project":"openstack/neutron","branch":"stable/train","topic":"oslo_lc_drop","hashtags":[],"change_id":"Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c","subject":"Dropping lower constraints testing (stable Train)","status":"MERGED","created":"2021-01-12 15:08:03.000000000","updated":"2021-01-16 12:17:39.000000000","submitted":"2021-01-16 12:13:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770355-oslo_lc_drop","meta_rev_id":"137b04f6418f8a06bfb45835f30784b0a17cdf51","_number":770355,"virtual_id_number":770355,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-16 12:13:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-01-14 09:10:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-01-13 09:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-01-13 09:50:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-16 04:32:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-01-14 09:10:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 17:49:00.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-13 05:24:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 07:53:33.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-01-13 09:41:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-01-13 09:50:56.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-01-14 09:10:21.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-16 04:32:53.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"14170ffb7711cafa560130ece864c858ba0bbe17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-12 15:08:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29ad9f0c301f52ea78a3feaaea54ced48b9d0d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-12 15:50:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"75742dc3ace977d843deb829510917407930a558","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-12 17:49:00.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770355/2 : FAILURE in 1h 57m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ffa8817901b92da9dd7d134aedc778af90ef9170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 05:24:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/4ece8270b70c4909b89a0e30b62986be : SUCCESS in 45m 35s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/557104819e4d4c0092c11cc503b65f9f : SUCCESS in 37m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/f3f7a2b6ea5f4ae59a9fc92fc5a9cf4c : SUCCESS in 1h 29m 22s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/ff7530f1a8db4966a14d09c2badd3563 : SUCCESS in 1h 40m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/255201ff2d574bb891c98d6fbcfaffbf : SUCCESS in 1h 34m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8ca9b2f11a34dbea72a0569fdb66292 : SUCCESS in 25m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f72e922b4c44ebb0ea43904d06c757 : SUCCESS in 11m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76a37f697bfe44e6a70fa958bdfda848 : SUCCESS in 25m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87bc605ec1234c9cba0a197bc0725a04 : SUCCESS in 21m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1092aff6a574b72b6c76745b121adc4 : SUCCESS in 26m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dfb41b5a7f54f609fb9ae7c87a13894 : SUCCESS in 7m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/325783cd084142b3964d29e411ac8800 : SUCCESS in 6m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/58aa2b88234f466c94244435edb0663b : SUCCESS in 54m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b1a672c328d64b0f970aa12f0e83435f : SUCCESS in 1h 01m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e614edd7df4452487c9f39b91bb6dd4 : SUCCESS in 7m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/90567b9df5894d1890516b67ebbc885f : SUCCESS in 40m 33s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/2113b831039e4509877ec2a430fdaf8e : SUCCESS in 42m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ee9e65c53dd44b1aaee504b7dbb60f91 : TIMED_OUT in 2h 18m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/92acf2b167d94cd6b80300a9e71e45c1 : SUCCESS in 1h 41m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2c1643543b6f49f6b957eaa2f6835f48 : SUCCESS in 1h 34m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d3f65e425e1e4b9894e1df5eb3f9631e : SUCCESS in 1h 20m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ee2a1adff4e446a89ae18f755e88fe8 : SUCCESS in 1h 04m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/86d3c5c60fe64621acd079e90012a6b6 : SUCCESS in 1h 24m 42s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9b28274aeb5943fcbb995075d9d4826e : FAILURE in 2h 56m 30s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a420af9627a492fb59503b59fc79de1 : FAILURE in 1h 39m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/736d05692835463ab8b3cf298f38b81f : SUCCESS in 1h 07m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ea0edf67f74c453882764d04df936d46 : SUCCESS in 56m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c942ebfec4ce43a7a5d62b8df89eb150 : SUCCESS in 1h 10m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2e7868df45440b7932917abb7cdda56 : SUCCESS in 2h 02m 03s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3a295330132f4262b33f18e1693616f6 : FAILURE in 33m 45s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/21165d2c56c84ad68e42882e3dcd29a3 : SUCCESS in 1h 44m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0cfc35d8a524b64bcc2f73dcabc5a53 : SUCCESS in 1h 34m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/63d204ea25104e4084d10a4c08b7a8dc : FAILURE in 1h 42m 04s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a7bb7e168ed45ccb033173e5c77fda2 : SUCCESS in 39m 34s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6f4d19a03b4485b869ac896058ef299 : SUCCESS in 1h 04m 45s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b5e394205d64010ac19d79382fb9dcd : SUCCESS in 1h 20m 42s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/5a770baf0196444ab19fc196d0b9bf89 : SUCCESS in 3h 19m 28s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fc98a69e5076415b989801f1c0cd1be2 : FAILURE in 1h 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"98707033cb69026225da965f0660c072988f4621","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-13 07:53:33.000000000","message":"Patch Set 2:\n\nrecheck fullstack TIMED_OUT","accounts_in_message":[],"_revision_number":2},{"id":"4127cccf0589864917f5bc3d209cd0fe0f984ff3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-13 09:41:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5ef209c2ee44c889e21dc64cdecbeff38737aa06","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-13 09:50:56.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b8eb9ad92b5acf14b1a1092ed9f5ddf1e1c684d2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-13 09:52:50.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770355/2 : FAILURE in 1h 58m 43s","accounts_in_message":[],"_revision_number":2},{"id":"35217511c4a550b985698653de14a968a1c21d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 14:37:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/a663efec40224a88948c9ebfc194a8f1 : SUCCESS in 49m 32s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/c4ab1b117b1f46ff82529b11634d6b50 : SUCCESS in 30m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/72b5b46dccfa402a92f25573bfd4c0f9 : SUCCESS in 1h 18m 44s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/c855c95c24184da18e35745b9c550673 : SUCCESS in 1h 47m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/89371009d73c438387d4ddbba2f11622 : SUCCESS in 1h 31m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b5e6388395a4224a572ef2074a914c8 : SUCCESS in 31m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f75db7b36a8b41538df2bc5df0768b05 : SUCCESS in 11m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d759f730ae7472284333131b6e06d3d : SUCCESS in 23m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c183273382094560b57f31337d412657 : SUCCESS in 25m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb85059327dc4ed29395c4d954c5d967 : SUCCESS in 27m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/768b79831bea474ba4c8c2332734b6a4 : SUCCESS in 8m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/524e9e8796a14c56930207e2a011addb : SUCCESS in 7m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/e3812bd2948340c9b05e490c7ba1e47b : SUCCESS in 1h 01m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8f6b0d03df7c424bb201610a8e3eef0e : SUCCESS in 1h 24m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9594303af2d42e0ab045177e4630850 : SUCCESS in 9m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b673dd72e7a8485d83008d9ccbaf9c74 : SUCCESS in 43m 56s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/e6549a77adcd4f148204202b9f15ef08 : SUCCESS in 43m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/58f0bac3c45643b1be8c320f4fd46aa0 : SUCCESS in 58m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/67116e6883e74357b51516453e7afd25 : SUCCESS in 2h 07m 45s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/03bac115ed8a49898c33ca9b587c6897 : SUCCESS in 1h 10m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a55661fdf38a41fcb1c9a698fe0e834e : SUCCESS in 1h 10m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/457ffa738d2f43c280baab167b6a234f : SUCCESS in 59m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e437369f6e694f0a9672c367dfef2c8b : RETRY_LIMIT in 3m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3c7ea3ab975f406dbd3c0fc8e5206b7e : SUCCESS in 1h 40m 04s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/50e467d6e53949a88ba742c6a6cf1cc3 : RETRY_LIMIT in 3m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/15c8c9e358c44275907829408464fec0 : RETRY_LIMIT in 3m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e6aaf8a45c62420380cf51823d5dcebd : SUCCESS in 51m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a65644e392d44d2e8dfb4a4f7300f52a : SUCCESS in 1h 19m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/82e846e6963e487b8b7cc7e8f3d60aa7 : FAILURE in 2h 05m 31s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/87164e4e7f3a43f7b269e5c66ea24808 : FAILURE in 11m 46s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d10d213c39ab4db6887c3c36e714fee0 : SUCCESS in 57m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/847cb0043f544973b66ccbd7a21fc11b : SUCCESS in 1h 05m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/46583d9726b84d71b8bdf8e7ba872b9f : FAILURE in 1h 16m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2daf4b379d52430da31261ae28d5284e : FAILURE in 39m 28s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1f49a22f6c014714babde3d2fca2448e : SUCCESS in 1h 00m 24s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/55b1ec953ca846e280b0a24e1f2670d4 : SUCCESS in 1h 14m 29s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/8ad00444873c48c89747ade55b9877ae : SUCCESS in 2h 33m 39s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e09e6a662b3a41aca6c4a6e14c459288 : FAILURE in 1h 03m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ce0206f7d94adb96eb5b8d2bfc868a73fc6bbb0a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-13 16:30:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f4cb0782dfefcb94a86b4546872c991193ef25f2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-13 18:38:03.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770355/2 : FAILURE in 1h 56m 59s","accounts_in_message":[],"_revision_number":2},{"id":"3d3e4eedcee7efd822b1e5cfddd80acc7fa939ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 08:52:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/6994ea7baa164a9a84ff9f46117d81d4 : SUCCESS in 1h 03m 39s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/23b98386b6f1437bb381e2713150a63e : SUCCESS in 34m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/dfe4e78088b9433ebdfd769134a95edb : SUCCESS in 1h 38m 11s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/1782fc9d87814d7d863ff2c604f4c27f : SUCCESS in 1h 27m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/c0ed533839dc43859815731c4af25061 : SUCCESS in 1h 27m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/265fd44b16234f38b550224e2695d90a : SUCCESS in 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df653fa74724703b86cea5eaa7a63c6 : SUCCESS in 12m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e456fce61aed4392a64690b874550def : SUCCESS in 22m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d4ee2097cb6431f9ffad954b2bfc371 : SUCCESS in 26m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b9bc60571ed495ebf8a1fd1f6e852c4 : SUCCESS in 24m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1657a5ae7bf44b1ebd4d4d84b4c12278 : SUCCESS in 8m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e4fdac66b8704db6a47e754b4263b405 : SUCCESS in 3m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/44e8bcb3519849758be4d816b7f45410 : SUCCESS in 1h 07m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/69b4521d26894a998b2018bb494f83fa : SUCCESS in 1h 16m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c37329bc5ce24456957db63749d91f92 : SUCCESS in 9m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c6b450b640d74c81b09176c5d2759e6e : SUCCESS in 43m 01s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b6c161640c944c229523e7b649fa254e : SUCCESS in 41m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/38578260cb574f5ca29645ef3a6592aa : SUCCESS in 1h 06m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d37405baa5a345c198fbf00efd523579 : SUCCESS in 1h 55m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ef2f3305fd7042aca5f7fdbfa620fc35 : SUCCESS in 1h 08m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8c235de758ce47d4bde14af4a80053ce : SUCCESS in 1h 12m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f145d86cd92147cfa8b7834c2fc0da53 : SUCCESS in 1h 05m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/92bc7be7877c43f3ab3f42f523f4ec38 : SUCCESS in 52m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cc16c45b9e9f41278f919ab6b37d0c07 : SUCCESS in 1h 39m 41s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4b68f5859ad94c1d9e44f476368f2d9d : FAILURE in 1h 53m 46s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5610ec7f28f40ba909934cd63bc92ec : SUCCESS in 1h 23m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6ae5cc04da5647c79f643af5cb72a9e0 : SUCCESS in 57m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/064f29f34ba34142a98d2049a6a93256 : SUCCESS in 1h 12m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d21d3ac6b88d49a3901b8e6c9d3adb9c : SUCCESS in 2h 03m 56s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/0c855c92a4024e3a972cc59f3f16992b : FAILURE in 17m 45s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8076f671ce8847d8be8e5c69bd3f8760 : SUCCESS in 1h 01m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/104a7d9ccf7e459e9cdedd2a9167a408 : SUCCESS in 1h 18m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ad4e9edc7069414f93fdc5e3f5d9dc73 : FAILURE in 1h 08m 46s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/37d12d7bc3b1402d9d36500cf0d3076b : SUCCESS in 39m 32s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/276256eb9a184bcb8f386a3c7c676169 : SUCCESS in 1h 09m 02s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/24225c12e30f4e149b06c04bbe2e1fdb : SUCCESS in 1h 09m 46s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/8769257fd5f84ba48667034654ecd96c : SUCCESS in 3h 15m 46s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2d1f462e8eaa4f2bb4d844896a00e96e : FAILURE in 1h 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"de03b92172f9aff826a9fc83f8d0b30bdda39782","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 09:10:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e9cb70a2b94d9b0f19d7b815232a6e63d891fb8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 09:11:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"462bd63e0a1207d8647bc2cf0bafca6f1736da75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:05:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/6d8a2376cf834a96b38a9d91ce2713d8 : SUCCESS in 1h 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d27d228b85e94d43b1645a011fe77160 : SUCCESS in 12m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35ae9159f9eb48d5b17637befc37c99f : SUCCESS in 28m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f705dd475e9641eaa13d20ed097474e4 : SUCCESS in 21m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50858d1725de4d7f8f78b85c784a3777 : SUCCESS in 26m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d226831cff145449df20fec98c27682 : SUCCESS in 8m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/04a7de94183d41cab8b8ed5e1a501a3b : SUCCESS in 4m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/506a65ed2fa141cdb37001d8d2a9080a : POST_FAILURE in 1h 12m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9bfe29cf508c4ad2b42f9e9c97309379 : SUCCESS in 1h 28m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f4042d5c0a547bdb73d900f7a2ed10c : SUCCESS in 10m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8fbf98c43d1a40b2aa668ad9f273b4f6 : SUCCESS in 54m 07s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/88dad9c32dea410791a388e8dda55ee6 : SUCCESS in 43m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/30a8620830af49c2af5c11a59753c351 : SUCCESS in 1h 11m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/53e114e8146c41589aef67d5d138f840 : SUCCESS in 1h 06m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/28d5b1fb181e4b8aaf5e4f2325d3f310 : SUCCESS in 1h 28m 39s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2575e631865546aea09aa50b4e240e24 : SUCCESS in 1h 30m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9055bba85c54107959dbf20f7688621 : SUCCESS in 1h 28m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ceb2b25b88ad4e599d3b39be4642b51b : SUCCESS in 1h 33m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3cb2b17036734412aaf7ce8feed139a5 : SUCCESS in 1h 14m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23a0f68f3e1f44928679c5be1f814da6 : SUCCESS in 2h 20m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a384b47cfe6d4db38b5f2a81187d7ac8 : SUCCESS in 1h 29m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/102e3bfa8b694abd87eb95d4a9706056 : SUCCESS in 1h 13m 02s","accounts_in_message":[],"_revision_number":2},{"id":"515997915dc4e09b87ffeb2f39d01789a3561760","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 13:36:29.000000000","message":"Patch Set 2:\n\nrecheck POST_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"4c02cbbad83b70a2f50e5360a8b8135ea9fd4574","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-14 15:40:49.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770355/2 : FAILURE in 2h 03m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e174f298cb434465fa6c60f7dd187941fc0d0d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:30:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/f04baa6d5dfe45ca9fa3d498e100aa60 : SUCCESS in 49m 04s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/03e23fbf5ba4471abce156beb70b13d1 : SUCCESS in 35m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/e1d421149813476c820014bb31b9e189 : SUCCESS in 1h 35m 08s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/77c60a0b7db54d2ca7f52a7d8c8aca3f : SUCCESS in 1h 23m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/dbb2a479ac7b42efb49c930b21366c43 : SUCCESS in 1h 30m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/510e24e9fb4c4c738e47e9fd0be80bf9 : SUCCESS in 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d9c735e7934276969996e548438d20 : SUCCESS in 10m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4134f90a1dd04f2f8b417a904f0add86 : SUCCESS in 22m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8b0ea7e23334ada97c327ee8f98139d : SUCCESS in 24m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a97818eacdfc452c9c1cc68562cd4d02 : SUCCESS in 20m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075724d6d83b45aea5d5228bef98ff65 : SUCCESS in 10m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b8b910ade9d3447e83ab010e7ab993f2 : SUCCESS in 3m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/1ccfc6e79c4a415faee7ef5dbd212aca : SUCCESS in 58m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6323339410da474c8ed2c0572c15223b : SUCCESS in 1h 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85c73b66cc084e47af9e661e98e68822 : SUCCESS in 8m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8c95c320ac9e46669af5e1939f57d1be : SUCCESS in 42m 34s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/6c4ea2f8f5ce4385ae36676e7696cb1b : SUCCESS in 41m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/dbac2dff95fc4c9498aa758c2a5cf627 : SUCCESS in 1h 12m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/60d60acacdc844f9890077b311a7312e : SUCCESS in 1h 49m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8dc6b46709fc4dcbb638a75b4a886ea1 : FAILURE in 1h 03m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4c9f6928dbaa4a9fadc5cb8808c5909c : SUCCESS in 1h 07m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed292363422c43c89d51b31647acb20f : SUCCESS in 1h 01m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3237e714c7b447918419f7d597374358 : SUCCESS in 1h 03m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/319d96ed8ba04810980d864d45ebb7ee : FAILURE in 1h 33m 11s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0ba83227924f492092837b7f07564079 : FAILURE in 2h 05m 38s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/034c0cf3854449e9a8099eda92d41312 : SUCCESS in 1h 09m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dd90347cde244f7f8e849ce8a0618178 : SUCCESS in 53m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c7145838009a4a209648119835977f2e : SUCCESS in 1h 08m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ad06ed72acd4267976c662963e2a2e7 : SUCCESS in 2h 07m 52s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/438b5b724f824205a823ec41405cbdd7 : FAILURE in 10m 14s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/44a2fc6d4ff745f7a3a31b6d2daf5603 : SUCCESS in 55m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00a26893bdc443a6a40a50bd0b6c302e : SUCCESS in 1h 00m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fe989abe011f482cb0b16cdd5788b059 : FAILURE in 1h 22m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62c7214aea6646e68e4baa570f5227bc : SUCCESS in 38m 54s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9dcf8bcfb3284ff186926046d266b8d0 : SUCCESS in 58m 15s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c78310622ff54762b2570f8d5c65e20b : SUCCESS in 1h 05m 09s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/b40ab5541a51486b888ff0bc1362dd6e : SUCCESS in 2h 30m 39s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d33955697adb4497bb0c820fb387f5fb : FAILURE in 1h 00m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b710f240f5687da68aeb3d614cf308f184afc30","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-15 10:30:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"806d307a36e8974dfbc969fbca66e17830e3093b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-15 12:45:01.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770355/2 : FAILURE in 2h 11m 31s","accounts_in_message":[],"_revision_number":2},{"id":"17041cb39c79ed4f3e623bb3ff6934cd0ba1a4db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 23:20:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/fc560e5e8bcb4ef9ac7abfbfde2871bf : SUCCESS in 44m 51s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/e1413fa06d614230b57815a5896ba6fa : SUCCESS in 27m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/357ba13fe8ed45159bb197ab85d94e5f : SUCCESS in 1h 59m 15s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/abe7f6e0fcf148609d7abf00d00591b0 : SUCCESS in 1h 09m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/f4c092f6203f4d6084e3333ee35b0b3e : SUCCESS in 1h 14m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbdceb01c86e4c759e89b29f22f0d1ea : SUCCESS in 30m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41c4efe8d1dc4c45be9bd8671eb52d26 : SUCCESS in 12m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b32552574c24406a997c4a00174463d1 : SUCCESS in 22m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/718528463cae456482a9f526690fdc83 : SUCCESS in 19m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51b1731721574b53a90b8e975fd05f42 : SUCCESS in 25m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe1afe030cd4b5facc6e4a84d399d99 : SUCCESS in 31m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3aaa2cd1e56d47b4bd34c4217ba7b78b : SUCCESS in 3m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/ff93bed0b4914c958516de9aeae9f115 : SUCCESS in 1h 05m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8f9a30abe376455d9a60668fc3900580 : TIMED_OUT in 2h 04m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1882c6f197644546ac403670abd08043 : SUCCESS in 5m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/50074d2bf86043878d8228ccf0fe3432 : TIMED_OUT in 2h 13m 19s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/f72462154d54459c87c3d8b0d504d092 : SUCCESS in 39m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/853007dbd2414b458017bd10c2278ba4 : SUCCESS in 1h 07m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/baf3e1a5e076409ba31a2f236691a099 : SUCCESS in 1h 43m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/22c62dd250fa46d483ae0d9950f2184e : SUCCESS in 1h 37m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0478432f7a2b428bb3b8618bccb5fe45 : SUCCESS in 1h 06m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2c9b8b77a3f40b1a6f5cc69bdb7dc45 : SUCCESS in 1h 24m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a4f2db73cc8a4c1495353a84b033d6d2 : SUCCESS in 1h 11m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/80dd86c36f0943fcb5a0e92c8214fb69 : SUCCESS in 1h 52m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d40ca4b675774b8cb6cf0bb2fe7a2907 : FAILURE in 1h 55m 39s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb60778e00b941c3b44d28c2b14fbb7f : SUCCESS in 1h 11m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/62e94465c5334ea9869bb3d1e2cd777e : SUCCESS in 56m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/54d9ac7ded944a64b6d1199c970a3fa8 : SUCCESS in 1h 21m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b7eae8b9bd8e471faf4d892e12a1976b : TIMED_OUT in 3h 04m 15s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/e81ea1509300444c8c6fe291443b330c : FAILURE in 15m 11s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/692be08199db4d90b64e957e43206c8e : SUCCESS in 59m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/789b680bd32d460dbd1d157c3e601a7d : SUCCESS in 57m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dac25344bcb04a1ca0c2a467be7f4c7a : FAILURE in 1h 16m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ec2d92667654f69b4632998a975d173 : SUCCESS in 38m 48s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5a98fc4fb27474799b019b4794a0c82 : SUCCESS in 1h 00m 12s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/70dde7643bab494fa8922b57b4f5249f : SUCCESS in 1h 07m 56s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/da4f5a0907b4473d815b14365b6e33da : SUCCESS in 3h 08m 09s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9f81c13f4c054c6cb61c3378a110f36b : FAILURE in 51m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74aaf4f6310e42da05bdee2540cb822b0d401144","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-16 00:05:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"36587d1e841e63c1e0ad1837d6dd77d023b225b3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-16 02:02:23.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770355/2 : FAILURE in 1h 56m 32s","accounts_in_message":[],"_revision_number":2},{"id":"5e2189b4b5399a6c099b48c22da41df4e63e2526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 03:42:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/d7cb7f9558bd441b9989847700229373 : SUCCESS in 45m 10s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/cb77c71cebc942ce9a036647a257e69a : SUCCESS in 27m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/9d7f692c8a974eb3af1502b9431992b1 : SUCCESS in 1h 37m 17s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/fab72b590df74956ae2636a471bf84cd : SUCCESS in 1h 49m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/a5516c19eff34427ae6a56067e7a3fbd : SUCCESS in 1h 28m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc5f0c9adac940109a8d8a0e7ee6ecff : SUCCESS in 38m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ce481fa131480eb5fb2b732745d629 : SUCCESS in 12m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1d48cac7cc549179fcf2128094ef5b5 : SUCCESS in 24m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf348eabfcac4065876047d14aa4e41c : SUCCESS in 32m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be0899b57db14c59a70cd60a912e79ff : SUCCESS in 26m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a45a48fbc44a0ba649b835bf5079b7 : SUCCESS in 8m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ef49830c24d40e3b4c165ccad43fcc9 : SUCCESS in 4m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/79b2749687a84ebb911bc75d09994b55 : SUCCESS in 1h 02m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e3477ea4446f41d7b8af32bfd3ecebab : SUCCESS in 1h 09m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/617b80fcefb6491289a72fe5e5cf2ac3 : SUCCESS in 9m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7338173e5ec14c8a867255331a8cb711 : SUCCESS in 37m 05s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/c7bfbc250ae84b6f9d5a1dffef0f4d04 : SUCCESS in 37m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2086d97b8f3c4a1b99e653bfc93a3a7b : SUCCESS in 1h 00m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f37cde2ab2da4f279d2a5eee822ee5f6 : SUCCESS in 1h 42m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2ef4e37e80724dec9dccad8b524b1698 : SUCCESS in 1h 05m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7e19631fe9354096b3d67774feaa30d7 : SUCCESS in 1h 21m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bab8d32735174ec7a633bcb335e56226 : SUCCESS in 1h 06m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a64e6b05f3ca4d2a8a7bfb0df0732953 : SUCCESS in 1h 07m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/94f7188be05c4ee99aecfedc63d038f8 : SUCCESS in 1h 40m 30s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bef0cbaf00984b889aa9004d0972bc17 : FAILURE in 1h 31m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b032ac898da42a08df4e9b322ccbd13 : SUCCESS in 1h 08m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/34f0a0105bbb4fcb9e3995b5c112e17b : SUCCESS in 57m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/633719d2c4914b66b83b03febed4c0e1 : SUCCESS in 1h 11m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/51a3fe9baab64c228b4916080c2ce2a5 : SUCCESS in 1h 42m 07s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/ceb1207ab0ff480b9c294ea2610c0830 : FAILURE in 18m 30s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/97d86ce6227f47f284dd89990b20408d : SUCCESS in 59m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd18ceab267a442cbc2961aeb7f60831 : SUCCESS in 1h 19m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f5d2496694194b478172b44ee79e0722 : FAILURE in 1h 18m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d73be4d7da994e00b2f3f19fe827b696 : SUCCESS in 51m 37s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8f3fd13a0e944c68eba96346849645d : SUCCESS in 55m 33s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fab4b06b6d2a431ab59da74aec276e01 : SUCCESS in 1h 08m 29s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/433064dcb2644824afe17bee8e785515 : SUCCESS in 2h 22m 55s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/077187a9a9d942209f833f3c8eb2e7b7 : FAILURE in 57m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"622316150def7aa46d8673790e19b77ca44f6422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 03:43:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9449b993ee212efaf61f46bb1b077e203981af4f","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-16 04:32:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5250f4e8458d139b17f8936d18c6928030eff08a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 06:21:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/7aefdd378a9d4ff19094910036524c15 : SUCCESS in 36m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9bc18f94d9c455cbabdbe0d0df095df : SUCCESS in 11m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e90953bfe5e46fc9362602067acf96c : SUCCESS in 18m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b36b90da2f544d9b6461cc784bf10dd : SUCCESS in 22m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18d9238b727449ad9082ec6f7f99ef48 : SUCCESS in 15m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37aa675516d43f0863c125d236e5276 : SUCCESS in 12m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/de617c02cf5c4dd5a2e5e9c7bcfa6871 : SUCCESS in 4m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/4d8bd645e45142a99b4d38c19badc697 : SUCCESS in 56m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bf5a10babb6d43f983139bd0bd09315c : SUCCESS in 1h 30m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7f99e35513a4aebac9744cbcbdec11e : SUCCESS in 8m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/640540b2e9d548d1ad549665788099e3 : SUCCESS in 38m 21s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/115886a99f3b482a893243389e021634 : SUCCESS in 43m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8d83f56cfef14436ba48e11ce8b56b88 : SUCCESS in 1h 29m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cf5a96522f5b4ef1a6e85aa426630bfc : SUCCESS in 1h 02m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e2ebdee673a94d0ab886e353626a935a : SUCCESS in 1h 11m 48s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8c7499e3751d40ddaeed30077f03264c : SUCCESS in 1h 03m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5852aa72e4214fe9bcafe265212727cb : SUCCESS in 1h 04m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c3f13de8fdf1433a925ece5fd85d17ff : SUCCESS in 1h 02m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9533b146333743d1a902317767c0b01b : SUCCESS in 46m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ef23a4516074cab9af8bf763fda637e : FAILURE in 1h 55m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/07db9354b68c4e1f842eeb60daca1765 : SUCCESS in 1h 03m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97f78124b6744e15b20e6f1485f9709f : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":2},{"id":"05a0634ebd3a27763b7f9b18634f0cf0453a85e9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-16 06:51:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"87f975f9d59aa440c02883eeae90c243bbdfe4be","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-16 08:54:20.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770355/2 : FAILURE in 2h 02m 29s","accounts_in_message":[],"_revision_number":2},{"id":"38323d5d00a7ca9011d72b37cc62a2b812311af5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 09:44:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/9859656b2b3d4d928141aca33d42574d : SUCCESS in 54m 58s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/4473c759994f4698be42121dbe3cb87b : SUCCESS in 26m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/ecc019fe5632413cad215146eb092220 : SUCCESS in 1h 39m 19s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/e6c16b470c5147ce85dbbf1b23102a67 : SUCCESS in 1h 14m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/5095692d0f7143399357e9f1deef1ef4 : SUCCESS in 1h 38m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86481d8d8e064734a9fca78e86f61d56 : SUCCESS in 17m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b308457add4826b8521dfb670e0a8c : SUCCESS in 10m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/516d4e3fc330422483f9b37a2e3f8fd1 : SUCCESS in 14m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f61f271b8290482fbbf694e751a6cabf : SUCCESS in 26m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb223dade71f49ffa17861810730792e : SUCCESS in 15m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a98ed2450d4fccbbef5888de4754ad : SUCCESS in 8m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4cf03ce748f4e0a8ece44b492c531f9 : SUCCESS in 3m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/ee9c41ad12ca4c699fc2c206b85762df : SUCCESS in 1h 02m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4c28cd1cd3514f1899e454cd4000d497 : SUCCESS in 1h 18m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8533e664bf984a52b63d40d59c30c0ec : SUCCESS in 9m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8b439a74fbc14d2e8d200c00765f11b9 : SUCCESS in 44m 29s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/97762f392e694e7aa4af73ae733ab327 : SUCCESS in 1h 07m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/beebc1cec1384ea69d8ad7d60ead5d91 : SUCCESS in 57m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/aaba320c708b42bfa90a6805173fe167 : SUCCESS in 2h 31m 35s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/98b68cebf04344a0b9edda523cd1ae6d : SUCCESS in 1h 26m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/68614b94718e47c4b8854dc39199bde4 : SUCCESS in 1h 40m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed922735dccd4c288172830639c54d5d : SUCCESS in 1h 05m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/839363add94442c4877535a22222fa79 : SUCCESS in 1h 04m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e558a923b3114b5887e93e0448426072 : SUCCESS in 1h 35m 23s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0141a5f536544926ae98f357456e7d72 : FAILURE in 1h 28m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48e76038965746e3a50e1f57d317aea4 : SUCCESS in 1h 03m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8c07e523b9cd43c28c524d5ac9eac64a : SUCCESS in 49m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6fdca24e1bb7440194350842470315df : SUCCESS in 1h 19m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b686a6db3b7408fa67892f291e63caa : SUCCESS in 1h 51m 16s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/53affb8353dd497c999b59a94dec7097 : FAILURE in 14m 59s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/459c8a3310a3470f80acb15af317162c : SUCCESS in 1h 01m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f832a4e274ea4f098a21f51040941106 : SUCCESS in 48m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/897837a32b684021a29606fae3f754c5 : FAILURE in 1h 18m 41s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/64696bc7ab124aae9f5ddcd4bc953859 : SUCCESS in 37m 52s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/75752867e71e47f5a366ccd947fa619d : SUCCESS in 57m 27s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a461e41e58d475b870b13b6884606fd : SUCCESS in 1h 07m 41s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/165f8353f01140cfb5b5baef7acddd1a : SUCCESS in 2h 49m 04s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dc38f643c4d144fda5313805075d9e15 : FAILURE in 1h 09m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9886ad20c5ed0aba5c7438ca02dfa06bb1e79bb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 09:45:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"35bc28316bf9e1e72e21588ef6535c1e0b048d5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 12:13:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/ad7f841a35bd4b40b980885f0e55be5d : SUCCESS in 45m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ddb16e4e0fa4e2a9068581a6caa7070 : SUCCESS in 12m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/982fb97cc2324745bb95fa4e2d82356c : SUCCESS in 23m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0425275ea80d42a9a16bdc69b669db01 : SUCCESS in 25m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5c37a2b07f049109f77c685d80bb8a5 : SUCCESS in 14m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4403d77097d42b4bacdf56aa55f4785 : SUCCESS in 9m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9568d196c52948389911820fff8260e5 : SUCCESS in 3m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/1342a297fc014beea3e1c01aa3a4732e : SUCCESS in 54m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c991023396884eed8e35cafe97c5621a : SUCCESS in 1h 05m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d837c11af0c483e92beae0767368224 : SUCCESS in 8m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/96090f386ed749449be2badec92f73e5 : SUCCESS in 40m 51s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/acdf2524a7e54eb1bf21a7c35892ae7f : SUCCESS in 38m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d55e292261ae4ca2b4562000e913823d : SUCCESS in 57m 20s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1694cb4d0cf747c88f222697f86bd1d5 : SUCCESS in 1h 18m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/998e3a5d1c6e4e6fb786b62fbc31d8e0 : SUCCESS in 1h 03m 31s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b3d53201fb1344dfb444b7696af4a43f : SUCCESS in 1h 34m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ab087d16f0346789edb6db77b57b009 : SUCCESS in 1h 07m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a8a28a968fe54cc29276c40cdf471d27 : SUCCESS in 1h 12m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9fbc746a958c4447820d620b95fc41f9 : SUCCESS in 53m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c08a4293d3e4ac190bbbf7bae3c8d38 : SUCCESS in 2h 13m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0703b6bb80c7483dbff4c5dcfd627fff : SUCCESS in 1h 02m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83f48968f5e84bb1a52330c27e7a0cb2 : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":2},{"id":"4bac08413bca73d098405a83d9b3f680d0d8b9d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 12:13:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"137b04f6418f8a06bfb45835f30784b0a17cdf51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 12:17:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4d0311088cb4cf992dc258204626f33 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b3c4e316b934c0280ff49fced2f291b : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef37b3a39a323cb1d2a1fffe1aa466e82da3a18b","revisions":{"bfa52384e86dab057b3599746112c9f3afae3c82":{"kind":"REWORK","_number":1,"created":"2021-01-12 15:08:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/770355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/770355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/770355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/770355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/770355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/770355/1"}}},"commit":{"parents":[{"commit":"c5b4e8889ed2017dbd3adf88582c999fb0783e8b","subject":"Merge \"Switch tripleo standalone job to content provider\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5b4e8889ed2017dbd3adf88582c999fb0783e8b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:07:35.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:07:35.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bfa52384e86dab057b3599746112c9f3afae3c82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bfa52384e86dab057b3599746112c9f3afae3c82"}]},"branch":"refs/heads/stable/train"},"ef37b3a39a323cb1d2a1fffe1aa466e82da3a18b":{"kind":"REWORK","_number":2,"created":"2021-01-12 15:50:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/770355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/770355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/770355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/770355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/770355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/770355/2"}}},"commit":{"parents":[{"commit":"c5b4e8889ed2017dbd3adf88582c999fb0783e8b","subject":"Merge \"Switch tripleo standalone job to content provider\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5b4e8889ed2017dbd3adf88582c999fb0783e8b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:07:35.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:49:54.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef37b3a39a323cb1d2a1fffe1aa466e82da3a18b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef37b3a39a323cb1d2a1fffe1aa466e82da3a18b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
