)]}'
{"id":"openstack%2Fneutron~770356","triplet_id":"openstack%2Fneutron~stable%2Fstein~Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c","project":"openstack/neutron","branch":"stable/stein","topic":"oslo_lc_drop","hashtags":[],"change_id":"Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c","subject":"Dropping lower constraints testing (stable Stein)","status":"MERGED","created":"2021-01-12 15:09:09.000000000","updated":"2021-01-14 10:44:38.000000000","submitted":"2021-01-14 10:29:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770356-oslo_lc_drop","meta_rev_id":"3ce81be1467db4609414e3f5d36891e1de81fafb","_number":770356,"virtual_id_number":770356,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 10:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-01-13 07:54:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-01-13 09:39:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-01-13 09:11:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-01-13 09:39:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 15:15:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-01-12 15:18:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-13 04:37:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 07:54:00.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-13 09:11:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-01-13 09:39:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"ae62abb71bc8cd977558a6ef25578fc0f3a15893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-12 15:09:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fe411c37562de169ea840c0048be00c77a4fcab","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 15:15:47.000000000","message":"Patch Set 1:\n\nperhaps removing l-c.txt?","accounts_in_message":[],"_revision_number":1},{"id":"c5d6f9c31a112fad40638a7aa12a454099d2ba20","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-12 15:18:36.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770356/1 : FAILURE in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"140954208a8b4fcce0d2236c64865534a38ac38b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-12 15:50:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a275a249a99fae232c0f2bda76b6100c5bc2a41d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-12 16:00:42.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770356/2 : FAILURE in 8m 39s","accounts_in_message":[],"_revision_number":2},{"id":"62485a1c9b7c1e232caeb0825c0699fcd800f48a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 04:37:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/9fc46c88541b41fdb24a2b2a3e523441 : SUCCESS in 1h 07m 49s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/a76eae33452f457ea0707e4338ddc2ca : SUCCESS in 46m 49s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/9e7e860e47434a39a143fcb13e831b3a : SUCCESS in 2h 00m 32s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/fd58a1337686424ab4fe2209a9a29427 : SUCCESS in 1h 12m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/90748f8528fb44899bebb44edf39ac72 : SUCCESS in 1h 36m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/defe3de408cc43bc8781213b48059670 : SUCCESS in 27m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb07464fcb204a4a84362cac9151e439 : SUCCESS in 12m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/449edfc132f64529892d74d63b5101c6 : SUCCESS in 22m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fe5348ceeefc495bac9ab0d861f48054 : SUCCESS in 31m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f9b3538062f4fa9a2d0d37aaf0f8d5b : SUCCESS in 27m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8be9be7290bc4f1aaf3ff4b3d2c90c71 : SUCCESS in 30m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862f907788ff4057a827ba874a7ec502 : SUCCESS in 5m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0f88dd0897434069bbb083cb22b521ed : SUCCESS in 3m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c29c8e9f2f524c888de5df414f296493 : SUCCESS in 1h 29m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bf274ef1c56b4edd977d367b7fc6cee7 : SUCCESS in 53m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a51e38027ab4712b56811a21bd923f2 : SUCCESS in 1h 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee3e5ddf1099469ab4b631df368f180f : SUCCESS in 9m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/238096e47a8748158251b9db08eab29d : SUCCESS in 35m 15s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/9b223009321642039f8c88d18f37c0ed : SUCCESS in 36m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/be0e84f434dc4a7283010b85fa69182a : SUCCESS in 58m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/76026095b12948e4b2ec90ec72c0f62e : SUCCESS in 1h 34m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7a071e07af2f449b8ed88c341e6caf72 : SUCCESS in 1h 52m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/24f10cccdb964ba4bc01f61c7f375771 : SUCCESS in 1h 24m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac976b46fbd9431b94c5db6485e188ff : SUCCESS in 1h 22m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d28b0168fe5e4e818d49f17f13302ff6 : SUCCESS in 1h 00m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2b9911e468e841738396a8f471ed1f22 : SUCCESS in 1h 37m 44s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/789b722ba03647babb6a7a1b670e7692 : FAILURE in 1h 47m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aa1fd971fa284b90b32299be655fa0e4 : SUCCESS in 1h 29m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bfb036d856a44186aa7020c53e6d00e6 : SUCCESS in 49m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b34101cb461f41d8b40f0944178ccd07 : SUCCESS in 1h 58m 11s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/77bbba8798594ead810527877b692111 : SUCCESS in 1h 51m 19s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3f06fb0dc69b4fbea22893ec51bbd85a : SUCCESS in 1h 25m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7c83c87a95fd48a084aa70eb0130d070 : FAILURE in 1h 31m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76eda9b1468943fec28036f00be7721cfd103677","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-13 07:54:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6097d155ee315f2e737701ca321755ec72391f68","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-13 09:11:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9fd3ac341f6e18c39fa6c61db326ec7abe1d0985","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-13 09:39:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3f9c1b97dbb348c78609e9cd5b45ce41a7749dd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 09:39:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eee89d3c253daf7a916a6b2d2816850d516a5a54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 11:47:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/0181e307c4a1477cab3868a603f172f2 : SUCCESS in 53m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7aacdeedb6b4aee911b671011fad0e4 : SUCCESS in 13m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c60263edcc014725829d17ad7fb2e46c : SUCCESS in 24m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9c9c332eeab3465eb0d6338086b75ac9 : SUCCESS in 29m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e56e46f3ba24456a86da675f04fcaaf : RETRY_LIMIT in 2m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a74ec6a6ebf74ef5b376ff2e39ffa974 : SUCCESS in 29m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77cd32e26204703984a125834a07ceb : SUCCESS in 10m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/10ec2ee9fc7744128f279722ad999bd6 : SUCCESS in 4m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5f5cf010ef304bdaae5255c1938ad216 : SUCCESS in 1h 32m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a8892627298a41e1af78051507f44c9d : SUCCESS in 1h 04m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b4db179b861442d983524d8f63a0017 : SUCCESS in 57m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9fc11a71d0e492588ab258a3732ea4d : SUCCESS in 11m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2f21b76b71c840ae96080b7651fe2653 : SUCCESS in 40m 02s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/f7f67aee8fe84bb09c4cd75a3f6d5563 : RETRY_LIMIT in 2m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/296b67188bcf4e839fd740aa704642d8 : SUCCESS in 1h 07m 20s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5f60f7602256430b9b8cc780bfa4e0c4 : SUCCESS in 1h 28m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d5934493dc564f05af725b8ca3a9d200 : SUCCESS in 57m 51s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6645719a3b3444b49dbf691f0ff87b14 : SUCCESS in 1h 25m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b72fc09782c44ecb9a2c8a317712794 : SUCCESS in 51m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b5aab961fc604a68adaed2e701988f58 : SUCCESS in 1h 15m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a2df8bc7120e4cb88fbc7602b8afa6fb : SUCCESS in 2h 03m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8b0fc998de12443e8d221677bb9ce30b : SUCCESS in 1h 08m 46s","accounts_in_message":[],"_revision_number":2},{"id":"918f87773d35dc1062399cc8f706385fcb98460c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-13 16:31:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"223199efec75535cc10a76b5c71442038a29ae34","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-13 17:35:27.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770356/2 : FAILURE in 7m 56s","accounts_in_message":[],"_revision_number":2},{"id":"dc6a56121369f81411d56600a3a372902c6f1e98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 07:56:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/00d51a7fbb7544bb9ff79e866c6fb511 : SUCCESS in 1h 04m 47s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/3faad465f2c7451285d2f7877421b441 : SUCCESS in 33m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/b75688119eb045f09a42d1f9ac4d9cff : SUCCESS in 1h 29m 54s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/cd0f6adc88904f9195b28142f112e228 : SUCCESS in 1h 28m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/eb287f93910a4fe4a5817b37d6329881 : SUCCESS in 1h 20m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c09812c9b82401c83642587338fffa3 : SUCCESS in 42m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c308f8d61f6d48c08dd528590aa985a3 : SUCCESS in 12m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bbf95d825ba47b7b2b0bfbf304a0c2d : SUCCESS in 17m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0240d1e59af44bf994a43b66318ef901 : SUCCESS in 29m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3b8762cc0544df18cfad6d254beb631 : SUCCESS in 24m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/994f40380bc644e9987de860eb009b23 : SUCCESS in 21m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff3d9d8630142879cd76a60b9407bdc : SUCCESS in 5m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/91cf4ef919b24eb6a33abfe91c8593f0 : SUCCESS in 2m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eb8786b481c44374985d1843db51efac : SUCCESS in 1h 30m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/64dc35437977444eb7e1412805248b2e : SUCCESS in 56m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aca91135c1b94e17b9310045a827ba4e : SUCCESS in 1h 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f036eceb208a4a0aba8932b1f110e2c3 : SUCCESS in 8m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/31553e741ac94b98add3d98da5afaf0e : SUCCESS in 37m 19s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/6c21da03e48548a8973ec93f0c444ebb : SUCCESS in 40m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a24d76f779054ec59c0801ac27e9e2ef : SUCCESS in 1h 01m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9e54535197f94166baaff94b8985e02a : SUCCESS in 1h 30m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8c4069425bd94703abbd28e702abda5f : SUCCESS in 1h 18m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1a522b9b28f343c59fb8ca01517d6484 : SUCCESS in 1h 30m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1db275f8afe04bce8803d99c53609904 : SUCCESS in 1h 00m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/27def3d642e74f95b85a5b96fac7b66e : SUCCESS in 1h 08m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5d75514a4488445193ac6fd0b29c89f0 : SUCCESS in 1h 58m 46s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/972edfc8d64f46d0995a2762aeddae7b : FAILURE in 1h 40m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7599bfa28b44d43aef0f9c0e5b9ff03 : SUCCESS in 1h 20m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/de2c1f42ed4f421ba4c2d82bbf2200d2 : SUCCESS in 47m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/faa6059da58644f9a62d53484220fa59 : SUCCESS in 2h 00m 19s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d437e251deb146d69042f3b4e78b95c5 : SUCCESS in 1h 05m 07s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/73be99eea11f419da6bd66a398a65d4f : SUCCESS in 59m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3ca91239f090423fbeeda399f6b04004 : FAILURE in 52m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29306e2cf14494a8ceef260eb40624041d81217d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 07:57:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2c56808d06b6074e431af93e46236cb5e6dc84d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 10:29:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/7c4c1ab1adec40e8b1eb69f36000a28a : SUCCESS in 1h 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6499482a7641998801d2f0a71155fb : SUCCESS in 12m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36e453b08aba4e2489e6447dc2223c15 : SUCCESS in 21m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6ee186ddd0774216a5bae61b2c02e2af : SUCCESS in 32m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/204f63cd51f4402381b9b526c97409b6 : SUCCESS in 24m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0d7fed8ee674885a75487c651ddb316 : SUCCESS in 21m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c6f7ebefc543e5ab708584b15afcdb : SUCCESS in 7m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f56c36e94ae24cceb07655c9428dd9df : SUCCESS in 2m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/edfd4a3c808d4320bc951cdb9d8bc457 : SUCCESS in 1h 28m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6e845595023c4a9eaa6ebf6e602e1885 : SUCCESS in 58m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a722269a22444081a1c8c232f6df4aeb : SUCCESS in 1h 16m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94209704b42e4cef9a0947f846cbb9b9 : SUCCESS in 9m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b49a3949b40a4ca4ac26d6eb96d12eaf : SUCCESS in 44m 25s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/89e4846d6dc64516855790a37d71c652 : SUCCESS in 47m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9e328ec7e6e04c5e8367c1fd1ff4985e : SUCCESS in 1h 14m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6b3d0f17717e4e2b888357a3ad7eac0d : SUCCESS in 1h 38m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6c58cbeea9ac41f08114b76ddde443af : SUCCESS in 1h 12m 15s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8281501f7b14153b841675f2537ec0d : SUCCESS in 1h 15m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa9aa4478bfa4e8fb4927f157befda80 : SUCCESS in 1h 06m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/39f8b51c9b6c4819884dab5f9475bc25 : SUCCESS in 1h 15m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cb52f77369dc4d38a22c6d9565f5e781 : SUCCESS in 2h 18m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b02255241ead47dfaa3b783e731ed257 : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":2},{"id":"1144602d6a8e87244b0f7fc0250da3e301cd1fd8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 10:29:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3ce81be1467db4609414e3f5d36891e1de81fafb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 10:44:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0acd107f74d84aeba1a7b24994dd3a03 : SUCCESS in 1m 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbf0433ba5e14cbdb24961300c572a87 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"593587e297efda2be2787c2e25056148e2aecc88","revisions":{"d14ba06d118d130db17a67d99ec5aed92ebfc24c":{"kind":"REWORK","_number":1,"created":"2021-01-12 15:09:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/770356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/770356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/770356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/770356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/770356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/770356/1"}}},"commit":{"parents":[{"commit":"10990578ad63b1ed0c6d2dcf9a0159d104af5ef6","subject":"Merge \"Rehome api tests for propagate_uplink_status\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10990578ad63b1ed0c6d2dcf9a0159d104af5ef6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:07:35.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:08:48.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d14ba06d118d130db17a67d99ec5aed92ebfc24c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d14ba06d118d130db17a67d99ec5aed92ebfc24c"}]},"branch":"refs/heads/stable/stein"},"593587e297efda2be2787c2e25056148e2aecc88":{"kind":"REWORK","_number":2,"created":"2021-01-12 15:50:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/770356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/770356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/770356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/770356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/770356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/770356/2"}}},"commit":{"parents":[{"commit":"10990578ad63b1ed0c6d2dcf9a0159d104af5ef6","subject":"Merge \"Rehome api tests for propagate_uplink_status\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10990578ad63b1ed0c6d2dcf9a0159d104af5ef6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:07:35.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:50:31.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/593587e297efda2be2787c2e25056148e2aecc88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/593587e297efda2be2787c2e25056148e2aecc88"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
