)]}'
{"id":"openstack%2Fneutron~770358","triplet_id":"openstack%2Fneutron~stable%2Frocky~Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c","project":"openstack/neutron","branch":"stable/rocky","topic":"oslo_lc_drop","hashtags":[],"change_id":"Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c","subject":"Dropping lower constraints testing (stable Rocky)","status":"MERGED","created":"2021-01-12 15:11:15.000000000","updated":"2021-01-16 01:30:12.000000000","submitted":"2021-01-16 01:25:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770358-oslo_lc_drop","meta_rev_id":"a7d0e10d3780fc732a7567f97caf2b5e9982d89e","_number":770358,"virtual_id_number":770358,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-16 01:25:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-01-13 07:54:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-01-13 09:39:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-01-13 14:10:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-01-13 09:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 15:26:44.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-13 04:29:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 07:54:34.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-13 09:39:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-01-13 14:10:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"1f5b082934eeaf2115579c1b46402aa4bf4650cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-12 15:11:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e05fcb490d92ecdd634fbee64333308b227a35d4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-12 15:26:44.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770358/1 : FAILURE in 8m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7a52cf907f0566715c6e48144691784f85add2ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-12 15:51:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fd6366f632df4e4a7cf11c9d107188966c670d7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-12 16:01:22.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770358/2 : FAILURE in 8m 42s","accounts_in_message":[],"_revision_number":2},{"id":"efa042552e6ca96e9bc2b12389bcd63f02872da1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 04:29:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/e1a1d6ca8413404294f046cb90154c55 : SUCCESS in 49m 16s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/f10c9e372c2948aaa0c124b4d534f340 : SUCCESS in 1h 00m 02s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/75834349745d4f85b761738d52ac8465 : SUCCESS in 1h 25m 51s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/284b5a1da615438cb9f1789271297223 : SUCCESS in 1h 07m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/3234351e45e046418998d444ffd0a261 : SUCCESS in 1h 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361f25de01d040efb2b3b6a81c351b91 : SUCCESS in 11m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0ff550263d64136a418b2b798bcf770 : SUCCESS in 36m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d10ae933c2b240489d74fb5fb89722cd : SUCCESS in 29m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/07bebb4e23694947a95e9b8665b4b427 : SUCCESS in 8m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b44770eefc94386915c0135010131eb : SUCCESS in 7m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/160aa48424684acd86ac1db4e121c4c6 : SUCCESS in 1h 19m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/da8a9701457d4bf5867e5aa76e2bdb50 : SUCCESS in 48m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/badc1bde798c427391f6f9bea563ca02 : SUCCESS in 1h 09m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7476c3137a1147799817d075be114619 : SUCCESS in 8m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/89c29cda2491460e8bcd26c66893df95 : SUCCESS in 37m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/69a454bc6e344519a36b06538cc3fb87 : SUCCESS in 1h 46m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/55b4aa43a7a148a99922c1656e5e81a3 : SUCCESS in 1h 21m 01s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ebc19466e7b74367b52d7ee0daf24403 : SUCCESS in 1h 20m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e52faa1e74f43ad8e0d69af879d04b5 : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":2},{"id":"a27413d413efd147a15dc21e3e3f52da8b6c73ca","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-13 07:54:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"544fbee6072149d1e27702cb19b055afa5f02c32","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-13 09:39:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"48c56b89300483439f6973e2188f7d63d17148a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 09:40:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"628b6ab016c9d3ccd77101744bd05d369a440ad2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 11:33:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/92c8f107e0284bb4903909539ef9c881 : RETRY_LIMIT in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0e6b71ed35403a96319bca68dc6710 : SUCCESS in 13m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2690cdfa695a4ecc9f03dd845982024d : SUCCESS in 25m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b466758e9aa34871ab5b80b36ce493f0 : RETRY_LIMIT in 2m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5f7f3292ac3d49a286072654e8539ead : SUCCESS in 7m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2960b0e38d944cb8a5a00c8b8c4ab1e6 : RETRY_LIMIT in 2m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/80a47ec0c048401b972cce8f9c6a827a : SUCCESS in 1h 29m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8c9d17423f4b46dba7a5e31776af6b9e : SUCCESS in 1h 19m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c320eb5c33f40f38df1b04a5a8e85ab : SUCCESS in 1h 36m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e4b9405ecde4cc3ab0ebd2dccc829f3 : SUCCESS in 11m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a880e7c62606415fae531528107f94ad : SUCCESS in 48m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a4a4389f5862475a87c1d4ece64b78ad : SUCCESS in 1h 22m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b237927e5d9041a5ac4a57c02175cc4f : SUCCESS in 1h 22m 07s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d5d0aa2280db422f96796fd498af52b4 : RETRY_LIMIT in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b3e0eca4e8d4670b2bb5fdf299526df : SUCCESS in 31m 50s","accounts_in_message":[],"_revision_number":2},{"id":"16e6793cd2db07bac8a1931e35250d6ebac9d281","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-13 14:10:37.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e2e19e6993b05dddb8eaead5176213f56b3f3aa3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-13 14:20:07.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770358/2 : FAILURE in 8m 17s","accounts_in_message":[],"_revision_number":2},{"id":"84586eadec2ccb32fe632bd7c297ddea1f935149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 03:49:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/aa2b2b75e6074ab791c052fe6a580650 : SUCCESS in 50m 58s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/6e4276956d7e48989141de47a6d1e80d : SUCCESS in 36m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/d5842f6f13d44cb1ba2268f8c616c131 : SUCCESS in 1h 25m 03s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/8941299d59af441c91b816899691d272 : SUCCESS in 1h 14m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/5f726ea6aeec476eb9dbbd996495da83 : SUCCESS in 1h 22m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/071a63c3e6fd4b13ab7f5ddd4617ef40 : SUCCESS in 12m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15eaecc61ee84ad1990ca189d3a7754e : SUCCESS in 23m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2c776b2e61ad47ebbc2332c539a9ad82 : SUCCESS in 34m 21s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5525bc22a3564d4ba7812a2dcc4159d2 : SUCCESS in 7m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/184c10d9a93c4c2098276d44e8f53f3a : SUCCESS in 2m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d3a2460d99c440eebb287457bacfe690 : SUCCESS in 1h 14m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/daf172269e9d47dcadc0d4288ef6135d : SUCCESS in 50m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83ffc28ed6cd44c5a84b02cc3fd7f989 : SUCCESS in 1h 20m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/190c069404f54e709a9af11b7de25671 : SUCCESS in 11m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d4ce7793551e4ada800d50b663d9489c : SUCCESS in 38m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c7574546d0684e0a8ba821f8027c1b27 : SUCCESS in 1h 29m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6589257f627a449287d090f06a17ae77 : SUCCESS in 1h 16m 18s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/579b247aa03d4c838137ce8efbc0d165 : SUCCESS in 1h 29m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/823fb6ac696a48358b11e5c8a462e892 : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e7e7a47f87dc80e1fcdd240030702597740573d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 03:49:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fa858b140a573be4b66c25fefdaf5127e6e86534","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 07:06:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/c926c82e29194ba485e3e62c819cb34c : SUCCESS in 50m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/665dee8213ad44c6bd1d9563ed5b766f : SUCCESS in 11m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e2b936b21ec4ee287e167dc95c0eda7 : SUCCESS in 24m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ac1c4ac66c04b51917a7836fd05d1da : SUCCESS in 30m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/46c2373482454e288ed47ae61283e26f : SUCCESS in 6m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ba250ddf2f24493fb12e9cbf74c44c20 : SUCCESS in 3m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b5dabbf6b7b4059b50c09d588a816c4 : SUCCESS in 1h 18m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c64ef86e5b894ac4a8c0ce7f5363d978 : SUCCESS in 56m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/addda3da3722479586e2586fed98db4d : SUCCESS in 1h 16m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2454342657af4e4b9f1f9cc2d5fa732b : SUCCESS in 9m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/37d6c597a0d54fb597cf18bd575ef1a3 : FAILURE in 46m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/18ea0281d49d44bca0bdd1d76d82a93b : SUCCESS in 1h 43m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2f321652e10240158f616331e2e0b499 : SUCCESS in 1h 39m 58s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4c91f70c74c6443795e5db0a78f286af : SUCCESS in 1h 23m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e461eb7147164d918d24417b237bd953 : SUCCESS in 35m 52s","accounts_in_message":[],"_revision_number":2},{"id":"30cb87187f8c0047ba7808f2c35540f31ddb7cc9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-14 08:25:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cb373264e59de4339c1b1b0f0572a4d088f5a4fe","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-14 08:33:18.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770358/2 : FAILURE in 7m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4576afbde8aa8bd346a7b6d358b3e3b34e654cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 02:15:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/dcf5c70ee7404467be80e326fa463d2a : SUCCESS in 50m 17s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/2ab2faaa84f641169e0002cf79d1845c : SUCCESS in 30m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/6e7082ff682043c1ba3c9770ccdc5d90 : SUCCESS in 1h 02m 11s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/9735190398b943119c30b53c66a2bc98 : SUCCESS in 1h 04m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/35798884bc0e493ab757067a37dc377f : SUCCESS in 1h 22m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bddb0d2f199745b08bd5aaa508c23364 : SUCCESS in 9m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe10356512ec49938d8ebd825671d25c : SUCCESS in 17m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/77b92e0a9edd4db0b8c7651486e218bf : SUCCESS in 15m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3fe661699be445c7b6b33086784c333b : SUCCESS in 7m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a6aeeb182fd4520a73a85f7ed0e58b5 : SUCCESS in 3m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/81c242cc933a42cab66ba619eda91cf9 : SUCCESS in 1h 11m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fbadaaf3fadd49f1a21522b4fceed01f : SUCCESS in 47m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f414056a7fb4449a2ddade564b54d50 : SUCCESS in 1h 20m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39f2e7af7ebe4cbba8a6de2ceefe8c4a : SUCCESS in 9m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e07192aaadae46e4baaee0f784f89d92 : SUCCESS in 41m 00s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a4d3d292dfc64a9cb8bf422604aab8f8 : SUCCESS in 1h 14m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a281db6a93534226a037d068e6d390c1 : SUCCESS in 1h 13m 53s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f9c443884e4a4292a21b19f36c3951e3 : SUCCESS in 1h 32m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4a7a4a725d448cd96ec3810a020a0ae : SUCCESS in 34m 21s","accounts_in_message":[],"_revision_number":2},{"id":"956982de475d0971c6d194458e1f9fc15bb580b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 02:17:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac9942a9a73da95190e70284bea31a6c312c7442","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 04:01:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/973413f369c54a8a9514a187b3521454 : SUCCESS in 46m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e72fc3dc843146b5962771438cc5fe0a : SUCCESS in 10m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/168f7b2047f74402ac146aa435c663ed : SUCCESS in 25m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/98f93aefa5c948a383b39a37eac139ce : SUCCESS in 26m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7adc5128f95e44b983ae670046f9cd69 : SUCCESS in 6m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/19d4c7c781e54077b03181b15643bdab : SUCCESS in 3m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7788c26d83d1482490e5111e8d074d4e : SUCCESS in 1h 29m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/37bab53b9f01493582c2598ad2e385ab : FAILURE in 1h 00m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5d6318539754ab6ab1fa7862086dc94 : SUCCESS in 1h 15m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bec4b0855fea4c799a08dab09cc7e902 : SUCCESS in 9m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/562e792a0f274a4596d286078472810d : FAILURE in 41m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/818c16ac4bb340f8ab30d596c7ec902b : SUCCESS in 1h 38m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7eae5907c8564ecfbc4d6282b50fa03e : SUCCESS in 1h 20m 33s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5c31035793284c5991e9af876198128f : SUCCESS in 1h 14m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9bc7217a5754591bb7f9ea92268af39 : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":2},{"id":"383b8f9b9a3722ed9d9d3ab375838d1152128229","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-15 10:32:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"90061c0cc9ef327bb7c8bdc029792b0cf835bc0b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-15 10:43:37.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770358/2 : FAILURE in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2289382e62ccb677e2138494888eccc15932f03f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 21:59:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/45a994de7cb9490b9b2202c5bf927c5d : SUCCESS in 52m 38s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/74f0da263ab14eb58805089442d9cd54 : SUCCESS in 27m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/e18ce483adc3462f8aedbbd9c86bb52a : SUCCESS in 1h 21m 19s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/dde1ac2baf584cd386b072af7bcfc74e : SUCCESS in 1h 13m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/a954827b9fee467fa6fdcc79b42d68e2 : SUCCESS in 1h 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f741203f5b24c5c8a82a6189e1a09a0 : SUCCESS in 12m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e64a3e19a7be4e1c9950247e2cb9a208 : SUCCESS in 26m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3fa139b65938417bb9126e60ffec0e5b : SUCCESS in 28m 21s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2555d49e3684402fac6c556a5e8844db : SUCCESS in 6m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dcf4e372c62b424391538ae20adafe07 : SUCCESS in 3m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5a8ff273dc424f98ad803bf840f4cfbe : SUCCESS in 1h 23m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dbbbf396c7bd456e88715995276421eb : SUCCESS in 1h 07m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41f8ccf0f8b545b6bc13028b63bce8cc : SUCCESS in 1h 03m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ac10ef0e38444c299d2c81b9faca189 : SUCCESS in 8m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ccf909933a4a4c939469d5a3ede7d733 : SUCCESS in 43m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0c3cfbad21b44350888fdec0a7e56437 : SUCCESS in 1h 15m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f8ce36e7c3144fe9964bb58f8c355e9d : SUCCESS in 1h 12m 34s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e148c13bfcaf4f7d89b3b34e433295fe : SUCCESS in 1h 17m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8457b579eb5e49a1a24e39a1a199c95d : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6ad91c215dd8affa966a73b1501f0c58ee46e9ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 22:00:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"83625f4191150fcee262653a916577f87b44b28c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 01:25:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/4557796f880e4f50a13592bb07f797aa : SUCCESS in 47m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f67efb3932a7468bafcaa86731bd6321 : SUCCESS in 13m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a06c3f3bfaf749a2a7ffc1fde3429282 : SUCCESS in 26m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d7d2ebe6a5be45f59218ab5c5696d45a : SUCCESS in 29m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b814ee5dafbd4e7fa0266c8767503459 : SUCCESS in 9m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/474b7c5af4ea4235927c2aef3c2b1fa0 : SUCCESS in 4m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ce4a96080a53446b92cb21f5963b3a72 : SUCCESS in 1h 16m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/da928c197c1042cc810eab2ae2af7f2d : SUCCESS in 1h 12m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/458ada5f29b9419aaa1222ee16226c8e : SUCCESS in 1h 20m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8812d793fffb4de59311a2031ee55416 : SUCCESS in 8m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6e91f9210ebb480894e43eadf01aea37 : SUCCESS in 41m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7e1f5888f94e4722b8b5b881a730d84f : SUCCESS in 1h 26m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5b165de907ec4edb9e6c53e6b83c3fbf : SUCCESS in 1h 19m 42s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/be13aed85c0a4d378f3967fdaecf8882 : SUCCESS in 52m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f66a9c56cae14e5f9579d1bca5c7a773 : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":2},{"id":"ce991cc40eb07f33e7baba362902a73009d501ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 01:25:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a7d0e10d3780fc732a7567f97caf2b5e9982d89e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 01:30:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d9679584420418cb4f4cfcccd9b2a5f : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9a945cfc79bf0003c09595876a5aaf80f01cd406","revisions":{"45ee49402e1f07f324cd4571abfa76c4ce58ab43":{"kind":"REWORK","_number":1,"created":"2021-01-12 15:11:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/58/770358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/770358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/770358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/770358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/770358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/770358/1"}}},"commit":{"parents":[{"commit":"8d3155410f047709315f977dc0501157ad4303d6","subject":"Rehome api tests for propagate_uplink_status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d3155410f047709315f977dc0501157ad4303d6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:10:42.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:10:42.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Rocky)","message":"Dropping lower constraints testing (stable Rocky)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45ee49402e1f07f324cd4571abfa76c4ce58ab43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45ee49402e1f07f324cd4571abfa76c4ce58ab43"}]},"branch":"refs/heads/stable/rocky"},"9a945cfc79bf0003c09595876a5aaf80f01cd406":{"kind":"REWORK","_number":2,"created":"2021-01-12 15:51:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/58/770358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/770358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/770358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/770358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/770358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/770358/2"}}},"commit":{"parents":[{"commit":"8d3155410f047709315f977dc0501157ad4303d6","subject":"Rehome api tests for propagate_uplink_status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d3155410f047709315f977dc0501157ad4303d6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:10:42.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-12 15:51:21.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Rocky)","message":"Dropping lower constraints testing (stable Rocky)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: Ibf9b20b993ab59b64bd4c954cfcd34f14e0e078c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a945cfc79bf0003c09595876a5aaf80f01cd406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a945cfc79bf0003c09595876a5aaf80f01cd406"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
