)]}'
{"id":"openstack%2Fneutron~770935","triplet_id":"openstack%2Fneutron~master~Iffdc0373ed72aea1320155ea8bec93dce797f27c","project":"openstack/neutron","branch":"master","topic":"bug/1911153","hashtags":[],"change_id":"Iffdc0373ed72aea1320155ea8bec93dce797f27c","subject":"[FT] Reduce \"test_walk_versions\" upgrade executions","status":"MERGED","created":"2021-01-15 14:00:29.000000000","updated":"2021-01-26 14:11:57.000000000","submitted":"2021-01-19 11:57:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"770935-bug/1911153","meta_rev_id":"ad5f64dae186a8685bd348660aa7c4eb22921f67","_number":770935,"virtual_id_number":770935,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-19 11:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-19 09:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 11:22:27.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-19 09:37:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 15:56:58.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-16 01:38:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 05:35:13.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-01-19 09:37:13.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-01-26 06:49:04.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"daadf652f461062e89fabce34b2660579f40424c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-15 14:00:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"472c30c5df6d85aaa7ee723c09dc08bb57009112","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-15 15:56:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/770935/1 : SUCCESS in 1h 55m 06s","accounts_in_message":[],"_revision_number":1},{"id":"17f700dce796aeb034ad0197752b2c7d53e2fc83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 01:38:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ce2981b1fc6541adba6113629e8ec044 : SUCCESS in 38m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/282488f0a58040db8630557289d2299e : SUCCESS in 33m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bdb030654bbe48b781133421e42ff742 : SUCCESS in 1h 54m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/097e278c5c2b42bea7c5367791332555 : SUCCESS in 1h 34m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2db13b614bfe44d68297a0fc5351feaa : SUCCESS in 1h 24m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6769d67003b447c9bef7a043b44d5a00 : SUCCESS in 1h 31m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/634c06ca42a24a5e816ff508a4f05649 : SUCCESS in 35m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdb4a72083a64137bc2762a25c09f6b3 : SUCCESS in 27m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00cb16a5e9214c37b0c7e6bd4136156a : SUCCESS in 13m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2df156b2f1b14406a532664a47d6a9ad : SUCCESS in 27m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b0b0c1d8b81418ba3d30826d0e2ef1e : SUCCESS in 28m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a07426dbdf9940a39872b35fc4a6a3f0 : SUCCESS in 24m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80ec62e79cb48f0be1538a1e50013b7 : SUCCESS in 8m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/299f6b46bc574079a5ee105032bc25fe : SUCCESS in 1h 02m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/55631d6fc79c4f21936010ca34cb7ab1 : SUCCESS in 1h 11m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f9f5fc48feaa4274a49a74bb91065378 : SUCCESS in 1h 34m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8bc8aa4f1a244c48dc0b8894ea06e3c : SUCCESS in 55m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fd8e8b1cf56042aeb9a626ad0a5253e8 : SUCCESS in 1h 04m 23s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/1a8aea5ab9784d848d0354bfad286561 : FAILURE in 58m 59s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/61bfd9246a054ff395e249ebf267d2b3 : SUCCESS in 1h 22m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/67e2050003fd46ef9e93363a411f43a0 : FAILURE in 1h 27m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/64f0ed3a59ce47149c464801322add65 : SUCCESS in 1h 15m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dc6896c04f6b40da99ca20103fb6afb3 : SUCCESS in 2h 05m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5e7108116df4257a501b538f116079a : SUCCESS in 1h 00m 28s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5943cc04803f48d39bbf70ed4d2a169c : SUCCESS in 1h 05m 00s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/38e86098773f4c49ad0d5510e1b2046d : SUCCESS in 1h 20m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ab3a3676bb1b431c862579c54d05c614 : SUCCESS in 50m 15s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/69870fa67ab54775b202fb32b04e7ae6 : FAILURE in 58m 10s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/961e7d9782684ccb975abfc4a0c30b8e : SUCCESS in 1h 16m 54s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6b52aabeb6cc4efd933a7b1c11f696bd : SUCCESS in 3h 55m 17s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/33f285e3c07d4792b0cb7825a8a21375 : SUCCESS in 1h 20m 36s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8ac26c6f263f4c1ca51cf9ec485336dd : SUCCESS in 1h 43m 52s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/49a3cad0e0724b5d901ebfac3fdff1d3 : SUCCESS in 2h 11m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f7ae122eb26e4af88035a5559fbd5d9a : FAILURE in 1h 43m 30s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/164a2fd66b7245318cb50bed8c663f6b : FAILURE in 1h 39m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9df2426ba3d8117c107c9be9168c6b6e63cdf8fa","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-18 05:35:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nSeems parameter can be removed","accounts_in_message":[],"_revision_number":1},{"id":"f0e2b1be315ab5d22190d321bab6f63bd2148606","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-18 09:53:11.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"09e9f4ec94c0d2fd89bc2d5448903aa6f2a55079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-18 11:04:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"daebdebec5cadc7ea427df3534809c86ab226725","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-18 11:22:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e5722c96d2412c7af2a602a0c3dd644524f6ba5d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-18 13:05:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/770935/2 : SUCCESS in 1h 59m 45s","accounts_in_message":[],"_revision_number":2},{"id":"cab969c69cbdae453b3ba19c2fc7fe810138cd9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 15:26:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2a1293590e0c4584b4e822ed91e1281a : SUCCESS in 49m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f47716b77cb54cc8bfd7e8c753bac33f : SUCCESS in 40m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/78d81ac8389546f4a968289567b05cd1 : SUCCESS in 1h 12m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/68867993bbff48308dc6fe0e681d768f : SUCCESS in 2h 04m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12fb9cca9c154339916725610768a4c3 : FAILURE in 1h 36m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/024d493980944c5b8372353c7fa57aca : SUCCESS in 1h 24m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e2b72daca13449db5944d3e0cb7c382 : SUCCESS in 32m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fade3467f4234ab5a4ea858ead34a3c4 : SUCCESS in 29m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfec8ee4910641a88078640b165662db : SUCCESS in 12m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb0b2b9d348f4c399a9c2d030696283f : SUCCESS in 25m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfd6b2ed70f34edfacaf2036b849341d : SUCCESS in 23m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48a8f5c0b6ed422d9bc4fd05795363ea : SUCCESS in 26m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328c9f3ddf804f4c9964dba73243d724 : SUCCESS in 10m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea97cfd5e15a46c28e498455ae168bdc : FAILURE in 1h 21m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea5f594d97414cf0a2396266c9f5d36a : SUCCESS in 1h 08m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b97ab1e1af11409f827bc43a8cd99b37 : SUCCESS in 1h 50m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f9435daad844fbd99c71c110af23613 : SUCCESS in 1h 19m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8c90586a48fa434b928209ba4d47a5f7 : SUCCESS in 1h 01m 22s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/d73b3d074b25401f811a48bd3671d8b9 : FAILURE in 51m 35s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/38a994fd19f24fda9e5ac79446b3894f : SUCCESS in 1h 33m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f24791b5708e4a9e9b21c35774757a71 : FAILURE in 2h 08m 25s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d4acd29411fb48a9852cffa722707601 : SUCCESS in 1h 11m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/88db030c949e409498fde49a2726d19f : SUCCESS in 2h 13m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c606b804b5c487897065a78ae6bf2ab : SUCCESS in 59m 25s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8f543348fb4148b8aa9ea933f5048908 : SUCCESS in 1h 06m 03s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/38c4517608d248c180d2d785d24dff17 : SUCCESS in 1h 06m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/91a7f71b299b4b538237a80cbbaa66a9 : SUCCESS in 41m 16s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/db82cfb16ca040ac8b42ace1ecddc5a7 : FAILURE in 1h 27m 17s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9f543d63ad734a9e94012f69add921c8 : SUCCESS in 1h 24m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d59bd59860a6414799d17b64e6ba9975 : SUCCESS in 3h 43m 44s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f1f68d0facc945fc8f4a52849134b61c : SUCCESS in 1h 24m 55s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c0f2cbca9d04d71bbdaa212e29b6a95 : SUCCESS in 1h 40m 11s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c6059c2eb44f43e88256dfd1ff0157a3 : SUCCESS in 2h 04m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/08a22cb20cee42899f0f00d0ba62a7e7 : FAILURE in 1h 43m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2d8ffbf1aae24b33889c7cb38b56e218 : FAILURE in 1h 48m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed0bca8710d7de4df6d3d5558ebff2cc868abd92","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-18 16:47:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7954bf5f0c342811568a47b9e99a5a38087b2dbd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-18 20:45:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/770935/2 : SUCCESS in 2h 05m 49s","accounts_in_message":[],"_revision_number":2},{"id":"38a30e423ece7db8682d47161d5dc153b1af5f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 03:05:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2201f8511a424c96a21e72b4c4ce0174 : SUCCESS in 57m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ffe1950b9464d118830ee196945d10b : SUCCESS in 43m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d77c361609ae49eab455bbfe799bc065 : SUCCESS in 1h 27m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d6484975cf674fb59074da0cdb0eaa7d : SUCCESS in 1h 30m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c83a7729aa5e470389fa3a2e5f9fbd8b : SUCCESS in 1h 24m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/da0daf18ef294cb486adb5594c35bca5 : SUCCESS in 1h 49m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8467de75b334a82adbd1eaa43c62498 : SUCCESS in 34m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae0f4c3c130a4f8b9e18e6e599d20a65 : SUCCESS in 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e3e860d632148a09bd3dc8a5b6b3191 : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1064f7211964fb1a81f246db051b665 : SUCCESS in 25m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb7084af131a4979bd005c1c9f769632 : SUCCESS in 29m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8280715e945c4f099a6d18fdf80e80f7 : SUCCESS in 26m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b33dfdf097b4790a4dfa9af033047f8 : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/474b3616d826413da37aab0e6b74bfbf : SUCCESS in 1h 07m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6269adb549b2493889edae79c49c922b : SUCCESS in 1h 06m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ed5c2dc1157d4dc298238f5e3ff0ef3f : SUCCESS in 1h 48m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/93a88dc99b33416e86297a2b6ad442b9 : SUCCESS in 58m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/85631d81713e4cfea6b1fc3fbac20c84 : SUCCESS in 54m 27s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/8b51828e77654afc81cee737b7e434bb : FAILURE in 55m 14s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e369f2337ce84082a018abb56da9827e : SUCCESS in 1h 24m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ff140c5e2728432ab333b4a917014431 : FAILURE in 1h 27m 45s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c0f9c78106bd4980a43c05c309ac5ea0 : SUCCESS in 1h 18m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e187ac601476440e97b7a7611d86c859 : SUCCESS in 1h 57m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f53022f51a8414ea1b5a0a89b83aa7d : SUCCESS in 51m 30s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/695a17d537c342d0974a79f9205afaa2 : SUCCESS in 59m 28s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d5c67c943fde40418be2565eb764d1da : SUCCESS in 1h 11m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9763ad8faf9c43cdb2b6757ae69b2f75 : SUCCESS in 40m 00s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/288301bd555540e7b9dd558f8b05f584 : FAILURE in 1h 29m 06s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a020a61a7b074c22ba454ad401b17a48 : SUCCESS in 1h 23m 16s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b666ff2fa3a44160be3eaedc9702e26f : SUCCESS in 3h 23m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cf59913ce9944d8aa34814502fa38f56 : SUCCESS in 1h 16m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/653b98f095744aba91eaf22c6e3d6323 : SUCCESS in 2h 12m 03s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/89f3a77771014cc49b00e25812afa7f0 : SUCCESS in 2h 00m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/87ecef10d0904413bd488b400531b9e2 : FAILURE in 1h 42m 27s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b0ae0df3b4934f448a5bb39245392e5d : FAILURE in 2h 00m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cadef86d78b0ebf519a7be46adce31a90a4a239d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-19 09:37:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"35e856e0ac6c9b33f4e400c0fdf146561dfeb27c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 09:39:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"becb19047ab9c9d7ca2f8ac449a61ddb899ebbad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:57:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/31c33104ea7347fbb9201467d15f3664 : SUCCESS in 56m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/75f26b63a13f48b8a56926cfe5ce87a7 : SUCCESS in 1h 39m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d0161d58b41a4797bfffccbf76aaf5b6 : SUCCESS in 1h 53m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6f055aff87c34ce5acdf6cb419536949 : SUCCESS in 1h 34m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6cadcfe2ac464e568ec809f5680ad7c8 : SUCCESS in 1h 32m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c9d87cca27e40f3af2c6eb68bdc975d : SUCCESS in 31m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c30197ee4c74a4f84d4e26e1d1c62fe : SUCCESS in 13m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/655bc23a77e047629277af7af0c6f666 : SUCCESS in 24m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fbbdaaf711743d6890396bd62ff8240 : SUCCESS in 28m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3861a10a774136bdc57146df978261 : SUCCESS in 9m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac15de3c2936473cb66c9081558b2121 : SUCCESS in 1h 13m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/427ee2cc00c040f0b0cc21e172876fcc : SUCCESS in 1h 08m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5a5cadf43977467ab6bf6b997089eb4d : SUCCESS in 1h 22m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/50a47af1b5a144f389aa196fe5f62e3e : SUCCESS in 1h 03m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1d5340127214411a9220a206f7deefef : SUCCESS in 57m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/135396b2774844aaa42c67775a98a960 : SUCCESS in 2h 07m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a186c7d4c47140b8956d70b6751ce757 : SUCCESS in 1h 23m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3ab10f5b51fd4c9eac490cb31e2d33bd : SUCCESS in 1h 26m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fdd3ee39e72a490098fff2146967ba80 : SUCCESS in 38m 46s","accounts_in_message":[],"_revision_number":2},{"id":"e3924a1c118be32acb7bbccb3a98babe7a6ba4d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:57:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9b2d4714c6ef3e5af9b260529ded3172e2a60000","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:59:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75dbdba4b1264985a2448403cfa31b9b : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f8edf19751e37af61cad8a0ab249e60ea263e352","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-26 06:49:04.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/35/770935/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ad5f64dae186a8685bd348660aa7c4eb22921f67","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-26 14:11:57.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/35/770935/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c84ae9762c2bae84a3f5f20cfb01b8ef00f46118","revisions":{"c612e684353b424df63bc9583fc63a7814b2f3cb":{"kind":"REWORK","_number":1,"created":"2021-01-15 14:00:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/35/770935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/770935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/770935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/770935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/770935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/770935/1"}}},"commit":{"parents":[{"commit":"64598ed51aee9d42af22275b226bf661ae58fa96","subject":"[FT] Remove unneeded DB migration tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64598ed51aee9d42af22275b226bf661ae58fa96"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-15 13:52:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-15 13:52:08.000000000","tz":0},"subject":"[FT] Reduce \"test_walk_versions\" upgrade executions","message":"[FT] Reduce \"test_walk_versions\" upgrade executions\n\n\"_TestWalkMigrations.test_walk_versions\" is inherited in other classes\nto check a specific DB revision upgrade. The goal is to, from the\nprevious DB revision, create a testing condition and then upgrade to\nthe related DB revision. If the upgrade is correct, the test should\npass.\n\nThis patch reduces the number of Neutron DB upgrade operations to\njust three:\n- Upgrade the DB to the previous DB revision to check.\n- Upgrade the DB to the related DB revision, after setting the testing\n  conditions.\n- A final upgrade up to the latest DB revision.\n\nIn a testing deployment, this reduces the execution time from 60 seconds\nto 35 seconds, without reducing the testing scope.\n\nChange-Id: Iffdc0373ed72aea1320155ea8bec93dce797f27c\nRelated-Bug: #1911153\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c612e684353b424df63bc9583fc63a7814b2f3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c612e684353b424df63bc9583fc63a7814b2f3cb"}]},"branch":"refs/heads/master"},"c84ae9762c2bae84a3f5f20cfb01b8ef00f46118":{"kind":"REWORK","_number":2,"created":"2021-01-18 11:04:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/35/770935/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/770935/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/770935/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/770935/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/770935/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/770935/2"}}},"commit":{"parents":[{"commit":"64598ed51aee9d42af22275b226bf661ae58fa96","subject":"[FT] Remove unneeded DB migration tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64598ed51aee9d42af22275b226bf661ae58fa96"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-15 13:52:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-18 11:03:56.000000000","tz":0},"subject":"[FT] Reduce \"test_walk_versions\" upgrade executions","message":"[FT] Reduce \"test_walk_versions\" upgrade executions\n\n\"_TestWalkMigrations.test_walk_versions\" is inherited in other classes\nto check a specific DB revision upgrade. The goal is to, from the\nprevious DB revision, create a testing condition and then upgrade to\nthe related DB revision. If the upgrade is correct, the test should\npass.\n\nThis patch reduces the number of Neutron DB upgrade operations to\njust three:\n- Upgrade the DB to the previous DB revision to check.\n- Upgrade the DB to the related DB revision, after setting the testing\n  conditions.\n- A final upgrade up to the latest DB revision.\n\nIn a testing deployment, this reduces the execution time from 60 seconds\nto 35 seconds, without reducing the testing scope.\n\nChange-Id: Iffdc0373ed72aea1320155ea8bec93dce797f27c\nRelated-Bug: #1911153\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c84ae9762c2bae84a3f5f20cfb01b8ef00f46118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c84ae9762c2bae84a3f5f20cfb01b8ef00f46118"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
