)]}'
{"id":"openstack%2Fneutron~770940","triplet_id":"openstack%2Fneutron~master~Ie02f9d40716af1b3b4efc3e9340acab7f9af113e","project":"openstack/neutron","branch":"master","topic":"improve-neutron-ci","hashtags":[],"change_id":"Ie02f9d40716af1b3b4efc3e9340acab7f9af113e","subject":"Add resource_provider_bandwidths for computes","status":"MERGED","created":"2021-01-15 14:31:58.000000000","updated":"2021-01-19 11:56:24.000000000","submitted":"2021-01-19 11:49:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"770940-improve-neutron-ci","meta_rev_id":"667fbc974037e1e2fb370147c898047139ca836e","_number":770940,"virtual_id_number":770940,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-19 11:49:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-01-19 09:18:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-19 04:47:04.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-01-19 09:18:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 14:45:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-01-15 15:06:43.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-01-15 17:37:45.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-16 02:35:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 11:10:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"93b4810a997c70a65d9205dca252715d1c2c3fb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-15 14:31:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c47581ef1215ba28a8929e4546c355afad81e24","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-15 14:45:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c52839c311be7829527ef99bc2c9e8127a3c70cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-15 15:47:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f5ea5bf367724caf98d76d364be74c2a91d463d2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-15 15:49:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f94e74c7bab56c6d667d64a3a43f2eeb92db90bc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-15 17:37:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/770940/2 : SUCCESS in 1h 49m 25s","accounts_in_message":[],"_revision_number":2},{"id":"735cb87e1e6060442dc8bd8185118c5a487a91ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 02:35:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a283e9221876438189d6d48b4a2bcaff : SUCCESS in 55m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/badd1a8092424f97bbe1fb18bac0ded9 : FAILURE in 44m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7525ea89996b4157a146f927c71a5c16 : SUCCESS in 1h 53m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/43ce838d65984c80afcbfd81f96ed225 : SUCCESS in 1h 22m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05c065eb21b047e6b3e8db2d8eaccb00 : SUCCESS in 1h 30m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9fcca1a9a86142ea96cc13bc18a287ac : SUCCESS in 1h 46m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e038d33db4d44fc816a3433d1871d11 : SUCCESS in 32m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffb5cc1994a148b782df141c39297ecb : SUCCESS in 30m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa18effd8fda4c9dae6bdce839f64bc9 : SUCCESS in 15m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ae06f8b51434d50a233b4d37c172d5e : SUCCESS in 38m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e99a9c80f0ad4a3eac88016fb64d3c33 : SUCCESS in 27m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cfca38c045a4280be8cf3fef5e9f514 : SUCCESS in 28m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5459eb61ac2407c897b32a0ffcf7719 : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1bd0dc892ac24ef59956698b94540067 : SUCCESS in 1h 16m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3f89ed6d15144839038c61816ab54fa : SUCCESS in 1h 12m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/544ca83ef2e3427b90571191bd085b1c : SUCCESS in 2h 11m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf31ca56954c424caf595eca8745e20f : SUCCESS in 59m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5778da517e364b23ab3568a8a8385b82 : SUCCESS in 55m 33s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/649a6bb9444441f7ab45102dfb8b03c7 : FAILURE in 49m 22s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/15199df5726640678f8ae4d73153d8e2 : SUCCESS in 1h 19m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e7d1594df17f4de8ab3a014e016ade3f : SUCCESS in 1h 19m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5ee731fec42544e284a783639b1bf601 : SUCCESS in 1h 10m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/49c12c998c0745b281d41afdcc7654bf : SUCCESS in 2h 13m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac6be50f2159405e80649a381fb34fd8 : SUCCESS in 55m 47s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/27094359cc6141c09895351528eb03d1 : SUCCESS in 1h 17m 59s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0270467ce6ad4bd1bc6df4c9483525cc : SUCCESS in 1h 18m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ca54b642d15a4df581e06fee526edecf : SUCCESS in 34m 11s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/c419f3c0e88e4a82a57228b377953ab3 : FAILURE in 1h 01m 29s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1b4fc8cd063946b2a0e2e3df0048c755 : SUCCESS in 1h 07m 16s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/649c18f59ccc4f3e8649432ab7daba5c : SUCCESS in 3h 40m 17s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/325ad688785147cdb80fdcd39922bcf2 : SUCCESS in 1h 11m 19s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ae8e960e263448c091e681747f4756b3 : SUCCESS in 1h 36m 13s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a66611b95ff34ccaa9496a27fa5df221 : SUCCESS in 2h 05m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b1acf8ea4ea44059b6a5f0ef5bdc5292 : FAILURE in 1h 51m 33s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/162fea5479964e9bac81497b4c1a1d73 : FAILURE in 1h 49m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ffa494291b3dc1540523dbca8e246864448d1d22","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-18 05:23:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"28574e1ee76e2c05480619eecba4db2f1876cc35","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-18 11:10:18.000000000","message":"Patch Set 2: Code-Review-1\n\nI\u0027m ok with that patch but do we have any bug related to that issue? If yes, can You link to it in the commit message and add TODO in job\u0027s definition file to reenable those tests when that bug will be fixed?\nIf there is no bug for that, can You open new one and do the same?","accounts_in_message":[],"_revision_number":2},{"id":"33d0331d914938cafbefbed5e592ea2b767d2d2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-18 12:12:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee4a11c1aa5b4db176e99f0a6f4316948e824805","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-18 12:13:32.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e I\u0027m ok with that patch but do we have any bug related to that issue? If yes, can You link to it in the commit message and add TODO in job\u0027s definition file to reenable those tests when that bug will be fixed?\n\u003e If there is no bug for that, can You open new one and do the same?\n\nLet\u0027s do another try, during the search for failures, I realized that computes haven\u0027t got the same config automatically as the controllers\u0027 agents.","accounts_in_message":[],"_revision_number":3},{"id":"ba7ad282bdba3c6ee255f93f6691d23694abb95e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-18 14:07:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/770940/3 : SUCCESS in 1h 54m 56s","accounts_in_message":[],"_revision_number":3},{"id":"a36623651c3901895516ab9a39be53b364e3a4fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:01:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/625548300931448b85c50bfbed90eb7a : SUCCESS in 52m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/394ce03a7ff44f7385b133bab754ebd9 : SUCCESS in 37m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/083a05d654b044989eba4216d7b9ca8f : SUCCESS in 1h 36m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fa477e10b0f54dad8c7512a143256514 : SUCCESS in 1h 23m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a2ee9cfac5dd4529880513d038b4027c : SUCCESS in 1h 24m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6a38d543431c4c6cbbacf166702eab42 : SUCCESS in 1h 32m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/689d8410cbd249ec9da5223d6fc22759 : SUCCESS in 33m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab678a3804d54dfa8215a9f223e968a7 : SUCCESS in 32m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4d7acc396364b4a9aaddd847b03c411 : SUCCESS in 12m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c6518130afe44ebb98738b31ee32793 : SUCCESS in 28m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d407840687e400aa0fff5ee181fd182 : SUCCESS in 24m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7962107aa3e747659033160ee4b1e9d7 : SUCCESS in 25m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7afa50cbfb484a4ab43f80ed76d84d07 : SUCCESS in 8m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a9bff93657a4e6d9827219a3f87cc01 : SUCCESS in 1h 10m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/265e9d87ae9943c880bd5787008ddf72 : SUCCESS in 1h 07m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/96dd1c1d0db1463b98b24a7ee1530ef7 : SUCCESS in 1h 34m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b0527748a304cddb94b3eddf32f9982 : SUCCESS in 52m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8cbed7d0d26a4329a80022d0afbc5dfa : SUCCESS in 58m 23s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/dbdbd51b23e24a0ead09cb4e21213537 : FAILURE in 1h 05m 41s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1a96577664954257b0de2e16370f48e2 : SUCCESS in 1h 22m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eab0463819894e7e9cb8b330714a4a1b : SUCCESS in 1h 30m 51s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2cde1bd0d8e5402f98ec270638b08a55 : SUCCESS in 2h 02m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9680be2b1f314594bc11833b7f2d883a : SUCCESS in 2h 14m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd20d5fab58b450c959f1a7701b293d7 : SUCCESS in 59m 46s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/485c903dae8449f4a463e8ee1ef20f61 : SUCCESS in 1h 14m 41s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7930b44fa3534fbca425ddd4a0dd01cd : SUCCESS in 1h 08m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c470b90086284029ad09ceadf8d61bb4 : SUCCESS in 35m 00s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/d5c101908b324279beef954b317b7dea : FAILURE in 1h 04m 15s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b49be6b3ca9b4995991f684fe3363c6d : SUCCESS in 1h 18m 27s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/70ee4e7976a1403b9abacbddec545d91 : SUCCESS in 3h 32m 17s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a713260e91eb4aac90c9a619a70f45bd : SUCCESS in 1h 27m 26s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/48245bbd44f9466fac74ad7ad878eb31 : SUCCESS in 1h 31m 41s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4bd284b0bb04417ea0e1f3476b85993e : SUCCESS in 2h 08m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/846977660a6b4ab9811aa96717b83701 : FAILURE in 1h 49m 21s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9cc4355c72124b63a17763aadfc826c4 : FAILURE in 1h 51m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"13c0caf3de881fc2695b1a6f201868d093304cd6","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-19 04:46:57.000000000","message":"Patch Set 3:\n\nTests are passing now, thanks!","accounts_in_message":[],"_revision_number":3},{"id":"3243f864f81dc64152ebfb9cb8981b216f784a8f","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-19 04:47:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7a1117fe0a2955c26ae91259746695788dd89aa3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-19 09:18:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2b7e79b21d32a60f429cb4cf22a4e5f6710acad6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 09:19:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1070bef7860168a2314f4961c036903f1202ff3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:49:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/49aee1dc57444cf8b4144ad511d999f4 : SUCCESS in 57m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ed29c8d0b69b4b3a868afc10841bbbb6 : SUCCESS in 1h 26m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ce36c411d6d74152844a5bc3cee34447 : SUCCESS in 1h 30m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/36039ae07adb4984903bf550f623b459 : SUCCESS in 1h 44m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f9e2192b0c4245689135e9ce76307382 : SUCCESS in 1h 43m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa8a8ff65b1a4695ad8670dcdfc9e67d : SUCCESS in 34m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb59476bd089494c8aa870c3be85217a : SUCCESS in 12m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/084d794348eb4fee88aa6955f2af590e : SUCCESS in 29m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e450851e0b84581b920994011959b50 : SUCCESS in 25m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3a400488d94c3f939e6fa2f104c749 : SUCCESS in 9m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77481c38227b489da86e009d3e6cbe10 : SUCCESS in 1h 03m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6479c9cdc76a4b65986c9dd975505b82 : SUCCESS in 1h 19m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c34ef64f968142fab57f406758e5f9e7 : SUCCESS in 1h 20m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98fd8094d1fb47e8bcfd345f150d5932 : SUCCESS in 1h 02m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/afe6b37ffe564a00a84ed1d87a9c3534 : SUCCESS in 1h 07m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a3e035316084ee7b57a90cc233a4db5 : SUCCESS in 2h 22m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0b46b12beb84e14bc367accbc8ef22f : SUCCESS in 1h 06m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ae6108852bdf43c38364a1990916844c : SUCCESS in 1h 10m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/afabfb2125364580b7d53f36b9bac7af : SUCCESS in 37m 12s","accounts_in_message":[],"_revision_number":3},{"id":"1c11881cb0ddad3167fc8162d4074faa60cbe50e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:49:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"667fbc974037e1e2fb370147c898047139ca836e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:56:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288d278fbd14486db66fce29fc7304bd : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7826661d6594e593e6f5c86753fac3e40ae5bbf7","revisions":{"22b6c527a27833f1ee4c49e143257cb340f0ade3":{"kind":"REWORK","_number":1,"created":"2021-01-15 14:31:58.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/40/770940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/770940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/770940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/770940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/770940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/770940/1"}}},"commit":{"parents":[{"commit":"1683ada81fc983bcdca1275ce52b54b4a5854f1c","subject":"Merge \"[FT] Remove unneeded DB migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1683ada81fc983bcdca1275ce52b54b4a5854f1c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-15 14:29:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-15 14:29:02.000000000","tz":60},"subject":"Skip QoS minbw tests on neutron-tempest-dvr-ha-multinode-full job","message":"Skip QoS minbw tests on neutron-tempest-dvr-ha-multinode-full job\n\ntest_migrate_with_qos_min_bw_allocation and\ntest_resize_with_qos_min_bw_allocation are failing on the job\nneutron-tempest-dvr-ha-multinode-full, so disable them to keep the job\npass.\n\nChange-Id: Ie02f9d40716af1b3b4efc3e9340acab7f9af113e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22b6c527a27833f1ee4c49e143257cb340f0ade3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22b6c527a27833f1ee4c49e143257cb340f0ade3"}]},"branch":"refs/heads/master"},"35f1b37d36b12179804095ae9af884929c1cf868":{"kind":"REWORK","_number":2,"created":"2021-01-15 15:47:27.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/40/770940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/770940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/770940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/770940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/770940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/770940/2"}}},"commit":{"parents":[{"commit":"1683ada81fc983bcdca1275ce52b54b4a5854f1c","subject":"Merge \"[FT] Remove unneeded DB migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1683ada81fc983bcdca1275ce52b54b4a5854f1c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-15 14:29:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-15 15:45:48.000000000","tz":60},"subject":"Skip QoS minbw tests on neutron-tempest-dvr-ha-multinode-full job","message":"Skip QoS minbw tests on neutron-tempest-dvr-ha-multinode-full job\n\ntest_migrate_with_qos_min_bw_allocation and\ntest_resize_with_qos_min_bw_allocation are failing on the job\nneutron-tempest-dvr-ha-multinode-full, so disable the whole test class\nMinBwAllocationPlacementTest, as we have coverage for it in other test\njobs.\n\nChange-Id: Ie02f9d40716af1b3b4efc3e9340acab7f9af113e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35f1b37d36b12179804095ae9af884929c1cf868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35f1b37d36b12179804095ae9af884929c1cf868"}]},"branch":"refs/heads/master"},"7826661d6594e593e6f5c86753fac3e40ae5bbf7":{"kind":"REWORK","_number":3,"created":"2021-01-18 12:12:05.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/40/770940/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/770940/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/770940/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/770940/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/770940/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/770940/3"}}},"commit":{"parents":[{"commit":"1683ada81fc983bcdca1275ce52b54b4a5854f1c","subject":"Merge \"[FT] Remove unneeded DB migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1683ada81fc983bcdca1275ce52b54b4a5854f1c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-15 14:29:02.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-18 12:09:06.000000000","tz":60},"subject":"Add resource_provider_bandwidths for computes","message":"Add resource_provider_bandwidths for computes\n\nneutron-tempest-dvr-ha-multinode-full job\u0027s computes seems to lack\nresource_provider_bandwidths values set, let\u0027s see if it solves the\nissuse with test_migrate_with_qos_min_bw_allocation and\ntest_resize_with_qos_min_bw_allocation.\n\nChange-Id: Ie02f9d40716af1b3b4efc3e9340acab7f9af113e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7826661d6594e593e6f5c86753fac3e40ae5bbf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7826661d6594e593e6f5c86753fac3e40ae5bbf7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
