)]}'
{"id":"openstack%2Fneutron~770950","triplet_id":"openstack%2Fneutron~master~Ifff470c00eae0be1f4a5a882d3b03fd5ccae9d8e","project":"openstack/neutron","branch":"master","topic":"bug/1911925","hashtags":[],"change_id":"Ifff470c00eae0be1f4a5a882d3b03fd5ccae9d8e","subject":"[FT] Wait for the \"neutron-keepalived-state-change\" init message","status":"MERGED","created":"2021-01-15 15:29:34.000000000","updated":"2021-01-26 11:44:45.000000000","submitted":"2021-01-18 15:55:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770950-bug/1911925","meta_rev_id":"264a24fae7e9de229ba973941ee5977da2932fd8","_number":770950,"virtual_id_number":770950,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-01-26 11:44:45.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 15:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 11:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 05:56:20.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-18 11:12:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 17:25:10.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-16 02:26:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 05:56:20.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-01-18 11:12:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-01-26 06:43:12.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3181e2f012f3bf4a63793ce7e8b0bf4ef79edce9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-15 15:29:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca0e2c8a74627130190bac28d96d5277e5437cf3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-15 17:25:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/770950/1 : SUCCESS in 1h 54m 26s","accounts_in_message":[],"_revision_number":1},{"id":"64c00f3343059c6fa461098f9c48b49f44b78a6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 02:26:30.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ddc9e47488e94c2a9df72919a22df8c8 : SUCCESS in 51m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e0c92362b104e7bbcc6bdd64ea0add6 : SUCCESS in 32m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/87f86fa5d61f4812a67f09518f33a094 : SUCCESS in 1h 49m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/72e463ab9dce432e98d94ba627dfd3ca : SUCCESS in 1h 44m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4c99c7862a7f46618691bcb1ee4eca09 : SUCCESS in 1h 31m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/684394bb46804138a99647e3f3b33f3f : SUCCESS in 1h 58m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95c6b24bdf7b4479a5e0c854fe14eb33 : SUCCESS in 44m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2b0880e15d54ff4aa8d73da4c94bfc1 : SUCCESS in 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e442f25b0495441289ddac43bb0612bf : FAILURE in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffdbef58e7cc4a089a221407c61e474d : SUCCESS in 25m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec0614d44fe344cc9580da150b08255c : SUCCESS in 23m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3453c192b7f84b9e970ca2e2cadf0905 : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be05ca3c791646cbbecfd51ebd031a7a : SUCCESS in 12m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/868a3e2e60bc407a85e5c176b96bc45a : SUCCESS in 1h 07m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/219b5f0cbc9e45f09586f58c7fb584b8 : FAILURE in 1h 01m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/983afe53fa2c4a40865b4e873a70c665 : SUCCESS in 1h 29m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa53f25fc03c41cf88e2319c490afbfe : SUCCESS in 52m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/007980ea82f745aea317dfdd24ae3806 : SUCCESS in 1h 03m 15s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/3002a8859e0f417b8c2deed0b6e2933c : FAILURE in 1h 01m 32s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/91c4111ad67643eeacf70a9ceba89491 : SUCCESS in 1h 22m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8b1244eeda9b4f03ac00c133034c6066 : FAILURE in 1h 17m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fb95a210949f495990bf3d0016577432 : SUCCESS in 1h 39m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9489f516ebcb477eb42d6a091c17dea7 : SUCCESS in 2h 07m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a42b3e60af044fa8c0f17b4260dbf64 : FAILURE in 1h 00m 17s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/9cfa6262095744c684d5e65f16a4dfbe : SUCCESS in 1h 16m 28s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d26a2eca547b4453869e70110e6d19bc : SUCCESS in 1h 14m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/62af590b9f8e4302a9b8b52066d0cf25 : SUCCESS in 1h 37m 48s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/7e2bff3a028c48d3b5b7b54578f89483 : FAILURE in 34m 12s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1042af85a59a4715987fd1e2ae8d69f3 : SUCCESS in 1h 23m 47s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1cd4b0ac94c64897a91c7c9798b2e09e : SUCCESS in 3h 46m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/840fb50ccf9940d58f54a5fa0e921a5e : SUCCESS in 1h 29m 10s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c60edff6f74b458982a59f57f7a219a2 : SUCCESS in 1h 36m 48s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8d7a6b206dae4887b75fa741ed54ddf5 : SUCCESS in 2h 10m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7834bd9629dc4981a7ddda89fbe314ce : FAILURE in 1h 52m 18s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/caca85cda0414ff2807f1a1edd335a27 : FAILURE in 1h 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c55ec7c45aa60814536f6e64a48d60074cf41fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-16 06:46:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c7faad28d003871328d87de13ba1a4d995614a07","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-16 08:43:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/770950/2 : SUCCESS in 1h 56m 12s","accounts_in_message":[],"_revision_number":2},{"id":"16d74d1437da68343adba456c8f98ddd727d5cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 10:18:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e04c172c02254caab525015260d1e5cd : SUCCESS in 45m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c2e4967f7e2b45378ed0b820ee681625 : SUCCESS in 43m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7fbf607b47074f3a9c1fe5c11d79bb53 : SUCCESS in 2h 11m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5c7e652862a5416288049a1d52f03ea6 : SUCCESS in 2h 03m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/083d8b88afac48bbb38b28d4fb18dedf : SUCCESS in 1h 49m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1acb762097544216a21e8f4e57ee7ab0 : SUCCESS in 2h 10m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9f240f753a94a01942681f6f3df2340 : SUCCESS in 36m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eaf760e14b2047e5a297e64ea2761b1e : SUCCESS in 38m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7de06e7a17e44aca8ca45b3903c3af77 : SUCCESS in 14m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9b35301b93a4e5ab70aaac1f0bd2834 : SUCCESS in 27m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c714618e94446f190796ed09766e2ec : SUCCESS in 30m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd7526f0ce6a4165ab903bdc082b09cb : SUCCESS in 28m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d858e1897d134b579753f24edf426f18 : SUCCESS in 8m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/46ec23b4d6f743139a5041d0974f37b3 : SUCCESS in 1h 08m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/85e2b1b75a2b4083971f902303f5c60e : SUCCESS in 1h 18m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ec5cda0fdfde45a1bbb8c79c4761e11c : FAILURE in 1h 50m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b9c8304bb614b199bed27f879bea1e8 : SUCCESS in 1h 00m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a3a4bd35b69f458487e8a8b1c83a6f07 : SUCCESS in 54m 11s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/7604fa2b63c149b2b65fb4ed902e487a : FAILURE in 1h 06m 15s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b1f98a2af63a47b8bb1dcc9637af58cc : SUCCESS in 1h 20m 58s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0386a920aa00440fa723c137f1ffbf59 : FAILURE in 1h 17m 57s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b956263ffb394591a9093c9127331c83 : SUCCESS in 1h 13m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/47cc00cb3e604c559bf5a7f843434fb7 : SUCCESS in 2h 09m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19f09c7846414d90bffb48b3dc1f3b67 : SUCCESS in 1h 01m 41s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/11fd8fe99c534f5ea71508a519cb6652 : SUCCESS in 1h 07m 07s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6b172e44c88247b3aaccac8b5b2661f0 : SUCCESS in 1h 11m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1951f4a4ca704ba2b8d9f62a342dfc72 : SUCCESS in 34m 56s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/5595d71b660349f18e816224a50bb982 : FAILURE in 1h 11m 41s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/57a89ef75aeb4168be4ae2248ff02fc3 : SUCCESS in 1h 19m 03s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0784e9ec456d46eb8785bcedcb072840 : SUCCESS in 3h 24m 37s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d50d8ecca8a54efe8c8dcdfe601490b5 : SUCCESS in 1h 17m 55s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/14442a9096184211a5032bc2dfa4b0a8 : SUCCESS in 1h 52m 15s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f0a54b5bbe794793a7b1d610bf44d03e : SUCCESS in 1h 59m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/95c57f5d599e4cea9975b1f08c7ce9d4 : FAILURE in 1h 48m 13s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ead4bea105a2475ab992838967d94e6b : FAILURE in 1h 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34d2e209daa30a05f9bb5f8b675d59d8f9a48536","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-16 14:45:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2c106c70572f63501beb287829688a8bc4a9035e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-16 16:41:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/770950/2 : SUCCESS in 1h 54m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f680224447bc26cee98b6eb538dcbd49c937bb7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 18:22:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d756127a223641bc9aa00f86d7692137 : SUCCESS in 1h 09m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9719add0a62e4805a3429f6d2938dae3 : SUCCESS in 23m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3d1fbb70a5fb49c6a31dfca1d92d980c : SUCCESS in 1h 47m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/967a8ce806cf4f96b1bd3f6bc32c5c0b : SUCCESS in 1h 27m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9d077357cf54c8c84dffa9b09d5a705 : SUCCESS in 1h 24m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fb9504ec3497432eb14a2bdb874e9cc2 : SUCCESS in 1h 36m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a767975ce88c40aa9d9791418aaf339c : SUCCESS in 39m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5a6ee0e31dc4aeabc13e3e8a9b8455c : SUCCESS in 34m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dba095240a704c2ab0d306de6ad7040e : SUCCESS in 11m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bb7a419f6cb43219e7d287d0071649e : SUCCESS in 35m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6476c09b9d9446f8bfc1543e1ab30fc7 : SUCCESS in 34m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03dffd2b9ada42e9a2c275faab046afb : SUCCESS in 35m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a3e3ff51e047798dc6dc1f1950dcd2 : SUCCESS in 9m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a0bc17724364c5c9579c689fedf2664 : SUCCESS in 1h 08m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/79e46808a18840af9a971c85ea922f13 : SUCCESS in 1h 08m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6a1d19058648471fa57dd30e312c8019 : SUCCESS in 1h 39m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/610f5e0d0fa44d61977dadd3daaac9db : SUCCESS in 1h 05m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ea4e3bf167df4f8cb69e6cf4e1bcfbcf : SUCCESS in 1h 07m 16s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/c043b8d41bb748b1b0d7355bc87f4d15 : FAILURE in 1h 01m 34s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b3f6acb71145475294fe75a849705aa4 : SUCCESS in 1h 29m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/98a21a42d1834ea48d0f0c112084544f : FAILURE in 1h 21m 37s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/56fdd5e5470e41d2a7dcfce107294d2a : SUCCESS in 1h 11m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4664988e924c4c10a2c6c38e8e33ae6a : SUCCESS in 2h 12m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/410e2168d7f34337a2a27ed17d347564 : SUCCESS in 1h 01m 24s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/66cdd8f50c20402ca425716731efd370 : SUCCESS in 1h 12m 47s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d558192428f24322906f923c86b932e7 : SUCCESS in 1h 12m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/034902da32a744f2a9da48b4c25f6b8b : SUCCESS in 33m 09s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/d0c912810c7441039408ecd2ffa9c1e9 : FAILURE in 1h 01m 20s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4a75958fbede4743a757b9a63f84e525 : SUCCESS in 1h 22m 07s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/103cb5af9b79405189a4e2c65b2981b0 : SUCCESS in 3h 32m 38s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0606ebfb815a415794d87e44f16c12ca : SUCCESS in 1h 16m 06s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6594b11dcfa54515b6d15030c53f55cc : SUCCESS in 1h 30m 54s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/852eee4253df464fbde95d8062364a06 : SUCCESS in 2h 05m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ef71f93597f04dda8667a3fd0346ee4e : FAILURE in 1h 50m 26s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/57f81824cc6840708e3ffb39415eeed2 : FAILURE in 1h 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"804ad79b605827f7fecc1e70722bfdcd66ab766c","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-18 05:56:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cb788a5ffb39650141ba2b73bf8d99bb991c3292","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-18 11:12:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"80e60bc7a5ae8f928ffdf3329c1425b57e15eeca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 11:13:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4b83c91300b42e29b0f855827f909f50c962d953","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 15:55:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1151939b8bf74f01b91faa9c64925d1e : SUCCESS in 49m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1cbc7ed4ba604b16a3fbad50b735c414 : SUCCESS in 1h 37m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5fcf272e728e491da0bbc909e8a45794 : SUCCESS in 1h 38m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6584fe3cf72f4bf28081e77f2700283e : SUCCESS in 1h 36m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8e50c643d2c4407f849012404ee715cb : SUCCESS in 1h 15m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/648180d21b3747d19919b3123d00335b : SUCCESS in 32m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c078c2f41e5f46789369972ed64302c7 : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f9ac0c0efc64257a0b8fcec63b32e00 : SUCCESS in 28m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db9de40668574707abd886dbe3b00ee1 : SUCCESS in 29m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5292b507b933408f89c9e45ab9e67a63 : SUCCESS in 8m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cde2401f6ab64da4a8988617a6a9e375 : SUCCESS in 1h 15m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e86b72e976f5470eb9afb604e619e1c4 : SUCCESS in 1h 08m 05s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bbb363aee622478a8d75f1ab5eaa83cd : SUCCESS in 1h 43m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/642a7fa7d09f4859bd6a879bbd3cba50 : SUCCESS in 1h 04m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/052afe7d1c444938b44de82364abf8b4 : SUCCESS in 1h 00m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/426683a16bf74d60af4a319dedea70a3 : SUCCESS in 2h 08m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbf0fa1cc847434ebdaa4448ce6d7171 : SUCCESS in 1h 05m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b0f6fc2aa5b54991b1f53ff35cf7a917 : SUCCESS in 1h 13m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8c76287c12c7480b9912e4973e4119dc : SUCCESS in 37m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8dcaa4b161e9a30635d4926a21ecd738cc1ae2cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 15:55:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8dd43f1e2925b8f0813a93e615cd9537a8f18835","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:09:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d478e7e5954afa8c6b04b2f275f60d : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":2},{"id":"911b9de554205e1e3e98d1bc17f76269b1fac96c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-26 06:43:12.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/50/770950/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"264a24fae7e9de229ba973941ee5977da2932fd8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-26 11:44:45.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/50/770950/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"66e045360cc23ff1533ad5df866e7a0574eabcbe","revisions":{"8f24d30478e0c0713ceb563f7737e97b1a9974e3":{"kind":"REWORK","_number":1,"created":"2021-01-15 15:29:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/770950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/770950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/770950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/770950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/770950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/770950/1"}}},"commit":{"parents":[{"commit":"1683ada81fc983bcdca1275ce52b54b4a5854f1c","subject":"Merge \"[FT] Remove unneeded DB migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1683ada81fc983bcdca1275ce52b54b4a5854f1c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-15 15:27:43.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-15 15:27:43.000000000","tz":0},"subject":"[FT] Wait for the \"neutron-keepalived-state-change\" init message","message":"[FT] Wait for the \"neutron-keepalived-state-change\" init message\n\nWhen the process \"neutron-keepalived-state-change\" is started, wait for\nthe initialization message \"Initial status of router\", printed in the\nlogs, before starting any other operation.\n\nChange-Id: Ifff470c00eae0be1f4a5a882d3b03fd5ccae9d8e\nCloses-Bug: #1911925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f24d30478e0c0713ceb563f7737e97b1a9974e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f24d30478e0c0713ceb563f7737e97b1a9974e3"}]},"branch":"refs/heads/master"},"66e045360cc23ff1533ad5df866e7a0574eabcbe":{"kind":"REWORK","_number":2,"created":"2021-01-16 06:46:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/770950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/770950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/770950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/770950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/770950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/770950/2"}}},"commit":{"parents":[{"commit":"1683ada81fc983bcdca1275ce52b54b4a5854f1c","subject":"Merge \"[FT] Remove unneeded DB migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1683ada81fc983bcdca1275ce52b54b4a5854f1c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-15 15:27:43.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-01-16 06:46:38.000000000","tz":0},"subject":"[FT] Wait for the \"neutron-keepalived-state-change\" init message","message":"[FT] Wait for the \"neutron-keepalived-state-change\" init message\n\nWhen the process \"neutron-keepalived-state-change\" is started, wait for\nthe initialization message \"Initial status of router\", printed in the\nlogs, before starting any other operation.\n\nChange-Id: Ifff470c00eae0be1f4a5a882d3b03fd5ccae9d8e\nCloses-Bug: #1911925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66e045360cc23ff1533ad5df866e7a0574eabcbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66e045360cc23ff1533ad5df866e7a0574eabcbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
