)]}'
{"id":"openstack%2Fneutron~772255","triplet_id":"openstack%2Fneutron~master~Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee","project":"openstack/neutron","branch":"master","topic":"bp/distributed-dhcp-for-ml2-ovs","hashtags":[],"change_id":"Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee","subject":"Config option to disable the DHCP functions","status":"MERGED","created":"2021-01-25 07:56:26.000000000","updated":"2021-03-06 15:03:26.000000000","submitted":"2021-03-06 15:00:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":48,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"772255-bp/distributed-dhcp-for-ml2-ovs","meta_rev_id":"194e3e08542b1172e289c54b02140bfe82095a54","_number":772255,"virtual_id_number":772255,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-03-05 14:31:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-06 15:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-06 15:03:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-05 12:49:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-03-05 15:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-06 15:03:26.000000000","post_submit":true,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-03-05 12:49:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 08:18:33.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-25 10:10:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-25 11:17:00.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-01-25 17:01:46.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-30 02:19:11.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 03:57:38.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 08:39:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-02-03 10:55:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-02-08 12:42:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-02-19 08:46:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-20 01:32:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-20 05:59:23.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"891c5e6df9a61eb0029b6eb6daf9229e9b70c0a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-25 07:56:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"353ce60dbcd2ffa97ff2f26d92849cd8b5b0c590","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-25 08:18:33.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e1ac7e4e23599c3475679b9135773835dfc40593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-25 08:56:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8687d21741edb73f8407c158c0a816064b3fd11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:10:04.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/341a8a9e405d46789bb2c9ad11a30c0a : FAILURE in 20m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e174e2179954481790616348a7064774 : FAILURE in 30m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/98abad9de61b4c6baace8a0972554213 : FAILURE in 27m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e79138c7817478da3763532ab76c402 : FAILURE in 26m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ad300a9cd04450687da0b668697d2a5 : FAILURE in 23m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b22c1151db114e3bb0876b4fe7354f9b : FAILURE in 25m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0474b0d89dce48bfa79b823ca32f3f16 : FAILURE in 28m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be9a1fd7e2734373b4a7ad19a7a9e4d3 : FAILURE in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517267586e834e6b81f1c637d4724322 : FAILURE in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e37416a99e54ddf8f6422460585372b : FAILURE in 21m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cbfce1cb9c648b8acbb75f72daf8b8e : FAILURE in 18m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3355a3ef2951477c91951fa6ecc007f4 : FAILURE in 22m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8172419286c4429934c58424d355349 : SUCCESS in 8m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2533987a49204b86a7851b0ede668be7 : FAILURE in 1h 03m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/571b094e786542f39288d62eaa8ff93a : FAILURE in 53m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c08b03b321374842b458abe3e4d50902 : FAILURE in 24m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8483cb032734c79a22427e16192c972 : FAILURE in 1h 02m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1e6b829745d344cf870b07a77341fc67 : FAILURE in 53m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2639b0e73a7241189dba9234535fff71 : FAILURE in 27m 57s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/badea697629a4b2aa0e600afef1c5cd9 : FAILURE in 30m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed973f4a068548ff8fdf2fd62101312b : FAILURE in 26m 47s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ae420c1bbcc4021bc410887bd7daaa0 : FAILURE in 24m 54s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5f85c0bf7ea2431c87901734fa7b79ae : FAILURE in 23m 07s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ff238ad09d9b452dbc0e39d822a01ad0 : FAILURE in 25m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/13947b15042741109abde7bb7deed667 : FAILURE in 34m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2917ce5af76045eaa5b2db60144b8b46 : FAILURE in 25m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e641b934243b266aa27ace3e5615dfea4acdbd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-25 10:59:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9ffb00cffc5d6a0ce10dc515be4bc1282985a96e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-25 11:00:46.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/2/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"a761729d941a21a402c04970569b9cadbd791f06","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-25 11:17:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c1b8a8478fccb89cbd7d8aae4550a89333185395","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 12:49:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e1090f4985bbd3b46153a4c0c121b0f47fe21ef6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-25 14:20:23.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/3/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/3/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"21c291a588fb51b6ec173981c3dc287a1b567659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 14:25:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2562cc2e873a40fe86601c8f5c7ab087 : FAILURE in 29m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/943967f5fb004ad18da02184403a096a : FAILURE in 27m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e847e975a7274f9e86e6d354b502541f : FAILURE in 22m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1dc17cc4220348529bef2e8569a581f0 : FAILURE in 23m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6e926862de524020a750980d4885c592 : FAILURE in 23m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c191cd1b7d2c49eb802a0fa257c3b139 : FAILURE in 42m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b65c705bfbf489ab7dfc364ca408059 : FAILURE in 22m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/238f8a5457674de8b815441680346cc5 : FAILURE in 21m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a28d9abe0f4ae69b52ca09242304f8 : SUCCESS in 14m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cfd8e866ded40e194000f47856cc59a : FAILURE in 19m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51bf3d17b3584ea8b55c81b3d5ef8319 : FAILURE in 14m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9940029a2c1b4987bb1f335e999b8bf3 : FAILURE in 22m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a627198b1014207bbe9f3354029e55e : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d05765d2aa144f6aa9ad8087ac2f744 : FAILURE in 58m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/145f7e6dcdc346e1bc9c001531b49c41 : FAILURE in 1h 29m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0d9fb70e1e7a44c4ab730cc3563a72af : FAILURE in 23m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb1666e022454ddbb6fbeecc1bc0097d : FAILURE in 51m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8ee22fe0e13f499fbe571606d0dd5f0d : FAILURE in 56m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/525f37c7371e400dbbe42d948def991a : FAILURE in 32m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0e03ce16c7d747359d33bd77c51c945e : FAILURE in 34m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c8eba4663b24635b4cb51b9d6ad5ae5 : FAILURE in 28m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2405f38bd514936a418292af64a307f : FAILURE in 20m 30s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/aefbac98b0e84711a0856c1aa95012c4 : FAILURE in 23m 32s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d0cc335276af4260ab1160c69641d756 : FAILURE in 25m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8cb838f76087486facba4b6906b7bb51 : FAILURE in 27m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a9c8b48a26b3465dbccab19e93468c0a : FAILURE in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e7a307ea028078575dbff59ab12a1bbc256ca01e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-25 17:01:46.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772255/3 : FAILURE in 6h 00m 53s","accounts_in_message":[],"_revision_number":3},{"id":"3d3d67d1b67805581fd2256746f6040b43ce3ad3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-25 19:31:48.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/3/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/3/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"5af0cc393b1338219b3db67cfc915427225e6006","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-26 01:42:51.000000000","message":"Topic bug/1900934-config removed","accounts_in_message":[],"_revision_number":3},{"id":"0bf0917a8ef37f16e4d58474f412f9b12d3af873","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-26 01:42:55.000000000","message":"Topic set to bp/distributed-dhcp-for-ml2-ovs","accounts_in_message":[],"_revision_number":3},{"id":"5a283b07a575c02f7d6ee6263d1c35aaad167ed6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-26 01:52:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4518446fd15b93b911b3fa0e896346b31c46f0ba","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-26 01:53:00.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"71edbd8c7bc79ffcc4d7019ec1284a186745689f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-26 01:53:04.000000000","message":"Topic bug/1900934-config removed","accounts_in_message":[],"_revision_number":4},{"id":"05af86d913aa0c0d30765435ca9b9475e936b313","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-26 01:53:11.000000000","message":"Topic set to bp/distributed-dhcp-for-ml2-ovs","accounts_in_message":[],"_revision_number":4},{"id":"07d8e9775f4c3df9513bc92ecc1ebb260a88b587","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 02:39:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0397eb77cb64e94a0118b4cbc149ecc : SUCCESS in 40m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18bdc0e70d3546caaed507c425ad076b : SUCCESS in 38m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2a9cdf5d6b35ccb46071585888998d1f3c6b64ed","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-26 03:41:02.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772255/4 : FAILURE in 1h 47m 40s","accounts_in_message":[],"_revision_number":4},{"id":"e31aa9d9edd68a8decc6b8ab95f243e7cc072ff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 03:59:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/acc289eb1eb247b09324c09f97779a35 : SUCCESS in 51m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c2de86229724bdb9fb2d8311dab168d : SUCCESS in 34m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/21cfedfb6a954808b25ceecd5466a114 : SUCCESS in 2h 01m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7a2e023b979747eda43f2e6c8b327af9 : SUCCESS in 1h 29m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7a0fa5d015854a78b087e4f37fa3688f : SUCCESS in 1h 16m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/469e73ab0259474ab835956d754863df : SUCCESS in 1h 45m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1795ff208e36470aa00d807c6dfeacec : SUCCESS in 38m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8a42947084e481ea4d5e083eb1bd793 : SUCCESS in 29m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4f264e2c7e24a8b8537ce6c75cda16d : SUCCESS in 12m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8e508dad7ba4f6490c3abd9f34101d6 : SUCCESS in 28m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e439fce732a4431bd5ce1c3c1b66830 : SUCCESS in 29m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a4b82610f1b4988962add1eb2fd1a5d : SUCCESS in 25m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3274b5dd658940e08a871e4fc934726c : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3de2c855676e4f948306d2e0cbf2542b : SUCCESS in 1h 11m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ec58e259d9c4481a2e53628cf9ae6c9 : SUCCESS in 1h 15m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3542cb8a10b14efcaa3a6b92faa7c8c1 : SUCCESS in 1h 43m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd0e1e1161474c56acbe14b0c0b213b5 : SUCCESS in 1h 00m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aa8427f2ab844a3c9352d01770ca8b27 : SUCCESS in 1h 00m 58s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e82f27f47de54e23b2bfdd40fb2820e9 : SUCCESS in 1h 13m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/243f049d5969461796ba5be89d658483 : FAILURE in 1h 12m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c28b4ddaf15943eaa98bd9e0f57c7fd3 : SUCCESS in 1h 49m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b87eb52b1b44b0799532f40c3ee2793 : SUCCESS in 1h 07m 19s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0d7d08f07ceb4074b6f501d12d4a98c6 : SUCCESS in 1h 09m 05s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2b6d73be189b4635aac25bae7664c372 : SUCCESS in 1h 12m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b91af4f170ff4f43a4d25c1e3d1ceaee : SUCCESS in 29m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e433bd09d88f47db9a88ffad09495e18 : SUCCESS in 1h 36m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e1984fab1509059909e77cd96a61d8e7eb93051","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-26 05:48:39.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nA couple of questions for better understanding the strategy","accounts_in_message":[],"_revision_number":4},{"id":"0141f0407f73ea72f2f188d9cd81fa50bc7d3e86","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-26 11:21:08.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cc60c74f96fbe00a91eca50b2141ff635c962aac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-26 11:28:43.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/4/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/4/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"a889dda0c9787618a387bcc52132aa54bba6d143","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-26 11:51:44.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"be16aeec2f8b9efb64fb14626f5cc9847e3d7ea7","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-27 01:04:27.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"21558c6f52d0f86d1f89fea677b768ea278a1711","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-27 07:52:05.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c9245ab9f059bcb6543ffd7439a8ac15b54304ca","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-27 08:38:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e069f15551033432ce9e8c2055c2bf1a52631801","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-27 09:04:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cd46795f68976d80e0b1fd1fd2103fd0ac3852a7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-27 09:05:31.000000000","message":"Topic bug/1900934-config removed","accounts_in_message":[],"_revision_number":5},{"id":"79a11bf92437bf5600c897c4b8fbefc62156f6f3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-27 09:05:44.000000000","message":"Topic set to bp/distributed-dhcp-for-ml2-ovs","accounts_in_message":[],"_revision_number":5},{"id":"079613801256f482fe81c5da0bebd398ae1a2046","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 09:47:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eebccf2e10344c22b80ef0b10ecf2064 : FAILURE in 32m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/619df1a3e1b145deb9aaec1bf426b71f : FAILURE in 35m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"28f67afbd0274e4171b28cde0eb0250a50bb49eb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-27 11:30:07.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772255/5 : FAILURE in 2h 24m 35s","accounts_in_message":[],"_revision_number":5},{"id":"aa6bb7bf8cee79cdf6a627248a6816c401f14fef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-27 11:30:58.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/5/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"1c6ebef8d3531e1150c5eab5fa293e4370023b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 11:56:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3bffdc8072444249b71ac32248e2e65f : FAILURE in 1h 07m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9100827e0bba425394754bd5820438a6 : SUCCESS in 28m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9668b0ace23f4e29982e71606126f25f : TIMED_OUT in 2h 50m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0b60e776a0994d3f958c68dc8976eb7f : TIMED_OUT in 2h 49m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4aeec9e9a3314d4386878dcea7a66614 : TIMED_OUT in 2h 49m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e8d2f3b91f824401b51a16f3380d28b0 : SUCCESS in 1h 47m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51d122eca1004dd58c0a62976ae2caea : FAILURE in 36m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20f63bdf534648798d4d8ad2e1590949 : FAILURE in 31m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abf940a6a08340f089cd20e3a09e30a2 : SUCCESS in 15m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f6cdec23e8a47eba86eed29eb3412b7 : FAILURE in 27m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f9d7551039145f2bc8e62ab915a98e7 : FAILURE in 24m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bce89d620dc44b35ad1eb6a793ed46bb : FAILURE in 24m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856edc4d776a441f8769318afe04eaa7 : SUCCESS in 8m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f156e6a14bde4641b8af0358f1a35d17 : FAILURE in 1h 04m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e41b2ce26ec43f7a1d637c412b1867b : FAILURE in 1h 04m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a8f5d2a4cd1e484fb78fd31a0d2f62d5 : SUCCESS in 1h 44m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b23e6d11ba634fa0a09d525d5c829fe3 : FAILURE in 1h 14m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ed3e58a1fa404f7999d95e6a45d073f1 : FAILURE in 1h 16m 33s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6309f1ad496046279f5ee719025ec4af : FAILURE in 1h 40m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8bf0d2ce8fec499887c11eafab081265 : FAILURE in 2h 00m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0b5b0cfc5614285bf022d9868823c4e : FAILURE in 2h 29m 23s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f31dbec718da4fbcbeffaf4cd6b4b2b2 : FAILURE in 1h 33m 36s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/3411e5410c55452588494a55436b67ed : FAILURE in 1h 20m 57s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d229eb9b38b74a489101c5edf010331e : SUCCESS in 1h 05m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/38a94f172bd94bcb943e57cb8db97540 : SUCCESS in 34m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5491fc452df54a4aa5c66843b6aee182 : SUCCESS in 1h 36m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2dd56fb1e1ae176f9716c703ba0d06c9f8d31ade","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-27 12:18:01.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"482d40ee2a6a4f0693e313ee6a0ccf2ba85ff0e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 12:54:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/036d73c98fbe40008bc3f2f233e2dd2a : FAILURE in 32m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de03b972e1af4b5ba5ef1b998e371839 : FAILURE in 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0bb0252854cd3f080b7cd224925bcf846081ff58","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-27 14:23:06.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/5/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/5/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"1f5a629200188fd8b87141ca418979c72364b9eb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-27 15:16:18.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772255/5 : FAILURE in 2h 57m 29s","accounts_in_message":[],"_revision_number":5},{"id":"298b4eca64c2083172b05ee172cf5a15ec8f558e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 15:53:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a9dacaf2d3ea474e98076160ed8cea52 : FAILURE in 47m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ffc40a0e4dde41b890de14484252ca7e : SUCCESS in 33m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b23f68e9bc874758bd5d3be6e9aa24b6 : TIMED_OUT in 2h 49m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/acd449b5933b4b0db28d56dc2fb23195 : TIMED_OUT in 2h 50m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d53b2f2f1c3b4c6fa9610896ac87df4b : TIMED_OUT in 2h 51m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/65476a331f014edb9e5150186db29574 : SUCCESS in 2h 01m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c41e7be7cce4eb0b011d7e880f5e506 : FAILURE in 34m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a50818f0961a4982b9d55667b0f98aad : FAILURE in 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0264e546598842c585cc1adedb56cc52 : SUCCESS in 15m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e911d7839eb47adad33dde661361f42 : FAILURE in 29m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccf7d0c8d44348aca559f31da2147d14 : FAILURE in 27m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b8df64bdd6e4d6188e8a6eb896f73fb : FAILURE in 24m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d72cbffce3451bb062dee97276455d : SUCCESS in 9m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c220aa17751f47a1be3d9bc224913615 : FAILURE in 1h 09m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9d64313309140529fc35c97aeb90cc8 : TIMED_OUT in 2h 15m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/900934167be54e4fa606a4c4b47991ac : FAILURE in 1h 40m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d325dd6fb225430c9f065de0a01c3b27 : FAILURE in 1h 08m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ff5cbb7fdcd64845b022887f1df6874e : FAILURE in 1h 26m 23s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6b308bc72efe4811a9a92c07790225e5 : FAILURE in 1h 30m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/de7a4272581e4fac9b0f30a9537706a8 : FAILURE in 1h 45m 37s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0714588eeee44d77b33a0ed116a1e352 : FAILURE in 2h 26m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6def2326c6642898fa8b0084dc1764c : FAILURE in 1h 35m 05s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0f64472bfa8241b7a4f0ffb443a06889 : FAILURE in 1h 15m 09s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1f43b58dd2ae454ca3a4167ee2e8b0e3 : SUCCESS in 1h 08m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7a046e31e9f743f9a3e11f582ac2c028 : SUCCESS in 35m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b654ddcc55f94b889479453f54f120d3 : FAILURE in 1h 50m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7aa62820ba7226eec478cea8f8476fb24dc692ff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-27 20:33:53.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/5/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/5/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"8f06bb411e6a4114603f43bb268875d7ae710f2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-28 01:03:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8cb937ca94a1f689a6b7ab87eaab86afc5bef4d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 01:41:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/485eb5758c4543b0b583f7d16b267f1c : SUCCESS in 31m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e0b2a8cd198d43cbb83da5f7dc92bc6d : SUCCESS in 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6d11e98b63f3ef4ab0125426ff0be2bed8da9dcf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-28 02:48:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772255/6 : SUCCESS in 1h 44m 18s","accounts_in_message":[],"_revision_number":6},{"id":"e2441a8c7d0a2659518a2984ed52bc9e0130ebbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 03:07:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/52a5162ebbbe4e4d8675be0cc28d42cc : SUCCESS in 48m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1343c4b3c9b74878a971d84174718cad : SUCCESS in 34m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5a06fe2806ba4b72ac6f5fcc7d61c63d : SUCCESS in 1h 27m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b8f4c55843154ab6874860b5e8042fd8 : SUCCESS in 1h 36m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/02bf06a985414e03a13952bd58430a9a : SUCCESS in 1h 38m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e0c1c0504dc1416f87cdffe5eca3083c : SUCCESS in 1h 58m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/340a700ef39d42019d4f03175f60df53 : SUCCESS in 29m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0958c0e0a9f64129ab336a18b32512bf : SUCCESS in 33m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a0e0105d19a4d3693a9486f8b02e4d6 : SUCCESS in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/170cd854ad8a4af88b9159e9951e150f : SUCCESS in 26m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a64334a6a4fa43febdc99ca923c77c8d : SUCCESS in 24m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ffdaae3d7de4794a95dcaba4221f6a1 : SUCCESS in 27m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046c0e1362424c2d905db66d2d3240ca : SUCCESS in 10m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/001eae11d97c469687ecedb103b1a849 : SUCCESS in 1h 07m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fdd1205f4fa341e4a1512fb01ed00c3c : SUCCESS in 1h 11m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c7d41373c5994fc8af290c5bfe3f068c : FAILURE in 1h 40m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd1d953efbba40e084990edc938059fe : SUCCESS in 1h 01m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/58846067033b4bf9931aad0ce767d686 : SUCCESS in 1h 04m 45s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6d3d6e2178ee44ac9ebf7b1b04f53329 : SUCCESS in 1h 15m 05s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a725cc2c3ad34df285b8ca4f8af9e64b : SUCCESS in 1h 28m 37s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3dbddc5479ff4b65a42eba0ec8ced6bc : SUCCESS in 1h 38m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af316038be4a467482775f5abfc6cb63 : SUCCESS in 1h 02m 05s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4673b3cbc3ec4e61bd236bcce4871b77 : SUCCESS in 1h 09m 19s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4cd9ae189b3a43a58ab46336ec359a88 : SUCCESS in 1h 14m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7680a98b84d243eaa9fd8f11a9f43da8 : SUCCESS in 36m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e7efabd23ae7459580b0faececbd8f59 : FAILURE in 1h 40m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ff16b484ed4b0e8bc1e24c1a7485321b0fda3261","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-28 05:41:31.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nrecheck rally suspicios failures","accounts_in_message":[],"_revision_number":6},{"id":"f429c989d6c3aac72ec888e18d402c2f8b5a1135","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-28 11:18:15.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"bfd7df01421d346a4539ec3c6697b63dcea652ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 12:03:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96fcbdfc2145479aa568226b70ece533 : SUCCESS in 39m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec8a3db059e740c2a569b53d8e5d78c8 : SUCCESS in 39m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fd1689ed63c0807392ec16dcf5a1be905903a35f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-28 13:38:27.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772255/6 : FAILURE in 2h 19m 37s","accounts_in_message":[],"_revision_number":6},{"id":"bb5e73a9d526026f2687cbd013f4d78e80db07df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 17:28:56.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2ad50d466edc4ed3874a16eb071e13dd : SUCCESS in 52m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2b2109f7888a42a5b71a555880054e61 : FAILURE in 33m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/50a9377d879b4bdebfdaefe207cac532 : SUCCESS in 1h 56m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/522123276bdb4a40b61bc566559b8118 : SUCCESS in 1h 56m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6f535aa11454eb6b571e8354bfee2e0 : SUCCESS in 1h 37m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/871dc42111944de5b07d0bec94fd746c : SUCCESS in 1h 37m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/682fb41df92d47d1a509ecd680b7b5a7 : SUCCESS in 37m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6d42972c0674bf09be174372eac4523 : SUCCESS in 27m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fffea62641644e058d0aa03f79a4cc91 : SUCCESS in 14m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5dafbc7130d4ba1aa62790765917fd0 : SUCCESS in 29m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c3b99f8f0554786bbdd0c89cadf63a7 : SUCCESS in 25m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce1bd848e2804e5d90710e4ff39e7def : SUCCESS in 25m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b5591268f14f4eb652a30194ad8e1a : SUCCESS in 10m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5351be8c3be41ebbed76e0fd93cc123 : TIMED_OUT in 2h 15m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c2ff6a6479c4957b196a068baeabfde : SUCCESS in 1h 28m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/59cc3aab4b6b4a5196ed9b9216933e69 : FAILURE in 1h 32m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/92c64cc241f84e4db9b48c3fb63e86f4 : SUCCESS in 1h 00m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e3fecd82fbef4c1dbca0030d33a77e84 : SUCCESS in 56m 26s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e98e9609248e43ad92e0cc38d74c0828 : SUCCESS in 1h 37m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0e428cd56c8a48839e49a54cd446fa45 : SUCCESS in 1h 47m 04s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/284de8546b114b948a45be78c998faaa : SUCCESS in 1h 46m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d4eb4c4a4da4094994fe958d9eb3e91 : SUCCESS in 1h 07m 43s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/70381127a52f4397873f8dd2456e1c0c : SUCCESS in 1h 01m 43s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/652ba4f0778547ad9338ee284c459c0f : SUCCESS in 1h 15m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6ee1dba62f6746fa93286792eb49972d : SUCCESS in 47m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/220d6bb03a87490a838e7a03ac6d2de3 : FAILURE in 1h 45m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0634fdd339079044f601579d8397fcbc3bb3107f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-28 18:14:30.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/6/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"942d1b2ae1e077a909d3e2b3cb5ddec3dd53040c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-30 02:19:11.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/55/772255/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"33aac1c1b8fa4821a1d0551e0199d6ac4b0c674e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-01-30 02:33:38.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b97c504724d58a1979199d7f068ff3e58583abf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 03:11:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98295c62537c464a8ca55789a94c3a96 : SUCCESS in 34m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2919778799374f87b46fe9135e6d03c8 : SUCCESS in 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fc6d0dc54b9905ddcf903b50b941632a21adb1f4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-30 04:24:54.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772255/6 : SUCCESS in 1h 50m 42s","accounts_in_message":[],"_revision_number":6},{"id":"c84f31b515a6b902bcbe2a2d3ab2cb6000bf2de7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 04:55:23.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dc1f10c4d85544bc9ce146a365a6b26c : FAILURE in 7m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d287a26de4154556aa87facce8bf0684 : FAILURE in 9m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/849a6d91e6594ba68f8599740c57d34d : SUCCESS in 2h 00m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0782ab780dbd4fed8b5c5acb802ec483 : SUCCESS in 1h 42m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0181c303141043a890821941b55c9ef7 : SUCCESS in 2h 07m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/73eec790b0f44e4d9e17ec3828d21559 : SUCCESS in 2h 11m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1912da90fcb94c0ea31368039c3c64e8 : SUCCESS in 42m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/065189bbd4fa4ef3b4b605eea239ee0c : SUCCESS in 25m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d3b5b91f534d32828eb702d0302470 : SUCCESS in 12m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69b260465f5e4a899cc9c2a6ad5558bc : SUCCESS in 25m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2ffcac98ce34d349e1d1b9fb840ba5a : SUCCESS in 26m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/167a27b00ae6425fb8393494d04c35fc : SUCCESS in 28m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b17c8df18448d4ab76ab25bab35816 : SUCCESS in 8m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b529c245db7b4e7bb2d3ccef69e1d346 : SUCCESS in 1h 10m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1aee289aa1664ee3b8479fb4b5caba25 : SUCCESS in 1h 10m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/34ae9adb22ba46fcbf9c2b0f0a268db1 : FAILURE in 1h 36m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c45a01626207462493c2db9479c0fccc : SUCCESS in 1h 12m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/100a807a9443427692a18e02a51c2d1a : SUCCESS in 57m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d917bbbb2d924c9f8a07c18a27a2e3e1 : SUCCESS in 1h 22m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb61221c4f8d4ff78739f8254ff42806 : SUCCESS in 1h 38m 04s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a296af0847d84bf0956304d18b9223a0 : SUCCESS in 1h 41m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbb22a0cce374c3eafad4f82fcdd05b7 : SUCCESS in 1h 00m 45s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a31812155f264dd99494d27eb58c5b83 : SUCCESS in 1h 15m 31s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/27fe35e1a9d541eb95d10afc5c762cde : SUCCESS in 1h 18m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/37d3252940e04999899729534dcab499 : SUCCESS in 35m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ac942ab70c7849fb8cb10528178ca32d : FAILURE in 1h 57m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"59bfe2380359f75be4e50b61789d314da8f90652","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-30 18:10:34.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/55/772255/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"d05755664eb58c8036d97c2548e172e429b181ee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-30 22:05:11.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/55/772255/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"3d67fd810c2c86b41a520dd5c71f392c5568a9bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-30 23:14:31.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/6/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"e9394ff802ebf7b1f619683dfd2d24cfdaf63f1e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-01 03:34:28.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f6a026e729b783a63ed4677128ca04868115fee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-01 03:37:46.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"70ebd8531d3dec22206069e685002e3ae7b6bb3f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-01 03:39:05.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/6/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"bd5ca39e4d8e4e3dc8851105ca844fc966aa0d1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 04:30:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9538f2a992c47399a57f1ddc91f8a23 : SUCCESS in 42m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/122e8733219847f5b79ad365ba2557a1 : SUCCESS in 42m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"efe76aa8d043cb89d9807fc27a07953a9afe223c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-01 05:33:54.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772255/7 : FAILURE in 1h 55m 02s","accounts_in_message":[],"_revision_number":7},{"id":"38ac97f5fa6586a808aae0f95237c08a162d624f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-01 05:34:46.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/7/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/7/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"a8a3c8ff9c4ec244d80ace34ca28511cafed9f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 06:14:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6401c592dd9b4e308d1e24a7d0a55ae2 : SUCCESS in 50m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2006a9d6757d43878361a24873cdc95d : SUCCESS in 33m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c5107f62f9f34f9f936bd2173e4078bb : SUCCESS in 1h 54m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/234a9bb3606c41d7833191f71f4ea017 : SUCCESS in 1h 37m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de2534e98cc746e980c1ed6216c79dad : SUCCESS in 1h 30m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/38357aaacf024f1f82243c2926fa3c5d : SUCCESS in 2h 31m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffefbb6c6b02479481152065972dc490 : SUCCESS in 34m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ba2b8dc7a1a46c491368ec840fa19e6 : SUCCESS in 25m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a3ba54ca3a46ba88144ad1f216de6a : SUCCESS in 13m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96baed5c7d994b9892bd6c039020834b : SUCCESS in 27m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cea097243e154bb2b540c36f2e650d54 : SUCCESS in 26m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20758fd019074894bd96a9b8d3eeb371 : SUCCESS in 26m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b825ba463314255b7c0eb6b7f695e97 : SUCCESS in 8m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a591d7dcde664f9c9520c76e8d5a067d : SUCCESS in 1h 12m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/43d2053d8ed44b9bbbebd762728b4781 : SUCCESS in 1h 13m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/45389917167c4bf59116387229f3364e : FAILURE in 1h 53m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b88ae602484464ea78633547574655c : SUCCESS in 1h 00m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/213fc92680ab4818827ab2e7ba07e970 : SUCCESS in 1h 00m 59s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c091bf0050164ed091206ffd4f849496 : SUCCESS in 1h 26m 51s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15e85f0450af4683bd0cb7c4826b4289 : SUCCESS in 1h 34m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d8432c91533d4f30aeae0cf2196fb5e1 : SUCCESS in 1h 35m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63b2f9b5592e4b148a0c9215f68fe550 : SUCCESS in 1h 03m 34s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/10a45eaa1f7e4e2080674750dddc9b6b : SUCCESS in 1h 02m 42s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/236d48c6516e4e788bf5b7ef1c4ba83d : SUCCESS in 1h 06m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d085cdf251244a8c92143d2228c5a953 : SUCCESS in 34m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/af514718f15f4aa0b7d6c0723ce85d8c : FAILURE in 1h 40m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9bb3c51fc5bd428b608d293dad35df9cc0e41101","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-01 07:40:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7114bebadb35c4de854f5529b6a03077cab49f69","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-01 07:41:10.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/7/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/7/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"14fb13d334386cab86de4164575530266204b23d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-02 03:57:38.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/55/772255/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"a1524f7dbae7b6bec3a80a10c1888968dbcad8a7","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-02 07:45:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4b02a3667778f65129199b618d89a0e9526c2354","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-02 11:22:43.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/55/772255/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"1eb67e824a58a15643d8b984ed9c8144ed40cf83","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-02 22:14:12.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/55/772255/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"9a7e6d9ce5f8b1d2c2035f5994da7f6b72d13815","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-03 04:22:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5da8942409d048765e5d57c17becd9402f0088f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 05:20:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdbcad6179fa48ac8ae7a066ef993128 : SUCCESS in 52m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5676492f0f104fecbc28b29004cbfc20 : SUCCESS in 48m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"60718fd3509c7939b16fed326de011ff19d4b4a9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-03 05:31:29.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/8/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/8/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"06b4c706f39986f3f5b6ce2023c8ca1d1b43b637","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-03 06:21:23.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772255/8 : SUCCESS in 1h 57m 35s","accounts_in_message":[],"_revision_number":8},{"id":"81f02c0de401cf5534d3bee97d83e338db0e7ebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 06:47:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/413aa29ed6b748d1953f23814cf00906 : SUCCESS in 1h 15m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e47b3c83ca07451997fcbd0e1c9eada2 : SUCCESS in 30m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0274ea8f30894a04b9e7261bda008c9d : SUCCESS in 1h 49m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3a10563099504361ae4613a8024e1e20 : SUCCESS in 1h 47m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/976be82054284e4d8b73b9d8972e2bde : SUCCESS in 1h 18m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6d5e37b7479e4bb2bf1eb3a95934cbd2 : SUCCESS in 2h 00m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa45be85553f4d0c991d95ff6905f341 : SUCCESS in 43m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbd216a050f44ed896608df87d0b98e6 : SUCCESS in 33m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59dda17c3b334562887c6fe2f816f2fa : SUCCESS in 14m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/234cb99cde6242a49f980a37735b1646 : SUCCESS in 24m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8301173d0fd849c1aca6f6f88d248311 : SUCCESS in 32m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c018c411e5447579c3c02f41f37aabc : SUCCESS in 24m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ace237d59d4a86abc3f22bc64828f2 : SUCCESS in 7m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9d21040c3bd4622a158b2da5eac367e : SUCCESS in 1h 07m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0af45c7d6e38496aa211c08bf93ccdaa : SUCCESS in 1h 04m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6b56dd2b16b24b61973b3fd137596c34 : FAILURE in 2h 21m 45s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9493a13f03ed4361bda4c327f4c608c3 : SUCCESS in 1h 00m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d666c7f4555f4ead8ed1e5dd1226b90b : SUCCESS in 1h 01m 03s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/88db4f1b0d544c7cb1fe5ea2084f78ea : SUCCESS in 1h 52m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0f43e3ee8baa4b019bc73820483c8eac : SUCCESS in 1h 28m 45s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9015382abf44abda55c1cbd8c08ba7a : SUCCESS in 1h 35m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b37a511d79dd492bbd21fa2c34cdf21d : SUCCESS in 1h 10m 05s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b8bae360319e477ca18ad7a66e7180d9 : SUCCESS in 1h 39m 04s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6524774531e5459dbb5947392cf946e3 : SUCCESS in 1h 14m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9b72fda0e57f4423b8ed62a045bf7424 : SUCCESS in 28m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/569d2de510f34499a0added96205b48d : FAILURE in 1h 53m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ca57ebde4475dbd0f6134b61eb864d805fb07a20","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-03 09:36:35.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/55/772255/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"07d82f85b429c4a3ea69e562193bd00d671c5046","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-03 10:55:54.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"558a529c2e8f6b2e82400e78603f07c7800c4804","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-03 16:15:50.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f5b65f635ed5ade8c9aca2f9e4b4a5bce2032a82","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-03 22:19:35.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/55/772255/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"0c8748e6b8a7833cfe5d917e4ccc38f709a08b20","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-05 03:14:42.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"124120ea826b011d6ecec7d1a229aee03a213e19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-05 05:50:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d59a76bff61ca8f2be76fb8e8b0ad82d2909b264","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-05 06:05:31.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772255/9 : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":9},{"id":"d89d8900cfce5424f23b1e447c3a94b041eece08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 06:55:06.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/704fa93450a64e23b4452ff585403ff0 : SUCCESS in 46m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/afd1a0bb08f940fba28bcf143db14dcb : SUCCESS in 50m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a420de9bf391e5058f973d15b3debab547c6a862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 08:04:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0a5eb6b09de74c7ba077ad8be952d975 : SUCCESS in 49m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/934dca33bca4466396870b11d180ed59 : SUCCESS in 40m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1355c4805c8f4e6a85f56b29e46232c4 : SUCCESS in 1h 28m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/550f721d15a44429aa821edbf5d37f3f : SUCCESS in 1h 41m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cb29300c911447eeba8da1bf0d900643 : SUCCESS in 1h 17m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7fe81f341ceb48b6bf0c54587cb1d800 : SUCCESS in 2h 07m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbd1b213c44d4f4f8084b247484a9131 : SUCCESS in 36m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/869a209eff5f4c00b6d3b00c843d123b : SUCCESS in 28m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/911383e0ee214f00aac6b40befe4bbf5 : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5a87a64479e40d4a1f2601b1303f3ac : SUCCESS in 25m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa863d3621d743e186c7aeb4d25d9133 : SUCCESS in 33m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e57c7760b6541a38a5bf1037f99afb5 : SUCCESS in 26m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a387a69f3e9b40ee8fb9c7bc7b01454e : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9331cada0ea146b48da87c513dd462d3 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/84d72ab3772341e787858b463762c89b : SUCCESS in 1h 06m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5b14a7a4bf743719a2ba8238dafe976 : SUCCESS in 1h 30m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/14b28a4af00e49ab89d2b4f633fdfbfe : FAILURE in 2h 03m 17s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f9e532b91484ca9a27bc87b19fb1270 : SUCCESS in 1h 10m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/66696b372d5d43309c09a886d9a742e1 : SUCCESS in 1h 18m 51s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2b1a0b4abf8c45faa8a1cd3e66475d17 : SUCCESS in 1h 27m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7f52f6ba0ba74652b85409039cd0a0f7 : SUCCESS in 1h 34m 39s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c294e4971e84b16895bf6ee64dd0c15 : SUCCESS in 2h 07m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d41f32de79941b68ed1a96bb3f6a87c : SUCCESS in 1h 05m 44s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/3027e8415d6742f59d4f56e21cbf553f : SUCCESS in 1h 23m 46s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/95c0ea894f20465aa67d07a5d91f62d7 : SUCCESS in 1h 35m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4e6bdf4fcc2c49e5a49a16a6bfd939c8 : SUCCESS in 33m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5cf534fdd5fe48479c15a89cc57d94b7 : FAILURE in 2h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cb4dae12da21a2e454b188b0974c9ab764b8bc93","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-05 08:40:30.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c47cd2ea49a57c65e3f543b4f3a859f38dc10e54","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-05 19:36:13.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/9/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/9/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"00cbeab1696144c0cd752a8c6980f8a03d83b0b6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-06 01:18:57.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/55/772255/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"b91a9ed41040d1a0dcd81724fede3f6c61e49343","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-08 12:42:46.000000000","message":"Patch Set 9: Code-Review+1\n\nI am fine with it, the documentation will be crucial for this new feature","accounts_in_message":[],"_revision_number":9},{"id":"344e227063637b00c1b5afe77eb709f020c2e7e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 01:54:59.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f7aacbfbd2f5bd7bdd5ed51f11189e6db0ebb245","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 01:56:27.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"3fcfeccdd305f0420635f84c04d62f812e492d44","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 01:56:40.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"555aabe36718a74958ce794152e9ae84538ba248","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 02:49:25.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9474c3d6aa6a405da2ae39265c45feb5 : SUCCESS in 53m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29595967cf45406dbc1330f305da968d : SUCCESS in 35m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7c35b85ad37c2a1ddcabe6140ad3e3db5490990e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 03:46:19.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772255/10 : SUCCESS in 1h 49m 45s","accounts_in_message":[],"_revision_number":10},{"id":"a7d98d4db753d4b2a38c04c9931c4ee04638e14c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 04:19:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6c25fc5b51a7402dbd253fe323d1e732 : SUCCESS in 52m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c874b34ad602414895f9eeaf1368dc91 : SUCCESS in 1h 26m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a50b40d8160d4895969141e3676c5449 : SUCCESS in 1h 18m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/26491be1f49f420694de2a5df76d4bf1 : SUCCESS in 1h 28m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/396fb9bb4f434ce3a443aa40ca3fc47e : SUCCESS in 2h 22m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdcab8e678014eb7bea93f999bdec781 : SUCCESS in 36m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60591abfd26f4d02b2b6f0b2d76b27c8 : SUCCESS in 33m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdda4d47454646e7b51db118b2602429 : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f9cf042c3e74278bb4f407d7b2c940a : SUCCESS in 32m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c9d48904ece4ce19a3fc621d69b565c : SUCCESS in 38m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9af5905127bf4283b78a994eca1cd2b9 : TIMED_OUT in 41m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143bad1cdc614ead8a824a5a8f1f09f0 : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1185e7c21c1646a7bb232ccb92cd08a1 : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e12cc879d1e247fe96c182b00e4ff52f : SUCCESS in 44m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/869831a568c64f9a9414f2e7f38d181b : FAILURE in 1h 34m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/94d3b26c4d8345af8dc0d8e81adf5e16 : SUCCESS in 1h 25m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac5a687a493d4fe1a4942a0c8aeed5e8 : SUCCESS in 1h 00m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/af6d0686f63d4c0ea45f21bf30d44090 : SUCCESS in 1h 24m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dbe551c6b3ba49819043ce703ed98f59 : SUCCESS in 1h 28m 14s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/767b38ed7c5a49328fd02b2c3db25f10 : SUCCESS in 1h 25m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df73d93f127b437a850d39fe9666659b : SUCCESS in 1h 32m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c60046eb9b204aa6a4e69e14b29f707f : SUCCESS in 1h 06m 09s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/36ecd18caaf24095bc7b54913d48590c : SUCCESS in 1h 12m 17s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c70e02f9763045faa04cc7167b010bdc : SUCCESS in 1h 20m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8992b3b36a9d4f36945be15138ca40b5 : SUCCESS in 40m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1549ed90de146758b1d3f1eaf2a5c6e : SUCCESS in 1h 42m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d44de365267a4ab8a823f543cc9fb351 : SUCCESS in 1h 30m 34s","accounts_in_message":[],"_revision_number":10},{"id":"7e7a79b92ac6774e24b9fe75b26b3a035ccdc1af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 05:33:31.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/10/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/10/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"bf498e21f715709c85832741a7806de2db802fb6","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 07:59:56.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"e8d253ba3f42a6511cda06763c98cb8066f3b50c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-19 08:46:35.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"eb4369818ed8c943073f7786a63be52835e6b704","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 09:12:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/54548b4850cd49d0ab9efae6d9ded2c8 : SUCCESS in 52m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7dc00429272a4106a1ae9be783d4e5f5 : SUCCESS in 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"20334537afa8469bc9070d060f40139f652f5c69","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 09:18:11.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/10/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/10/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"97ff54b794b30faafa18dae7ddb9892d71eed144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 09:49:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/85df3f71295e464b9504909291d7a8f5 : SUCCESS in 42m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/908e4e4d41454604b92122b8c981571e : SUCCESS in 1h 47m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/af16c9dd1d4946af922a178f3881a53b : SUCCESS in 1h 20m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3abeffafe1c487cb22686b02ceb3b87 : SUCCESS in 1h 13m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/52c58b702cad400882ae7f42b14aedb6 : SUCCESS in 1h 33m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8651db538ce049d9b5f44a93589787ab : SUCCESS in 45m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/970e30ed4f924daf8fa32a4e70ccd882 : SUCCESS in 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3521e2f176ad4d0581ae30913398e958 : SUCCESS in 13m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4a9aea883e44568b2d648f9c3cdd02f : SUCCESS in 32m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a73f0c85f3164ffeaffd61871c02d516 : SUCCESS in 19m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f3f015c8900443da33e3c3366d8c860 : SUCCESS in 26m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24e3347342943ed8a30e85c314f436c : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c05ff4f690b4bc8b127316d4a3a535b : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f374c822c2df48b4b4078085fce88803 : SUCCESS in 1h 00m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9bb3b91c6f9b4a819a7e210d9d11f2a1 : SUCCESS in 1h 14m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fab15e76a7e3422eae7909b293e9483d : SUCCESS in 1h 36m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c075ff400864ea1973216c47646e06c : SUCCESS in 1h 02m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/24fc29048b3642a4aa3e936e977d54db : SUCCESS in 1h 01m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ab2c200d98dd4282a274f59fa258c463 : SUCCESS in 1h 15m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2423ba37d4b5418d8e98a47aa9652e2e : SUCCESS in 1h 32m 02s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f053462f66a4beab6d29aff59183893 : SUCCESS in 1h 38m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cef032dbe9c43c4b8353cc971f197f3 : SUCCESS in 59m 19s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/770d4f21852e4673a38affe80ecefe69 : SUCCESS in 53m 39s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d94d61264326433080120ce4de5df99a : SUCCESS in 1h 08m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cbbcf441d1084338b0d7050bacbdadc1 : SUCCESS in 30m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9c348524d614426af2636d91a76ec21 : SUCCESS in 1h 09m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/86baa7a830e84566a410a4ddbd4b1523 : SUCCESS in 1h 24m 34s","accounts_in_message":[],"_revision_number":10},{"id":"15923e63f37614854e70954fc9370ec377c600fb","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-19 09:52:19.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5132512048678901bfd3a922bdf157dee50cdd88","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 10:04:28.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772255/10 : FAILURE in 2h 02m 46s","accounts_in_message":[],"_revision_number":10},{"id":"b1921f7ba081213b151b649673dfd3ab33e7b0f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 15:23:10.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/10/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/10/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"aa9e096f7b63cceb5ee849baa2939ba552cc7974","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-20 01:32:59.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)\n\nlgtm","accounts_in_message":[],"_revision_number":10},{"id":"555e0e706d753fda6afba0a5b0b9b35456532d1d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-20 05:02:25.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/55/772255/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"9541fc5f9c272b5820a69eb9cd5fe4f61b9ebc61","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-02-20 05:59:23.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"a11d88a4ce2d2937553e3da8c82f2852643f6aed","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-20 12:37:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e59a06a4312f121b6f8a15dd83fef66968fd9ed4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-22 09:01:31.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8ff7fc84bc38b95aa93b5d47e694a353d6e878da","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-22 11:51:42.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9b6af9b07526c216079a02133648ea1eebd1d52b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-22 19:46:35.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"8f4b08e12ca573ac7b2e865fb059c3149be416f6","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-23 01:43:20.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ddfd33c9a30c22c7df0878b3ee7471d064b35587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-23 01:43:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"12712004554c8168f25d2646a85257208375edeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 02:36:20.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fda13ac791c54f4090c948aa0a24bc51 : SUCCESS in 46m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa46690bb81f4d7297c23c4e40e8be3a : SUCCESS in 41m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"fa4bec6f6a73d1b4f4cd80735f5f0149e62514b8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-23 02:43:33.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772255/11 : FAILURE in 57m 36s","accounts_in_message":[],"_revision_number":11},{"id":"3db5e0b9235702ee1f7c1a5952f04cbf934fdad8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-23 02:44:41.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/11/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/11/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":11},{"id":"e77bfca88f85998509dba2b04cdcb9711b7453c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 03:35:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/71e71d94a4e145e8a7519510cf146dbb : SUCCESS in 55m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/909cecdb61ac465b81f5f9d26391e8ff : SUCCESS in 1h 46m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f56f94d3a14843d68c6de8de1e1434df : SUCCESS in 1h 39m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/03f46e9feaf0424387f8edeb153fb073 : SUCCESS in 1h 36m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1daabefe641544c496853487e447ab64 : SUCCESS in 1h 47m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b056e405f1a4db38c427c455fc5bee1 : SUCCESS in 45m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/652c20088b254a17863cfd40e1387d3d : SUCCESS in 34m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5438da6e62f34e4c952c97634978fae2 : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/095403ac3bc141a6a89871cb6e648a6e : SUCCESS in 31m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a0b4a2fdf83409cbc4726ff28a8aa51 : SUCCESS in 31m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6b2be9cfa614a30ab55ff6dfbb920ae : SUCCESS in 34m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/672ce34d864d47658ead1325ef9ba239 : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37b7e6422fd04bf0b9da8f93a242b804 : SUCCESS in 7m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1092af2c6f0e436fb3480a0f77061eb4 : SUCCESS in 48m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9418f2ff33b42d5b1bc8f7822ce6ead : SUCCESS in 1h 12m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cbedae3dafb441dabd0f30a3fb62590c : SUCCESS in 1h 28m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e2d997be31042e1bc9119ef10f8fa4b : SUCCESS in 1h 05m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/91274b92be7f45f9961e0871ac61f095 : SUCCESS in 46m 32s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/723974690d6146268d451c5d1b37ac1d : SUCCESS in 1h 07m 15s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eab8cbd60737440ba66c93ec86acd3bf : FAILURE in 1h 28m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e249fc6d25d243378472ff3b3f1439eb : SUCCESS in 1h 37m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cf2969a365744d8b0c18ef52e55bd30 : SUCCESS in 1h 15m 20s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e24a136e7872471b93c61536252d355b : SUCCESS in 1h 10m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3aeb107f19ad4a8a95228aeebe853234 : SUCCESS in 41m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4717df94a9694a0d9ea737cdd3a53be8 : SUCCESS in 1h 33m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1920776de3254a64b73287461d2e3098 : SUCCESS in 1h 32m 34s","accounts_in_message":[],"_revision_number":11},{"id":"4f2e2083b9b2db0ac83d26f6fd4a0ad5e03c8947","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-23 05:21:45.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/11/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/11/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"eb0987c96d2310c7ce8e54ef7909a8efe7fc6ced","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-23 10:23:34.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f428fb4dc03dddee0b815afcb534d43bdcc19d7a","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-23 12:25:54.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1aedc14a7b5c3686237629466f54a68bd184739c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 10:17:10.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/55/772255/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"02af73056cc725eb7afe32c5089ae57aa5984336","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-24 19:23:23.000000000","message":"Patch Set 11: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"2e059308191c2ada97cf32d283a0502de380aae0","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-25 03:11:37.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"2de78be6371a1aa0d4b6fc396ee469efab9b4ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-25 03:11:49.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7ca84c05f041dba7a60ea953ceb3ee05f39c6c71","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 03:13:37.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"7fb3639c58255f49967cccae137f69438fe33911","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 04:05:25.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ddcadebda7ff41a39c54e8e11e6f3b7e : SUCCESS in 42m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b70fc8114f7041da8785e25d6e1a97ba : SUCCESS in 42m 58s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4fca7ae055dd9e41a85f568b865d2ee4dd0f5a62","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-25 04:34:11.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/55/772255/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"15213d705542a16e8d3a1d5b8a33e29340f5cb55","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 05:44:50.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/12/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/12/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"71e9a329aee1915e3298d137a449e6fd21c0fa1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 05:59:00.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cbcf38c11c494739ab4c4d9452deb5c6 : SUCCESS in 55m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9a41bc6f3e114049b3e27155aa64c8f1 : SUCCESS in 2h 05m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9a1a949004e54fbcbc3fde55ddaa0439 : SUCCESS in 1h 20m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7cbb698644e14ef1ad7d7676ce05885f : SUCCESS in 1h 49m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f899ec28ca72462e9c2c03ccf0815d9d : SUCCESS in 2h 11m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6adc8fdb82e04c9c8cc0e487d8f18315 : SUCCESS in 39m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c94b5d0e9844422b03de03bdee17d48 : SUCCESS in 35m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a7d26aa5374ecd9f6abbee64c40e6c : SUCCESS in 13m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4299065441184cf496a80937530b440f : SUCCESS in 29m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce9105ee239c4b969aa64dcf5b621654 : SUCCESS in 27m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddb01e95ccad4237a380dc2d467b8d24 : SUCCESS in 28m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0bc62b159a34afbb3d79b3635264912 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/282d0495d6384b2d808da0a0950a3207 : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/829cc33046da43f1bbdedfe37d82f96c : SUCCESS in 54m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d0a9c7f3a1b349efabafb99ef6091219 : TIMED_OUT in 2h 16m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5755a26dd45b41b5b88c727c2bc68e1d : SUCCESS in 1h 33m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3cfe8b8bb7e94058bbc62879bd5adce8 : SUCCESS in 54m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ea737e7ab08045e7a17e3913df36b68c : SUCCESS in 1h 04m 33s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c17230ae4c9345f98e2a76ee9a34355e : SUCCESS in 1h 24m 49s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/96a48a70e42a4845a30180b65c85017e : SUCCESS in 1h 33m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/622c6d68ca4b4b3d9031307a0194f657 : SUCCESS in 1h 51m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5310e2370ff49538f59d1e0b207f0af : SUCCESS in 52m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9cb1df9e4fc74e7face9152f8e49b61c : SUCCESS in 1h 26m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0627cb75a0194752ae86e32ebbe7fae2 : SUCCESS in 49m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e0f4eb493fb340fe8246414cae84a16a : SUCCESS in 1h 33m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7d8afeb58fe247bcab7646e1a9b5314b : SUCCESS in 1h 22m 22s","accounts_in_message":[],"_revision_number":12},{"id":"8d29b4eec1e7a8ed1d3d430f81f0590e38660642","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-25 08:44:03.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"82939a55d8aa26c36494f48c33b42ba67c4435b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 09:33:06.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/85de2dd759f64855baf9a43f00d02f46 : SUCCESS in 41m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e42032758f2475f9d9439422c119075 : SUCCESS in 43m 17s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ac9133d6c5685932a6fed479d4b3c413e2f33abe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 12:29:32.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f5ee601162a040cc94065c4ce4b1f38d : SUCCESS in 52m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ea040c63092c495ea28a4a8621156b9f : SUCCESS in 1h 59m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aed8928ef12d48889c061a6c12eb85a0 : SUCCESS in 1h 39m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05200e07b6ba4ef089d78c55138c06b0 : SUCCESS in 1h 52m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3377b7831a5b4eb5a1043db37e3db0ae : SUCCESS in 1h 32m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b8f1c40e3cc47498190ec76419efc52 : SUCCESS in 36m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f0825ba021e4b13b388df7094734fbe : SUCCESS in 30m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db837b8b9c844cea52c77a2dd9e7758 : SUCCESS in 12m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7f458b8176a4ad7942fa5ef9a495cb7 : SUCCESS in 28m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b2c1e1fab1a4219a0124686bf23637c : SUCCESS in 27m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eccba08627634bd1be453a5eca8810ed : SUCCESS in 31m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb1a70ba1fa483fa8ebd6ce35e3bf7d : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f13be3899834ab18bdf6fb753a832d8 : SUCCESS in 11m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4766f320d9e04af5939acacdc5ea11b5 : SUCCESS in 50m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3474653dcc2d4080a5f254adf727eca0 : SUCCESS in 1h 11m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/58b82df55e6445e482f353f01e3d0a7f : SUCCESS in 1h 58m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/03b6acb8be89449dbe9a485cc6460df5 : SUCCESS in 1h 07m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9b5d7a04ff9647eab88db74a1e26195f : SUCCESS in 1h 11m 48s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bc4613e42942423b9c9276804b7dc1c2 : SUCCESS in 1h 20m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/830648b2c35648a99849c1d873f473b5 : SUCCESS in 1h 29m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b5872acd0344a8d8b88b88891c2bbf7 : SUCCESS in 1h 36m 10s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c17ba5ac197424eaa9ec27764f6ddb4 : SUCCESS in 59m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/25c73c13d7ba4047920c41a3b246c247 : SUCCESS in 1h 06m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2b538ae8a2244c82a32d862cff07862e : SUCCESS in 38m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fae4b96a3efe4ccbaebd1c76419c6765 : SUCCESS in 1h 32m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/58140932548344219abb2e6a433ffce8 : SUCCESS in 1h 26m 26s","accounts_in_message":[],"_revision_number":12},{"id":"38d54e841330358714a137435c9b93819f7a9708","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 12:46:26.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/12/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/12/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"a35caccf4b03a56b8626a38c876f941fbc05bb56","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-25 12:59:34.000000000","message":"Patch Set 12: Code-Review-1\n\nCommit message to be updated accordingly","accounts_in_message":[],"_revision_number":12},{"id":"12737d7d011bfa1860282fd8a89ae299ea8a0ad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-02 06:31:20.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f37a1b48c71edfeab34d9b528da1e4b9316a62b3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-02 06:32:08.000000000","message":"Patch Set 13:\n\n\u003e Patch Set 12: Code-Review-1\n\u003e \n\u003e Commit message to be updated accordingly\n\nDone","accounts_in_message":[],"_revision_number":13},{"id":"6b347b4efe3578df13578a78540b2277429a29a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 07:21:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b828c9b0d46a4cf79cc167c2b36831c6 : SUCCESS in 37m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43851115f4b446d8a884b653cf7c8e2a : SUCCESS in 42m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"15a204e27065caf499b5a4406bb709245ec1f393","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-02 07:53:48.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/55/772255/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"3746d733b02eed19ad69440a882aeb3642f643a7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-02 08:45:26.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/13/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/13/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"b1f6b94916635e5f2c236bd3474d747a166aa3cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 09:20:16.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/473de06352864a0780ae48c5cd6c24f3 : SUCCESS in 1h 06m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4708d057908d4c30a5f32f2dc2889f94 : SUCCESS in 1h 39m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f5d59db9706c4fdf9ead2f5e7b5650f8 : SUCCESS in 1h 27m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e9317a748edf43fca1456e1d8296e23c : SUCCESS in 1h 21m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/277c3230bc2746cba1eda2f6375b5a13 : SUCCESS in 2h 28m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4196ad165e87416dbcebca782d5136b8 : SUCCESS in 43m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33d81f95f10241af98acd0d15633937d : SUCCESS in 34m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd5025e94c85461a9a1d41b9b2f90c47 : SUCCESS in 14m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94d6a5892737473597533c820471e897 : SUCCESS in 30m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1750792283042dba9ec5ab04f5b3b5e : SUCCESS in 31m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4262cc2632744c149649ded16fa4a6bd : SUCCESS in 27m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d48956ab8c41b994996d4031c413ac : SUCCESS in 15m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71073ba432c54b8cb2a3ee7d334a1f05 : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/838778e9988748058e2ee1ddc9d6c9e7 : SUCCESS in 50m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/87f7cdaaa85c49058ff9977b11a7d6b6 : SUCCESS in 1h 20m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/439d0536681d4fccb1a43119d6e951d8 : SUCCESS in 1h 25m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/88f38d1837b044ca866f2909106419a6 : SUCCESS in 59m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d5187027af5d42178b1405928590d71f : SUCCESS in 1h 06m 37s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/15418965d22f44a09d9b451a9dc5781a : SUCCESS in 1h 17m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4c9def8bd4d947c6b52ce7a02fc8c132 : SUCCESS in 1h 27m 03s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70ea92cf0e8a461880d659ccd0e1a8d7 : SUCCESS in 1h 38m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71b7dbf98c194fa4b95067c6496f90a2 : SUCCESS in 1h 00m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fe8b6009e16d414382cec2c420850528 : SUCCESS in 1h 07m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/542a1eec37dc470bb343d084be36ea79 : SUCCESS in 41m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9f6e3e1b3a4445e7a741be2a00d61205 : SUCCESS in 1h 32m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e38589e62f2640fb8f5ca3198ef0454a : SUCCESS in 1h 25m 30s","accounts_in_message":[],"_revision_number":13},{"id":"6c5f50484e2ea53937e83370c6fb097b5cac3b5c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 09:50:14.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"ee97b94bb07bf1b6109c3aa7a55f145878adbf7b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-02 12:07:18.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"bbb8b18d9c37a540df4f629bab68b892cf097421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-05 10:08:25.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"d302a6b6e118f041bdae25d4557d77ffc6c19e5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 11:20:26.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e85a9a7bbf8a4d2f97d072336e8e6cc6 : SUCCESS in 45m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a08257c43ea4c65a97f755480f9d856 : SUCCESS in 49m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"5371e3593e37ee6ac94e227feae71d3771ac7754","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 11:23:06.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/14/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/14/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"1d5f60f1d3a32c50a31fbd17105d1e6879c6d95c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 12:12:25.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8a55e317608a47ef8604f98c7a06efbd : SUCCESS in 49m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9665ef8575154752b271d5c921a182cc : SUCCESS in 1h 35m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3a933f30ef2f4c33b5bff63b9ec071ad : SUCCESS in 1h 50m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9909d97e038a4c00a9a047f148ffcf81 : SUCCESS in 1h 59m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fdd6d651e19e4b75a83f8c92a1c6709d : SUCCESS in 1h 40m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d08fde85cb8c4bb89fc7a118dda16af7 : SUCCESS in 39m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b2e1564a4474f55abfb858e562919ac : SUCCESS in 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a1c6724adce45e897251c2134a3bfbb : SUCCESS in 13m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a142d8f4809418ebc8794f0a5c5265f : SUCCESS in 37m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf8e6b55a3bd49499fed84d10080a69c : SUCCESS in 38m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea159a2197b240a097ee06e78652efd2 : SUCCESS in 31m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32cc1770cb524e4790ec3fc809f98f9b : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fb4c1b3f6ef4859b3c41b015648765e : SUCCESS in 10m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9dbba4ed3a8241ecb487f0dbc56c77c1 : SUCCESS in 50m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/741e8ab146ca4097a3584efd2e36e653 : SUCCESS in 1h 27m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/132f6cb3679f45f59c50c599247e1db2 : SUCCESS in 1h 32m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/22f869f4bbba4a88835841932f92bf99 : SUCCESS in 1h 09m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b17a347d978c49a2b355e9e2c3594177 : SUCCESS in 1h 03m 22s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b3208bdf62da4aa5b56d9ff9112b2e54 : SUCCESS in 1h 28m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3c67b1aa5c034d7b96b309ceee7a634a : FAILURE in 1h 05m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f38c01734f747e7b89de91ba8f148b2 : SUCCESS in 1h 43m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e65573e487414523be2ebd9b0c48f468 : SUCCESS in 1h 04m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4263a3243dda4482971a7c32a6d0d2c9 : SUCCESS in 1h 08m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/46731faf03de4e3eadc55df37079d339 : SUCCESS in 36m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f324f6857074d7aa2fba881a0628b62 : SUCCESS in 1h 35m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bdcd9860a92e439e802d4f45b37a1534 : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":14},{"id":"a27a2086dff62f6005901a0f49e4d485b5b41397","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-05 12:49:19.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\nI think we discussed all aspects of this new cfg option","accounts_in_message":[],"_revision_number":14},{"id":"765fd9a7b7443e7f5bd5616ea4845c1b63cc72ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 12:50:01.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"e63d634c2395c14a69b127948697a63df597e700","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-05 14:31:38.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/55/772255/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"5ccb9b6b435bb23dd0b0b6a0b519881028cd49b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:59:23.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aa6ab746011a48bab3a70da09348c87a : SUCCESS in 52m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2221810fcd17485da6ce6361ededbe12 : SUCCESS in 1h 49m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f5949e2a176b4cf6bf5cdc8546096939 : SUCCESS in 1h 48m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/feeb4183db2f4c2d80ff463d896cce77 : SUCCESS in 2h 06m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c4eca3cf54a4fa6aad7b63054fb0a3b : SUCCESS in 1h 28m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/015883ec71e94ebc8de96f79fb5ee39f : SUCCESS in 26m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef4127f27ca46e98ec725e832a230a9 : SUCCESS in 13m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2049a45851a41669923dd0f4201bf68 : SUCCESS in 28m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/780a6088881642cfb905942007ebef3e : SUCCESS in 32m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4d6ec92adda4763a1ee77f07de3b826 : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54abedf40c9344158b4ee96ef1fe52e6 : SUCCESS in 9m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e26ac36f9c542998beaad0a8595b90c : FAILURE in 44m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4edce5fb54e74f70b7749d9317ba956e : SUCCESS in 1h 16m 09s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bc560ad73d214beaaa4f0833403b908c : SUCCESS in 1h 47m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba2199b38a734fbc8690a7ad7bf22404 : SUCCESS in 1h 14m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/503b3b5b693d407aae251c0a7507a8e9 : SUCCESS in 1h 02m 09s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f31dbfe646943cab0aeeb137f3601e2 : SUCCESS in 1h 45m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/207b37e24bf94d5fb0070cae2c2d47be : SUCCESS in 1h 07m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eb600fbc5a304442b1a39bfa1aa363dc : SUCCESS in 1h 14m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/88c1a2a58cfb4fd1a31526e1bf4b12b3 : SUCCESS in 28m 49s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/475bcd5734044c73bbfddf33e88c337e : SUCCESS in 1h 31m 38s","accounts_in_message":[],"_revision_number":14},{"id":"f94cfeb82cc32b85ad791850207bc6c7bb0cb6f1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-05 15:28:46.000000000","message":"Patch Set 14: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"792b71baf6a6b25c407de9135f0e0c05572fbaad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:21:29.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c9ef55721174014a2d3d15326630b40 : SUCCESS in 42m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54296c7acc294f889c5b3afcd3dd27a9 : SUCCESS in 41m 52s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4041e00783a4f0e303a140b9209fa63f8e539422","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 18:25:44.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/772255/14/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/772255/14/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"554ccf91e017d840bad8dfa224066ae6f3c85f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:06:31.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/731210f1d32e4e4187a009d994d28f22 : SUCCESS in 55m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c00baa58ecb34126a9ca9c620b3b5697 : SUCCESS in 1h 44m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/766b3ff2189744b5a4ede5f4a7ac7f9b : SUCCESS in 1h 50m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7436a1b300d04f369115723129b7b993 : SUCCESS in 1h 40m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8f8371216d524729b053cdfa3eed1585 : SUCCESS in 1h 42m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa11b433e03e4c0b9fa892ff6596d714 : SUCCESS in 36m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33e7ddf2ab484e3e887bc124f5830462 : SUCCESS in 26m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf960f02a45c4b3bad1b80418700c8ba : SUCCESS in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/414eaafcffb34940b5257964c8a22857 : SUCCESS in 32m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4da7dd84a1d4094bbcea99859041bb6 : SUCCESS in 30m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8372d2f28ff8435eb95f4e94f98831e0 : SUCCESS in 28m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354df307fa7242bbbb670d7469b48322 : SUCCESS in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7201d2d462346b68a62672c8f253dcf : SUCCESS in 10m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5111374bf6e4409b921f27c6a95fd876 : SUCCESS in 49m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/df40762a19ab4a50a188893765e11fcf : SUCCESS in 1h 26m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3e77407b54854d7a81c05f064ca4791f : SUCCESS in 1h 28m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/06d3d4e088504260b39a981aa88ea724 : SUCCESS in 59m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/057a1b3c70b34ad281edc25bfe3877ad : SUCCESS in 59m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8fa0b7ba0fc94b0086a5318880ffbbf8 : SUCCESS in 1h 05m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/54c32b68a3c648a78f8cd984858ed525 : FAILURE in 1h 13m 48s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/64a4b87b342a47e59c6433367ebf3780 : SUCCESS in 1h 21m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa05a9a3cc034c218327754fb15b696f : SUCCESS in 1h 00m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/32e4c541ca7a483caafadb4f63f9d4ce : SUCCESS in 1h 07m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/84d808637bc048b8ac57873a8a0c9704 : SUCCESS in 34m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/10518ac25487425b99463d6af874d31b : SUCCESS in 1h 30m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/abbbcaaa550a487e8bc91c815afa967d : SUCCESS in 1h 27m 52s","accounts_in_message":[],"_revision_number":14},{"id":"98b8e9f65bb3798a7a1a36fe9dc91e2061e2b820","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 02:24:49.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"d9eb63fd90ba6b1b909515439a41467d8c3b8163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 15:00:56.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/258d1c31557e454a8814c3e208918713 : SUCCESS in 52m 11s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/aacd94063c3942dbb9cc4d4cc69e172f : SUCCESS in 1h 10m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/508a4936cc21459f911d93b4bfeada74 : SUCCESS in 1h 26m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e24cea8998bd4cd39708e9f365f4fc0e : SUCCESS in 1h 41m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d75c9171a4f842cc868bf9c9b9b60618 : SUCCESS in 1h 18m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cd5f3604bec4065aa7446b68271ce53 : SUCCESS in 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3af30a825f94daea77c615537db11e0 : SUCCESS in 12m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66f4ed47b1e1486999f1bb512db15a80 : SUCCESS in 26m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffae6769cc9346e7b71c5b9d9ef801c4 : SUCCESS in 37m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2cb430efacd4011bb661a65f53066b3 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91c5fce3189141e3bfbe4cf7dc9921c2 : SUCCESS in 8m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/348272f3b41b4dc980320b165dd6efc2 : SUCCESS in 47m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/73cab6221d5d4782aedb1040a5592dd0 : SUCCESS in 1h 21m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5e8499808174436f9f4876bb049ada0c : SUCCESS in 1h 19m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38c5edfe22d843e4af55f13441b63c8d : SUCCESS in 1h 05m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cc1f1d99f2ff4a14a95aa29f221b1370 : SUCCESS in 53m 19s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/974d7c095ea34cf8bfe89c3e7f4de628 : SUCCESS in 1h 49m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bc05f1e750f4e3a8ec1f5ba5f41062e : SUCCESS in 43m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4a57506292cf46e09e55b1e00fd67998 : SUCCESS in 1h 15m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/611daa8dc1ba4d1b858b2a5f9a84e48b : SUCCESS in 40m 33s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/77b2f6477b9e49149b3892c517e6f0d7 : SUCCESS in 1h 27m 08s","accounts_in_message":[],"_revision_number":14},{"id":"fac65402c74d463b1aeafd496efaf7df40eaf797","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 15:00:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"8c1bff99bf75dbef0cc057b410196e3a48c510b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 15:03:26.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15906715344047afb4a7da1c36214afd : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5069190fedc24a8a96b86d0a143eded6 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"8e3a83c21304c28054bb0134add1688e9f3a9140","revisions":{"273844b670db3d66d87e861467a300323f3ab6c9":{"kind":"REWORK","_number":1,"created":"2021-01-25 07:56:26.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/1"}}},"commit":{"parents":[{"commit":"2bf558f6ec1dab70a81ba988f78d587433f43539","subject":"Merge \"Improve DHCP agent\u0027s debug messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf558f6ec1dab70a81ba988f78d587433f43539"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-25 07:55:04.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``enable_distributed_dhcp`` [1]\nto disable:\n* DHCP scheduling for networks\n* DHCP privisioning block\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/5/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/273844b670db3d66d87e861467a300323f3ab6c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/273844b670db3d66d87e861467a300323f3ab6c9"}]},"branch":"refs/heads/master"},"557222141f46de2f5cb18174ecce0afae2b4a797":{"kind":"REWORK","_number":2,"created":"2021-01-25 08:56:08.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/2"}}},"commit":{"parents":[{"commit":"2bf558f6ec1dab70a81ba988f78d587433f43539","subject":"Merge \"Improve DHCP agent\u0027s debug messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf558f6ec1dab70a81ba988f78d587433f43539"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-25 08:55:24.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``enable_distributed_dhcp`` [1]\nto disable:\n* DHCP scheduling for networks\n* DHCP privisioning block\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/5/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/557222141f46de2f5cb18174ecce0afae2b4a797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/557222141f46de2f5cb18174ecce0afae2b4a797"}]},"branch":"refs/heads/master"},"b1f7f3f261257074494b4b7b267241f4cf76e206":{"kind":"REWORK","_number":3,"created":"2021-01-25 10:59:49.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/3"}}},"commit":{"parents":[{"commit":"2bf558f6ec1dab70a81ba988f78d587433f43539","subject":"Merge \"Improve DHCP agent\u0027s debug messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf558f6ec1dab70a81ba988f78d587433f43539"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-25 09:46:01.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``enable_distributed_dhcp`` [1]\nto disable:\n* DHCP scheduling for networks\n* DHCP privisioning block\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/5/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1f7f3f261257074494b4b7b267241f4cf76e206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1f7f3f261257074494b4b7b267241f4cf76e206"}]},"branch":"refs/heads/master"},"68b7a45fb05beda60ef19360b7faec9a199eb85c":{"kind":"REWORK","_number":4,"created":"2021-01-26 01:52:36.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/4"}}},"commit":{"parents":[{"commit":"2bf558f6ec1dab70a81ba988f78d587433f43539","subject":"Merge \"Improve DHCP agent\u0027s debug messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf558f6ec1dab70a81ba988f78d587433f43539"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-26 01:50:32.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``enable_distributed_dhcp`` [1]\nto disable:\n* DHCP provisioning block\n* DHCP API extension and scheduler\n* Network scheduling mechanism\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/5/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68b7a45fb05beda60ef19360b7faec9a199eb85c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68b7a45fb05beda60ef19360b7faec9a199eb85c"}]},"branch":"refs/heads/master"},"f9730803fb2ee12e662143c0508fa43aef30bfed":{"kind":"REWORK","_number":5,"created":"2021-01-27 09:04:44.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/5"}}},"commit":{"parents":[{"commit":"2bf558f6ec1dab70a81ba988f78d587433f43539","subject":"Merge \"Improve DHCP agent\u0027s debug messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf558f6ec1dab70a81ba988f78d587433f43539"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:02:05.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``disable_traditional_dhcp`` [1]\nto disable:\n* DHCP provisioning block\n* DHCP API extension and scheduler\n* Network scheduling mechanism\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/7/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9730803fb2ee12e662143c0508fa43aef30bfed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9730803fb2ee12e662143c0508fa43aef30bfed"}]},"branch":"refs/heads/master"},"52d315ddf32e0fd28e85e9fb9cdcdd07c4def599":{"kind":"REWORK","_number":6,"created":"2021-01-28 01:03:25.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/6"}}},"commit":{"parents":[{"commit":"2bf558f6ec1dab70a81ba988f78d587433f43539","subject":"Merge \"Improve DHCP agent\u0027s debug messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf558f6ec1dab70a81ba988f78d587433f43539"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-28 00:58:04.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``disable_traditional_dhcp`` [1]\nto disable:\n* DHCP provisioning block\n* DHCP API extension and scheduler\n* Network scheduling mechanism\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/7/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52d315ddf32e0fd28e85e9fb9cdcdd07c4def599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52d315ddf32e0fd28e85e9fb9cdcdd07c4def599"}]},"branch":"refs/heads/master"},"3b7e56f27b782ffa57f62dbfca2100ba60fdf14e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-02-01 03:37:46.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/7"}}},"commit":{"parents":[{"commit":"0fb63f729751d7e62ac107d3275d440632936978","subject":"Merge \"[ovn] Metadata agent: fix checking datapath type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fb63f729751d7e62ac107d3275d440632936978"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-01 03:36:57.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``disable_traditional_dhcp`` [1]\nto disable:\n* DHCP provisioning block\n* DHCP API extension and scheduler\n* Network scheduling mechanism\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/7/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b7e56f27b782ffa57f62dbfca2100ba60fdf14e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b7e56f27b782ffa57f62dbfca2100ba60fdf14e"}]},"branch":"refs/heads/master"},"43eb67a5e8749dff5bdbae2e0c06c78e4c27f4f6":{"kind":"REWORK","_number":8,"created":"2021-02-03 04:22:31.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/8"}}},"commit":{"parents":[{"commit":"00d5f7f2abed82bcd071b03cb186bdf1fff390eb","subject":"Merge \"Fix unused \u0027timeout\u0027 in OvsdbMonitor start\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00d5f7f2abed82bcd071b03cb186bdf1fff390eb"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-03 04:20:37.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``disable_traditional_dhcp`` [1]\nto disable:\n* DHCP provisioning block\n* DHCP API extension and scheduler\n* Network scheduling mechanism\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/7/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43eb67a5e8749dff5bdbae2e0c06c78e4c27f4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43eb67a5e8749dff5bdbae2e0c06c78e4c27f4f6"}]},"branch":"refs/heads/master"},"60572115e65c12cb6d13f59940f926958c23b658":{"kind":"REWORK","_number":9,"created":"2021-02-05 05:50:33.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/9"}}},"commit":{"parents":[{"commit":"09078a9d695ea93f9322ee15fcaf6f15d566c30b","subject":"Merge \"Fix losses of ovs flows when ovs is restarted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09078a9d695ea93f9322ee15fcaf6f15d566c30b"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-05 03:11:19.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``disable_traditional_dhcp`` [1]\nto disable:\n* DHCP provisioning block\n* DHCP scheduler API extension\n* Network scheduling mechanism\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/7/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/60572115e65c12cb6d13f59940f926958c23b658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/60572115e65c12cb6d13f59940f926958c23b658"}]},"branch":"refs/heads/master"},"58a11f2b9f0ba700ee50c4bfbaf5c8c2fef56739":{"kind":"REWORK","_number":10,"created":"2021-02-19 01:54:59.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/10"}}},"commit":{"parents":[{"commit":"90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13","subject":"Merge \"Bump neutron-lib to 2.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-19 01:53:54.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``disable_traditional_dhcp`` [1]\nto disable:\n* DHCP provisioning block\n* DHCP scheduler API extension\n* Network scheduling mechanism\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/7/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58a11f2b9f0ba700ee50c4bfbaf5c8c2fef56739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58a11f2b9f0ba700ee50c4bfbaf5c8c2fef56739"}]},"branch":"refs/heads/master"},"f919ff2e40226da7139d1b4af2915628b23c34d0":{"kind":"REWORK","_number":11,"created":"2021-02-23 01:43:56.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/11"}}},"commit":{"parents":[{"commit":"872650e4639cbc50b92f38c0fd4f07e5cec7e873","subject":"Merge \"Don\u0027t update AgentCache when Chassis_Private.chassis \u003d\u003d []\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/872650e4639cbc50b92f38c0fd4f07e5cec7e873"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-23 01:26:40.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``disable_traditional_dhcp`` [1]\nto disable:\n* DHCP provisioning block\n* DHCP scheduler API extension\n* Network scheduling mechanism\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/7/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f919ff2e40226da7139d1b4af2915628b23c34d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f919ff2e40226da7139d1b4af2915628b23c34d0"}]},"branch":"refs/heads/master"},"1bdeab6a5323492f54af64626c26ba94c6dffd7c":{"kind":"REWORK","_number":12,"created":"2021-02-25 03:11:49.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/12"}}},"commit":{"parents":[{"commit":"084fa99874e882649aa6895941fa913b2b852868","subject":"Merge \"Limit execution of the neutron CI jobs on some irrelevant file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/084fa99874e882649aa6895941fa913b2b852868"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-25 01:58:24.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``disable_traditional_dhcp`` [1]\nto disable:\n* DHCP provisioning block\n* DHCP scheduler API extension\n* Network scheduling mechanism\n* DHCP RPC/notification\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/768588/7/specs/wallaby/distributed_dhcp.rst#185\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bdeab6a5323492f54af64626c26ba94c6dffd7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bdeab6a5323492f54af64626c26ba94c6dffd7c"}]},"branch":"refs/heads/master"},"f216a864d33bd21cd4ee9f5a86d1bab3ae09172d":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":13,"created":"2021-03-02 06:31:20.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/13"}}},"commit":{"parents":[{"commit":"77ee0847f5fe886ae6a1c4a9b9949d3590567e0d","subject":"Merge \"[FT] Reset \"AgentCache\" singleton in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77ee0847f5fe886ae6a1c4a9b9949d3590567e0d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-02 03:37:40.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``enable_traditional_dhcp``,\nif set False, neutron-server will disable:\n* DHCP provisioning block\n* DHCP scheduler API extension\n* Network scheduling mechanism\n* DHCP RPC/notification\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f216a864d33bd21cd4ee9f5a86d1bab3ae09172d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f216a864d33bd21cd4ee9f5a86d1bab3ae09172d"}]},"branch":"refs/heads/master"},"8e3a83c21304c28054bb0134add1688e9f3a9140":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-03-05 10:08:25.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/55/772255/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/772255/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/772255/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/772255/14"}}},"commit":{"parents":[{"commit":"cb64e3a19fdddb3eac593114a482c9dd69be68d5","subject":"Merge \"Add new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb64e3a19fdddb3eac593114a482c9dd69be68d5"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-15 06:24:44.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-05 06:35:29.000000000","tz":480},"subject":"Config option to disable the DHCP functions","message":"Config option to disable the DHCP functions\n\nThis patch adds a new config option ``enable_traditional_dhcp``,\nif set False, neutron-server will disable:\n* DHCP provisioning block\n* DHCP scheduler API extension\n* Network scheduling mechanism\n* DHCP RPC/notification\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: Icfbfc9691c5cf837406ff4291b3e3ed4970b26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e3a83c21304c28054bb0134add1688e9f3a9140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e3a83c21304c28054bb0134add1688e9f3a9140"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
