)]}'
{"id":"openstack%2Fneutron~772402","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Iad29cb34825adaa8c766d01b192a6bbe9992148b","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1905726-stable/victoria","hashtags":[],"change_id":"Iad29cb34825adaa8c766d01b192a6bbe9992148b","subject":"[QoS] Get only min bw rules when extending port dict","status":"MERGED","created":"2021-01-26 07:09:59.000000000","updated":"2021-01-28 16:12:05.000000000","submitted":"2021-01-28 16:07:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772402-bug/1905726-stable/victoria","cherry_pick_of_change":764433,"cherry_pick_of_patch_set":7,"meta_rev_id":"44320927e04dbeda11e6633cb097dd0040c791c7","_number":772402,"virtual_id_number":772402,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-28 16:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-01-26 09:50:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-01-28 11:10:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-01-28 08:06:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-01-28 11:10:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 07:25:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-01-26 09:09:37.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-26 09:50:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-01-26 14:02:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 11:10:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"cf27f664f7178d3869a913c5aa8a2f743cb85c84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-26 07:09:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b209249fda87776a36fa7a7073593089ec002935","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-26 07:25:46.000000000","message":"Patch Set 1: Code-Review+1\n\nClean backport","accounts_in_message":[],"_revision_number":1},{"id":"ae6550a5391eb71a8f954cc28f04de2e2cedd624","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-26 09:09:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772402/1 : SUCCESS in 1h 58m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6d42c55fdcf7b7f09b1096368db7e37abf27e121","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-26 09:42:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f62173e63104cd808c581836c3e996204902d8ee","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-26 09:50:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1e30b3c67defe4c42f38e250b453a6961e9d3b26","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-26 12:03:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772402/2 : SUCCESS in 2h 19m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a76ae2adcfa244ce52af6eed3e9042f02c4f086d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 14:02:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/4ebd63ed7b944528ae15e66b80c53706 : SUCCESS in 44m 35s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/75affd9678984e1e98d283b5918a0224 : SUCCESS in 28m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/bf664a85091d4954ab1249323bd715d4 : SUCCESS in 1h 39m 50s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/07138b9cbb074f81bcdcca7808fb40f0 : SUCCESS in 1h 36m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/36eab4841a5e413dba91d9b6ee7ec58e : SUCCESS in 1h 30m 42s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/2171d2d877e74396a3aa09bd1568c557 : SUCCESS in 1h 32m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/077e24e0ad2640dc98613680a12c2894 : SUCCESS in 31m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94e62d89ff447cc9cddc4c311247de9 : SUCCESS in 11m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88111675dd244681aa5c9d49843cf505 : SUCCESS in 24m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ad17e8b6dd94e3fbc1b1c731a349c36 : SUCCESS in 28m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb53096298a24bacbcd16dacde530757 : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42cbf9fbaaa24385ba98d72163ccf92f : SUCCESS in 1h 08m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb3ebba13b634ea4bed723009453ae36 : SUCCESS in 1h 08m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/66a9992cc30a49128c77193715e9b744 : SUCCESS in 1h 49m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c1e4f0c31314fe4941707c779544af4 : SUCCESS in 1h 12m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/36175f9a5636412280d52d43ebfa4f09 : SUCCESS in 1h 08m 40s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/3b11cbc675ee432598bdeed594c0f7ae : SUCCESS in 1h 26m 28s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ff0cf3c3a7724687b1d40f8e1e0ceeca : SUCCESS in 1h 14m 31s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d9220ca1aea45b89711e8177ba2ee97 : SUCCESS in 1h 26m 27s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8e3c892bd35f4e24a77e24bcf3e4d8a9 : SUCCESS in 1h 34m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a133507877248cfb5b53ecc20b1847c : FAILURE in 2h 00m 53s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5361baa2848e4043ade53f8a53bd3a26 : SUCCESS in 57m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c8e9099ed0414ef98537afc6e78264ce : SUCCESS in 1h 19m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d132e7e1756d4552a8af552bacb12ea6 : SUCCESS in 2h 06m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34cd95e9b1904bc49f65e6b698568567 : SUCCESS in 1h 10m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2f6890cb98904071a3b7c7c77de23186 : SUCCESS in 1h 25m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/916d8554299449d283d1f18c03925601 : FAILURE in 35m 48s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/096298a5addd466db250ac6d42a2f1dc : SUCCESS in 1h 17m 41s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4dc21ac328ea47e9adab8883d670400d : SUCCESS in 3h 27m 05s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e445f7bd6725450bb0f1e3a32f688a69 : SUCCESS in 1h 28m 13s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2b7fdc6b96ef4d1a9199722101a63ac9 : SUCCESS in 1h 09m 14s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/152c69128ac74bcdb6207dec695d9e60 : SUCCESS in 2h 02m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/84e287b4fb9148e1b223189bb2314606 : FAILURE in 37m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16b36951e0e66a3413276abf6104389ca196c5a4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-28 08:06:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5ba001b3b5dc1f5dca5da6949dd4d55f248e30e9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-28 11:10:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"96464bd76e11da034663ef7c19558323df23f1d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 11:12:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d417aa03401457915e1e5c522247a56bc641dbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:07:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/887676fa37304969a9708aff83ee019d : SUCCESS in 42m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69b1e59f0abf4fae875505163d0bf42a : SUCCESS in 12m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55b05a81ff3841098ee06ebf9123f9b7 : SUCCESS in 26m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eddd7734f424c58a03fb8bc5a2862d0 : SUCCESS in 27m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f8f5ff9d614dfb88071bc3acec685d : SUCCESS in 11m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bca6a11695ce4134af5887fc76628900 : SUCCESS in 1h 10m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a6c9b9d7a6142f38206cb9275cca595 : SUCCESS in 1h 10m 39s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb26fde58bcd4acea2b627f2a28c1d2e : SUCCESS in 1h 21m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1eedce835e0249c4991d23b1c70964d1 : SUCCESS in 1h 08m 53s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d05de38901b849aba840aacae4a7cf9a : SUCCESS in 1h 11m 29s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b89bf10f1a9642deaf55ca83167dd96f : SUCCESS in 1h 41m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8164c5b48fe746849199f8dbd3900e52 : SUCCESS in 1h 04m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5fa0307021194c2ea99db6e1af5815f7 : SUCCESS in 1h 12m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/599d994914fe4c77add3962a0658e565 : SUCCESS in 2h 10m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96907d6a4b394b37a0b73f0dfaf15851 : SUCCESS in 1h 02m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e6d450a52db2479e986b12bf4e41e2e9 : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9a816afd3bd308dba62084548eaf8981eb9c3c2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:07:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"44320927e04dbeda11e6633cb097dd0040c791c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:12:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4cdc38f7b1c46beb88567cac5158152 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5e7d7c8f1d93996a4535588830b5b7b190f7eef9","revisions":{"be8f30dc925f74210b9918b47690468dc6c02284":{"kind":"REWORK","_number":1,"created":"2021-01-26 07:09:59.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/02/772402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/772402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/772402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/772402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/772402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/772402/1"}}},"commit":{"parents":[{"commit":"0a02297ea944896328bbb5722d05d8225609f36f","subject":"Merge \"Use consistent filter API syntax\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a02297ea944896328bbb5722d05d8225609f36f"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2020-11-27 09:37:42.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-01-26 07:09:59.000000000","tz":0},"subject":"[QoS] Get only min bw rules when extending port dict","message":"[QoS] Get only min bw rules when extending port dict\n\nOnly min bw rules are required hence no need to fetch\nall rules of QoS policy.\nAlso no need to get qos policy from DB\n\nPartial-Bug: #1905726\nChange-Id: Iad29cb34825adaa8c766d01b192a6bbe9992148b\n(cherry picked from commit 334fa1ba276e0d375724b0203999643e04a5724e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be8f30dc925f74210b9918b47690468dc6c02284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be8f30dc925f74210b9918b47690468dc6c02284"}]},"branch":"refs/heads/stable/victoria"},"5e7d7c8f1d93996a4535588830b5b7b190f7eef9":{"kind":"REWORK","_number":2,"created":"2021-01-26 09:42:37.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/02/772402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/772402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/772402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/772402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/772402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/772402/2"}}},"commit":{"parents":[{"commit":"0a02297ea944896328bbb5722d05d8225609f36f","subject":"Merge \"Use consistent filter API syntax\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a02297ea944896328bbb5722d05d8225609f36f"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2020-11-27 09:37:42.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-01-26 09:42:31.000000000","tz":0},"subject":"[QoS] Get only min bw rules when extending port dict","message":"[QoS] Get only min bw rules when extending port dict\n\nOnly min bw rules are required hence no need to fetch\nall rules of QoS policy.\nAlso no need to get qos policy from DB\n\nPartial-Bug: #1905726\nChange-Id: Iad29cb34825adaa8c766d01b192a6bbe9992148b\n(cherry picked from commit 334fa1ba276e0d375724b0203999643e04a5724e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e7d7c8f1d93996a4535588830b5b7b190f7eef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e7d7c8f1d93996a4535588830b5b7b190f7eef9"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
