)]}'
{"id":"openstack%2Fneutron~772406","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Ib2f3aca08946e584156d092c37e1ea5ed5ca81a6","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1902998-stable/victoria","hashtags":[],"change_id":"Ib2f3aca08946e584156d092c37e1ea5ed5ca81a6","subject":"Auto-remove floating agent gw ports on net/subnet delete","status":"MERGED","created":"2021-01-26 10:55:25.000000000","updated":"2021-02-02 04:49:01.000000000","submitted":"2021-02-02 04:46:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772406-bug/1902998-stable/victoria","cherry_pick_of_change":761498,"cherry_pick_of_patch_set":3,"meta_rev_id":"e31f6140aa89dd7cc1f77fa51623aaf3e8c3e27c","_number":772406,"virtual_id_number":772406,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 04:46:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-01-28 11:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-02-02 01:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-02-02 01:31:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 13:14:55.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-26 15:09:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 11:15:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-02 01:31:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"f9d01778a9050cf2e28d8a547c8afb421a19ea16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-26 10:55:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"03e375dfe3c966cbef14e698f99eddb3fc534234","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-26 13:14:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772406/1 : SUCCESS in 2h 18m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1ca9bafd6a0016072ccc316ccc63dcf419c427bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 15:09:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/900f66ec594547af99773a108ad74f67 : SUCCESS in 49m 37s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/0ad177fc6477436ab6e63df33253e7f0 : SUCCESS in 34m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/f4945c2efbd647f08fb168db838aaac7 : SUCCESS in 1h 52m 05s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/08258c5f3c9e4814a678ce1ad081548c : SUCCESS in 1h 41m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/945e44ad1c8f4504a1434de0465493e6 : SUCCESS in 1h 27m 50s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/96e3afe2232a405d81f2ba0190798646 : SUCCESS in 1h 27m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88798f29879c42f09daa0389e4236359 : SUCCESS in 40m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd0591945d343718d55aab722f0e6b4 : SUCCESS in 13m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a48174c6eb4433e9397f3cd1095fcfe : SUCCESS in 26m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/375fcdf399ff485aab183e4d8753be0a : SUCCESS in 26m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2ed5305bffa4bd3aaba8c9b51ce6494 : SUCCESS in 10m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/528670830382480e9d46eda25bc143e7 : SUCCESS in 1h 04m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/26305c53f6704a06ab42814acbd89683 : SUCCESS in 1h 19m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b178172a6042422692275861d412d2bd : SUCCESS in 1h 47m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbdf285ea646462e907fe778160b6d2a : SUCCESS in 1h 28m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0fc5f10b517247c797c100aaaac7cbb7 : SUCCESS in 1h 10m 19s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/63888cb08b1a454586e6c2be41cd48a5 : SUCCESS in 1h 37m 22s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7c596d9711574415ab76ac7c1c7bbdb9 : SUCCESS in 1h 08m 48s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ab84448870443beb235731001b27a9a : SUCCESS in 1h 27m 22s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6498d5f3f1f84488b8b01e47990ca3eb : SUCCESS in 1h 28m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ec6ba05e61944bc09af7aaf3219d77a2 : FAILURE in 2h 03m 29s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/737a59cda7de4cac99ed1e983ca4fd96 : SUCCESS in 1h 18m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2e91124442bf425da691bc9331404d65 : SUCCESS in 1h 16m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f672a88a440f491abce55f2102a65a07 : SUCCESS in 2h 27m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56da4072de884d6aaeece0dc564fa13d : SUCCESS in 55m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/378e32777af74f4a956ffb48d57a0df9 : SUCCESS in 1h 16m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5817f28032f3485cbed2e839e0685ca1 : FAILURE in 35m 13s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a937908ccfc2446899deff654dad95ce : SUCCESS in 1h 27m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e53a714b94664a7bba7110e88acb711c : SUCCESS in 3h 09m 45s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/055a5be332bd464e985afe1119a673f5 : SUCCESS in 1h 14m 49s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/34314d23a74d4ee2a9b42feb4be233ca : SUCCESS in 2h 04m 48s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d6455c9bb034439289f60817ff269f03 : SUCCESS in 1h 55m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/832798aea18e4ca9b91c3a0eb1a46a4c : FAILURE in 56m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b47c53be7f4fceaef8aeb190672ab7886a2414cc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-28 11:15:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"073ccf1daf9fa2d954680d69cb991fc41f047686","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-02 01:31:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3674fa88bcaf52a4640e7a2e19a87fd70ef739f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 01:31:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7007ff5145a28e54556f5da878b48d23740ee6cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 04:46:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/89fd4440cdfd4daebccc716a511103a9 : SUCCESS in 58m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82556df439474d6dbb9d350d9c52751b : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84bf8d16a88944afb7c97c3870e8a435 : SUCCESS in 27m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fd6588873744311bf249510a4c4ac19 : SUCCESS in 25m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3684580dd5a74efa9130053db1174523 : SUCCESS in 8m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/88648a22ab1244728fcd7ce00b76398c : SUCCESS in 1h 08m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f995362bed784353b117c6d37d39ff7a : SUCCESS in 1h 08m 35s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f8c1cf1993c478ca043917ba561d2c3 : SUCCESS in 1h 12m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2e23a91c48db48efa31afba3eadf3b40 : SUCCESS in 1h 08m 10s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3dc980d2d6be4ed4b6fee5556f1ca914 : SUCCESS in 1h 21m 47s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f807f9a52d3c4af191ead64449c8e03e : SUCCESS in 1h 29m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e43e292b1894da89061a53c4f649535 : SUCCESS in 1h 06m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/165f334cb7784e108577c2af1e24eed8 : SUCCESS in 1h 15m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0eac4368964f489f93ec7220ce9ae671 : SUCCESS in 2h 04m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b85bcf67084448c3b047f10a3d90e3b6 : SUCCESS in 1h 00m 58s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/acce0c61900b4f48b498e67c192706d6 : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d05b1415587b073e80a8ec20071fb82797df170f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 04:46:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e31f6140aa89dd7cc1f77fa51623aaf3e8c3e27c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 04:49:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26a838a892d442bbdbd197547205ea9 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b4211d5323f44b32ceb38351383247a8cc5f44d","revisions":{"9b4211d5323f44b32ceb38351383247a8cc5f44d":{"kind":"REWORK","_number":1,"created":"2021-01-26 10:55:25.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/06/772406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/06/772406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/772406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/772406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/772406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/06/772406/1"}}},"commit":{"parents":[{"commit":"0a02297ea944896328bbb5722d05d8225609f36f","subject":"Merge \"Use consistent filter API syntax\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a02297ea944896328bbb5722d05d8225609f36f"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2020-11-05 06:41:15.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"oleg.bondarev@huawei.com","date":"2021-01-26 10:55:25.000000000","tz":0},"subject":"Auto-remove floating agent gw ports on net/subnet delete","message":"Auto-remove floating agent gw ports on net/subnet delete\n\nfip agent gw ports may be left in DB after router removal due to\nrace condition between l3 agent and server: when server processes\n\"router delete\" - l3 agent is still processing \"router add\" and creates\nfip agent gw port after server already removed the router.\n\nThe patch also adds handling of external network delete event\nto cleanup fip namespaces left on agents due to same race condition.\n\nChange-Id: Ib2f3aca08946e584156d092c37e1ea5ed5ca81a6\nCloses-Bug: #1902998\n(cherry picked from commit b97a8eb488aec0ee9afed0ceeb3d45caeb37ffdb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b4211d5323f44b32ceb38351383247a8cc5f44d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b4211d5323f44b32ceb38351383247a8cc5f44d"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
