)]}'
{"id":"openstack%2Fneutron~772460","triplet_id":"openstack%2Fneutron~master~I9d9e2bd4add5bb6fa4105352bfda739340932571","project":"openstack/neutron","branch":"master","topic":"bp/address-groups-in-sg-rules","hashtags":[],"change_id":"I9d9e2bd4add5bb6fa4105352bfda739340932571","subject":"Allow sharing of address groups via RBAC mechanism","status":"MERGED","created":"2021-01-26 01:09:05.000000000","updated":"2021-03-03 03:17:05.000000000","submitted":"2021-03-03 03:15:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772460-bp/address-groups-in-sg-rules","meta_rev_id":"45845cf12b75c0071f12171c28e2c5ae49989566","_number":772460,"virtual_id_number":772460,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-02 02:02:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-03 03:15:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-02 14:14:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2021-03-02 09:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-03-02 14:14:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2021-03-02 09:28:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 01:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-26 02:16:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 05:16:29.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-26 16:27:44.000000000","updated_by":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"reviewer":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"state":"REVIEWER"},{"updated":"2021-02-02 02:47:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 09:51:25.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-15 15:22:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-18 02:05:01.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-23 14:22:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b3704149525e0a9e345d4e4a7dcf50a8986b1dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-26 01:09:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66936efdc93952dfbe3c03e2d420f84c8c7a0722","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-26 01:10:16.000000000","message":"Topic rbac-for-address-groups removed","accounts_in_message":[],"_revision_number":1},{"id":"38b26855577c01ff805927cdeaf3667911521dc8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-26 01:11:39.000000000","message":"Topic set to bp/address-groups-in-sg-rules","accounts_in_message":[],"_revision_number":1},{"id":"c78133799968d728169524381d129d3299188fd0","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-26 01:11:50.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1a2f8a6807bcac4da45cd4e781e3a6adc6c92249","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 01:48:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b893f6cebeba499192bb25940ab132bd : FAILURE in 28m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a928dc21fe164f0d855cb4a476ee5632 : FAILURE in 28m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"560cd166057fae8bac87788a6215e25e170b6961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 02:16:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ef75727dee2a46ce8c71480a81dafcaf : FAILURE in 24m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8d291f4b054e45189c5fbae56607c900 : FAILURE in 23m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5e500c442c3e45fe936556d35e513dda : FAILURE in 19m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/85b3656b45504673b891786d01bda6ef : FAILURE in 22m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ba04e1af0ba411394f4203ae5b58284 : FAILURE in 20m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/666e77699344437fad077619dde94572 : FAILURE in 24m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6fb2963981849c8bc1f4c2b43d1539b : FAILURE in 22m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2d170a10ae54e44872c7539453b6613 : FAILURE in 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cffa8c0678a44d8cbe7e26091dff7594 : SUCCESS in 12m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/988bd0646cb6457787dab484d3f8fe27 : FAILURE in 19m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfca5ede2a62461aa15906fcc8b3c6c1 : FAILURE in 20m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea617dee905f481ebee853952f55579a : FAILURE in 19m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2369bf0bedbd493a87d8f04ca9e8f50e : SUCCESS in 8m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/81ef1621105b4412bc11b7069e2a12f0 : FAILURE in 58m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/86789a416082488ebfbbbe1cffb3d004 : FAILURE in 53m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/87b8b17a1afc40a6b6702fd5f4733894 : FAILURE in 23m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c45e676fbc44d308e21163996677703 : FAILURE in 55m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5398f487920c4a81b1568c911c76de42 : FAILURE in 53m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b70db1f9a3e648efa04527d6743c301c : FAILURE in 27m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ae77aa534d8145c8bf6355ee48a69570 : FAILURE in 28m 50s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87e93122d4f347df9e7440f735e1cbe4 : FAILURE in 29m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abffb3ef42634be8abef98f321bad88f : FAILURE in 30m 43s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/eaf092539325423b9ae6d4a8999f614c : FAILURE in 27m 06s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6a4e32fd7c4745b6b43df97e36873d47 : FAILURE in 25m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f38c9e29d17742d98bc5cdc53ef33692 : FAILURE in 24m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f67f5eb577fc4d95bc903aeeb0e0f1db : FAILURE in 27m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7da0845216011304c66faf5718ab0ca013359827","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-26 05:16:29.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772460/1 : FAILURE in 4h 06m 20s","accounts_in_message":[],"_revision_number":1},{"id":"206cb9efed6775f472e45c5f22e56da89990f414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-26 23:54:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41efca4bd210eb85e4305652905930d236de9ba0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-26 23:56:21.000000000","message":"Topic rbac-for-address-groups removed","accounts_in_message":[],"_revision_number":2},{"id":"80d45c40f9c2b6b0a37b6a2065511ef00cd8c697","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-26 23:56:31.000000000","message":"Topic set to bp/address-groups-in-sg-rules","accounts_in_message":[],"_revision_number":2},{"id":"3671e2bf8f25d9b34ddc9433a4fbbd5935f03018","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 00:37:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a575060d588a4034854868ba1729852f : FAILURE in 35m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f1221cfa4504e1d9d8e2eab17d3bc2a : FAILURE in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"840d974ec6f4129d07c3dab5b7647423064756c7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-27 01:39:47.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772460/2 : FAILURE in 1h 44m 39s","accounts_in_message":[],"_revision_number":2},{"id":"58e2e2f4445ae581c799c475d2509ec03a487164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:06:01.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/521e23dbcd9f498aa7a4b69578bd96eb : SUCCESS in 57m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/49fda169810a44a49d99825d6e0060d6 : SUCCESS in 27m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/178b8fc55ea142f9bbdaaf6177867ca0 : SUCCESS in 1h 18m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a3b461c6b5ea498eabd1c0120dcd9a2d : SUCCESS in 1h 17m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e281d2e03b9e44eb974871ca703386e1 : SUCCESS in 1h 14m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3a629d193a0f494a934f9ab158257d1c : SUCCESS in 2h 08m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/294fbbbd12614722bbc27bce537d7114 : FAILURE in 34m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20507f7016e74e36a61f8a99a8e7cf21 : FAILURE in 27m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/413208504593419fbbab793bb16ec4f1 : FAILURE in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76a0819e4ea242ec96390a2ac6b26561 : FAILURE in 26m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7527256aba5648bdbb7d1cea7e58505a : FAILURE in 19m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f42ddb837a1341a283234ef280fc4bdb : FAILURE in 18m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eddb036e882e48619ea18b7ba89d5c06 : SUCCESS in 8m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6256a8aa369d45f192405e1121b3fe73 : SUCCESS in 1h 03m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f9f163919f4748429874f98e5cf10a2e : SUCCESS in 51m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c69855d2dd37403a9e3ff407dfd3d127 : SUCCESS in 1h 31m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/389913de968a43f2ac6374d0ea551b13 : SUCCESS in 57m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f7d4d1bd4c1747d7ab3763697fad13b3 : SUCCESS in 53m 55s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b41c6ba7d44d40789fb2bfb8e816dfb2 : SUCCESS in 1h 12m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9677736094024dc7a3013729f152e475 : SUCCESS in 1h 34m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4e5be0e8d9e4049baac6f163eb91515 : SUCCESS in 1h 34m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/154a2ca372354d5b9d50c34ac41365b0 : SUCCESS in 59m 26s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2b81d30e16fd43f0848e666de2e95a7b : SUCCESS in 52m 14s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dc099a22519a44f8aa2d05ce5b3a1070 : SUCCESS in 1h 05m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7f23e50af13d46048102aea71281c1be : SUCCESS in 34m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/772d520c7ad047d1b97b33f44dfd3668 : SUCCESS in 1h 18m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0fd5b58117b0387a501ced8e5504ed522abb2c41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-27 22:34:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87118acf4e071b4022730bc48bd37c2cc1ce2d13","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-27 22:35:31.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"c406774fddd8a5d26506d8d145294f84c0f35a17","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-27 22:36:42.000000000","message":"Topic rbac-for-address-groups removed","accounts_in_message":[],"_revision_number":3},{"id":"c78b9c8bf5c78973593c45bd1b15690054ff95e7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-27 22:36:58.000000000","message":"Topic set to bp/address-groups-in-sg-rules","accounts_in_message":[],"_revision_number":3},{"id":"f4c131450fec0e10a620e18f7795bcee7d43ff9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 23:18:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01d94d784bbb4dd58550ef8bb76b9355 : FAILURE in 35m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c27eca0a379b427d94129dfa6df78371 : FAILURE in 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"37db58d39548e55592e41b0be3b7e418dda01c84","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-28 00:27:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772460/3 : SUCCESS in 1h 52m 23s","accounts_in_message":[],"_revision_number":3},{"id":"2cb507f184ff95747eaab2aa20faed014cedcb35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 02:14:03.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e1901ce80d014cc0a58d25093e9bc13e : SUCCESS in 50m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/52bdd538568e48c59b1eae643dcb218c : SUCCESS in 34m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9d7e50e793b349d4a902c9f14f055d3c : SUCCESS in 1h 32m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/101cdf2d803e4041a8d7ea8ccb5d75d4 : SUCCESS in 1h 46m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7f28a2ca843940789826f3c2e1ea21a8 : SUCCESS in 1h 31m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c87041bea66d4a5c8257b8e2fd616f65 : SUCCESS in 1h 53m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41532da6bee548beadf7944c21bc0e84 : FAILURE in 31m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96da41fd6678422d85390e2859ca7eaf : FAILURE in 27m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2732965e457246798aa6d06e803c4439 : FAILURE in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd3bac31fdcf413caf63c4536ac36d29 : FAILURE in 28m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b72fef80350d4fa29ffa6ffa5dba722e : FAILURE in 29m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65a3078757e7464da49ebb8aa91dda07 : FAILURE in 26m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/317c05ac75c04f228f64283917f6daac : SUCCESS in 9m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb7293d649b348c2b640e98217f9a283 : SUCCESS in 1h 08m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/525fd6ff089b4c37b9ef6ca422448edf : SUCCESS in 1h 10m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/afafe27dbcfd43b6af70da7bb2f1a934 : FAILURE in 1h 38m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c73e42c03904a17ad1f559f71ef96e9 : SUCCESS in 1h 12m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/851d70254eee4650b9b91072329fef9e : SUCCESS in 58m 14s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/64061e5c55e84f60b7a881c11d14b4c5 : SUCCESS in 1h 21m 33s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99d3336c271042339292c9c2b6d4655a : FAILURE in 1h 08m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/02a2a4a8ad28490fb612217da9249983 : SUCCESS in 1h 33m 15s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82cf34e38a724b02acadd6bc78f6225b : SUCCESS in 1h 06m 07s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a86f720dc0a6443b8079787b6e6443af : SUCCESS in 54m 36s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2181d3e06a0040d39f406c0e05fc8716 : SUCCESS in 1h 05m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6dcd3eb73f5340dcba8e438bdae9689c : SUCCESS in 31m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/521f30d4a00342f0af6ba4e650999c31 : FAILURE in 1h 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dfdf924fb9a3abd7b329dc865747e3ab0dc6d6f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-28 18:18:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b837634cc63de7a15e1a7e01c0fe440090ba5208","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-28 18:19:59.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"4098c3d4f903c6e804d714ea09d5e29f0fce04e4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-28 18:20:33.000000000","message":"Topic rbac-for-address-groups removed","accounts_in_message":[],"_revision_number":4},{"id":"3ff25410dba0da3412d2703f02a4b0043b6886dc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-01-28 18:20:43.000000000","message":"Topic set to bp/address-groups-in-sg-rules","accounts_in_message":[],"_revision_number":4},{"id":"fc8200f35790f3f6ff8d5ef34fcdd5f2d4161f40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 19:21:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05fd640ac92b49669af47165fceb5253 : TIMED_OUT in 1h 00m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b57f094b1da5463fbf9c61a6349a8327 : SUCCESS in 32m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ccd0182d2d0a7a5938a3025a60e60f68b642bbca","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-28 20:37:08.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772460/4 : FAILURE in 2h 17m 30s","accounts_in_message":[],"_revision_number":4},{"id":"22511260f1c0ea96be698d6776ce150166e35d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 01:24:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3a7a69e345a643e6ac95477aee6261fa : SUCCESS in 50m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5c0a3df2503d463092a93dc1214dfd75 : SUCCESS in 34m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/85273d29ae214f6fadfdf0ef58449944 : SUCCESS in 1h 28m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c6e91d45051c4417b924d03e17af3359 : SUCCESS in 1h 41m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ed858ac0744f4afcb5a427ae80fc1b9a : SUCCESS in 1h 29m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/38ee1298843444b7abc44b641e50a9bd : SUCCESS in 2h 01m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9407da4547d4b81a175cf92f5468b1a : SUCCESS in 34m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b26fd1be0bb497ba8d135de3c1f999c : SUCCESS in 25m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c939c0639f648578e881213d4b8e895 : SUCCESS in 11m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5bca1c14d674eb9ab5b35852c2487e6 : SUCCESS in 26m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e585c391bce419f807d6beceb6cbaea : SUCCESS in 27m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54c0b1d2c0744962aadc44fce9a3a4f7 : SUCCESS in 32m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6e5117cd1749b3a1a39e04b27b8921 : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9902159c16a43979b5fe9dfbcf391fb : SUCCESS in 1h 08m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/32dfc30064a148d3b09cd5b773ad439b : SUCCESS in 1h 16m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5d7fd9240a0246649abedaaf1111cb3e : FAILURE in 1h 44m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c229dbe21f242c6af18796cf5b45060 : SUCCESS in 1h 02m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c6c1c25696ed48269c67489bae16ac3c : SUCCESS in 59m 01s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f5ffc4362724451690b0acf29135c03b : SUCCESS in 1h 13m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/26cd7494ad7143099abc1d3f2d1cd230 : SUCCESS in 1h 31m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5c74a90d20bd47bd86c689042aa8042b : SUCCESS in 1h 35m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8350528cd1eb4ce8988d5f8a60d75ac2 : SUCCESS in 1h 07m 41s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f0ce4a4418e649b4aed88f2a3fa9c185 : SUCCESS in 50m 43s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f83bd7479ee942988737557ec2e64357 : SUCCESS in 1h 32m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/23219531448b4adaa2cf7f74720567a5 : SUCCESS in 36m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ecf32390b21d41e0aeefa99a34f6580a : FAILURE in 1h 52m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0e053bde99a4f70b3c17aec01dee145df31ea0ec","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-02 02:47:28.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"566e53d0417740da599cdd024e9eb033c2d3f393","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-02 09:51:25.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"f05fe52cf0ceb81b2350de4d36acad4abcfcbf6e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-02 20:56:39.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"15be0c4addf0e9ce6dd4870c873ae11ce913b79a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-03 05:39:11.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f21a3fac2a75064d9ba73c4c3742f17c04b9ba56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-08 00:26:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cf0f130c2dac4a1862b80ba4dcee4dc90727e997","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-08 00:29:38.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"85aeef600d77121d00773b79d9c0f2f96480e270","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 01:05:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/86a48106a2234fd395373d3e7b00ff92 : FAILURE in 31m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c53de05c395c4f6894629a7d560fd9e9 : FAILURE in 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"792ab1f3b69f77f5d42d45321df2a69ed3a09da0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-08 01:56:33.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"fba7b6427942d25feff2b0ed54b7796cce62d17c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 02:19:55.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/496c464cbb4745069336f2e854f03c87 : SUCCESS in 48m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c2f5912cb13c4d669f3b87d297d02fe3 : SUCCESS in 50m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/70ac1c94dda040fa80be6f5f952d4f85 : SUCCESS in 1h 25m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dc75dfc49cd042c5a27b7c79a4a14e4b : SUCCESS in 1h 33m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/765a75d061664d47826d480507899790 : SUCCESS in 1h 40m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c7ed02cb59c9486981e41b745e391503 : SUCCESS in 1h 50m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/945c9e84a65b4890b8a87c4b654991ea : FAILURE in 32m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e78897d1297a41d997601667d8c8f701 : FAILURE in 29m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc6e553a62142fe8cb26a6d3667bf9f : FAILURE in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53a65de8269e4ac18914ccf0c44a53d0 : FAILURE in 25m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79cb320e170d457cabf26212ac075765 : FAILURE in 23m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/609b4753962448aeb8a2a8875fb4cd79 : FAILURE in 24m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e7f91613564ef4b6069ff8cd344d3b : SUCCESS in 8m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/beb7fbb117db4e018868fae53f32abf2 : SUCCESS in 1h 05m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7352ec7aa51543e49863c47a420a0258 : SUCCESS in 1h 21m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8d58a31458654e8c891fc27bc7704947 : SUCCESS in 1h 30m 43s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/169996140b7f4526b1bf9d84da1255a0 : SUCCESS in 1h 08m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c1128b26062e4851ac2949c77019cce1 : SUCCESS in 52m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d7c0362529704dfba5ad5210aa3e5cbe : SUCCESS in 1h 22m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b56de235796c4f62ad5bcd2426e65de7 : FAILURE in 1h 11m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/00ad765bbe104629bfbe9a23fdd9b99e : SUCCESS in 1h 39m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/714766357ab14944b2eec92de179fece : SUCCESS in 1h 02m 51s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0fab62a91c0e4d6c8ecdfd7c21b32cac : SUCCESS in 1h 02m 33s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/748997eb464442f087fbea9e27260e30 : SUCCESS in 1h 09m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a5113a6a7c1d46f282439a585aca0b3b : SUCCESS in 36m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a964882f2f8a45efaf14e5a866ee192f : FAILURE in 1h 38m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bf0715dfa6b6a46fcb8536e6961b02fdc68ca5ff","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-08 02:28:01.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772460/5 : FAILURE in 2h 00m 28s","accounts_in_message":[],"_revision_number":5},{"id":"16b59c08b7a6b31933a165da49e25644782f515b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-10 01:04:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8b5f886f648890d8a0cbac081ed80f652a39a9d9","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-10 01:05:35.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"3bb9b95002f303b061ccee6d70bf5af1be7255e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 01:44:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a162ff92ecf40a1a642663796ffaa7e : SUCCESS in 34m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4eac2ec774764b508c4882a5f6a2de88 : SUCCESS in 35m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"701ea0e53501d2c6ef181a946e35b8f2096944a9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-10 02:32:52.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"50e36dc3ce92d619eb660e21f88b2a05b0b7dece","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-10 03:07:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772460/6 : SUCCESS in 2h 01m 33s","accounts_in_message":[],"_revision_number":6},{"id":"e9d633203ec94693da5e2303502dc5fc3ca88833","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 03:38:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/caf5eaee89be4c868999088e295afa38 : SUCCESS in 53m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c53002c341554d05aa212e37100c6886 : SUCCESS in 40m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b52dbba321a449f5863ac46cd0c41b88 : SUCCESS in 1h 57m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9a91e78bfc5540a1ac1014bcf0d7e727 : SUCCESS in 1h 35m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/97e432ef8636417c98b09af1dc14ca9d : SUCCESS in 1h 30m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dbf433993090416695d23c8198299193 : SUCCESS in 1h 26m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4de8b9515873464f938627c64ae3ee0a : TIMED_OUT in 1h 21m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c4f4ddd7a6542279058b7285e8e3054 : SUCCESS in 26m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e43cdccbe6244657af696a8683943dcb : SUCCESS in 12m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cac58bb23cb546d09e2dc1b9524e4cad : SUCCESS in 31m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ad96825350247d0b2e001b185464c6b : SUCCESS in 26m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e1b0028e8cc4999b4e0fc62c5c8a474 : SUCCESS in 32m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fee575ef7e24f028005eccd8d480ff8 : SUCCESS in 8m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecb5cb0460824c17a94d19165dedda9b : SUCCESS in 1h 06m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d44ee5191384c329e8fb131c6c6f61f : SUCCESS in 1h 08m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/717765d189874424b2675cbda89155f6 : FAILURE in 1h 52m 54s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/714f55102da44f8e99d6ac45daa2701b : SUCCESS in 1h 01m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e4c694f6e67b45f4b1b494120fdc07cc : SUCCESS in 55m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f64aa35c3a084739bbe6c44ea5a8a68b : FAILURE in 50m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/480dc0f3380b4e458ef0e48afb983eab : FAILURE in 1h 15m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87f774a2bcd44c5183e386a4a425d818 : SUCCESS in 1h 40m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45a7e279ad8c49dbb6f1109b8ddfcc9e : FAILURE in 50m 54s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/fac3d4e82c5740b9aff4c373e3421771 : FAILURE in 52m 13s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/807520bc45de4bceaa9a89d2c690a7e0 : FAILURE in 58m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/267ab0bc2a0b4be586460f710495e151 : SUCCESS in 33m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e53c98376fb84a93be7b0038e6d4e09d : FAILURE in 1h 50m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"94a2217fe5fd4cdbbf8dc5215321c9009a403375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-10 23:51:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d85b8c70e6c031d765596babfda467b3e3c09830","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 23:58:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8fa7a0c937ca8f82b77ffe412e616433923df5a7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-11 01:23:47.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"c85b3f69566602164f5fb314290ebc338b1d29e9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-11 01:42:38.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772460/7 : FAILURE in 1h 49m 54s","accounts_in_message":[],"_revision_number":7},{"id":"b4a7256e222c296402526e90214a054ae81bebc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 01:48:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/75aed8b58d0e4b97b2a2c522d1ff713a : SUCCESS in 41m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e2d02b84131a4158819912450d88c5f3 : SUCCESS in 33m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f7f683eca7ef445c9e311daac11a7721 : SUCCESS in 1h 32m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/220b0164e33944028601c279bbb228c2 : SUCCESS in 1h 08m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8c47667af9044f9a9a77249651ae98b : SUCCESS in 1h 02m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c7d65a8ce48040c5809fdff7639e24d4 : SUCCESS in 1h 54m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8103111d9594251b5f2405545a75ea8 : SUCCESS in 33m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dafd6b33f8c4e5794c3aef24218728d : SUCCESS in 26m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4732a84b87db487fb0c43d0f16798815 : SUCCESS in 12m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87744af7c6cd46c2bdd149d2e75c02d4 : SUCCESS in 17m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bb883fdfe2b47f38d14260fbdfde063 : SUCCESS in 19m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fb90cd6434d48ab9bf8e2d0e9487979 : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3ac8fd856b46d98363500615340bfb : SUCCESS in 8m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fdb513cf3fa47d3b4d5a14360fb85b5 : SUCCESS in 1h 14m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/97f720ee005547b7bbe0a45514420082 : SUCCESS in 1h 06m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6451d5e200ae4e39954e213080459700 : FAILURE in 1h 48m 41s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/508c3e304fe3429cba2336f3f5f792cc : SUCCESS in 1h 08m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/53f38532b045468bb33d854a9b3c3897 : SUCCESS in 46m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/51e925d303c94d36a3984b707356b7c1 : SUCCESS in 1h 25m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39de2bde76b64693ac50c512447b06a4 : FAILURE in 1h 13m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/788ea29c22e64c46bfad632d86b152c9 : SUCCESS in 1h 46m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18d5bdd7fed146b48b670673a6b8b4e5 : SUCCESS in 1h 01m 30s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/39910203f2ef4a58ae4b976be402f22f : SUCCESS in 44m 02s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6b63144cfce3431bb5bd4ceb8a82f2ba : SUCCESS in 1h 38m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f54da94d847e4e719b85960179fb910d : SUCCESS in 35m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/62fc39f3014b4e7bb0842f63e0172964 : FAILURE in 1h 45m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6f7424ee223f855bec2ef69799957a0593f8e655","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-15 15:22:34.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"82c93cb6262fd902460651a897d9007971bb3aae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-18 02:04:16.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4c2b94448e10935427017b480fc9bdc851f76484","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-18 02:05:01.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"749adfa9b546ae13ffeaf4828a805c2227a57701","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-18 02:12:57.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"376eca5359bdf65ca14619f5f4bad5db7ce05269","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 02:47:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f51d1e8ac871439aa049aa7efacd5204 : FAILURE in 35m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66a9e5caa379478fbf6ffed9d5702ebe : FAILURE in 35m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"44579383e929a5aee585c789e8d57d804147d544","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-18 03:34:17.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"63b5c865f9e1fb437cdbe58c54236a8084731c4a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 04:02:30.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772460/8 : SUCCESS in 1h 57m 14s","accounts_in_message":[],"_revision_number":8},{"id":"435f555a317d8ee0d9d65c3d724a182e48d8ccd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 04:09:58.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1c1569782b2c4c1dadcbcd8b9b459e83 : SUCCESS in 54m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f80855d8a5e1480182da7e8c911e4580 : SUCCESS in 1h 41m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4dcc5cd0410e46c393056a51e6a761e5 : SUCCESS in 1h 12m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b2c3776a7b2c4c5c95f9bcd597e554e5 : SUCCESS in 1h 37m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/08ad6a3d8ed742c0a40240727a865956 : SUCCESS in 1h 58m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2036299890644edbb3cd97dbec581de : FAILURE in 42m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3ffe3c3c4e44d0185da92e68fdcc3fa : FAILURE in 28m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135c5546cf56426f94896cfc278c9971 : SUCCESS in 11m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b6dd216d3df496db727b8e6561d90ec : FAILURE in 17m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17094b556eaa4da090193cf0780a4333 : FAILURE in 25m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da1ad481bd0a400880af840cc4084e56 : FAILURE in 18m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578bc706da964b1f8346de5a4f47475a : SUCCESS in 8m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2787f32f84d744feb6155df295485e85 : SUCCESS in 44m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3d2bbf2e2c649d387b2bbb0810b090c : SUCCESS in 1h 14m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/48397dd9a15d45259d458ca77d894149 : SUCCESS in 1h 26m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/57ae0266bc3b4994949bfa398ec9b907 : SUCCESS in 59m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/17846a2049344a94b771ab8d6765bd53 : SUCCESS in 47m 20s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/07c1ac983a6246fa8f32f4e0df640326 : SUCCESS in 1h 19m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/69cf2a5e1eda43fb86bb49835e30d1d5 : SUCCESS in 1h 24m 55s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f18c4fd82f94493b3f08c3b9a64e043 : SUCCESS in 1h 37m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2edb8680240e42569df4d10434d88824 : SUCCESS in 1h 00m 48s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/26898aa80f5a47c59bb98398fd98a8c0 : SUCCESS in 1h 04m 25s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/61cbac86e98144b7830ffbe20297eb93 : SUCCESS in 1h 06m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/685bd79ba9da47beba171acbd95989ea : SUCCESS in 31m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/015707e9695247e397de33c73532fc21 : SUCCESS in 1h 28m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a2905bdf57ba4ff3b851475efa1aa32e : SUCCESS in 1h 16m 45s","accounts_in_message":[],"_revision_number":8},{"id":"6a623e36fa4a8044f2f64481dccc3911a0a12913","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-23 00:56:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a6498a6e0b821a8014207c3f13d580995aa11254","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 01:10:06.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba2d47727f464db1a5e444e6e72b4789 : FAILURE in 7m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b7f095fccfe440fab2a810ce692ac5c : FAILURE in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a76d8d9f5c28119aa63bc165e7d3e6a20e793b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 02:09:29.000000000","message":"Patch Set 9: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b7405b4f89854d1289bebd3608629a8f : FAILURE in 24m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/057b447e1fe543e2aaf601c174454282 : FAILURE in 22m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/36eedc37d9d949b28c7bfd94309dd872 : FAILURE in 22m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a5baaf9cd3734c609cd0ace4fee159dc : FAILURE in 23m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d540236cc5d24d3595afacc2cd1559a4 : FAILURE in 22m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f46b83d0ad7b4482bdde619b79714de7 : FAILURE in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/390f07195f5942acbafa829ce1108b8d : FAILURE in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85c1aa4a118a42b8be774c8d154d5c30 : FAILURE in 12m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc1d37d641d04e478ff9f50dfa47e88d : FAILURE in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e633d1b988245bba0bd0e8a9dd923ea : FAILURE in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67b24b6b8d614016ade79a5363c83b63 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c21af721244f9c8a215dcff08980f5 : FAILURE in 5m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8eca7fd7c4ea42c2a2c91caad6fb8e3e : FAILURE in 19m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/15c7bc7b334148f9bfd930e8e4f3ecda : FAILURE in 18m 17s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8d904c3467b44d7280f9ba83783b1765 : FAILURE in 19m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d81e33c1b6c4f49abaa713171714b53 : FAILURE in 53m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/89caf5f8302a447b98fca86a16a57b1e : FAILURE in 52m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ac9225960ac74757821147c3afaed8f7 : FAILURE in 26m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c44ac43e868e4ef19846c7289a3d45d8 : FAILURE in 24m 03s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28e12e3e38464fe7a8f893ae3d56e5d9 : FAILURE in 24m 12s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcdcdac581c24bb4a03c0ca4e487540f : FAILURE in 26m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d8e8ba011aec415bbe0806dfba7e2e61 : FAILURE in 20m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/241bbc52c9624669b7ecf2a0826e0176 : FAILURE in 21m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/08482489efcd48deae114eacdea92659 : FAILURE in 26m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/696ee03a58f64cdb8a3046dfeae08e73 : FAILURE in 27m 07s","accounts_in_message":[],"_revision_number":9},{"id":"d60c04240e171dd7453ba381caa256e1b1ca899b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-23 03:56:37.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772460/9 : FAILURE in 2h 59m 12s","accounts_in_message":[],"_revision_number":9},{"id":"9db4c1e15281ac3a44229a41fdc75f913df2fd8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-23 14:18:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6e96b61405fa04f45da56fc02578db412b655d86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 14:59:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79f7c6a57fc14175af31f7d825e42849 : SUCCESS in 36m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b4fc4175dd149f899e24be6316bbb0b : SUCCESS in 36m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"80e2c7a34836ac59d1aefdec5d069002416863a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 20:36:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c6dd06b448884df5a48dc5d0a77c3573 : SUCCESS in 49m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7f54efb8a8094971be7771551dd3e3ec : SUCCESS in 1h 53m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/478a3cf01df2422481066924209e5460 : SUCCESS in 1h 34m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/870765d771c24d8dbdae24ec2e19761a : SUCCESS in 1h 59m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/522f272634b34bc1834aa2c6d7a51692 : SUCCESS in 2h 07m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/293b16e1235e4041a147682ada7a1743 : SUCCESS in 40m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4589c56b2a544de3a06d79b8d8a31940 : SUCCESS in 34m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8f34342f4f840839a897b45be6e8a10 : SUCCESS in 15m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbc651e6a5e248f683b505270ed697cd : SUCCESS in 38m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5104fb7105074b599522b00b6fbe653e : SUCCESS in 32m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0aa661a9cd54cbbb9a09005ccfea6e4 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a06db51e764b4d128925d699fc495e72 : SUCCESS in 8m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f5013bdf9b7479e815d3058ccd4cb6d : SUCCESS in 1h 02m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/42b04600835b4c3aad3901cc28cb029e : SUCCESS in 1h 19m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/90585d87bff84d798652caefa3a6e7cc : SUCCESS in 1h 51m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/013200b931d84764ba61b0d57ab6dc13 : SUCCESS in 1h 04m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7e5327d8bd984914aa76fa1bd53b45d4 : SUCCESS in 1h 08m 31s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/16bf33e719fe44458e74f5b59ef69589 : SUCCESS in 1h 29m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d3c9cb48212d4d7eb2f067cc6f9babea : FAILURE in 1h 36m 38s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aacf07f12db449c080b2acc30dc6f71b : SUCCESS in 1h 59m 12s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bd12e34bb084f46b2c34d349e1aa3a1 : SUCCESS in 52m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aab294b42ae240b29cb4d77f8186e85d : SUCCESS in 1h 11m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cf29feac85744b30893db6ee6d65d7ca : SUCCESS in 32m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/846cd81b625049528fbe3b67a0dd4988 : SUCCESS in 1h 31m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8af475bfd51a4e0fb11ad94d7977e411 : SUCCESS in 1h 24m 39s","accounts_in_message":[],"_revision_number":10},{"id":"5939a39a810708173d884e2c80d70c65fadf4c81","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-23 23:09:03.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"d710e8d2c26427fa07d8ceec76c9ba8b1b6be90e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 09:14:30.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"eee9bffbe60226d20df1874cbb08ef809e12505a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 17:42:15.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"af94d9f6fa565227778d7c6ae7177eb8fb469be1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-25 09:18:13.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)\n\nMinor nits, looks good","accounts_in_message":[],"_revision_number":10},{"id":"1413ff43386f3fc081555456e420d77639799a29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-26 00:10:28.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4b17e957ed98f1526f9ab5cd10bcce549f1037a8","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-26 00:11:14.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"83cd1a5cfc6df8eb623401093d6eb6d655444c3d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-26 00:11:37.000000000","message":"Patch Set 11:\n\n(4 comments)\n\nThanks for the review. I addressed all your comments","accounts_in_message":[],"_revision_number":11},{"id":"1b3a7e180a6f0139743a18d585dcb1764e945fed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:53:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcd5e5e0aeb04bcdaad30b0ed1e7b118 : SUCCESS in 37m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/557979d3906b492b91d9ef27f873b06d : SUCCESS in 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"59329228280be96c3778b3f28c412cfcf25ee415","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-26 01:46:45.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"5c22587af64e2f84efd44c8f65725b2ceaa3cf4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 02:06:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a92885e61dbd4202813aad2241479b10 : SUCCESS in 41m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bbfae75ba1a54d5d928fdf13b50a78bb : SUCCESS in 1h 40m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/48f6b72cdee94a84b73844d08cdf8b35 : SUCCESS in 1h 13m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b2fb890490648c78448b35794bb3633 : SUCCESS in 1h 27m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0cea5a8626874e84abfeb8efc7649f04 : SUCCESS in 1h 53m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e2448706ecd4465ae31adcd9a1fa3c9 : SUCCESS in 25m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86416f6d42ac4228b5302a84acb77607 : SUCCESS in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb3f90c0b5c467183a227681d45da53 : SUCCESS in 9m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bec3ae4ce8db4338956d877a7d5bd748 : SUCCESS in 19m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4abcff5e9ebf4fb08bfbd7298f5cdf2e : SUCCESS in 18m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d135f12e4014d1c8a0bda0235da2d43 : SUCCESS in 22m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb12d25739034847a92527b29fe9812e : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/597b35b27b1f455dac5b3549541593c4 : FAILURE in 51m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa88d2da59ac4c4c9f0e5d8877d9f929 : SUCCESS in 1h 10m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bb8b490dae8d4b019100596c4be9b512 : SUCCESS in 1h 30m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f322cf031e74c0796d1f2372f0ce5da : SUCCESS in 48m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/87b159bc3f9345d299f4721ebddd4e12 : SUCCESS in 1h 06m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/18d8b0fbe4ed48cf9f26145054d77edb : SUCCESS in 1h 18m 05s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7816b1681d3a43e59e65743d8fabe22d : FAILURE in 1h 12m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b4276f171154ccaabdd3c4839903667 : SUCCESS in 1h 41m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6db653e30a124d4d919148355df04cd8 : SUCCESS in 1h 10m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5ac137413a3d45d8a9019f7cf40cb5ce : SUCCESS in 1h 09m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a133061214e947d0aa8f0733b48eda33 : SUCCESS in 36m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fb92e68404cf43babacfc297439a0803 : SUCCESS in 1h 38m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0df4007bc101460a9508ad6a01f4f9fb : SUCCESS in 1h 32m 43s","accounts_in_message":[],"_revision_number":11},{"id":"24f924d49365991b1e4314b34fe9da3a4308d713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-26 16:45:29.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"502860303f803168d4a6fe513b98c73679de9c4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 17:25:27.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/33a220c1dffb4fcc899108ef2b892e08 : SUCCESS in 33m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11b63a1cae9748269661596bc1c85692 : SUCCESS in 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e667da629446f0e7a42eed2b298dd2ca823799fc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-26 18:18:14.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/60/772460/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"ecd9b168cd5f1e809d887a52bcbd683207ad954e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 19:04:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/60234b1b5d4b415fa7b0e79ec108fdb8 : SUCCESS in 56m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/38aea839f78b459294a7a0e661147b07 : SUCCESS in 1h 16m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3bdcf648136044dcaa1e2e9d13f0c938 : SUCCESS in 1h 37m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c27cf9abb4a644588466103ba6151515 : SUCCESS in 1h 25m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2e6c49f8b3314f1a8ae02a92e88872ce : SUCCESS in 1h 36m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1877210c9e204f658b60c50990b74db2 : SUCCESS in 35m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3099ead56273445b8c7d8879ac616045 : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d3e4d8ed874dffb5d1612a32db827c : SUCCESS in 13m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f47ca5b69233442ab7f054cabdb7e25a : SUCCESS in 30m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24c9845df65e493b90465076ebc93f37 : SUCCESS in 28m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01144019e2af45939eb0d72462e6222a : SUCCESS in 30m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a0fab4d0b347ed84f1a36e58b6be88 : SUCCESS in 13m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbc7dc5215794b6491780a43d203788a : SUCCESS in 51m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e59807b4d8c4ac0a944849bf0ef393b : TIMED_OUT in 2h 17m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f4d45642356d4e728e131d0d408b4073 : SUCCESS in 1h 40m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a3ab20aa39e4f80b543cd9acc89f06d : SUCCESS in 1h 03m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8b77af48e0384e2ca64dbe8c26483916 : SUCCESS in 1h 21m 19s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cb9ddf43b4e84eba80ed046ce59def93 : SUCCESS in 1h 33m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b4576d128df94cf5a6f5204d0126de26 : FAILURE in 1h 15m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/41ce5dd216cd40bb9b4550142b43fa3e : SUCCESS in 1h 35m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/847c865f226943e39e543bcb40d55bf6 : SUCCESS in 54m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/00a1d8e1b33845568a76556c2617847d : SUCCESS in 1h 10m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dae401aeb3834c8ba09e7c998236e393 : SUCCESS in 38m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cc3441c6462c4582ba38380c645db9d3 : SUCCESS in 1h 20m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ded5cd2c356c423d87759c35f82416a5 : SUCCESS in 1h 19m 38s","accounts_in_message":[],"_revision_number":12},{"id":"f5534886b3b4febe64ca41f42a9e516934839482","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-02-26 22:57:08.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"ee777504409945be398b55acf61d116beda387ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:46:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cc4dcff0fbf497b82aeee1800b604dc : SUCCESS in 43m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79eb72b7f4f543ec8edaece28ed4b832 : SUCCESS in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"94f481b132f94e116568b3652f91568b82f1953b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 00:57:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/15d75bdb16a5422899134c6ce4345d0b : SUCCESS in 56m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ccc45a17174d43dbbc21cd2791b0424d : SUCCESS in 1h 31m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1df339bfe149483596a41bc5833a5c73 : SUCCESS in 1h 24m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4b9bc4e822854363adb63db4d7e23c24 : SUCCESS in 1h 38m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e43aced0569040bfa5d2809387cf2551 : SUCCESS in 1h 43m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee6ae9dad03f46f18ea7b55212244bc1 : SUCCESS in 42m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5cb6016456f4d73ade1ff5271aba1ad : SUCCESS in 34m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1685f657b2364f40966131361c552fdc : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f3917cdfd604818a5a7a7a99838276d : SUCCESS in 28m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdcee9f752cf4af08fb6aac59262da14 : SUCCESS in 30m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9557ea0385624b6f95c5e200db0b828e : SUCCESS in 28m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4fbfa86d809432e816624ae83066622 : SUCCESS in 9m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2c6e5ede05a4cb98113f837dc386394 : SUCCESS in 43m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/42b1ba69e92541969d681dd12fcbe018 : SUCCESS in 1h 15m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c80fa65d36ec4ae3adfb815a9baec0bf : SUCCESS in 1h 41m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc1d65a7309b41668a13b136efb8d572 : SUCCESS in 1h 05m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4e0ce2fc609d4dc480c5f7b6eef68deb : SUCCESS in 1h 09m 54s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/912e68288d1447b0b75d1bf72746503b : SUCCESS in 1h 17m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c7b4988c02564c31b04647b683abcc08 : SUCCESS in 1h 48m 42s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08f9e35fe0e34b0db7d1fa97e2b01ad2 : SUCCESS in 1h 53m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/030872bebc654d2083401c65f9441d77 : SUCCESS in 1h 01m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ba4e917ff18e48b1b42dda793d4d3439 : SUCCESS in 1h 08m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d4d4cd8f882943f39a7a7f1858973688 : SUCCESS in 38m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6df4afa6844740308670f11a718002ab : SUCCESS in 1h 31m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3da96d5d402f464d9bd8f7b9fca4b216 : SUCCESS in 1h 27m 19s","accounts_in_message":[],"_revision_number":12},{"id":"07b212d1d83abdc445428655b1038dd0d64b107d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-01 07:59:26.000000000","message":"Patch Set 12: Code-Review-1\n\nPatch LGTM but please also update docs. There is something to update at least in doc/source/admin/config-rbac.rst and maybe also in other documents.","accounts_in_message":[],"_revision_number":12},{"id":"28665dbe938ef3e6038830e36f4c17b41edd5914","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-03-02 00:29:01.000000000","message":"Patch Set 12:\n\n\u003e Patch Set 12: Code-Review-1\n\u003e \n\u003e Patch LGTM but please also update docs. There is something to update at least in doc/source/admin/config-rbac.rst and maybe also in other documents.\n\nDone","accounts_in_message":[],"_revision_number":12},{"id":"645f5a531f1f59299066934d1e42c299b6a1999a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-03-02 00:29:23.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"2ae1c938f8b11be49dfe39df372fdf3de0096caa","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-02 00:30:16.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":13},{"id":"f2c74e1d597957dda276876ee22f2207c4212dc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 01:09:19.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a628e6f034f247179f3d8cd60ee36703 : SUCCESS in 35m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18fed3f4e56d4b8da6bc1e11ae5420a8 : SUCCESS in 35m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"14548d53a19ed11a31ad74d065c61a87f66fa429","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-02 02:02:39.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/60/772460/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"ee824e80bd8b2cf328afb8f9686e1141bdd3f93c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 02:57:28.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a6e762d97f4d451e9988d12619e2088d : SUCCESS in 46m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2bf6838987894f16b3179a0a5559d1ff : FAILURE in 2h 07m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1a5b70d2d6d647b188c692cd0c75a147 : SUCCESS in 1h 23m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/64fe38e70d1b42ddbf66ccfd70f318ea : SUCCESS in 1h 18m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9ce4d09f55924b2f8c0f918bec4c8eaf : SUCCESS in 1h 43m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7431e6c5c1c4621842f4eab4df78642 : SUCCESS in 36m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d259c7197aa14f898f3c479ce922b670 : SUCCESS in 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a16f196f645e420cb3a22fac0835de59 : SUCCESS in 12m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bb7b5fb242b4bc68b10fd638828702a : SUCCESS in 23m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0ef21c07c7f4cc2a897b43efb688fca : SUCCESS in 18m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eea73979f48472eb32fc97afe6504ed : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4197a8ddd8f741d5873813b5678b358a : SUCCESS in 9m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0181e4fdf6234278963ca6096d427e62 : FAILURE in 1h 03m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8275b3ffc0d74cbcb57e545b589e7bd3 : SUCCESS in 1h 10m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/da8f3a8485ec4f55b4fe95307d7fe44f : SUCCESS in 1h 46m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1833719d02d24cadb2a3ec21dcd5a3e7 : SUCCESS in 56m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a5a4325c0a544b0d940bddf25fa16550 : SUCCESS in 58m 58s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/469dd188f0bc41218485be649725a4e5 : SUCCESS in 2h 14m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2c592ee108ce42038042e8a11eef1834 : FAILURE in 1h 09m 22s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/56133cb3894343c8a19969bb657a2204 : SUCCESS in 1h 30m 02s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b458c4f654648138c7e4b36e898e94a : SUCCESS in 1h 30m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c483d2b0241e49fa877738e5401efbee : SUCCESS in 1h 38m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9a3258602763452eaa8c766b777d02e3 : SUCCESS in 1h 09m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a6187ad7cdc24f6ebf9eb8c01cf02fc2 : SUCCESS in 1h 29m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6d58c76a928c4f83aa57a5c6d5f1804a : SUCCESS in 2h 25m 38s","accounts_in_message":[],"_revision_number":13},{"id":"4c9da07693bbf19e36829cc9819772e24e9025c4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 09:28:16.000000000","message":"Patch Set 13: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"625cea646014d46f71b5a3ca73c654fd6928c336","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 09:28:24.000000000","message":"Patch Set 13: Review-Priority+1","accounts_in_message":[],"_revision_number":13},{"id":"879374de7edf06ef3eb844661307310dbb313ca9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:18:38.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd1df5bffd5d440dbba5edf183d2a628 : SUCCESS in 41m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bec105760c0f474888ab91108946f1f3 : SUCCESS in 43m 27s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9d06a2e919ea82686e2b0d7f5a95210122a5b0cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 11:19:43.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b63991e3c34c47bdad9b20cc8464f993 : SUCCESS in 39m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6ef82e8ca1074e86b492d3d7cee92997 : SUCCESS in 1h 36m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/951115efe3aa41bd8e02f2b2bffa9b93 : SUCCESS in 1h 11m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/318afebaba5147959373d628ed4bb06f : SUCCESS in 1h 29m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/08eb9a6c0b1a4044b0cb0d0b7c5b8005 : SUCCESS in 1h 24m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a62f42ebc74a4fd7a3daf2b92f5d25e8 : SUCCESS in 47m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29576a5d9aca4a8ebb33f6694d76832f : SUCCESS in 45m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f99c82bff97451192a2d2d971a1e222 : SUCCESS in 12m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ace4d6a4632f4ee1897e18809b30179a : SUCCESS in 44m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77659ca21b5f41a28a83bb2ec5023fcb : SUCCESS in 24m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bcb2c68e9944fafb34855c2f652a91f : TIMED_OUT in 41m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b011df126547f68e29f94a7374ded1 : SUCCESS in 8m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fd327eb158e41b1b80d99a049f74584 : SUCCESS in 49m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a024a5a6ef046298e614b807948ebb9 : SUCCESS in 1h 08m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2b68c00ae7ee4c0d8dfb171de679b089 : SUCCESS in 1h 32m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/26901f3cc9324d329f729e03173a72a4 : SUCCESS in 59m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9fd1e7520508467dadcd914f2dbc4966 : SUCCESS in 1h 00m 23s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/24a05a35f92a4b9d87a5796740f11532 : SUCCESS in 1h 22m 47s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/59804e8555af44df9ff92a5c047d19d1 : SUCCESS in 1h 36m 07s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7a43f371288488693db082278f81273 : SUCCESS in 1h 42m 23s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/237aa4bd175e4bdbbe9522f81563c38d : SUCCESS in 1h 33m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ceb3d50e6fd4431991495720c1a1a433 : SUCCESS in 1h 12m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5c4c0b0215e945cdac10feff826e96c2 : SUCCESS in 27m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d2d38898ee414b67914086f497d71b40 : SUCCESS in 1h 45m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/02aea441989c4248a6809d6e10efc520 : SUCCESS in 1h 33m 16s","accounts_in_message":[],"_revision_number":13},{"id":"12770a9e5745dff06a589f2944287055def4c938","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-02 14:14:08.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"caf2a1967a273dae481fd0345e266b7fa754d0dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 14:18:15.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"2929085cc0fe97f269dc96ed5408495e4da54628","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:35:53.000000000","message":"Patch Set 13: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1c40a974c14d4ae9b8aa30724aaaf9b0 : SUCCESS in 50m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/aef6a4836c35404191487fb03e7e171c : SUCCESS in 1h 22m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0ce64c55856243908bc6c81a296b0caf : SUCCESS in 1h 35m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a95c7f1a7fea4ed0a8af7dd2684058ac : SUCCESS in 1h 46m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e95ae197d127402990ca71d56fda42b2 : SUCCESS in 1h 26m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b11eabd8ac774ac6ab1aab3922c42ce8 : SUCCESS in 31m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2b0484fc79241be8f1eb4d6a36ea7f0 : SUCCESS in 13m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/931e9796409447bdae023dff67e110e7 : SUCCESS in 30m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03d733df14904e4cb81fb687f7a85aed : SUCCESS in 18m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5165142386fe4b7f883f5e1317dee310 : SUCCESS in 9m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd54bb31ba284848b6f2f53892a35b28 : SUCCESS in 55m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5e9df46799641b58af9ea923121c333 : TIMED_OUT in 2h 14m 54s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/52153531fa6c40e3b55fcd4edee95d0d : SUCCESS in 1h 33m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d09b0d3df04442208387dfa6ec7c1bb5 : SUCCESS in 55m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/95cf7c2d72a5490495c017c9ba0979b1 : SUCCESS in 58m 51s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f88b33715d6c4b9d918d9c6c863272ad : SUCCESS in 1h 42m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b62e5bad92442a385cf96fb8f575e58 : SUCCESS in 1h 06m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a3f24316a79542169600a3911970d06d : SUCCESS in 1h 05m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4ca5e9096ecc46298d7c1a7adb69d879 : SUCCESS in 36m 46s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b342d10c8e454f8785394e389d96543b : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":13},{"id":"3a013ee9de0922b897c8334dea0bfdd04e994f50","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-03-02 17:48:27.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"e56ed26ffa34e1b15c9cd839a051a2470df2c308","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:37:04.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d2a2d976f034f4cbada1728d87e6de0 : SUCCESS in 42m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b648903095e4796b81167e308dd479b : SUCCESS in 43m 10s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"57f2c22d01742cb66cf733be9c57eb2466db5de1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 20:37:10.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ba4a00da957943259e10d296e761a603 : SUCCESS in 47m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9980b4a504f546849832277a04dcd29c : SUCCESS in 1h 44m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b8877f2e5e4842259e7901e1512cfd97 : SUCCESS in 1h 30m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9c74254143c74f479892771a8dad244f : SUCCESS in 2h 01m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/57c798cad06b434dbf59b0ee38d381ea : SUCCESS in 2h 29m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a34d215d7e446cd9ed936b356b85a0b : SUCCESS in 40m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a366b2e6f0d548ec8c1bd9bf78c6c444 : SUCCESS in 35m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f68fbe392b48c8a5db33908bcdffec : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e65f18edc9e34d0fb9e9724dc7cb0d15 : SUCCESS in 30m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b76b07e8701f4b5696c18dd5a6193517 : SUCCESS in 20m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91229d8b1ff54ebeb42c876323458e77 : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79409992cfaa41e69acacb1665cd435f : SUCCESS in 11m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8988a42dfbd34405a1a297073dad8754 : SUCCESS in 49m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fdb123fb9384e23a95dbd6ef61ff791 : TIMED_OUT in 2h 16m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8ebcf00a324849b9a30141d2f72ad573 : SUCCESS in 1h 28m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d1e2dabeaac4ba4a50bc25a16da9460 : SUCCESS in 1h 09m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ada0d7a64f4b4d618ce40716129083f1 : SUCCESS in 1h 05m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/74f51ce3633d4cd7bada4c5a51f8f7aa : SUCCESS in 1h 20m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e4670232a3b48e8980b97b7cc37c44b : SUCCESS in 1h 25m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a5da2566c12f4ad283f95b681a2703d8 : SUCCESS in 1h 31m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/601c9bed50594f4bb437a1b6a1bcd740 : SUCCESS in 1h 02m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/54e06e2a07ad413b943e8d288ce26623 : SUCCESS in 1h 03m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9779969d978740f0af5a2cd33900a8e6 : SUCCESS in 34m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d247447dc9d04b4bb803a68071542ae4 : SUCCESS in 1h 26m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ea6ac8319cc3447eb25e958610c48ba1 : SUCCESS in 1h 30m 33s","accounts_in_message":[],"_revision_number":13},{"id":"45da57f1e2defe0790e9aaf37927f71ec5b26454","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-03-02 22:25:31.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"57eb0eab3511ebe34923c874c7a567bb72cda8f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 23:11:35.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff7c0bbf82e04433822dcb5b0f165737 : SUCCESS in 35m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a44cdd91f5f4d289c68bf7bb9f12d28 : SUCCESS in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"2796469d577204c75338cabb9e0c1445b95bbdd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 00:21:28.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e93119e15d8d4af49c4118ea714626d4 : SUCCESS in 44m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/715720d2c62a4ffa95467a06cf80a938 : SUCCESS in 1h 21m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4b116b932c3240cd87bd010dd7ee1dcd : SUCCESS in 1h 24m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bef128f5826d49149f4627080e317f09 : SUCCESS in 1h 10m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/caa4b4043b434d218c62a0ff43399703 : SUCCESS in 1h 53m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5211b4b72bc24d00b3ccb6bcc28764f2 : SUCCESS in 31m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbf3e681cd2842429a65e6355a8a05a7 : SUCCESS in 31m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f0e46a0cf34c4c9f220a15a380f255 : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c23b706964434d31aed32ab925a02a44 : SUCCESS in 24m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a38d72ccc9d041b898922d607c3a5f1f : SUCCESS in 30m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c4d8b4acfab475ab09a4944badc4042 : SUCCESS in 30m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5746ad8b91b3403b9d0217a5c345c83b : SUCCESS in 9m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c357312bc614f07acdff8426858ee46 : SUCCESS in 44m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3046209706584f7585f604a2ef42d498 : SUCCESS in 1h 19m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/20e4921011c54f7eb74fe144cb37f0ac : SUCCESS in 1h 41m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac9b72fd91c5439cb94de56e2de4394d : SUCCESS in 58m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/64c9953d53cc4e27ac848f815aa25263 : SUCCESS in 1h 02m 32s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ea041b1b44a84c9ca0d6bcdae26bf088 : SUCCESS in 1h 17m 14s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b893621204774b7187bb9c1d79881936 : FAILURE in 1h 13m 23s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c9db9460285d4827bed19d8bdc30f737 : SUCCESS in 1h 29m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cb1f31cfd8544e387435400d8f9ed78 : SUCCESS in 1h 03m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/15155235c48a42b896fc29ed8ddb084a : SUCCESS in 1h 06m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/66c468d11c2b4a48aa79dff5df843537 : SUCCESS in 34m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b436175ad27346a285091b5cb4f1fb85 : SUCCESS in 1h 22m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/22563eb042e14e829ec9b4b94cf19250 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":13},{"id":"27e691d5b5ed52460caeabe907384bdd62ad03c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 00:21:56.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"8868c6b8a903f9ffe402aeb475097f52bb2ef56e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 03:15:37.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5ba6ac0da9f542a1b52a70a359e3fb72 : SUCCESS in 48m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7d2808f2232549818ad780bffa091ad6 : SUCCESS in 1h 50m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/df20f04663fb431d8ebaf77ca9adb002 : SUCCESS in 1h 42m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f5d4d59e55d240efb7c77111da74a658 : SUCCESS in 1h 51m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5fcfbfd66e29418a90d76047770d3ee2 : SUCCESS in 1h 25m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6e92ccfffda4db2b7095ec7f1ce4a4a : SUCCESS in 33m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea6bee0f95d4d55a96e2bd14a5d5d20 : SUCCESS in 12m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/415bbedcaca3428682e4cfe1538d0b46 : SUCCESS in 29m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cce3340455344be0b550135139f82f44 : SUCCESS in 31m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62346dbdb12f4afb8ee9265fc2c7963c : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/22473bce49c542428f27c8c00bdc8343 : SUCCESS in 52m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/daa0dd106f784bea99f8b43ce5221b3c : SUCCESS in 1h 19m 17s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8b06901b49364d42a17471d7db9316b0 : SUCCESS in 1h 01m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1cf10b5e90884b0ab7fc3ee753fa445e : SUCCESS in 50m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8d98b7c27bbe499c883cb84e4a07f4c9 : SUCCESS in 51m 46s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/69484f0f6d224bce912ab4fa800875d0 : SUCCESS in 1h 27m 15s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/decb8a4709744cde8ff8f95a0319d6fa : SUCCESS in 46m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ad99c70d052e4d9eb85966d8be1c83e2 : SUCCESS in 54m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/95bad1e5eab641a39425062d34bb2849 : SUCCESS in 24m 48s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dd143ccc960c492b87d87b0055c61c61 : SUCCESS in 1h 10m 59s","accounts_in_message":[],"_revision_number":13},{"id":"57180f35716f8be3779aeaa9d8675d5c17b0a438","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 03:15:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"45845cf12b75c0071f12171c28e2c5ae49989566","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 03:17:05.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7ded193d02c437399285a7f7256b56b : SUCCESS in 54s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"8094b524f693180fe9f435ef0d1f23b6a44259ae","revisions":{"0c767e1d7e0dafc236add259aba6a29ecad084b5":{"kind":"REWORK","_number":1,"created":"2021-01-26 01:09:05.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/1"}}},"commit":{"parents":[{"commit":"329f3788376d49d37a9079cc4244bd005e269896","subject":"Merge \"Use consistent filter API syntax\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/329f3788376d49d37a9079cc4244bd005e269896"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 01:08:16.000000000","tz":-360},"subject":"Allow sharing of address groups via RBAC mechanism","message":"Allow sharing of address groups via RBAC mechanism\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c767e1d7e0dafc236add259aba6a29ecad084b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c767e1d7e0dafc236add259aba6a29ecad084b5"}]},"branch":"refs/heads/master"},"771d0963c056e82e2efaac4813b4957af3e0d96b":{"kind":"REWORK","_number":2,"created":"2021-01-26 23:54:23.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/2"}}},"commit":{"parents":[{"commit":"329f3788376d49d37a9079cc4244bd005e269896","subject":"Merge \"Use consistent filter API syntax\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/329f3788376d49d37a9079cc4244bd005e269896"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 23:53:59.000000000","tz":-360},"subject":"[DNM] Allow sharing of address groups via RBAC mechanism","message":"[DNM] Allow sharing of address groups via RBAC mechanism\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/771d0963c056e82e2efaac4813b4957af3e0d96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/771d0963c056e82e2efaac4813b4957af3e0d96b"}]},"branch":"refs/heads/master"},"1a4b2dcbf160362079c0e3242b4e80720350d696":{"kind":"REWORK","_number":3,"created":"2021-01-27 22:34:43.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/3"}}},"commit":{"parents":[{"commit":"329f3788376d49d37a9079cc4244bd005e269896","subject":"Merge \"Use consistent filter API syntax\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/329f3788376d49d37a9079cc4244bd005e269896"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-27 22:33:53.000000000","tz":-360},"subject":"[DNM] Allow sharing of address groups via RBAC mechanism","message":"[DNM] Allow sharing of address groups via RBAC mechanism\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a4b2dcbf160362079c0e3242b4e80720350d696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a4b2dcbf160362079c0e3242b4e80720350d696"}]},"branch":"refs/heads/master"},"f087381d4cf1000e968cced5631f2b3bce7944b3":{"kind":"REWORK","_number":4,"created":"2021-01-28 18:18:51.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/4"}}},"commit":{"parents":[{"commit":"329f3788376d49d37a9079cc4244bd005e269896","subject":"Merge \"Use consistent filter API syntax\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/329f3788376d49d37a9079cc4244bd005e269896"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-28 18:18:14.000000000","tz":-360},"subject":"[DNM] Allow sharing of address groups via RBAC mechanism","message":"[DNM] Allow sharing of address groups via RBAC mechanism\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f087381d4cf1000e968cced5631f2b3bce7944b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f087381d4cf1000e968cced5631f2b3bce7944b3"}]},"branch":"refs/heads/master"},"5e6a45eae6b6fd8dbdbfb2694ecf5a1152070934":{"kind":"REWORK","_number":5,"created":"2021-02-08 00:26:49.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/5"}}},"commit":{"parents":[{"commit":"61a72322fe38710db9f88c507f3b75bce9284a8b","subject":"Merge \"Add OVN mechanism driver to the ML2 config document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61a72322fe38710db9f88c507f3b75bce9284a8b"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-02-08 00:26:23.000000000","tz":-360},"subject":"[DNM] Allow sharing of address groups via RBAC mechanism","message":"[DNM] Allow sharing of address groups via RBAC mechanism\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e6a45eae6b6fd8dbdbfb2694ecf5a1152070934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e6a45eae6b6fd8dbdbfb2694ecf5a1152070934"}]},"branch":"refs/heads/master"},"9418b9f71bf34dcfea573b961ba59cd07d5c39de":{"kind":"REWORK","_number":6,"created":"2021-02-10 01:04:36.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/6"}}},"commit":{"parents":[{"commit":"c4ac136c82722271b0df54454a7d58e0e9c5fa0f","subject":"Merge \"Stop metadata proxy gracefully\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4ac136c82722271b0df54454a7d58e0e9c5fa0f"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-02-10 01:04:06.000000000","tz":-360},"subject":"[DNM] Allow sharing of address groups via RBAC mechanism","message":"[DNM] Allow sharing of address groups via RBAC mechanism\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9418b9f71bf34dcfea573b961ba59cd07d5c39de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9418b9f71bf34dcfea573b961ba59cd07d5c39de"}]},"branch":"refs/heads/master"},"05ec5102be721ff020bbac48aea8904d8ed1554f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":7,"created":"2021-02-10 23:51:53.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/7"}}},"commit":{"parents":[{"commit":"a3dc80b509d72c8d1a3ea007cb657a9e217ba66a","subject":"Merge \"[OVN] Fix RowNotFound exception while waiting for metadata networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3dc80b509d72c8d1a3ea007cb657a9e217ba66a"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-02-10 23:48:28.000000000","tz":-360},"subject":"Allow sharing of address groups via RBAC mechanism","message":"Allow sharing of address groups via RBAC mechanism\n\nClient: https://review.opendev.org/c/openstack/python-openstackclient/+/775045\nTempest tests: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/773274\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/774929\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05ec5102be721ff020bbac48aea8904d8ed1554f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05ec5102be721ff020bbac48aea8904d8ed1554f"}]},"branch":"refs/heads/master"},"4f882cfcc6be8ddcd128549ed868feba3c1978a7":{"kind":"REWORK","_number":8,"created":"2021-02-18 02:04:16.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/8"}}},"commit":{"parents":[{"commit":"bd725020d88d42ee8d2459828e2641fcb3760578","subject":"Merge \"[OVN] Re-enables neutron-ovn-tempest-slow job (voting)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd725020d88d42ee8d2459828e2641fcb3760578"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-02-18 02:03:49.000000000","tz":-360},"subject":"Allow sharing of address groups via RBAC mechanism","message":"Allow sharing of address groups via RBAC mechanism\n\nClient: https://review.opendev.org/c/openstack/python-openstackclient/+/775045\nTempest tests: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/773274\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/774929\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f882cfcc6be8ddcd128549ed868feba3c1978a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f882cfcc6be8ddcd128549ed868feba3c1978a7"}]},"branch":"refs/heads/master"},"c46ab34d7054c089946525a98089aca35bf7935d":{"kind":"REWORK","_number":9,"created":"2021-02-23 00:56:22.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/9"}}},"commit":{"parents":[{"commit":"872650e4639cbc50b92f38c0fd4f07e5cec7e873","subject":"Merge \"Don\u0027t update AgentCache when Chassis_Private.chassis \u003d\u003d []\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/872650e4639cbc50b92f38c0fd4f07e5cec7e873"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-02-23 00:56:07.000000000","tz":-360},"subject":"Allow sharing of address groups via RBAC mechanism","message":"Allow sharing of address groups via RBAC mechanism\n\nClient: https://review.opendev.org/c/openstack/python-openstackclient/+/775045\nTempest tests: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/773274\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/774929\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c46ab34d7054c089946525a98089aca35bf7935d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c46ab34d7054c089946525a98089aca35bf7935d"}]},"branch":"refs/heads/master"},"7571925dd9be98d2af9f36af03a42404b0c45350":{"kind":"REWORK","_number":10,"created":"2021-02-23 14:18:12.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/10"}}},"commit":{"parents":[{"commit":"872650e4639cbc50b92f38c0fd4f07e5cec7e873","subject":"Merge \"Don\u0027t update AgentCache when Chassis_Private.chassis \u003d\u003d []\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/872650e4639cbc50b92f38c0fd4f07e5cec7e873"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-02-23 14:15:34.000000000","tz":-360},"subject":"Allow sharing of address groups via RBAC mechanism","message":"Allow sharing of address groups via RBAC mechanism\n\nClient: https://review.opendev.org/c/openstack/python-openstackclient/+/775045\nTempest tests: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/773274\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/774929\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7571925dd9be98d2af9f36af03a42404b0c45350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7571925dd9be98d2af9f36af03a42404b0c45350"}]},"branch":"refs/heads/master"},"0e455812c6e307d87a0908c581c72fe0dffb9782":{"kind":"REWORK","_number":11,"created":"2021-02-26 00:10:28.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/11"}}},"commit":{"parents":[{"commit":"598117a77da6c79f7721f93daab713ffa27093f9","subject":"Merge \"Remove rootwrap execution (5)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/598117a77da6c79f7721f93daab713ffa27093f9"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-02-26 00:10:11.000000000","tz":-360},"subject":"Allow sharing of address groups via RBAC mechanism","message":"Allow sharing of address groups via RBAC mechanism\n\nClient: https://review.opendev.org/c/openstack/python-openstackclient/+/775045\nTempest tests: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/773274\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e455812c6e307d87a0908c581c72fe0dffb9782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e455812c6e307d87a0908c581c72fe0dffb9782"}]},"branch":"refs/heads/master"},"7108be4d7d3166ff821babeb10de89ecf40147ca":{"kind":"REWORK","_number":12,"created":"2021-02-26 16:45:29.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/12"}}},"commit":{"parents":[{"commit":"598117a77da6c79f7721f93daab713ffa27093f9","subject":"Merge \"Remove rootwrap execution (5)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/598117a77da6c79f7721f93daab713ffa27093f9"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-02-26 16:45:08.000000000","tz":-360},"subject":"Allow sharing of address groups via RBAC mechanism","message":"Allow sharing of address groups via RBAC mechanism\n\nClient: https://review.opendev.org/c/openstack/python-openstackclient/+/775045\nTempest tests: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/773274\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7108be4d7d3166ff821babeb10de89ecf40147ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7108be4d7d3166ff821babeb10de89ecf40147ca"}]},"branch":"refs/heads/master"},"8094b524f693180fe9f435ef0d1f23b6a44259ae":{"kind":"REWORK","_number":13,"created":"2021-03-02 00:29:23.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/60/772460/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/772460/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/772460/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/772460/13"}}},"commit":{"parents":[{"commit":"77ee0847f5fe886ae6a1c4a9b9949d3590567e0d","subject":"Merge \"[FT] Reset \"AgentCache\" singleton in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77ee0847f5fe886ae6a1c4a9b9949d3590567e0d"}]}],"author":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-01-26 00:54:33.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2021-03-02 00:28:31.000000000","tz":-360},"subject":"Allow sharing of address groups via RBAC mechanism","message":"Allow sharing of address groups via RBAC mechanism\n\nClient: https://review.opendev.org/c/openstack/python-openstackclient/+/775045\nTempest tests: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/773274\n\nAllow sharing of address groups via RBAC mechanism\n\nChange-Id: I9d9e2bd4add5bb6fa4105352bfda739340932571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8094b524f693180fe9f435ef0d1f23b6a44259ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8094b524f693180fe9f435ef0d1f23b6a44259ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
