)]}'
{"id":"openstack%2Fneutron~772492","triplet_id":"openstack%2Fneutron~master~I63221507713b941c261cdf88781133149da8ab8d","project":"openstack/neutron","branch":"master","topic":"bug/1913269","hashtags":[],"change_id":"I63221507713b941c261cdf88781133149da8ab8d","subject":"Don\u0027t configure dnsmasq entries for \"network\" ports","status":"MERGED","created":"2021-01-26 10:25:13.000000000","updated":"2021-02-23 19:50:14.000000000","submitted":"2021-02-23 19:42:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":37,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"772492-bug/1913269","meta_rev_id":"6cf5c21e184a7512f6e78da1e53fd8cbd41d13e0","_number":772492,"virtual_id_number":772492,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 19:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-02-23 11:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2021-02-18 21:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-02-19 08:01:14.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-02-23 11:59:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 10:42:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-01-26 10:59:10.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-26 11:09:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-26 12:32:21.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-26 13:09:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 13:14:32.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-01-26 17:25:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-01-27 17:49:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-02-02 03:22:24.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 16:00:01.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 11:11:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-02-06 00:29:10.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-18 09:15:26.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-02-18 21:05:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"6fdcaf4742cf2edccded8844469ce5a36789dfaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-26 10:25:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a7c9d14360108cd1ebce3683eed5653912b9011","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 11:09:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb5b721f896c4626b4f22cd513dfbcc9 : SUCCESS in 37m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76da8386b31c4f769f1a3aa055d3b8ae : SUCCESS in 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1150eb5c2a08effdee368430e54669b53607be8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-26 12:32:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772492/1 : SUCCESS in 2h 06m 19s","accounts_in_message":[],"_revision_number":1},{"id":"321ea92c281abdbfccda666ff697e38b1578af2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 13:09:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1c11ec997e0c4fc7b4c5142b76d489b5 : FAILURE in 44m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/731f5dc107ea4632adabdaa222c3d9ba : SUCCESS in 34m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a8b7c86804fe4cd395dec26bbccf1de1 : SUCCESS in 2h 01m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b53fc014ae624a07942c6a5071ab2426 : SUCCESS in 1h 29m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/93a15f1767614d9d8e3ad2e3ce9dd066 : SUCCESS in 1h 36m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/20c02e313706462cafbab2aaf45df4e8 : SUCCESS in 2h 07m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f22392649da4449acebf738e5e3caf7 : SUCCESS in 21m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f45de215b58c4aef899a0903d5efa8af : SUCCESS in 26m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ecdd3536624360a67d3ba3ac96b480 : SUCCESS in 11m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/332c11e88793465585924476717dd98e : SUCCESS in 25m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1caac50e4afd4fea9cb34193c6d8652c : SUCCESS in 25m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13895d9581c643ba92275f3b01b5245f : SUCCESS in 25m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7648c7b08d744f468ed27036c08d1847 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/35d6b646d60a48e3bfe8ec205314ad94 : SUCCESS in 1h 08m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/268f7a913ee244048a0b0f9fe2fc4309 : FAILURE in 56m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/80301e91c3c7404a974797397dc6decf : SUCCESS in 1h 50m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86ad59792aec4cc29cc241843c666171 : SUCCESS in 59m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/25c603b9c872419ab34c884f22d0acae : SUCCESS in 1h 02m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0b1abb8a916748e6a2369f128ba4e158 : SUCCESS in 1h 21m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c1616ee696b44714bb9bda1161dcd69d : SUCCESS in 1h 44m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c9e10e1f07184968a4f47a08c6f26cca : SUCCESS in 1h 35m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d85e500e548c40aa879bdd06d16b8ba0 : SUCCESS in 1h 01m 37s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/82d3457a0fd24320a8183e038a303ebc : SUCCESS in 59m 04s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2ce2c4f215f84c46b285a06826db7ddb : SUCCESS in 1h 13m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bdaf76fb9492408d85a34792edabfdf4 : SUCCESS in 35m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d7b8d03aec5e49f793df37b982447e26 : SUCCESS in 1h 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4209313a8ccdc8351b21b8f4ba1c49eed18b04f9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-26 13:14:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)\n\nNice optimization! I can\u0027t remember where we\u0027d need DHCP service for Neutron owned ports, should we check with stadium projects owners?","accounts_in_message":[],"_revision_number":1},{"id":"de5ef73ea6be3ae175af078e7ee01af29bc23f90","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-26 13:14:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7c557fe3c357e0513b3d92f788e896de649987ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 13:56:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc8f9ea86b634a2eb9de224388929bcb : SUCCESS in 35m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3989f4a3c994fdc877cc4b7323b0f6d : SUCCESS in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a1f711553ecb8e555e87f1792280d8e36bba233","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-26 15:25:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772492/1 : SUCCESS in 2h 09m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2169d319199ccb61fb4910819bb602c98ed9ab82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 16:47:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8c3f64e38a434dc7a00357d81d97d0ee : SUCCESS in 50m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ef2f5314001487ca72a626d066b328b : SUCCESS in 32m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/440cedc789da415f8b0455bcd1bfb658 : SUCCESS in 1h 50m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2d88a9709ed74af6a13087a2a8ad9d1c : SUCCESS in 1h 28m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/feefebe149f14bd395737ebed220c9ad : SUCCESS in 1h 28m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e5eb3ef0e7c04b59956b052bb3ca2507 : SUCCESS in 2h 02m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e31393d0beb548b19de13152042840c6 : SUCCESS in 29m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fded43ae6ce84015937c0276abf04bf1 : SUCCESS in 32m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6334671050b4fa5b477f0bd46bae15d : SUCCESS in 12m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/366e79baca84486fa426b13c3a41f807 : SUCCESS in 29m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cabfe6d74b814dfea4ac89d0390aaf96 : SUCCESS in 24m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea7349e0cbbb4b8e86e694c360f091bc : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0627c48967294c1b8f282cc2541983d4 : SUCCESS in 11m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6fc4200266bc4fe8b8be994e8c9492c1 : SUCCESS in 1h 08m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2654d92d00a34410b9106e8e0edfde42 : FAILURE in 53m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b491f5f938a44bdeb0f1e6bb6e0cd3b9 : SUCCESS in 1h 36m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/527bc76c8de34197b0299e4fd7f72cbb : SUCCESS in 57m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f45ed1b9106d44468eeba8a05322646d : SUCCESS in 1h 14m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0783c0aa125a4e38b40b4a8008014329 : SUCCESS in 1h 28m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a279f875d0b04b4ba4af12e8dbd0b742 : FAILURE in 1h 09m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94d5b03b505b4345a5b6fcd7160c0c60 : SUCCESS in 1h 34m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/128e97add1274e919bbbb7bcca4df865 : SUCCESS in 1h 10m 40s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/28cf8fcb10274af7bafafbb08b35931e : SUCCESS in 1h 06m 59s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/36c5dac4f45d48fabd2ff1764934ab28 : SUCCESS in 1h 20m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5811bbab177346138e02502edfe4b90a : SUCCESS in 34m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8fdd0d82a28f4ad78ae86acdf9270c1c : SUCCESS in 1h 35m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72b1ad4de0b6ee40412ff6462de3af644adf6c64","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-26 17:25:24.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nQuestion - can/should we filter these on the server side?  And reduce message overhead?  Or maybe that will break other backends...","accounts_in_message":[],"_revision_number":1},{"id":"98c28a17c86b8290e543a58bc6dba634942ef2b2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-27 10:05:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b2a955aeca2310b05821147a3df46296dfe68f1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 11:02:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c1879853a024900b9a726184cff26e2 : SUCCESS in 45m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4016901b3f7945dba7f492ad52d91725 : SUCCESS in 51m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c33a67540ae76a8bc567097905c773a9dc9829c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 12:13:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6f518be4e24243868b53314505c112ae : FAILURE in 52m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/17813d3881f94b25b270e6284bb75eb8 : SUCCESS in 56m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/392102095cc3441c90e645519263ba35 : SUCCESS in 1h 36m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4756f30cd2444b15989a1e548f32b4d1 : SUCCESS in 1h 57m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e922d83196df4bb1b4f7d7edce3cc513 : SUCCESS in 1h 42m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4a860f2108c948d4a10733cf6d9fdd43 : SUCCESS in 2h 02m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f584d1b871a4247854126a8555f31c9 : TIMED_OUT in 1h 21m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18bd6c584d0f4847adce3bfadf2e2436 : SUCCESS in 31m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e4349f8743942ae8a55ac9fb1b1f76e : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/535fa37e50e4433e95b6d80b12f53dbe : SUCCESS in 26m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a980d0383f749e7bd35502347993b3a : SUCCESS in 25m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/733fd95cf1ec4846b155e1be33c2725f : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54dd2a25bb6a4f02ae52517b0be3a1f2 : SUCCESS in 8m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d762d73d72344f6dbfc6ca173dadaf1b : SUCCESS in 1h 14m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/366dbf967d4344da96463126950f0242 : FAILURE in 58m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/186c72f7377f4132b3c0043bd4cded56 : FAILURE in 1h 43m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d6cd8be2c5641a5839f4f53acf557ab : SUCCESS in 58m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fa717137c73747949d1f382cce82f584 : SUCCESS in 1h 05m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3323523882c64bb7bcb0ba6461d04cb5 : SUCCESS in 1h 22m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/94234c0f0d344f83a658eb3ae2bf1393 : FAILURE in 1h 21m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d5c2fafac9184806a8e50f357f65fd4f : SUCCESS in 1h 37m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78aa3c9cd83e4493890548abee4e703a : SUCCESS in 1h 06m 19s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/35d74943b9ca4da895e0ddb5dbd5eb7e : SUCCESS in 1h 02m 31s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/202b69f647e24543aebee5d5f83d2edf : SUCCESS in 1h 13m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/690df0ce4fbb484dbe13807502d59b72 : SUCCESS in 32m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ec374b9099fb4ab7b67ca7b976cc02c3 : SUCCESS in 1h 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c143df1acad2e8c3a485d50dfaebafba665e5a0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-27 12:13:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772492/1 : SUCCESS in 2h 07m 47s","accounts_in_message":[],"_revision_number":1},{"id":"564075c907113770fa34de31763b4ca63ed16b3b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-27 13:29:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"69a73be7f999f79d742bb88f80573bbd6e21dcc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 14:06:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/532877adf33d48b892ab5656eecf8e83 : SUCCESS in 33m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c8788202cbe49bca0c10be4dbf586c9 : SUCCESS in 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b503d63e7a75798f75a0215260885ca595cf87d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-27 15:21:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772492/1 : SUCCESS in 1h 51m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2d90f289f154f4c527e85b725f99e9287d3754b3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-27 15:31:30.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"87908a579537658edba64f97c99ee99bcea56879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-27 15:55:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5efb74b9496c17aee5340074bdd8f610d5d8e50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 16:54:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98c1a730a6fa4a16b0b042cad3972376 : SUCCESS in 50m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1cbd33d280a74151ade4d2a3dbc9e946 : SUCCESS in 51m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae5b1d5f63ce68b735f15996ed0c8d7f4eaf3667","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-01-27 17:49:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nI\u0027m thinking this might be worth a release note, maybe there are some people who make use of those entries?","accounts_in_message":[],"_revision_number":2},{"id":"ff5e041764f086cb45d22b001b589e95e9afe1c7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-27 17:56:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772492/2 : SUCCESS in 1h 59m 02s","accounts_in_message":[],"_revision_number":2},{"id":"1ea1f010bcb18168bf5153b774c8cf5926a129d6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-01-27 18:14:32.000000000","message":"Patch Set 2:\n\nActually thinking about this a bit longer, the entries at least for router ports likely should not be removed, because that will cause broken name resolution e.g. in traceroutes and possibly delays due to that.","accounts_in_message":[],"_revision_number":2},{"id":"0d8fdc416fd6a93e8eca212003f811b692d02bdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 19:40:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/91d2ca7668ff4705957a9568ee4e8ddb : FAILURE in 52m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/78e32159e4e4439bb88b17ceedd165cd : SUCCESS in 36m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/87ee246d6ed34e5f8de7bafc8174f2d5 : SUCCESS in 1h 34m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fa1b8a6efb814089979ae898ec1199fa : SUCCESS in 1h 37m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a4790b331d444330aaafdca1735d29a8 : SUCCESS in 1h 27m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2225758aa0c1475c8cee388a5550ba2e : SUCCESS in 1h 37m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2329b72c4edd444d8ebcfb3afd6c8824 : SUCCESS in 35m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d5289e3d09546dda59ccb45b98a1247 : SUCCESS in 26m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/772315d374854d3998a58f594839ce37 : SUCCESS in 14m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a95bb51e3344bb798f1e81c13331ec3 : SUCCESS in 35m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3367b1761df49398d19ccc49f7deeee : SUCCESS in 27m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdeb8fb21f024c37a7179eeba0db701f : SUCCESS in 25m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d595b334e34a59a8a0dbfb4d74d0ca : SUCCESS in 8m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/acb867ed1140474a9698879d86103ba2 : SUCCESS in 1h 10m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e066e0e3d87e487f89001fc278e1c7a8 : FAILURE in 59m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a11704e5bbef4d6bb78d21432f2b6891 : FAILURE in 1h 41m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0423dd1728754b768338ae08f7e52c34 : SUCCESS in 57m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/358e3f0a7394484ba57c881d7db4790a : SUCCESS in 59m 08s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d4288c3eacec424780bf14c541393fbf : SUCCESS in 1h 18m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1b40311787744e04b48ab3aefc6959b8 : SUCCESS in 1h 43m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d325ee03865d427486ade28e7684701a : SUCCESS in 1h 32m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/148d767e1f744dc4b1d5d21ab7e3a9d8 : SUCCESS in 1h 01m 20s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/142f368cc69c4a25b6e8bfc454b60ad1 : SUCCESS in 1h 07m 23s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/86018647a6ec412abb42199ecca677e3 : SUCCESS in 1h 21m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ab14a74818ea49aa9c8f7beb4374717a : SUCCESS in 35m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/469b2507db194c558e56cf7d7d05a335 : FAILURE in 1h 41m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ec0c32d48a55ae130b76dedf6e2764c8e89db26","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-02 03:22:24.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/92/772492/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"01cec9b4713e037a324c238033e273b1cd8c6fb4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-02 16:00:01.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/92/772492/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"4cb14974c3fbfbb65c20c530b5e52dce57484a23","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-03 10:36:57.000000000","message":"Patch Set 2:\n\n(2 comments)\n\n\u003e Patch Set 2:\n\u003e \n\u003e Actually thinking about this a bit longer, the entries at least for router ports likely should not be removed, because that will cause broken name resolution e.g. in traceroutes and possibly delays due to that.\n\nTrue. I limited it to only some device owners, not all \"network:\" owners","accounts_in_message":[],"_revision_number":2},{"id":"a0b8fc67fb44c906a17f3eb2687797b2258a6c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-03 10:37:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f0f00d51553cec297d9bc18d3e4efb416c0eb068","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-03 11:11:18.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"94e1083f2320c8556ba9c2114f36e64a39de9b99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 11:22:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/405d316135444f938c3b6f38c167f4f8 : SUCCESS in 33m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54eb43839dfc402ba7622f8e4f8b4ad9 : SUCCESS in 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f9d87e0952dde0485e02c9fa31e1f6dc18ec523e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-02-03 12:46:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nI think this also needs a release note, as it will change user-facing behavior, sorry for not coming up with that earlier. \n\nAlso not sure whether this is documented anywhere which might need updating?","accounts_in_message":[],"_revision_number":3},{"id":"cd9e5aff30de0dff0ec5d442a2bed270087ab953","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-03 13:21:08.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772492/3 : FAILURE in 2h 42m 46s","accounts_in_message":[],"_revision_number":3},{"id":"08516c202caea20001fd2ee1463c1e016b554499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:16:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c08aa21ed03b4b1096ceb1002a0b4063 : SUCCESS in 39m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c6dfd5296175463a9100f82ead5d3bbb : SUCCESS in 35m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/87c0c286d19848f0ac86223b35932941 : SUCCESS in 1h 17m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d6474baea13c4887a4705ae24dd86392 : SUCCESS in 1h 29m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1d8923d8d63e4443921018ab26d5373d : SUCCESS in 1h 30m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d4d4e962144449548715eb2839ec2185 : SUCCESS in 1h 55m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e327d66642574db9a262be8b1e56bf89 : SUCCESS in 34m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8d5440f7b3c4d1f85017c6ca03c157a : SUCCESS in 28m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2945009253c4661be5d3879d3d369e1 : SUCCESS in 13m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb325923fb4e48358672a311aa67a9c3 : SUCCESS in 26m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8df8710929c4a77b19e1e52f77bcb8f : SUCCESS in 24m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c8d6cbb69924187bb469c65f7773768 : SUCCESS in 26m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b352fb06bf42a6933b21195b006aa6 : SUCCESS in 8m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e49b149c3dfc48c5a41828b9db8762d2 : SUCCESS in 1h 07m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4af4a6bbd474e95bc6a74d7afd48fe4 : FAILURE in 52m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c0123146a1f342dcadbb6dc9265dce06 : FAILURE in 1h 39m 49s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23c513c016124223863c84f0784a864c : SUCCESS in 1h 01m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/58cff38963254df08e102bcbf83d7d57 : SUCCESS in 49m 45s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fee2143e552e4b0f89ec046d0e18258e : SUCCESS in 1h 20m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6f68767a6e9747d2bef4889283970498 : FAILURE in 1h 13m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b8c0964ded049aeb7bae15c8151f2d5 : SUCCESS in 1h 45m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a17bb70f4561423aa0c20fd9020744ec : SUCCESS in 1h 08m 51s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/12fbf210f43e413d98891b129d565559 : SUCCESS in 1h 06m 03s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/81e062991f4945fdb25dbc1c4bc9fa6f : SUCCESS in 1h 07m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/39d8ea9114ad4d809787c5a0f128ffe3 : SUCCESS in 34m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/14545f7b39eb4867ae39eb4b89eeb3db : FAILURE in 1h 54m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b5bb00b56608629546948ff662b4daaf6634cde","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-04 06:57:00.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/92/772492/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f0bbd14a42e375f497de288b99f3ff9a8b132c0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-05 15:03:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0c33c2db94859264af490c10c4efe47a6ebabfb5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-05 15:03:57.000000000","message":"Patch Set 3:\n\n(1 comment)\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e I think this also needs a release note, as it will change user-facing behavior, sorry for not coming up with that earlier. \n\nDone\n\n\u003e \n\u003e Also not sure whether this is documented anywhere which might need updating?\n\nI didn\u0027t found any place like that.","accounts_in_message":[],"_revision_number":3},{"id":"a5c8966fdbf75ed3fd4302c67514732b0046be47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 15:48:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7596f038a524cadb07466bf49519800 : SUCCESS in 38m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b58e5283bcc74fda84a8f46ab0e2e0ff : SUCCESS in 40m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8e471dd9b02dd5b330b9c2ef8bb38a93ee66a296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 17:16:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c41aed135adc4f2aa756ca4e7411e902 : SUCCESS in 1h 13m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b412cd36201a42f99b05c4eab94211b6 : SUCCESS in 33m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/58f578ca0df84a288c714da823853747 : SUCCESS in 1h 33m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/95dc94fbaa5d4b23a26bfa882397d801 : SUCCESS in 1h 22m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2bfa3cf2b02f426faeba76f54682f2ad : SUCCESS in 2h 06m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8dea5813ae8445c492e71c498efc6dff : SUCCESS in 1h 55m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d39d9eb1bbb4159bed84c77cac02b17 : SUCCESS in 39m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cfaef401612470fa215e1d5b48e1814 : SUCCESS in 26m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/338ee40a92b74ff7984c227786797b09 : SUCCESS in 14m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5ca18397a544218a14ee5bec883dc61 : SUCCESS in 29m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bbd60252a1f4bfc8f8825b71ebeb682 : SUCCESS in 29m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de46a49c77d144c8be0d39f81633694b : SUCCESS in 27m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5acf0f700b934aa19b04c595d9882614 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9edb897dd439436893b5cfce2ee7e634 : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c59eac73f1a94d3da65d82efe6aa79e8 : SUCCESS in 1h 07m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f554b46adfd942a48ba5f6613e4f78db : FAILURE in 1h 08m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dadda0cb2add450cbb419f53a7585529 : FAILURE in 2h 00m 48s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac1a7660a4274ebe89f26688e6bee95f : SUCCESS in 1h 02m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e84ac7ea2d1144bfa7fe151c8c84c199 : SUCCESS in 1h 32m 00s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/da9aa2f1023d433bb609833c79dba2ba : SUCCESS in 1h 18m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bae62cac83084f6a9b794ddbe7d14a8c : FAILURE in 1h 14m 41s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa1fb87d98f4429e8a42516493c040a4 : SUCCESS in 1h 35m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc2e4fed81d2422595c3af6119dc1e83 : SUCCESS in 1h 05m 40s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/7d49069ddb8b4efd9012e5a9a989960b : SUCCESS in 1h 09m 05s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/81b618b61b5546c190fc0e8678e54885 : SUCCESS in 1h 14m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9473375175fe44efa7bba3e5eef81aff : SUCCESS in 40m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bf56cd7f71c148dd82a0e6629332618d : FAILURE in 1h 46m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"27f86a1d20d3cf268f45157bac2f5301bbd0290e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-05 17:40:31.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772492/4 : FAILURE in 2h 35m 46s","accounts_in_message":[],"_revision_number":4},{"id":"2659e505d875f2e749b7d724444a7c2c132cd4ed","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-06 00:29:10.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"a3d7de738eb2945a272444595aaf82f4bc5682c2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-06 08:32:03.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/92/772492/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"24608ce9cbb1f3ef689129a6a65cce35ec82be61","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-08 08:48:41.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"819593695dad32f09e29854b0077aee0c2b51f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 08:48:57.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"98ca3902086c4f2d28975cc2cda11efb65496478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 08:49:00.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"f7d974548c29bd467589054c81eb26d59b4e29c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-08 08:52:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e1f01b561553d8ae980542abf89892d21bb756bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 09:31:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fe324b616a44ce591337466ec7efd0e : SUCCESS in 35m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90e44c67e09d44dd9e96b048d71d8901 : SUCCESS in 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7cce934bc638e6c2902a417c8cf0a72b3d4e07da","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-08 10:30:49.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/92/772492/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"a408120cfa2abcb29d0ce720573f99c847f1aa74","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-08 10:45:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772492/5 : SUCCESS in 1h 52m 33s","accounts_in_message":[],"_revision_number":5},{"id":"600d2cf0f76d984c8d78a24f51c8c0718d7c724d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 11:01:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a0beb9b4aae94f67b88fc6113c7d4d8e : SUCCESS in 49m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/79ecc8fa16cf463d950eaefe46e9cefa : SUCCESS in 31m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b0ecbbdd1bc643c3a3c750378fa09444 : SUCCESS in 1h 29m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3c4232ed6cf44204b1076a25219675c9 : SUCCESS in 1h 34m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/16ee98bca5bd4b3aa4eac421288e21a4 : SUCCESS in 1h 14m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/aec94a7dd8ca48a6ba335460d6f71c24 : SUCCESS in 2h 04m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e085ffd976c4db89a5781b94c53fa01 : SUCCESS in 36m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7ae7ed804c8407c85cb771b01740f05 : SUCCESS in 34m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/765cff31f63f49a1bd19b86d60f8705e : SUCCESS in 12m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3bc4c8bde294a578839e395481e6509 : SUCCESS in 25m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/165158f9d49c4162be7f3a7e4a829c04 : SUCCESS in 24m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9a356f4a70249e0be784f520fc70f02 : SUCCESS in 25m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/352ad40c6c604f789614d7ae3f617b89 : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0c1dfdad2e94496bae480d93543086d : SUCCESS in 8m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/90c5037f37fb4aaab62bbfebec36f2bf : SUCCESS in 1h 05m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d02d3e6a1ba74ad7bc9c1c6a6f60bbf1 : FAILURE in 56m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5d0fe47c84de469599665b3c5b3e4b73 : FAILURE in 1h 39m 16s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/94900cfe7d8c43309c5d1562b85574f4 : SUCCESS in 58m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e41070517c454ba5b27bba50376e48a9 : SUCCESS in 58m 29s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9f809db016394874b1d729f9b032b578 : SUCCESS in 1h 15m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3df9543a6d574fb68ce02a09d788b6c7 : SUCCESS in 1h 27m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/632cd5eae8374b2b96298e82f7d8b8d2 : SUCCESS in 1h 31m 23s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecddb3c54fe044e09bb3e442aac224f9 : SUCCESS in 59m 15s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/3f9757052a634d748691dcb4317962e3 : SUCCESS in 59m 13s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c9777b14496a4e6ab8846f2d381e4696 : SUCCESS in 1h 08m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/82ae138063d149a2a08a26554a722b12 : SUCCESS in 33m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b54fae839f0a462e85822cd575cde3bb : FAILURE in 1h 39m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"06c71f57d7713523d1bff1b0d81fbf7157ac7511","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-08 14:12:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ae8a830cbfb37a6f5b1c6f3184239c055343d2ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:09:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1d645567473482ea217c0890eef2f60 : SUCCESS in 41m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d084540602d405abb04f35dbe786b4d : SUCCESS in 39m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c7c4e1253b90e7379e81f9a57c8dd29a477fd0e8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-08 16:19:36.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772492/6 : FAILURE in 2h 05m 01s","accounts_in_message":[],"_revision_number":6},{"id":"2a75bb02a493e2341eb35af33a43157ada3473e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:34:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b9b10eac0af34f29820bbc7e074b5bd0 : SUCCESS in 56m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3175f883e093437ab04c5a3ac3f288db : SUCCESS in 28m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9be0354ce14a45eb97b830d1a39019ab : SUCCESS in 1h 40m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1b16ba6cc60d4d569a4d6525448ef1ff : SUCCESS in 1h 45m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/225b3d3b9f4d4ed0abd026e409c8618f : SUCCESS in 1h 13m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e6a7ff43ce09498bb104fcb8be1f31db : SUCCESS in 2h 06m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c8a6167f02345f99e97b872e7ae958a : SUCCESS in 36m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73887c0fe59349f7aedb1c5eee6d532c : SUCCESS in 37m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102cdc1c4147431db54b8bac9787779b : SUCCESS in 11m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40887abbec2d4ad98b44c4bef90b12ac : SUCCESS in 37m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29742cd343ad4a2c9b1cd570b4ce1d96 : SUCCESS in 23m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1a0c4c9f27948e9933af466706c4f0b : SUCCESS in 36m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb29b8c14364662a1ef3a16b16481c6 : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/196273eb8ed34deca3ae26d05f31fc35 : SUCCESS in 8m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd5273913e0b4fb0a84adb9fa2da8c79 : SUCCESS in 1h 03m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbf368b2ed3f41259d12499d735633e9 : FAILURE in 1h 14m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2a289970c34641f2bef2518a7bee4aaa : SUCCESS in 1h 49m 41s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40534de3ab48429581d53d92dcb443bf : SUCCESS in 1h 38m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/97de8fb30b1c4c2195583d38fe2f89b7 : SUCCESS in 58m 54s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d9fa8201a9d441a583712ee430774a06 : SUCCESS in 1h 38m 14s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0b66dac00aed41749398b1b1c902f3bd : FAILURE in 1h 37m 30s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0e790f8937dd4ff8b5fb65947e42eaad : SUCCESS in 1h 25m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5ed06246df8416a8ce17e36ad1d8553 : SUCCESS in 1h 09m 38s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4bfb946214684cb388d1985ec75551e0 : SUCCESS in 1h 00m 45s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c9c6dfe2a0d847c0ae49675f1219b537 : SUCCESS in 1h 04m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a57659dd9abb4cb18f93bb859d5b6266 : SUCCESS in 33m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dcf970ece7a4476b84459c79c19d717d : FAILURE in 1h 53m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ef8c94643f032f04d81e20015d4a0a869a0b561a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-08 20:09:49.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/92/772492/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"0d55e3fe40e851bd8f167a6a3a90e060f86d1b1a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-09 13:41:22.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ab0170bf84494c17cc24c8abc133dca66261a740","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 14:27:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04b3e112eea74b99821493cfd4235028 : SUCCESS in 39m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f99fcb2e0bab45ea9b24c536e84e772c : SUCCESS in 39m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b50bb752920feec5c0d5a8a024107b7d923c955e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-09 15:27:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772492/6 : SUCCESS in 1h 45m 24s","accounts_in_message":[],"_revision_number":6},{"id":"a5fb19c3915cc82ee725e53d6d10315185c613cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 17:31:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/957ff25b99dc4061a7baad57041735f7 : SUCCESS in 51m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e68d9eb18a34a959c7490aea53df609 : FAILURE in 32m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4cfaff1f99124a4882ddecfff7be0282 : SUCCESS in 1h 44m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/142d223b12ff42fb9c916b940c010e33 : SUCCESS in 1h 45m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20b5e430a2d8472290832060a64179ab : SUCCESS in 1h 31m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3f4cb8d17e764c59ab433932f6943df6 : SUCCESS in 2h 15m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2a2f4a853c949aa9ae80ece6c035d5b : SUCCESS in 22m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e6e2623f06046c4a1cd5bc88b8c8cef : SUCCESS in 17m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8cceb1943324a00a9ef3b43a6a1c02f : SUCCESS in 11m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de76c0f441c744c4bedd6983e105ecb4 : SUCCESS in 20m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aec1c0a7170422b824ab5707a747baa : SUCCESS in 18m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f30239f16b8a4359b716ebbc93e27458 : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255e4d56214c43879037544b6426f229 : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e61c828e87cc411290499319f629fdd9 : SUCCESS in 9m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef49d3991d874492b5b97e6003c7b6af : SUCCESS in 1h 10m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/650c8703b03e432f80ed89664ef14c80 : SUCCESS in 1h 10m 01s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f9e2febe68504ace90c3970ec46f22cf : FAILURE in 1h 39m 53s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/95789e3914e34166a9cab3335763929a : SUCCESS in 58m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8605e40402fe4e80b2bf0bf3f2a317af : SUCCESS in 58m 13s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/67747e7f50244122a39f2a99e76bfa19 : SUCCESS in 1h 19m 57s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/56969f240f0341d8b7a21d6713da85b7 : FAILURE in 1h 11m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87d6f755e48249cfa73cf605067c6b24 : SUCCESS in 1h 38m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c4f5a4a06f242bbb72f927145717987 : SUCCESS in 1h 05m 50s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/623c65b22b2642afa359432b97e6b79b : SUCCESS in 1h 02m 24s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/585017aaf403476cb7c938be6bcacc0b : SUCCESS in 1h 10m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d95d40f1988a4cab83215a4b5b9585dc : SUCCESS in 32m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0d9e9c4295d04ad0bf929f2682da5261 : FAILURE in 1h 40m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"61334e117a7bb63f8285c6b1dbc6d59c1f63cf5c","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-10 06:06:10.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f33c838c13fadb4bef3642bbd208c688554df017","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-10 08:06:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2c3b82a7ad273a7b4ea209ee196300301f3cc1ba","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-10 08:22:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fae6ae62db72aa6e53ed0fe9f1ef7695bf804050","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-10 12:16:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6c6470b38dabff4129fae8dacc4013f7552a44b5","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-10 14:29:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"657d19aed09cdad3f02b7b0c1f116745b0d90cd3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-11 10:54:04.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f663c691486032a3a5d06c3bf504c865e1a69487","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-11 17:49:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6ec0bb5231f460c0e8b31a6e57b4611c6882ed03","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 08:49:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5c8f8488e4b195cde3bfc764f90122a5f55a8a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 08:49:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f6d5fd831f019b1505f49085c5281d52103f9e29","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-18 09:11:52.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"382f7eb8fb85a52079fe91ea6aebb6b1b5ce7435","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-02-18 09:15:26.000000000","message":"Patch Set 7: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c6fc3db8110e75d89d67f8223e005ac00e284d9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 09:56:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/167cae3a4e204589867cd07a798ad18c : SUCCESS in 59m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24aaed34fa0a41b892db94702b78cbec : SUCCESS in 57m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ee7b28abe9e49d4f03b6e4f1cb25a719240bf713","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 10:40:48.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772492/7 : FAILURE in 1h 49m 50s","accounts_in_message":[],"_revision_number":7},{"id":"4817516fe6924f992e9ba24ca450149321755164","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-18 10:43:35.000000000","message":"Patch Set 7: Code-Review+2\n\nAs I see all concerns were discussed, perhaps let\u0027s see Brian\u0027s vote","accounts_in_message":[],"_revision_number":7},{"id":"7652e7eafba6214a133197efa81085f498460aff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 12:18:49.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"23c8e5849f8ab782c27932f776ec2d2a3845ea38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 12:19:00.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7dc35953f85c9e2afbb26e57ce1a39918478a592","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-18 12:55:42.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/92/772492/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"8b92fc5ad8cda3a15c4dc74da044fab525b44f75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 13:35:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d51d40da9614f6eace18e0e0d360495 : SUCCESS in 53m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02085673671f4c0eab01243f1de0e581 : TIMED_OUT in 1h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a3f23c2de8a525a14e1917f5db4208d68c4b6d0d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 14:10:52.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/772492/8 : SUCCESS in 1h 50m 40s","accounts_in_message":[],"_revision_number":8},{"id":"fa0c8aea30c1ee7cd83959ba91e7f9cffc4ebb66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 18:44:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8d019c1345f64495881da4078a2bf17d : SUCCESS in 1h 44m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3a81038f0774421a9f6f837b4ab942ec : SUCCESS in 1h 26m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e636d5d774e49929b5a8e9c9acd45ca : SUCCESS in 1h 18m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/516afb6d46e8470cb6a6ff672fc46a58 : SUCCESS in 2h 08m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14a46e72761e44fd8b03e5cb8bb35bf4 : SUCCESS in 43m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58d5c1038ec54b7284a3269e52c1d2f7 : SUCCESS in 30m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d25a4c22c8504701848d6fec5bdc6ccf : SUCCESS in 14m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/077523f1c6ab48dbb83ff48713421af6 : SUCCESS in 31m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7db7585a489f47818d9c253cde4d6c7a : SUCCESS in 27m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34795e9ad95545a68476ad230634783a : SUCCESS in 31m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348d48c5d0f44d5089adc4417148ba48 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4647768ec7c4e4583c58f474cdf6358 : SUCCESS in 10m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff2daae1e0d64ba09d61bc9496f68a11 : SUCCESS in 49m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/20f307089403400d907afd3141acfd0a : SUCCESS in 1h 10m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b4fc496f8aad4928bbe3d7128565956f : SUCCESS in 1h 34m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e212f62687c4a2581acc635f5ccc312 : SUCCESS in 58m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6f72c8178060412d854ea522098ba586 : SUCCESS in 1h 01m 18s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e0f55b85c8ae4644ac30d3dfa16b4906 : SUCCESS in 1h 30m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7eb5bc0413764b5bb9d7993ce13caca4 : SUCCESS in 1h 38m 03s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b41e12cbe6ba4cd1ab3f1f40420be95d : SUCCESS in 1h 40m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09f6e383c74d429397c30fdda5b98769 : SUCCESS in 58m 22s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c74a9a24b2144a58a169a26864792b5c : SUCCESS in 57m 01s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/77796a4bf7ef492bb21186ec041d044e : SUCCESS in 1h 08m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cc7be609377d4638ad15698cd0d86e0c : SUCCESS in 38m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5083fb3ddfde405ab9d9859093e0abf2 : SUCCESS in 1h 46m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/61b324865d57404e9f3dc6541c60769f : SUCCESS in 1h 27m 10s","accounts_in_message":[],"_revision_number":8},{"id":"d3165234a21f437d7a0a3477a5a75330be6b2fdf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-18 21:05:03.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"18d530a2fa9b4456ee8600a6c97c64330110f6cd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-18 21:24:53.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/92/772492/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"878c85559b7c081008a870dcbdb64cf02f375f6a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-19 08:01:14.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"d1492e7b8d4973598e0923d6b6ae44f219f2f39d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-23 11:59:29.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nAs I see all aspects were discussed of this patch","accounts_in_message":[],"_revision_number":8},{"id":"2c1f77adb212c707520966c78cedc990b3ffdb5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 12:00:10.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"9a602b19689b2d034ba3d1b1942dfa6786ed7e6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:41:57.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0fa05f38c51d4d08ade1a2671d38712d : SUCCESS in 1h 36m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4c6ec1987cf74f9db2230ae5b246c068 : SUCCESS in 1h 40m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f480e83b4e7542f18ba05e3c9cfa1579 : SUCCESS in 1h 43m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac8f61c79fc7436da67c7183ae8900f2 : SUCCESS in 1h 23m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03913ea645d34778a10611be8b5060db : SUCCESS in 30m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cf400149d4e4be5a4731ebdac9447e9 : SUCCESS in 18m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b7f9db2538049318f06d18656a4fd4b : SUCCESS in 29m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd91dd319c844305a796a32f9f41d69e : SUCCESS in 27m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/476ab63d09a74652b5a92731e787762a : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8911fb6771942f584523c88dae47444 : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f9012caf4cd4161bc94d2c38274b589 : SUCCESS in 48m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f141e462829545c8a7bb8c3b186666ba : SUCCESS in 1h 16m 02s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ed16818079e248c7a48a4cda38e69144 : SUCCESS in 1h 31m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b01f5d51655c4dc598eab1377d2352b5 : SUCCESS in 1h 01m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6d8762cee2b844c081dbc4273ce6d401 : SUCCESS in 1h 09m 47s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c38e04b2c6ac4b57b403f6fe345f0f7c : SUCCESS in 1h 39m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2a274a736b34adfaafd0701619a7a93 : SUCCESS in 1h 03m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b25bc65fa9704121912d4879dca77507 : SUCCESS in 1h 08m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/aaee77da932747f3ae29f1d190790ac2 : SUCCESS in 32m 43s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/075ffac586c14d30a226b036cf002770 : SUCCESS in 1h 23m 53s","accounts_in_message":[],"_revision_number":8},{"id":"339eb442eeffaf9166a5d5e2126454b92c869736","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:42:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"6cf5c21e184a7512f6e78da1e53fd8cbd41d13e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:50:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5126bfdbff924401a75ebfb7043c1cd3 : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f679b9accbf4b918c073bd0ba98a034 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e4bbeee2060058d079885339e99ef7e6e366de16","revisions":{"9029c5240dddc1d0d1e062b387ac66c898435723":{"kind":"REWORK","_number":1,"created":"2021-01-26 10:25:13.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/772492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/772492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/772492/1"}}},"commit":{"parents":[{"commit":"2bf558f6ec1dab70a81ba988f78d587433f43539","subject":"Merge \"Improve DHCP agent\u0027s debug messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf558f6ec1dab70a81ba988f78d587433f43539"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-26 10:22:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-26 10:22:39.000000000","tz":60},"subject":"Don\u0027t configure dnsmasq entries for \"network\" ports","message":"Don\u0027t configure dnsmasq entries for \"network\" ports\n\nPorts with device_owner like e.g. floating_ip or ports which belongs to\nthe router don\u0027t need to be configured in the dnsmasq file.\nSo there is no need to reload dnsmasq every time when such port is\nadded/updated to the network.\nThis patch adds skip in such case which should improve load on the\nNeutron DHCP agent.\n\nCloses-Bug: #1913269\nChange-Id: I63221507713b941c261cdf88781133149da8ab8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9029c5240dddc1d0d1e062b387ac66c898435723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9029c5240dddc1d0d1e062b387ac66c898435723"}]},"branch":"refs/heads/master"},"9b23b4f0c88429ab785cdf997a248bbac47f5b77":{"kind":"REWORK","_number":2,"created":"2021-01-27 15:55:19.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/772492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/772492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/772492/2"}}},"commit":{"parents":[{"commit":"2bf558f6ec1dab70a81ba988f78d587433f43539","subject":"Merge \"Improve DHCP agent\u0027s debug messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf558f6ec1dab70a81ba988f78d587433f43539"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-26 10:22:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-27 15:54:46.000000000","tz":60},"subject":"Don\u0027t configure dnsmasq entries for \"network\" ports","message":"Don\u0027t configure dnsmasq entries for \"network\" ports\n\nPorts with device_owner like e.g. floating_ip or ports which belongs to\nthe router don\u0027t need to be configured in the dnsmasq file.\nSo there is no need to reload dnsmasq every time when such port is\nadded/updated to the network.\nThis patch adds skip in such case which should improve load on the\nNeutron DHCP agent.\n\nCloses-Bug: #1913269\nChange-Id: I63221507713b941c261cdf88781133149da8ab8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b23b4f0c88429ab785cdf997a248bbac47f5b77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b23b4f0c88429ab785cdf997a248bbac47f5b77"}]},"branch":"refs/heads/master"},"7b9bcc80846fd7aa045f19f2a9ff943ef75c906f":{"kind":"REWORK","_number":3,"created":"2021-02-03 10:37:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/772492/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/772492/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/772492/3"}}},"commit":{"parents":[{"commit":"2bf558f6ec1dab70a81ba988f78d587433f43539","subject":"Merge \"Improve DHCP agent\u0027s debug messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf558f6ec1dab70a81ba988f78d587433f43539"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-26 10:22:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-03 10:29:58.000000000","tz":60},"subject":"Don\u0027t configure dnsmasq entries for \"network\" ports","message":"Don\u0027t configure dnsmasq entries for \"network\" ports\n\nPorts with device_owner like e.g. floating_ip or ports which belongs to\nthe router don\u0027t need to be configured in the dnsmasq file.\nSo there is no need to reload dnsmasq every time when such port is\nadded/updated to the network.\nThis patch adds skip in such case which should improve load on the\nNeutron DHCP agent.\n\nCloses-Bug: #1913269\nChange-Id: I63221507713b941c261cdf88781133149da8ab8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b9bcc80846fd7aa045f19f2a9ff943ef75c906f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b9bcc80846fd7aa045f19f2a9ff943ef75c906f"}]},"branch":"refs/heads/master"},"3fa84aca7b9a367770db8b55cd13fe8977ebac1c":{"kind":"REWORK","_number":4,"created":"2021-02-05 15:03:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/772492/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/772492/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/772492/4"}}},"commit":{"parents":[{"commit":"2bf558f6ec1dab70a81ba988f78d587433f43539","subject":"Merge \"Improve DHCP agent\u0027s debug messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf558f6ec1dab70a81ba988f78d587433f43539"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-26 10:22:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-05 15:02:45.000000000","tz":60},"subject":"Don\u0027t configure dnsmasq entries for \"network\" ports","message":"Don\u0027t configure dnsmasq entries for \"network\" ports\n\nPorts with device_owner like:\n* floating_ip,\n* DHCP,\n* some types of router ports, like: HA interface and HA replicated\n  interface,\n* distributed ports,\ndon\u0027t need to be configured in the dnsmasq file.\nSo there is no need to reload dnsmasq every time when such port is\nadded/updated to the network.\n\nThis patch adds skip in such case which should improve load on the\nNeutron DHCP agent.\n\nCloses-Bug: #1913269\nChange-Id: I63221507713b941c261cdf88781133149da8ab8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3fa84aca7b9a367770db8b55cd13fe8977ebac1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3fa84aca7b9a367770db8b55cd13fe8977ebac1c"}]},"branch":"refs/heads/master"},"b535c8b4b92eb5006c1038dc25e71a43cb5add5e":{"kind":"REWORK","_number":5,"created":"2021-02-08 08:52:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/772492/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/772492/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/772492/5"}}},"commit":{"parents":[{"commit":"61a72322fe38710db9f88c507f3b75bce9284a8b","subject":"Merge \"Add OVN mechanism driver to the ML2 config document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61a72322fe38710db9f88c507f3b75bce9284a8b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-26 10:22:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-08 08:51:49.000000000","tz":60},"subject":"Don\u0027t configure dnsmasq entries for \"network\" ports","message":"Don\u0027t configure dnsmasq entries for \"network\" ports\n\nPorts with device_owner like:\n* floating_ip,\n* DHCP,\n* some types of router ports, like: HA interface and HA replicated\n  interface,\n* distributed ports,\ndon\u0027t need to be configured in the dnsmasq file.\nSo there is no need to reload dnsmasq every time when such port is\nadded/updated to the network.\n\nThis patch adds skip in such case which should improve load on the\nNeutron DHCP agent.\n\nCloses-Bug: #1913269\nChange-Id: I63221507713b941c261cdf88781133149da8ab8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b535c8b4b92eb5006c1038dc25e71a43cb5add5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b535c8b4b92eb5006c1038dc25e71a43cb5add5e"}]},"branch":"refs/heads/master"},"78d9107dda5bf50eab96fb6fd8e379979d7cdccf":{"kind":"REWORK","_number":6,"created":"2021-02-08 14:12:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/772492/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/772492/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/772492/6"}}},"commit":{"parents":[{"commit":"61a72322fe38710db9f88c507f3b75bce9284a8b","subject":"Merge \"Add OVN mechanism driver to the ML2 config document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61a72322fe38710db9f88c507f3b75bce9284a8b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-26 10:22:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-08 14:12:04.000000000","tz":60},"subject":"Don\u0027t configure dnsmasq entries for \"network\" ports","message":"Don\u0027t configure dnsmasq entries for \"network\" ports\n\nPorts with device_owner like:\n* floating_ip,\n* DHCP,\n* some types of router ports, like: HA interface interface,\n* distributed ports,\ndon\u0027t need to be configured in the dnsmasq file.\nSo there is no need to reload dnsmasq every time when such port is\nadded/updated to the network.\n\nThis patch adds skip in such case which should improve load on the\nNeutron DHCP agent.\n\nCloses-Bug: #1913269\nChange-Id: I63221507713b941c261cdf88781133149da8ab8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78d9107dda5bf50eab96fb6fd8e379979d7cdccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78d9107dda5bf50eab96fb6fd8e379979d7cdccf"}]},"branch":"refs/heads/master"},"ca32981425cc1f1a2b1a4d101bed24e81c596516":{"kind":"REWORK","_number":7,"created":"2021-02-18 08:49:56.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/772492/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/772492/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/772492/7"}}},"commit":{"parents":[{"commit":"61a72322fe38710db9f88c507f3b75bce9284a8b","subject":"Merge \"Add OVN mechanism driver to the ML2 config document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61a72322fe38710db9f88c507f3b75bce9284a8b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-26 10:22:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-18 08:48:05.000000000","tz":60},"subject":"Don\u0027t configure dnsmasq entries for \"network\" ports","message":"Don\u0027t configure dnsmasq entries for \"network\" ports\n\nPorts with device_owner like:\n* floating_ip,\n* DHCP,\n* some types of router ports, like: HA interface interface,\n* distributed ports,\ndon\u0027t need to be configured in the dnsmasq file.\nSo there is no need to reload dnsmasq every time when such port is\nadded/updated to the network.\n\nThis patch adds skip in such case which should improve load on the\nNeutron DHCP agent.\n\nCloses-Bug: #1913269\nChange-Id: I63221507713b941c261cdf88781133149da8ab8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca32981425cc1f1a2b1a4d101bed24e81c596516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca32981425cc1f1a2b1a4d101bed24e81c596516"}]},"branch":"refs/heads/master"},"e4bbeee2060058d079885339e99ef7e6e366de16":{"kind":"REWORK","_number":8,"created":"2021-02-18 12:19:00.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/92/772492/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/772492/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/772492/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/772492/8"}}},"commit":{"parents":[{"commit":"61a72322fe38710db9f88c507f3b75bce9284a8b","subject":"Merge \"Add OVN mechanism driver to the ML2 config document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61a72322fe38710db9f88c507f3b75bce9284a8b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-26 10:22:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-18 12:18:12.000000000","tz":60},"subject":"Don\u0027t configure dnsmasq entries for \"network\" ports","message":"Don\u0027t configure dnsmasq entries for \"network\" ports\n\nPorts with device_owner like:\n* floating_ip,\n* DHCP,\n* some types of router ports, like: HA interface interface,\n* distributed ports,\ndon\u0027t need to be configured in the dnsmasq file.\nSo there is no need to reload dnsmasq every time when such port is\nadded/updated to the network.\n\nThis patch adds skip in such case which should improve load on the\nNeutron DHCP agent.\n\nCloses-Bug: #1913269\nChange-Id: I63221507713b941c261cdf88781133149da8ab8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4bbeee2060058d079885339e99ef7e6e366de16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4bbeee2060058d079885339e99ef7e6e366de16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
