)]}'
{"id":"openstack%2Fneutron~772498","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ib2f3aca08946e584156d092c37e1ea5ed5ca81a6","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1902998-stable/ussuri","hashtags":[],"change_id":"Ib2f3aca08946e584156d092c37e1ea5ed5ca81a6","subject":"Auto-remove floating agent gw ports on net/subnet delete","status":"MERGED","created":"2021-01-26 11:08:19.000000000","updated":"2021-02-02 04:48:50.000000000","submitted":"2021-02-02 04:46:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772498-bug/1902998-stable/ussuri","meta_rev_id":"b8024be5a321b36d391ab7b5f0717cb79730cd83","_number":772498,"virtual_id_number":772498,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 04:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-01-28 11:15:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-02-02 01:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-02-02 01:31:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 11:16:54.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-26 16:41:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 11:15:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-02 01:31:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"f0d3168e818fc50a7e7866c9066af45a4e724082","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-26 11:08:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19027c0919384b5570d5992673c0e27ffae42090","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-26 11:16:54.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772498/1 : FAILURE in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f439fbcc71ff205dd96d72a1c545b76b422f9215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 16:41:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/168a549b238f4246a8a4d4faeab35737 : SUCCESS in 1h 08m 09s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/1d45006c7cc04cd288f5cd34fee4aee4 : SUCCESS in 34m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/4e3eb4ca80834534959e0e1e51218eb2 : SUCCESS in 1h 30m 08s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/e2fb87727c72491883c826dec7288876 : SUCCESS in 1h 49m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/703cc4cb3b99486d9f0cb1aa454d9539 : SUCCESS in 1h 37m 21s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/e480258ea3f245daa16839f557c4858b : SUCCESS in 1h 41m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e08695807064128981346851aad52f6 : SUCCESS in 42m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dedbb6d9a5ca4a56a0e74c104cce9b9d : SUCCESS in 13m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17508c89fa0d4003a579b786ace17b90 : SUCCESS in 31m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36b54f96d2dc4ea894508f432115663f : SUCCESS in 26m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cae67c5f577344a9b2248d33f68ec2d6 : SUCCESS in 17m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32e27c93b3b845dab99c67d07ecdeb7d : SUCCESS in 9m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/80859165a21047ad84ff9e3fec96251d : SUCCESS in 55m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5bdf8ee987724d91a38b2b84f658b692 : SUCCESS in 1h 07m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d03516cd4924eb996480e9b8f40bc36 : SUCCESS in 1h 13m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cdcabad5e2a34da4962cf7f698270fa3 : SUCCESS in 2h 03m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e611ef6aeace45ecbf17eaab944a05d9 : FAILURE in 23m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0f0a4f1fb41c4be7a1cc8b9c680acb01 : FAILURE in 23m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/56be83900bf044a6b08fec287dcc48e5 : SUCCESS in 1h 28m 26s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2919038cb27343e1b37f84fb6a2fc67a : SUCCESS in 1h 13m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5be1872dd5414c0298fe67e420fda2cd : SUCCESS in 1h 27m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0535206917e344909846ce43b838cd4e : SUCCESS in 1h 44m 22s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e5b727c1cd1a43e9b48dee5fee909dba : FAILURE in 1h 35m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/445ffd4e957840d5bd68cf7c424179be : SUCCESS in 1h 15m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ed704eb8d4ef4de88780d9003b180518 : SUCCESS in 1h 16m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d4d6483d35741f485876fbc77fcdab2 : SUCCESS in 2h 02m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f93b03376404739bc5e451b7eeaf156 : SUCCESS in 53m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8da2fef649fb46a78952db803fef67bf : SUCCESS in 1h 18m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c7c915aaca834886946f32427c8ec14d : FAILURE in 1h 00m 56s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f36c79bbcce8440496b191f9abe31ffb : SUCCESS in 1h 17m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/36b50d68ffc0423eba377e75f00a30f5 : FAILURE in 50m 40s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9c8d5e8484e24b1abb44d8129633edc4 : SUCCESS in 4h 02m 05s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4fb0839b944c4c03a71b9b2246f2742a : SUCCESS in 1h 15m 41s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c34db3a61a3547979999c8c6619d474a : SUCCESS in 1h 44m 49s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/332757ba80fc47edb6e76f65f7eff163 : SUCCESS in 2h 09m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/460854e9ebd8490cbe31491a57a323f5 : FAILURE in 1h 49m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42f1a5caa205be5931331abf7541f90388cfd153","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-28 05:00:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5b2705a0c156914df37bd951905c486bf32120b8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-28 05:09:46.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772498/1 : FAILURE in 8m 52s","accounts_in_message":[],"_revision_number":1},{"id":"57eb2ef0e8d5c5c87f3981d251c994ac36d3ec3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 08:33:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/1ffc0e7ee9434615b9b76dfab0a5e548 : SUCCESS in 37m 29s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/2513554b27b84202ba1e8451e5b76700 : SUCCESS in 31m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/16eaf66049214965838056623d5a7f1a : SUCCESS in 1h 07m 23s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/470c4a7e43a24ef78f3547c6b3ae8cd3 : SUCCESS in 1h 36m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/f4e059f70ad94debad5e51234a5562b6 : SUCCESS in 1h 38m 27s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/30f7fd0c281746d894ef0019ec0f13fd : SUCCESS in 1h 38m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b33021be066241a09f4fb70c8e35b308 : SUCCESS in 36m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2870c63e51746c087a27183db445159 : SUCCESS in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ee42acf03154989abb74ba9d3ab8ea5 : SUCCESS in 30m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f6942920a4944b9b132c1a29568f5e6 : SUCCESS in 30m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcd9622e6cbe408aa8989c1221f2b853 : SUCCESS in 27m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f5d7f0a26a4afeace6c9e52e8d00ab : SUCCESS in 7m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/61352a68f80c45eab65931b68afe7e21 : SUCCESS in 51m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/51fdc0565b1c407ab2762a3006a4af24 : SUCCESS in 1h 11m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f5f558606ef4af9abb691c679fdc202 : SUCCESS in 1h 27m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0f624edc804c4f18bf8828d8f5a5e8aa : SUCCESS in 2h 06m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9f5ad508dd74f3a884188f90ac400cf : SUCCESS in 1h 23m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/06467aec567641268d02019bae46e0e8 : SUCCESS in 1h 33m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9f32d116dc824dc7bfbd8c7b380a3405 : SUCCESS in 1h 26m 09s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/cffe58390dd34e3a866f42f648be1375 : SUCCESS in 1h 04m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9063bbb295ca484e97cf36e85ad3a344 : SUCCESS in 1h 06m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c0894ffb5c204145a45a656c14d863ac : SUCCESS in 1h 34m 46s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7fb85257c9b245669c9d32fe9682fa7e : FAILURE in 1h 56m 38s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/94554c5704ef436e95b07aaa5b02db61 : SUCCESS in 1h 09m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5bd7100c938242ef901b009c5f28a46e : SUCCESS in 1h 02m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/83c76f254c524c86aec835b748aa457a : SUCCESS in 1h 58m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a97e54f4428479c9014ec515505e590 : SUCCESS in 57m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c0435b2366244b6099e8b5bd301ecb2d : SUCCESS in 1h 14m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/98aa3d5615224217994652b534af77cf : SUCCESS in 45m 09s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f2f472a76598405198ae3e43d962ef6a : SUCCESS in 1h 28m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d7d0da30d4e4ae9afe379361b116493 : FAILURE in 53m 05s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7faf6d8c9a6548d7aa8d04d5b2dc77b0 : SUCCESS in 3h 28m 17s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/df94a9a289ed4d278a0f3f79cef1b643 : SUCCESS in 1h 11m 37s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d4e5c53b9eee4383bd3041b870a89564 : SUCCESS in 1h 45m 45s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/08ea209699684cb9bd135be960431cfd : SUCCESS in 2h 06m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/78bf6151aa374407b3e426285e009880 : FAILURE in 1h 40m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49879a3550922770caec320dac2e1d5e58c12b43","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-28 11:15:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3322e244e45fea1fd521e81248ed25e5d685f773","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-02 01:31:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"297c90ab63e9c46991d9334b88060c1d4ec885e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 01:31:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00a444b1c4847c51975404b354b907fbcd1318a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 04:46:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/ef4627cec20540beb4d4de714e68a4c8 : SUCCESS in 45m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f02a396a5d34e899d83bb9dde572093 : SUCCESS in 13m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b396a8e2a374f17ac45fce2dcb267ef : SUCCESS in 24m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5597a7df124b4177b2c76218e5f82ece : SUCCESS in 28m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c9773ef6e3400a89d95c327516e9ef : SUCCESS in 10m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ae1ce8e043384788bcd698c0308409d6 : SUCCESS in 52m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0d239bdd83044344925ebd9f191679d7 : SUCCESS in 1h 13m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c32755e7aa0244dead849ee799f5e1b2 : SUCCESS in 1h 14m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8cace60f89f34e18bb1af7850cf34570 : SUCCESS in 1h 11m 48s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6b0923d32a42478f8c8148db1901b07e : SUCCESS in 1h 02m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/11681664c0a94b978f93ab5531de2752 : SUCCESS in 1h 09m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bcec787c0ba64c25abe07ae55ec34cde : SUCCESS in 1h 09m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/807a001a4a7141dda06917a60846b728 : SUCCESS in 2h 03m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a88045c92b3246b7b872b51ee85e18e0 : SUCCESS in 52m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dc71ad9f8abc4c148152128225fdf488 : SUCCESS in 1h 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"968c2c525ebf2a5509ecbf2e16d3107877801938","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 04:46:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b8024be5a321b36d391ab7b5f0717cb79730cd83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 04:48:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440bb1e123e34a1bb148157f1e23381e : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2f312d6fffe0edae12afcd615f22ca1cbe90dbc","revisions":{"a2f312d6fffe0edae12afcd615f22ca1cbe90dbc":{"kind":"REWORK","_number":1,"created":"2021-01-26 11:08:19.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/98/772498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/772498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/772498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/772498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/772498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/772498/1"}}},"commit":{"parents":[{"commit":"df9ef3df03b9c8494e3a7dea210d411ce01c8168","subject":"Merge \"Add WaitForPortCreateEvent in BaseOVSTestCase\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df9ef3df03b9c8494e3a7dea210d411ce01c8168"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2020-11-05 06:41:15.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2021-01-26 11:06:08.000000000","tz":240},"subject":"Auto-remove floating agent gw ports on net/subnet delete","message":"Auto-remove floating agent gw ports on net/subnet delete\n\nfip agent gw ports may be left in DB after router removal due to\nrace condition between l3 agent and server: when server processes\n\"router delete\" - l3 agent is still processing \"router add\" and creates\nfip agent gw port after server already removed the router.\n\nThe patch also adds handling of external network delete event\nto cleanup fip namespaces left on agents due to same race condition.\n\nChange-Id: Ib2f3aca08946e584156d092c37e1ea5ed5ca81a6\nCloses-Bug: #1902998\n(cherry picked from commit b97a8eb488aec0ee9afed0ceeb3d45caeb37ffdb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2f312d6fffe0edae12afcd615f22ca1cbe90dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2f312d6fffe0edae12afcd615f22ca1cbe90dbc"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
