)]}'
{"id":"openstack%2Fneutron~772983","triplet_id":"openstack%2Fneutron~master~Ib02ee77c69e0c014ac5af7ad7fcb42f3bdcd6bae","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-12-27 11:47:01.000000000","reason":"Michal Nasiadka replied on the change"}},"hashtags":[],"change_id":"Ib02ee77c69e0c014ac5af7ad7fcb42f3bdcd6bae","subject":"ovn: set allowed_address_pairs to addresses also","status":"ABANDONED","created":"2021-01-29 09:16:51.000000000","updated":"2021-12-27 14:12:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"217b634a9054dba925fc86ffe85d0ba0516c1620","_number":772983,"virtual_id_number":772983,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2021-12-27 11:47:01.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-29 09:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-29 11:22:45.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-29 12:13:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-29 13:45:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2021-01-30 03:03:58.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-03 07:03:27.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 13:31:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-12-27 11:39:24.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"a271b869f5fc9975355c6ee9d6c942643487100e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-29 09:16:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd246b54d3a21f74e510312a711a2fca3b13d808","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-29 09:18:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b0429a8a08f408bcdd3a9a06105049dc08b2ee8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 09:59:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc650fef005544f1b1e45955f678743d : FAILURE in 35m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/817dd60499f1453d8d9763af16cbda60 : FAILURE in 36m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f318e1b34697a1d289df09225e278fa69d6ef16","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-29 11:22:45.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/772983/2 : FAILURE in 2h 03m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d4e066345b716adaefb40277b7b203de68d16372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 12:13:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6d3aef6ed1b84c48a633bf0340a1e241 : SUCCESS in 49m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8491c980b614b44b3d328474c7e4072 : SUCCESS in 30m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/107b4129ab67469389b56f42aef6fc74 : SUCCESS in 1h 15m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c7523fbe28bc4bf29f60b832a00f6ad3 : SUCCESS in 1h 20m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e07f78e2393c40bea70fb3e8cca06fe5 : SUCCESS in 1h 32m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/db5c1808471a43eba4327b53799465ff : TIMED_OUT in 2h 50m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09749541bd22490f9d8675761eac21c7 : FAILURE in 32m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11c1ef14a8db4401950cf898cf328103 : FAILURE in 26m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adfd22b713ec45dbbcf1be7769883371 : SUCCESS in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd162971d03a486b951497d424c30a49 : FAILURE in 27m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9c73e7452d44119adc3b7752db5af59 : FAILURE in 32m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d0a33299e734794ac1ebc91a712cc6f : FAILURE in 25m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e85d90ca05484395fdf5e1e526d642 : SUCCESS in 8m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/179c1c83daf449828b3d74624422838a : FAILURE in 1h 07m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fb8a2fc20bc1491e921575a032dd7fea : SUCCESS in 1h 08m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f10f5a286fe54dc1bd1ef0c66f7ab007 : FAILURE in 1h 51m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e651558256048649e70ea37584c47df : SUCCESS in 59m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5b3a0d5fd98e431f91f92d385ab49a0e : SUCCESS in 1h 04m 45s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5ede8fbbefd14359a92ba1f4d0138a96 : SUCCESS in 1h 17m 10s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/26de595258f047d4bafd01adaeb1d8e5 : SUCCESS in 1h 33m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad231044887a428285438902b1320cac : SUCCESS in 1h 35m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17b2d752a56946acb770be04ed1e7636 : SUCCESS in 1h 04m 42s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/880990eda31d431b9c5d6e6f2c332518 : SUCCESS in 1h 03m 44s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/af7c7c50f4cf448bb6aa2e6e2aae8aec : FAILURE in 1h 10m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6536bfefed6849c18e3b9aa10306ac38 : FAILURE in 44m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/800f30276b70406da6e1c3bf7c6add62 : FAILURE in 1h 42m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"90d5210601f9b7986bc386b111dce42f6f95a32d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-30 03:03:58.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/83/772983/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/83/772983/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a8aa85572babc62ba098228f83970116930f12a6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-03 07:03:27.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/83/772983/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"27cfd00ee475e160fdae4dda8344099fc1bb2e2d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-03 08:02:33.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/83/772983/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"15c0b7bd36b0ae31947437aea11fdc06d1729247","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:31:43.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2},{"id":"635e2c58c181fa34bb21473d86c82cb1f27130b8","tag":"autogenerated:gerrit:restore","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-27 11:38:32.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"466966920eb8affef1f22906444106b65c00e7bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 11:38:49.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"2164eb5b86079e16209443387ab5764609ceb690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 11:38:50.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"613807626327173c3e6abd7234e793a50ebcf20f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-27 11:39:16.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1f3f0d8174cbb1ec04b69b76ef8634736225fa5d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-27 11:39:24.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"c56a9ac88bc9504aeaaceabd54cb4d6244e6f706","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 11:39:36.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"21a925c83c99d1b795c00434dae7de568e46c310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 11:39:39.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"5635709e1c7dcb566c999e6907be24d589cd26fd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-27 11:39:42.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"63324799493a9d4b14b386d8a6cfa8209c008a8a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-27 11:40:53.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8e22e108a8ecebc91c3476deba79fb2eca35acc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 11:41:12.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"a6144389b44b9d4ced9a74241fafdc10e4c3ef2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 11:41:13.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"34adc2e741340541e521aac61d6289590ef9669c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-27 11:41:18.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"c9aecca81718d52219023791c0cac8b6b56d567e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-27 11:45:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c9063786f38f117ba7299c934c41f1e9653713f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-27 11:45:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"9e216ba2a08d946b801e90ddbaece1bd07f22ce8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-27 11:47:01.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"4984287a52d39f004a612892f87f0354ffef0475","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-27 11:54:55.000000000","message":"Abandoned\n\nseems to not be needed anymore","accounts_in_message":[],"_revision_number":5},{"id":"2eca7f7ede192bbf685e7db57c7bc1c8ddd39ac6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-27 12:12:50.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/83/772983/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"6dbbe039352254a8323b8010dd9f54cef47ca5a6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-27 12:51:21.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/83/772983/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"217b634a9054dba925fc86ffe85d0ba0516c1620","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-27 14:12:22.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/83/772983/5/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/83/772983/5/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9975d20f2a38cff185b8019cc87fb7fa305340db","revisions":{"62fb400f1bce5b8c9c912e98fc53d21900243006":{"kind":"REWORK","_number":1,"created":"2021-01-29 09:16:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/83/772983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/772983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/772983/1"}}},"commit":{"parents":[{"commit":"0fc3a698b3e760323362091a25d2deb3a625f346","subject":"Merge \"Fix mock of the _is_keepalived_use_no_track_supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fc3a698b3e760323362091a25d2deb3a625f346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-29 09:14:08.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-29 09:16:45.000000000","tz":60},"subject":"ovn: set allowed_address_pairs to addresses also","message":"ovn: set allowed_address_pairs to addresses also\n\nCurrently Neutron only sets allowed_address_pairs only to port_security field\nin OVN NB DB, but only adds new macs to addresses field - this breaks\nallowed_address_pairs functionality.\n\nCloses-Bug: #1908382\nChange-Id: Ib02ee77c69e0c014ac5af7ad7fcb42f3bdcd6bae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62fb400f1bce5b8c9c912e98fc53d21900243006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62fb400f1bce5b8c9c912e98fc53d21900243006"}]},"branch":"refs/heads/master"},"4a579f29ad9e67b7b715396ff78bf9d5872f8355":{"kind":"REWORK","_number":2,"created":"2021-01-29 09:18:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/83/772983/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/772983/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/772983/2"}}},"commit":{"parents":[{"commit":"0fc3a698b3e760323362091a25d2deb3a625f346","subject":"Merge \"Fix mock of the _is_keepalived_use_no_track_supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fc3a698b3e760323362091a25d2deb3a625f346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-29 09:14:08.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-29 09:18:05.000000000","tz":60},"subject":"ovn: set allowed_address_pairs to addresses also","message":"ovn: set allowed_address_pairs to addresses also\n\nCurrently Neutron only sets allowed_address_pairs only to port_security field\nin OVN NB DB, but only adds new macs to addresses field - this breaks\nallowed_address_pairs functionality.\n\nCloses-Bug: #1908382\nChange-Id: Ib02ee77c69e0c014ac5af7ad7fcb42f3bdcd6bae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a579f29ad9e67b7b715396ff78bf9d5872f8355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a579f29ad9e67b7b715396ff78bf9d5872f8355"}]},"branch":"refs/heads/master"},"2bf998cf752098f5a80c93a731680058b8ac9b10":{"kind":"REWORK","_number":3,"created":"2021-12-27 11:39:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/83/772983/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/772983/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/772983/3"}}},"commit":{"parents":[{"commit":"0fc3a698b3e760323362091a25d2deb3a625f346","subject":"Merge \"Fix mock of the _is_keepalived_use_no_track_supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fc3a698b3e760323362091a25d2deb3a625f346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-29 09:14:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-27 11:39:14.000000000","tz":0},"subject":"ovn: set allowed_address_pairs to addresses also","message":"ovn: set allowed_address_pairs to addresses also\n\nCurrently Neutron only sets allowed_address_pairs only to port_security field\nin OVN NB DB, but only adds new macs to addresses field - this breaks\nallowed_address_pairs functionality.\n\nCloses-Bug: #1908382\nChange-Id: Ib02ee77c69e0c014ac5af7ad7fcb42f3bdcd6bae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf998cf752098f5a80c93a731680058b8ac9b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf998cf752098f5a80c93a731680058b8ac9b10"}]},"branch":"refs/heads/master"},"7a20b1286ddfad269eab6bdf0e1d17390aacc8e5":{"kind":"REWORK","_number":4,"created":"2021-12-27 11:40:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/83/772983/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/772983/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/772983/4"}}},"commit":{"parents":[{"commit":"0fc3a698b3e760323362091a25d2deb3a625f346","subject":"Merge \"Fix mock of the _is_keepalived_use_no_track_supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fc3a698b3e760323362091a25d2deb3a625f346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-29 09:14:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-27 11:40:47.000000000","tz":0},"subject":"ovn: set allowed_address_pairs to addresses also","message":"ovn: set allowed_address_pairs to addresses also\n\nCurrently Neutron only sets allowed_address_pairs only to port_security field\nin OVN NB DB, but only adds new macs to addresses field - this breaks\nallowed_address_pairs functionality.\n\nCloses-Bug: #1908382\nChange-Id: Ib02ee77c69e0c014ac5af7ad7fcb42f3bdcd6bae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a20b1286ddfad269eab6bdf0e1d17390aacc8e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a20b1286ddfad269eab6bdf0e1d17390aacc8e5"}]},"branch":"refs/heads/master"},"9975d20f2a38cff185b8019cc87fb7fa305340db":{"kind":"REWORK","_number":5,"created":"2021-12-27 11:45:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/83/772983/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/772983/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/772983/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/772983/5"}}},"commit":{"parents":[{"commit":"ca90d8c041151efce19dea10758fa46a7d32cc39","subject":"Merge \"Allocate IPs in bulk requests in separate transactions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca90d8c041151efce19dea10758fa46a7d32cc39"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-29 09:14:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-27 11:45:17.000000000","tz":60},"subject":"ovn: set allowed_address_pairs to addresses also","message":"ovn: set allowed_address_pairs to addresses also\n\nCurrently Neutron only sets allowed_address_pairs only to port_security field\nin OVN NB DB, but only adds new macs to addresses field - this breaks\nallowed_address_pairs functionality.\n\nCloses-Bug: #1908382\nChange-Id: Ib02ee77c69e0c014ac5af7ad7fcb42f3bdcd6bae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9975d20f2a38cff185b8019cc87fb7fa305340db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9975d20f2a38cff185b8019cc87fb7fa305340db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
