)]}'
{"id":"openstack%2Fneutron~773017","triplet_id":"openstack%2Fneutron~master~Iecf1b9ce1472ae73dee9642e949e3fd14d69ed68","project":"openstack/neutron","branch":"master","topic":"bug/1460177","hashtags":[],"change_id":"Iecf1b9ce1472ae73dee9642e949e3fd14d69ed68","subject":"Fix typo in \"DhcpAgent.enable_isolated_metadata_proxy\"","status":"MERGED","created":"2021-01-29 11:52:12.000000000","updated":"2021-02-03 08:07:40.000000000","submitted":"2021-02-02 08:07:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773017-bug/1460177","meta_rev_id":"824035c041e1b63b94aeffac94e63c8a82a371c7","_number":773017,"virtual_id_number":773017,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 08:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"recommended":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"all":[{"value":1,"date":"2021-01-29 14:02:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":2,"date":"2021-01-29 14:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-01-29 13:54:03.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-02-02 05:40:55.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-29 12:35:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-29 13:48:38.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-29 13:54:03.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-01-29 14:02:56.000000000","updated_by":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"reviewer":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"state":"REVIEWER"},{"updated":"2021-01-29 14:06:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-29 14:21:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-30 05:25:56.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-03 08:07:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"520321d60c7b562e8103d786d1b2bb48dc84682f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-29 11:52:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4299307014960eee7178af72a9c04ccfd4f776ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 12:35:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5418cae6f45944f5a3c9b7ab940629a3 : SUCCESS in 33m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5eeb438f602452c9a0841156649f347 : SUCCESS in 37m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98637041c86c5f6b48736e47ee9c00534285810c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-29 13:48:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/773017/1 : FAILURE in 1h 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b19b7d6400069ecb42f3961cc96a764c40d84989","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-29 13:54:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a712136c8c050cfb998034d73172572ac33a0b2f","author":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"date":"2021-01-29 14:02:56.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good.","accounts_in_message":[],"_revision_number":1},{"id":"254e7694959e0d2af2e60b0efdd7ac1c4df53955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 14:06:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7a9edda055724d0caf8ef6a6aa8b07a5 : SUCCESS in 1h 00m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/600d657de4ec4a30a8cef0f581e96e5a : SUCCESS in 32m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/03bfcc0992f84dbb9fee9cab41907ac5 : SUCCESS in 1h 57m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/25dda1ff7d854dc7a359410ebcec270b : SUCCESS in 1h 20m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1c8d54d1ce214c4dba294169477bcb32 : SUCCESS in 1h 06m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/188279ed76b14a13aec6b67d57d9c314 : SUCCESS in 1h 27m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d67cc06cd5734ec8ad266c399efe0cad : SUCCESS in 22m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8531bac5c55643068c44d7ab3c971fed : SUCCESS in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2616b9dbd1c433cb9d61815fe1f5302 : SUCCESS in 13m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac7426253f754360ab2b7d78ab586691 : SUCCESS in 21m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0991a39f053843e8b76cb55f51b00fa2 : SUCCESS in 16m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25bef61b0fbc448a9d9b85ea1fd1b91c : SUCCESS in 21m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474fe4ccba09449ab2be206b592b5054 : SUCCESS in 8m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/557ed6d198ac4dee8e8e044046b089b6 : SUCCESS in 1h 35m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a4bc38bc44b4f6e88b9b58571efff96 : FAILURE in 2h 12m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6c71c05a1c614d448a800f87a699f07a : SUCCESS in 1h 16m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/43d6089cf40947ba939f24e49e31b751 : SUCCESS in 59m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1b70893c14014b5e96add5153e8cae84 : SUCCESS in 1h 03m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/16e32eb16d4b49d4ae46d5d3d73441ef : SUCCESS in 1h 35m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ef80f5a230024c12833983845aeb181f : FAILURE in 1h 12m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/97bf04e669e74c34af42e36d9af2464e : SUCCESS in 1h 56m 41s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10351ddb00854ccb8c257c7a79e83fe6 : SUCCESS in 1h 02m 43s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b8c3d158985f4939b5ea94140a368499 : SUCCESS in 1h 20m 32s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f2829c4953d644c588f0fbffe31c5385 : FAILURE in 9m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/38cd4bb8156c46efb52c390b2fe77b8a : SUCCESS in 33m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/856f1f8b7ea54291906c5f2a47a94838 : FAILURE in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6dab8ebe76fe47b8e2d9bdf39958f5135bddd78f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-29 14:21:50.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck mirror issue","accounts_in_message":[],"_revision_number":1},{"id":"ed642c289201b73a3a709303d1d7b77008b83d36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 15:18:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e4bf11e95034a4a8f428dbd8aa81c8a : SUCCESS in 42m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8845621a1ce944c8a5f50227f4ed7709 : SUCCESS in 47m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab1841c470af89a460bc373c326d64394964e6e1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-29 16:10:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773017/1 : SUCCESS in 1h 47m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5184656c95ce0de4dc1058743db1658643d74f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 16:36:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/02c2c086fb6b41b496f1b38aa9477147 : SUCCESS in 50m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9be4bf69cf6342f6902e54d75a066fbd : SUCCESS in 35m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a720698996764eefaa847263fffde4b2 : SUCCESS in 1h 36m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/906def1080e54a27bab0fa366ccb8805 : SUCCESS in 1h 23m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f9f79e185a8a4178993b33115e16acab : SUCCESS in 1h 20m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/544f2688d6fb47a1b324348d8ff96707 : SUCCESS in 2h 08m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13d0290a88a040ff803c9f27fab8a870 : SUCCESS in 45m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f0926f8a3c3478582c6e4fb7167ea15 : SUCCESS in 38m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c433f97afcb44b8bd88e8aaa5460885 : SUCCESS in 12m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83a5c9f1f74f42a0b0fdb631cd8865d9 : SUCCESS in 37m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/679831c1254e4dc8ac2f5ccb5261bf5a : SUCCESS in 24m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f8a9ee4bc44456c9877a536229590f7 : SUCCESS in 28m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c98b65a36ca4e43a981ce5a40a95fb8 : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc988d113e024275b338608c7b310676 : SUCCESS in 1h 10m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/517f9abde59e48318e6b6c68130e695a : SUCCESS in 1h 14m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f9d07c127e5e45c4890c0feeca7f4c74 : FAILURE in 1h 45m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b79116566e1a4c5cb6ff2b547db382a3 : SUCCESS in 59m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5832eb4776a44fcfa3c05189a9db3b38 : SUCCESS in 1h 01m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2a011f97cd6b4d79a63729d658ec0d0b : SUCCESS in 1h 19m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ef5010ced0cb477eae60c3533e0fac9b : FAILURE in 1h 21m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/98adb707ac904cbbbe9369e38d18cc81 : SUCCESS in 1h 52m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1aac25cf863497cbcd21aa414e764b5 : SUCCESS in 1h 11m 13s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/db0c4839ebb141adb7b3bf65151a0b88 : SUCCESS in 1h 01m 19s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/61b96d0a12f445fea6f2ef796f480332 : SUCCESS in 1h 15m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/20e3d0a7e88649a695778dc4f7404d8b : SUCCESS in 36m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1db26bcd019a4cfbbc32d098bfba7471 : FAILURE in 1h 39m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26ae412edbe48e7c9281f7b9aba784f3c683d489","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-29 18:06:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1948b40bc835e4f3f1e426dbac483afe5ce8ade1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 18:52:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2ad326c57f24ae0b0fd2f858a59ef20 : SUCCESS in 38m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8de982c10c8c43ae9e043597c28d9f65 : SUCCESS in 38m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61c1f34f3e0005af62cf2c7ea5b65e29fc1edbab","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-29 19:54:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773017/1 : SUCCESS in 1h 47m 24s","accounts_in_message":[],"_revision_number":1},{"id":"26924e6813017bacfb48fbb26f8882a2f9e653be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 21:10:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/efbbd0ce2688408b869f028bd198787d : SUCCESS in 53m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/431f9f022c2e46569af11c9e6acf1ada : SUCCESS in 35m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b1c120bfcd704c7aabecab3f9ef047cc : SUCCESS in 2h 03m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8054a59375a744338028e68456a224d5 : SUCCESS in 1h 22m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f3c312bca4cf433d90e3365da09ab999 : SUCCESS in 1h 39m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/63300a2be32e4e1f905b5f896d6d4f5a : FAILURE in 2h 01m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e95672cf7f83458584c03849d925fcef : TIMED_OUT in 1h 20m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2147abbad5994aa8a74d6047fb32f180 : SUCCESS in 27m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c1300904dee475283627f0369d4cf8e : SUCCESS in 12m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d441e753a754f57a3305f64799c0f75 : SUCCESS in 28m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/108e385856924ad1b79b3af438d908d2 : SUCCESS in 25m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be9bbedb15984aa38d21233f48f0e39a : SUCCESS in 31m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c08673f07b34cc6a00eba78a0777c19 : SUCCESS in 8m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/67ef80e1eb9e426dbac6f75deb6d9ea8 : SUCCESS in 1h 07m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7f87c99f25d420ba3fef3a926c2e929 : SUCCESS in 1h 12m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8c0bc3c5f2554fbeb0e8d40d926f9581 : FAILURE in 1h 44m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f13ebbfb0ffc4a6997ed925a4bab37e2 : SUCCESS in 55m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/86b780e316124b7ea05f1ea707f1551a : SUCCESS in 1h 04m 21s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/34fdf81954d441498559f307e3baf19b : SUCCESS in 1h 14m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eee471a07776409e9d4e6be1cdd63d65 : FAILURE in 1h 13m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/56112e6055944a2084c57ae11bae712b : SUCCESS in 1h 46m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6efb146fb61a4a4eb3fd2771b0460de4 : SUCCESS in 1h 06m 34s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ec8404e3f56c43889577e5269efc9f9e : SUCCESS in 1h 01m 20s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a0890cb1dc9c4fb88037cdeda9f0f2cb : SUCCESS in 1h 06m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e540f9c6f75542699ee497867ed59416 : SUCCESS in 35m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a0dc28030f7846728aebc1ab03bfb986 : FAILURE in 1h 38m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30847a1e6cc275c7a5431423d318a7aec71cb163","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-30 05:25:56.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/773017/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/773017/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"5dede50f125066f4f3228de1a201d2f91d47e1b1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-01 17:01:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9f4c63e73d2e383e225d3d00d2df8f22004308ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 17:48:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27f4137f6f05455885eabcbfc874534e : SUCCESS in 37m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9125c6268c2d4ee28e1c3afee15be9fc : SUCCESS in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a41f985f01b1b6627c9f475fbe711c6f2e176ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-01 18:54:36.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/773017/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/773017/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"db8ac446195160689e3ff926c64d1116ad1d0718","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-01 19:15:18.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/773017/1 : FAILURE in 2h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a0359d1a48e1df76857c4bba4bd0bb3ab6f2fdd8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-01 20:24:32.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/773017/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/773017/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"7bb508410060864a4b798882a19c741127a4f69a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 22:44:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/363a1e8f84554080b905dd8465dcecc4 : SUCCESS in 59m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f922cc663e1c434bb47e57ef7f7f432a : SUCCESS in 28m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2569003eb6334bbbaa39fb6907a95ec6 : SUCCESS in 2h 13m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1124482a958847689348917b6ec6d6e7 : SUCCESS in 1h 23m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b73a88f7bed34130be40f49ed296f4c5 : SUCCESS in 1h 30m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8ded7ff11eab40608a10bfe66db60672 : SUCCESS in 1h 49m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/943ee0d1f5b240009fc249cdb693198b : SUCCESS in 47m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2099bbd1c0f4e478e4cc676c8b5babd : SUCCESS in 33m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f87e33db7de423eaaa7065e3d4f57b8 : SUCCESS in 11m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9daaade606249448d93870dada560ce : SUCCESS in 27m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50a6b7a9b186433d8873ba88c83cce1e : SUCCESS in 27m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93d34a74fb9d4913953d11a3fde7a130 : SUCCESS in 28m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8634692a471d449e9c8cdf7c6f60c560 : SUCCESS in 8m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/75f2dddbd13e4bdcadf18f09411ca790 : SUCCESS in 1h 10m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1e8b7383cd540c884e82b4bdbb95a40 : SUCCESS in 1h 09m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/19525dd1f0a9476bba6969476f755c35 : FAILURE in 1h 38m 01s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2918d4bf0d94162bb107ab0d8f70ebb : SUCCESS in 1h 03m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fcf90874086b45fe8b59bef2eae7075c : SUCCESS in 1h 07m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/37a0fd14f2ac4becb7607d09da93aa6e : SUCCESS in 1h 37m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1fae4020376d47f7aea2fc58af638498 : SUCCESS in 1h 35m 22s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e5948c9cd5004e358cd26ff4a3c1e136 : SUCCESS in 1h 41m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/351b2004ab2f4c2b869777a786018be8 : SUCCESS in 1h 05m 48s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d68ee1c4996a419a9544a43a1882a400 : SUCCESS in 1h 04m 27s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8e901c128205440387c42588effee935 : SUCCESS in 1h 05m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1f3a950e77724d0e80edc966fe8b7ece : SUCCESS in 37m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/951d1ec328624ea29df1fab7fb8d2115 : FAILURE in 1h 52m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1c2ab00dfe7d0fa3db952e31ebb857e35b0f0c8","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-02 05:40:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2d15d13335f6b8314b89f7c3e254719b917ba77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 05:41:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"738ae211d26acbd19fa90f9e75f8aae0de74dc23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 08:07:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/46d939bf48a34bee86652480215d0533 : SUCCESS in 57m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6637ea92b0184d91a08cb5a21979f85b : SUCCESS in 1h 28m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dd3bcb8898dd4845b85a7e522c3291fa : SUCCESS in 1h 48m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dd29a5bfe80f47b4ad4cf82cac61f107 : SUCCESS in 2h 20m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9006eda77ec42c1ade987ba0dfd93e7 : SUCCESS in 1h 26m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f8618189e0241c8a461a4b0319aa3cf : SUCCESS in 25m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a2244684bd645f6970358802ec032dc : SUCCESS in 13m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0440b22fb31548eca61bf61c618edc7d : SUCCESS in 26m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9674356b3c3c40faa333fa145a97ded2 : SUCCESS in 27m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e64edfb2e704c7e991271d8a5fb8b62 : SUCCESS in 7m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/718f511974de411d8f3f6a729283412d : SUCCESS in 1h 10m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c35f072c0e6e4cfab9b7c30ee18cb227 : SUCCESS in 1h 12m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9da00db7d12c4159b191f3d196de1cca : SUCCESS in 1h 26m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d85b41c4c2c455e90bbbb41d89dde19 : SUCCESS in 55m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/389a9da1ce0d4aa193b188b15f7b5bef : SUCCESS in 1h 04m 39s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8d470859b6db48818eb53ecdcae58c8a : SUCCESS in 1h 40m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66c39d5c69a3434ebf210bc934f3f4c6 : SUCCESS in 1h 15m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b422cbb1c08949148c0d9b32c5635537 : SUCCESS in 1h 11m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/931ae5aadee1407eb42c62eacc2a1c59 : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2050b831aa374bc56eeb69a7287cb93db8426d25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 08:07:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"70d4c49046cf33966733a9ff3f47a22c6b437d22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 08:10:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507a4cf90cd541f48b9a21fc027246b5 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1},{"id":"824035c041e1b63b94aeffac94e63c8a82a371c7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-03 08:07:40.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/17/773017/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bfb8ef01990fc8a5ee0a32c6207e61b4eb11d526","revisions":{"bfb8ef01990fc8a5ee0a32c6207e61b4eb11d526":{"kind":"REWORK","_number":1,"created":"2021-01-29 11:52:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/17/773017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/773017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/773017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/773017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/773017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/773017/1"}}},"commit":{"parents":[{"commit":"0fb63f729751d7e62ac107d3275d440632936978","subject":"Merge \"[ovn] Metadata agent: fix checking datapath type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fb63f729751d7e62ac107d3275d440632936978"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-29 11:51:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-29 11:51:14.000000000","tz":0},"subject":"Fix typo in \"DhcpAgent.enable_isolated_metadata_proxy\"","message":"Fix typo in \"DhcpAgent.enable_isolated_metadata_proxy\"\n\nChange-Id: Iecf1b9ce1472ae73dee9642e949e3fd14d69ed68\nRelated-Bug: #1460177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bfb8ef01990fc8a5ee0a32c6207e61b4eb11d526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bfb8ef01990fc8a5ee0a32c6207e61b4eb11d526"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
