)]}'
{"id":"openstack%2Fneutron~773282","triplet_id":"openstack%2Fneutron~master~I8b0466263388085bb1949ea37f4788579c46093a","project":"openstack/neutron","branch":"master","topic":"bp/distributed-dhcp-for-ml2-ovs","hashtags":[],"change_id":"I8b0466263388085bb1949ea37f4788579c46093a","subject":"Add the DHCPReponder for IPv4","status":"MERGED","created":"2021-02-01 03:37:46.000000000","updated":"2021-04-21 18:10:45.000000000","submitted":"2021-04-21 18:02:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":49,"unresolved_comment_count":13,"has_review_started":true,"submission_id":"773282-bp/distributed-dhcp-for-ml2-ovs","meta_rev_id":"fe0e85e64dec3d5b79dadf657bf6ce53eaf24756","_number":773282,"virtual_id_number":773282,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-03-30 13:51:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-21 18:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-16 07:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-04-08 14:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-21 18:10:45.000000000","post_submit":true,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-16 07:17:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 04:28:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-01 05:30:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-01 05:41:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-01 18:12:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-02-03 08:40:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-02-03 10:07:17.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-03 23:57:33.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 02:01:09.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-05 07:55:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-11 08:14:17.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-11 09:10:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"81571b556714104b7d8a2c13e3297d76ed88ae0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-01 03:37:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40965b26a121cb55ff7c4b4738f65083888d9a97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 04:28:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e38a44b7172a438582a3433dab3d80d2 : SUCCESS in 40m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24e157f4e3a44dc8a7efd1cbe0249cbc : SUCCESS in 43m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2beda001a9dae6221332277a54e6ebe1791eb1fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 05:30:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dc0fb242b6af466baee7ae19fc363a63 : SUCCESS in 50m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/66d8285347d7452983118f769486e494 : SUCCESS in 32m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f7309ed5a8084038b7abfeb14da7777a : SUCCESS in 1h 31m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9a88fbb35fd54f868a15387c0f353e17 : SUCCESS in 1h 34m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00ae02a65355420e944ea8adf88a39a2 : SUCCESS in 1h 19m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5511e35b7d4242bdaafe5e49cc937ab2 : SUCCESS in 1h 47m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e02a37878b1c44dca9a1e3d0eda2c041 : SUCCESS in 33m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6999210ad0442c6beabdc4b10df02b9 : SUCCESS in 26m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d0890c5e149489aa760dff298eeb8e6 : SUCCESS in 12m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03a9e5088d9e4a1e80b64ae6a2234492 : SUCCESS in 27m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94aff2c8d2c64567b71aa4fae2a9fa60 : SUCCESS in 26m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2bad52658ba4d26a8c1840dc715b1ca : SUCCESS in 26m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb65e43f3dd46678181eaa97ac9c03e : SUCCESS in 8m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/24ff8ab5ba1547b4a666b41d51a10038 : SUCCESS in 1h 07m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/70c465745ce941f6b0c0f54f3760ce5c : SUCCESS in 1h 07m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c2885e56fcad46c3869d328cff888ba1 : FAILURE in 1h 39m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/62575a10b69c4a9593853ff41176d087 : SUCCESS in 58m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/671ab85a5d7c4075abaaa2a9173c7c8e : SUCCESS in 1h 00m 23s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fbe56c5d3479465d9f94160f04bad220 : SUCCESS in 1h 14m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c54554bd58c248278c0310cc60866bc9 : SUCCESS in 1h 26m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4c63710763d94099bf65241e3e22998f : SUCCESS in 1h 35m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/660769a813164f40b2486f2d5fd7dcbd : SUCCESS in 1h 03m 36s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/626bc9d3e2bc49e9bca5f1a38850ad4b : SUCCESS in 1h 00m 08s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a809f2a73a0049148d07e2c4f41caf7d : SUCCESS in 1h 07m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7b9f907217e1429497f921dd1f1455c9 : SUCCESS in 34m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c1d3b936a94440a5ab50c9f59dafbb71 : FAILURE in 1h 38m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e0fd7b1c5587a85e4420ce4274a839d5adf62ba","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-01 05:41:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773282/1 : SUCCESS in 2h 01m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c003bc98bdc58c542a36237e224b718e7b2d4028","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-01 18:12:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"236c26ea6c789cbd38a020c53b07eda0d202d801","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-03 04:22:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"888e0d9d034e5fa00d3ebd565fd39ef4b1cfc78c","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-03 04:23:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"960d5b56d51668982ea842eecaad262040981562","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 05:21:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4da14fa0b344376a605f341b00bdeae : SUCCESS in 45m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c15da12006b427a8062308ce2029c6d : SUCCESS in 47m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"801ff33e1f79064404c233eb9c2404032dd0d1ea","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-03 06:16:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773282/2 : SUCCESS in 1h 51m 17s","accounts_in_message":[],"_revision_number":2},{"id":"1e648695eb33db81a51ebece10e6bf78eddf019a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 06:46:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ca248ec546764b0e810384035d97870e : SUCCESS in 42m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b6a75b3b5d034e5baf028ed47dae9936 : SUCCESS in 32m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cd1fe39ee0b341a5a6a3f2a9ece424bb : SUCCESS in 1h 35m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7433f027ea8a4534a43271b38ac6a6a8 : SUCCESS in 1h 39m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eb0b29500b1a4387bafdf23f6074634a : SUCCESS in 1h 11m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/027595fc1bdc4f6dbcb3865b36aeb8e4 : SUCCESS in 2h 21m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9eb9a406d5c4fe996943fc582598462 : SUCCESS in 40m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fa58e2e704c439fa3e4f744c7d3908b : SUCCESS in 25m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b98d66bd0345fd86d28aa5197554f0 : SUCCESS in 11m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d77857831e6448d1876d7557a3a8ac58 : SUCCESS in 27m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0373f77b114447e93f9e3a833bc2196 : SUCCESS in 24m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b447d7bd1bd04a8682fac6e6089f6fa8 : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/834f32fd040643c198d874c656f5e211 : SUCCESS in 8m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b6611adb6fe4da1acb2f21087262759 : SUCCESS in 1h 04m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/425899e8355347f7aabe5db93a18c88c : SUCCESS in 1h 07m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/40fda2d6e79a4995a22208309d3cc694 : FAILURE in 1h 33m 59s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb16d7ec88f04a1fb6727ea6aa2dbc1f : SUCCESS in 52m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f7700a95b66042ceadcd80dc671764a6 : SUCCESS in 56m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ad027179b4aa4047bddbc23be2a4af13 : SUCCESS in 1h 13m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5b50f71046941e58e837a6d5e1204e5 : FAILURE in 1h 14m 08s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29632f1702a94fb6b83ba733632771fa : SUCCESS in 1h 37m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e64b4c5c8691474e98da17deeda04805 : SUCCESS in 52m 47s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/307b9cdd4ad845a7a4bc69e0a816870f : SUCCESS in 1h 12m 21s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4d5447a24403469e933c9a906ad1d54d : SUCCESS in 53m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ca26ec2d50f54f17bec47519f6206b1b : SUCCESS in 30m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0486bba506f8434e9239cbad82d2deb4 : FAILURE in 1h 43m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4d9f6d05ec528fba06016fdd78bd15a6689f5b7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-03 10:07:17.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/82/773282/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7db905a9ddd638f2b4101860d5f1e128c93c14ed","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-03 23:57:33.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/82/773282/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"ec33ba7a86418916efa51c6cca0a9f7a88959c07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-05 05:50:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"fa6129948f8f76cbef54320e9beaa67e38c8d3ee","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-05 06:06:29.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/773282/3 : FAILURE in 13m 44s","accounts_in_message":[],"_revision_number":3},{"id":"af3264c28cc5b1d5040e904d6f559d885a0f8f85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 07:01:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc562ca36bd74ff5a1feb712ea96d0f4 : TIMED_OUT in 1h 01m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4f5b7f4c756490ea84c3378a6cca03d : TIMED_OUT in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3ea6351eebd3c8035cd664668d413d540bd6b9bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 07:50:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6d13a8f08774427980369b72e33d558c : SUCCESS in 51m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e637ad00241346bb9424dd1986b5bb62 : SUCCESS in 36m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e184835cfd3c453b824738c9d17bff7b : SUCCESS in 1h 33m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b3fcb800f71e4144ba5959ec4e1bdbfd : SUCCESS in 1h 38m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fb48c7e7218f4e28a8cc01250a86945f : SUCCESS in 1h 41m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/68a6d09fb2d64b0eb16d2cb0178acbbc : SUCCESS in 1h 45m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/070e8d19cf9f45b485f1b3ab66bc4739 : SUCCESS in 42m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/125c60af103f4ff788d305154f3d4d2f : SUCCESS in 36m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b02693e5c6b4a1097fd82855fcf07c9 : SUCCESS in 13m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d82a8e39b9df4847a8c31d37d02c6cd9 : SUCCESS in 26m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6cfa7c71bbf446cad5cbe81e2b4f8ca : SUCCESS in 24m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec0c16939cad4812bc9f65f1e7f27884 : SUCCESS in 27m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6236bd61b254ebfbdafa0bcb480c555 : SUCCESS in 8m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/92eea7568e964fad86dbd1e3af1da8e9 : SUCCESS in 1h 15m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e3eed16fcb9439b817085d5241aa769 : SUCCESS in 1h 10m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b531391b638741059e340808dc6eb69f : FAILURE in 1h 44m 44s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/64f09114805a4a33b1ec38e7de87f219 : SUCCESS in 1h 12m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7833cbdcf209472fadb0fe8b80e6b626 : SUCCESS in 1h 11m 43s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/72c917d81c954a5e8c737eb16e6fd056 : SUCCESS in 1h 22m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/59b593284cfc43ac98f4bedecb2ee5c4 : FAILURE in 1h 17m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ffdcffb242644b458c456f306560d565 : SUCCESS in 1h 44m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0029649d1bc14df4afe6b97e3e8258d1 : SUCCESS in 1h 00m 12s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0308e6fae8094e3f93f2a3b285744d56 : SUCCESS in 1h 06m 00s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aba84f1140d346189baa2e3d537aa4d6 : SUCCESS in 1h 17m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7f1d52330a3540b48e08e0e3991defe3 : SUCCESS in 34m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c9072a400d1e49d0b8fcfdd13f8b7e77 : FAILURE in 1h 53m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b01b3c801f1e8f874b9a7c4dccb0a02d8518948c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-06 02:50:58.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/82/773282/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f08cdbe6a615b93683e0eec661c8b4e6d18d55e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 01:54:59.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"325bb4048d0a38ee85f2ab2efeb75d6296ac409a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 02:01:09.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"432edbf25a6c846175eb895eb197384cf45a6f27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 03:07:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7dd318b287748c78fd52f655ada6995 : SUCCESS in 59m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b71861fce46e43878fbe01a888fb8ddc : SUCCESS in 59m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c7243609c3cca1369b7b9772f2695027c5c32bc7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 03:51:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773282/4 : SUCCESS in 1h 53m 33s","accounts_in_message":[],"_revision_number":4},{"id":"25b33d3ee33d39ed3166f817943dc05fb64856d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 04:11:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/100b3eca06eb49bcbc175fdb0d5c8ca3 : SUCCESS in 1h 12m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cfaa5ccf8e574b85949d3d18904ea9b7 : SUCCESS in 1h 30m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/48cc4106bd1f459bba3650a961d837fb : SUCCESS in 1h 01m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1947a6f2f7004db09f5fd249d16c4518 : SUCCESS in 1h 37m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bb95222946a94eceaf270b347f07d2dc : SUCCESS in 1h 42m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/174026d88f3645ad8b7edb40900f848e : SUCCESS in 41m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dcb145179164da389bf1a8b04b40fb6 : SUCCESS in 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dcccbeee91d4b7d95c9e40425fe8692 : SUCCESS in 13m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07b59427ac534a7bae536c614f8ff467 : SUCCESS in 21m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f74bb5f627d4494c99394e299e44e42d : SUCCESS in 17m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58b6837d120745ec944efeb4a1381985 : SUCCESS in 30m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c52548b4c13402282d3a84d05060d0e : SUCCESS in 12m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc7fde70b61343e49b8d630d429ca505 : SUCCESS in 48m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d683a094b674de3a41704397d07f273 : TIMED_OUT in 2h 14m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7a1a8ebd680f4abcb3329746303d25c9 : SUCCESS in 1h 33m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1860d5fe23ac4ea98e10d669dbf08b5a : SUCCESS in 1h 12m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e1e28767236340b982d58169cd386aea : SUCCESS in 1h 05m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9ba26d2f689a4724a6aec7017b84ee57 : SUCCESS in 1h 27m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a54fe9ba300b407bbaae427980673b5f : SUCCESS in 1h 52m 07s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e6a8670abfd49d297c23554648ce1e7 : SUCCESS in 1h 39m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/557395f7fb4141c4904ef6786f6be291 : SUCCESS in 1h 00m 57s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/7ee481d12bf04680a9fe55d298cd02e5 : SUCCESS in 1h 32m 12s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8aad763a969b4cd2a2d23ca04728f612 : SUCCESS in 1h 22m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7dddc58aac6a4cb3bd95cd5b19348881 : SUCCESS in 34m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/20738ca84f8645928d809e5cb9e5d222 : SUCCESS in 1h 31m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8fcccc10a9bc476e80e32f7950b6723e : SUCCESS in 1h 28m 56s","accounts_in_message":[],"_revision_number":4},{"id":"79a1b7fbd8d5f0ef84567269ea52c91a68d2959d","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 08:00:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5ca50678f2790298fe6bb7b4aebd5a64efc75742","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 08:48:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95f5916c12094b608675e971fd96a341 : SUCCESS in 36m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0457727e30634cd0a7683b63caf72b4f : SUCCESS in 39m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b85146a5eabcf4164f7c52e98873862e885633f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 09:53:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/41079d9dbae84b0e8ce76d091c9b9cf0 : SUCCESS in 44m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bafcbf43abde4380b8091e2d4bbd9b46 : SUCCESS in 1h 38m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2e8dc8e5be424caeb58346eccc88913b : SUCCESS in 1h 29m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/53075bc52da34833923ccda534ca7216 : SUCCESS in 1h 36m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f99d2364c35d4a42b2f7e6ce296bb74f : SUCCESS in 1h 49m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4a100f3713c404996c731109ec4189b : SUCCESS in 38m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f4614268ca64367804f948784bb27e8 : SUCCESS in 31m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95605ca9b9994b0fb2d510bc848e4bb4 : SUCCESS in 13m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1adac9797e1c4c95af9e470226649687 : SUCCESS in 29m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27a64320a7da477191198cca0c66a097 : SUCCESS in 25m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/148777a369174f0fbf7d6c6c1130d12d : SUCCESS in 27m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0a5ada8be6471e9beef6b7f7f04105 : SUCCESS in 10m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/24eaab5a52254bd1b1b0dfdb270a45d8 : SUCCESS in 45m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/88a0944b09894794bd054956bbf1f7f7 : SUCCESS in 1h 15m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ec62535d3e2d4c9288c51d917b399f84 : SUCCESS in 1h 29m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9d84059687a480d8470709234cc8caf : SUCCESS in 1h 05m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7c94bfb02484ada80e73102e2f63224 : SUCCESS in 1h 00m 30s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c53dccd713e74ed09ebb382a9af7e6f3 : SUCCESS in 1h 27m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9067cceb11c44499950b76fde665c036 : FAILURE in 1h 15m 18s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4090d43782d1485398b69c41012ae564 : SUCCESS in 1h 32m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c3b43bb6b924018857f502d4f719f39 : SUCCESS in 1h 04m 29s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/7b51e002328e4ffa859aecc88addeaa7 : SUCCESS in 59m 41s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b2dabfacdcb645c3a886605225805f23 : SUCCESS in 1h 10m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fd093bd638a94cd1bce33fdc4ce481d3 : SUCCESS in 36m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa2ed7ca1edf46569bd0c65400d2dfd9 : SUCCESS in 1h 39m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/367c0092c5bf4879b10051ff006db4b9 : SUCCESS in 1h 48m 07s","accounts_in_message":[],"_revision_number":4},{"id":"cbfebaa94be6910d1f62d075bd47aaf27057d41e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 10:14:43.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/773282/4 : FAILURE in 2h 12m 25s","accounts_in_message":[],"_revision_number":4},{"id":"ba013a605e985d4b5014f601fa1248b1bc6485bf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-20 06:27:53.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/82/773282/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"03b03382f21c3c135e74d6c14fe96887ed70f40d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-23 01:43:56.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"86883556d102cebbb703baa3adc348a7c38c1bb9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-23 02:36:12.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/773282/5 : FAILURE in 49m 14s","accounts_in_message":[],"_revision_number":5},{"id":"c67c433fe26c7f6d9b3957b2d5bf25b1acf61f09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 02:57:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c517124521b348b4bca686421d0a4bd7 : TIMED_OUT in 1h 00m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a97609b7ef647f38a09511f866dc492 : TIMED_OUT in 1h 00m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9b5c15a5e49b44120eb63b235693678908b7920d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 03:55:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/66b75575acfb4354b0967149ba9393ab : SUCCESS in 52m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/47fb2b51478a4c6eae5da364123c9a14 : SUCCESS in 1h 31m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2248486f9ad54186988334252c32f138 : SUCCESS in 1h 28m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/47f126b0a9e44e608a848591d46efb82 : SUCCESS in 1h 49m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dd35c1b70397405e8a8dbea043d19eb4 : SUCCESS in 2h 08m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ace978f4b664d59918bf6d3ea29f4b1 : SUCCESS in 35m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fb843fbe2cc47e99e2147a055b168de : SUCCESS in 32m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0745d53c65754c45a3766167328d096c : SUCCESS in 14m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79fe48a73edc43b9944b086919a35947 : SUCCESS in 30m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/982465de9f5c4dbfadc5b8c0d108d9e9 : SUCCESS in 28m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d783869ce2b4fc9a5b2417d71a8cef3 : SUCCESS in 32m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07efa2ca5e0e46fdab446cd8ab632e57 : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/576f6547a4ea48818bd7e065a9fd669c : SUCCESS in 51m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9214d9dcdc4c467da94f579feb3947e0 : SUCCESS in 1h 20m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/03052f7495594da6a7f83224dbff2a54 : SUCCESS in 1h 45m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/130c35f99a9e47daa198bc5db2ecc08c : SUCCESS in 1h 03m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6677146bf30646d8abd99cd482685887 : SUCCESS in 1h 06m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/790fa802da9c46518edb3f27dfc266e4 : SUCCESS in 1h 19m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b09420041aec4bc1b90443a393255147 : FAILURE in 1h 24m 59s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb56d0d406114a31ac8ad67514fe184e : SUCCESS in 1h 51m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c79ad60d78c4b8692a270938f8fee5b : SUCCESS in 1h 05m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a45629144caa4b999a7e54a28638fc3f : SUCCESS in 1h 14m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f1b52659c36c4faebc1a8a126b295d21 : SUCCESS in 34m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e71afd61c10412599375a5f8993895a : SUCCESS in 1h 32m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/30fdf28fa8f64e1da4ded6e7deed7802 : SUCCESS in 1h 42m 20s","accounts_in_message":[],"_revision_number":5},{"id":"5abfe8f2a0e86ec21181e50684f1eeb4ad968b6a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 11:57:15.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/82/773282/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"bc07aeb9444479647453f3b4e5a7905b66cd1739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-25 03:11:49.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"a4ddec24c3e98490025829b6c199eb70ed832da4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 03:19:21.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"1f6992354e7a570f3b66e47c9d0404507cc9ac28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 04:10:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71652cc6e371494eb736c0de72bcf728 : SUCCESS in 46m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f043950c4bbb4cb98644b482d505b44a : SUCCESS in 43m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"49be99166b4b44c29c8475e79fe075bbad2c7671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 06:01:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/38f5e19edb074092b59f083fb227cc44 : SUCCESS in 48m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8a6d11d1b1044e98a35eff24c2f5fc7c : SUCCESS in 1h 49m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/57a2384349c6498ba3aa0cc313181530 : SUCCESS in 1h 38m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f8420af10c544e339ff7ed9c6149a3bf : SUCCESS in 1h 24m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2377cf151bc6425491e8e7fbdc973709 : SUCCESS in 1h 30m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f7b8c3e7c244df4b7fde1ebba812031 : SUCCESS in 40m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f8f43f0e2e345cdb79a302b4d3a31c4 : SUCCESS in 30m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f449fd1ac147809eeea8c5e15cf246 : SUCCESS in 11m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fa60e151fec4fcc9a3b3d50120e55da : SUCCESS in 28m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b606e33b811a400cb5bb77df350be1aa : SUCCESS in 26m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48bf2df3068440689ba0791d9aa6014d : SUCCESS in 39m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d1f7dfd8264482a9c2b8dc3a89d5ee : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/556d457d632f4478a6314cfb388d2d7a : SUCCESS in 46m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecd00375a82b4fa4ab23639d5a54cc22 : TIMED_OUT in 2h 14m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1438f922267840d08a1891a8cc7b2bfc : SUCCESS in 1h 40m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6996cbe70402434e9f94c1fb528b8524 : SUCCESS in 1h 02m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f0b3740de62947448d35983703023bac : SUCCESS in 1h 05m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4edea68ac51a40fcb74d3f481b67d8f9 : SUCCESS in 1h 26m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/889667e750b0490b8d755c50934e2375 : SUCCESS in 1h 35m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/342960e826034fef9abe1be02226537c : SUCCESS in 1h 37m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a25e0b86fc1462ca3d0a92bab13eb1a : SUCCESS in 1h 13m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/02f310071efd4f22922d1dbced5c9eb2 : SUCCESS in 1h 10m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cdf14c1e2d5c43ac98ec1eff4d91ca1d : SUCCESS in 37m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/da9ae6c827b540f980b0afeb06bc5218 : SUCCESS in 1h 18m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8d05670d26264b68a9078f73f502225d : SUCCESS in 1h 36m 00s","accounts_in_message":[],"_revision_number":6},{"id":"bbf418cfa35eafd7d938088d63b7bdda1b9d2070","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-25 06:02:37.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/82/773282/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"52ffb19d6f31e51c03abe728c404e852b8925d54","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-25 08:45:38.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5ce95155e7e00fd18c4171224054c06e5daefe6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 09:40:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5197ca5473637c91fe747eb6a28b5acfecc784c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 13:13:29.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9cb488da773041079e41f3bc26dcf6b3 : SUCCESS in 57m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/50a183db198d4dd2bc14e68816bb46c8 : SUCCESS in 1h 42m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/df8f2b393de146eba073424e072b38dc : SUCCESS in 1h 42m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/705698c4cf1049bc81acf010825a1ac3 : SUCCESS in 1h 50m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6f69cd001f274fe69865572ace38a1d0 : SUCCESS in 2h 02m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29dbd7ef940e41e3b7e6b105b9f3873d : SUCCESS in 41m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9954620daf543ef924133512815ae09 : SUCCESS in 23m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7c2830e3864c0495bd37196ff9704c : SUCCESS in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a01bf34bbd694ea6aa3dec978ff37db5 : SUCCESS in 26m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4618e9d115e4dbc9a58a8d544556da8 : SUCCESS in 30m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3521d011e50c4477987ee368e47072e2 : SUCCESS in 31m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f74ed1137a4569820fc772258edd3c : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/65f26ba131b74fa0a07566b802f3cd03 : SUCCESS in 42m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f135467d0344953adccd4ec6d9360f9 : SUCCESS in 1h 12m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1c16cb14b809439788e4d0f916d702d7 : SUCCESS in 1h 34m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/81dc2584232e498cb4288d1849c80588 : SUCCESS in 1h 17m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/af509fe4a4384506b39709213b9ff3f6 : SUCCESS in 1h 08m 15s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/68aac3e8e70446a2a54c0b86dbb9db13 : SUCCESS in 1h 23m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4bab77673038464a9686ecdfe70653f6 : FAILURE in 1h 18m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/95e6bea5b1f149cd9c5a0df24ff0649e : POST_FAILURE in 2h 11m 02s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88f9209555cd4dff9aea839c62c984f0 : SUCCESS in 1h 02m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4043dbc0f0b54d839b7bb7e227aa8423 : SUCCESS in 1h 06m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e4627a760fc943808a81f8b3d7f0f0df : SUCCESS in 35m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3a691b7cc954419f8397bd7ef183d8cd : SUCCESS in 1h 52m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ddca650fc80846688fd2f3864deda81d : SUCCESS in 1h 25m 40s","accounts_in_message":[],"_revision_number":6},{"id":"a98cd4218caf08b52d2d3d132fc0ebab906c1c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-02 06:31:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b114ca57724fc0e783f0f9583b89b146e63c31b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 08:12:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b83e25c70a1439ba169b46f50c4ffb5 : SUCCESS in 38m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/559655a35d4b4e0e80dc623c500374d2 : SUCCESS in 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"84d8aadc3f6d52edd8dd675669f5981410e98c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 09:11:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/85cd915a5c8544d6babbb7408973755f : SUCCESS in 38m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7243df496fbf41c2b0c9b00a7ca4c9ff : SUCCESS in 1h 36m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8cf0cf765c3a4d418dc922ea3b00d041 : SUCCESS in 1h 33m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9e054e5133a4a90a342bd4b15d6e46f : SUCCESS in 1h 42m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/60247117b7ee4033a3b4cb5b08a725d2 : SUCCESS in 1h 51m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a46182ae840489fb1f06490b99206c1 : SUCCESS in 37m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d9fd1e8ad0d4a02bf6b581d766d5171 : SUCCESS in 32m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5562d4e933f54c46861c9073609eca07 : SUCCESS in 14m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/031160a642ab43ef9fdddf7408e189a6 : SUCCESS in 26m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47091370dd12483f8d15f51595240789 : SUCCESS in 24m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cd481ad287a4f289f28d94a1720d924 : SUCCESS in 28m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da46785aa6794b14999b16b1113d1cb6 : SUCCESS in 11m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/983015e63406443e9f60dca25162246d : TIMED_OUT in 2h 16m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdc807bf57184ae6ba9760c8cdcd4416 : SUCCESS in 1h 10m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7017d917d6254db7bbee4565c5f4adab : SUCCESS in 1h 32m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2591bf27057a4e3eb2d356c141df49ce : SUCCESS in 57m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6bcbed98727746ddb17f06306d1d2cfe : SUCCESS in 1h 02m 22s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c88ac0dab65a4d1195799e960bed6c7d : SUCCESS in 1h 23m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3100b7da98054de7b9922f87191f4b29 : SUCCESS in 1h 49m 02s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4b4cf4f391d94a04816db33eb8828468 : SUCCESS in 1h 42m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f31416fc108b4d6fb693d82cd17f41b6 : SUCCESS in 1h 11m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c336815d5f6d4b4d8c077e7d4f053019 : SUCCESS in 1h 21m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1f9906f44c7648919dd9b7c5baca870a : SUCCESS in 39m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6d4ec5f057b348eaa3650bf515ebffe2 : SUCCESS in 1h 40m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b2d310c401914f2da65bc138621bc47f : SUCCESS in 1h 32m 58s","accounts_in_message":[],"_revision_number":7},{"id":"2b046235de7b9f134ec03b59ce207147f5a621be","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-02 09:26:40.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/82/773282/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"d49945c8d1f304821756f5aed45b92b2285f2936","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-03 00:58:27.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"7038807479d6d1daaf93ac414093f76985acc72d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 01:48:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/868ec2dff4b2486fa34cce790311e8b8 : SUCCESS in 43m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c30f10855fa34e05af83f7072e1171e6 : SUCCESS in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"927a9293c805ec47a02f93dab3cfa49a2b65e5c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 02:56:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c3feb79f35c54e478b5e2d71e39d3573 : SUCCESS in 54m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e4eef09c88e44d5385b339272de4d879 : SUCCESS in 1h 27m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8485ee4b3a0c4a18adc68f675164f257 : SUCCESS in 1h 26m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dfbc9314ca8e4fd4b30e19b2ce9b6060 : SUCCESS in 1h 16m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a6fb32374c6f49d89e357e4cb1fae3d1 : SUCCESS in 1h 31m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9cd44cac02d458c899aaf5330fedf83 : SUCCESS in 34m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a724553a91c842f4a32be484c14d21f9 : SUCCESS in 26m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9c3b001cb34438b062af5dff512834 : SUCCESS in 14m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8a076d1092d452b8e88a39df877b1f7 : SUCCESS in 25m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3065992045544b64894eeeb47b74755c : SUCCESS in 25m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59ec53c90409432bb929c479fd0d5d5c : SUCCESS in 34m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60add20b20f144dc97bcae66c0b70ad4 : SUCCESS in 10m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7df29d912fa1469a80b9ea8043949516 : SUCCESS in 49m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/144731f2047e47b2bb6904e85155adff : SUCCESS in 1h 36m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/10467c36e86642848e5421371561cf83 : SUCCESS in 1h 29m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee93cf32b88e43fca1609aa14b2d4bfc : SUCCESS in 1h 21m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1a3b075537ed4d83b879af47c6074f8f : SUCCESS in 1h 00m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fce104e7ac684182b418857465812284 : SUCCESS in 1h 25m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/274a773cd81644e199b7b6dbd1bc5c7b : SUCCESS in 1h 22m 37s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/49a0cea7a4104e4bb28115a6a852b1cc : SUCCESS in 1h 48m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fb2b9501b8b4cbbbab9beab42da4118 : SUCCESS in 58m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8ec643a8783047a18b659f9528699707 : SUCCESS in 1h 07m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5a4390846d5d4ba3906f7059de5814e5 : SUCCESS in 29m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/361a2b842d6f470883a4140e034a36ad : SUCCESS in 1h 37m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/67e7c7ab60b947989f453b0cd50a1cf7 : SUCCESS in 1h 26m 49s","accounts_in_message":[],"_revision_number":7},{"id":"8cbe6ba8c02268ceb81290f7c96c7f82330a2f44","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 07:55:39.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9e88fb9e9f20d881228017f6c23b85ad7086f5ef","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-05 09:23:22.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"12f3992d146ca68d221b04f341b3e2a4f37f03e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-05 10:08:25.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3e0afc58cc291b5c9d50fe11c9d34155b1470330","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 11:18:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9666ce11fd5f4bf49f0371d481f4e051 : SUCCESS in 53m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49c614e3c0bd4b9bad212e4627ea85fa : SUCCESS in 41m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c43bd8f30065acd22422d699b901dcf5fcd9987f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 12:26:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0677a73cfcb94681a024a5899413876c : SUCCESS in 45m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ae1750bc59294f9a93d88095f10d5e0e : SUCCESS in 1h 29m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e81a213dbb8c4d159c8d6d2f5500c1fb : SUCCESS in 2h 09m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dd17d4959f9d4034ab5a3654c0fb848b : SUCCESS in 1h 25m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d8edac428dfa4829a7b716de206fbdce : SUCCESS in 2h 04m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4604806031ff4d009ee29bea15d80d84 : SUCCESS in 32m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40fdbafa40374214a060d24dc97a4029 : SUCCESS in 34m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6664ebb2e0644e91839bd526c0860a32 : SUCCESS in 12m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d99108c2434670a15a88bb63c5b7f4 : SUCCESS in 34m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/377a5b7ff56b4156a15dca3ee1a68e3a : SUCCESS in 37m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ead57448c636450d8ede258bc5bf4a8b : SUCCESS in 33m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95344215da124642aba83fbea53f0788 : SUCCESS in 12m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a596d91d76c43f6b3bb5c740bbf03ec : SUCCESS in 48m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f7b5cbcc38ed4d60babb5a54e5fd6aea : SUCCESS in 1h 29m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/44258c218e0c431caa561b041a7ff3b5 : SUCCESS in 1h 29m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e8ef7024dd44fe9a4cbbaefe6c70573 : SUCCESS in 1h 05m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/13e629fd854d49c8a18f3dc1201d7441 : SUCCESS in 1h 01m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/213d1987ba9c4982b57b5e27f0cfe25c : SUCCESS in 1h 27m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f3751b69b97c486293e022a437eead17 : FAILURE in 1h 14m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6f44ed561824c53ad08da8f56da87d8 : SUCCESS in 1h 36m 20s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/500054579cae42d1b848f1a7c2fb6782 : SUCCESS in 1h 22m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1a68f78fddb043ada3391c44150b511a : SUCCESS in 1h 17m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c9e4eb2bb27e45d8a44c421eafacbed3 : SUCCESS in 34m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/60252c403e784d79a2f63dc0c6770711 : SUCCESS in 1h 33m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5e87f039c4b84e28b7a195bd60a7887a : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":8},{"id":"7474bc1e8902ab89388de0b6de8799034cb5895f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-05 17:38:07.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/82/773282/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"748971c53f10e55503628a1d096e558d94b1021f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-08 11:24:50.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0ad16c16ef126148494672a43bcb3d84205523c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-09 09:06:15.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"bfa8f29534f7bcc55328e05ab6a24d7b88ba706d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 10:01:08.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f01941146b6f4f2c826bab982f82229a : SUCCESS in 48m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77beb3220e0249b3a5823c8624a391ca : SUCCESS in 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1ff2c4be9e598c9b235f5c089dbcd18a0e1389c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:40:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2f49bb09aa33455f9ce16b79de7458c5 : SUCCESS in 54m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1e34c1e6f9a04ae7ae5b5fba7eda644b : SUCCESS in 1h 57m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/785fce2bac8c4a019f1cbf72407ec84c : SUCCESS in 2h 04m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a2cd32c881e4459ba843485f43d2b26d : SUCCESS in 1h 53m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c36128aea87e4823b84bb91be67e043c : SUCCESS in 2h 34m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed10904e944f4004ac62161d68e78eee : SUCCESS in 40m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b3e670f72384ad98e32bd4df7990a3c : SUCCESS in 39m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e9fb92d3cd43b1aebe3a0546083e95 : SUCCESS in 14m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f785f0dc153b42788337172ded398684 : SUCCESS in 36m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a7cc06235704ec2b18dd6ad187de69e : SUCCESS in 41m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76b644d4ef1944bc8dcbdac47ae1d489 : SUCCESS in 31m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920c8470b6d34bb2a1a4531f04b3a592 : SUCCESS in 10m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c009455d93848f9a1bea5e18d960f0e : SUCCESS in 47m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c2a27063d2b42c483a2d0f88c4158bd : SUCCESS in 1h 26m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8eb47d576f8244fbb21a90d9fbc12745 : SUCCESS in 1h 29m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4abeb1bb691a4384b566e9daab985231 : SUCCESS in 59m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0daa199913d34a888d6f2da0d3eada56 : SUCCESS in 51m 47s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cbad82c9c92443929fbe2ae50bdb2c7c : SUCCESS in 1h 25m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a9c4f5498654c30a77f6afd50c43887 : FAILURE in 1h 20m 21s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08f09aca9f1947ef81c5ddc08f130a3f : SUCCESS in 1h 41m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfea84614e9d42e598e55d72342781ec : SUCCESS in 1h 50m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bc5fff3244714ae99cc907246fd59b77 : SUCCESS in 1h 07m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/22d86a3a5b9e4ca1b4c3442ab60516b8 : SUCCESS in 33m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e968f7cdf8904722826e2be5a46a6e38 : SUCCESS in 1h 29m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ab68e69949b643baaf72601c303c167a : SUCCESS in 1h 22m 15s","accounts_in_message":[],"_revision_number":9},{"id":"d4d2764b8e59ac6e2f26e1ce933fdcca2a1679ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-10 03:03:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"127f6f2a55ca25ca28297ad5a0cd45809f8a8cfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 04:03:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ec45929de0a485cb5c8c13d885573a3 : SUCCESS in 47m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2109a2772b9143b5aadab7dcea9fa575 : SUCCESS in 49m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e2cf9e83b0d14beb6c3def04e49a9232bda537ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 04:43:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/98ed4e0a866546c788516e410a6f6741 : SUCCESS in 53m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fb1c14a74a484382b43b401568093f53 : SUCCESS in 1h 33m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2ed5a5dd999b4c7a99680ec1f9e610fd : SUCCESS in 1h 11m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a9119b362ba24c32a85c765892c798eb : SUCCESS in 1h 26m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4670ef2324fb462a8deb842e1bf71d9c : SUCCESS in 1h 37m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/312ee9bab7a346d3be0e5f95ce0c82e0 : SUCCESS in 43m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a58d8b70e40d4bff8209b62cbd162f3d : SUCCESS in 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e29aa4eb9092407588c7a7a6b274f896 : SUCCESS in 16m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a6f94a1472b45db9c74cd260feea6be : SUCCESS in 38m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fcc1b31f8a042fbbed32e2b3c70c2a6 : SUCCESS in 39m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5895a92bb8574bd28c8c0e465dbdd568 : SUCCESS in 39m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb2891ff6e34fb0bc6a25bc645134d4 : SUCCESS in 9m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef3bd970d45e4b7ba2f34589dff9f91e : SUCCESS in 50m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d968957ef6fd4544b2034cd596141584 : SUCCESS in 1h 17m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f487b8480371449b8c66c6a434fce681 : SUCCESS in 1h 26m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ef4e0c3b5b845a694df0e6dc2525682 : SUCCESS in 59m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/507be2cbe5a841728c335915b14c383b : SUCCESS in 53m 45s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/81dd83812c094b9dbeba2e3e5ff179df : SUCCESS in 1h 09m 58s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a38eb10bb30e45baa372c18235c9e337 : SUCCESS in 1h 16m 47s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0bcf439993dd4ff8bb04bd9b005a5ef2 : SUCCESS in 1h 25m 41s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ad80df0414343c3bb7b786fd4214c90 : SUCCESS in 58m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8b550cf81b50418cacb8a2ea22587451 : SUCCESS in 1h 02m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b7a18a9407a940c2b77baa4320a82e53 : SUCCESS in 34m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/29f2ebadd7804ffc83b5b09a75fff425 : SUCCESS in 1h 25m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7d455c4d65c047dca68ce4d2adc00138 : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":10},{"id":"21605a5288f3fb6938ec03c7c9e476eae51b2ce3","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-11 08:14:17.000000000","message":"Patch Set 10: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d186a4ff78e7c4918e254c6c03564f46faeaa5d9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-11 09:10:13.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"9cdce0f1d8a74d2b4e00a8b8c6df2da16a8511e1","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-11 09:26:21.000000000","message":"Patch Set 10:\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"34053227ef2d36e992ab3625aa4cfc008ff72b05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-11 09:32:43.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9adcc28fbf19cc094a0cc53517fe97ad2aadcd99","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-11 10:30:07.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0cfd02e709999222337a9645696462d14a46f646","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 10:34:58.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d655ac1ea374e2092822a08ccbc9595 : FAILURE in 1h 00m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a2febe24ab748c6b4aa876f38999ad0 : SUCCESS in 45m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9863d43580c0458465b790f5c075d91f0ab26d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 12:02:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9b3cd6d24f7a4036b5e52d8c4f86ec9a : SUCCESS in 39m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/abedb4fb76bc436e8473c1c85a55f168 : SUCCESS in 1h 01m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/acfcedcc2b694528a176983c4d838ea2 : SUCCESS in 1h 24m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d86e90402782497c82ea229236eea139 : SUCCESS in 1h 06m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9f1145bb08cd4a629363460173a9e816 : SUCCESS in 1h 35m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e226e352d5a04237a90fb892125c17b2 : SUCCESS in 24m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa6240b37f6847ceab12e2d0eec0ddf3 : SUCCESS in 33m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df307986a2e445ce8ed8c2e8be483c3f : SUCCESS in 15m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30427fe7f3ad439b84a2acd2f518b08f : SUCCESS in 50m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c0a1c7fcd7043b8ae1efd8a5b1bdecf : SUCCESS in 39m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b76dc1044d564a19bf4ee609debcc60a : TIMED_OUT in 42m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd14ac7424a4227832599995e382f07 : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/35feccf250df4c97a9059b8f00b15d7f : SUCCESS in 48m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1deb3a22b867482b9f6d74d7780dd03e : SUCCESS in 1h 25m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/38b4469525af4910a1c746de4f35184c : SUCCESS in 1h 27m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e2f944c76e741988fdca2c76ebdaee0 : SUCCESS in 1h 00m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a8135b50649c46f4990742b9b1a7bc4c : SUCCESS in 59m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b0a2299418234b12919d1ca34954e928 : SUCCESS in 1h 52m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3cf8842d654b4775860d9efe8cc640a3 : SUCCESS in 1h 34m 55s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03a6d9b41b004273abb49f071152a0a2 : SUCCESS in 1h 35m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa0520d4e16a485ab4ab52cb327d0ef7 : SUCCESS in 1h 09m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ffbe42e1bb8f44e2b50748d8d991518a : SUCCESS in 1h 21m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1f60d40032464a9796c83f1c2bdcb2ed : SUCCESS in 41m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/20aef04f4625463bb65a5860e337096b : SUCCESS in 1h 52m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/26e25d375f8f417fa17bcfb7fca22f67 : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":11},{"id":"91570ee6690854bd6f4f23533669e8e42f49fda6","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-11 12:35:38.000000000","message":"Patch Set 11: Code-Review-1\n\n(4 comments)\n\nSorry for not mentioning it in my first review","accounts_in_message":[],"_revision_number":11},{"id":"38fde28c0068d24937250d1dd5d08bae1f83abdf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 01:11:36.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/82/773282/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"7577eb413b1619198017c64f7c6fa32d60f73793","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 11:13:08.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/82/773282/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"9c56342df09a12c88f285a4c0fc9a2299bdd3631","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-12 11:37:02.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"260cf137ad8f9aa2453adf340bf556af15523a82","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-12 11:49:26.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5783aa66c69eb0e138e5c52d24c6046e1d27bfdf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 22:40:05.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/82/773282/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"309262b4174c0caf9843efcde88267f33cc33e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 02:38:58.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a824c513e1102f6f3b689e33d1635fb8c3b105bb","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 02:39:10.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"9423c14ae9f5bf4f640ca30d839b8e20fe7943f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 03:44:20.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c1d2f5ab699432480123769e6122ce7 : SUCCESS in 47m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78d81ca3724e46f7b17b60cc8fbf4217 : SUCCESS in 57m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f50b19f22c8001d9351a3f7e5d72bb01c810c9ad","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 04:14:51.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/82/773282/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"7fe528bd444a4268806a39a1a4ef8ae0c52ffb96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 04:42:51.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/91e93d2e0bad4c98b28e7627acd7f4d3 : SUCCESS in 1h 00m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/92c65af8c5e24904a8d854f6f5ac375d : SUCCESS in 1h 14m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/33f2c78852b5442192091c798c5df415 : SUCCESS in 1h 41m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b924937d92ba417d976458aa39589fa3 : SUCCESS in 1h 24m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0ad7e3b0a87f4c898610f0ff7386ac36 : SUCCESS in 1h 57m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e081cbad501b422e9e977594c32eb57f : SUCCESS in 43m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e146b13a877424f94e841aecf44ca22 : SUCCESS in 39m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54ea28eb6a3f4bb4b4cc2863b2cf17f7 : FAILURE in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d46c106d0343487f96b488ef89f103ac : SUCCESS in 37m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f925ce011bcb41b6b30bd3b97943b9af : SUCCESS in 39m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40968272c27a49dbb72668cebfbccb8f : TIMED_OUT in 40m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b395e8e75e94b818407ee58206567da : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3fbe02c98cca4a83a955968f984a79ac : SUCCESS in 54m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/38e88aeb472f421c96f3cf69bd279a6f : SUCCESS in 1h 22m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fe39f0d357014abf90d822c072cb6f82 : SUCCESS in 1h 39m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c33de26026f441578f402f20c15980b2 : SUCCESS in 1h 13m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9558a8b99e046fa8e7e27f8f1848186 : SUCCESS in 1h 15m 23s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4611154a32c2410e8486cdcb564ccfaa : SUCCESS in 1h 23m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f90e748b04a74ba5b61c38d0752c8901 : SUCCESS in 1h 40m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7693997c721747c3b980b65167e7f301 : SUCCESS in 1h 46m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0e7651991124f7da1aa564175bcb4a7 : SUCCESS in 1h 10m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f933cf17d40f4d0096633e2037a9942b : SUCCESS in 1h 11m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/234348abfbac4727904960324688d48b : SUCCESS in 40m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3f9718360f1c422c924a0bb9aaf5598f : SUCCESS in 1h 35m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6411e28a711444dfa918f72431aff2b5 : SUCCESS in 1h 31m 46s","accounts_in_message":[],"_revision_number":12},{"id":"96b53d56109e575c57aec681fdd4673336b6689d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 05:34:33.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8851604d93deb7d4d23e0725dc6ff9494139f215","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-15 05:50:39.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"8909c631664328c11bbf976e6a5ee8d10551d364","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 06:25:48.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/486091137a3c49b88bdd9a5d5da71382 : SUCCESS in 50m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46d341492be9474fab31a2aa371bbb79 : SUCCESS in 45m 21s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c16176aaa69ffc3f8cf8ec376ebecccb1eefffdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 07:52:35.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7b3aa28baae549d79272c4e3daa19d2d : SUCCESS in 48m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/152d0e06d3ba4edab49eb5f8e9fcec57 : SUCCESS in 1h 34m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b5f6d5409a664ec9ad306bf1304f563b : SUCCESS in 1h 35m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/65ba9731eb244ea094d35b487acb9162 : SUCCESS in 1h 35m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/77afbf59a49c49b9b270d33f9168bab2 : SUCCESS in 1h 51m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/041024845ca147eba59ed535667c1f3e : SUCCESS in 40m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7155d3ce50246feb752356b2a0e18ae : SUCCESS in 33m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efa7ebc873ce4d68b82d9cde01eba82d : SUCCESS in 16m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a2fa4905dec453eb8b91d3ba565a9e7 : SUCCESS in 45m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac23386c874846cca5ea84d638da1f24 : SUCCESS in 32m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9f70a4758f44eb38698ec33f8826014 : TIMED_OUT in 42m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ec0328a610747bab9c35f0739a96b3f : SUCCESS in 12m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/adda624a1717485db991a132ebc61e5b : FAILURE in 53m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f879258b5ecb47429596b1e379dc1e17 : TIMED_OUT in 2h 14m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5f14911af97a4b10b6af0dd8702e7f7d : SUCCESS in 2h 05m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c82f20e80501477b87815b9fe9e15a96 : SUCCESS in 1h 31m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/966c5ded524b4adea7d848d022875d3b : FAILURE in 1h 12m 54s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6264b641132d4ebaa0844e0019f9ad56 : SUCCESS in 1h 40m 32s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3cff303c45f9415697693156faabe792 : SUCCESS in 1h 52m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf3cb9bd9933405e897aae0e191c78a6 : SUCCESS in 2h 03m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62c22bbefcbe4093a5f040bd82b0cac2 : SUCCESS in 1h 16m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b63b54960729406c9cc93a11821158f3 : SUCCESS in 1h 34m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/59d3ca549fd34591a0659cbba0276468 : SUCCESS in 1h 02m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e54ff1770e734a3e9b93e953687da384 : SUCCESS in 1h 55m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4bff7570ee4d4a7db9b865b22c7b131c : SUCCESS in 1h 47m 28s","accounts_in_message":[],"_revision_number":13},{"id":"d657f47938e32b170e059d89a347d214e8161e4e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 08:11:09.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"63a6523efd90115c5ac3815cc46d8ec860678369","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 08:27:57.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/82/773282/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"24c998c7c829c6a2108f988f67afe1adfe3381c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 09:21:41.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0aee8aa04e34642874d97d430d0eb1a : SUCCESS in 41m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3878827de4e477c8fa188ad5b449adb : SUCCESS in 56m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"2849fbcdc399391ae4b9bd1aaf22e5812cb937a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 10:21:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7b60e06e38164aa0a159910b98e53467 : SUCCESS in 1h 04m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3312c23749b3417a9928a42a29fc6bf9 : SUCCESS in 1h 26m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c06fc2a896ed4a998e37fb1a79517610 : SUCCESS in 1h 41m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28acc046900245dc9e43f6748cb10678 : SUCCESS in 1h 36m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c5ce4286cf164f0b95f9d7f02a21ff4b : SUCCESS in 1h 59m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7601002d029e4b8ca8cf7f6a74924e5f : SUCCESS in 44m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fcbe727742443b59032c56601771996 : SUCCESS in 35m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6df033e9f5c4455abd1feccde3f7de6 : SUCCESS in 14m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4faf994b7dd148b29d43eeefc6e30f3b : SUCCESS in 36m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/132db62487b443e1b3d43c503116648c : SUCCESS in 35m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4fcffd8d55f4bffbf6ac41dd6d43e74 : SUCCESS in 31m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c19f4920d24ca3b0e73dfe45c2d942 : SUCCESS in 10m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73caf1f5b90d4fbdafb2ec2c256ef970 : SUCCESS in 56m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b22d4068fac94785ab3aef131aad4595 : SUCCESS in 1h 25m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/74d812a8023a4079b160c75184d535c3 : SUCCESS in 1h 59m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa28b4ba586146c8b69686f93d8d0cc1 : SUCCESS in 1h 05m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b6589c61aee445a79889326cd18d70f8 : SUCCESS in 1h 11m 55s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2682f004f1b14e02a845bf4c53314e14 : SUCCESS in 1h 27m 36s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/270a28cfcdfe42b48980929a876c717f : SUCCESS in 1h 43m 02s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cf82b220145b40408010dce1f4064a50 : SUCCESS in 1h 53m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3ea81c7d00e41199f07d620ac314fce : SUCCESS in 1h 04m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e9fca7bbacd2435fa1095bb818c7b212 : SUCCESS in 1h 13m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0840cdc29f304926a4d6f6e56c22ef58 : SUCCESS in 52m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/efb8d95309d041ebb1983e11f93cc212 : SUCCESS in 2h 00m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e11177a055a749d7921546f9c8712ee8 : SUCCESS in 1h 24m 32s","accounts_in_message":[],"_revision_number":13},{"id":"a674f55d4bec42ea999cca1572727b910be4a1a9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-15 11:41:45.000000000","message":"Patch Set 13: Code-Review-2\n\nWe are now passed feature freeze deadline for Wallaby. Let\u0027s wait until master will be opened for Xena release to continue with that.","accounts_in_message":[],"_revision_number":13},{"id":"e6dc92c693d87ee23effe69446180c5127785bf4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-16 11:21:50.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"80bc5038425f019e4181d6cf18b26f23fc2fc994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-30 09:26:52.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"c8212f538b189939e663bb937abc326a3eef92ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 10:19:22.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0bc9896cb6c44ef79fb590fb59b4a013 : SUCCESS in 45m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aea8864d031c42acb406ed6657becd81 : SUCCESS in 45m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"64e284ab9b485fbdbd994caaae64597447a8e13e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 11:41:30.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/01d84f9a2a514b5cae77ce642c390132 : SUCCESS in 43m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8732d54fd0654315b008311b85d06b47 : SUCCESS in 1h 55m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7fa8f4bbcda24974ac593441a48ef91c : SUCCESS in 1h 42m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a5b8a44f7b3a4a7ea24137890f855c2e : SUCCESS in 1h 29m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8be60fbb37ce42eeb8dbffa7a55c9933 : SUCCESS in 2h 10m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e1626b3577d4ec596d818077cd5a3d0 : SUCCESS in 42m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/599146b3a2124137b7a0d8d5a7d76693 : SUCCESS in 34m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d30e3e07675440abaec887cc0ed7112 : SUCCESS in 15m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/125c357df219497b923d613ac0363356 : SUCCESS in 27m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dc436092cfb474594cff866a9044dde : SUCCESS in 26m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53b26ac7e82749909b7d43e57295c24c : SUCCESS in 29m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6ea2bb14784cb0ada9bc155fd61074 : SUCCESS in 11m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a74a65c9819348d1891b9c2514055745 : SUCCESS in 54m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7223d7f56e1842b78b4d293aabaaec31 : SUCCESS in 1h 19m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4e37f7fda9c84b229532e6755dbdb62d : SUCCESS in 1h 31m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/60148106bfa94a76a2f90c5a1f9723dd : SUCCESS in 1h 01m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c64172c1bcc1430e9425e3d5a3198730 : FAILURE in 42m 57s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/55adb81b5e3142d7a03fa1a1b1c40f8a : SUCCESS in 1h 16m 38s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f566bf2f985143d584529d282f8aaa6d : FAILURE in 1h 02m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f58de9cd12004e7d911debaaf7ab7011 : SUCCESS in 1h 21m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/837d1a1750bf4feeb9a04ffd995f0a9d : SUCCESS in 1h 06m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/40731a953c5c4391880b8f832423e4d4 : SUCCESS in 1h 07m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/289240127cd74e238950186e541ba43c : SUCCESS in 39m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b0f62ed67904eaf998df023bfa6202a : SUCCESS in 1h 27m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/61bc791cfa3f4bb7970f1a1ad508feb2 : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":14},{"id":"fcb63dad70c23c6da5919a35d0ccc9bd04cbda03","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-30 13:51:05.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/82/773282/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"5252e9e9a336d352b501921ce8a9df1c27c06c54","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-07 14:45:30.000000000","message":"Patch Set 14: -Code-Review","accounts_in_message":[],"_revision_number":14},{"id":"acd0cbc4e1c74abfe1d54180c747e8e44e146f9d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-08 14:39:15.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4cb4a89dcae93155182a9b41d2909b8270d1e4e3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-16 00:50:34.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d9a9c97fe16d1b20da748f53d93e832b49ea2eac","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-16 06:37:30.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"76923822ccf7d4330e0cc9cab6c39297835888e2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-16 07:17:55.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4f959d7eba52f478be47a6ca730314f0808b77eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 07:18:15.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"599a72016d4751d25a74cc95c257e8f2c026e58b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:40:13.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/38b767ba818847f98f69cfb2168d005a : SUCCESS in 57m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7993a471ad4a47f482eed25deabd3845 : SUCCESS in 2h 34m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/45a45b9deef44ba2a26ebacdfdc10870 : SUCCESS in 1h 59m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d468b5ffcaf44e5fa6a7133cf2902a1c : SUCCESS in 2h 09m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7a2a9fbc7e8d484bb5b8191c6508acc8 : SUCCESS in 1h 21m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/356f85ff1e2f4f04ab723cbbbdf1ea6a : SUCCESS in 38m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c645e9885ace49e18108afa56cccb6ee : SUCCESS in 12m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b69e227e4d504fedbe482a54bef8f69d : SUCCESS in 19m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c913d04d480454f9c47a166b2d7f4de : SUCCESS in 26m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b5a6a42942346caa8e89c0a009a9a58 : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d8891c344d043f699af3a984b95b53a : SUCCESS in 56m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/38428d32d09143bf81ed76acbced2bc7 : SUCCESS in 1h 29m 09s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/be4b05e57c864e6f82a41d31d94256b4 : FAILURE in 1h 18m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/05797814907346a7867ffc4d9d3affbb : SUCCESS in 1h 03m 57s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7e2eea58d834457ead0745acb8b6a586 : SUCCESS in 1h 47m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dab572c9776145cdb8938a29864a1707 : SUCCESS in 1h 02m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/11ab308de1274ef88f5346511891a9c6 : SUCCESS in 1h 14m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/626648cd818d4b33a5adf252108cf1f0 : SUCCESS in 29m 43s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/600d8e8dde074cb98c32cc2c3071ba44 : SUCCESS in 1h 19m 59s","accounts_in_message":[],"_revision_number":14},{"id":"913d456694bb7f587fb2b886e97fb6f6b2252f26","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-19 02:09:19.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"8d2476f081285e9c0c0d65a5883caaeb78be3f71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 02:51:10.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e5d92f726aa48d89482fc90128048fc : SUCCESS in 31m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81d1bc9a6c994ae4b45c2146d26cb316 : SUCCESS in 37m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"bef27512800c221c8560427c9e95895aec1efb64","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-19 03:49:07.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773282/14 : SUCCESS in 1h 38m 57s","accounts_in_message":[],"_revision_number":14},{"id":"857a07644ae33304ea66729d08a029512cb65030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 04:48:17.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8a62cc4d396a4fcc9b5f691b4db754c3 : SUCCESS in 48m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f99f6f34b7bc43e1bc9093cea1aa919c : SUCCESS in 2h 13m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5c3ad51d45104aec936a1f371d625ea1 : SUCCESS in 2h 00m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7aa072ffecad4d4f8ba44c12ebb3bb59 : FAILURE in 1h 22m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1f8eea2c67b04ee6b09c3af5fae49663 : SUCCESS in 2h 33m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65acda98d7314045af4f1eb028fdbc23 : SUCCESS in 37m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/649f5f35f2144ab59a5c972fb3b0f1c0 : SUCCESS in 38m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05fcf9f4e3074c82a42b84be3121be9c : SUCCESS in 13m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aedc7c92466d4de89c5623f3515d1dd8 : SUCCESS in 56m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcb90466f5ac4892b7e5c38438e4e763 : SUCCESS in 29m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76239fd8ed1f4941b8203de1e95b36ea : SUCCESS in 29m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67681e91bc1e4f7aad6b6530135493a7 : SUCCESS in 8m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e96165d3c38e4d7494c4f1ec5810a2e5 : SUCCESS in 53m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6971492dfc0f410094f0a12c76adba1a : SUCCESS in 1h 22m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d540e9f8af8f4d24a3d744664481d7ef : SUCCESS in 1h 34m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7410d7797b2465998c1130944f260c1 : SUCCESS in 1h 00m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/86714542f68a42b7b531cece416aa8bf : SUCCESS in 1h 01m 05s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2ae29cd6ff794613a0232bb86b6cccb3 : SUCCESS in 1h 12m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b41c7e3e07244726a46ad41eceaf8ac0 : FAILURE in 1h 03m 54s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/80b5e650bb384d2697e35983b57e4d4a : SUCCESS in 1h 30m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85c429db2039423ebf31ebe3cebc12ad : SUCCESS in 1h 01m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/08a4d827412b41579fbdae9282ebc174 : SUCCESS in 1h 04m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b3c9fd333ae4448794115a7020bab4cd : SUCCESS in 35m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2dc8651908564696a41518d45e772252 : SUCCESS in 1h 13m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/089910383a024a8e8e1ac831f116dbd0 : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":14},{"id":"fa8bc2086fc0ff48d73a8da9fd8a3100127b3c72","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-19 09:19:56.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"ee18fa3c3e32394f0768b7f49d492e42f597a7a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 10:11:09.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/249ff77af8f648d1a62712b301a98278 : FAILURE in 37m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50147bf84b8149aca8766a1b4bbafd72 : SUCCESS in 48m 06s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"99b002696f1eb7b581aaccba9d85b4fb1b3450b4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-19 11:02:57.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773282/14 : SUCCESS in 1h 42m 14s","accounts_in_message":[],"_revision_number":14},{"id":"58ba9ef5cb00199f8d958fc5a11f7966fbcc99f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 11:28:47.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1c5c39551c8043fca7a71903c422b522 : SUCCESS in 51m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4ddf621cbefc49969753bf6be0c1ad1b : SUCCESS in 1h 55m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dae4be37495e47e7a9111644dcbf56d4 : SUCCESS in 2h 02m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2fd53e80703d4382985f64217e5a0b88 : SUCCESS in 1h 21m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ecbf52f205384bbc8fdfb0390204b985 : SUCCESS in 2h 05m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59c6819c76ef44eab5b89df214fd00fc : SUCCESS in 44m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec2a1897956a41648fa4500a403ff544 : SUCCESS in 34m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd69543c2e14aaf9d23691fd9c8f013 : SUCCESS in 12m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5a61c9cfb8046879758a4bb3be8e95b : SUCCESS in 31m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b7ee6fe902248f4a3aed8d45355ec2b : SUCCESS in 29m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7350f25048cb4757bd7fd790173f02cf : SUCCESS in 22m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d350c392ae146908aee1a6ea68680a7 : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/65fce8baeaf8413685a2dad1628193f1 : SUCCESS in 48m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/72a33729ca094eebbb4733b44549856a : SUCCESS in 1h 22m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a753df33c0f24b08bd65970bb5642f38 : SUCCESS in 1h 22m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cdfda283190c40f791eebff7757b9915 : SUCCESS in 1h 03m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f624438c0a941bea26bdb92d6b63e61 : SUCCESS in 1h 01m 58s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/23c03a4c1dd64d808082cf7f3910f44e : SUCCESS in 1h 11m 32s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/35779e8a48b44037a2d7f990ecbd3b9b : FAILURE in 1h 06m 49s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7039963750e54a67aed334cc5d5a0c9b : SUCCESS in 1h 30m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/554dc363a5e24c11bd16ca37ab46395a : SUCCESS in 57m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5f475d6ce6fc45388fc4d7a208148e1e : SUCCESS in 1h 14m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/14d2f768151246c8bebddcd6168ff4df : SUCCESS in 33m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a28422d5a03405a813c79fad824c9b6 : SUCCESS in 1h 22m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bf3a9947c7b549f8ba1b3371d83aa70f : SUCCESS in 1h 23m 09s","accounts_in_message":[],"_revision_number":14},{"id":"4c1dea093e4315fcc8161dcdf9b845cd2b3dfcdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 11:29:09.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"af561e46fc317eb48f432d6606c5b25753e78f44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 18:16:14.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/af828afe2190437aaf0e8636f8877ad9 : SUCCESS in 39m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b238aa174c814929b2b5f06aeb60f103 : FAILURE in 1h 41m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/00671a31a31a4d08a5b7cd508a40a4f1 : SUCCESS in 1h 13m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2baa642fa0634986b7ff3e8b3958ceda : TIMED_OUT in 2h 56m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ec46818e35c34fb6998f1198a168cebe : FAILURE in 1h 04m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42b34d68f97f4ae2a8c0bb9139783bab : SUCCESS in 36m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b41e9e5c5df41c4be7d1ea2774f6e3a : SUCCESS in 14m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/495fc1beeff04a4eac6184f47bd7332f : SUCCESS in 30m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/190b5d3ed98b4d92b17adc7c243012a5 : SUCCESS in 25m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d77a2e1f932416f91e06d6e4f5bd1a9 : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0900ace82c0e4e33b6097cf51a106921 : SUCCESS in 1h 07m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e64ac921b78442ccb74e636a972ecf20 : SUCCESS in 1h 33m 42s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ee458b078972468e959a7be2c3bcacd9 : SUCCESS in 1h 02m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89bb078fce9a4d309b9d06451eaef0f3 : SUCCESS in 1h 02m 40s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/527251f44bd34a149db78a5d454b79f3 : SUCCESS in 1h 35m 19s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2746c2c8196245beba65eac25666521b : SUCCESS in 58m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c88e5b3875134dbba15807d2e6360c62 : SUCCESS in 1h 14m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/738457b8307543f288d5d0941a7443e1 : SUCCESS in 49m 57s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/67f56a3505714ab0813497d1c1fdec7b : SUCCESS in 1h 28m 59s","accounts_in_message":[],"_revision_number":14},{"id":"bb1727811b06d6f1738676880c286d2775e477b7","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-20 00:35:44.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"fce3a162de7c6647263dbf067713de76be4723d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 01:20:23.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f067cfe7a404811941d3cc3906be5f5 : SUCCESS in 33m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff2beb4d968e4ec58f8fe9727c8bf74b : SUCCESS in 39m 15s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f6440f41771b2b897441d69019e38ecc79e38092","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-20 02:26:38.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773282/14 : SUCCESS in 1h 50m 10s","accounts_in_message":[],"_revision_number":14},{"id":"8d35d4fad5484905700ead0fc5dabdf57f92dc5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 02:38:50.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6eb55863c8cc4426a3169664116649c9 : SUCCESS in 46m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a82cb897a5ea468e81421a3515a3d53e : SUCCESS in 1h 57m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b96215719cb44041a61bd3963d757437 : SUCCESS in 1h 50m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bc8c51a9e78b412f97eee6dc55b47b44 : SUCCESS in 1h 07m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5e471e499daf4125a9c154e0c4cb19d7 : SUCCESS in 1h 31m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93537dd023b44693bae1f63f4e96be81 : SUCCESS in 49m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb8f28c9844f41598e5d8deb6460f81f : SUCCESS in 33m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9423e76efab4eb5a2a1f396d046e4e9 : SUCCESS in 13m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2177e8ebddaa4f1f9104ae5658fffb07 : SUCCESS in 30m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3b281b68e0b4d9c9b807f574f6f3bfa : SUCCESS in 30m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8d71e3a0d14440da0e97eb54fb26c19 : SUCCESS in 37m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38644f2ffa8c49668d1ba5943772c0bd : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/820dd1e1840d49489056a593f7fae808 : FAILURE in 54m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/45f4aca030d24cf9b85d2ab5b86a2f37 : SUCCESS in 1h 22m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e4e35c3981234705b3618c56f0b1d1ef : SUCCESS in 1h 37m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/13ecdfcaed7a459ca218aa7565932411 : SUCCESS in 1h 00m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6fae6c3497fb41f3b028a59d2c40a2a7 : SUCCESS in 58m 31s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/96cbe66037404c0fa2c68e872686d17a : SUCCESS in 1h 17m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b13ea2abab1e4344b28195cf1c653037 : FAILURE in 1h 10m 59s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3b4fab616e2c4b0e8c91e0e9bdf15819 : SUCCESS in 1h 25m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed6eb3bef83e414697bb0ff5787f4571 : SUCCESS in 1h 13m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2b6cbdbdb86c422eb4824db08d361547 : SUCCESS in 1h 04m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d95da0c7193e47de816179621091c868 : SUCCESS in 30m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b7339d88ecfd4bf0a420bb1db38845e5 : SUCCESS in 1h 39m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4c31e301b1ad4b3c8364124734a8717f : SUCCESS in 1h 19m 35s","accounts_in_message":[],"_revision_number":14},{"id":"e06729755f093503573cadfec67fd2a88e8e304a","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-20 11:03:57.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"f41d20a32242712a5d39378c7544e0e3a7da19ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 11:53:28.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2713ac92c00349b09e6081b0b047e811 : SUCCESS in 44m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1fbd11f4e1e04edea31b0906b0f929f1 : SUCCESS in 42m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2e9c901c1d3d9b12149761cfe9d94371052d8001","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-20 13:02:14.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/773282/14 : FAILURE in 1h 57m 29s","accounts_in_message":[],"_revision_number":14},{"id":"b6df324f696704ff5a1dcaccf34810b6d254a72e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 13:43:02.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7af8d397ff8d434eaad30f471d636cc8 : SUCCESS in 34m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7c9bf97d77464558bfd7cef2ddfca8f4 : FAILURE in 2h 01m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/42de27b0b8dc4bc1a197bab8e21b5c7a : SUCCESS in 1h 21m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/24bacfcabaa742398b4384a241c37452 : SUCCESS in 1h 38m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4f1b9899ebf948359d362dab63e206a2 : FAILURE in 2h 31m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5d6cb6dd1784ce8bc6b0dbac3098c7b : SUCCESS in 34m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ae900869f60470385aed4929295f054 : SUCCESS in 26m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aec85cf0ff842bdb8ada4fa63836af9 : SUCCESS in 12m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/874d5935a8e64d60bf6fae8452dfd613 : SUCCESS in 29m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2ba30b7f6fb44379ed3903c591f5b5b : SUCCESS in 29m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55b2c48a74be483b8232948d031e6336 : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f124936125a40249b7509a3d3dd8e34 : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/953afeeb81aa48b5a5ed39f0c27d11b6 : SUCCESS in 46m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74bfbde3026f4f42bf02fa1034214209 : SUCCESS in 1h 16m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9fba63168dc94b1ab3c106128bb56459 : SUCCESS in 1h 19m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/00a46c4c9b8545c683cd041f7c49028c : SUCCESS in 1h 00m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/20d4a60b3cf645cea412d6b65d7e0d5b : SUCCESS in 56m 03s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/10d792d7218e46a0962332e1d6a923a9 : SUCCESS in 1h 01m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/24a64dca5b8649c388c0950d5c3c16bb : SUCCESS in 1h 17m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e3c6aef989b14fe68de83df63687bf70 : SUCCESS in 1h 33m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f117dad8a8e345a7837b2abcde4debdb : SUCCESS in 54m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8cf46afc0bc04fe58a3f3e072aeafe3e : SUCCESS in 1h 07m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/11f219e294dc47f68c79a967829e43d6 : SUCCESS in 28m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4243fc06a0fc41afbff26d84777bf175 : SUCCESS in 1h 06m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2919b154509f49eca38a04d7c8ba3b9d : SUCCESS in 1h 15m 54s","accounts_in_message":[],"_revision_number":14},{"id":"2e08d08105e68e020b2c1c50fd1191cb0332f951","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-21 00:31:48.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"1806d1cca5da71a87ad04191eab2dfffa624c2ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 01:21:40.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be3f2e4ef8e542439a9a658df819b6b5 : SUCCESS in 43m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25962e1c40784465b76f6577bd89263b : SUCCESS in 45m 36s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"dc1afa0e3ac101111254a7cf146d5539d30a0ace","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-21 02:33:09.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773282/14 : SUCCESS in 2h 00m 33s","accounts_in_message":[],"_revision_number":14},{"id":"3b6d737699ff3f291c5847830ecfbd3e3718daa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 03:14:00.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aed3b2d7a9b14991951a8ce3040637d8 : SUCCESS in 44m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/04566ab49a8a4772baee28d8d8644ad1 : FAILURE in 2h 35m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/550e068515e944008788566a82010643 : FAILURE in 1h 49m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/35b5b39937694d9094f9be1b9f2b8f23 : SUCCESS in 1h 14m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0bf949b658dd43ceb8dae9073781ca0c : SUCCESS in 1h 45m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ba65a985e454b0a83bc2d1cc4719310 : SUCCESS in 39m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/716ab4ca65564f5e99830c4675be8b99 : SUCCESS in 33m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc5ac863a2740e4b3cf6983b43c7bfa : SUCCESS in 12m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53079ee250d54afc8a39df850bd97d47 : SUCCESS in 32m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/933fb3a8793c443fa97a7bbd0b729a8f : SUCCESS in 28m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be0c907af99e400f8cad9df22228bee2 : SUCCESS in 30m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b090aea619ae4a36a1f9bef21de864b8 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e7b077dd04b4b699aa720cc43791a39 : SUCCESS in 1h 03m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/61b6c14d749040a8a9242b17596f826d : SUCCESS in 1h 24m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1573744a7a7e49d0b73f0984213839aa : SUCCESS in 1h 32m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea6871bcef6c4c50aa8be269141857cd : SUCCESS in 1h 23m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/abd488b5c1d34111820e0b83d86fb257 : SUCCESS in 1h 02m 58s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6c7da3070c444569bcd3734bf2de6af2 : SUCCESS in 1h 12m 15s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1e3479638f914302bff411a69cea7b1b : SUCCESS in 1h 32m 21s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/331b664a256944ab835b109d415fc6c1 : SUCCESS in 1h 09m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0a278364da348509f63c904a0da8901 : SUCCESS in 55m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/71a2124a0d354a1cb6ef915dbc40fa0e : SUCCESS in 1h 27m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1e94205f279a46f79fcbe761995be06c : SUCCESS in 30m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d5b245ff98145afb630a6136fbce185 : SUCCESS in 1h 44m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d030ec18808b4eeea6c4fb3d42873e91 : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":14},{"id":"9c650fa99f637ac4ce2e05d3ac1074635f7cf27e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-21 06:00:18.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"628f7fb2f9e8b06123298f2582caed371817c5ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 07:14:39.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c490d9178ba473ca81b73e65b82d9fd : SUCCESS in 54m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"622fcd224d4a161eada26cec0cc8c9bde80500a6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-21 07:49:20.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773282/14 : SUCCESS in 1h 48m 00s","accounts_in_message":[],"_revision_number":14},{"id":"03ef57d26243150395601192796cddd4826ebf74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 09:43:38.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/411a644777464b65b8487b502f976229 : SUCCESS in 42m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/01e20031be514361b0acac1125000b81 : SUCCESS in 1h 50m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/49124911ec1b4d60903e534126fb5a9e : SUCCESS in 2h 37m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/99d17b9bd5134fb789433c131fb68f9e : SUCCESS in 1h 28m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/53c2db7f3317437898da2e92488c448c : SUCCESS in 2h 21m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0214d12ab5d488c82981b8cf865b8c3 : SUCCESS in 43m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3f237dfb6f84ad4aacd97be14271594 : SUCCESS in 39m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8194ce949ab42e2859c004c8c76e26d : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21df30bff2d24143ba453e2b4f6457e6 : SUCCESS in 25m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e00a38d10c4f95844383cf68cc3463 : SUCCESS in 35m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5af2725417dd4ea98e8200d400f034ed : SUCCESS in 34m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c81f01b9d648829aabd9199fa5fbc7 : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f35bb2c74b44471af6e290467092edb : FAILURE in 50m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/06de780980664762bd0512fc5204269e : SUCCESS in 1h 22m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5f1a8f25d5064bf2a4e5d642f9e9b0da : SUCCESS in 1h 17m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab01699acbbf405daf6f3bafe8e709d5 : SUCCESS in 1h 01m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/53b05e8578b6472d986530ead91f3a91 : SUCCESS in 1h 06m 25s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/25bd901ffde34281b746e89e803362e7 : SUCCESS in 1h 20m 33s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/06dbbe44de774ac6bd062be088494658 : SUCCESS in 1h 36m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a92372403ed642d593f84632cd268261 : SUCCESS in 1h 25m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc7162c02539419ba1b292fdb9212239 : SUCCESS in 1h 16m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/66c9a94f0a7a434fbd732ce4fa94bc77 : SUCCESS in 1h 06m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a5d87497f9974fe1989e91ce31f07106 : SUCCESS in 31m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0dfbb1f97942471397f33cfc484f7b57 : SUCCESS in 1h 58m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a475fdf7529e4a5aa1901c88c60ef509 : SUCCESS in 1h 35m 21s","accounts_in_message":[],"_revision_number":14},{"id":"4979ded4b0d9bf891afba18364daed661b6fe33f","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-21 10:16:37.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"bd9c57204a8b74b8d2500f63e89a391307ec05a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 11:02:29.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51b3e7363acb43d0a4534a4bc6424968 : SUCCESS in 42m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd03c608090b4d2284a283b99acf7417 : SUCCESS in 41m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"fffd675dd36cf10898362a8f5337275b9216b526","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-21 12:12:35.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/773282/14 : SUCCESS in 1h 55m 13s","accounts_in_message":[],"_revision_number":14},{"id":"e695b8f842d91f82524172920d2746764927786b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 12:55:11.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/424ea2556e014b7685c350163d712de3 : SUCCESS in 39m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e393b2635ffd42ffaa166e6087233451 : SUCCESS in 1h 54m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aa89c81b671d4246b680dd61b15d3ec2 : SUCCESS in 2h 05m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/11fd636feb7e424796502fca837a6383 : SUCCESS in 1h 14m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bc664debbfa04824a3c824238467b4ce : SUCCESS in 2h 35m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aa5ab0766d541ab8646f867a44cb088 : SUCCESS in 31m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6530465eff9d42ff81aa7598773b4ad9 : SUCCESS in 26m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5940914eff914817afd59ab4437587c6 : SUCCESS in 12m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d462c908a1a348998845371f3ea587f2 : SUCCESS in 23m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16c9882fb0a94309959db045d5ac9ae3 : SUCCESS in 22m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48ce179f0b0f4198a92d765d5b93de09 : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b78bece4fd3f48a99d046b9cf363dcb4 : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a87813eac4ff4c90a700e6930d955cc6 : SUCCESS in 45m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/26eb42d7e6e84436823101666ef9498e : SUCCESS in 1h 20m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fc4e1b1f814e46e999b24678f245ef43 : SUCCESS in 1h 17m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/10877f6c8b4a4e2fb9bd98c6e18441e6 : SUCCESS in 59m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8455d112c06442f1b2b7cc1c48d99e6a : SUCCESS in 1h 02m 50s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7ccda33feaf1426d8b996da655c26288 : SUCCESS in 1h 10m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f80542b9f6d0431b87db78ae59a5f68c : SUCCESS in 1h 29m 45s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4fe62b82a09a4500903f8fca2032fdee : SUCCESS in 1h 42m 19s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/577d261400d14734bff45e1dad75694a : SUCCESS in 58m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/03fb666be60b4791b7fdafdea9e18e44 : SUCCESS in 1h 07m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d4cac89cd8e743a8b08daefab1430699 : SUCCESS in 28m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c8e4895bb41461788be4a54eb69fcca : SUCCESS in 1h 38m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b33e1d63b4d9485f9d8d27de79f2661b : SUCCESS in 1h 35m 17s","accounts_in_message":[],"_revision_number":14},{"id":"acad604c2b90009da262269787580dc560197bc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 12:55:49.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"03fc89b1dd1f354cae9152a593c7df0f88eefa9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 18:02:47.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ccf4eef94a414b079a5829d6fdaafc10 : SUCCESS in 44m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/27b71868a8f14132abbc9fff12b68e1f : SUCCESS in 2h 10m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1965a7ff2b3f4531a70691c9efb38b67 : SUCCESS in 1h 44m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2c9939def8f74332a0a05f5d47244e4e : SUCCESS in 2h 31m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d296ce6def1a4b07946ac32a1fdc56c7 : SUCCESS in 1h 43m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/435b9c4eaa684a72a4dcb7386f4404b8 : SUCCESS in 40m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70bea5dd7ad44adc8775bb61bb793e8f : SUCCESS in 13m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d71a3477974574b2cde8360d815cc5 : SUCCESS in 34m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/480acf5e6d0941d9b8c4d17da5a362a3 : SUCCESS in 27m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/926d5814049344e0be158092fee783bd : SUCCESS in 15m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/397936adea39406f88ceb1c13649ca00 : SUCCESS in 54m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c595515f50d2428899b830d03e09a90f : SUCCESS in 1h 34m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/69c2f8b989094303809637830b3aab12 : SUCCESS in 1h 18m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e50ab0cee80c4905894f1ce191ce08cb : SUCCESS in 58m 55s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2331f0eab6ec40eb8bb2d4e4ea6f70d6 : SUCCESS in 1h 26m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca2d6b2aa9804011a49229de6ee53020 : SUCCESS in 58m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7ded9873fe044b75a14f6aad14440858 : SUCCESS in 1h 06m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/aed023d159f34094abebd208bab8bfec : SUCCESS in 35m 32s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bc4fadf0601c4ca88a0cdb93ce77c930 : SUCCESS in 1h 20m 24s","accounts_in_message":[],"_revision_number":14},{"id":"3b8346808b9c7bcc1829c7a651f3aecb3b035439","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 18:02:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"c75149ddf5c18c9507fb139092dd486df851d863","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 18:10:45.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b59733f25944d44b3853d10316af2a0 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"b0022c51fed2c824ab85c0ec8f5dc2b9f5a13c59","revisions":{"601bd6d5b90c87c8a95eb93fcdd5a06fc5e255f7":{"kind":"REWORK","_number":1,"created":"2021-02-01 03:37:46.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/1"}}},"commit":{"parents":[{"commit":"420d68c7923a32536b071bf5e4eaaaf545d96b82","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/420d68c7923a32536b071bf5e4eaaaf545d96b82"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-01 03:36:57.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/601bd6d5b90c87c8a95eb93fcdd5a06fc5e255f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/601bd6d5b90c87c8a95eb93fcdd5a06fc5e255f7"}]},"branch":"refs/heads/master"},"a6f76cb687caf735ca376f79fe347090cce0347f":{"kind":"REWORK","_number":2,"created":"2021-02-03 04:22:31.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/2"}}},"commit":{"parents":[{"commit":"23cbaac4217c27b4049e8014d469c1977e49ebc2","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23cbaac4217c27b4049e8014d469c1977e49ebc2"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-03 04:20:37.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6f76cb687caf735ca376f79fe347090cce0347f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6f76cb687caf735ca376f79fe347090cce0347f"}]},"branch":"refs/heads/master"},"47f1127b3ab4361f3ff14915c706b5f146a33564":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-05 05:50:33.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/3"}}},"commit":{"parents":[{"commit":"90b662b5253555a9a812df06118c6fcc67362e29","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90b662b5253555a9a812df06118c6fcc67362e29"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-05 03:21:07.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47f1127b3ab4361f3ff14915c706b5f146a33564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47f1127b3ab4361f3ff14915c706b5f146a33564"}]},"branch":"refs/heads/master"},"9d066adb6767da0e708093b98f86017b39a6a23b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-19 01:54:59.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/4"}}},"commit":{"parents":[{"commit":"055ae2bbb0e80d7b6bff1b0146a85f0396a0ccb1","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/055ae2bbb0e80d7b6bff1b0146a85f0396a0ccb1"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-19 01:53:59.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d066adb6767da0e708093b98f86017b39a6a23b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d066adb6767da0e708093b98f86017b39a6a23b"}]},"branch":"refs/heads/master"},"6dae578a241cbc281803a773ed12cc0de64991d8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-23 01:43:56.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/5"}}},"commit":{"parents":[{"commit":"917288dbe460e3fca52411c4161b35d8936a34bc","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/917288dbe460e3fca52411c4161b35d8936a34bc"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-23 01:26:46.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6dae578a241cbc281803a773ed12cc0de64991d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6dae578a241cbc281803a773ed12cc0de64991d8"}]},"branch":"refs/heads/master"},"b59af9cc7a25477773aed30bba1619f3acd8532f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-02-25 03:11:49.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/6"}}},"commit":{"parents":[{"commit":"07ff02b5522da8ffc4ded3c9303c057d8cd1b7d0","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07ff02b5522da8ffc4ded3c9303c057d8cd1b7d0"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-25 01:58:33.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b59af9cc7a25477773aed30bba1619f3acd8532f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b59af9cc7a25477773aed30bba1619f3acd8532f"}]},"branch":"refs/heads/master"},"eaf78784c52d7c9126499c701ac501cb4a291952":{"kind":"REWORK","_number":7,"created":"2021-03-02 06:31:20.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/7"}}},"commit":{"parents":[{"commit":"3bc4a357549d0e9a87f10ec798f59bc0fab1e792","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3bc4a357549d0e9a87f10ec798f59bc0fab1e792"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-02 06:06:39.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eaf78784c52d7c9126499c701ac501cb4a291952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eaf78784c52d7c9126499c701ac501cb4a291952"}]},"branch":"refs/heads/master"},"923a8e278622137539223bb408c366911359f0f2":{"kind":"REWORK","_number":8,"created":"2021-03-05 10:08:25.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/8"}}},"commit":{"parents":[{"commit":"22310ca64ec4f5159d09606f9722cc5cfc97834f","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22310ca64ec4f5159d09606f9722cc5cfc97834f"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-05 09:27:29.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/923a8e278622137539223bb408c366911359f0f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/923a8e278622137539223bb408c366911359f0f2"}]},"branch":"refs/heads/master"},"5262f07966d89665fc8467142d94521de0a04ebf":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-03-09 09:06:15.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/9"}}},"commit":{"parents":[{"commit":"bcbb8c848db0dec9c5be3d423c07fb369495c677","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcbb8c848db0dec9c5be3d423c07fb369495c677"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-09 09:05:18.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5262f07966d89665fc8467142d94521de0a04ebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5262f07966d89665fc8467142d94521de0a04ebf"}]},"branch":"refs/heads/master"},"71b5543b40d07a1322c938f105cc42877a09f5c1":{"kind":"REWORK","_number":10,"created":"2021-03-10 03:03:31.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/10"}}},"commit":{"parents":[{"commit":"50242af1b3632804c991e834786c0a88b1042b7c","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50242af1b3632804c991e834786c0a88b1042b7c"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-10 02:36:06.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71b5543b40d07a1322c938f105cc42877a09f5c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71b5543b40d07a1322c938f105cc42877a09f5c1"}]},"branch":"refs/heads/master"},"87506765e1baa4cf3f9532be3bfe042b47b855dd":{"kind":"REWORK","_number":11,"created":"2021-03-11 09:32:43.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/11"}}},"commit":{"parents":[{"commit":"50242af1b3632804c991e834786c0a88b1042b7c","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50242af1b3632804c991e834786c0a88b1042b7c"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-11 09:29:22.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87506765e1baa4cf3f9532be3bfe042b47b855dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87506765e1baa4cf3f9532be3bfe042b47b855dd"}]},"branch":"refs/heads/master"},"fecc359c87ba18e9aa871e81b0802132c95b9a8a":{"kind":"REWORK","_number":12,"created":"2021-03-15 02:38:58.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/12"}}},"commit":{"parents":[{"commit":"50242af1b3632804c991e834786c0a88b1042b7c","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50242af1b3632804c991e834786c0a88b1042b7c"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-15 01:31:05.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fecc359c87ba18e9aa871e81b0802132c95b9a8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fecc359c87ba18e9aa871e81b0802132c95b9a8a"}]},"branch":"refs/heads/master"},"346f3157b4469a118e8fc003989a35f7e5978dee":{"kind":"REWORK","_number":13,"created":"2021-03-15 05:34:33.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/13"}}},"commit":{"parents":[{"commit":"50242af1b3632804c991e834786c0a88b1042b7c","subject":"Add the base class DHCPResponder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50242af1b3632804c991e834786c0a88b1042b7c"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-15 04:54:41.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/346f3157b4469a118e8fc003989a35f7e5978dee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/346f3157b4469a118e8fc003989a35f7e5978dee"}]},"branch":"refs/heads/master"},"b0022c51fed2c824ab85c0ec8f5dc2b9f5a13c59":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-03-30 09:26:52.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/82/773282/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/773282/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/773282/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/773282/14"}}},"commit":{"parents":[{"commit":"493286511d227c7ee8189838199c673110af1ee8","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/493286511d227c7ee8189838199c673110af1ee8"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 09:09:33.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-30 09:25:59.000000000","tz":480},"subject":"Add the DHCPReponder for IPv4","message":"Add the DHCPReponder for IPv4\n\nThis patch implements the responder for DHCPv4. The main\nwork of the responder is to assemble the DHCP packet with\nits protocol required options based on the information of\nneutron Port/Subnet/Network.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nRelated-Bug: #1900934\n\nChange-Id: I8b0466263388085bb1949ea37f4788579c46093a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0022c51fed2c824ab85c0ec8f5dc2b9f5a13c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0022c51fed2c824ab85c0ec8f5dc2b9f5a13c59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
