)]}'
{"id":"openstack%2Fneutron~774045","triplet_id":"openstack%2Fneutron~master~Ie1d837d17bf50a4675bec4fe48ab13dde22ebdc2","project":"openstack/neutron","branch":"master","topic":"bug/1600788","hashtags":[],"change_id":"Ie1d837d17bf50a4675bec4fe48ab13dde22ebdc2","subject":"Drop meaningless translation marker","status":"MERGED","created":"2021-02-04 07:04:11.000000000","updated":"2021-02-06 04:08:38.000000000","submitted":"2021-02-05 14:50:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774045-bug/1600788","meta_rev_id":"2e17ee9becacd6331aebb228585bb8b87f544574","_number":774045,"virtual_id_number":774045,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 14:50:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-02-05 10:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-05 12:56:00.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-05 12:56:00.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 07:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-04 08:56:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-04 10:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-04 19:28:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-04 20:15:58.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 10:30:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-02-05 12:56:00.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-02-05 22:29:20.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"406852a5c5730be4b43c891299cf53fd47f7a021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-02-04 07:04:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e7f6d13f7af6c245c66fff0c36b71f43eaf9d91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 07:53:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5226b69a011b4359b164a2558251f5a7 : SUCCESS in 43m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02479c123a76492c8571b33632bb2341 : SUCCESS in 43m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26399dc6b48becf0a174ccc9b0f84dc8081e2628","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-04 08:56:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774045/1 : SUCCESS in 1h 51m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e9f31879d4ffe61efa80a5fdae15ada398fb3b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 10:00:43.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b66f874ee5904c6a85fb210be1134d15 : FAILURE in 12m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/369714723886493294aa12f2f1b8bea3 : FAILURE in 12m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d6ab30fedf71478fa85df9d0dde78c2c : FAILURE in 11m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aa4f9374a6a041c4937213c3081882d0 : FAILURE in 11m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5b9467d752f4068ab5528ee171f1f1c : FAILURE in 10m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/13f2790a296d4cc28fc6453d6a3ac382 : TIMED_OUT in 2h 49m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04438a0e08984f8a995cce45b36b4b5d : SUCCESS in 36m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dddb35f211f9431a97ea754b975f4ba1 : SUCCESS in 28m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85abf8dfeca34f8ba1abfd809428ce79 : FAILURE in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f4ca5b9c1304fba95e6bffad8dacc04 : SUCCESS in 29m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83438a4470414cb99dd77a3cf673482e : SUCCESS in 24m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7fe20ea05a34362bbf196d8b3705e21 : SUCCESS in 26m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b53c3d7dc3ce48209aeb1e6be03236bd : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c9af1d6dbea433480a0f33fb33c6086 : SUCCESS in 1h 15m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa4bd182a5ac44038c3ee41e533275b9 : SUCCESS in 1h 13m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1535635c65ae4957ab19aab95391df77 : FAILURE in 1h 49m 22s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68e9801af3ea4d8386018ad3b4e81a0d : SUCCESS in 59m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9828c52b20c6467a83402827511ce72c : SUCCESS in 1h 01m 33s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b107a95b932e4fd790597f5421298f64 : SUCCESS in 1h 17m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/90af3c14799c42bc8f1b3096f9f804ac : SUCCESS in 1h 25m 03s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f180f53f8a8946a7affc5c269567f5e1 : SUCCESS in 1h 37m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5698ded5e73454a9e6216222264b1ba : SUCCESS in 59m 56s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e6dfb68451964b2282ca417c48230008 : SUCCESS in 1h 07m 12s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7e8b82b2836c4587aae6a931983726db : SUCCESS in 54m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/47981946b8e64125996cbd235fdcd5f6 : SUCCESS in 29m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c564b4ae3f6b407db84b9a92d32f6fd2 : FAILURE in 1h 41m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cf431a17163add403f4ba16a373b0a952d6b748","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-04 20:15:58.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/45/774045/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7141f76471d7f188b94059806b5f537545af9e38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-02-05 10:15:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8d845877efa801ddbf92d8c56defda8fd5cef655","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-05 10:30:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e5c9611176bb42ebcea29ef04a1ae901f876623b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:58:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b900af3c70a4300ae6165468577d7a8 : SUCCESS in 34m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3f6d5f2927014a3bbe480a67a4d2a1f3 : SUCCESS in 37m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f24f21e21825543cae95bab5bd144c133ea60f0e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-05 12:26:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774045/2 : SUCCESS in 2h 10m 16s","accounts_in_message":[],"_revision_number":2},{"id":"8462820936cf31d6059771496ccaac98dd2e926b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 12:29:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/974f4a56f73a413fa935def1fce3f329 : SUCCESS in 55m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7279cdc28f46446e8929992c12fb4f6e : SUCCESS in 37m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7b94fd05fb8d45fcaa4cdf9305cd9c99 : SUCCESS in 1h 44m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/23a0f6450b8c42c7bcc68d04a0c91fb0 : SUCCESS in 1h 37m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6edff0b19d3a4e71820d4aded56ab304 : SUCCESS in 1h 44m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3184db4e0f5a416c93d4d623dfe8a100 : SUCCESS in 2h 07m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9765a42472e54af2854e891fe8ad16a8 : SUCCESS in 35m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f81918d9503437294a0d676089f7a97 : SUCCESS in 27m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7980e3beeaa44d28a5f0625beacc2a6a : SUCCESS in 11m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df2bbbecad704c0c9f167bbc7a2fa890 : SUCCESS in 18m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a29ff783689f4416a625ac483dec708f : SUCCESS in 22m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72ce0fc120674e76b0de242970c5f0ae : SUCCESS in 25m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b19546639756414db9cacc7a2643b962 : SUCCESS in 9m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/abe6f33cf57e499e9b8f17299cbafa97 : SUCCESS in 1h 12m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c04e8694863947d991c04f33d295b1e8 : SUCCESS in 1h 09m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a81bb599e83143228c329ac2b0105718 : FAILURE in 1h 44m 06s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/854afd2ba6744a988f9a052a6041481d : SUCCESS in 1h 04m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/63f1edf5107b47fca026819e53deb70b : SUCCESS in 58m 58s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f77ed6ab964f4bf39bf0cd8fac97df14 : SUCCESS in 1h 21m 49s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/613c8a060a6843d78f00fd7f3923201a : FAILURE in 1h 15m 18s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ffb692764ef74283840c4c7e7f340dae : SUCCESS in 1h 32m 23s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1fb21aad0134056b22b8db23080cc84 : SUCCESS in 1h 03m 52s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d94ff73ff7074e2ab6a4a5a9dd38fc4a : SUCCESS in 1h 00m 28s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e504697262bb47539659695d6feaca23 : SUCCESS in 1h 02m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a3572c2a0a714afb887ce7c05a40c743 : SUCCESS in 30m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fbc421fe3e55466ba64e11990216dcc0 : FAILURE in 1h 31m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05a60c61aa3a95f14711b70414431ea18f597fbf","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-05 12:56:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1ee66fd9943b7e09d4f4ddb1427b175e316a0284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 12:56:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"068bc3ff585f68ebe1baa5c0344faac3098ef660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 14:50:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c9666ad889234951baf439e065394dd7 : SUCCESS in 47m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ceec8fcb5fce48efae7561a87811d4c4 : SUCCESS in 1h 49m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2271b5a9a54141089fc52022172ca32a : SUCCESS in 1h 32m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c8e6817761e6452c8fab503c9f9ece04 : SUCCESS in 1h 38m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bfdb3a2771514700b9ce9edfdafd5375 : SUCCESS in 59m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbe02936860b459f8d7e1794f2a9d977 : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba18819e692c4cb78be308a59c785f19 : SUCCESS in 15m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca54b43bf145418e884e724136299c68 : SUCCESS in 27m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6eab35b85f5c43859c8f1bad236c05f7 : SUCCESS in 26m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f465dbc630c94546bfc3b66bc9deb4e2 : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/52d3f665099d44659e29b02cd2ffe7db : SUCCESS in 1h 09m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/34c64b06addc4d3fa4a9c590b0b0745f : SUCCESS in 1h 14m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7faf26ff93714fc98bf8795d2f955ddc : SUCCESS in 1h 18m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79205ef3363644f09b6562689823388f : SUCCESS in 1h 05m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c769853eae6247eba4592202082aa9ed : SUCCESS in 59m 10s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/177f24298c2c47b5b1750d15236f15e0 : SUCCESS in 1h 44m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0632d41ee5174e5fb4e7c00f57da9c1e : SUCCESS in 1h 02m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ed3689efa43d4ea99b7881086e31fede : SUCCESS in 1h 18m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5e3aef41315745bcb833c60f6740c835 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":2},{"id":"eb54fde079d6b9ec0fc5bedf0530b1343be58497","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 14:50:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c3a7134a992e82a994b1f9aa724713bf9da86b9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 14:52:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d31f45d54cc0440cbadfe3a3f1c5bdb2 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d59cb780029d97c34e756b9790e89b2f6c162bfc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-05 22:29:20.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/45/774045/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/45/774045/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"2e17ee9becacd6331aebb228585bb8b87f544574","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-06 04:08:38.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/45/774045/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"257a51769bb070600ca3d6bc169cf530bd33f1a3","revisions":{"cebc980fee12ea3b231a017dda2e8b54f1ecb026":{"kind":"REWORK","_number":1,"created":"2021-02-04 07:04:11.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/774045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/774045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/774045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/774045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/774045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/774045/1"}}},"commit":{"parents":[{"commit":"07aa04b955473333aed3f8dd34eb1532a6272d8a","subject":"Merge \"Change how ovn mech driver checks if vlan transparency is supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07aa04b955473333aed3f8dd34eb1532a6272d8a"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-02-04 06:37:23.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-02-04 06:37:23.000000000","tz":540},"subject":"Drop meaningless translation marker","message":"Drop meaningless translation marker\n\nThe translation marker _(..) wrapping a variable is meaningless\nas no string is extracted and there is no way to translate them.\nOnly rare exception is the same string is defined in somewhere else\nbut I\u0027ve never seen such situations so far.\n\nIn addition, we now only support translations in user-visible messages\nused in API, so we can safely drop translation markers in exceptions\nin the agent side.\n\nChange-Id: Ie1d837d17bf50a4675bec4fe48ab13dde22ebdc2\nRelated-Bug: #1600788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cebc980fee12ea3b231a017dda2e8b54f1ecb026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cebc980fee12ea3b231a017dda2e8b54f1ecb026"}]},"branch":"refs/heads/master"},"257a51769bb070600ca3d6bc169cf530bd33f1a3":{"kind":"REWORK","_number":2,"created":"2021-02-05 10:15:09.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/774045/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/774045/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/774045/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/774045/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/774045/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/774045/2"}}},"commit":{"parents":[{"commit":"07aa04b955473333aed3f8dd34eb1532a6272d8a","subject":"Merge \"Change how ovn mech driver checks if vlan transparency is supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07aa04b955473333aed3f8dd34eb1532a6272d8a"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-02-04 06:37:23.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-02-05 10:14:41.000000000","tz":0},"subject":"Drop meaningless translation marker","message":"Drop meaningless translation marker\n\nThe translation marker _(..) wrapping a variable is meaningless\nas no string is extracted and there is no way to translate them.\nOnly rare exception is the same string is defined in somewhere else\nbut I\u0027ve never seen such situations so far.\n\nIn addition, we now only support translations in user-visible messages\nused in API, so we can safely drop translation markers in exceptions\nin the agent side.\n\nChange-Id: Ie1d837d17bf50a4675bec4fe48ab13dde22ebdc2\nRelated-Bug: #1600788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/257a51769bb070600ca3d6bc169cf530bd33f1a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/257a51769bb070600ca3d6bc169cf530bd33f1a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
