)]}'
{"id":"openstack%2Fneutron~774061","triplet_id":"openstack%2Fneutron~master~I0fc4ffd5db2673cefcd01307555d1a3900bb12e8","project":"openstack/neutron","branch":"master","topic":"improve-neutron-ci","hashtags":[],"change_id":"I0fc4ffd5db2673cefcd01307555d1a3900bb12e8","subject":"Move tempest job with neutron-lib master to the experimental queue","status":"MERGED","created":"2021-02-04 10:37:01.000000000","updated":"2021-02-19 12:34:24.000000000","submitted":"2021-02-19 12:29:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"774061-improve-neutron-ci","meta_rev_id":"f246af6c97d06585e7c401e47d154b09dba51907","_number":774061,"virtual_id_number":774061,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-19 12:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-02-04 11:12:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-02-09 15:05:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-02-08 13:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-02-08 13:44:18.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-02-12 10:01:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 11:12:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-02-04 11:17:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-04 12:45:54.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-04 12:57:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 13:41:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-02-08 13:44:18.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-02-09 15:05:26.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"c81f311f34d3781df39d5ca52b66ccc3536189c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-04 10:37:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f127002231e18581892f03f3d0c9207a12c3c8fe","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-04 11:12:11.000000000","message":"Patch Set 1: Code-Review+2\n\nGood clean-up and reorganization.","accounts_in_message":[],"_revision_number":1},{"id":"daab9d4f8609ad27fc4284b1a3d88bb555bbe762","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 11:17:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a42ad912165e4687be1ff505f833f606 : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da96e11c4cba4c85a98fb4da85521f2d : SUCCESS in 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7acd5d5015c327da6eb2cd3f88c831a6bdf0737f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-04 12:45:54.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/774061/1 : FAILURE in 2h 07m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3d1a2b2d91690e529768f5036d6a55443ef15817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 12:57:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/edd30dab32fb4918a1388c1a1056465f : SUCCESS in 50m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/28e0704c404142a69180f53abc697ca5 : SUCCESS in 28m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5f3952d7ee6b4164b3e2b8df21283177 : SUCCESS in 1h 20m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fcb1e893a82e411eb6de2a609ff3775e : SUCCESS in 1h 11m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/58cf0e55d45d47bd8c32c6b742333eab : SUCCESS in 1h 48m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6cd0db420add40bebaff66b71016f78e : SUCCESS in 1h 57m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f307bfa4b0143cb97dcf3c8699faf31 : SUCCESS in 33m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7f3c9b2548744dcabd2dc530f64ffcf : SUCCESS in 33m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/784cc8fa44fd4797a4ce09dc935812f0 : SUCCESS in 15m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3e5457e29fa4b4d8fca2995a578383b : SUCCESS in 25m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c632640280f4a55804243304226de66 : SUCCESS in 26m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/057fb67123b440589d4a4fd04ec2a6b1 : SUCCESS in 29m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7805a451714fc1b968b83d9ac92930 : SUCCESS in 9m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e51b5f9f1e6d4d79bd01e5e5223bfb7e : SUCCESS in 1h 08m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/99184cb435fd4c21a1ec593e1b38d4e6 : SUCCESS in 1h 11m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dbb3c61d5023499b8c76b77a68730033 : FAILURE in 1h 58m 35s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccaa39643d5643ba97661c3cbfa1f49d : SUCCESS in 1h 04m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f4202f7b31c491f9e06c1c363bdaa9f : SUCCESS in 1h 12m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1893c152e30049688c82ab32c0e80727 : SUCCESS in 1h 36m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/63238e57af9e415a862bcd0d717ad551 : SUCCESS in 1h 59m 13s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/42a1f1975d874fda93a463161394c5ac : SUCCESS in 1h 35m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7cc41dbccc24aba9641517fdb7c0bb0 : SUCCESS in 1h 01m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/297a710c1c77415995286ec00f8da455 : SUCCESS in 1h 04m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/da4c74ef202f4e4290e57b3f7dad3f3d : SUCCESS in 28m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7314b21576934532ab20b15a29743831 : FAILURE in 1h 42m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a6a08577801792b15cb70cc350739daa78e71ea","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-08 13:41:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nLet\u0027s wait others as well.\nI will miss the neutron-tempest-with-neutron-lib-master, but it\u0027s ok to move it to experimental, as the volume of patches that need current neutron-lib is low.","accounts_in_message":[],"_revision_number":1},{"id":"5b78a612a980541d4faa89fa9ea23107a22dce85","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-08 13:44:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"418212c579333ad991015ea2a640530b95dc4db0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-09 15:05:26.000000000","message":"Patch Set 1: Code-Review+1\n\nMakes sense","accounts_in_message":[],"_revision_number":1},{"id":"6f9bf0b4a71ccd360d083d02198f78b0137adb3a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-12 10:01:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28d36c1cbc565e4c0136ea011db76a5d8765014d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 10:01:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a41fb37eba0a499bf1ed6631f50741665fc7e121","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 12:19:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fa36786d3949484a86b15c36fd3079b9 : FAILURE in 23m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7368b07f178a49b0b345549c95184609 : SUCCESS in 2h 02m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8951e2914a624e6aaa2b6bfc8df0e9d2 : SUCCESS in 1h 29m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5386fd60b36043d6a2540fe60006e252 : SUCCESS in 2h 15m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ce826eb46eb4d1e81455c0362c37ae3 : SUCCESS in 1h 27m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ac5402035bb4963905c4896736a02a4 : SUCCESS in 25m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08efb8d215174d539b64598def13a3df : SUCCESS in 9m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a924d90de8144b9ba2d0c02bdbb7d1b : SUCCESS in 18m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50240eb333e74834920408596a4f4204 : SUCCESS in 17m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab7b878bd2ec4e80a85e910ed9b660a3 : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f707b90a7e94910ba4095dd4efaa27d : SUCCESS in 39m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/20fd0d5fdb004a68a4a6040e406274f0 : SUCCESS in 1h 03m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0634955ad7f54d3180edd34c187543a1 : SUCCESS in 1h 05m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a75b0a6005d24c909a9b6db93acc64cc : SUCCESS in 1h 14m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a11065a20b16442b99009bfff48e6acb : SUCCESS in 1h 05m 09s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ec47fbed09a04e0c9456b763137cd488 : SUCCESS in 1h 41m 23s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5e7360986d244f1b122d3d3e22221df : SUCCESS in 1h 08m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8dd7e423c24845a9b1fcffc4c65ed4cd : SUCCESS in 1h 19m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6a6f9cb927df4e199196dad5d75a2f22 : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1556cfc72a71f6ef09614d38fb62ab5a12281465","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 08:50:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bdc085b6e00723959faf4bbafe78b8be2006ce8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 09:56:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4048f5eac99c4bb2b010331ac08c4072 : SUCCESS in 57m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c151680b7d264ceabd605ce4eb525a0c : SUCCESS in 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63553af0a80fd55dbc21cc2e91b8c83da1e935c8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 10:52:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774061/1 : SUCCESS in 2h 01m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5ede2103bce1bde60291dc995c91fd102cace9b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 12:08:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6d2cc583f965448983a532b3282484bb : SUCCESS in 53m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/43e9e7dc3f574d9d9a2430506fd2c085 : SUCCESS in 1h 42m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0958b8c0bcae49f8bcd5b572ed0cd339 : SUCCESS in 1h 22m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04b67fb965364912877b48c2e54f404f : SUCCESS in 1h 06m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d9d0c67e12344594bff1ebc85beb3695 : SUCCESS in 2h 03m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/351e7f0f97ca41d7b77cc543e0b239c7 : SUCCESS in 25m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62c7eda2833d4e9c8761d5052fc0a291 : SUCCESS in 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84904e9475054179a53a62cfce0b8dbd : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f2705d595b74d70988fc100874713a2 : SUCCESS in 36m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6878a7e0adb848d9bb243b9c606a3d68 : SUCCESS in 27m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb08197ffba045cfa4f605e4cacb8f05 : SUCCESS in 34m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6565d766beca4d4f8489864651db14d2 : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9dcccf2b4cb74a1da5109f4353794305 : SUCCESS in 53m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/493827f6de534adda12e90510cbc8caf : SUCCESS in 1h 12m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4ae76ab09de24bfdb5cf1da1ddbf877c : SUCCESS in 1h 40m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b240e3b1df854e59ad67d3be52441963 : SUCCESS in 1h 14m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dbb9c43582934c52a8691ce106f1a40a : SUCCESS in 59m 23s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f47326124a30499ba37927ecea886d33 : FAILURE in 1h 08m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa38bcd87c48480193445aaebc349613 : FAILURE in 1h 21m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2f9d89fd4233400d8e6753c6a130d028 : SUCCESS in 2h 05m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d116c879ea4e4b16b8dd02476bc933ce : SUCCESS in 1h 04m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0dc821bf00ce4b12945621f953712d7d : SUCCESS in 1h 11m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2bbe1e7c4b354705bdac9923f60ab18e : SUCCESS in 55m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/85bde6336b9c4dbabe5874827539e3a3 : SUCCESS in 1h 47m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a8910b6d3c4246b29824f1e55b45f4c2 : SUCCESS in 1h 21m 33s","accounts_in_message":[],"_revision_number":1},{"id":"27d0a489fd2e104798d12d2f5a1427579c74e642","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-19 08:25:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5e4be9e254af93ff1a3c6f0897ae9d978660f1fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 09:23:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/970b88bed5e0433ea118436c1cb38c6f : SUCCESS in 57m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e84bef23d874555879958ec10a5577b : SUCCESS in 38m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef7d31ed55a651c4f1a12b3ecff16f22140b1a46","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 10:27:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774061/1 : SUCCESS in 1h 48m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7df8977d6f44d0fa5266059688d3380f741319b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 10:35:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/137735dccf1f45c6af8e399711dcaeb8 : SUCCESS in 50m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/14d362d0b6f446638d965eb53cff6b03 : SUCCESS in 1h 40m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5366f95a8cb14f29a9a51993c25b4363 : SUCCESS in 1h 39m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/864a7944716e4b6d826f281099122055 : SUCCESS in 1h 28m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/67024e7428e146088227d3a18afd28d5 : SUCCESS in 1h 42m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53b2ffe2e10b4374809b81558fa46aeb : SUCCESS in 38m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43574ddffc58431584e7fe0c8401d419 : SUCCESS in 31m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c19b83185fe4592b0497822291863c5 : SUCCESS in 13m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cccb83b6fced407489ebdf464017eb6d : SUCCESS in 32m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c8406cc9fd943e6a66259b98977e4f8 : SUCCESS in 30m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b069fd76c44d4f11920a5e597c1529a7 : SUCCESS in 28m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e9a8e9b04474003b2d35c9594228788 : SUCCESS in 10m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e99983a400f4aff83ade51c15fa56cb : SUCCESS in 41m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0c6e7d5a2e74cff894804ea0b100e40 : SUCCESS in 1h 06m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/913c227fbbe9442e9f9dbf07dcaf228f : SUCCESS in 1h 40m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f9142a1b439410bb35c0d53fb6a6590 : SUCCESS in 1h 15m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/809ca79acca44319a8e7e35fe8299964 : SUCCESS in 1h 28m 20s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ced4ff6beba246e186d64b9373dee131 : SUCCESS in 1h 21m 56s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8319ad3c4af14b448a2e6980202da46d : FAILURE in 1h 19m 20s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8cef22ddea14475a9d978a94df9265a : SUCCESS in 1h 39m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8630a6556c6443b6b0643b097cd6532d : SUCCESS in 59m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0800c7e3d890427bac067c967f296491 : SUCCESS in 1h 16m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d56567b066344d4bb0f46f542ae94a57 : SUCCESS in 34m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/49e46aa7d88642129efb31b1ed3560e7 : SUCCESS in 1h 31m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/179dc5d889c34c56a6cdb224cd1176e2 : SUCCESS in 1h 33m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b8718921e1dbcffaf8b315a1b8edef88fa602b29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 10:35:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30d655febb3d9c4e4ad96c154cbaadebca08371e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 12:29:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b55b4245cf344cc3bb698d369d568361 : SUCCESS in 46m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/346e4820e9924b28bda4c4ff2d2f5e5a : SUCCESS in 1h 27m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/75c174f7ca7d484184dd216f83acba36 : SUCCESS in 1h 47m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/796f3ee0b7854e2cbaf0e4a97edf4bb5 : SUCCESS in 1h 37m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1bcfe5fabf04d388fb36554434da345 : SUCCESS in 1h 23m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b318473c7c54bd59bb6a00335197897 : SUCCESS in 31m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf495674e741449495a1a1f7ff450e18 : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a625a1d253f401fbe18568bc1fa469f : SUCCESS in 30m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25c320586bd747ec9b95f91b7d280af5 : SUCCESS in 26m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9e80371c3f741e29c5116dc7f75c861 : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e2b53fe366544ab9b2ef65337614cc8 : SUCCESS in 45m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/29a183c656574ea48094627b7f30411a : SUCCESS in 1h 10m 42s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a768c9f93e9f4edbbfa7efe2fba5a7e6 : SUCCESS in 1h 16m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9db60e1732e44bd882c33bd14670bf70 : SUCCESS in 1h 03m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e1a265b9248d44049bd9680015153d81 : SUCCESS in 1h 13m 53s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/227a66401b984f889d0e4b69f1bf5131 : SUCCESS in 1h 39m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3579325c61e44a9faab942b2b5748176 : SUCCESS in 1h 06m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/498534c6c8d54fd0b4753dbd8d077794 : SUCCESS in 58m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4abae419412144eebe2fe10c951eba19 : SUCCESS in 32m 42s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2aa9dbfb1e43470a85f08203edca031c : SUCCESS in 1h 17m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6672e339a9e358c85b2491b1b61e5e281e50fc0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 12:29:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f246af6c97d06585e7c401e47d154b09dba51907","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 12:34:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acded43d03ae4522a549e401cabba20e : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc4e6ce1fe749ac9a6fb5b5ec55cb0ee90447cdd","revisions":{"fc4e6ce1fe749ac9a6fb5b5ec55cb0ee90447cdd":{"kind":"REWORK","_number":1,"created":"2021-02-04 10:37:01.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/61/774061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/774061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/774061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/774061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/774061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/774061/1"}}},"commit":{"parents":[{"commit":"72b79969667214b1bf7e21df30edb077ebbd1c08","subject":"Merge \"Add note for openflow firewall\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72b79969667214b1bf7e21df30edb077ebbd1c08"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-04 10:24:09.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-04 10:34:48.000000000","tz":60},"subject":"Move tempest job with neutron-lib master to the experimental queue","message":"Move tempest job with neutron-lib master to the experimental queue\n\nThe job neutron-tempest-with-neutron-lib-master is mostly to check if\ncurrent neutron changes will work fine with some changes done in\nneutron-lib.\nIt isn\u0027t really needed to run it on all patches and can be rune on\ndemand by running experimental queue if that is needed.\n\nThis patch also removes job neutron-ovn-tempest-ovs-master from the\nexperimental queue as we already have there similar multinode job.\nAs job neutron-ovn-tempest-ovs-master wasn\u0027t used anywhere, this patch\nalso removes its definition.\n\nFinally this patch improves ci_scenario_jobs document and adds there\ninfo about experimental queue and what is there currently.\n\nChange-Id: I0fc4ffd5db2673cefcd01307555d1a3900bb12e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc4e6ce1fe749ac9a6fb5b5ec55cb0ee90447cdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc4e6ce1fe749ac9a6fb5b5ec55cb0ee90447cdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
