)]}'
{"id":"openstack%2Fneutron~774134","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I5b5cf563f0c3d6ea352303f76323177cffc57c6e","project":"openstack/neutron","branch":"stable/queens","topic":"bug-1912651-stable/stein-stable/queens","hashtags":[],"change_id":"I5b5cf563f0c3d6ea352303f76323177cffc57c6e","subject":"Fix losses of ovs flows when ovs is restarted","status":"MERGED","created":"2021-02-05 12:12:18.000000000","updated":"2021-02-08 13:36:06.000000000","submitted":"2021-02-08 13:36:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774134-bug-1912651-stable/stein-stable/queens","cherry_pick_of_change":774132,"cherry_pick_of_patch_set":1,"meta_rev_id":"d89a65b41dced1cf5f29cb0590907cd06fce6c57","_number":774134,"virtual_id_number":774134,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 13:35:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-02-08 11:22:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-02-08 09:23:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-02-08 11:22:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 12:12:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2021-02-05 13:49:10.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-05 15:22:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-06 00:06:39.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-08 09:23:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-08 11:22:14.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"fa595deb23d992eff248630939d069a0e1269893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-05 12:12:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"24611ee2af4f9a2ea0a49ba41ef53a623b4db9fb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-05 13:49:10.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/774134/1 : FAILURE in 8m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ab55c327d97913128e7a3d4b1dacb853872d39dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 15:22:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/3d27780bf5af46e981f5cf8ef482a39e : SUCCESS in 42m 09s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/dbb71d7b9f4e49dcb50fe36246217751 : SUCCESS in 41m 43s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/51e120a1ed5c4050bd6addf43108c4c0 : SUCCESS in 53m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e6f180fd5c4e83ba1c9300668717de : SUCCESS in 11m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bdc3905f3e0e464e937e781b52237f72 : SUCCESS in 22m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/04573285c26947578424da8043934da6 : SUCCESS in 25m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0b21cbc2a9ce4a36a1b42d14a7a4b18a : SUCCESS in 8m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/00d6f559c10640f9b6b2fec60132e8b2 : SUCCESS in 1h 17m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9b0d8cf6c73b4e7094631ba3783cedbe : SUCCESS in 53m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be9573187c0648cdb4fa3077e5926155 : SUCCESS in 1h 18m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f7af1a3b8e44436da67c9a8a5c226edb : SUCCESS in 35m 54s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/1f91084be8fa4ff184321b7a018da0c5 : SUCCESS in 1h 56m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bd839763a732420883bc2745aeb83b42 : SUCCESS in 1h 21m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bd5e7a985b0d4bfdb1c8977868067298 : SUCCESS in 1h 08m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56f1974b099f4d7baea397511c134104 : SUCCESS in 29m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a59178949f184897c82a3018a79617f5f3167ea4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-06 00:06:39.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/34/774134/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/34/774134/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a4ce97cf5456aa21982cd675fd985adcdd4c65e0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-08 09:23:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"193541372cecff43f7940e350eb9352ffe4ed379","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-08 11:22:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"362e74c454da93e9b50ed206522d0395c4bc2236","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 11:23:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a80ee240d9c3b23244270a257b9f69ea2bbf6a29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:35:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/7d26d4ae9c6b4153a209615687d90729 : SUCCESS in 37m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aea44499a42a488fa862a30add055199 : SUCCESS in 11m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf96011d72ef4783af9e0e33bfe95170 : SUCCESS in 26m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2e5ef686c2a4423d98d061488f09ab39 : SUCCESS in 26m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9bfd944ca7634bbfb8e863526db81584 : SUCCESS in 6m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca83a02c1bfc404183583ebe37e561e3 : SUCCESS in 1h 38m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0d02b4fad1f548c8818559f03d528a52 : SUCCESS in 50m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34dfcea585de42268cda44981143fed1 : SUCCESS in 1h 28m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fa5da2613e8d4f26a07e158245f4fda0 : SUCCESS in 35m 45s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1a5c7059cfce48c7ab6d5970a0472c1e : SUCCESS in 1h 21m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0751ad5e782d41b4a3a210590dc1f7ab : SUCCESS in 1h 17m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e620d8499824f609a8cd90d57acbbc1 : SUCCESS in 30m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d89a65b41dced1cf5f29cb0590907cd06fce6c57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:36:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e52d3f554240b74ef528f1d6bdf173f9563393a","revisions":{"2e52d3f554240b74ef528f1d6bdf173f9563393a":{"kind":"REWORK","_number":1,"created":"2021-02-05 12:12:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/34/774134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/774134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/774134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/774134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/774134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/774134/1"}}},"commit":{"parents":[{"commit":"aed188c349794849105c034ffbabf3053d2fcb32","subject":"Improve DHCP agent\u0027s debug messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aed188c349794849105c034ffbabf3053d2fcb32"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2021-01-29 18:21:43.000000000","tz":-300},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-02-05 12:12:18.000000000","tz":0},"subject":"Fix losses of ovs flows when ovs is restarted","message":"Fix losses of ovs flows when ovs is restarted\n\nReinitialize conj_ip_manager when OVS is restarted.\n\nCloses-Bug: #1912651\nChange-Id: I5b5cf563f0c3d6ea352303f76323177cffc57c6e\n(cherry picked from commit 412160b97fc398f105c3f7386b928eeec2d9e60a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e52d3f554240b74ef528f1d6bdf173f9563393a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e52d3f554240b74ef528f1d6bdf173f9563393a"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
