)]}'
{"id":"openstack%2Fneutron~774153","triplet_id":"openstack%2Fneutron~master~I1ce783add19cc2ed81117a894e377a21e5ca25de","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I1ce783add19cc2ed81117a894e377a21e5ca25de","subject":"WIP: Add OVN support for stateless ACLs","status":"ABANDONED","created":"2021-02-04 20:10:03.000000000","updated":"2021-06-09 18:01:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a834d5f3c690ca2e1b408c70ecbe17fe19db260b","_number":774153,"virtual_id_number":774153,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-02-05 16:00:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-05 01:12:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 20:49:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-04 22:24:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-05 01:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 14:23:39.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 13:31:00.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"288ce252e3aa016907c906614b2503804055cede","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-04 20:10:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a008f2fc712f4ab4aa12491c58847fa49820e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-04 20:10:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c57e637e8c4548fe7fd4dd6c82d7be5d46117b09","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-04 20:10:57.000000000","message":"Topic bug/1901527 removed","accounts_in_message":[],"_revision_number":2},{"id":"5b74812725ea6f4f759145be1061e74eac4d890f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 20:49:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/205f0213d87d46459669ec9444a2703b : FAILURE in 34m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e2ed67d00fe4b699b2c17406faf5afc : FAILURE in 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72bb26ac5010e10d472455a004a859a807d1a15b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-04 22:24:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774153/2 : SUCCESS in 2h 12m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1532713dad3524c5a704cc0d3280a2f8a0a5988d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 01:12:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ac3a78b735a046e8b7d4eac84b41bf10 : SUCCESS in 48m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b655eb5220664a3ea0799eab35ae7918 : SUCCESS in 30m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4dc09d4206c345dbb12ecb98cfe6fbec : SUCCESS in 1h 38m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7ce2af11d7f045a585fa889d9a7cac07 : SUCCESS in 1h 44m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/041b4b9b41e3437389d482b70bdd2e97 : SUCCESS in 1h 14m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1de573fc8d934afba36ac83cea9a6269 : SUCCESS in 2h 04m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cf22d483cd44a7bbd9c08c666f5ee98 : FAILURE in 35m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11f7b961dddc45e3888a9f496b9d0b14 : FAILURE in 28m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c84fa7df53ac4a9385f245aa8b68ae66 : SUCCESS in 12m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a199b3f568544d094196fdd3ad0f624 : FAILURE in 27m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9883e08312614c2d98d429f79ee0d9c0 : FAILURE in 16m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08d8aa10c045472ba828589e267b5ecd : FAILURE in 17m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da987a76d5764abc9f886986be3702e2 : SUCCESS in 8m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4711d21f1fa246fa9b224b693c18ba6a : SUCCESS in 1h 16m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e51da29743324520b222234e619ac540 : SUCCESS in 1h 19m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a0b24a425b08490a8686b7b219fcc629 : FAILURE in 1h 47m 29s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7759d980628e4ae49e1f72a13c3f00bb : SUCCESS in 1h 00m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6b68332d0ca44270926afc5222ab7e69 : SUCCESS in 48m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/51e70086ac1a495caa118f74271acfd1 : SUCCESS in 1h 22m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/57657cc217bc49e5b4294bcd1aa5c38a : SUCCESS in 1h 43m 18s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fea7b1f01a0d4af7818dc7fa94314a21 : SUCCESS in 1h 43m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ebcd66e5d4a46c18a0d46eec04552fb : SUCCESS in 1h 05m 58s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/06a1c1d7ea4f4d4384dc52fe0ca175bb : SUCCESS in 1h 10m 12s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/029c1a6b932a440daf560139521d1c53 : SUCCESS in 1h 08m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6f7705beb97b43dc9e27adcad82c87ec : SUCCESS in 27m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b14ce057d9e7453db4f02bcf4bd35877 : SUCCESS in 1h 10m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7da815bb60118fa947e573c25149bb3b8a8552a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-05 14:23:39.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/53/774153/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"5e023b16258b2d82d163c5c52736e3b00c2d169e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-05 16:00:39.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/53/774153/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"1dd942fb4e799608aa86b73b5c04d99fd21baa4f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:31:01.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2},{"id":"e1b5452a5695171e039a639525090012523f3d97","tag":"autogenerated:gerrit:restore","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-05-13 14:38:39.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"1a79c8ab0c623a3efa59fd1266033796971b1d54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 15:26:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/493736691c4647f38510fcf02fbb2386 : FAILURE in 34m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b1b86f5d1834c57a2fdb279f0f4788b : FAILURE in 46m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7a6bd5c2dc58408465f37a7787bda88c4c8a4d1d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-13 16:39:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774153/2 : SUCCESS in 1h 58m 50s","accounts_in_message":[],"_revision_number":2},{"id":"33a93b67d067c96b46bbb8c1023c9d9b2e27309a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 16:57:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/096dfbaca9f64966a3f7c1abfd61b063 : FAILURE in 39m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/40309fbb16424ce3a671cef994aa76c4 : SUCCESS in 1h 46m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/06526dfa36034a5ab0c5c84570614c95 : SUCCESS in 1h 37m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b683a7af883a4aa699c8076c442fcf80 : SUCCESS in 1h 27m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/147035c60d664d71bdbb5a7f0647fc6d : SUCCESS in 2h 15m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8545a247f96d4d3dad65ee1742a475fd : FAILURE in 31m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61b52a5bf7e44a288769efac7bddaf4d : FAILURE in 30m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e90b9c32b104566aff90f9582888a4e : SUCCESS in 14m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1a75dcc3f2040e49a3701cd8b4011d5 : FAILURE in 25m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb2777e0d688445ebf3c235e457e614b : FAILURE in 28m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f21cca22c83546689c5808ea50640a93 : FAILURE in 24m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b868373108134c49a5914479c743b51b : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d439cf6e501d43d5ba0d9c4c7b8ba14d : SUCCESS in 50m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1113255908849c4ab75d33b50d54052 : SUCCESS in 1h 23m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/18964637c43040cfb1e446860b1e4a0c : FAILURE in 1h 34m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/075b464bc50b43a09c355fcaf409fa9b : SUCCESS in 54m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/962ee024ca014a1687fd338af4ae9602 : SUCCESS in 1h 01m 56s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d04aaf397a3b4247a6c5ff4286c9b27b : SUCCESS in 1h 26m 15s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/625ca358a98044f6aa4a5bb49fcd06cf : FAILURE in 29m 24s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5df81edcd6254798bcb7cef0e994476e : SUCCESS in 1h 29m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53144622c644454090cc3d287f401842 : SUCCESS in 1h 04m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ac65c004427f4a9e88157a29396bdcb1 : SUCCESS in 55m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ffaa2ff1a45041fd8639a08528952495 : SUCCESS in 30m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/de5dd7a4005c4e28b8d6e13d9eeb3e56 : SUCCESS in 1h 21m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d2db3af39fdd4cc2b375dad0d54b4cda : SUCCESS in 1h 20m 30s","accounts_in_message":[],"_revision_number":2},{"id":"a834d5f3c690ca2e1b408c70ecbe17fe19db260b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-06-09 18:01:12.000000000","message":"Abandoned\n\nAbandoning in favor of Ihar\u0027s patch https://review.opendev.org/c/openstack/neutron/+/789974","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a9f15d6c70803a3bde36b6402bd5e53d9612541","revisions":{"4f4f50085af5e9f40f3ecd8798c5976377eccbd4":{"kind":"REWORK","_number":1,"created":"2021-02-04 20:10:03.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/53/774153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/774153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/774153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/774153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/774153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/774153/1"}}},"commit":{"parents":[{"commit":"96eae957b06e84dce45159faf1008de9a69a26e0","subject":"[OVN] Allow schema checks to happen before starting connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96eae957b06e84dce45159faf1008de9a69a26e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-02-04 20:09:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-02-04 20:09:42.000000000","tz":0},"subject":"WIP: Add OVN support for stateless ACLs","message":"WIP: Add OVN support for stateless ACLs\n\nChange-Id: I1ce783add19cc2ed81117a894e377a21e5ca25de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f4f50085af5e9f40f3ecd8798c5976377eccbd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f4f50085af5e9f40f3ecd8798c5976377eccbd4"}]},"branch":"refs/heads/master"},"6a9f15d6c70803a3bde36b6402bd5e53d9612541":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-04 20:10:50.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/53/774153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/774153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/774153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/774153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/774153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/774153/2"}}},"commit":{"parents":[{"commit":"72b79969667214b1bf7e21df30edb077ebbd1c08","subject":"Merge \"Add note for openflow firewall\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72b79969667214b1bf7e21df30edb077ebbd1c08"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-02-04 20:09:42.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-02-04 20:10:50.000000000","tz":0},"subject":"WIP: Add OVN support for stateless ACLs","message":"WIP: Add OVN support for stateless ACLs\n\nChange-Id: I1ce783add19cc2ed81117a894e377a21e5ca25de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a9f15d6c70803a3bde36b6402bd5e53d9612541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a9f15d6c70803a3bde36b6402bd5e53d9612541"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
