)]}'
{"id":"openstack%2Fneutron~774375","triplet_id":"openstack%2Fneutron~master~If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252","project":"openstack/neutron","branch":"master","topic":"bug/1914842","hashtags":[],"change_id":"If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252","subject":"Make phynet paramter also is optional when network_segment_range enabled","status":"MERGED","created":"2021-02-07 10:04:13.000000000","updated":"2021-06-05 01:05:30.000000000","submitted":"2021-06-05 01:03:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":11,"has_review_started":true,"submission_id":"774375-bug/1914842","meta_rev_id":"157bb67e1a6993b1bbe43b4c7a19617f95dd0007","_number":774375,"virtual_id_number":774375,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-05 01:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-06-03 16:16:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-05-21 06:46:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-05 01:05:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-06-03 16:16:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-07 10:05:10.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-07 10:48:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-07 11:33:30.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-07 12:06:49.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-07 12:12:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 08:06:42.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-02-10 12:24:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-16 10:08:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-21 06:46:42.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"4e86cf8bc92cb53675de611470915f14d6a33598","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-02-07 10:04:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ed17b567d623f74539cf728b530bd3406da054c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-02-07 10:04:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"55dfa1fb50a1a5bc59b2ccfa4df6c6135cc853b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-07 10:05:10.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/75/774375/1/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"a4996a5f7e7ea37baf02017c9b471c4cb5433bb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 10:48:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20abff12fb624b17acba98a473cd2ceb : SUCCESS in 36m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13938e4e515443159ea945f0de79657b : SUCCESS in 38m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe7e31e5594ed222222e13681fe8b46dd4eba677","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-07 11:33:30.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/75/774375/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"53349776c9d1f28e386826739dda6d12f29987d5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-07 11:38:05.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/75/774375/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"cd71cb8b13eaf7c1be337cbb7590ebc62e515054","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-07 12:06:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774375/2 : SUCCESS in 2h 01m 16s","accounts_in_message":[],"_revision_number":2},{"id":"267b6afcd41250cec1cbda2e9f527c5b4fb15f41","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-07 12:09:13.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/75/774375/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/75/774375/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"834aab621628832e71fb6b0b6e466e0661d858c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 12:12:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fde840c04c0d4942b9678286effe2cd0 : SUCCESS in 35m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cecf48c246344f2cb60f6e31c98b8606 : FAILURE in 27m 22s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f7a7566dca794714a97dedfe58addf30 : SUCCESS in 1h 34m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/93c00b1f9a9247aca7b63cf1e348ed85 : SUCCESS in 1h 45m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2fdd3fed98d44a49c0033b7e5597a29 : SUCCESS in 1h 21m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/24cb826578f94557a3b57201e3d7baa3 : SUCCESS in 1h 35m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3547652998d64bad9b8ae6e4b6401acc : SUCCESS in 40m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c4d7820e2f94dce93fc70f574eabc9b : SUCCESS in 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6764d05af7e040e885323f87ff27349e : SUCCESS in 11m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf02e32f02142bbb020f86ab65b2d70 : SUCCESS in 33m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11b6bdf4e3314a158c4f81a4cf0b4dc7 : SUCCESS in 31m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d44327d79bd4199bd04c078c733c5e1 : SUCCESS in 24m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa2ad51de514e4480e7110cdd4d971f : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d9201a6fac14e6b9d8f44b9a0ad9ea5 : SUCCESS in 1h 05m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5cf65d15042742979eef05929a4027a0 : SUCCESS in 1h 02m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/09c1d7ab9f1c4ba49aaea87587779ab5 : FAILURE in 1h 38m 09s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ef3171882604cc780cf7c53efd5214a : SUCCESS in 1h 03m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8560ec1b8ec142d7a47c815e75d9ce66 : SUCCESS in 1h 07m 40s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ab5b88c86456468b9dba890d9095a6d1 : SUCCESS in 1h 30m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e2204efc75614d7fb72dd56ee7f04793 : FAILURE in 1h 14m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0815f50fa578431280844078c64782fb : SUCCESS in 1h 35m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47d10bda9d6e4c1d96f79b448347d5f1 : SUCCESS in 1h 02m 29s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e4dfeb8a35574fbcb37d287da1cd3f7e : SUCCESS in 1h 01m 20s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9ea66e40812343e4929937924d974227 : SUCCESS in 1h 11m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/58dac4e29d2c4f9d90fdc69372a68ce1 : SUCCESS in 35m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ffb405253d224842a43d4973e4f75b05 : FAILURE in 2h 04m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e244db8da485bb60ac2464be14283098af0a47ff","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-08 08:06:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nConsider adding a unit test for the fix","accounts_in_message":[],"_revision_number":2},{"id":"6a84886c36caf6bfd7db1ac15971b5ee3a6611ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-02-09 07:10:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f37065f10f518bbf5690ea87124f7391d5abe919","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-02-09 07:11:17.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"1087c6d21de57f872b16086391bfe189002c573c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 07:47:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f33653189bc04041ad3ef19c6570377b : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ac42ace913c4a33829828becc5b6367 : SUCCESS in 32m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"065190284612eae49f702cf48523c44814e97337","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-09 08:13:56.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/75/774375/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/75/774375/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"ed20d59ec68ee6ce0f08ff02d87653d99cba4fa3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-09 08:17:34.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/75/774375/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"fcfaacdca0a5b76b3d5cc153666550e852aef095","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-09 09:10:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774375/3 : SUCCESS in 1h 58m 48s","accounts_in_message":[],"_revision_number":3},{"id":"7fa2291c8d2124285915905eea9a6ba68066c5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 09:30:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/455176b497874b4db5d95131317125ea : SUCCESS in 1h 01m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eeb9b949e03644918b06d07e1d38ee0d : SUCCESS in 29m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d308eaa8b09d4dbd9478c098906c583e : SUCCESS in 2h 01m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/20dad42c22df4022a2b469078540deda : SUCCESS in 1h 35m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7cd5b809aa904f2cb83d0abeeef69b76 : SUCCESS in 1h 26m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4b34bfd7890e432486030e6e67085a4e : SUCCESS in 1h 48m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f18e7598e7d42e9aaf4f4efb9f7dd11 : SUCCESS in 33m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6fdef24ba244480bf464f4df0a2847e : SUCCESS in 28m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e01f461b762a49b79a59ed2503461f74 : SUCCESS in 12m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/564507b8663648589d7ccc74ef6ecb0f : SUCCESS in 21m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8c03ac77fa3425db9f580bead7c637b : SUCCESS in 25m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1dcfc1c917240f99ed238c13bfa7a45 : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b761f9e23944dd4ba46d5ba9f264754 : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d90c3c991cc745739fb1846dcace276f : TIMED_OUT in 2h 17m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a674fa2116c4a1dbbeeac81f2d8f3f5 : SUCCESS in 1h 21m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/31630116c57f45c2909bc5e7fd7ce5ec : FAILURE in 1h 38m 12s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3a4caa1cb5d489f8b911ada4b0a0e31 : SUCCESS in 56m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e95c6cd09cb34671929335595b308d23 : SUCCESS in 1h 29m 21s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/235f4cb88d604f57a2bf47535f3af376 : SUCCESS in 1h 23m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3dfa0848cb5840f5b8d141835c01eeb3 : SUCCESS in 1h 28m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b1a3a703bde446199c0fd72b66228a8 : SUCCESS in 1h 50m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/630c3003bfa041cea936fc03c3bbc9c3 : SUCCESS in 1h 06m 55s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/fe79aa9b3ed9468bba0e6f537712c19a : SUCCESS in 1h 08m 10s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f5cbd96d195540d48acc0f77f8bac543 : SUCCESS in 1h 06m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/04b1e05ce9e4419697a99c58f8751166 : SUCCESS in 39m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/823ff1cfab304a1dabf22d73e45f3c06 : FAILURE in 1h 37m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"87dca807b913ab2ccd847e7e72bf87a923555139","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-10 12:24:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nYou should also add a UT in NetworkSegmentRangeDbObjectTestCase, something similar to \"test_get_segments_shared\" but only for VLAN networks.\n\nThis test should create an environment with at least two VLAN segments, with a reduced set of VLAN IDs (could be just one). Then you should request allocations without the physical_network. In each allocation you should check a different physnet is provided. When the allocations are depleted, the method should return [].","accounts_in_message":[],"_revision_number":3},{"id":"993771c776b79ece1e977c4a33f7ba3f6e50ebc1","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-03-14 03:27:37.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e You should also add a UT in NetworkSegmentRangeDbObjectTestCase, something similar to \"test_get_segments_shared\" but only for VLAN networks.\n\u003e \n\u003e This test should create an environment with at least two VLAN segments, with a reduced set of VLAN IDs (could be just one). Then you should request allocations without the physical_network. In each allocation you should check a different physnet is provided. When the allocations are depleted, the method should return [].\n\nSorry for my late reply and thanks for rodolfo\u0027s comments. I will add related UT as soon as possible.","accounts_in_message":[],"_revision_number":3},{"id":"8430736ba83b4256df05996983041cc5a3e60860","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-03-14 03:28:10.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"e22cce8c4cacff01ee59ce99fed5335ab433a812","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-03-15 11:01:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c4b1071b79700b357aa3911d99aad08362cad5a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:54:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a739d30311547edabd4b76040e582d7 : SUCCESS in 47m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d252fbbd44cf49f794527c5fa894b795 : SUCCESS in 44m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3db579b9dc69c60e005d1ab33751cbd4edc40d4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 13:32:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0e08ec90650a48979af52958cd5dbd9e : SUCCESS in 51m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7df59d49297d4d3baf565203d64fed60 : SUCCESS in 2h 21m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d9312a49cc36438aaf6b426301854a9c : SUCCESS in 1h 42m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ced7f77ccd3e488c9105921449f99576 : SUCCESS in 1h 37m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/163fabde932e497eb3c1f2af90eacd96 : SUCCESS in 2h 24m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63b973fd2be440c5853dbfa8050bb55a : SUCCESS in 44m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8061c8249ca743d2a4a723d9e3380325 : SUCCESS in 39m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ddede3de73410fb10d164087c52637 : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86c11c3affe1423e9c3c1fcb1e4eb82e : SUCCESS in 42m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4649c58101e74d5fbb3990199cf3cc16 : SUCCESS in 35m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e1b4ca568594610a3ee9aada9a49c1f : SUCCESS in 34m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac914f49a8248c1aca7c64cc54dc388 : SUCCESS in 9m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0bd009fdb8724838a41b2f921cc0619d : SUCCESS in 54m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/501f25e007914b6cb6ba2b258a5cde10 : SUCCESS in 1h 32m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7fc8485947b24da891006ba08856158e : SUCCESS in 1h 34m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b0d9fb4970c4377b3a2476d5fa46d6b : SUCCESS in 1h 09m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1ae621f509b64f0b8246eb65939e1806 : SUCCESS in 1h 07m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bcf05b2213564c9e83e53f54694a3128 : SUCCESS in 1h 26m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4590bd96d35f417494a2278bd9002679 : SUCCESS in 1h 26m 30s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad49ef4e5a1f4e9a9bc031b7e51385ee : SUCCESS in 1h 35m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e06d8aa39fe34d68a4b6a928bd3e849b : SUCCESS in 1h 02m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/025808f0ed31445cb1e8b5522047f2cd : SUCCESS in 1h 11m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f093a4d67c864f4db8beaa31bb6fa2fb : SUCCESS in 34m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/62abfe91e6784b6c8bef48ae16193fdc : SUCCESS in 1h 31m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e1d82456db2e409fa76ce4ceb91c8b43 : SUCCESS in 1h 25m 06s","accounts_in_message":[],"_revision_number":4},{"id":"80fd45a424835f02c1c41d3316141293531a64db","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 17:46:27.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/75/774375/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"159caac40a25b5aee029a1c2519f2a4cbf23d64e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-16 10:08:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d117ee93b70ff1623ede7e75a244406bf23eed0f","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-03-19 07:18:24.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5da87471e26ea85d4f947a4232143c0dfd28d595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-03-19 07:29:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"679cc8e36b8a3275ee12186653d88a13f5a332d6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-19 07:53:31.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"be512f1890e71042eee71b5a5626bc7cf46b45ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 08:27:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8307ec872f8d463ab1d71e5c4eed95ae : FAILURE in 41m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8d42ae54cea47b18f7662eb208a0db9 : FAILURE in 53m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5840c08b1c5375f22852d2379309d334498c88ea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-19 08:53:37.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/75/774375/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"4eb12849a938fdd65baf971f5c02f5c04e058472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 09:47:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f5e01fd93c70434981ac155802b2c4ad : SUCCESS in 59m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a1f5b33277a543fe8dcff2b66009b15a : SUCCESS in 1h 25m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e0e3164f587943e282ea8c0ecbb3400f : SUCCESS in 1h 21m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1cd5b3f8b6f34a19bb7d4ed711644755 : SUCCESS in 1h 16m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0e69471756ae4d8092864a1ab9bb4957 : SUCCESS in 2h 12m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0cafd1d5063405db0be37db1029a72c : FAILURE in 41m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f6d92f573344bb397eded61f7ad61a7 : FAILURE in 37m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc85a03684224ca8a0ec9638be532a5c : SUCCESS in 11m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e363f18507fc41fd975527e99c56b759 : FAILURE in 32m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c5056f7c29244ada91dd54efcaeec28 : FAILURE in 28m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e570754775f4822a6fa3d8a897df838 : FAILURE in 33m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382c1e2ea4754283bbd9e734d5501e3e : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccd9134be86442d49644a2a307a38e13 : SUCCESS in 47m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/56a3851ac3a14777979285047d3902fe : SUCCESS in 1h 28m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2fd6b773546046ccb0921d78f55ba57f : SUCCESS in 1h 52m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b0044998b084b52895324fc5782cc8d : SUCCESS in 1h 00m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e05598dcfc8245d2af1271fe1b92d193 : SUCCESS in 1h 18m 57s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e9849d8449184563b902ab827a79e049 : SUCCESS in 1h 16m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/375dc12974814ad59662876b3791e67a : SUCCESS in 1h 24m 46s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e889d8489fa64566a322c2136be8582c : SUCCESS in 1h 22m 31s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6c6760b2ee54464a41b800818cf8cc1 : SUCCESS in 51m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a0adb16fe94a4bbc93b3fa26eb52f585 : SUCCESS in 1h 04m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f12da8f4531b4ec0bea1f48ae5be82a6 : SUCCESS in 30m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d20ea5ca4ff146f4a3ef99ffdfd6f073 : SUCCESS in 1h 30m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4f6226ae09904a2da8b7d1e358870773 : SUCCESS in 1h 22m 02s","accounts_in_message":[],"_revision_number":5},{"id":"f6a50b2fc9476a310bd00d5c1d05873a6f7b5878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-03-20 02:59:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"055cdf1d73ffc50aff03ea139c2ba15a0f9edd57","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-20 03:00:48.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"9caaaf41480e8138937e7c5099bb736d18bf3a50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 03:45:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/024948748dbb418c98b8f2f0404eaec5 : SUCCESS in 40m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b491e8aaec74a13802d5f91778eaeb2 : SUCCESS in 40m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"aafeceb1d9bb3ecd8976fc3441b2363a5b2511e8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-20 04:35:43.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/75/774375/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"aa3b7291baf4bf423efd20f5d40c70d6547927e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 05:07:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fa823dbef6bc42e684caf5f812d59524 : SUCCESS in 50m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5fb5de6ed9f348e29faa034fe3e475bf : SUCCESS in 2h 02m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3b32d853382e44a4bef63723d8003dc6 : SUCCESS in 1h 33m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f454732bbeec4c699e8e99f82698c55a : SUCCESS in 1h 53m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/25a9eb3a8aef4c31bc02075e7c4390b6 : FAILURE in 25m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a613bd600b5e4f22aa5bfe891d85ff50 : SUCCESS in 40m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89c3168d9c054d4797f53a23118d3cb4 : SUCCESS in 36m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f8247cea4a4fe9add2d7ccf11a59f8 : SUCCESS in 14m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2136925d95e94ac6ad49c1b593273edc : SUCCESS in 35m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9b662b2872042318aeb68bcd5bd424e : SUCCESS in 36m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d86ebb6d13549a1ac3c802ba7ee30ec : SUCCESS in 36m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d0455bfb434e3094fb36dbe2cd4690 : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfb431311c5946e692470826a975cd3c : SUCCESS in 51m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bacf09bf30046b0aae373ccbe2b7f22 : SUCCESS in 1h 28m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7f1bd1692355496e937c64d645cb4af4 : SUCCESS in 1h 29m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/57738a13a2384dc2b1c5ed8edbc835e0 : SUCCESS in 1h 09m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3045d29bd0734d5bbd34fac23711c572 : SUCCESS in 1h 08m 21s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/29fe65983ec54bf483f4a68532923b0f : SUCCESS in 1h 23m 35s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c15e8037795c41d6bb636c000b29f645 : SUCCESS in 1h 22m 45s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/06f9fcb14fea45bc97b8c9834d666dcd : SUCCESS in 1h 51m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/baa2cc5d86e943a0b1a62fe8516bd0ea : SUCCESS in 56m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9099fd94e933487380eb64c5fea0ad0e : SUCCESS in 1h 16m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/02b8e9961a534968b351691b0b39708a : SUCCESS in 29m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0937d98a426b4afaa077ea7f9657712b : SUCCESS in 1h 03m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6ff7027bd235402bb76d0adbe7b78aaa : SUCCESS in 1h 22m 10s","accounts_in_message":[],"_revision_number":6},{"id":"4354fcba845fc61e1e0c9d3524766c2c4951f7d5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-23 13:19:58.000000000","message":"Patch Set 6: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a5b461dd166cf0d3d700ac4ba8a280bb0c25d429","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 14:01:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e33838be90e41eea19a2e191c0642cf : SUCCESS in 40m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ee4eb9c655146dd81855ff4712fb514 : SUCCESS in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"49f438fac6918b4c339204eaed0ef907a9c2b92f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 15:34:08.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e9d294b18fc74ac78cf543d6635a76d6 : SUCCESS in 47m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/75c7b04ecb95488496b2d5776ff2b41e : POST_FAILURE in 2h 12m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/081f3368105e419481cd7066a8e9b24c : SUCCESS in 2h 03m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3cc79d4839024ac29ed30174349e6d86 : SUCCESS in 1h 21m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2fafa4e27e644661b85ca744ad38651a : SUCCESS in 2h 01m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae041762dbdb4289872de32160e156f8 : SUCCESS in 45m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d6c214b130f4713a75153f92e957e0a : SUCCESS in 27m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8ab64a853e34c5a83add2745068f1d2 : SUCCESS in 12m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15c98be0e8094205a5b853c7b6ad7356 : SUCCESS in 24m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1765acd8486444ea915d7e5354be8b95 : SUCCESS in 29m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4c54c02255240f79227c2ca45e7e96f : SUCCESS in 24m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da219ebf151b4eabb6c979825b05e706 : SUCCESS in 8m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6be31534eb7498d89c761b261e5d3b3 : FAILURE in 51m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c17b2065009b4ed880059ddb959423b3 : SUCCESS in 1h 31m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/168d6cbc1f7448f5904051793d26e6e7 : SUCCESS in 1h 38m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/44639999fbe843c794c1d627cb6fa552 : SUCCESS in 58m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ae65ed3b98c74c4d87f5fe9aebcffa90 : SUCCESS in 1h 05m 00s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/896af5fc50ff4cbaa80828f3e30f4528 : SUCCESS in 1h 36m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c865b42454da47fbb214513b23c2f9b5 : SUCCESS in 1h 55m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/498fe6c393b54bae9c121cb1a2eb997d : SUCCESS in 2h 06m 41s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1728e1f792cf4a9394e985b43d3cdc3f : SUCCESS in 56m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4342180094b74999a423b4d4dab853ea : SUCCESS in 1h 04m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7aa9cb4bf0a541948e7659688042da9a : SUCCESS in 34m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c3f131528a154226a7b3c1f10b77011a : SUCCESS in 1h 21m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bcfb7c7bd06a4479b6d71ae4b1b6a69e : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":6},{"id":"7a3250051cbec042aeeb72cb5061e7a7f8914c04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-09 02:34:37.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"0c68129247d85714923b7ff6b0284482431fbf64","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-09 02:49:08.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/774375/7 : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":7},{"id":"36b9367699c4d16ecb9c2b4e3e9e718d1c6999c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-09 03:18:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f21ad9ccee7b4a0390bdcad5c17d764d : SUCCESS in 31m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a3eb948ada64af181575ad49055eb05 : SUCCESS in 42m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dd20e81207266f450356fa7ffa405ab8a3189050","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-09 04:06:16.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/75/774375/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"915082733b44f0bef596705b40468f1545126f6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-09 04:36:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/30d8e124217143a6934513002e13742e : SUCCESS in 45m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6ef9e88df79b4afb93989103f04ae6fa : SUCCESS in 1h 16m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4d286ecf1a0d48ae9b4c1e448a64b9aa : FAILURE in 1h 33m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/14e4165f1aa04019bebc0ac97c65ef53 : SUCCESS in 1h 31m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/20f41b97fece4a5cb0fa76939a274340 : SUCCESS in 1h 43m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62102463e77746ce990a98035b69bae7 : SUCCESS in 34m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b10dff9e89b4de0be29ecc9c5c604c9 : SUCCESS in 29m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee77eeb9b3a14c7291717d4da827083f : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0083f280de014a3ab834fbf05130080a : SUCCESS in 27m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53ce25aee92143a0abd63383d145faf6 : SUCCESS in 25m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/824016eafd454d52bc67831448e387d3 : SUCCESS in 25m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b86f23753f42f98df2b230efafa996 : SUCCESS in 12m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c43f33fa3085423cac77963c433f3a2c : SUCCESS in 46m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5edc7937c6824f0d9a445a62d834beaf : FAILURE in 1h 11m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/be4e4a07ac1c426fb53479b199db5387 : SUCCESS in 1h 16m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8456d3ffb6f44f7a4c18b0114bd34d1 : SUCCESS in 57m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f2a86a59c5cd499b8e2fbd42e8137154 : SUCCESS in 1h 04m 40s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9baee5a7668d44e8aad045b23c6b3f2c : SUCCESS in 55m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/27d5815e916341b3b9e93d7ef7c948ae : FAILURE in 1h 03m 58s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6cf7ae509a7445a84b6350a0de065a2 : SUCCESS in 1h 44m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96b86a0fd31a424aae8a186e13a845fe : SUCCESS in 1h 04m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/95d38dfcda74424eab43abf850dcb0cf : SUCCESS in 1h 02m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a203a7c65c5042d295a102c1972b035b : SUCCESS in 25m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d4d15a7c81594a1a87aa54ac9e950bdc : SUCCESS in 1h 06m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/26b5602595fe4343b6c52481199d5bcb : SUCCESS in 1h 20m 22s","accounts_in_message":[],"_revision_number":7},{"id":"00ce1df042d2f529edf34b169ef7b9cafcd81e41","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-09 04:48:14.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"5420e11da1d2dd3688368f68ad6bd64c501e5ae1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-09 05:02:38.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/774375/7 : FAILURE in 13m 47s","accounts_in_message":[],"_revision_number":7},{"id":"99ab79480804ca375f41916d96d1d6054fd1de15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-09 05:41:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5c43563428043278d104d5ffa2006e5 : SUCCESS in 43m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d50aaf0b0176448489600c2f4e7d36a5 : SUCCESS in 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6ace77a515502fe72438b798f037e177e902d96f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-09 07:12:07.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/010c8d86fe6142cdac50cbf236d10441 : SUCCESS in 46m 55s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f0ed986efc474b2dbcf1d43edc94d2de : FAILURE in 1h 46m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ddacd1c9f6394cc281e6a1197f16ea17 : FAILURE in 2h 10m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00db1580916242fd8105299dd4f73b97 : SUCCESS in 1h 56m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/56214f003d8c4d99a7d22be892450a02 : SUCCESS in 2h 19m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/849e11df3fc84939b0d2ebe55b3f1a49 : SUCCESS in 38m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8750fef98abe4e2d864865fff52eba76 : SUCCESS in 30m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d3d0ea2da540dc92d24dc00f0f366d : SUCCESS in 13m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c8c00c7b1f74ed2b73b0af52a3de385 : SUCCESS in 31m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa2c179c636d449782879a28287b9dc4 : SUCCESS in 24m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4df55da64e4c4bcdbd4b57b126e23cf3 : SUCCESS in 29m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d32ef5e9b8c4d259e501b2bd587afb1 : SUCCESS in 11m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f92020c3a8244cd8958f96db2a988e3f : SUCCESS in 46m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/286413defd514cd197191b124e5935c6 : SUCCESS in 1h 22m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e829283a05704da6bb5e606b6f0fee9d : SUCCESS in 1h 19m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/06c3dbc1af6a4f748a76b0560405c10e : SUCCESS in 1h 03m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9dd975a174d14542ad195023ac452871 : SUCCESS in 1h 08m 27s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/05efa859b28d4d8e9cde727ff3d602d8 : SUCCESS in 1h 16m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/de9754f90888420fba8cecabeae01fa6 : FAILURE in 1h 09m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15bbf5ae721041e08ff0cb660c996bfb : SUCCESS in 1h 31m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84a10ffa6a7340ea963ed7697ff1886a : SUCCESS in 58m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2af481ed7e2c426ab830c4f845b80992 : SUCCESS in 1h 06m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7db91a87d8314f219c311e4535068d4d : SUCCESS in 32m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c0969376e86d4dd0af432335d3593dd8 : SUCCESS in 1h 22m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/23c23d23b9f14945ae577f2f415d439a : SUCCESS in 1h 19m 50s","accounts_in_message":[],"_revision_number":7},{"id":"3381e1901d07a34430d137a798562351eae4f7ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-13 06:09:36.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"22e8d4a87b9e509e335f0c88f5fa3ab51ad5f5da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 06:57:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db53f931dc594f6a822e1324825aa242 : SUCCESS in 42m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67a38dbc06a944f3b9a78241043e7862 : SUCCESS in 42m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"727e40aa7e9bf4d99953e376c88e897f04a69322","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-13 07:08:12.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f2c40b56030b270be2763c7d5b96dabd28f2a83e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-13 07:10:37.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/774375/8 : FAILURE in 59m 54s","accounts_in_message":[],"_revision_number":8},{"id":"887707f90ff95a03ec57437f4d0809cfc196af4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 09:09:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9cc303ed76c74c06a4b42d59904dcf31 : FAILURE in 45m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/27d703528ada4d99a9e2935da8f8563b : SUCCESS in 1h 32m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d92576595eb94d928f8307d5533b3e72 : SUCCESS in 1h 41m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b696f239f80744b9a86cebc74b7beae3 : SUCCESS in 1h 29m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5ce1d4f60c1746e5965e84f5b4464c10 : SUCCESS in 2h 20m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a69f5aceb314968bc88056b33f71ed4 : SUCCESS in 33m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf1225007fdb46fe857a27bc77c2c086 : SUCCESS in 30m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acb9e2767fac413fb2711ab430cd4723 : SUCCESS in 12m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7050b0ac0af40b3a4c57848787336c2 : SUCCESS in 23m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db269161b54c43d29c6b5166cad7b0b3 : SUCCESS in 20m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaf4eab0f534472ba53440343eedb873 : SUCCESS in 20m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a66655e8887439095b8e4d35bf7e9e0 : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d4286402de94f47a9b6e9f4fb56e505 : SUCCESS in 52m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e4b9ecda99044cd863b4ce09a787460 : SUCCESS in 1h 23m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/18d728f1a7ea4cd7996d25d643a69c66 : FAILURE in 1h 45m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fa078241f2f4401bdf1cdb59841eab9 : SUCCESS in 1h 06m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3b737d9dd785417581ed470c524bd99a : SUCCESS in 1h 09m 15s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6da8a389440e4ca08930f825fcdee3a0 : SUCCESS in 1h 16m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d7d30a17615e4e4895cff2abff77e48f : FAILURE in 25m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f43d8637ba04bf6a74cc2e006c651fc : SUCCESS in 1h 43m 49s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb523f95598a406da362f024392b1859 : FAILURE in 46m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4d1031acf16a4395a05d3a20e2b58cc4 : SUCCESS in 1h 05m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/525c03da7efe42739f445e23178ce6cc : SUCCESS in 32m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/97bb4a4167504db9a39e3b7ce25ba94a : SUCCESS in 1h 15m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ca7459f27c5248e0a897fc6537999b6e : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":8},{"id":"94b9b5d173de07fb198eb81f92263a390860736c","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-14 03:58:41.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"99d9fbeb89e4d20fd1c398e5a5cf755149118bd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:43:37.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7acd50a055a0466fa808457e85756dbe : SUCCESS in 43m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b013f88927a47a38f14d353c0067ae5 : SUCCESS in 43m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a72812602e1db47fcd154dab52a1aaf698bb1700","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-14 05:33:22.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774375/8 : SUCCESS in 1h 34m 06s","accounts_in_message":[],"_revision_number":8},{"id":"b058c7791b79ddfa65545e7e069a9648865024c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 05:50:51.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/65ed4c678f04497eaa03c4db5b79253c : FAILURE in 40m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/328efecec8da4fe4bec700f2975d9053 : SUCCESS in 1h 51m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b2a1b6f0c7a244ffa497287cdb082537 : SUCCESS in 1h 33m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a6c18df5e5a24e89a274b8dd6b38dcaa : SUCCESS in 1h 37m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7d7b7b12712e4a42800bd552f9206f74 : SUCCESS in 1h 44m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a8fa6791af449bc9b4670577c600344 : SUCCESS in 35m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d14dcffdc8d4e2484332a3dc31871c2 : SUCCESS in 29m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2c133f8a2249a2bd1440d480f8afcf : SUCCESS in 12m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/571bd10b9d6147e79e61bacfa54f03c2 : SUCCESS in 30m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ea86874eb224c66ab546f653d510666 : SUCCESS in 24m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79fdf7b2b19e4423a2fe2afd5baefe1d : SUCCESS in 27m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db6709633354391a6715b7f1e2f28a5 : SUCCESS in 8m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c721afc94e1e4def87367967ff0513a7 : FAILURE in 45m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8de47dcb17dc444bbd15eefed7c1cc20 : SUCCESS in 1h 14m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/320686df779148539c2fc0ae2de98fb4 : FAILURE in 1h 42m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0eaefeba4d6844589a33241b6a452275 : SUCCESS in 55m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6d551b5ba35f4062b73fce1023798530 : SUCCESS in 58m 36s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1ff5520e9f4f4e569585346f853edd3c : SUCCESS in 1h 25m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9cad3baf7d3b4dde915800e9604942e6 : FAILURE in 22m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ac2f32ab2144d57b3e73792faedeacb : SUCCESS in 1h 26m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/102c333f0b0744eebb3265d00cab7b5d : SUCCESS in 52m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d31f6209fd134d94bdf73d2b26e2699b : SUCCESS in 57m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c60af9b2711740a0a07942a01d3d96c2 : SUCCESS in 25m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ccdd50bb69494305b96d402e8d7aacb8 : SUCCESS in 1h 10m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7dc6d7079b4645a3be14a4062fce232b : SUCCESS in 1h 29m 10s","accounts_in_message":[],"_revision_number":8},{"id":"2ee8188536ecba3b6b33c32554bacdd4b3077d63","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-15 06:27:00.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"977cbc8a945143776fc7a22fff198da7fdc84798","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-15 06:42:32.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/774375/8 : FAILURE in 14m 48s","accounts_in_message":[],"_revision_number":8},{"id":"a7a8fb87c4e805f2bacc57308b641d441320aac1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-15 07:12:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99a13e2b540e4559af78bf39e83316ea : SUCCESS in 42m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5460857a53584e01a9725f622c994c02 : SUCCESS in 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6c1f97f007ed8737716a24d803094a0e27cef8e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-15 08:09:19.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/59d018bcc4ad4401a6a1b603babd5e40 : FAILURE in 25m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/30c7bffd238f42568b74bb73ecdee038 : FAILURE in 27m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c8a2f60dae5b4dac88c90524109528d3 : FAILURE in 26m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5be1b665567439aaa8632b49fdac608 : FAILURE in 28m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f9fe32d6fd424299af638a83f7d2a720 : FAILURE in 29m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22ec864ee68148eab90f1eaae74b161e : SUCCESS in 38m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/257b941e73e046b29eb90f6dfdf579b2 : SUCCESS in 32m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791273aabc934adf8f3e06ba3f239aef : SUCCESS in 14m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c3f656bfe2a49b6ae6339b57ca9ba6a : SUCCESS in 28m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9490f20294d14bdd9104deb5cb3d0062 : SUCCESS in 27m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c892b46fd79d40a686c1796cc401a2ad : SUCCESS in 27m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d82bd3bd621e4944aea7c28e9c4d2dbd : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/49f67ece6ef44a64b0738f16e7257efe : SUCCESS in 46m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d77239f16b7b45999f26c7172c0a425c : SUCCESS in 1h 21m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5017de69000d45019366665e40299584 : SUCCESS in 1h 25m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/099312115b33435d9898752175fee8a2 : FAILURE in 32m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3b25e6c3cf614e8688f47493d7ac2d8d : FAILURE in 32m 10s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/81620c124ff24e3889485b6021bbffea : FAILURE in 27m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/440442e83d854982a050f27c414f50f3 : FAILURE in 23m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/133966e4b8864f159d974140898b34c5 : FAILURE in 23m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45e08425e78c4ee6bc25e80d8676558d : FAILURE in 21m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b1935354f5324c7e9492c796b3c5e98d : FAILURE in 22m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/97b5e08f46cd4c96a9d30aa5fa145d7d : FAILURE in 29m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/31f8065b5bd54511b6931c96dc6d599a : SUCCESS in 1h 35m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/08e82b9579ba4a18ad8196259488e697 : FAILURE in 29m 55s","accounts_in_message":[],"_revision_number":8},{"id":"636055bb7f00f0f8ad8d21837389b3531e5cbb43","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-16 09:09:52.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"e10554f38518cbe1223a556f7bdbc5c082c1dcc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-16 09:10:59.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"5cb60e60caecdd8420c5f264fbed9a08b6699bc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-16 09:53:07.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d882bad476b44c609ee9b4683f37730f : SUCCESS in 36m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2609e0fde8824506bc92989868c40073 : SUCCESS in 35m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"644b01a2f74b26f4ad5d2956c610bf58638c1bbd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-16 10:45:12.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774375/9 : SUCCESS in 1h 33m 34s","accounts_in_message":[],"_revision_number":9},{"id":"c9769b9bb45c74c659106c7253c6d1b3cd8c6971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-16 11:35:01.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ffa143cc72614cad87264a215562a2da : SUCCESS in 55m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f58b889ab6b44f88bf5483dc2875f222 : SUCCESS in 1h 28m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/202f9f09487d47fbab4e1e34468f942c : SUCCESS in 1h 51m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ad637f8c0254417a91f02b69bdc8a491 : SUCCESS in 1h 42m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ae5fd0aa3f5d4a7cb643bef79a262be4 : SUCCESS in 2h 20m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3eca813dcfb4007b062060fac54cf68 : SUCCESS in 33m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa090793eb454c67b7fd5634aaf35b62 : SUCCESS in 41m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bc79297a1ea40c9a444421b05cfa604 : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/754fcbe8c35746d6a6aa1b2cbaec6156 : SUCCESS in 42m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6510a8c1e878408d8b4fd249ed6b47c6 : SUCCESS in 39m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7cc8443c5d24045b703cb740fae54be : SUCCESS in 38m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e24ac2061994a0cb4dcc0907f1cbe50 : SUCCESS in 7m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ef992f2d6744bf485322a9c02252648 : SUCCESS in 37m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/59b9a48bd83549ebbae13990752e7d8c : TIMED_OUT in 2h 14m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a5a9a191ce1143748e19d50a28d530d0 : SUCCESS in 1h 30m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ac53486cff642289e72de60ae746bb4 : SUCCESS in 1h 13m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/82489d07f5e34116876ef9487779977e : FAILURE in 1h 23m 38s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/20f09f91d2274bfaaedf2af2eed63a41 : SUCCESS in 1h 05m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/80537becf67e4ee9a00b59d9c4234d4f : FAILURE in 53m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d1758da606a54b27a7cc4d5f3e41fa77 : SUCCESS in 1h 08m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e6e15a582af453583c984760cd04f30 : SUCCESS in 40m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b89d19e7164442b0aea1e82e56c1774a : SUCCESS in 1h 14m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cad64029a41d42af99ff89918942263d : SUCCESS in 24m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4e0a47122e724b15aaf508ef993d919e : SUCCESS in 1h 33m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/83d7f670086b47e6be50cf1719a4a983 : SUCCESS in 1h 26m 22s","accounts_in_message":[],"_revision_number":9},{"id":"505bcd8e6b30eda08c5b7ce68fa6f362ec7a71a9","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-05-17 00:27:36.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"74f5a9fae57b810376578e88eb7e4419c5188ad5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 01:18:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35ea3bad2f2b4c73ac04c939427978c4 : SUCCESS in 48m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/404bb25c2f32423182485e0c71a5a70c : SUCCESS in 44m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f3cf06e16a308b6f4dae5b0d4e78452ba46f0640","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-17 02:25:07.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774375/9 : SUCCESS in 1h 56m 56s","accounts_in_message":[],"_revision_number":9},{"id":"5f557c740b81519218d993cb58d1550ce5a73858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 02:37:07.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cb427e2d201f49ecb558e71d7ee0d231 : SUCCESS in 38m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e57543ea49c44ab2b03bd69d2e189645 : SUCCESS in 1h 22m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/edfebfe5c257414ba32734b4080787b2 : SUCCESS in 2h 04m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a092a789d9bd43aebd81308be0f162c9 : SUCCESS in 1h 27m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f3c0d7af08834ea89c8a26af7580cccf : SUCCESS in 1h 54m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3473442964554d3da03723e8bcf6fa01 : SUCCESS in 36m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d934319c85e4b84b0e77ddf485157f1 : SUCCESS in 24m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29f48869912c486f8aa469220318431f : SUCCESS in 12m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37ba9db51bad4d68ba212d2d2f2fd139 : SUCCESS in 21m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa70b686eb014537a445fbf307b7a83e : SUCCESS in 22m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19f01ae5802344aab7dce71de75e0c8a : SUCCESS in 27m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8691b3771d04b7eba7f437a0c61804b : SUCCESS in 7m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5611444347654d758d3ffe14ce1cd9f1 : SUCCESS in 45m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8fe187445b2b4bf8bdbf90ec880e3de6 : SUCCESS in 1h 14m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/64f66127acc440e0998dc791e6e2f9e0 : SUCCESS in 1h 14m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/63ef16e1268747248f94352bc10aab2f : SUCCESS in 49m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/77a8e4af2a274a0fb7ae672e69cf9387 : SUCCESS in 53m 14s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f4237145258b4e2291943a5805cd2e5b : SUCCESS in 1h 00m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/539d7ad3198d43599a7d3301e8c33d9f : FAILURE in 59m 00s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e14b527e19140e189dece610c317c32 : SUCCESS in 1h 11m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/806dc84b6efc41e5b39c12b59011d2d4 : SUCCESS in 49m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ee5b47bbb8094b77b729c114c96406f1 : SUCCESS in 1h 00m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0ed6edf815484b2ba5064246fd8b26ba : SUCCESS in 22m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/36622d14798946bb837986359d45ebf0 : SUCCESS in 1h 21m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/469171a31afb4e2eb2a94a3f0fd2a5ba : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":9},{"id":"c2341e7118cdb972923275a70bf71de5f157ef38","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 06:43:35.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/75/774375/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"516b024e9d9edcdb72d7c26f335e220d6030c1a8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 09:11:56.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/75/774375/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"cae9efb69f7cbb0db516e0a5accc7433b7fce038","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-05-21 06:46:42.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"dbad894e4a2b26d4572eaf19c3c3ca5121ce9d4b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-03 16:16:21.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nI don\u0027t know why my -1 was still there... it was for PS3","accounts_in_message":[],"_revision_number":9},{"id":"80c5ea117b9866110d5699b09886ae738b3159f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 16:16:56.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"df44ae52d110b0b4763060bde4df0913574c43b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 20:06:41.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0016f2c751174906bdca82262a0a46c4 : SUCCESS in 40m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/49f836e525e14eed84b819b5c4319ba0 : SUCCESS in 1h 50m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b6f4e23cbcd940a09ebd00ef5b4a62ed : SUCCESS in 1h 56m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/29bb5b129deb4c63b287f845624657cb : SUCCESS in 2h 06m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aef8c575aef9418eb74bd4c68fbd7c2f : SUCCESS in 1h 42m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/515c50df3039433bb4e6569249020969 : SUCCESS in 33m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/387fa44494774ad28c4081a82ae424b4 : FAILURE in 16m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c811cd1a355442178354bc4a3fc79f3d : SUCCESS in 29m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7a3717334ad4cc893bc91d7aca819a8 : SUCCESS in 41m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b3bd86fcf841a3a01389ac6fb6ebd3 : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1066fd756d52473fb86c01367e6abe22 : SUCCESS in 42m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/13ffa02cba3a4f129d680c392040dd21 : SUCCESS in 1h 26m 42s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/eda4f26b63f245f49e986d1f8e941aef : SUCCESS in 1h 18m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b731ed8aa6f40d2bb298cf67a3f66b9 : SUCCESS in 1h 00m 54s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4320779fb97458b97d06be7421243a3 : SUCCESS in 1h 09m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7fa6338144dd496b9de73b98ed2e39c9 : SUCCESS in 1h 06m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ba4a591864a947e3bd469496217c8210 : SUCCESS in 29m 13s\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":9},{"id":"7e56511cda7965f87417e07d2bd5c69a3ffb89ec","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-06-04 05:57:02.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"ddfd687830b276c7949bbcba2c36b2bd232a5c8c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-04 06:11:31.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/774375/9 : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":9},{"id":"4b8d8562506975faeb93e45f3a2c8412f755da22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 06:42:21.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/949949e5a7d44a53b6f8c77d1e32eb6f : SUCCESS in 43m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a729eca2c7064980a90daa26f61f8dae : SUCCESS in 38m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cfe96884b8f2851608db4ebd89b617c2f5727e81","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-04 07:37:15.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9: Code-Review+2 Workflow+1\n\u003e \n\u003e I don\u0027t know why my -1 was still there... it was for PS3\n\nRodolfo, you also left comments on PS8, which were not addressed/answered","accounts_in_message":[],"_revision_number":9},{"id":"465be834a878cb542e0d5360089d353721ecfc5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 08:04:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8aa20aba957e4a558f3f3fd530a96339 : SUCCESS in 42m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fbd8a3c9bc984512b9c66e9cb7b8245d : SUCCESS in 2h 05m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7b42b68647b54aa49c1859c021190af1 : SUCCESS in 1h 32m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a4f4d6643dd64b7eb663ea247e5c61e7 : SUCCESS in 1h 41m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c73e2c1c0f054a0997a4cb9cfc37443b : SUCCESS in 1h 48m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbb7e32a712b4de0a2b47246cca0743e : SUCCESS in 32m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f07e4fa22f84c2c9dcda3c17f27b8ca : SUCCESS in 28m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae60e0e164d442ef822e66914c0a1f0b : FAILURE in 15m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b5b0fa160bb4af0a6d5a524ea5a08c2 : SUCCESS in 26m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4483c46861c43be9911747bb47095e7 : SUCCESS in 23m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93b8f8d779d548fe9afedfbc1bb4dd7e : SUCCESS in 23m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d0c6994550d4976bc112562634eab78 : SUCCESS in 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d51e3f7c6a264e79b9a1e2a42cbafe6b : SUCCESS in 46m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/92f2ab32a8e24c0588344da253178c38 : FAILURE in 1h 01m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/db81387e9c6d40e4a889b185a0e587bb : SUCCESS in 1h 11m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5e5f622aaa24e1cb53b0291a16be1e7 : SUCCESS in 1h 03m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/29c0176a1515453bbc8f8c8b694f2aa0 : SUCCESS in 59m 11s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4db20c8fa4cd46d1897a6da8f90ade10 : SUCCESS in 1h 28m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/22f07ce1d6dc4330a5f746368af52452 : FAILURE in 24m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d99f90ada59549209d5d47a05cdf16d5 : SUCCESS in 1h 41m 10s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f572dceab1b4c80b8a49709f6fd93e1 : SUCCESS in 1h 02m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/81e49350f94e42ec8fcda6049f2c6f14 : SUCCESS in 1h 00m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/da012ceaa9ab4275be29c3ebb3dab38a : SUCCESS in 42m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4c355ecc8a6d4af79b0c21d1d41355da : SUCCESS in 1h 16m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b29930d5db23413bb223253fb6323d09 : FAILURE in 1h 19m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":9},{"id":"7ed88e34b86aa15cfc96c86c71dd9ee304097f1c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 23:01:50.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"8558c11949603c8d58a1ddc3941336768ea568ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:04:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f56df1a67464300b49c8d07a911b732 : SUCCESS in 31m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30926da0af424541879921a88bf59ea2 : SUCCESS in 40m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c54c15ef6f845bf804919bde883e376 : SUCCESS in 11m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6e715aee02d4b1fb1d7547ae3fee8e0 : SUCCESS in 30m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7d502a87c0b4a55b2a150439a6623f9 : SUCCESS in 24m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2671eeacef242ccb8a374c702f48afc : SUCCESS in 36m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3e8bd9bf9e4c92afd43b5cdad98603 : SUCCESS in 10m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d27d47f916e4371895eb10423bc6f9f : SUCCESS in 58m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9b08596d53d04dc598c93331a9ec6442 : SUCCESS in 59m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1c525500826f9d6fe844c20d39eb3ae3017947d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:04:37.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d6d2c14598ce2751ed53e85dd7cb7d74f258b9ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:04:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/16dec6a1041f4df08acddb0ba00ffad8 : SUCCESS in 48m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ffb3c2ffd7f4c5bb4796eb88e32c4c4 : SUCCESS in 52m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"effa7a5d072ccf1faeb7a90dfa54fdf3187b837c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-05 00:55:24.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774375/9 : SUCCESS in 1h 45m 35s","accounts_in_message":[],"_revision_number":9},{"id":"54f6f0168ebff59c6f786edfead13738a3537016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 01:03:44.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dba140ac7ef5426898e574bb018147dc : SUCCESS in 28m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c12893098614d1baa6b8f9fd42c37ee : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dcbf422385e44d3afc3c28b7163e43c : SUCCESS in 31m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/813dae03987c41869e1220ae5c7e797c : SUCCESS in 24m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a0b0006cd5467095093e6e302b61ac : SUCCESS in 8m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea4d57005a6d42ba9c89bf13cf42d204 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":9},{"id":"76cb16c5372412382a77d85f203063a322bc9b4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 01:03:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"a47c4f69ac64f5eb3c80e9ab1d4841590ea0a7ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 01:05:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e173d3e5fdfe4e8d91815b01f53fda73 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"fa7c7282f019d3e0dd4d492f5ff2c47bd45b0ce7","revisions":{"324c7289041f895dcc21bb91a656e12260562aa8":{"kind":"REWORK","_number":1,"created":"2021-02-07 10:04:13.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/75/774375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/774375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/774375/1"}}},"commit":{"parents":[{"commit":"f9c9d102aa2264c59ebe421146fef5779e6ceadd","subject":"Merge \"Process DHCP events in order if related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9c9d102aa2264c59ebe421146fef5779e6ceadd"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 09:59:07.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 09:59:52.000000000","tz":0},"subject":"Fix bug 1914842","message":"Fix bug 1914842\n\nCloses-Bug: #1914842\nChange-Id: If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/324c7289041f895dcc21bb91a656e12260562aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/324c7289041f895dcc21bb91a656e12260562aa8"}]},"branch":"refs/heads/master"},"1a397f7ad19b1abc78e3de87d00da7f6b3dc53a5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-07 10:04:47.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/75/774375/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/774375/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/774375/2"}}},"commit":{"parents":[{"commit":"f9c9d102aa2264c59ebe421146fef5779e6ceadd","subject":"Merge \"Process DHCP events in order if related\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9c9d102aa2264c59ebe421146fef5779e6ceadd"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 09:59:07.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 10:04:24.000000000","tz":0},"subject":"Fix bug #1914842","message":"Fix bug #1914842\n\nCloses-Bug: #1914842\nChange-Id: If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a397f7ad19b1abc78e3de87d00da7f6b3dc53a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a397f7ad19b1abc78e3de87d00da7f6b3dc53a5"}]},"branch":"refs/heads/master"},"08f3570c5ccdca464ccee6412c07410cdf372041":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2021-02-09 07:10:26.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/75/774375/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/774375/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/774375/3"}}},"commit":{"parents":[{"commit":"c4ac136c82722271b0df54454a7d58e0e9c5fa0f","subject":"Merge \"Stop metadata proxy gracefully\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4ac136c82722271b0df54454a7d58e0e9c5fa0f"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 09:59:07.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-09 05:47:51.000000000","tz":480},"subject":"Randomly select one when physnet not provided with network segment range","message":"Randomly select one when physnet not provided with network segment range\n\nCloses-Bug: #1914842\nChange-Id: If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08f3570c5ccdca464ccee6412c07410cdf372041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08f3570c5ccdca464ccee6412c07410cdf372041"}]},"branch":"refs/heads/master"},"5ba8cf97c2c80455b760d8d0ef937416fe13a032":{"kind":"REWORK","_number":4,"created":"2021-03-15 11:01:03.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/75/774375/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/774375/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/774375/4"}}},"commit":{"parents":[{"commit":"5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3","subject":"Merge \"Implement secure RBAC for the l3 conntrack helper API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f6fbcb155e19d0a46cef0e7bbfee553f0472ef3"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 09:59:07.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-03-15 10:50:19.000000000","tz":0},"subject":"Make phynet paramter also is optional when network_segment_range enabled","message":"Make phynet paramter also is optional when network_segment_range enabled\n\nWhen network_segment_range plugin not enabled. we can create a provider\nnetwork without physical_network parameter. In order to compatibility,\nWe shoule support don\u0027t provide physical_network parameter when create\nprovider network with network_segment_range plugin enabled.\n\nCloses-Bug: #1914842\nChange-Id: If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ba8cf97c2c80455b760d8d0ef937416fe13a032"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ba8cf97c2c80455b760d8d0ef937416fe13a032"}]},"branch":"refs/heads/master"},"4950ef426b64f6bc355258f44dd4bd618eba51cf":{"kind":"REWORK","_number":5,"created":"2021-03-19 07:29:40.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/75/774375/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/774375/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/774375/5"}}},"commit":{"parents":[{"commit":"cfe16da2aaab210cebc9a02c9beeb3843f01d8f6","subject":"Merge \"Fix VLAN driver tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfe16da2aaab210cebc9a02c9beeb3843f01d8f6"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 09:59:07.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-03-19 07:28:01.000000000","tz":0},"subject":"Make phynet paramter also is optional when network_segment_range enabled","message":"Make phynet paramter also is optional when network_segment_range enabled\n\nWhen network_segment_range plugin is not enabled, we can create a provider\nnetwork without physical_network parameter. In order to compatibility,\nWe should support to not provide physical_network parameter when create\nprovider network with network_segment_range plugin enabled.\n\nCloses-Bug: #1914842\nChange-Id: If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4950ef426b64f6bc355258f44dd4bd618eba51cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4950ef426b64f6bc355258f44dd4bd618eba51cf"}]},"branch":"refs/heads/master"},"1caeb1104502d7644749296832219e000840cdcf":{"kind":"REWORK","_number":6,"created":"2021-03-20 02:59:34.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/75/774375/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/774375/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/774375/6"}}},"commit":{"parents":[{"commit":"cfe16da2aaab210cebc9a02c9beeb3843f01d8f6","subject":"Merge \"Fix VLAN driver tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfe16da2aaab210cebc9a02c9beeb3843f01d8f6"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 09:59:07.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-03-20 02:59:09.000000000","tz":0},"subject":"Make phynet paramter also is optional when network_segment_range enabled","message":"Make phynet paramter also is optional when network_segment_range enabled\n\nWhen network_segment_range plugin is not enabled, we can create a provider\nnetwork without physical_network parameter. In order to compatibility,\nWe should support to not provide physical_network parameter when create\nprovider network with network_segment_range plugin enabled.\n\nCloses-Bug: #1914842\nChange-Id: If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1caeb1104502d7644749296832219e000840cdcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1caeb1104502d7644749296832219e000840cdcf"}]},"branch":"refs/heads/master"},"0565749b7e96aa132ac6a97166a8effc0b530035":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-05-09 02:34:37.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/75/774375/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/774375/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/774375/7"}}},"commit":{"parents":[{"commit":"9241c76b04e6745cc648ee42037cfe6ddad3600a","subject":"Merge \"Add enable_dhcp filter in get_network_info rpc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9241c76b04e6745cc648ee42037cfe6ddad3600a"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 09:59:07.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-09 02:34:37.000000000","tz":0},"subject":"Make phynet paramter also is optional when network_segment_range enabled","message":"Make phynet paramter also is optional when network_segment_range enabled\n\nWhen network_segment_range plugin is not enabled, we can create a provider\nnetwork without physical_network parameter. In order to compatibility,\nWe should support to not provide physical_network parameter when create\nprovider network with network_segment_range plugin enabled.\n\nCloses-Bug: #1914842\nChange-Id: If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0565749b7e96aa132ac6a97166a8effc0b530035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0565749b7e96aa132ac6a97166a8effc0b530035"}]},"branch":"refs/heads/master","description":"Rebase"},"a4457190abcfc9fcff5e7646a7fdaed09ceeeddd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-05-13 06:09:36.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/75/774375/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/774375/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/774375/8"}}},"commit":{"parents":[{"commit":"2c192b78a2017da761e351321b996231ed641893","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c192b78a2017da761e351321b996231ed641893"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 09:59:07.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-13 06:09:36.000000000","tz":0},"subject":"Make phynet paramter also is optional when network_segment_range enabled","message":"Make phynet paramter also is optional when network_segment_range enabled\n\nWhen network_segment_range plugin is not enabled, we can create a provider\nnetwork without physical_network parameter. In order to compatibility,\nWe should support to not provide physical_network parameter when create\nprovider network with network_segment_range plugin enabled.\n\nCloses-Bug: #1914842\nChange-Id: If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4457190abcfc9fcff5e7646a7fdaed09ceeeddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4457190abcfc9fcff5e7646a7fdaed09ceeeddd"}]},"branch":"refs/heads/master","description":"Rebase"},"fa7c7282f019d3e0dd4d492f5ff2c47bd45b0ce7":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-05-16 09:10:59.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/75/774375/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/774375/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/774375/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/774375/9"}}},"commit":{"parents":[{"commit":"3b3398b8c5c0d597e0995ab7fdccef6d64d456cf","subject":"Merge \"use payloads for NETWORK callback events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b3398b8c5c0d597e0995ab7fdccef6d64d456cf"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-07 09:59:07.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-05-16 09:10:59.000000000","tz":0},"subject":"Make phynet paramter also is optional when network_segment_range enabled","message":"Make phynet paramter also is optional when network_segment_range enabled\n\nWhen network_segment_range plugin is not enabled, we can create a provider\nnetwork without physical_network parameter. In order to compatibility,\nWe should support to not provide physical_network parameter when create\nprovider network with network_segment_range plugin enabled.\n\nCloses-Bug: #1914842\nChange-Id: If19a6a9b0f2910baefba9d7ea9955b1f0ad7d252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa7c7282f019d3e0dd4d492f5ff2c47bd45b0ce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa7c7282f019d3e0dd4d492f5ff2c47bd45b0ce7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
