)]}'
{"id":"openstack%2Fneutron~774487","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Ib1a95dd41d25f39f3378a6728d752a6589b9b61c","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1912596-stable/victoria","hashtags":[],"change_id":"Ib1a95dd41d25f39f3378a6728d752a6589b9b61c","subject":"Fix incorrect exception catch when update floating ip port forwarding","status":"MERGED","created":"2021-02-09 04:52:41.000000000","updated":"2021-02-09 16:16:56.000000000","submitted":"2021-02-09 16:13:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774487-bug/1912596-stable/victoria","cherry_pick_of_change":774325,"cherry_pick_of_patch_set":2,"meta_rev_id":"4f7954ecfc14092eef5457ddff440953fb673ead","_number":774487,"virtual_id_number":774487,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 16:13:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-02-09 09:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-02-09 11:47:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-02-09 11:47:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-09 06:41:27.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-09 08:23:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-09 09:28:08.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-02-09 11:47:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"6d7a800ef2595b889d3ceaa52c373c787737bf88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-02-09 04:52:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2615219528e28598374058b08aa46ad323de6702","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-09 06:41:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774487/1 : SUCCESS in 1h 47m 41s","accounts_in_message":[],"_revision_number":1},{"id":"784a04c20bafff0f5ff83ed7924f7efdc582ce49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 08:23:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/d10b9e700d5d4d1ea1667bd00a4a9937 : SUCCESS in 52m 50s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/2a4d8d9a5d7e44279188fe9e942c515e : SUCCESS in 36m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/5640dd7266c84dcea16c66a3f34d6a2b : SUCCESS in 1h 43m 27s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/88a17a743b0f4bf7bf8c03129373900f : SUCCESS in 1h 55m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/224d840867784b2d817d5315f6049c58 : SUCCESS in 1h 34m 56s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/559bc66ed4504e72a731d04794d8902a : SUCCESS in 1h 36m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/723de129ee4747cd833649efc3552dd0 : SUCCESS in 34m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a284ac03c6a4931aec8a1aee98fe37c : SUCCESS in 13m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/217167d998eb4089a5df0586d50459e3 : SUCCESS in 26m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2650b7ccf514bd8b96df00bc4e8881f : SUCCESS in 28m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc4502db6994ccba1886b3da7f7b8d4 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d56006abde7c4511beeca0db7642eac5 : SUCCESS in 1h 07m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/585ab1589d1f486eb433f51275afce6c : SUCCESS in 1h 16m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a3a76cfad7a840cfba65c0fb3f2b31aa : SUCCESS in 1h 44m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38eb07b4ad3a491483c81607c749d301 : SUCCESS in 1h 02m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0864d04a74f24a42b004a6fae7bf3253 : SUCCESS in 1h 12m 45s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/d7b34ba39d6b45d08155a6aeeb9c000c : SUCCESS in 1h 08m 00s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/866c4cf8e3c44f9bbba1d62c854cf8d0 : SUCCESS in 1h 21m 45s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fdb0f07a105444b386f8f53badd029bd : SUCCESS in 1h 00m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/de6e55b2b0024719b11c9b72808f3be2 : SUCCESS in 1h 36m 15s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a9afafe66a564bc49fc3ed5cc0575e32 : FAILURE in 1h 46m 50s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7110098c9bbb4991a175921ff0b2eddf : SUCCESS in 1h 19m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/27612208d03343e89884d0153f87c898 : SUCCESS in 1h 16m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a27bd1707e1b4bb09080efb2ce57e014 : SUCCESS in 2h 07m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37c1fee71a3842a7ac36756852b6784f : SUCCESS in 56m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f3db23eaa7b64bb19de522428910c3a9 : SUCCESS in 1h 16m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/af5ee3a5b6e14003bd823f4794becebb : FAILURE in 41m 39s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/97096015ef324c388d6c330d0b64a77f : SUCCESS in 1h 26m 25s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/09655a00ceb04e218739bfa8a119da1a : SUCCESS in 3h 24m 44s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4cd3958efb114fc7a405ff5834d7f785 : SUCCESS in 1h 21m 16s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c96c252ac03404d9db8570c10c30f9e : SUCCESS in 1h 39m 45s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ac61379121de4d969c0cedcd221f6a6d : SUCCESS in 1h 59m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/51a38eb8bc534020a42f123dbd7aca15 : FAILURE in 35m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"932db7e2366e36bdbca6c7a000d7fd6b8ad663a2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-09 09:28:08.000000000","message":"Patch Set 1: Code-Review+2\n\nClean small backport, nicer error message for linked issue","accounts_in_message":[],"_revision_number":1},{"id":"6236e0db29ef7ca3fb6b869bdfcbf6e972e7172f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-09 11:47:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c78463dfbe53f6c854b4ec0daeba79467563ebf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 11:47:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e2eb4eacf704128f280c49f4a6d8441a967f420","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 16:13:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/49e7017329824f46bb805ae6281e3adf : SUCCESS in 48m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a56d1d914a424bbfdf0226afe0175f : SUCCESS in 12m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c55b5f8b607c4bd9944a0fc032cef75a : SUCCESS in 31m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ab7d32442734744a9e6efd962d58b18 : SUCCESS in 29m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4385ff5a10b1494b8140470ec5623ece : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1af1583d20b545eebe256d10ba84b81d : SUCCESS in 1h 11m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e2e5a8470bd486cb35110c96c32f7dc : SUCCESS in 1h 18m 16s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0139ecf5bf364dfb84e107a0e480149c : SUCCESS in 1h 14m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/71f7bcbb2f1841c9bbe7a1cc2428456a : SUCCESS in 1h 11m 05s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3ef46b18cef342079b714b524595ab45 : SUCCESS in 1h 29m 25s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/40b0b6aa15034ca0bbc66555f72dcf09 : SUCCESS in 1h 53m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/541dfeb64cc54fd08a10ade1fba1f4ab : SUCCESS in 1h 01m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/250058cf26f14028b574d537f94f1456 : SUCCESS in 1h 15m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0e50abe85e74c95a22c56d832103bd6 : SUCCESS in 2h 14m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/625a0eeb075e4cf984d440a860a97838 : SUCCESS in 58m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/18f52f53cb494c56a06bdf548f331cbc : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d3306cceb8ef2b51fdfb782a0fc1b31cfe5fd138","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 16:13:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4f7954ecfc14092eef5457ddff440953fb673ead","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 16:16:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7dd282ee7b4332b8151b9a1c33cc12 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2bad8cc18247d39403ddb826b4ff6ae7808cc11","revisions":{"d2bad8cc18247d39403ddb826b4ff6ae7808cc11":{"kind":"REWORK","_number":1,"created":"2021-02-09 04:52:41.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/87/774487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/774487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/774487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/774487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/774487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/774487/1"}}},"commit":{"parents":[{"commit":"db935d7d31fc30d01d827547ea0b060fa6398b8f","subject":"Merge \"[stable only] Wait chassis creation before retrieving agent\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db935d7d31fc30d01d827547ea0b060fa6398b8f"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-06 02:13:04.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-09 04:52:41.000000000","tz":0},"subject":"Fix incorrect exception catch when update floating ip port forwarding","message":"Fix incorrect exception catch when update floating ip port forwarding\n\nCloses-Bug: #1912596\nChange-Id: Ib1a95dd41d25f39f3378a6728d752a6589b9b61c\n(cherry picked from commit e9e4395d578e40bb59272b409c7ca3617ec1e6e3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d2bad8cc18247d39403ddb826b4ff6ae7808cc11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d2bad8cc18247d39403ddb826b4ff6ae7808cc11"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
