)]}'
{"id":"openstack%2Fneutron~774491","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I2e89bcb183a75f39ba6aeeef6f1ea355fca59fc8","project":"openstack/neutron","branch":"stable/ussuri","topic":"add-conntrack-helper-extension-unit-test-stable/ussuri","hashtags":[],"change_id":"I2e89bcb183a75f39ba6aeeef6f1ea355fca59fc8","subject":"Add extension unit tests for conntrack_helper plugin","status":"MERGED","created":"2021-02-09 05:30:34.000000000","updated":"2021-02-09 14:33:42.000000000","submitted":"2021-02-09 14:27:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774491-add-conntrack-helper-extension-unit-test-stable/ussuri","cherry_pick_of_change":774344,"cherry_pick_of_patch_set":1,"meta_rev_id":"1dd355ab2bb62ad9707dd3b24ab68056e699df38","_number":774491,"virtual_id_number":774491,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 14:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-02-09 09:42:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-02-09 11:42:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-02-09 11:42:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-09 06:52:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-09 07:48:10.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-09 09:42:17.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-02-09 11:42:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"7494940a2ca4141f0dfea7536283f4330fb72f73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-02-09 05:30:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a2d301c04ee5a172f48ae1017fa1655af26e7529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 06:52:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bdd2c7a80154ec39fbe014113f405b3 : SUCCESS in 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66fad7c9b3d14d6199f0b194d6b5b4c6 : SUCCESS in 9m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd9bc32564424b29aa41192d78a4296a : SUCCESS in 18m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d704345928d9494c998041f2fbf4f43d : SUCCESS in 17m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b605a736f96427288b312c43c8b748b : SUCCESS in 30m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7431764c30ca4232adb6c7126bbfedaf : SUCCESS in 9m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/485bc3599b4b46ffaa9ecc98c30ac1e8 : SUCCESS in 54m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/012405374f904dcd8f83bae4293fb360 : SUCCESS in 1h 13m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d47c248ced314e919067efdd91967f61 : SUCCESS in 1h 13m 43s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a71fe7174c194960b5532f18a1214df7 : SUCCESS in 1h 17m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7759afa0ce8d464a889d61174acb48f4 : SUCCESS in 55m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/94b228ada22e4917a9ea67a9db05f0ad : SUCCESS in 1h 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7c6e2de96b952a0772a31d27c27aeaebcf9d52f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-09 07:48:10.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/774491/1 : FAILURE in 2h 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"096164732a24c8f8efff2fc68f911c159d43af4f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-09 09:42:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c825e575b0ad82f7ebbb093cb133231c00a82a0b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-09 11:42:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"939d355f822b9dd01f735b6e865841934fac3d46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 11:42:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd260d3fe2d6511a2212c9c91f4f6f0def447476","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 14:27:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af13abcdd7d24a5397834e828da58f98 : SUCCESS in 14m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0eaff151b02e4f59ae72a882383477bd : SUCCESS in 33m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e5bd2084ab34efa8433a8719c0de240 : SUCCESS in 33m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a141c0ef18b14f9592540781ebbd07ca : SUCCESS in 9m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/27064030ec0f4c8687911c9b7fb937c0 : SUCCESS in 57m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e048f98ada9248d1b5db5e75bc4a8f6c : SUCCESS in 1h 16m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8fda39010a07472e8fac1b33ef596fff : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6599382b6291cc24a16451db42ff878ce665e34c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 14:27:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1dd355ab2bb62ad9707dd3b24ab68056e699df38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 14:33:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b7bff38ab8b4f97838b5a214db24438 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75b8fa7e7b6761fa7a78183d0b77499464ea2bdf","revisions":{"75b8fa7e7b6761fa7a78183d0b77499464ea2bdf":{"kind":"REWORK","_number":1,"created":"2021-02-09 05:30:34.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/91/774491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/774491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/774491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/774491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/774491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/774491/1"}}},"commit":{"parents":[{"commit":"4b7b59efa75b693fcde25094bf6cfff7eefd6342","subject":"Merge \"[OVN] Update metadata port ony for requested subnet\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b7b59efa75b693fcde25094bf6cfff7eefd6342"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-06 02:32:19.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-02-09 05:30:34.000000000","tz":0},"subject":"Add extension unit tests for conntrack_helper plugin","message":"Add extension unit tests for conntrack_helper plugin\n\nChange-Id: I2e89bcb183a75f39ba6aeeef6f1ea355fca59fc8\n(cherry picked from commit cbfac6b7a849a5b7344c525d8fd070e48bdbe6b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75b8fa7e7b6761fa7a78183d0b77499464ea2bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75b8fa7e7b6761fa7a78183d0b77499464ea2bdf"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
