)]}'
{"id":"openstack%2Fneutron~774529","triplet_id":"openstack%2Fneutron~master~Iad01db7f33d6c61b29c552d2dc8893e2946a369a","project":"openstack/neutron","branch":"master","hashtags":["ovn_agent_delete"],"change_id":"Iad01db7f33d6c61b29c552d2dc8893e2946a369a","subject":"Don\u0027t update AgentCache when Chassis_Private.chassis \u003d\u003d []","status":"MERGED","created":"2021-02-08 18:44:23.000000000","updated":"2021-02-23 14:38:45.000000000","submitted":"2021-02-22 23:35:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774529","meta_rev_id":"586ede3beb0a554608817759af32e6ac032fb199","_number":774529,"virtual_id_number":774529,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-22 23:34:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-02-16 18:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-02-09 11:53:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-02-17 08:38:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 19:26:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-08 20:22:46.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-08 20:52:45.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-08 21:51:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-09 02:34:38.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-09 11:53:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-10 14:04:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-02-16 18:19:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"2953b2f1858a9d43eb8bbc3c99691af4248917e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-08 18:44:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"994e1782c8c12fbd6636ce6ee6ef6c8926b03764","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 19:26:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3d3c03016684886b4fc5b6abf48989a : SUCCESS in 32m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/733f838aeb5648c8bfda1a5af8b3015d : SUCCESS in 35m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"536a18777645470264b0b29b24d460ad3b367f10","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-08 20:22:46.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/774529/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/774529/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cd703210f25dd09c50616d705f284ab51b5861ab","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-08 20:52:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774529/1 : SUCCESS in 2h 07m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f3561d9087ecf7504b6e8a3a6fa5e899129067a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 21:51:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/26ec898fe178497e8b113e33a753a864 : SUCCESS in 37m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/87c94d8003e848949858e2372e05ad64 : SUCCESS in 31m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/08e59481541a4a6e95ebbde77f1df94d : SUCCESS in 1h 31m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3b0f9a341a1b4904be594affcf3f522a : SUCCESS in 1h 50m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de61e1200bdf45a2bc85aced1c4f60de : SUCCESS in 1h 40m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/25997776cd28428ab28694e7e7ff6945 : SUCCESS in 1h 54m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03310691436947b08f7ec43b6208e8cc : SUCCESS in 32m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/606adf863bd4451e8c5a66848520834f : SUCCESS in 27m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091e35a2e72941c391ec83fc67a70bea : SUCCESS in 11m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd3f7fb404854c9f9d3c9c7d702c3d07 : SUCCESS in 27m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e754f6fb8c24f2eb58e617712dace14 : SUCCESS in 26m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fd3de70f67847b597a869ce161c8529 : SUCCESS in 26m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/689e241f65d743eb824789d343ed2909 : SUCCESS in 8m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42f0a9b522cd458fa507481e4dc17368 : SUCCESS in 1h 11m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e50fd9ed03e41dc8679748b3bc5743d : SUCCESS in 1h 07m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3cf3d5e47fe447f6a1fb52ef50d097c4 : FAILURE in 1h 41m 14s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9acb125b6f847bba1f7500368c7ecf3 : SUCCESS in 57m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/515cb5fb75f34fa8bdbbd3dab8dccc02 : SUCCESS in 57m 22s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/39ef33c9e3664a89b369311365507e03 : SUCCESS in 1h 47m 56s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/874d882d6ee646829670b64475d66cd8 : FAILURE in 1h 08m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2792a095aa64aae9b3e29f9a58762dc : SUCCESS in 1h 34m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/057975edac3242c4a98a97d891aeb920 : SUCCESS in 1h 05m 20s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/97842806bb0e47ab9ed5f058d12385de : SUCCESS in 1h 11m 15s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4db1580012e04e729322fdb29218cd3b : SUCCESS in 1h 10m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/212db8d6d5904c2f9331cabc2437a35b : SUCCESS in 40m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b5bf93f4a09044a987302aaf69e77142 : FAILURE in 2h 06m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05f9a0f2c5e76339534ab91945ee615ca8c5c636","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-09 02:34:38.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/29/774529/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"8b4cebc158a0a6180f87550725ed63bef73a3367","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-09 11:53:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c22f9ca1227c111dc5d3c30374920ff936aba36","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-10 14:04:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36e81357c3ea8b2091aaa8b815acdaa5f2fe8adb","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-10 21:13:25.000000000","message":"Patch Set 1:\n\nWhile one *could* create a transaction that deletes both at the same time, the events would still come separately for each row. The Chassis/Chassis_Private is only ever deleted (to the best of my knowledge) by ovn-controller shutting down gracefully. Otherwise, it\u0027s just when we delete it (and we only delete the Chassis). If ovn-controller is still up, the Chassis will be re-created.\n\nBasically, in the \"real\" case where ovn-controller is ungracefully killed and we delete the Chassis and ovsdb-server also clears the Chassis_Private.chassis since it is a ref. We get DELETE event for the Chssis and an UPDATE event for the Chassis_Private.","accounts_in_message":[],"_revision_number":1},{"id":"97c47fb699160f482986fdea8572da54cfd1b17f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-10 21:20:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72d48f20afc7b886f64ef4ce14ac96474c87d63d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-16 18:19:05.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e Patch Set 1:\n\u003e \n\u003e While one *could* create a transaction that deletes both at the same time, the events would still come separately for each row. The Chassis/Chassis_Private is only ever deleted (to the best of my knowledge) by ovn-controller shutting down gracefully. Otherwise, it\u0027s just when we delete it (and we only delete the Chassis). If ovn-controller is still up, the Chassis will be re-created.\n\u003e \n\u003e Basically, in the \"real\" case where ovn-controller is ungracefully killed and we delete the Chassis and ovsdb-server also clears the Chassis_Private.chassis since it is a ref. We get DELETE event for the Chssis and an UPDATE event for the Chassis_Private.\n\nThanks for the explanation!","accounts_in_message":[],"_revision_number":1},{"id":"b4c3c98a6648d5357152f7b9823086af119a001c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-17 08:38:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a446175794296610cc39ae5a0b01d3d7941be061","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 08:38:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bde34a3cdf615e59e6de1587d2d4d41a44ee8ec6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 10:54:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2ade3ba3b5b4443cbabdae7d4e2639d8 : SUCCESS in 46m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2f56a8b5d553490da0b9ff3f81217e0b : SUCCESS in 1h 38m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/236b1ef291564703bea107b757fad8cc : SUCCESS in 1h 22m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3aa277fbe9ca4b459724a12861b89a5b : SUCCESS in 2h 14m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ee06613e1f38401fa5874f3aaf37a55c : SUCCESS in 1h 16m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65075905034048ecafd5ca1beb44a828 : SUCCESS in 40m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9c329e86e2940a8a7124a60dad09b50 : SUCCESS in 16m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4a697422584406f984569eb8a43be1c : SUCCESS in 30m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19b064459d2e467aa10eadde245ef72b : SUCCESS in 39m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f52bd6328aa4124b723d4f5df60014e : SUCCESS in 12m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/daaa5600bcf54ad691ba81369d28cfed : SUCCESS in 44m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eac3dbeee2504593993b048cbb6b35ef : TIMED_OUT in 2h 14m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8487e4aede6e48908bf50e77d0750d4f : SUCCESS in 1h 26m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f724da7277a84e32851fd83b5c5154b9 : SUCCESS in 1h 08m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4c8ca28eda6f44c7963c768e5289fe66 : SUCCESS in 1h 16m 44s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/766a9841492a4df8ba677c76579fac89 : SUCCESS in 1h 37m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff36be5bccae4d8eb9beeee1a5147450 : SUCCESS in 1h 06m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b7997d5693654e6295a02007c9a2b972 : SUCCESS in 1h 03m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4386936e4c4b403c828c63eb5ec5320f : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fb39186d27ec854b9f7102fcef980a4d8254d98a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 08:46:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd8a1768e9916c3e086c038913e45d95ae32de12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 09:29:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b16b4eed6285430e874ece6fff44e8ef : SUCCESS in 40m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/99416c417cb548d68b0dd493aa903740 : SUCCESS in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"728427b376532eeb9c4d8f8f1831fe87cea69d0f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 10:37:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/774529/1 : SUCCESS in 1h 48m 41s","accounts_in_message":[],"_revision_number":1},{"id":"171247e7aab3fb1d0bcc4ded5991e7b499320055","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:51:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dec52bfeb32c4e49a39c9be7c9a01835 : SUCCESS in 52m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1149856f41ba4159ab48ad7d5eca86d7 : SUCCESS in 1h 50m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2bd4e4b004f148afb6110244480d5c75 : SUCCESS in 1h 25m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f66ce40da2824bc19c2c855a241e0576 : SUCCESS in 1h 18m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/838eecccdb5f4c20afe86c41ceda724c : SUCCESS in 1h 47m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d34f6ab052dd4e188b50b1683600f679 : SUCCESS in 24m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdc258aff5d042de88b2d0cb29b4e6a6 : SUCCESS in 20m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93686c573978494486f219ce19ec05f9 : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84d4314ec7734b97967e9d1727261d3b : SUCCESS in 29m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c74b932d80e42b3b43be2fd1876edaa : SUCCESS in 17m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a424ece3fcc4b80b835ee4673d03f82 : SUCCESS in 29m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e4d5aef46b493480a29a6052755bca : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd366412067e4556b804c9d4b1c230b0 : SUCCESS in 44m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/30f3475efcb3419893feb88549d88126 : SUCCESS in 1h 10m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/21090bf608e343bd9adb9984b7a651fa : SUCCESS in 1h 30m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/663357229acc4d4faabe33140ac145d8 : SUCCESS in 1h 01m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6302c9ef20d6447fa4a716e217750778 : SUCCESS in 1h 04m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fabf40c1bed244e88ce8497693e34d63 : SUCCESS in 1h 24m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/70598f179d11474bb3ad74e0f36db4ac : SUCCESS in 1h 38m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a4d21c04cb014ec3b3ad2d27c35a18f7 : SUCCESS in 1h 40m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/473f607d534d448eb17f59b1ebe9dd59 : SUCCESS in 59m 41s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c8d7007abf744877ac8d80023035c772 : SUCCESS in 58m 05s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1beac9cb57d3466b9e15bc4b88f2c2d4 : SUCCESS in 1h 02m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f5222c002671447d8d17407927042285 : SUCCESS in 38m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/251ac8d9ad1244a5b23e0e6ff70ec8bb : SUCCESS in 1h 28m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/81b229d6f8d94208b8d562437e41cc9c : SUCCESS in 1h 28m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a5c593214ff198a5cef950a14f47e857d95db249","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:51:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f036d56c9901654b8d23aca7a9c42de8d50da5ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-18 12:21:12.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/774529/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/774529/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d77973e2b7b81fdf5c1cb6f9736325160c7a3515","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 20:57:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5739cb1f2aaf4c34ac83999649c9c55a : SUCCESS in 51m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/344bbcc8cbc348e68f4d2566a63e2edf : SUCCESS in 1h 53m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc4928a1d2b1406e829c2b667cb10d16 : SUCCESS in 30m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b1efb46acd48aeb7904a6c4361c73b : SUCCESS in 14m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea191f6e7df0429ca7850651bcceb340 : SUCCESS in 29m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a35c87c7dc5c48efb8d8997d39bd61cb : SUCCESS in 26m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2969b80439c42668d1f3369a10ac3ed : SUCCESS in 9m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab9d62c7774b4225a9d82279eb661139 : SUCCESS in 49m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea60ee30ee4d4a6681ff550330e1dce9 : TIMED_OUT in 2h 15m 52s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/aeb552d4461c411e9e915221012261c5 : SUCCESS in 1h 42m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5975ba9988a04cc49db4d888574738c2 : SUCCESS in 1h 02m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5bf50ca2c7ca4c5fb42610697201235e : SUCCESS in 59m 33s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94ae31fec85b493a90dd1da71f5cc57a : SUCCESS in 1h 31m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e8f76b1b6ef43baa5e0fdf7fcc77d42 : SUCCESS in 1h 12m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5f514043ec574a8389d63513e14cc704 : SUCCESS in 1h 01m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0135a44df9614d48a5fe80a2fee20716 : SUCCESS in 31m 27s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0cc3a9701f1a408bbdae8ab642e2676d : SUCCESS in 1h 31m 07s","accounts_in_message":[],"_revision_number":1},{"id":"73aa42a3a88114f1db2251a2bff3330d8dad7aa9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-22 17:30:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"06389bfc837a267464872873439c9fc493f8c261","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-22 18:15:43.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/774529/1 : FAILURE in 44m 35s","accounts_in_message":[],"_revision_number":1},{"id":"43301b3472de5ffbc69dc1cd694983d4e804d727","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 18:17:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/023ba7bfe128404d95b7e021f7bb0fb4 : SUCCESS in 33m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5fde9ef1bc0741e5ad706bf0133fd762 : SUCCESS in 39m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d15b4522e6993272a79ba617ec978d1e2c7a2d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 21:15:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0734ff0dbfca4ac1be2cf1e95dc6d861 : SUCCESS in 47m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b45c72e6ad8f4674a1055b336ca0ff4d : SUCCESS in 1h 44m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0e6914ffc4947e4a749dad5c8dc6281 : SUCCESS in 56m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94f1181b3ea14c059d77a345a5180525 : SUCCESS in 30m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66c7ca7e14e34fee8547fdac883af8ce : SUCCESS in 12m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fec857549a74e38b38355a426a6df70 : SUCCESS in 27m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5323a89b4cb34029a545e956c199ed8a : SUCCESS in 30m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/234fa5fe63414ca59a0f1083c172e0ac : SUCCESS in 27m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8e9c7ef0104f8aa0246e278d461f6d : SUCCESS in 8m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9382a482dabb40d3bad19cdbc12e7ed5 : SUCCESS in 49m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6d6e6799c174b6d9cd985299731d667 : SUCCESS in 1h 42m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/37d82248ee0041af949aa4c2836329bb : SUCCESS in 1h 46m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40dfa34c263f4663b015b125df93f11b : SUCCESS in 1h 04m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e9a5c074255d4de9bd1fe447d9d2fa8b : SUCCESS in 1h 11m 24s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c5f7cf59dc8d4eaca466885ee2f12b85 : SUCCESS in 1h 18m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d3ac23b9415248298afc4242067ecffe : SUCCESS in 1h 29m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5fcf96510ccc47acb86a67204b68d05a : SUCCESS in 1h 54m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8d07b2430f240a1a8d74fcf7675dabb : SUCCESS in 1h 02m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/93e5e50a1f0043aaa797bbbc60dd825c : SUCCESS in 1h 10m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a2609afd58a24fdbaf750a5a20431831 : SUCCESS in 40m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/480f5d2fad0e49e39b5da08e5a62b564 : SUCCESS in 1h 35m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/995703815fec40c78294c7052a9497d7 : SUCCESS in 1h 25m 12s","accounts_in_message":[],"_revision_number":1},{"id":"cc67afd5981b3c26b71a6d0a5aadc350a4c269bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 21:15:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74905df2f73b35dfad8e0743d7a638aa53156fae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-22 22:06:11.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/774529/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/774529/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ceaa18ae3d3e92f78efc9ecbefa3f5e5f995bcc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 23:34:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6433c23f4285470782335a3014669cdd : SUCCESS in 1h 00m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d1f278ea126f40a2aa565f2632f8be62 : SUCCESS in 2h 13m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84d5b160cb6a4316963bef28510528fb : SUCCESS in 33m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/671f0ee6126a4ca59f59353de35b9511 : SUCCESS in 12m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39eea7b365424e33bda03eb17dbf40f6 : SUCCESS in 32m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e15b56eedf3d4f5fb3c6a28d6943c338 : SUCCESS in 31m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e78eadcf27d4d7d81ae909a1b89c82d : SUCCESS in 9m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8309097b8a04628893bb2e838c3e5e0 : SUCCESS in 1h 00m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a774372006b41b1825239beacbf4006 : SUCCESS in 1h 00m 55s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7ba258a4329d4f20ae1a0849a2a6be56 : SUCCESS in 1h 24m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/01c2427ec6424d21bb757c4e2135cf76 : SUCCESS in 1h 02m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/216f0eb38fb246df9e98c66b27fc390b : SUCCESS in 1h 02m 25s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f522489107cb4ad2a069f83b72fead54 : SUCCESS in 2h 07m 42s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/757045e3d83c463d8a3a6f58216cbc3e : SUCCESS in 1h 19m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/16d5fa1f93004905bd9caa7a9aea20bc : SUCCESS in 1h 03m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8f246913b71344b582b79fde9d5d0bed : SUCCESS in 42m 47s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/02126a884a834367a7a757a9020c89b4 : SUCCESS in 1h 46m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c1748a3810ca5c6582462321b100dbce3f9abe49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 23:35:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"93359bce93a1483e4c4cc4456cf7fa3be2edd45f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 23:37:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb68924fb5324072946e380de0358eb4 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1},{"id":"586ede3beb0a554608817759af32e6ac032fb199","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-23 14:38:45.000000000","message":"Hashtag added: ovn_agent_delete","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f06268a8ffb5a0cb3d234c7c13a6808b9cce8263","revisions":{"f06268a8ffb5a0cb3d234c7c13a6808b9cce8263":{"kind":"REWORK","_number":1,"created":"2021-02-08 18:44:23.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/29/774529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/774529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/774529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/774529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/774529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/774529/1"}}},"commit":{"parents":[{"commit":"27255fce30f570b549b026346fb10b7d0dca9039","subject":"Convert OvnDbNotifyHandler rows to frozen rows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27255fce30f570b549b026346fb10b7d0dca9039"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-02-08 18:40:36.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-02-08 18:40:36.000000000","tz":0},"subject":"Don\u0027t update AgentCache when Chassis_Private.chassis \u003d\u003d []","message":"Don\u0027t update AgentCache when Chassis_Private.chassis \u003d\u003d []\n\nWhen we delete a Chassis object when deleting an agent, the\ncorresponding value in Chassis_Private.chassis is automatically\nset to []. Even though we use frozen rows, since we listen for\nupdates to Chassis_Private, we will replace the row in the\nAgentCache with this new \"chassis-less\" entry, which will break\nthings since Chassis_Private.chassis is used to display information\nwhich is only stored there like \u0027hostname\u0027.\n\nThis patch avoids updating the AgentCache when Chassis_Private\nhas an empty Chassis field.\n\nRelated-Bug: #1914757\nChange-Id: Iad01db7f33d6c61b29c552d2dc8893e2946a369a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f06268a8ffb5a0cb3d234c7c13a6808b9cce8263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f06268a8ffb5a0cb3d234c7c13a6808b9cce8263"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
