)]}'
{"id":"openstack%2Fneutron~774540","triplet_id":"openstack%2Fneutron~master~Ibfbdd712e49d1fb8522bcc39357be0f311ce3f5c","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"Ibfbdd712e49d1fb8522bcc39357be0f311ce3f5c","subject":"[OVN] Update sample local.conf to mention OVN_BUILD_FROM_SOURCE","status":"MERGED","created":"2021-02-08 20:47:58.000000000","updated":"2021-02-26 07:08:54.000000000","submitted":"2021-02-26 07:07:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774540","meta_rev_id":"7d0dead07921a4aa73cc8a16f027e13b0834194c","_number":774540,"virtual_id_number":774540,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 07:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2021-02-24 10:28:09.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2021-02-24 12:59:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2021-02-24 12:59:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 20:52:04.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-08 20:52:04.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-02-08 21:30:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-08 22:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 22:55:45.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"}],"messages":[{"id":"7e19105f8984d9531c3e1cea8a8051d9e83215b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-08 20:47:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f69b44e8f18ee85bdf0e84e97cae23445c1f1176","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 21:30:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cfa72db9c7f43cba5c112a697e09c3e : SUCCESS in 35m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12b16fa87f174eebb9125c9172ca2113 : SUCCESS in 37m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98238e6999b4f2255edd282fe783184a047c8d30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 22:32:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/52056f15c9d2446abd0af85dcb084e09 : SUCCESS in 57m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/148817a06bb342fcb5f4f332dbf5547a : SUCCESS in 31m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/698810ddf9424ff1aca47978d63c9eff : SUCCESS in 1h 30m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2635d346a732434bbc659322357e1a96 : SUCCESS in 1h 21m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b7fcd015d07c4eb184efffb827d8ea07 : SUCCESS in 1h 41m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cc5fe4439e2f4f0a99307b73ba425767 : SUCCESS in 1h 33m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a3e449c2f674197add109563fcfc35e : SUCCESS in 36m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b48ffef8efba4d36b78f82f87d341434 : SUCCESS in 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de5e559c3e164c12b070fb89460b0812 : SUCCESS in 12m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58c88dd03e9646989241351a26a08dba : SUCCESS in 23m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/746ca075e27d4801bc40784e75a4916a : SUCCESS in 27m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f56219c5af744d8806de3cec7708b1f : SUCCESS in 28m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b3ff924bc949b98c8a41c625153c05 : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":1},{"id":"aeedb1e07d3fce272c0ac816f1ba3d24ce4245b2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-08 22:55:45.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/774540/1 : FAILURE in 2h 06m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c97210891c395d12e8b3ffb2946ec5c3ea1aca22","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-02-24 10:28:09.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks much for this!","accounts_in_message":[],"_revision_number":1},{"id":"1beab503da0dc2a0831f5223b9476f6ab40b7098","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-24 12:59:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef54a41f8e1bdb54e39d8ca9983dabf1f17f6288","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 13:00:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1786917a2d94093fac7e3b4d13aa57d60dd7bab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 16:09:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ca07bcad2979410caa2bbe95d3921767 : SUCCESS in 39m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b74ac71e1e7d4abd868475807476a6ee : SUCCESS in 1h 04m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d24bc8c99bc943e3b8dc168acb54eb46 : SUCCESS in 1h 39m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4ff5e6c6768640dcbfab0a7651fbcd16 : SUCCESS in 2h 13m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/340e0c5e881444e8a05cac75a25c188d : SUCCESS in 1h 10m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c8a64ac014b409eb96de4513fa0f04c : SUCCESS in 29m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5465009327b445dd9e8a97e1fa7783b9 : SUCCESS in 14m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a488b1efe33545c49653ac75675730a0 : SUCCESS in 34m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a665db70335d47b59c031f7f4a2d48df : SUCCESS in 26m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda8a3c87dab4e8395289a3ef72085fa : SUCCESS in 9m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9eb01f42f6314a0a85b64e13b0f3f791 : SUCCESS in 1h 15m 46s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/133e47d0102a4d959f15fe1acbe0b275 : SUCCESS in 1h 19m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2544742961f4655b36f76b55410b8e3 : SUCCESS in 1h 06m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/50ae3016402e45d8adeb47cc4858d539 : SUCCESS in 1h 04m 01s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/81304b2eb26a4a89a2e9c1224d020ff5 : SUCCESS in 1h 34m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91f42be65a10433aa8cb1f16ce68d2b4 : SUCCESS in 1h 04m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0b73247d5f424a4380f5d8cc286e2363 : SUCCESS in 1h 06m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/04292c936c0f4c169cb18dd41abaa904 : SUCCESS in 29m 53s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6e976bb29135473f8b59dda7efebb035 : FAILURE in 1h 19m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6ed435938bbc067231a3d8fd81164570518caff3","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-24 17:17:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"860b51fcfb66e3acb6645c5fda5affb8ed179b21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 18:01:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcb09c2928a249edbe83f763f9d0275e : SUCCESS in 38m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5cd3d7bf9f32418c9d248194bb95cec2 : SUCCESS in 35m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c869f24eda3fc5ba245d9a330abf9accde1b590d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 01:11:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5fb68e99226a4de8aa21e77bc83009a5 : SUCCESS in 1h 06m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/efd1bc80a3e540d982faf5858ae0e6de : SUCCESS in 1h 40m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9aa48f9a81624b3eb02de77e56cfd018 : SUCCESS in 1h 55m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8e1b49c359443afbe38a98db855925b : SUCCESS in 1h 23m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8f2b09845cc246cb98abd88adbac7ff1 : SUCCESS in 1h 55m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45fa2456d19240ceab0523cd9f8c9185 : SUCCESS in 43m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4a51860241f4182b7a777796fbf3e83 : SUCCESS in 30m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42288b8b99d448fda5708ac955036326 : SUCCESS in 11m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40448537bb5646d486ad4766850770a6 : SUCCESS in 27m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c7a084ee3844c3cbc2890b10f87982b : SUCCESS in 27m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef0e466ce87e4321b8c0d7fef5c5dd44 : SUCCESS in 28m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb7f6c14b02e45cc8f509dd490fee9a6 : SUCCESS in 8m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/98218ab74e1f43e9aa7e424cd590b864 : SUCCESS in 1h 20m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3bba2aba572c4a97bbeb2bf6bcde02f1 : SUCCESS in 1h 38m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b41fb2b7a0bf4f4e884e28f2ce25849e : SUCCESS in 1h 03m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/685d25a06b9c445f89a409ebe5217852 : SUCCESS in 1h 10m 42s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1eb371aabb0d46e4adc80a3706e4a03f : SUCCESS in 1h 22m 33s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3080e250a27d429c810d99eea97588a2 : SUCCESS in 1h 46m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/db623b0f9a6840c588c48dd4aa78cb6d : SUCCESS in 1h 46m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cdcdb7f26404572b68cca3fdee44476 : SUCCESS in 1h 09m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5bb5c71b88774a71af4676b632a029f4 : SUCCESS in 1h 17m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/565281d53a1441d7aa099b0bec0fa04e : POST_FAILURE in 42m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/807bb7a3516c4da29628898c2b4f1e15 : SUCCESS in 1h 51m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9baecd8cdfd94b369e1aaa7c719f79b3 : SUCCESS in 1h 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ef6dbf5d88ed94fbea7e3df8e6c7c8549f1025a5","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-02-26 03:02:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba6b506d2b043bd81a7085c9e46affc79e8d12c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 03:44:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5b9448d5e054e948ef56af3b1ab78c5 : SUCCESS in 37m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f891751017a149b0875f09440e0dbec1 : SUCCESS in 36m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d596c2d3ff39e7b67dcecfdd01a654b28a8bcf12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 05:04:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/46fbe86773c64582a3a5dfb460286505 : SUCCESS in 49m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4893b762af374683b80166459e2396cc : SUCCESS in 1h 30m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aa20867732494125bf098b4a6c34e655 : SUCCESS in 1h 18m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5e4e19e775364ea8ba2358ed31872003 : SUCCESS in 1h 28m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/67bff6ed515b47f09433204333832da9 : SUCCESS in 1h 59m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8348a46806c149639d4cc4836796cf63 : SUCCESS in 30m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38cb8c84417f4d35ab125d5bcbae95f4 : SUCCESS in 33m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e307aeddcd64190a2345c1af21010e0 : SUCCESS in 12m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efb3f1f9238c43b39de54109d30e1297 : SUCCESS in 31m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d494dcd63c147499059e48a3086293d : SUCCESS in 27m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34a6a0d087354075935465b1159b19d0 : SUCCESS in 30m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ad0882d6524f608d8c3cdee04ad8b7 : SUCCESS in 11m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c6394131324416e967e9e45a235fa6a : SUCCESS in 1h 17m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/39664118772e463c9075a1cbc3e624a3 : SUCCESS in 1h 35m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a49f0241fb8a40dbb2fddbec98546741 : SUCCESS in 1h 13m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/780b795c91af47eb8233ba4a21a8816c : SUCCESS in 1h 06m 27s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/aad368be8a6848eebeba53dfc9c700cf : SUCCESS in 1h 19m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/65c83f7686d04ee085215c53ab38b7eb : FAILURE in 1h 16m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e3d98ea0268d4ec09b20e55a37acdbb0 : SUCCESS in 1h 38m 11s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/409588562ad3494bbfd639e19eccd706 : SUCCESS in 1h 04m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1725e6bbc01b4bc9b59796699c053ba2 : SUCCESS in 1h 14m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/01f085371217448e8f5294130616a167 : SUCCESS in 29m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dd02d5aa88a24dcaad8114bc35cef49f : SUCCESS in 1h 38m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/66e9c3b385e54c78ac4808843d1afd23 : SUCCESS in 1h 24m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0ad190f8004eb63f4395d9d2f1f0056a239c096c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 05:04:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99a0c8721f6b325929792a80e8c52c5dd1af3e76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 07:07:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2fb4da564e2b4249870fd43847980cdb : SUCCESS in 54m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f7b6faccd28541bfa2e6e6f6211d68dd : SUCCESS in 1h 54m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e2b6770162134810a1d13690d04c04c6 : SUCCESS in 1h 23m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/60ce79ef12ea42cfa6116bc71f9aa8d9 : SUCCESS in 1h 56m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ce55a34a4ae84d16b30d7dc71ba87c11 : SUCCESS in 1h 41m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f087788c9364b0ba919c04abcf3c9d8 : SUCCESS in 31m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/447b4bb8b38340729dbdd469984847c3 : SUCCESS in 13m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32945e55eee948df978c577175ef8432 : SUCCESS in 29m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/152d3e6702b1426e915b710c460cf011 : SUCCESS in 27m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d359c0dba494a85921dd9bf7d9523c7 : SUCCESS in 9m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac8e596fe8aa4606a6a33ac993617fd2 : SUCCESS in 1h 26m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a23a993bb6c74371929f744f5955fda9 : SUCCESS in 1h 29m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/77160fb9321d4c0f9b7ea14f63d33721 : SUCCESS in 1h 05m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bfd4ebcf65e94ffa9d0070815a876fe8 : SUCCESS in 1h 09m 24s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72aa5bb0478f4dc8b3b1f6857bf7a592 : SUCCESS in 1h 40m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1aaa48f429b84e0e9a721056da5a2700 : SUCCESS in 1h 27m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/06b147c77b4e429380517e1fae0d3028 : SUCCESS in 1h 13m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3175ba72c2174a9aa6253e92bb5ddbd5 : SUCCESS in 42m 32s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/64b8bc5bed8f4d738d08d5606461d140 : SUCCESS in 1h 29m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0f49b1c6dd24fe38e6f72556ac7cd56d0c6a2f81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 07:07:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7d0dead07921a4aa73cc8a16f027e13b0834194c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 07:08:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c73c5213b464c83b36213b8c48f87da : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3b6cc06c3c55bfbbea64caa01a66f64afe04c0b","revisions":{"b3b6cc06c3c55bfbbea64caa01a66f64afe04c0b":{"kind":"REWORK","_number":1,"created":"2021-02-08 20:47:58.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/40/774540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/774540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/774540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/774540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/774540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/774540/1"}}},"commit":{"parents":[{"commit":"c4ac136c82722271b0df54454a7d58e0e9c5fa0f","subject":"Merge \"Stop metadata proxy gracefully\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4ac136c82722271b0df54454a7d58e0e9c5fa0f"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-08 20:37:37.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-02-08 20:47:38.000000000","tz":-300},"subject":"[OVN] Update sample local.conf to mention OVN_BUILD_FROM_SOURCE","message":"[OVN] Update sample local.conf to mention OVN_BUILD_FROM_SOURCE\n\nUsing OVN from packages was introduced in devstack via\nhttps://review.opendev.org/c/openstack/devstack/+/763402\n\nThis change updates the sample files to make users aware of it,\nespecially when specifying the OVN branch to be used.\n\nChange-Id: Ibfbdd712e49d1fb8522bcc39357be0f311ce3f5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3b6cc06c3c55bfbbea64caa01a66f64afe04c0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3b6cc06c3c55bfbbea64caa01a66f64afe04c0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
