)]}'
{"id":"openstack%2Fneutron~775089","triplet_id":"openstack%2Fneutron~master~I59ec5e301c012f6c61e066db862fa144ad967510","project":"openstack/neutron","branch":"master","topic":"remove_OVN_SG_NAME_EXT_ID_KEY","hashtags":[],"change_id":"I59ec5e301c012f6c61e066db862fa144ad967510","subject":"Remove OVN_SG_NAME_EXT_ID_KEY constant","status":"MERGED","created":"2021-02-11 09:57:47.000000000","updated":"2021-02-15 11:40:50.000000000","submitted":"2021-02-15 10:51:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775089-remove_OVN_SG_NAME_EXT_ID_KEY","meta_rev_id":"d660f068e3219deb9581bd268b194ddac249c55b","_number":775089,"virtual_id_number":775089,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-02-11 11:21:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-15 10:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-11 13:52:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-02-15 09:01:53.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-02-15 09:01:53.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 11:21:26.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 12:15:10.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-11 12:15:51.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-11 12:15:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-11 13:07:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 13:52:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-15 09:01:53.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"}],"messages":[{"id":"b1d1ad4d7d054d1c7c2889c0c7196bb08eb8b723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-02-11 09:57:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"686789725354269db6e7d3ff20749257e3f07ae4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-11 11:21:26.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/89/775089/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b1881b6b2736a5cb6a8790122a44b141445cb682","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-11 12:15:10.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/775089/1 : FAILURE in 2h 16m 15s","accounts_in_message":[],"_revision_number":1},{"id":"79c865b7ba79162cb9bc4f2ea906c75b8085a5f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-11 12:15:51.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/89/775089/1/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"a9721a6ce088555e3beb93350b666e15456008cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 12:15:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a20f0891250b4110b5911ee1e6d75a63 : TIMED_OUT in 1h 00m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb102ea3b2bf855179df536649f8de6598affe68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 13:07:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c8dbb63dcf76400ea501551c693af70b : SUCCESS in 1h 11m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/99d38ec8a5c14d66bcb8dd0ec7073abf : SUCCESS in 28m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a039fde06bf64af4b35076a169c4eaa8 : SUCCESS in 1h 27m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/33958617f2ce4936a9edd48f3036b7bc : SUCCESS in 1h 29m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5da22dc999f34d46aec9cb643eeeee76 : SUCCESS in 1h 46m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fb2f8618d7f84d979a565d2a2b9e60a7 : SUCCESS in 1h 49m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1177a5ddfaff4fb6be68523bfae9cdf0 : SUCCESS in 35m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caf88de839004c1ca15cabdd75c1d027 : SUCCESS in 29m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/124e599a52054e26bd1711a6eef583d2 : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ff07520c32b4183a8dfe4b39cc466d3 : SUCCESS in 28m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad631a2d3e124312adf6f42365b420db : SUCCESS in 25m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/583e797952784d90b0f62ae268bd471b : SUCCESS in 28m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7881a074e744fb84ea041e27fbc0d3 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48eaf266d72748b8af66d522f6645cf7 : SUCCESS in 1h 10m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6b415e61d4644a297f5a0f155ff5daa : SUCCESS in 1h 10m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8458010c6e3f46fd8d13f807f57dd6f5 : FAILURE in 1h 45m 10s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d91ca75fe6784994834bec00f2c51ed1 : SUCCESS in 1h 14m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7abc659269494d08bfa38a5fbd4788c3 : SUCCESS in 1h 09m 02s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d7b6a1b7f4d84ee29ff95b04625db4e6 : SUCCESS in 1h 31m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/41fae2702157490d9143cfceeaa11741 : FAILURE in 1h 19m 14s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0c9966c4655e4be38e293a38fa16bf8f : SUCCESS in 1h 40m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba539f16aab8454e90c881df1618d180 : SUCCESS in 1h 18m 03s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5cc6f148baa84adc99fcb2f785327b7f : SUCCESS in 1h 10m 17s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/51fa8df916fc42d8b0e26c877e952140 : SUCCESS in 1h 15m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9a9dd602268b4d479cb6fbe8d4a586ed : SUCCESS in 34m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8ce45cb126494626b4b9c21f2eabaab5 : FAILURE in 1h 51m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e16f7743482598568d44a9238392e5830fa69b9c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-11 13:52:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"db1b6b6a2aac171d2318002c87d49d5180902c4b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-11 16:28:06.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/89/775089/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/89/775089/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f84b279ef8ae8aba483fe7e154f1ae4154e87eab","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-02-15 09:01:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d25cdf20137c8fe22b75d19925adc109a485bf49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 09:02:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"684c17a0c33022bf53bf78224a832a0dd15faabe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 10:51:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/848d5bf648964fca8d349176074fea9b : SUCCESS in 53m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/89a0d973def142d7b28383459770efd7 : SUCCESS in 1h 22m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3bc7fc2efd1f416b93c11384d75d664a : SUCCESS in 1h 28m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3a724a95018d49f2bef00c1372566701 : SUCCESS in 1h 44m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/73e184d405af41379bc66e3fbb89a27b : SUCCESS in 1h 20m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42e9783c499846feac794ed0d3c77122 : SUCCESS in 26m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466fc37ba9dc412e8245f0837add8d21 : SUCCESS in 11m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0d337e9ad7b4bc08e8f38d11f73f487 : SUCCESS in 29m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cb4ebb9fc99400f9ed73b66f63e76fd : SUCCESS in 26m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3acc041b0d954b1389e6cfff76ca06d1 : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/635699de5257491fb1747ec259337ed3 : SUCCESS in 50m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5f18f3a4f6244a1ad52965071e3cd33 : SUCCESS in 1h 16m 00s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1b81401ae671430f87342aa57b109d9d : SUCCESS in 1h 29m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/00cc8151f95c47f8b27e70beae731825 : SUCCESS in 56m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a81617f4041947148504d545e6c77d9a : SUCCESS in 1h 01m 01s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/438afc576f4c492ca2f7eae16b47555b : SUCCESS in 1h 36m 43s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30b224acda154f20b6ea8a37c1f654fc : SUCCESS in 59m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d11021bda6c445d49ac41530d0ebef9c : SUCCESS in 57m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c2be987359a54b8882c112c1f9411a96 : SUCCESS in 26m 46s","accounts_in_message":[],"_revision_number":1},{"id":"164447288c4c4ef899e50b1630329a98fc47f50e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 10:51:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"370a138102f36484217e5b6560e0ef96e96895b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 10:53:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e84fbf4bd544ed09bba29ed10f863ab : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d660f068e3219deb9581bd268b194ddac249c55b","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2021-02-15 11:40:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a0caaaf7084d1fc55ac56feb033ad21bd6fa95b","revisions":{"5a0caaaf7084d1fc55ac56feb033ad21bd6fa95b":{"kind":"REWORK","_number":1,"created":"2021-02-11 09:57:47.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/89/775089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/775089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/775089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/775089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/775089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/775089/1"}}},"commit":{"parents":[{"commit":"a3dc80b509d72c8d1a3ea007cb657a9e217ba66a","subject":"Merge \"[OVN] Fix RowNotFound exception while waiting for metadata networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3dc80b509d72c8d1a3ea007cb657a9e217ba66a"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2021-02-11 09:56:48.000000000","tz":60},"committer":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2021-02-11 09:56:48.000000000","tz":60},"subject":"Remove OVN_SG_NAME_EXT_ID_KEY constant","message":"Remove OVN_SG_NAME_EXT_ID_KEY constant\n\nIt is no longer needed after the Rocky Release\n\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I59ec5e301c012f6c61e066db862fa144ad967510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a0caaaf7084d1fc55ac56feb033ad21bd6fa95b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a0caaaf7084d1fc55ac56feb033ad21bd6fa95b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
