)]}'
{"id":"openstack%2Fneutron~775413","triplet_id":"openstack%2Fneutron~master~If5fc1c83aec5f6694abd22f127f1720acad82498","project":"openstack/neutron","branch":"master","topic":"bug/1922892","hashtags":[],"change_id":"If5fc1c83aec5f6694abd22f127f1720acad82498","subject":"[WIP][DNM] Use nft alternative binaries","status":"ABANDONED","created":"2021-02-12 16:34:11.000000000","updated":"2021-04-22 13:39:20.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"41fee10201eb8cbb6143f8dfd070e831afb65d1b","_number":775413,"virtual_id_number":775413,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-12 09:48:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-12 17:37:37.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-12 17:41:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-12 19:16:33.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-12 22:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 23:17:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 15:47:49.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"f24301961aafbde60ef7aba4b811d110e71a589c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-12 16:34:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"711c990072be2a365f78b22905178a17cc806d91","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-12 17:37:37.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/13/775413/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"671dce5127c459be2ba17d3bc5503ffe3d97ed7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 17:41:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8638b3605d244746a6d8517bb02994ee : FAILURE in 15m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a578459cecf47338f7138bbf0092889 : FAILURE in 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cf4dd831faf8020ad78d0b04f33967f99a4993e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-12 19:16:33.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/775413/1 : FAILURE in 2h 41m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4fca19a263363d29fd613ed88a7bf099f0392183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 22:38:22.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8eafc793d655421bbd757c4e8e14fd7c : FAILURE in 23m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03dd7870a0a14cbba585fa99d4a6fe4e : SUCCESS in 28m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4b2a7a5989ae40f2bffe8cdd54d091cf : FAILURE in 21m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b414d6a499e64b3f8f0e67ba3ea84db6 : TIMED_OUT in 2h 50m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6fec9e2aeb074adf9bc7c827e8e5a158 : FAILURE in 21m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cbb4ecec12354f67b18b1f9901d40f23 : SUCCESS in 2h 05m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d104b4fe423e4eac9d243c6f642489fc : FAILURE in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de859578dcc8470ea11c56461b5588cb : FAILURE in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93216f01770417e8a3f070ad5e44716 : FAILURE in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ac998b8943a415f91a1d4bc9d427b76 : FAILURE in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37bebe9899634f7c98f664d5da89f441 : FAILURE in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7eaaf90a75c8466cb79a24a2dad89140 : FAILURE in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86ab25114d9455e92a93d948a31a31c : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2910104c66f4701b9e83c1e87e02cba : FAILURE in 34m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f24ae330a8fc419dbb6ac6f6391931c4 : TIMED_OUT in 2h 14m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/04f3cedf209e45c985e04a033be36a2b : SUCCESS in 2h 17m 43s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/33f3e6e361224f35b82246c400edf016 : FAILURE in 1h 13m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1f7803dc939845bf8bbebf5344e4a904 : FAILURE in 1h 06m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/faa4d99ea54244408fb8c8ac279e9151 : FAILURE in 1h 40m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e2136766e914b43a97fa0f9875b040b : FAILURE in 2h 07m 50s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0b8f94eee9641389262b83f8a973503 : TIMED_OUT in 3h 03m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/492ac4a3bb114100b6965a6a3645b65b : FAILURE in 1h 03m 35s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f2531b49f1a54add966b9b2b6ba0dfea : FAILURE in 1h 39m 36s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/15167a23202d406caae77fb3d9f4b5c2 : SUCCESS in 53m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2d382713e1ae488ab3a6619a91ce4886 : SUCCESS in 38m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5abcd6c9d6ff463a846c90ad3ef7d7dc : SUCCESS in 1h 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4dd7fdb2e6168ab5d8f29d6ad10bb520da0d01a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-16 18:14:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"931983305e9919edb45f55b667cd65c6462f92ae","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-16 18:14:22.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"98a16e894016c53dd9528d1b828bde09e7140ad8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 18:30:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b41874161f84d3884420c847be787dc : FAILURE in 9m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed540d108e904773ba9cc0a9fa83a6ee : FAILURE in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e5442110a36c82a3a838818fd2935b83cec625d2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-16 19:22:38.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/13/775413/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"77574104b9458fa05748af5ef2ff9a4c6de56443","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-16 20:50:10.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/775413/2 : FAILURE in 2h 34m 51s","accounts_in_message":[],"_revision_number":2},{"id":"33c00be4424cd078dda93bded416fb8d6beaf17d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 21:10:46.000000000","message":"Patch Set 2: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aef265cd1fb04f5d816a9bb3ef1e9ce5 : SUCCESS in 53m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0e705cd43af84d6686ca9305a49193dd : TIMED_OUT in 2h 50m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/93b5b1e3cb074288942321f916168c06 : SUCCESS in 1h 47m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b236adcf67944c04a766665cb9ddab6e : FAILURE in 2h 25m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2ceff0015b614ba3903a46c21e6c2ad5 : SUCCESS in 1h 39m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c830fa1813b04670ba6fde96fd5abe64 : FAILURE in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b50a957b51b4e6b8480ef197f77e9f3 : FAILURE in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8ce26919c7645e09099ef2352785aad : FAILURE in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c679dee21c894cf3afdfa29e8b43f6c7 : FAILURE in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5362d19aa2f043b8bcbd08223732bef1 : FAILURE in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/760cf19d58ae48358e77e447cfe0106a : FAILURE in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7264e10152594a9d94f3350ecd732f29 : SUCCESS in 9m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/189d14e57d164057a863bdf3d19f2e5a : FAILURE in 50m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f40b958f532942ea8068b3aacdfda7cb : FAILURE in 1h 00m 01s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/299fb0cc592a42368dde99f2fcdcf9f4 : SUCCESS in 1h 36m 25s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ad61458d19f4e30ae40de641e289ed2 : FAILURE in 1h 30m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9b97ad620b0c48a2ad9c9e3c406f9de7 : FAILURE in 1h 02m 03s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6bd96c439bb84b118fd0a6a6ea3eb51a : SUCCESS in 1h 15m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7779509f4476496eb1cc953134880615 : FAILURE in 1h 09m 04s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8a31357fe454622ad81897307bf54ba : SUCCESS in 1h 38m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b70035ad0d244985afd6471c9df8dcd9 : SUCCESS in 1h 04m 19s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4dd69336767241cfabc2cc4d403f64b3 : SUCCESS in 1h 02m 12s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/da08fff4454046be8502712e14c9ec91 : SUCCESS in 1h 17m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/21190d23b5cf41aea6901002ef23234b : SUCCESS in 30m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a3006e55276343a3baf9b7128a390db4 : SUCCESS in 1h 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a0b39215eb974fd4a80a517f9270277cb2bb5f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-22 13:59:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4facb09ad0bb5d611b1f56dac148ab1f667ad7ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 14:20:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3a4e61ccc794d09ae277d6e7927339d : FAILURE in 11m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea3e9a9708814b97bcfdc3d7ebfbbeb9 : FAILURE in 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6c1c615547b220e9cef3ccc003ad8a65250a408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 19:11:41.000000000","message":"Patch Set 3: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b03710a345d54c39a22fdab731987fd4 : SUCCESS in 1h 17m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d6039692635249118a48ff121116dfdb : TIMED_OUT in 2h 52m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f11286899d4548fd86abc78bd7cdb934 : FAILURE in 2h 39m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ac19760bfb5412abcf37ced61315841 : TIMED_OUT in 2h 54m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0d893added9f481a93952fa97cabea08 : SUCCESS in 1h 35m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/128f2b65774f424c965f81d023bcf680 : FAILURE in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e63a168f2faa4e998b36790ef59e42eb : FAILURE in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34002ca25837498fbd8ab46ac9c3769d : FAILURE in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54e7021d62214c2592e3fed3cafc3815 : FAILURE in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e7f6e261ffb42b6b945174a800d6e7f : FAILURE in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73ecb257522a4125a2f94f0ca5cf7cc8 : FAILURE in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477f07579ace4d8594fb1eb5299b97d6 : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2eeecf0dc831469485d86edf4a5e8cea : FAILURE in 49m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/10dad1bed6614e91a02331d32c5f9296 : FAILURE in 1h 10m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d57e0bc5b3d240869d7f50dd6687ef57 : SUCCESS in 1h 32m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86e6924676864679a687d87fffaa3dd2 : SUCCESS in 1h 05m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/087525161d9447a8955dd1eca0cff8be : SUCCESS in 1h 06m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/27a993f2bbd64d03803f4627f537e300 : SUCCESS in 1h 31m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/744df0abd4fc4e078473fcf0fa41298c : FAILURE in 1h 24m 22s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87b7ce08b0754889813f5a87ac26a90a : SUCCESS in 1h 40m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3aa8f44eb3da4337a1cd82d19f5713c0 : SUCCESS in 1h 00m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0eaf426588b845279030f80301d5bca7 : SUCCESS in 1h 13m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b8a0bca03ece4a348a1bf15b7844bc84 : SUCCESS in 36m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7386eea609854dd3ab72575f6035a49e : SUCCESS in 1h 28m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c4f7fa0a01e54e4182f2fc7bfccf27f1 : SUCCESS in 1h 23m 33s","accounts_in_message":[],"_revision_number":3},{"id":"2ed6475eeb0c2fd7ffc859ad100e1c2275276403","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-22 23:17:00.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/13/775413/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"637d567614ab0b508cfa49652fcca3b5d9585630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-23 10:38:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9224ab90a394ea71edea0c4c088a96940d393a07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 10:39:01.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027base\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-tempest-plugin-scenario-linuxbridge_nft\n      base: neutron-tempest-plugin-scenario-linuxbridge\n      pre-run: playbooks/install_nft.yaml\n\n  in \"openstack/neutron/zuul.d/base.yaml@master\", line 99, column 3","accounts_in_message":[],"_revision_number":4},{"id":"92e688642e110884086c420b18c8ea7152df9602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 10:39:09.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027base\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-tempest-plugin-scenario-linuxbridge_nft\n      base: neutron-tempest-plugin-scenario-linuxbridge\n      pre-run: playbooks/install_nft.yaml\n\n  in \"openstack/neutron/zuul.d/base.yaml@master\", line 99, column 3","accounts_in_message":[],"_revision_number":4},{"id":"d7fbe438e57f90ffd2f615dddb95b588b8ad8664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-23 10:42:45.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"eb56042ffed6f60d56a34b9c0549d5d85bed5400","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:41:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/267a94e99f944225ab68bc338c96c954 : FAILURE in 37m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba4dc69717814fffa3b91b652cdca77c : FAILURE in 50m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1f5555ca448d8ed8a83a8e32e68919965d53f748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:28:14.000000000","message":"Patch Set 5: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90ecf88edee54c85b08df78d685d1e40 : FAILURE in 23m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b550e2565fc4941bb2f7aff3487499d : FAILURE in 21m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c88e523ee5430b95774b63b03f0068 : FAILURE in 12m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21f7c520dd7746049fed5507ca916c51 : FAILURE in 27m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48442c78f52d46b3ade4e1f8e6c440f4 : FAILURE in 31m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f961c2e99244935b51f00f860d88d1b : FAILURE in 24m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fa131e8c2243eaa44209231dd58860 : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3275cfe012f49c1a63867bdb258ad7a : FAILURE in 47m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d87887f1e73c4dd2ac2b8d8c8eaeaa0c : FAILURE in 59m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a7ad53355f3247b190967c1d729fccd3 : SUCCESS in 1h 28m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69dac7306aaa4d5f9ccf9dc3d73d08c3 : SUCCESS in 1h 03m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0bdee43e5f414c71a4ee3dcce2e1ea97 : SUCCESS in 1h 04m 01s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7fead621908a47f8b9aab6708a65c150 : SUCCESS in 1h 16m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/544f0ccc1dc0443b8aaafb92a86a6252 : FAILURE in 1h 20m 37s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/02ec4e1dbcef4df481c1348b3b7c4d74 : SUCCESS in 1h 18m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45cf5315c93f45d38d76ee38a661020b : SUCCESS in 1h 08m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3d686f4ca00c48f5b3c188ab2d13b522 : SUCCESS in 1h 12m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/91d8e8c4e6f9457fa0f9ef3cb29a970d : SUCCESS in 37m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ccae93ffe6b642fda92e8a62d277b600 : SUCCESS in 1h 33m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/adfb4d3319d84870a2b399469fd241ca : SUCCESS in 1h 22m 49s\n- neutron-tempest-plugin-scenario-linuxbridge_nft https://zuul.opendev.org/t/openstack/build/d711f1e86bb348049e233d34b3836e05 : FAILURE in 2h 42m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid_nft https://zuul.opendev.org/t/openstack/build/4310b5d153034c04b2a0f02c3d925233 : FAILURE in 2h 18m 56s","accounts_in_message":[],"_revision_number":5},{"id":"60895a2663e73b1aa001fc318555811e215ccdb3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-23 14:21:22.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/13/775413/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"7098f12f7dbd26fea94aa9e375fb277da2f4bf7c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-23 15:53:32.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/13/775413/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"d57578460bb1a61ef9cfbc07cad8e5b82f62f183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-26 15:37:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7a9dbdf503b868708bf47509987eb99b66117b54","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-26 15:47:49.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"b7a4952845cb0014ad625b3e06d7ef3cc76a087d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 17:17:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5e8f4db9fe7cc207cc000f62b36cea171b01af27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 18:38:38.000000000","message":"Patch Set 6: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9b574d6d0ca4ecb868d32c12ae177ce : FAILURE in 35m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27eb2046c9cb441f91bcd748d3cb9d51 : FAILURE in 53m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccad1157f84046388c80c6be4624a7df : FAILURE in 10m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6912cfe1853a4469800713dcd9d688c9 : FAILURE in 34m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42276e8c49a2490c86ebdbeb8a0ecef0 : TIMED_OUT in 1h 00m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00da075f82f34d7982fd9d52c701f6e2 : FAILURE in 27m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34747cce009f4a1b9c8f9dbe93aef0a7 : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a12a734be77e4220af5d1f4372113755 : FAILURE in 57m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c45bdfda539b4e35be7c0ed449990a45 : FAILURE in 1h 04m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/37732424e82b42e9868202621f6ad6fe : SUCCESS in 1h 35m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0079af5082234a209219e9aebe8cd561 : SUCCESS in 1h 00m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9492f2c5a13c46b9a016c9b6a55969be : SUCCESS in 1h 26m 36s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b0338482d3d44662aa79d480fa191550 : SUCCESS in 1h 27m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e096a88964e4ccc8b2046a9ee548dab : SUCCESS in 1h 37m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fe97428efd8e4d30858ff4efbec542bd : SUCCESS in 1h 29m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f50a0a726094bcfb2a04c3836806219 : SUCCESS in 1h 13m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4c29608d440f42e6bbe5c13f17c7ab78 : SUCCESS in 1h 34m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/340843835742494b9afd619220d3933e : SUCCESS in 31m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/58ab9f5660b74e94bfba66b13108b6c0 : SUCCESS in 1h 28m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/de450d00785e4969bd1f2cd84c16f4a9 : SUCCESS in 1h 43m 58s\n- neutron-tempest-plugin-scenario-linuxbridge_nft https://zuul.opendev.org/t/openstack/build/aa1890f1e4f74fc98ab27cb0c07133a6 : FAILURE in 2h 47m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid_nft https://zuul.opendev.org/t/openstack/build/5ec4be49e2534ac494686c8ec9853461 : FAILURE in 2h 31m 27s","accounts_in_message":[],"_revision_number":6},{"id":"b1725f25d5bb71d49b7108174020f84d2e3e42ba","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-26 19:14:18.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/13/775413/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"bd310e16f3b0d9207e81f1da12a5c6a7e6c20b1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-27 09:24:26.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"69d4c781f87bd4db6f90587d6fc7249d4a7e1c41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 10:03:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/032b8d02623d4d3b94f313f72655b5a9 : FAILURE in 32m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4149c87d50164934a599336c01ccb8d2 : FAILURE in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ef92fece4450d9726046687f87ad5997109bac7e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-27 10:51:43.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/13/775413/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"a4e7744dc52725902493644ab7592559e976249d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 11:17:51.000000000","message":"Patch Set 7: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c2aa3ca733b40fca4d08b1630cb014d : FAILURE in 53m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7629c8b6ec964b3e8bf0aff0093e0db2 : FAILURE in 34m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e777d1eed59944de9228d38170581e5a : FAILURE in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41224e37670648e4ace3ceeb79809fb4 : FAILURE in 32m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/776f7133e7ad4eaa950c6420cc4b909a : FAILURE in 40m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c47d888c85147368187b49536dbf291 : FAILURE in 27m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d24b55651edf4eb4b5a3f1b7df6d4e56 : SUCCESS in 8m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/03d35e3f9afe41ed88080b1da544abeb : RETRY_LIMIT in 10m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b74f9687648490294bc4252143137e8 : RETRY_LIMIT in 5m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5e43c3e24ecc44fda195189807849143 : SUCCESS in 1h 42m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/11165022f25f4aa392979aef776870ff : SUCCESS in 1h 04m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/69669e720b174a49b37931aafc34b309 : SUCCESS in 1h 07m 20s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7ada990dbc70487585339e11de2c8538 : SUCCESS in 1h 29m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/46b1ed5dc308497780c10aa09764aa41 : SUCCESS in 1h 33m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26a942cdd9a04950ae5223e16109a351 : SUCCESS in 1h 35m 55s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c55d9d7229804038bfc302e3288a437d : SUCCESS in 1h 10m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a0b2376cc7d14efb8e28544038057024 : SUCCESS in 1h 16m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/43dff92adec8467d950d700d78aad879 : SUCCESS in 32m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/87df4d396e93432d8faa683517e44efb : SUCCESS in 1h 39m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/00a4f0dfd72e4b18ada79c32ade9851c : SUCCESS in 1h 37m 41s\n- neutron-tempest-plugin-scenario-linuxbridge_nft https://zuul.opendev.org/t/openstack/build/4eaeccf03c284de08e1e08cd446726bd : RETRY_LIMIT in 7m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid_nft https://zuul.opendev.org/t/openstack/build/51f99c44419f4bf0ba9b8387aa522aa6 : RETRY_LIMIT in 5m 12s","accounts_in_message":[],"_revision_number":7},{"id":"27a0ac3ea99067200099d38ee8905a0ffa713a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-27 17:15:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8ad3ba2880619c0588d5210cf3587f9bbb20de01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 17:56:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e16942bfc1db4578aec4032045f25286 : FAILURE in 36m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c85af6ea59c146dbbc7dd9631ddb9670 : FAILURE in 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7a26eb7429a0f4c409efabc436834001f339443d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-27 18:39:04.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/13/775413/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"f142516b1aca9ea36852606ddc367daeb92eb5bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 19:51:30.000000000","message":"Patch Set 8: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be8ced66891b41e493e2577f9b35b2f3 : FAILURE in 38m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cf2aa0d76c34ca2bad1e1d1d2bbc804 : FAILURE in 33m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c5141b8d22143c989c90c32360cfc2d : FAILURE in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/090b3434f7b549f4be7acaf82213496e : FAILURE in 28m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b3cc28f452e47aabb2b9b9af37f7aeb : FAILURE in 26m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d7089879e944e9a98fd5a3da9a6088e : FAILURE in 27m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db9254509f894d2aa9c2da4e377f1a81 : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ead1a0a777b4c27819a7afa70a3f1ab : FAILURE in 49m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba804cfa92794cbda126eea10abc7982 : FAILURE in 1h 48m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/00a88235c63c4fb496951da913fe648a : SUCCESS in 2h 07m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6773f668bc564e2ea3c257b90ded9d5c : SUCCESS in 1h 04m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7c894eadadc94f54bb58845c9d13659c : FAILURE in 1h 35m 45s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fd63caed870a4ac4b7a3664d08dea210 : SUCCESS in 2h 05m 36s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6a9c6816add944c786bf136329401ad5 : FAILURE in 1h 28m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ecd4b5df92ea4afca812311cf858bf76 : SUCCESS in 1h 36m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b3b6f8f48c14c0b996decbc2d48882b : SUCCESS in 1h 08m 20s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dcab94166d554ad3896ec9feb51e992a : SUCCESS in 1h 47m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2f889589477740ddb0ffe299b6b0bc40 : SUCCESS in 1h 08m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/09acb139ecc64517864a66da57414bd1 : SUCCESS in 1h 37m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c4025819d52b4fe7acdfa79270f8a9e6 : SUCCESS in 1h 57m 41s\n- neutron-tempest-plugin-scenario-linuxbridge_nft https://zuul.opendev.org/t/openstack/build/a9a0bcb4ed7e4c058e0ca8d76eb08533 : FAILURE in 2h 34m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid_nft https://zuul.opendev.org/t/openstack/build/b85389a8c5314c0b907ad4adbb248e0b : FAILURE in 2h 15m 02s","accounts_in_message":[],"_revision_number":8},{"id":"7df6bb50dcbd997a4e08009cb774eadd1cb11df5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-29 08:39:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"32f3d4bcb837d54e4af3856050970d1cf3fd153f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-29 08:39:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d4b7aab5927636286987637e976f1267549539cf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-29 10:58:00.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/13/775413/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"9931766301cf3142aea1d2e9f7827f86f715397c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 11:32:12.000000000","message":"Patch Set 9: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/817b0be0565541bebd3f5b586c395d63 : FAILURE in 38m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f3291a15a614a6b848464bca846ddf1 : FAILURE in 36m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5be68abe87a422bbc5d4f3a775029d5 : FAILURE in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df0b99e574804e158672a254e0983582 : FAILURE in 31m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4345beb69ffc4a958dba8b31097e8361 : FAILURE in 26m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba834097503c4287ba6ed4a08f24b3fc : FAILURE in 27m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe314e9df514f34b25f1ceed5efe03a : SUCCESS in 11m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d809eb1b4214c8688b93eabf6400c2c : FAILURE in 49m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d76aee4e68049a2a335331db858b663 : FAILURE in 1h 35m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/06d6f0f6d1f1445fae126309e0b7ab6f : SUCCESS in 1h 30m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec68deb60bb94d35a281ee87947458b0 : SUCCESS in 1h 02m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c887f3883f57467fbbb4a7093d50a052 : SUCCESS in 1h 04m 04s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f049d368427f460a81307065421a3134 : SUCCESS in 1h 19m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/27ac0bb8fb724e9fbb7d24e711fe0c3c : SUCCESS in 1h 28m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cad22d7fcef74c569166dc69c91d4623 : SUCCESS in 1h 32m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e606a25df32f438c885d556e0a12ddce : SUCCESS in 1h 04m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/23d5bdbab2444ed884534908dbe15e9d : SUCCESS in 1h 05m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1fad36fac60443d3896d086a5f617133 : SUCCESS in 35m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/487b17d917ae4efb9ea0cf8c3cd3fe50 : SUCCESS in 1h 24m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b0330622e78e49e6a9eee817b0d249fd : SUCCESS in 1h 21m 39s\n- neutron-tempest-plugin-scenario-linuxbridge_nft https://zuul.opendev.org/t/openstack/build/4ed07225b25f415e8033e36b4ce9c105 : FAILURE in 2h 50m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid_nft https://zuul.opendev.org/t/openstack/build/3d8af129f12441f2809efef7d776e0e4 : FAILURE in 2h 10m 28s","accounts_in_message":[],"_revision_number":9},{"id":"1d6be115221eb316cf6ecea43de3261a1be0c227","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-30 11:44:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3be5ea20afe8cf647d066679857dc3d8d3de3725","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 12:50:26.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c6b5baaf4f14bbcb03116d7bbc9b7a6 : FAILURE in 35m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/efd8900211b64c698a194812b90f69df : TIMED_OUT in 1h 00m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"877bff5824e7111b2e1bfbb3c4649e9a7b914189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 14:30:08.000000000","message":"Patch Set 10: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f10ed556229b424db62afb10719fd831 : TIMED_OUT in 1h 21m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25d20e7194864b9bad0eec7cf2d7dfa0 : FAILURE in 20m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c236812bce1d47f7987920c8f2ddf85d : FAILURE in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f94ec22972f499ba9c3d2f4307c1c76 : FAILURE in 32m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99921c500f684437b1799ca7f34044f8 : FAILURE in 18m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db43b466f34741148473dac400bb7952 : FAILURE in 22m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03fbbcc62d04221bf01f47647fa96ff : SUCCESS in 14m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a790cc3cdf4f45c99b5181dd81586cba : FAILURE in 1h 03m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2ee6287121f4fc8a6dac59e8058774f : FAILURE in 1h 32m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/70b19c54ebd74d559d8da87fa1d56859 : SUCCESS in 1h 10m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea5ac389a87147fe82f1b7ab80dab937 : SUCCESS in 1h 12m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a835ad67af7e4594bd3032d639d62728 : SUCCESS in 1h 08m 17s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/24673575148f4efaaa0add01f23f3d0d : SUCCESS in 1h 21m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7accb0b8d96d4cffbf3fe22f04bbb2eb : SUCCESS in 1h 11m 23s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a2fe39a78a254610964f1a2b8df0c50a : SUCCESS in 1h 44m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbe373c1689f4e6dbbb542f8ad50e424 : SUCCESS in 1h 03m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/830270651a024df68fda13f2e111c992 : SUCCESS in 1h 00m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/04d88298da2146059434395ad7b05845 : SUCCESS in 39m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7121e687844444c189f08dac41779462 : SUCCESS in 1h 33m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/89011d6c3f784c708803aabc02dcf56f : SUCCESS in 1h 33m 54s\n- neutron-tempest-plugin-scenario-linuxbridge_nft https://zuul.opendev.org/t/openstack/build/15f5b39bf0a74540a28fd1ebd5176c80 : FAILURE in 2h 38m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid_nft https://zuul.opendev.org/t/openstack/build/f5c3eda86bf84c1bbdba280f7a6169aa : SUCCESS in 1h 38m 43s","accounts_in_message":[],"_revision_number":10},{"id":"40bcfc37f1fb3a0c9533e33927b8abafe2359602","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-30 19:13:29.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"3ae25f69b3d6cb839fd3fe79c89c6f59591841d6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-30 23:27:13.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/13/775413/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"6d8aea9a7b90ed467cb71242b53255b61736919c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-06 10:49:09.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"0045470b6b4edc2c6315ce615480532ad221c900","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-06 11:03:49.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/775413/11 : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":11},{"id":"171a7b8eb8914d66b45e07e6488016313f2fcfae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:37:56.000000000","message":"Patch Set 11: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cd3a90f8a2f4be69521eb655f65c26a : FAILURE in 30m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35af5ec6982b4a45b53af2938493add7 : FAILURE in 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0745b2a76f64e4a9445096c8d0f2028 : FAILURE in 9m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0cafdbf72ad4db898da78f4417dda66 : FAILURE in 21m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be8e1edaf5dd47e4a499a56a32d18b25 : FAILURE in 32m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf47cee00fd644da87ffa88e588f0971 : FAILURE in 29m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2845f797498c4d0ab1686f2f53ac72d5 : SUCCESS in 11m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/29d4ec6c05244c3ca6b45f9bcb0bc2bb : FAILURE in 46m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8090b08ddc44cc7a40b1b7675502c13 : FAILURE in 1h 38m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2283b4ff44c649a19088a224c21d7320 : SUCCESS in 1h 42m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c750a7fd04234051a088bfd12d2824de : SUCCESS in 1h 03m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a80ab6d25de5490584f38227130e5531 : SUCCESS in 1h 07m 15s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a5fe5bdf08944295af8082ba18d2a4ec : SUCCESS in 1h 44m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8863cc3c6de44b95be6d478cd58875dd : SUCCESS in 1h 31m 02s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d414874311844de3b5fd0c1cd2eb59c0 : SUCCESS in 1h 54m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e7a4d7a65274943a415425ffc1f7bef : SUCCESS in 1h 11m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fb770e9660b94ecb9313cf621f07e80e : SUCCESS in 1h 10m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/81a945424d7d4964a406c9ec0a5885cb : SUCCESS in 33m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/21bbb59598b54d86b39b78ce0f0e8d20 : SUCCESS in 1h 26m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d396e3a3b218451eb008f292bae3b66f : SUCCESS in 1h 24m 54s\n- neutron-tempest-plugin-scenario-linuxbridge_nft https://zuul.opendev.org/t/openstack/build/706fc739c11f4a7ea42e1cb61049f9b3 : FAILURE in 2h 27m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid_nft https://zuul.opendev.org/t/openstack/build/c6bc627dfeb44c99982497b5037a22f5 : SUCCESS in 1h 30m 56s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_iptables_firewall.py","accounts_in_message":[],"_revision_number":11},{"id":"8e991ffbb3294d1a10255c99a300798a94a2c402","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 16:05:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd9e4bf43c74489091ac636962e22945 : FAILURE in 56m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/665dcbced8164fde9415353734aa15db : FAILURE in 42m 24s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"144fe72c1b3374fce2b8c0ab73976a5b68beb37d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-06 16:55:57.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/13/775413/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"b529149de090dd55a5afae72770ab7ee736078cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-07 09:07:41.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2db5e6e48e71bccac8baf17b965d439b7aaebedc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 10:19:32.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fbbead24acb14a2f84f4548596152279 : TIMED_OUT in 1h 01m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ded9336dd4a84835837518000959e69c : TIMED_OUT in 1h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"13062d0b3f19bc5ace6c57a07bbb4520bc827eb2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-07 10:55:49.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/775413/12 : SUCCESS in 1h 47m 12s","accounts_in_message":[],"_revision_number":12},{"id":"eb8f327c15707153f2f232b4f413b83e648d7fa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:02:24.000000000","message":"Patch Set 12: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cc2025a2a4b842a0abdb42e88e3653d3 : SUCCESS in 51m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b7a3dea9ca864f0faa7d34482d435fc5 : TIMED_OUT in 2h 50m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f487d4a4f1364c9c8c6f94980035ec79 : SUCCESS in 1h 47m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/35b4389eb65f45d9a5a742d577713839 : FAILURE in 2h 21m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d8f448d50c2740eaafb39e6f5dca5bbb : SUCCESS in 2h 09m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa9f579507854623a7103d353b87456c : FAILURE in 39m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ac6668f9fed42eeb8b3e91bbd30bac5 : FAILURE in 31m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d641a0a18c4f648082fafcafc79039 : FAILURE in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03ee96d237644e53a12e628eab2b677f : FAILURE in 28m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c7fa7c2bd5d430f843a840bab09a4e8 : FAILURE in 19m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e938c363beb45bf994b8f9c1fa8a218 : FAILURE in 21m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763afb063c184b35a73f1acef6d93b3c : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/452945b672804c218b52bee49d17dacf : FAILURE in 43m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8175995a1070468c8a00d5dc3fdf8c19 : FAILURE in 1h 41m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8ea9d68e254e4d87acd7cac82b5f2521 : SUCCESS in 1h 31m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/72df5eab17c245959ad5960f3b2fe509 : SUCCESS in 1h 16m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9e0580ef960f46c6a3b6876c4ba21bd7 : FAILURE in 1h 05m 49s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4517ac0b41bb41f7873ccbd5c49b44b5 : SUCCESS in 1h 27m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/50722da96de3490a97c94a34f8f1b1a9 : SUCCESS in 1h 58m 13s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/388416b0e6ea4dd8b9eb75dafa81661c : SUCCESS in 1h 44m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b6a46caf73240a9bb5bf43cff11f630 : SUCCESS in 1h 18m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/13d0f049f39140439c8d47a899a2f646 : SUCCESS in 1h 12m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c09f999a38cd42db8b686df552eec547 : SUCCESS in 32m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cb893a3ef98843629afce45527b07c74 : SUCCESS in 1h 43m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4e35005377e34118a3fec0773e676941 : SUCCESS in 1h 34m 03s\n- neutron-tempest-plugin-scenario-linuxbridge_nft https://zuul.opendev.org/t/openstack/build/a1683dea32944269b8bf2c6eddd013e2 : FAILURE in 2h 23m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid_nft https://zuul.opendev.org/t/openstack/build/b76a41727e004df3b8af1591948d3051 : SUCCESS in 1h 34m 30s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_iptables_firewall.py","accounts_in_message":[],"_revision_number":12},{"id":"1e964497edebda6897464967faa5aa502621dd13","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-07 13:06:49.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"99b2e829daf323f7e39c3c27247744a007fbe9cf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-04-07 13:11:53.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2021/Arista_Tempest_Testrefs/changes/13/775413/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"6fc7d869a7aee86f564ec4257f9c148575246bf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-07 13:32:47.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4308f1e504c2008afcfc19232f35c9bd2cd20399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-07 13:59:53.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"498ed1b25a99867d51ff07ae2f14c96df9d65216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 15:21:28.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-nftables https://zuul.opendev.org/t/openstack/build/b54fb4b17f324de3bde2e70ee3d2f018 : FAILURE in 25m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/bac11e89358a416688012a9e867eed6d : SUCCESS in 1h 14m 05s","accounts_in_message":[],"_revision_number":14},{"id":"f1898500643b2bcb9a9865e2438afdd7a20852e2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-07 16:41:04.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/775413/14 : SUCCESS in 1h 46m 54s","accounts_in_message":[],"_revision_number":14},{"id":"45122c6575d100aeedafa94461cfb1ebe1e8064e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-07 16:54:32.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"6d7bc125c7ea58bf92534d961cd33fb6aeb0947d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-07 17:24:49.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":15},{"id":"6dec06b15bd422be476c94451d620f8b9adc74b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 19:29:59.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-nftables https://zuul.opendev.org/t/openstack/build/ab8a5f6412d546b7b2fe4ac66b887acd : SUCCESS in 2h 03m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/15df12803cfc48eb828da097dfdc5c66 : SUCCESS in 1h 29m 31s","accounts_in_message":[],"_revision_number":15},{"id":"0332fe65a6a112483af1f639e9c8f56ff2ba3292","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-07 20:19:19.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/775413/15 : SUCCESS in 2h 04m 08s","accounts_in_message":[],"_revision_number":15},{"id":"46b0f7527dce4400c03dbf79921e34c022c6ef60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-12 09:46:46.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"f3b0ea9c3e7eaafb45d32b8717edca2eb02f1871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 09:48:17.000000000","message":"Patch Set 16: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":16},{"id":"5bd0a6df8d51c6142580c5f9ea957e711d801435","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-04-12 09:48:21.000000000","message":"Patch Set 16:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":16},{"id":"41fee10201eb8cbb6143f8dfd070e831afb65d1b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-22 13:39:20.000000000","message":"Abandoned\n\ntested patches already merged","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"e40c72617efb21ec2770d3fb21ff7db0f1024bc0","revisions":{"ce89ac0da6d56b7fa6f79db653d77ea474828ab4":{"kind":"REWORK","_number":1,"created":"2021-02-12 16:34:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/1"}}},"commit":{"parents":[{"commit":"c89c1f53db288143ef95362971d8db86b8f978ff","subject":"Remove rootwrap execution (1)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c89c1f53db288143ef95362971d8db86b8f978ff"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-12 16:28:27.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce89ac0da6d56b7fa6f79db653d77ea474828ab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce89ac0da6d56b7fa6f79db653d77ea474828ab4"}]},"branch":"refs/heads/master"},"f802b6a149d17a54e96da2ecd3b913c2a22cb5a8":{"kind":"REWORK","_number":2,"created":"2021-02-16 18:14:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/2"}}},"commit":{"parents":[{"commit":"6c75316ca0a7ee2f6513bb6bc0797678ef419d24","subject":"Remove rootwrap execution (4)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c75316ca0a7ee2f6513bb6bc0797678ef419d24"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-16 18:13:22.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f802b6a149d17a54e96da2ecd3b913c2a22cb5a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f802b6a149d17a54e96da2ecd3b913c2a22cb5a8"}]},"branch":"refs/heads/master"},"0644b93d1de70c47140049506922573032a51a5f":{"kind":"REWORK","_number":3,"created":"2021-03-22 13:59:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/3"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-22 13:59:04.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0644b93d1de70c47140049506922573032a51a5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0644b93d1de70c47140049506922573032a51a5f"}]},"branch":"refs/heads/master"},"455113bf942697afad89c9c0b2093180aad476cb":{"kind":"REWORK","_number":4,"created":"2021-03-23 10:38:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/4"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-22 18:02:50.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/455113bf942697afad89c9c0b2093180aad476cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/455113bf942697afad89c9c0b2093180aad476cb"}]},"branch":"refs/heads/master"},"8a7b4d9f8b17f8699521a8a16cf69696f9cce733":{"kind":"REWORK","_number":5,"created":"2021-03-23 10:42:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/5"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-03-23 10:42:37.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a7b4d9f8b17f8699521a8a16cf69696f9cce733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a7b4d9f8b17f8699521a8a16cf69696f9cce733"}]},"branch":"refs/heads/master"},"5eec715b63f89db88090ac5a6570a5a00818f7e0":{"kind":"REWORK","_number":6,"created":"2021-03-26 15:37:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/6"}}},"commit":{"parents":[{"commit":"fa3eab85319895bdf40c1c31a6844c6062719b4b","subject":"Merge \"[OVN][FT] Mech driver stop accessing DB at exit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa3eab85319895bdf40c1c31a6844c6062719b4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-26 15:37:36.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5eec715b63f89db88090ac5a6570a5a00818f7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5eec715b63f89db88090ac5a6570a5a00818f7e0"}]},"branch":"refs/heads/master"},"de0c174d1efa2742dec730e7e4060f84e16b4e23":{"kind":"REWORK","_number":7,"created":"2021-03-27 09:24:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/7"}}},"commit":{"parents":[{"commit":"fa3eab85319895bdf40c1c31a6844c6062719b4b","subject":"Merge \"[OVN][FT] Mech driver stop accessing DB at exit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa3eab85319895bdf40c1c31a6844c6062719b4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-27 09:24:06.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de0c174d1efa2742dec730e7e4060f84e16b4e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de0c174d1efa2742dec730e7e4060f84e16b4e23"}]},"branch":"refs/heads/master"},"bc3b39c3d6f07cb399995de1c3f2e67ed04cd95d":{"kind":"REWORK","_number":8,"created":"2021-03-27 17:15:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/8"}}},"commit":{"parents":[{"commit":"fa3eab85319895bdf40c1c31a6844c6062719b4b","subject":"Merge \"[OVN][FT] Mech driver stop accessing DB at exit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa3eab85319895bdf40c1c31a6844c6062719b4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-27 17:15:17.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc3b39c3d6f07cb399995de1c3f2e67ed04cd95d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc3b39c3d6f07cb399995de1c3f2e67ed04cd95d"}]},"branch":"refs/heads/master"},"94d8742a63343f2ebac7c28ed594f2ae8f4daa3a":{"kind":"REWORK","_number":9,"created":"2021-03-29 08:39:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/9"}}},"commit":{"parents":[{"commit":"fa3eab85319895bdf40c1c31a6844c6062719b4b","subject":"Merge \"[OVN][FT] Mech driver stop accessing DB at exit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa3eab85319895bdf40c1c31a6844c6062719b4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-29 08:39:09.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94d8742a63343f2ebac7c28ed594f2ae8f4daa3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94d8742a63343f2ebac7c28ed594f2ae8f4daa3a"}]},"branch":"refs/heads/master"},"156b9db42a857f4b137a6d9b2d03beeda04ed0d5":{"kind":"REWORK","_number":10,"created":"2021-03-30 11:44:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/10"}}},"commit":{"parents":[{"commit":"fa3eab85319895bdf40c1c31a6844c6062719b4b","subject":"Merge \"[OVN][FT] Mech driver stop accessing DB at exit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa3eab85319895bdf40c1c31a6844c6062719b4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-30 11:44:02.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/156b9db42a857f4b137a6d9b2d03beeda04ed0d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/156b9db42a857f4b137a6d9b2d03beeda04ed0d5"}]},"branch":"refs/heads/master"},"44f3c5f997df30bddd59f774d4517a522bcebf81":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-04-06 10:49:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/11"}}},"commit":{"parents":[{"commit":"5b5aa6d78f4b8ed59ff4e5f2637c1d14599abe39","subject":"Fix \"_get_sg_members\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b5aa6d78f4b8ed59ff4e5f2637c1d14599abe39"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-06 10:48:42.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44f3c5f997df30bddd59f774d4517a522bcebf81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44f3c5f997df30bddd59f774d4517a522bcebf81"}]},"branch":"refs/heads/master"},"55f1d7cef2d31b3368589a7fdb43aa7b43bf8e88":{"kind":"REWORK","_number":12,"created":"2021-04-07 09:07:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/12"}}},"commit":{"parents":[{"commit":"5b5aa6d78f4b8ed59ff4e5f2637c1d14599abe39","subject":"Fix \"_get_sg_members\" method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b5aa6d78f4b8ed59ff4e5f2637c1d14599abe39"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-11 16:18:50.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-07 09:07:17.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\nRelated-Bug: #1915341\nRelated-Bug: #1922892\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55f1d7cef2d31b3368589a7fdb43aa7b43bf8e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55f1d7cef2d31b3368589a7fdb43aa7b43bf8e88"}]},"branch":"refs/heads/master"},"707aabb5eb88ed2d246478f14b1ccfbd8469326f":{"kind":"REWORK","_number":13,"created":"2021-04-07 13:32:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/13"}}},"commit":{"parents":[{"commit":"4c216c3c9b49a02b0f2654965b9fcbed0e1aff79","subject":"Add periodic jobs to test \"nftables\" binaries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c216c3c9b49a02b0f2654965b9fcbed0e1aff79"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-07 13:32:15.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-07 13:32:15.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\nRelated-Bug: #1915341\nRelated-Bug: #1922892\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/707aabb5eb88ed2d246478f14b1ccfbd8469326f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/707aabb5eb88ed2d246478f14b1ccfbd8469326f"}]},"branch":"refs/heads/master"},"5f5ad0db354529d61014a45aaa1e862e51122118":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-04-07 13:59:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/14"}}},"commit":{"parents":[{"commit":"a4d0c2e903c67873fee3a6ca6774b7ab47a0eb82","subject":"Add periodic jobs to test \"nftables\" binaries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4d0c2e903c67873fee3a6ca6774b7ab47a0eb82"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-07 13:32:15.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-07 13:59:35.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\nRelated-Bug: #1915341\nRelated-Bug: #1922892\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f5ad0db354529d61014a45aaa1e862e51122118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f5ad0db354529d61014a45aaa1e862e51122118"}]},"branch":"refs/heads/master"},"2d7dcb5d8e80c0007b155355a79dcd66e96731e8":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-04-07 16:54:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/775413/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/15"}}},"commit":{"parents":[{"commit":"8ed5fd679d76da86ace7d1eeda3c5f50872a1326","subject":"Add periodic jobs to test \"nftables\" binaries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ed5fd679d76da86ace7d1eeda3c5f50872a1326"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-07 13:32:15.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-07 16:54:09.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\nRelated-Bug: #1915341\nRelated-Bug: #1922892\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d7dcb5d8e80c0007b155355a79dcd66e96731e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d7dcb5d8e80c0007b155355a79dcd66e96731e8"}]},"branch":"refs/heads/master"},"e40c72617efb21ec2770d3fb21ff7db0f1024bc0":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-04-12 09:46:46.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/13/775413/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/775413/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/775413/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/775413/16"}}},"commit":{"parents":[{"commit":"0d4aed529bf61975200c032e3b98144bb387efa1","subject":"Add periodic jobs to test \"nftables\" binaries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d4aed529bf61975200c032e3b98144bb387efa1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-04-07 13:32:15.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-04-12 09:46:46.000000000","tz":0},"subject":"[WIP][DNM] Use nft alternative binaries","message":"[WIP][DNM] Use nft alternative binaries\n\nTODO: well, this is not a patch to be merged because we cannot\n      change the binaries to *-nft. We should provide an configuration\n      option and, if possible, a migrating strategy.\n\nThis patch will make use of \"iptables-nft\" and all related commands.\nThat will make use of netfilter \"nftables\", replacing \"iptables\",\n\"ip6tables\", \"ebtables\" and \"arptables\".\n\nBecause \"ipset\" does not have a migrating tool, the code is entirely\nmigrated to nftables.\n\nRelated-Bug: #1508155\nRelated-Bug: #1915341\nRelated-Bug: #1922892\n\nChange-Id: If5fc1c83aec5f6694abd22f127f1720acad82498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e40c72617efb21ec2770d3fb21ff7db0f1024bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e40c72617efb21ec2770d3fb21ff7db0f1024bc0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
