)]}'
{"id":"openstack%2Fneutron~775795","triplet_id":"openstack%2Fneutron~master~I84718b8ac4d8d6e39426e1c4485c2d7fe28185dd","project":"openstack/neutron","branch":"master","topic":"bug/1915530","hashtags":[],"change_id":"I84718b8ac4d8d6e39426e1c4485c2d7fe28185dd","subject":"[OVS FW] Clean conntrack entries with mark \u003d\u003d CT_MARK_INVALID","status":"MERGED","created":"2021-02-16 10:23:44.000000000","updated":"2021-02-19 21:28:14.000000000","submitted":"2021-02-19 21:26:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775795-bug/1915530","meta_rev_id":"26a0e39ab53ca3e1cf372a98fe276292c6b0ad14","_number":775795,"virtual_id_number":775795,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-02-18 11:39:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-19 21:26:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-02-18 14:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-18 15:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-02-19 15:54:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 10:37:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-16 11:14:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 12:18:27.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 15:19:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 17:14:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-02-18 15:03:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"c5f606ad2e7917a39ab9fab0a1123618802caf99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-16 10:23:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36c4238932d3b2238d68fc075defaf5460f6d03f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-16 10:37:36.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/775795/1 : FAILURE in 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"67afabf2849c624b1600a1b5ba8a724c76b46b18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 11:14:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d214c3904434ad98627de4b9fd5e932 : SUCCESS in 46m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4003ea578ab5494faed7a5dce071e3d3 : SUCCESS in 36m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"561e9c8ef82e4494673bf2248ff22bc9b4371af9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-16 12:18:27.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/95/775795/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9a453dbed7fe4d1e5217a643a7ab7d4e503313cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 15:19:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ab453cd0674042ac9cc251f079b68bc8 : SUCCESS in 47m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/10ffda8f8fe54e5f9f2a6ced279d78a4 : SUCCESS in 1h 40m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9bd05503517b4aeba4f77aed274b4982 : SUCCESS in 1h 48m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/71d2b55e41b54048b1134c73e8905147 : FAILURE in 2h 12m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d7218b2b77fd4cdaa9e7e5d184d01007 : SUCCESS in 1h 52m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7292f54d98c4c21b7e87ddb517b9823 : SUCCESS in 41m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fca999619abd468399f2d5851329eca0 : SUCCESS in 31m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430014bc152340a38a6dc3840544d70e : SUCCESS in 10m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e559a2fdb5e44ea8ba360b719db3a72 : SUCCESS in 21m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a99490cba62049f49479305f8e52b68c : SUCCESS in 28m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4c90151ecb4440984a16c6df5b9155e : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f677f6472a34645911846261c638dd7 : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ce879ffa01745efb3006af838e752e1 : SUCCESS in 44m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c91178332ae84582af2d3ae1bc5c6315 : SUCCESS in 1h 11m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/db161e82b7414e0cae229eb1aebb790b : SUCCESS in 1h 32m 25s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d15a41ba0c6451a90ad9085a859c4f9 : SUCCESS in 58m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/99c7f98f64e5402283fdf4a3919f8257 : SUCCESS in 1h 00m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/74765b62cb4344b681562e723b286d5d : SUCCESS in 1h 14m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8c99f71ec17c44ad91b6992852a9f7d2 : FAILURE in 1h 17m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/297a7c0c5c4e4ad2869a942e52d2fc32 : SUCCESS in 1h 35m 54s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/395e8230bad6400ba663ade994d145a1 : SUCCESS in 1h 06m 21s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d6f26b9839ad440dbc62d76b128d6416 : SUCCESS in 1h 05m 02s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/524a3978a19c47a8a5f17def4826e3d2 : SUCCESS in 1h 07m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8a65e470f16d42e39b4ab3f387b901e1 : SUCCESS in 35m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1ebae30dd77040bfa952aa079fbd0245 : SUCCESS in 1h 35m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e9a32da32b6e0dc716aeae4923bb161bc5741cc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-16 17:14:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nI can reproduce the issue without the patch. Without removing the conntrack flow, the traffic is not recovered and the entry persists [1] with mark\u003d1 (invalid).\n\nWith the patch, when the SG rule is re-applied, the conntrack flow is removed and the OVS creates a new one, with mark\u003d0 [2].\n\n[1]http://paste.openstack.org/show/802702/\n[2]http://paste.openstack.org/show/802703/","accounts_in_message":[],"_revision_number":1},{"id":"3fd3627ccbfce2e6365b8e51774f1fab06c4b985","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-16 17:14:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4662b4b29ea930d1536091f94916e5b1e65dc40","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-16 18:06:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/775795/1 : FAILURE in 50m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6d70f292cd5782db39ec6002b38fb546ddf6b757","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 19:25:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20b9fcf863bc4b21b37f9cbeb439e2c5 : SUCCESS in 46m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0dbba6e0807f40059d1d6294a32b94fc : SUCCESS in 37m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75d60f5f4baba9187cc522123ea5764417e5ba51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 00:58:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/49bcc6adebc94241817809ed871537b6 : SUCCESS in 50m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ebf4d46161754f9b9ff1e7512228ff24 : SUCCESS in 1h 28m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a7be51785bf949aead65afb60c8c5f2a : SUCCESS in 1h 30m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d33e1fbc6e1b425fb312d1a7d26e750d : SUCCESS in 1h 32m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c2512f6dd270436398684eeb4c96000e : SUCCESS in 1h 51m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ccfaffcde8449e7b400d99c1b52d52e : SUCCESS in 39m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a42c7a6342a4f608e54ade961e2f5b4 : SUCCESS in 31m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d9fbf66f9084d45b6cf6a2c1abec812 : SUCCESS in 11m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50ae463b435d43dbacef7760b2972dc1 : SUCCESS in 30m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e135fd8a1f984ea2a1dd9677cf35c6e2 : SUCCESS in 29m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc216bc4bb9641929c0a24b06771ddb7 : SUCCESS in 27m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6711fc3bbe1c4738aaa556155b3588d8 : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3a8f7007311493dbb6141eab91175ac : SUCCESS in 46m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6486f46253774c099ae23bcad3f76278 : SUCCESS in 1h 19m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e7790cc6e184423598069bdafb525414 : SUCCESS in 1h 37m 35s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca5ba442ba464e7688e1d464234d00b7 : SUCCESS in 1h 17m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f51f4ade84194d88b8af9054e4e51beb : SUCCESS in 1h 01m 46s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6ce06cc12beb44ccb634464b5243fa6d : SUCCESS in 1h 22m 10s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5b04174c036141458516eacb5a61fb5a : SUCCESS in 1h 42m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa416f99491a4878b2ed42d160681b70 : SUCCESS in 1h 14m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dec3447036024345a73fb52ec2f4a7f9 : SUCCESS in 1h 09m 42s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4ce53bed672e407384dba641c9431d84 : SUCCESS in 1h 06m 03s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aa069728e9b9439395bd698eaa5aca6e : SUCCESS in 1h 03m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/60e890debdc048c88439c108b76ed3a0 : SUCCESS in 32m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fcf22982229e4db79688dc736f8ee6d1 : SUCCESS in 1h 41m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94fddc5f1f1f6d26ffeba960467aec6e664671db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 08:15:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4918283b793692af6e4cc62693cc2c161dd266e2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 08:16:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5fdf6a484714b3e73c08ee62341e7a84c64c6bd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 09:01:52.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/775795/2 : FAILURE in 45m 39s","accounts_in_message":[],"_revision_number":2},{"id":"76f3dd63f89e2204cd25a7223bff00c2e0b13010","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 09:02:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e431cab2b54452eaf7f302dfba80fe8 : SUCCESS in 46m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19e98aa3b8d8485584d682afe6407cdb : SUCCESS in 36m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f96d09345965c6dce79599895aeb81b44e939da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:16:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bf29b2daa7d74949856412e68b36c9db : SUCCESS in 1h 06m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cc678ef4241e4dfd84bb248e6e2d5478 : SUCCESS in 1h 48m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/30b03fb27e3d48ae8d30ed0611bf8714 : SUCCESS in 1h 38m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6ed89ec363a3482686690f7c23241afd : SUCCESS in 1h 21m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/40f1d5df5a7b46899a04186b49fcab93 : SUCCESS in 1h 54m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc4cb91bbf604890bdd09956607c88d4 : SUCCESS in 39m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0d41784976f416b8567c85574f83f84 : SUCCESS in 33m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf868ba9857c482297201802d3949170 : SUCCESS in 13m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d40bf0d3e6204b90924a280665e94494 : SUCCESS in 28m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b43c405d4624da3b4388bc3a9d5fd28 : SUCCESS in 27m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e77346994c02455c8c80280281a9ae85 : SUCCESS in 27m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed05e4eaa844580bc298ae212822ab2 : SUCCESS in 10m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b8513d8ce954a8da2e3efcae96a6a63 : SUCCESS in 52m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/76dedf50868842b4947dbd5c5493d3cb : SUCCESS in 1h 10m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4304565b4cd64e4b8b96f20bc58ee407 : FAILURE in 1h 57m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dfb323f82f5443589fce201459f81dc6 : SUCCESS in 1h 00m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/68b7dd4f8f684b60af8f3742ae00055c : SUCCESS in 1h 21m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5db6596e0cb744d899427a17ebfb7709 : SUCCESS in 1h 28m 18s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1e95693ff0334598825e987333b513a5 : FAILURE in 1h 38m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d856a38d363499193b4b1c90943268f : SUCCESS in 1h 45m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8908e34e794465196f753624d25ff52 : SUCCESS in 1h 10m 16s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4c9fc1b271c74b72a15a34b350f90eda : SUCCESS in 1h 06m 23s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2b8e03cc71104f32ba7a0ad6329ccf9a : SUCCESS in 1h 06m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a9c171d82b0e4320a9a1649e2dc41af0 : SUCCESS in 42m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd94fa97abcf4c3ebdcc5113afa62cbb : SUCCESS in 1h 21m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/68bb2eb60ed94ae8a76b3efdaf899534 : SUCCESS in 1h 43m 34s","accounts_in_message":[],"_revision_number":2},{"id":"dbc247bcc29c71a46f9604767db533f2e55be7ee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-18 11:39:36.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/95/775795/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0784fb2b2ce8988e0fca1bfaf00f9dfc88fd06a9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-18 14:26:01.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a2f6e4a6281d97378cb74b56f121e84013f23480","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 14:49:42.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/775795/2 : FAILURE in 13m 45s","accounts_in_message":[],"_revision_number":2},{"id":"422439b5d4d5f44fbdff25b7a27556f74e115107","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-18 15:03:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f72e9d40944a475813634dfa03f2b8e1322049be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 15:38:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37c38e932f8b44fc9132ce94dabdb34f : SUCCESS in 40m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2911c21dd22748f9bf5a40b6601d72c6 : SUCCESS in 41m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c62deacbfb75a0dc82efd159df2876b99d251b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 22:45:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1473946ea2d64b11906e02bded0a9e62 : SUCCESS in 2h 09m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/96188c2f8b414c989c9ea8790a5d0c29 : SUCCESS in 1h 20m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c623e109027547309322418111266d01 : SUCCESS in 1h 20m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ddfb16f354f6432082da19f57cf9e522 : SUCCESS in 2h 04m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd318e39d79744d1a174afb0c2524c90 : SUCCESS in 38m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d313d23cd9647d08bc88c8ca0be07d5 : SUCCESS in 32m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7550b7919693429eb20c16197e2a2510 : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08b1834f32d24b26ab10d38310082b66 : SUCCESS in 35m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06259b72afd64ec1a20c99692b33aac0 : SUCCESS in 19m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31a029f778ff47bbb9763b9569c8673a : SUCCESS in 28m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/096f06497d6b42ce818f9f9772e2b0aa : SUCCESS in 11m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7822254d51e04ba5bb14286df70fa04c : SUCCESS in 44m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/59e0dd11da8c48dba63b4b2b04bccafa : SUCCESS in 1h 12m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/237d094f767c461aa22a4458f64775bf : SUCCESS in 1h 42m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c67bc0da4f4485da5b2d9e41a57fd66 : SUCCESS in 1h 06m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/72b2d99485484af0bb9f98db840f3854 : SUCCESS in 55m 32s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b2f8af743e554fc9985c5dd0c30d4634 : SUCCESS in 1h 17m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/80b192f3aeee4dfdb7a44b0c0f1ffaef : SUCCESS in 1h 36m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37809b0553404202b4d64cd52796e0b8 : SUCCESS in 1h 39m 47s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/791d4fafdac348b4920adf699eeb5f66 : SUCCESS in 1h 30m 13s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/133ac3432631489f935b26af6f4668a5 : SUCCESS in 56m 59s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e88818393b7446408331c7c08facb052 : SUCCESS in 1h 08m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/18005a08a61b4d5287b0b49f874a3ea7 : SUCCESS in 31m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e074814a2a8946cf85619624e43fd8fc : SUCCESS in 1h 12m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4d1b362a1b3a49d081ef77d33102e296 : SUCCESS in 1h 37m 15s","accounts_in_message":[],"_revision_number":2},{"id":"115a2e69fd0bd97c227f338dac09a0e54321b815","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-19 15:54:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b228c24f81f852823fb15e176ba60ca782681b7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:54:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f8468725424928ddfc1f3ef8e497a74fa4cdfa75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 21:26:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/308b6a94fa2143649c77797e33f76108 : SUCCESS in 1h 22m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f7bd916e7f874181973d74f86bae78ba : SUCCESS in 1h 45m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/526955b35dcd4b6083a0ee0402c38003 : SUCCESS in 1h 47m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f546a7422d50479faa236af4211172bc : SUCCESS in 1h 35m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2fe46b3b89f40cfacde29c0cd656ef3 : SUCCESS in 31m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ceeb22af8324a92b8f700647178b86b : SUCCESS in 13m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd33f689dca24b5a93b0086743f15800 : SUCCESS in 30m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5eb390e674c643258f0feac71940d23a : SUCCESS in 28m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5473010a8e45b6b62fd60f5c100e1e : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a4745a24d2e4bb4a7acdebce1e94e0c : SUCCESS in 47m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/29cb0730ed63450e94779040c2ad8528 : SUCCESS in 1h 07m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/75226d99494644c4a5141c393d46a7e7 : SUCCESS in 1h 15m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/13833c41d3684f2b8b277d557c9b6dbe : SUCCESS in 1h 00m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf876ecc06224e64a2b600ce6f8794f4 : SUCCESS in 1h 07m 36s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/60e5526d8c0c43548aac80d0c62bdb6e : SUCCESS in 1h 43m 41s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06c8db8632594ca7bd2d93f2993506c3 : SUCCESS in 1h 33m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f9276acf19d248bba381599d352b360d : SUCCESS in 1h 12m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/56dd1397e9fd4f15b1a3a452441aa3ef : SUCCESS in 36m 05s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a6e820113c2c4ac59af253f2ff08fd0b : SUCCESS in 1h 23m 03s","accounts_in_message":[],"_revision_number":2},{"id":"daec311a58e1987f4289067435e01a4fdfe2a678","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 21:26:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"26a0e39ab53ca3e1cf372a98fe276292c6b0ad14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 21:28:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06bbd499d1b645ea9a5793e491e9aa66 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e220d7136965acbfe50215d982cac0bd3dcf9c3","revisions":{"4bd296667697567a244405ddd3f259325c5cd7cb":{"kind":"REWORK","_number":1,"created":"2021-02-16 10:23:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/775795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/775795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/775795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/775795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/775795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/775795/1"}}},"commit":{"parents":[{"commit":"924c40975a34c006f9d2f3df3aa16a24d62548ef","subject":"Bump oslo.policy to 3.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/924c40975a34c006f9d2f3df3aa16a24d62548ef"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-16 10:19:44.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-16 10:19:44.000000000","tz":60},"subject":"[OVS FW] Clean conntrack entries with mark \u003d\u003d CT_MARK_INVALID","message":"[OVS FW] Clean conntrack entries with mark \u003d\u003d CT_MARK_INVALID\n\nIn case when security group is removed from port, connections\nestablished to/from port which were hitting some of the removed SG\nrules will be marked with CT_MARK_INVALID in the conntrack table.\nDue to that such existing connections will be stopped.\nBut then if SG will be added again to the port, those conntrack entries\nmarked as invalid will not be removed and those hang connections will\nnot be revived.\nTo fix that, this patch adds deletion of the conntrack entries marked\nwith CT_MARK_INVALID and related to the port of which SG are\nupdated/added.\n\nCloses-Bug: #1915530\nChange-Id: I84718b8ac4d8d6e39426e1c4485c2d7fe28185dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bd296667697567a244405ddd3f259325c5cd7cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bd296667697567a244405ddd3f259325c5cd7cb"}]},"branch":"refs/heads/master"},"1e220d7136965acbfe50215d982cac0bd3dcf9c3":{"kind":"REWORK","_number":2,"created":"2021-02-18 08:15:13.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/775795/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/775795/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/775795/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/775795/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/775795/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/775795/2"}}},"commit":{"parents":[{"commit":"924c40975a34c006f9d2f3df3aa16a24d62548ef","subject":"Bump oslo.policy to 3.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/924c40975a34c006f9d2f3df3aa16a24d62548ef"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-16 10:19:44.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-18 08:14:23.000000000","tz":60},"subject":"[OVS FW] Clean conntrack entries with mark \u003d\u003d CT_MARK_INVALID","message":"[OVS FW] Clean conntrack entries with mark \u003d\u003d CT_MARK_INVALID\n\nIn case when security group is removed from port, connections\nestablished to/from port which were hitting some of the removed SG\nrules will be marked with CT_MARK_INVALID in the conntrack table.\nDue to that such existing connections will be stopped.\nBut then if SG will be added again to the port, those conntrack entries\nmarked as invalid will not be removed and those hang connections will\nnot be revived.\nTo fix that, this patch adds deletion of the conntrack entries marked\nwith CT_MARK_INVALID and related to the port of which SG are\nupdated/added.\n\nCloses-Bug: #1915530\nChange-Id: I84718b8ac4d8d6e39426e1c4485c2d7fe28185dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e220d7136965acbfe50215d982cac0bd3dcf9c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e220d7136965acbfe50215d982cac0bd3dcf9c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
