)]}'
{"id":"openstack%2Fneutron~776426","triplet_id":"openstack%2Fneutron~master~I85a71b389442581acc5fdd6ec18ef420e683fa2b","project":"openstack/neutron","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I85a71b389442581acc5fdd6ec18ef420e683fa2b","subject":"DNM Rally test of all secure-rbac patches","status":"ABANDONED","created":"2021-02-18 12:33:08.000000000","updated":"2021-03-05 13:58:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"301019bab19ba00e544a0ecf64a8a58a0947c65a","_number":776426,"virtual_id_number":776426,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-02-19 16:15:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-18 20:43:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 14:36:34.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-18 15:04:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-18 20:43:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 16:15:38.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d314c3810607a32e32465ca6c5188b5a59b0600","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 12:33:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90a7ffc802e03d119c0863e2a56f2d14382ecc4d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 14:36:34.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776426/1 : FAILURE in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b2dc6b98c727b063c909312963219a95bfd18649","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 15:04:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4530fcdfdef44fb983246e95a043c449 : TIMED_OUT in 1h 01m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d2e968f001b41f18c5886b57e7909be : TIMED_OUT in 1h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0b23428240883421a9f5ec627eb52b062aeac0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 20:43:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f42974860e27473e83f439c2c6f2ad3d : SUCCESS in 50m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e1eaa68898594c75905df0971dda0588 : SUCCESS in 1h 54m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c057c0a10b934a789d51c335bc3b0ef0 : SUCCESS in 1h 30m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/66ef1cba7eec4fbab9553d1e35b0ef4d : SUCCESS in 1h 27m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c91101f498c846bb8c032690b3fd0a24 : SUCCESS in 1h 49m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a848af351c6746689fe208221d6e73af : SUCCESS in 27m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f57154ca6444b37b141f078061cfb16 : SUCCESS in 31m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c7872853124c389f755e9fec039dec : SUCCESS in 14m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2106cc9da23346e9a3bfc9956ba9aaa9 : SUCCESS in 36m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9674ff65de7d4646947450f89e04e5b9 : SUCCESS in 27m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5c035db4332454fa235096304d4794c : SUCCESS in 28m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49b27dc3db5412497c0a786ee7e29b9 : SUCCESS in 9m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdbb1c46f58d4961802a1d34b09f640e : SUCCESS in 47m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1a2c7011c57477bbf4e0bce6939f829 : SUCCESS in 1h 12m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/da8efc61127a4bf1a530a8a22df113ea : SUCCESS in 1h 08m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/88cb1cdb759a46dda41f86ae42096a75 : SUCCESS in 1h 02m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2fa39cedd5c24313ba3d5ec9c7830f5a : SUCCESS in 58m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f340935cbfde43078e522a2d0cd05e0f : SUCCESS in 1h 22m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6e9493a04c66438781fdf961c68078f4 : SUCCESS in 1h 36m 58s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2024d2c02dd942cdb0640f542b68ae19 : SUCCESS in 1h 45m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ac5691844194f01a225ba902a58536a : SUCCESS in 1h 00m 13s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/220216210c5543078f130dddc59bdfbb : SUCCESS in 1h 02m 14s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/74b8cd7523dc4ed2a52e3fdc414e7efb : SUCCESS in 1h 12m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/38cc09dc6ba14d24a7ecf28973da53c3 : SUCCESS in 30m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/154299464f154f4c8b632aaaa037afd1 : SUCCESS in 1h 30m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1ad4080c06be493db7f42d1f32ae93da : SUCCESS in 1h 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b6934e3abec1c68a955bb1d852df140bdfc79df1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-19 16:15:38.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/26/776426/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"301019bab19ba00e544a0ecf64a8a58a0947c65a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 13:58:29.000000000","message":"Abandoned\n\nnot needed anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de1d7b3897f097b4517b44ed660cc717fbe8adf8","revisions":{"de1d7b3897f097b4517b44ed660cc717fbe8adf8":{"kind":"REWORK","_number":1,"created":"2021-02-18 12:33:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/26/776426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/776426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/776426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/776426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/776426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/776426/1"}}},"commit":{"parents":[{"commit":"90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13","subject":"Merge \"Bump neutron-lib to 2.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-18 12:30:53.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-18 12:30:53.000000000","tz":60},"subject":"DNM Rally test of all secure-rbac patches","message":"DNM Rally test of all secure-rbac patches\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/764980\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/765344\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/765847\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/765526\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/765846\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/765845\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/765844\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/765840\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/765417\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/765416\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/771850\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/765391\nChange-Id: I85a71b389442581acc5fdd6ec18ef420e683fa2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de1d7b3897f097b4517b44ed660cc717fbe8adf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de1d7b3897f097b4517b44ed660cc717fbe8adf8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
