)]}'
{"id":"openstack%2Fneutron~776501","triplet_id":"openstack%2Fneutron~master~Ib7d30b954bef3bc3551f1ca206873df354d1ab23","project":"openstack/neutron","branch":"master","topic":"task/41558","hashtags":[],"change_id":"Ib7d30b954bef3bc3551f1ca206873df354d1ab23","subject":"Remove \"IpCommandBase\" base class","status":"MERGED","created":"2021-02-18 17:15:23.000000000","updated":"2021-03-02 13:15:56.000000000","submitted":"2021-03-02 13:10:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"776501-task/41558","meta_rev_id":"992b07384097ed4d4fa7d839f0b917f9c4d74cdf","_number":776501,"virtual_id_number":776501,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-03-01 13:54:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 13:10:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-02 09:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-01 14:20:08.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-02 09:45:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 18:23:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-18 19:23:44.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-18 23:47:03.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 01:59:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 08:04:43.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2021-02-20 02:13:34.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 14:52:24.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-24 07:20:20.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"db41d33c6066d1ea56955e4df591dc25c9473f75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-18 17:15:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6899341750f8fe4da0290f82832f6a8b351cf9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 18:23:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/411779cf6009405facb317e4bc911f2a : TIMED_OUT in 1h 00m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db18b8854a5942e4b9c013e655ac1a13 : TIMED_OUT in 1h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"951ed9d2ead4d48e29d85640b9a550e4f528be6b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 19:23:44.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776501/1 : FAILURE in 2h 06m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0588c59176037dd053db20d42ce7aaefb5af43ae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-18 23:47:03.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/01/776501/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/01/776501/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"202bfc9479cd2ce92f70fc907772009d7d97d9bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 01:59:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a28597afa20b401fa46f7573e74bbe12 : SUCCESS in 1h 29m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/140d9c3b057747e89e553f32875b536d : SUCCESS in 1h 31m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dcc9a98c29124814b92e8f49ff6e3c5a : SUCCESS in 1h 36m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/52caff45877f4c30aae359585a9d1576 : SUCCESS in 2h 42m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb4678b7f323438fb572b8dabb2984c4 : TIMED_OUT in 1h 21m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2636c7cfae540c792d4aa4586fb8ca9 : FAILURE in 37m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f4290e904246e087e488a813e1e0e8 : SUCCESS in 11m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d98a2fc76e04b66b607d82c22a604dd : TIMED_OUT in 1h 01m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a10e437ac2d424b869b0ee0acb53ef6 : TIMED_OUT in 1h 01m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/128b4b67acc049c2a7cdf7dd60fee1f5 : TIMED_OUT in 41m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5257be9b94547ea8f83aa490a0ec58a : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2ef58edf524468fa0dfa3926fc41d46 : FAILURE in 50m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c621b7c0be44cd580b4817e35b90e43 : SUCCESS in 1h 11m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ef7616f3a67643afa2bb2df0d238cf10 : SUCCESS in 1h 42m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/421479a243554beab4c4569f945ed7c9 : SUCCESS in 1h 01m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d43d6a7fd2bd4cc6962a72f6315cbbab : SUCCESS in 1h 03m 45s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8d0dcdcabb4649b0a7a1edb154d61e78 : SUCCESS in 1h 23m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/434646e73429472396feb442f0b98885 : SUCCESS in 1h 31m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d604a8e8d42e405f88f649c187d7ce4c : SUCCESS in 1h 40m 19s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3d4f9f3baed4976a1453036df1c8669 : SUCCESS in 1h 02m 58s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5108f33cfb4f44aca286654e69134ee9 : SUCCESS in 1h 14m 41s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/53941f2994f74d64bf20e42918e2a5a9 : SUCCESS in 1h 08m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/346f90e8e04e456db3605cc68db11056 : SUCCESS in 36m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/62ae18802b2a4a208edce8b17cba7e7f : SUCCESS in 1h 41m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/10e6880af644462a91080fc1ac8e866b : SUCCESS in 1h 26m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5ebb95bd08d2cbda7e47d92f0d2c85dd223733f1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-19 08:04:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a785c35df8a45988d2f9736d89b1d0874de1e551","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-19 09:47:13.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7791b5ff814c1e67239a129b499ac52d8af82738","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 10:43:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a22f1be400242fe8bd99b9b009e0337 : FAILURE in 51m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a352d1b9dbd47bd8d5b9f156d8270a8 : FAILURE in 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08740b4b3258ecc64466af48289533b9b6952ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 12:32:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9bf6f8245e8f44088b7c04075186317b : SUCCESS in 1h 30m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/66ec650e43374ec3a404149d9ecf9b74 : SUCCESS in 1h 35m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c213e6ebf284826a91d7de0a58d212b : SUCCESS in 1h 27m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/be9095b8588d4b0c819185c99983cfd9 : SUCCESS in 2h 05m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae0c527c557242f4aa16d7ecd451a271 : FAILURE in 35m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/764227c68d9f42a081972c898328de57 : FAILURE in 33m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bedca4592470478aa673f65570978596 : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c81d01b9b28c42e384b416f3dc2b89b1 : FAILURE in 28m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ace93e08d7a94b2f91ad8bd0cd6adec8 : FAILURE in 36m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2422211173441598e32ac459893bd7a : FAILURE in 28m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e46d7727fb404b9a982bf3039b5df2 : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa71825367d248da9169d07008951e63 : FAILURE in 43m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a66de73b825f41beb10c395551230ddf : SUCCESS in 1h 36m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4e04ef19dc4e4fd18bbf418105a465f5 : SUCCESS in 1h 49m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40c2e6fbbf754f0fbb121507345d47ba : SUCCESS in 1h 06m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e3eef1482fe448089a6fc36e27b0f5ef : SUCCESS in 1h 09m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/665c412529fc43e8a87aaa6e22b62f81 : SUCCESS in 1h 29m 36s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b097b9ead8ba47879df46a09b8db0a1e : FAILURE in 1h 07m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a041327d68114761a2c1b3bd903a138e : SUCCESS in 1h 43m 45s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bc5ff6e068b41b9a71576c3a0070d5d : SUCCESS in 1h 05m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/123ee57d423046fa8e6f6891c3388291 : SUCCESS in 1h 12m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/907e72cf87774553ad4bbac6c3a7f203 : SUCCESS in 36m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad9739a70c5f4a2795ee5b1448026a8a : SUCCESS in 1h 31m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2d6022f76ea443cbb3bd8c5daf71b960 : SUCCESS in 1h 31m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a934878dd2c86f91b6b08084f804fa879f2e3615","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 12:49:53.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776501/2 : FAILURE in 2h 22m 11s","accounts_in_message":[],"_revision_number":2},{"id":"fbba60d9bf0d934f6514dba89324ba66f0ecb882","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-19 14:19:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ea376a56095a4488981c730520370d0f26bd79e6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 14:23:59.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776501/3 : FAILURE in 43s","accounts_in_message":[],"_revision_number":3},{"id":"db540221b17d1d12a3a1207e335de719d55cf32d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:33:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c304008ac9f5478fb1bb4736a86c7489 : TIMED_OUT in 1h 01m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4eb4fd7ffcf4653a2183e63208ce3ac : TIMED_OUT in 1h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4d9e39c685e5a4eac76f8636ca34e3fca10670b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 17:04:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/75cd284c9f5647378865f80a8f9455db : SUCCESS in 1h 23m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6ed46efb002047f1920f5587f3835296 : SUCCESS in 1h 30m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6ebcc083c93452c9956b3636a63f364 : SUCCESS in 1h 22m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a92eb72b4c6b47c4a4a40448533e7612 : SUCCESS in 2h 16m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c9f0668edc74719a4e759b44ebeb379 : FAILURE in 35m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2827e57732e948bda1a12840590154d0 : FAILURE in 26m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0cde4ce69ea4a58996ec5fe9aa4ad23 : SUCCESS in 14m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c32f88185b644928a1a2b473853d8654 : POST_FAILURE in 31m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9de4fd2880164808aff30e7f09a84f84 : FAILURE in 27m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/467c4aa4de5642eaa9622c51c881705a : FAILURE in 30m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5646916c7ee419887695559acf3df78 : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8fb8ca118c7a44a080cc35e06c66e1f3 : SUCCESS in 46m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe0f51d411514064a85f0d2b376127d6 : SUCCESS in 1h 10m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ba1a4b9d601d4435b2da999a5c1cdf67 : SUCCESS in 1h 29m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eacaf4ea1366441ab3656235f433ff5d : SUCCESS in 1h 04m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5c02545968b542539c7a077971ae8887 : SUCCESS in 1h 03m 18s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b63f027edd0647cc883795d2937674f0 : SUCCESS in 1h 45m 19s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e1061f1ab8b847e193a454c308efb788 : SUCCESS in 1h 36m 46s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/413b020beb5c4af68fbf784dcd55967c : SUCCESS in 1h 47m 45s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca031e2a547d447387f3844a9e8e8328 : SUCCESS in 58m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c181e42809b64269ac6dbfc0e1461093 : SUCCESS in 1h 06m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f4f882cb41e44e3b99fcbdfbc35e7e79 : SUCCESS in 34m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/60320f71abb944eaac8cdc54771e84fa : SUCCESS in 1h 37m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7cc07b386bee47f783c2ca7e892c96b7 : SUCCESS in 1h 21m 47s","accounts_in_message":[],"_revision_number":3},{"id":"57fd9450418ea76cc2abbeb89e4a121958d1af85","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 18:46:36.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/01/776501/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/01/776501/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"533117c582a72ad6220267a01bf1eda85dc611c5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-20 02:13:34.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/01/776501/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a29367391ef36c0190c4426faa0f4e67cf2b0e84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-22 09:16:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7e383e199c492d0a38e6ab19e92e8404f44d82c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 09:55:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac19562d2dc745ae969df291db99689f : SUCCESS in 37m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a32b03ef9446481eb475c1c4a42bebad : SUCCESS in 35m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"de43397eae79fd5a7d8b1fab2e44e8561ad8621a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-22 10:14:14.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776501/4 : FAILURE in 56m 49s","accounts_in_message":[],"_revision_number":4},{"id":"ef0020772d0009390646a34e3b354677d50cb544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 12:06:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1364c9b021c44232a0fd5245d696b316 : SUCCESS in 2h 19m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e5b93d60755b40d59b0f51aac354ec53 : SUCCESS in 1h 25m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1a15d97d13244c1e8861fab8b931d507 : SUCCESS in 1h 28m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/be209e5739de48e5bdaecba27d9adf3f : SUCCESS in 1h 46m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f05917ab5360437595b5b64429f64324 : SUCCESS in 38m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c850f36f8274427caa0b0976276fcb30 : SUCCESS in 30m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5eb8725969343059ef03f91aaaba054 : SUCCESS in 13m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ebb8223a11f4f5fadbf9bb8b265519c : SUCCESS in 29m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a50d577c008c4ee09705b55a8b1d489d : SUCCESS in 27m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/925c34ddabb3441eb67a6dd2559278e4 : SUCCESS in 29m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f60608bb32f746d1b3524a00fabcb06c : SUCCESS in 8m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/12df9e657d774ba88f1aab33e24db46b : SUCCESS in 49m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e76f12ba7ac444f48abdf2355637ad92 : SUCCESS in 1h 06m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2ec141acbead4fbca12fe9604bd37c5a : SUCCESS in 1h 46m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/45bf4ca438504fdca9bcad6d10d6baa9 : SUCCESS in 1h 05m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/63329f344d714cef99712b1f48649976 : SUCCESS in 1h 03m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ae8a0e5425d14e6998ff73afb53483bb : SUCCESS in 1h 17m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e64567839709486eb8e5c68fa6721374 : SUCCESS in 1h 34m 20s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b2f652b08d34886af135104050b9f64 : SUCCESS in 1h 58m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83aa30377dcb4553b7c8e84d5f219c1f : SUCCESS in 1h 12m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9409a1a624a34cf980cdc40b336b9085 : SUCCESS in 52m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cb52cb0b247a48d4a76cd641f7a8b05c : SUCCESS in 36m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/224e13f3700d477e8486ac3777fd0d4e : SUCCESS in 1h 37m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7b73c708d7b14aadafd8e97f23428710 : SUCCESS in 1h 22m 46s","accounts_in_message":[],"_revision_number":4},{"id":"50b60473216c79f8ec5ab808f9b649525c6e14b8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-23 03:11:13.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/01/776501/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"c377b6409c52dc288fbf77fb3e58b0373e4a6019","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-23 14:52:24.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f2fe54a213206f394a0f284001b9c33b656b4a79","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-23 22:22:54.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/01/776501/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"9aae02b13a3b68edeb4aa7b33c006ddbd20fd19b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 05:51:33.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/01/776501/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"ac8963caa4c5630335d6131d6a6dcffad9d77460","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-24 07:20:20.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\n-1 just to make a comment more noticeable (unlike the one on PS1 😊)","accounts_in_message":[],"_revision_number":4},{"id":"4a4c2839377592382ad1875e1c3a2cb18b25143f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-01 12:11:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"22a433e1f40637082192a8ff8d5554c97130935c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-01 12:19:00.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"ce74519ef8969b6e1803613b7228a7f3f59be688","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 13:05:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa28e1816b1047d09d3ba1f529310df8 : SUCCESS in 34m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/708b64b3dcd8496a9ac88e7d121788a4 : SUCCESS in 39m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"938bc2d15e19ff1935cedf1eec534566eeac4b5e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-01 13:54:43.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776501/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"99ff8437997c74d761cbd069ae2cb763d13ba1df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:14:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c233ba5953ca4ac9bf68809d5af2dfb0 : SUCCESS in 1h 50m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ec62d60e2a3440b9a6bbd0b7e58ee9e0 : SUCCESS in 1h 28m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eba1a14af47e45dcbcbe93eca508894f : SUCCESS in 1h 30m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a52c38ec858c4582b896acc8b2302a31 : SUCCESS in 1h 49m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/565bb1c2b1d24150b32fb4d346da7e78 : SUCCESS in 55m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f61bbf2b46a49f9be3460ea0ed7b522 : SUCCESS in 31m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/366132362ab44e83b93d777bf960d235 : SUCCESS in 13m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b0fb791ab914a64b2be05033c43e963 : SUCCESS in 25m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/836760ae2dca4bc7a8bc1634b04cdaef : SUCCESS in 28m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0d58b60adf349b99106e657c35294c9 : SUCCESS in 31m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa9dedfb0de41d7979441f1469da415 : SUCCESS in 9m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0c1e4e58caa4ed491b383b8e9325523 : SUCCESS in 50m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74481aea2c9143c68cd004d679261594 : SUCCESS in 1h 07m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fdc588194a52449aa7744165f30aa6d5 : SUCCESS in 1h 48m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e4f007fbc304df3b45f931f20223df0 : SUCCESS in 54m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7c31a86d6b39472fb8844f1bc5cfa9f7 : SUCCESS in 1h 01m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/83ee023e847b49d8afefe9f484102dad : SUCCESS in 1h 43m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/08ae3b80fbe24d029ff2b1604a8fc16e : SUCCESS in 1h 27m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/33ee0e206a6544afbd000c349cc125e4 : SUCCESS in 1h 27m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a3909d7816149798a3801bd332fc3ef : SUCCESS in 50m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/425e832adcbb4a34a6f8a86e96359570 : SUCCESS in 57m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d7b0546720a84f9590a2af1eae2c10eb : SUCCESS in 33m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/558bedbe51464f43840c985d170c1d31 : SUCCESS in 1h 06m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/75f87e0c2d164ea487fa9e7cee5b8a13 : SUCCESS in 1h 26m 24s","accounts_in_message":[],"_revision_number":5},{"id":"c98397dd4007a95bcf0cdc7a45d24b3115c9d9ef","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-01 14:20:08.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"6a84a4ee5077477c89890843ab77d335bde98a9b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 09:45:29.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9d333eb9edcb3ed7c966b34c5fd3b261528fd3d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 09:47:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"70d80e20b992b47648c97a0834085a76643a299e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:10:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7290106a71244604a9b7e85c2611628b : SUCCESS in 1h 34m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0cbe68e8ccd04c5aae52771ee157e613 : SUCCESS in 1h 23m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/15eb31f45e064c6aa801a499101d11a9 : SUCCESS in 2h 17m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1feaf275df5040fd9582c06056595dea : SUCCESS in 1h 34m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f85bc2d64d8442ea923ca808b95ca74 : SUCCESS in 34m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec1913b11d74b84a7a33905d3fd58cd : SUCCESS in 16m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a2a9066ce984f2e8da06ffb8eaf1c00 : SUCCESS in 26m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f956d2b3d55b489f81962fdfba8ba194 : SUCCESS in 29m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db79fe92bd4e491aa94bc507ddda470a : SUCCESS in 11m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d547c9c35faf4ba8a6e06ef10951cacb : SUCCESS in 58m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7a450ffdc914dd290a5e5b47569f7ab : SUCCESS in 1h 09m 43s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ffa02431b72741b3b4e718eb74c53a98 : SUCCESS in 1h 40m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d6e8bf8ad204937bf2c5c415a83fe69 : SUCCESS in 1h 15m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c1f75b38be9e43359182c088c392d6f2 : SUCCESS in 1h 16m 23s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/560a093d1432498eb3c295c0ef279df8 : SUCCESS in 1h 49m 31s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af995c3270e141f1baa82dca90222f4c : SUCCESS in 1h 14m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4da2f7d31cfc4b518ca51c8a67925730 : SUCCESS in 1h 26m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/13c34d64ec9249b881410c0fa1c3e9ee : SUCCESS in 47m 06s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a9cfaae4ee64470fbe4c00ca1a5aac8e : SUCCESS in 1h 37m 37s","accounts_in_message":[],"_revision_number":5},{"id":"6be50e86a1974ae9b2610f7a34c8df25a53e1795","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:10:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"992b07384097ed4d4fa7d839f0b917f9c4d74cdf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:15:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b28455b3c65c4889a0512ab2632ea218 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0a4dcb18bbc96b347f14d98a959a238c008bf5fe","revisions":{"c130e8449a739b49b66b710664bcec5b4e66d75f":{"kind":"REWORK","_number":1,"created":"2021-02-18 17:15:23.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776501/1"}}},"commit":{"parents":[{"commit":"5f0169d28303492bd9b08e0eedfc60878ca63f06","subject":"Implement \"ip neigh flush\" with Pyroute2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f0169d28303492bd9b08e0eedfc60878ca63f06"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-18 17:07:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-18 17:14:54.000000000","tz":0},"subject":"Remove \"IpCommandBase\" base class","message":"Remove \"IpCommandBase\" base class\n\nStory: #2007686\nTask: #41558\n\nChange-Id: Ib7d30b954bef3bc3551f1ca206873df354d1ab23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c130e8449a739b49b66b710664bcec5b4e66d75f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c130e8449a739b49b66b710664bcec5b4e66d75f"}]},"branch":"refs/heads/master"},"388f278bb523ed8b15de939b690faf1113d94f18":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-19 09:47:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776501/2"}}},"commit":{"parents":[{"commit":"0709f1545b348d58ef06881e8bd7d83888f8d6a0","subject":"Implement \"ip neigh flush\" with Pyroute2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0709f1545b348d58ef06881e8bd7d83888f8d6a0"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-18 17:07:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 09:46:53.000000000","tz":0},"subject":"Remove \"IpCommandBase\" base class","message":"Remove \"IpCommandBase\" base class\n\nStory: #2007686\nTask: #41558\n\nChange-Id: Ib7d30b954bef3bc3551f1ca206873df354d1ab23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/388f278bb523ed8b15de939b690faf1113d94f18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/388f278bb523ed8b15de939b690faf1113d94f18"}]},"branch":"refs/heads/master"},"38ee5b3f328a6634ac1369a4aeb0aa66fd018aef":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-19 14:19:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776501/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776501/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776501/3"}}},"commit":{"parents":[{"commit":"591715b86e086bc42065a9bae1690927414100ad","subject":"Implement \"ip neigh flush\" with Pyroute2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/591715b86e086bc42065a9bae1690927414100ad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-18 17:07:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 14:19:18.000000000","tz":0},"subject":"Remove \"IpCommandBase\" base class","message":"Remove \"IpCommandBase\" base class\n\nStory: #2007686\nTask: #41558\n\nChange-Id: Ib7d30b954bef3bc3551f1ca206873df354d1ab23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38ee5b3f328a6634ac1369a4aeb0aa66fd018aef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38ee5b3f328a6634ac1369a4aeb0aa66fd018aef"}]},"branch":"refs/heads/master"},"b47d03317a1ac9ec96d8a457dd7830a2959e37e8":{"kind":"REWORK","_number":4,"created":"2021-02-22 09:16:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776501/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776501/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776501/4"}}},"commit":{"parents":[{"commit":"591715b86e086bc42065a9bae1690927414100ad","subject":"Implement \"ip neigh flush\" with Pyroute2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/591715b86e086bc42065a9bae1690927414100ad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-18 17:07:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-22 09:15:51.000000000","tz":0},"subject":"Remove \"IpCommandBase\" base class","message":"Remove \"IpCommandBase\" base class\n\nStory: #2007686\nTask: #41558\n\nChange-Id: Ib7d30b954bef3bc3551f1ca206873df354d1ab23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b47d03317a1ac9ec96d8a457dd7830a2959e37e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b47d03317a1ac9ec96d8a457dd7830a2959e37e8"}]},"branch":"refs/heads/master"},"0a4dcb18bbc96b347f14d98a959a238c008bf5fe":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-03-01 12:19:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776501/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776501/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776501/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776501/5"}}},"commit":{"parents":[{"commit":"591715b86e086bc42065a9bae1690927414100ad","subject":"Implement \"ip neigh flush\" with Pyroute2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/591715b86e086bc42065a9bae1690927414100ad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-18 17:07:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-01 12:15:40.000000000","tz":0},"subject":"Remove \"IpCommandBase\" base class","message":"Remove \"IpCommandBase\" base class\n\n\"IpCommandBase\" class was implemented to provide a common interface\nfor all \"ip\" command subclasses. This base class provided a COMMAND\nclass variable, to define the \"ip\" shell command subparameter and\na two execution methods, \"_run\" and \"_as_root\".\n\nNow all \"ip\" command classes have been migrated to Pyroute2, this\nbasic interface is not needed anymore.\n\nStory: #2007686\nTask: #41558\n\nChange-Id: Ib7d30b954bef3bc3551f1ca206873df354d1ab23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a4dcb18bbc96b347f14d98a959a238c008bf5fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a4dcb18bbc96b347f14d98a959a238c008bf5fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
