)]}'
{"id":"openstack%2Fneutron~776567","triplet_id":"openstack%2Fneutron~master~Id8a4c501daad7c2185e6d69441182666ef987e61","project":"openstack/neutron","branch":"master","topic":"bp/distributed-dhcp-for-ml2-ovs","hashtags":[],"change_id":"Id8a4c501daad7c2185e6d69441182666ef987e61","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","status":"MERGED","created":"2021-02-19 01:54:59.000000000","updated":"2021-06-26 12:14:19.000000000","submitted":"2021-06-26 12:12:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":87,"unresolved_comment_count":20,"has_review_started":true,"submission_id":"776567-bp/distributed-dhcp-for-ml2-ovs","meta_rev_id":"c183f41d55768a756b03b4fc5b3dc7de39fc45ed","_number":776567,"virtual_id_number":776567,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":25058,"name":"huangshan","email":"huangshan@fiberhome.com","username":"huangshan"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-26 12:12:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-06-24 13:33:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":25058,"name":"huangshan","email":"huangshan@fiberhome.com","username":"huangshan"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-24 14:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-24 07:56:48.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":25058,"name":"huangshan","email":"huangshan@fiberhome.com","username":"huangshan"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-24 14:11:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":25058,"name":"huangshan","email":"huangshan@fiberhome.com","username":"huangshan"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25058,"name":"huangshan","email":"huangshan@fiberhome.com","username":"huangshan"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 02:06:41.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 03:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 03:37:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 04:32:32.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-20 09:25:08.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 02:25:11.000000000","updated_by":{"_account_id":25058,"name":"huangshan","email":"huangshan@fiberhome.com","username":"huangshan"},"reviewer":{"_account_id":25058,"name":"huangshan","email":"huangshan@fiberhome.com","username":"huangshan"},"state":"REVIEWER"},{"updated":"2021-02-23 09:46:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-24 09:14:39.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-05-25 08:12:32.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-05-25 14:35:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-06-18 03:56:35.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-06-18 12:02:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-06-18 13:11:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"32f9fe1ee12c6b14e619cd6514dd8142aa50031c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 01:54:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb6ec9b6edfb896a6aa6efad4a6cfd2429d50431","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 02:06:41.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"7b1f26ef9744dda87f2933ef6d7dec12fcaeb39a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 03:03:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c72925887f5447793a897bfcd1234d6 : SUCCESS in 53m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/741aa0667c6d4028a8fb0d6eef0571aa : SUCCESS in 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49998fe004007bfa68cd50966b7a239e09f882c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 03:37:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/66ce01ec2cc04d9e8aa343a1b7419e23 : SUCCESS in 3m 27s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a7706ab1b2cb405c88799f2392957867 : SUCCESS in 47m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/067d21e9d9c2496398fd8f28c7c78cee : SUCCESS in 1h 36m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9888334542934f9bbcaee50d83bc740e : SUCCESS in 1h 34m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/057cb82721724c0ba414d8dd32b3d3cd : SUCCESS in 1h 09m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42790a0bfcab4c88b850650adf65c5b7 : SUCCESS in 33m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/101be1c0b744439789b1ada0697d72af : SUCCESS in 37m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc7accbed25482cbaf1937f68037b9d : SUCCESS in 12m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4acb0ac19b445eeb218231c09717a63 : SUCCESS in 33m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e8c04ee442341c48d1694fe7c57d063 : SUCCESS in 31m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/206c684ad51a4346a97a9ca876e60736 : SUCCESS in 33m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff19a5ba92ec46aabf36c27c55dc65d8 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e54b2e306de54d75be04224cbcd10fbc : FAILURE in 44m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7618a6150a7841c2bce1cb445ddd8127 : SUCCESS in 1h 13m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/87c8ad95229741b39f2400f13dbe8149 : SUCCESS in 1h 29m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a67035ca18e4029a0d9afb5d74e1a27 : SUCCESS in 57m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0f62f03236f04563bbdd0fe09bac0488 : SUCCESS in 59m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/041c6de4b79b40a19e7b3b525ce6bc5c : SUCCESS in 1h 14m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d8e3b0f86b254dd09ee073f489b7abfc : SUCCESS in 1h 27m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fed8c275d0d34ed8aaab70d8f44e155d : SUCCESS in 1h 28m 45s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d43fdd998a1c469ca539b3095dfbce2a : SUCCESS in 59m 27s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/cb5b78b586b0497aa14079ded3ed244e : SUCCESS in 56m 02s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f01e1f31efe04370a0e7e51b333869da : SUCCESS in 1h 06m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e7c2c5f0411f44698de06fb7dce50b01 : SUCCESS in 29m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2567df42a74247d3bf5c6de1be507176 : SUCCESS in 1h 28m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a6c0c073bbd945b9b29d4cf55dfeb94a : SUCCESS in 1h 18m 25s","accounts_in_message":[],"_revision_number":1},{"id":"aea6b8ad90bde7ae37a9d90ad0691454e58db170","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 04:32:32.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776567/1 : FAILURE in 2h 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9e01199964ad63da1bcfc81b796b87e58389149f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 04:55:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f81e122c6458242b0d30662fc653450b78eb43dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 05:53:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/236f794b26be4e65a34d437add6fdac9 : FAILURE in 47m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60c5a4dcb2e34df38991212024b5f687 : FAILURE in 48m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3721e9b128bd22a2ea923e8bb285f37a58ee748a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 06:51:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b5724ea2ef8d4e85a614b27ab608398d : SUCCESS in 4m 04s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7eb96b5cb3ba4a989dc716a705515bbc : SUCCESS in 53m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e7a8eaad1ab0455186dd58d821448db6 : SUCCESS in 1h 15m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4d41f0107ef646b694e7f130e53815ae : SUCCESS in 1h 21m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/44d7657020d14cd79f9f70ed0be8330c : SUCCESS in 1h 21m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f2374e5bec74125b527f058d1aa35ed : FAILURE in 34m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/486d638c8c09423ba51a66359d64429a : FAILURE in 31m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7b6c5d776114c4cb77e920d473366c8 : SUCCESS in 13m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d20354715bd48979f6a053d29b37999 : TIMED_OUT in 1h 00m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b11a787c59504e6c9fd0f6b13314ec5e : TIMED_OUT in 1h 00m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99eae7f107bd4274a5f174188aeb6df5 : TIMED_OUT in 41m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ac2038bba3444593c9ad81e9f2fec2 : SUCCESS in 8m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b28b114312d454a88d508682a4c8a1c : SUCCESS in 1h 04m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7db3f505dfa465282a22b4a9eaa719c : SUCCESS in 1h 14m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5ae8a01e7b634924969e7dd923c11bc8 : SUCCESS in 1h 34m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9acdd8a350394d0d8e9856e5f2e46471 : SUCCESS in 55m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cd8d788b75b34fc1a3686106a56a8a06 : SUCCESS in 1h 05m 57s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d6195bc61f2346dba1d7fdac6c78c009 : SUCCESS in 1h 40m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/490b1e5a29614e0bb7872f6daedc06f5 : FAILURE in 1h 30m 30s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d8880c73b31846a5a408f5a864229eae : SUCCESS in 1h 49m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/624cb0f4ba7c4a62be79dfe7ff96a7cd : SUCCESS in 47m 26s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0a5eea3b8efd475a9f002966d659bad9 : SUCCESS in 1h 06m 29s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4a84dc1f912d4ad7b240bf2616adbe5f : SUCCESS in 1h 23m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5b7f76c61890451fae82712e512fd389 : SUCCESS in 31m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/35624e3e22bf44638514129fc4e83a8c : SUCCESS in 1h 46m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/67e54087cac544428aebe5723ad458cc : SUCCESS in 1h 32m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9618f8700f0f81fe562210437522c69f65ab1a93","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 07:01:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/2 : SUCCESS in 2h 03m 56s","accounts_in_message":[],"_revision_number":2},{"id":"31e446a63f1b3f3ced87c99ec18a2f3d6575121f","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 07:59:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b877e8778896adf0e07675a15c03a9dbac24c354","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 09:15:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c3f122e1c7b43c4b009d97767f51a89 : SUCCESS in 53m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0af4805f63ce424c941425776c27e351 : SUCCESS in 55m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"efcbcf1c28c7555eb57c23f820bb3bf4c4f8931e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 10:01:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/2 : SUCCESS in 2h 00m 24s","accounts_in_message":[],"_revision_number":2},{"id":"215a928776772c80eac6839dd650e87555fae764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 10:18:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ad41fcea2ac44edaf3e589bebdc0a22 : SUCCESS in 3m 15s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e8e18c653a93466d9dcec3d458f9c2de : SUCCESS in 55m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8a7b7ff0d0c443bda1cedbcc5f810a1a : SUCCESS in 1h 56m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f2521faffcda4371bb9436fc38db54d5 : SUCCESS in 1h 41m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/89f6b39581d048a38de6b555670a4092 : SUCCESS in 1h 23m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af6b32d52415486da14a765bb7992c83 : FAILURE in 35m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1055cc68f68646d289e16a936aa9f7e1 : SUCCESS in 32m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94042c5d5d404173aa1de4815825f2b1 : SUCCESS in 13m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b00bd14c2bad42e79e3999305effaac4 : FAILURE in 30m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a143da4a355f4cb7b11bedec241ca595 : SUCCESS in 30m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5ec6a0fed144783acb6e78ea49be98e : SUCCESS in 27m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08521306590f41c6885a162c7c7ed8cc : SUCCESS in 10m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dc711d278cfe4b56814ee9b8e370bc43 : SUCCESS in 45m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/83c2c2294b0e4f1e91393ab57af163bd : TIMED_OUT in 2h 15m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/70ffa99fbe3548dab7308343b896de2b : SUCCESS in 1h 39m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d0a1fed1ec4448f94f8b026876a49f3 : SUCCESS in 1h 00m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6d3182b2afa54c3083c10e1fce538911 : SUCCESS in 1h 01m 57s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/30fbdde718aa4d4b8d53ced0b5ff6203 : SUCCESS in 1h 20m 50s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a2a8b2d3472c48e6ae615b1438c71d28 : FAILURE in 1h 09m 47s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f0f29972395f4484a0e64a443040691e : SUCCESS in 1h 39m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bd2e27f4d624260ad89b608bf015ca2 : SUCCESS in 1h 17m 01s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f26689bf270840978c0549a74c15b8e6 : SUCCESS in 1h 07m 12s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2b3dccfd291e4b9f98812c35c4129742 : SUCCESS in 1h 06m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2bf6c0b6de0449578e1f89e11ba462ae : SUCCESS in 32m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cf633885b4004655b63b04af4313d42f : SUCCESS in 1h 42m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8c2001a124ae4108adc485ea6161a1e2 : SUCCESS in 1h 29m 19s","accounts_in_message":[],"_revision_number":2},{"id":"767bc513ae0bf6f11707b7133aec69d813be79c0","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 15:57:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"68836cb773045edbf4e234013b45382461832a0d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 16:00:38.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776567/2 : FAILURE in 42s","accounts_in_message":[],"_revision_number":2},{"id":"e1992a5cd928df3b5dbee07a3eea0dfe80a4d5c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 17:00:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52496b0e8d1f4f58a3b7e31ecea6e030 : SUCCESS in 55m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c0fff3e0d11468f95c05a222eba423b : FAILURE in 55m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01833d7b01bf8f318ab36c9dc376ef380e1d6452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 20:13:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1559636494b445708d2abec7fd1aaf21 : SUCCESS in 3m 13s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2851ae02c6b44ca4809db3eeed891bb1 : SUCCESS in 51m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/33b1df9b1bc2478e8be4b3ecd0a5264f : SUCCESS in 1h 25m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8dd5dc71c49547baa63289abf478ff2e : SUCCESS in 1h 56m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b8fa175ad5334e298c57d8d890da752e : SUCCESS in 1h 40m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b46c503243e8451e923e35caa38e8150 : SUCCESS in 43m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d54ad917a2f842f68d96eee1de1e26ff : SUCCESS in 35m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28032a6f97aa4068804a62a6ff93a213 : SUCCESS in 14m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a33eed20482c48c3b6387aec84243c44 : FAILURE in 30m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8190776fa84424e87f19eb4c83b0761 : SUCCESS in 27m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0bf42ff148e4800870454cc4b8538c3 : SUCCESS in 29m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3558c0989f849a38c23624135bb9d5a : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ced713b8f234458cb67648fabc0e8892 : SUCCESS in 48m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/065ae862d744448d99b4c1703c388fc0 : SUCCESS in 1h 11m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f985e7e56285479cb9a50bf921dac2dd : SUCCESS in 1h 41m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/12140892510140bea64feff897ad7e74 : SUCCESS in 1h 13m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5e8e7dc9486b47e5a98ae39143b67aec : SUCCESS in 1h 09m 03s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b12ba672a6e94dc09ef93ec22ffdd009 : SUCCESS in 1h 27m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a8f67cc2bc354cdc93fade4ac7329775 : FAILURE in 1h 20m 49s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2aa7ce029bb4ba59657a092cd4cfc7b : SUCCESS in 1h 47m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8ca0e8438fa4b8999e5d165fc94f61e : SUCCESS in 1h 08m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/225b0690bc9e407b926469588357b245 : SUCCESS in 1h 15m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/803b8f9b677d4c1392d037b7a88422e5 : SUCCESS in 33m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6202f8f8679d4a3b8c0153222769e2c0 : SUCCESS in 1h 25m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0c69895bf1b44ee8bed81758bc70e2c1 : SUCCESS in 1h 27m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e1df370cc9f3373f738c28656ab6bb62fe89dfa8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-20 09:25:08.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/67/776567/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7c2b77d5bd676f155300a035d2096124b966242f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-20 12:11:51.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/67/776567/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"d0918e6049e1a50bd21ed1108e6dfaac7178b82c","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-22 01:33:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c694e1b2989bba2f3f7531d8f9f11d9b89cbac46","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-22 02:13:15.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776567/2 : FAILURE in 38m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a14ed50e1972c2c5ddcbffaca5e4517cdc74ac0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 02:13:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7eb86a916d947b3afe3659d71d07c4f : SUCCESS in 38m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4cf400b12c884b7c8b25fb5237b545da : SUCCESS in 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3709d0572b88010820609e73723656c16dd421fc","author":{"_account_id":25058,"name":"huangshan","email":"huangshan@fiberhome.com","username":"huangshan"},"date":"2021-02-22 02:25:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8642a45948f6c6a66602b2a8dd4166df44ff86d8","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-22 02:50:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18ed40109f6f339b7d52141307c734510fa15575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 04:43:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a6aaa99de6f84bee8ff7b58ababf2ae7 : SUCCESS in 3m 40s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dcf242132b65420d9ec9f76f5ea56826 : SUCCESS in 41m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1ff676765fe24be8b23b19585088c414 : SUCCESS in 1h 39m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/24c34583d0cc45528a673ed81d2f9c0d : SUCCESS in 1h 03m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2277ec1339204060b431e8bee960d46d : SUCCESS in 56m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bba2f3cea354dfe9036f04324db4ca1 : SUCCESS in 23m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/343e36c4ed354e5f89ab4d7126ade621 : FAILURE in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7461821d3824f138f72993ef76403b5 : SUCCESS in 9m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/434eb94db3214458ae9847290235fa43 : SUCCESS in 19m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3172be5594b5413bb9360dff293c37c2 : FAILURE in 17m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/286e66f2df5f45edbfda89d97fd1f5c3 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9125c0b895d6407ca966496daf0049ad : SUCCESS in 14m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea63950c27e6488fb079d3df96f76097 : SUCCESS in 39m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa943edbef7b4da1872b6a14c10aafc2 : SUCCESS in 1h 01m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b7eb7e4f996a4f45b927788e6510ff42 : SUCCESS in 1h 01m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd9e5fcf2a8d44699a0df2754b3ce517 : SUCCESS in 48m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e01e893b805345d2847016adf05650ae : SUCCESS in 48m 10s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/25f386dd528b46f6be980c03bad77001 : FAILURE in 1h 05m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ec49a82ad5214eceabadb0e85bcada96 : SUCCESS in 1h 54m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b9c9a549855468490511070f9bbc8a6 : FAILURE in 1h 26m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a52702aab5e747c5baf1ca0e55707d72 : SUCCESS in 1h 03m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cad9e57af7ad417fb78b5e7edd330ef3 : POST_FAILURE in 1h 29m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e7f2343142854cf5b900bea0f62c60e1 : SUCCESS in 32m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/501be7f2af9747389bee972858bd8ba4 : SUCCESS in 1h 53m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7f4ec56faaa4470ca72092481722211c : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f9e256e7b2f36ec2d7a629ec49135eec33e7f1fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-23 01:43:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9dd4c43e9eee2fe5adae29a574321336309dc0de","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-23 02:37:45.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776567/3 : FAILURE in 48m 36s","accounts_in_message":[],"_revision_number":3},{"id":"6deb8496f1251c528a2fd1d1ff6ba0fda25644ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 02:58:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/281b008d092e4ebd8f68cd6bbbad06b3 : SUCCESS in 35m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71aed23add5d4cbba0a255acf81e33a6 : TIMED_OUT in 1h 00m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f2ea5c7f33121425cb906eb8e8be44ea4b94028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 03:22:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/143133ddaf4442868b3ce6d7653a3c15 : SUCCESS in 3m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0a8877c5d5d64f2db5bcea97d5593684 : SUCCESS in 1h 01m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2af3b5d0958e440599d9370f1405ed29 : SUCCESS in 1h 21m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/324f19d6a1ec4ac2b6c4bd4a01f1a4cb : SUCCESS in 1h 10m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/673e666e711f45f5947a6bb40de08577 : SUCCESS in 1h 13m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2297fcc7ceec46faac1bdaec2d5898df : SUCCESS in 31m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82ccd600a54b4056a6e71a28b73cf76a : FAILURE in 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04504cb75f7543188168e029cc62f22c : SUCCESS in 12m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/266a0b88ffb946deacc7310cde106bab : SUCCESS in 28m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04207b1deef74a9e995758fc0cb5452f : SUCCESS in 18m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/149c5f2260024b688bfb13c1e7f25493 : SUCCESS in 18m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5614e3e95752497aae95f7b8f885fae9 : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b3d543847914c01ab37435100acffbb : SUCCESS in 48m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/64aa1e46a3fc4a9fb181e3cb8a7a00ce : SUCCESS in 1h 06m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/172954e8357346e3a82f0d72d476b468 : SUCCESS in 1h 32m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c557304416d34f9e98c21c4feb3b095b : SUCCESS in 50m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/278ea290035a42c7ae40668099d28cc4 : SUCCESS in 1h 01m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/630db875440f425dbd954189ac3d7047 : SUCCESS in 1h 16m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/868cd4a6671142a18909c6c8154f01f0 : FAILURE in 1h 16m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/614ce42623c948d8b10436ae24b31ce5 : SUCCESS in 1h 20m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9c47e529ec14e5f869d97fb5151e5c9 : SUCCESS in 1h 05m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/59b7f659155f4e11861046a9304489e5 : SUCCESS in 1h 09m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1b72a2a637c64cd49fec85ba8315dfae : SUCCESS in 35m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4d1b19c54fc54c32a6f5a7f2f5629101 : SUCCESS in 1h 25m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2ab7201cf0a344b6ace426a03dec56f1 : SUCCESS in 1h 16m 20s","accounts_in_message":[],"_revision_number":3},{"id":"5f4039a4a85fae5307351e2030d78f8491eb8412","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-23 09:46:22.000000000","message":"Patch Set 3: Code-Review-1\n\nLower constraints job\u0027s failure seems like may be related to that change. Please check it.","accounts_in_message":[],"_revision_number":3},{"id":"fc247cac43326ff93237506eec1b28c7725ea4b7","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-23 12:03:59.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e Lower constraints job\u0027s failure seems like may be related to that change. Please check it.\n\nYes, I\u0027m working on that.","accounts_in_message":[],"_revision_number":3},{"id":"bddf03c4347a42fc654b6f79e96446110cd9e863","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-23 12:08:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"afbfc0f6df58d59e76f8887d7424ea0eab9f4cf4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 12:54:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc3ba55adcab444f8eedd7def4dbe504 : FAILURE in 35m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b7e3a78700846ffae82f5cd3c819923 : SUCCESS in 39m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4700d4466a42b1eddf07dc7abd25f7431eb81aa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 16:23:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9aae817291124197af9a1274e8c0c76a : SUCCESS in 3m 36s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/820188cffb5c46f1aab68b5ea22a2483 : SUCCESS in 1h 07m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b4dea66d628c472597a0e00ad25a4d91 : FAILURE in 9m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cc360fdab27c4aa5ba587b61d08f9737 : SUCCESS in 1h 57m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ec4f0019d78c4df49ca393a1d6ca5cb4 : SUCCESS in 1h 22m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dca478733fe4d5f8f1266ee5ac006ea : FAILURE in 37m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/465bb13ba9bd49b893d7ba9dfce9af3b : SUCCESS in 31m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa32dd6eb2c418d8134aaa164a151e3 : SUCCESS in 13m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6fc604acb114266966875ecfcc90156 : FAILURE in 35m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab6db2a9b9b347a7b6684ec354debb71 : SUCCESS in 37m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57fbed0abe974d939e2d184953638275 : TIMED_OUT in 41m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ccafd70b7d42ef9a34d8da079141d8 : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/21a7afc3939642c5b6552fa52e4cefb1 : SUCCESS in 45m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d47931107ff84fdd896d2d52a4aa6c7a : SUCCESS in 1h 09m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0da5b4c68c094a7e897e2670fa216b81 : SUCCESS in 1h 29m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23de384d9f65476c812654db7598f811 : SUCCESS in 59m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/565c81ed7170481094b9a124c3eae338 : SUCCESS in 1h 02m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7fb8a7332375444d8b39f31dfd6a720d : SUCCESS in 1h 21m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/defe025b9e614dfab412b412b42660c4 : SUCCESS in 1h 28m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cbc6c9cb8261426b92aab3123bc647b4 : SUCCESS in 1h 34m 20s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69ef23ca271e461bb5e6978305906b75 : SUCCESS in 55m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ba6ad529c6c04a3a8b884ac354cb1876 : SUCCESS in 1h 09m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ef5a7a06d9364dfda8f7a1f51a04a037 : SUCCESS in 34m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5ecdf63f54e4f9d9d93608c29301012 : SUCCESS in 1h 06m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4ed64c7d8f0c4708a5ff9cfedf2ca905 : SUCCESS in 1h 30m 31s","accounts_in_message":[],"_revision_number":4},{"id":"b4c69ab6971d83caa7ed76bc3ec96555656800db","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-23 22:34:39.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"00408b74812f3711ac48e9d7d88aad51df40bd15","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 13:10:18.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/67/776567/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"04fe5516e7abb396b776900cf974f2773d956061","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 15:53:10.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/67/776567/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"60ef6f489ea943c9256e3a777cc9b608731467cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-25 03:11:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"22d9a2cc97541cb5a87a1f325840509daa3a4699","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 03:31:57.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"f13ba321fed25bafbbce1854e52f0a86476468ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 04:06:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbb70a9474fd437eaec50f5298e0ebb6 : SUCCESS in 42m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e37185a5b30a495191959483f6735e66 : SUCCESS in 36m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b27cb5875bd227f260378cea7d464462abbf3316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 05:59:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0b451371b8294055ac2db0dee0dc203d : SUCCESS in 3m 33s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7d76e36e0e1f44fbabd621edb095b41b : SUCCESS in 56m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e410a9488aa549dc9e75f97797afd70a : SUCCESS in 1h 27m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d566e262001243f8be91a80999817cd5 : SUCCESS in 1h 34m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/411c7fecdb134733aca311365a2a3bca : SUCCESS in 1h 49m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3416db19b14240638efc1f339f478a8b : SUCCESS in 53m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1f0bf89ffa54169b0bc09ac616f2356 : SUCCESS in 36m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32285e2a0189497a8ed259bf781c042a : SUCCESS in 14m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56cabcb22fd34d5f99747a99a393fd05 : SUCCESS in 41m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4803819a474e40bf9b99fc422601e298 : SUCCESS in 29m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9c7526bb3c64b1f9dc93e604f87d754 : SUCCESS in 33m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf6c5edc6e7483f8c7344d5e528fe2f : SUCCESS in 10m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b894fbf63b8e4a74870d63f91be5f5cd : SUCCESS in 49m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f39152fc06845859e10b278413a1e70 : SUCCESS in 1h 09m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/97eb7defc7e64dfe9d30e8e18685b68f : SUCCESS in 2h 03m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/840cefc1832d48e4bd90b32def5c875b : SUCCESS in 1h 02m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/833f662121ab4fc1a0c070a8b35e5a0d : SUCCESS in 1h 09m 33s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e0f06c881fc14d03a38eeb398334a848 : SUCCESS in 1h 21m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/980fc27382a94f05ba49fe50bf69d33d : SUCCESS in 1h 26m 50s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eeea5a23fb36449d8ad15f8dd92d1cd7 : SUCCESS in 1h 42m 23s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39d1fd2e590a430398ea5a63e229d264 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":5},{"id":"917b47f273605f6d1b6b075b9b98439a30401872","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-25 07:21:49.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/67/776567/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"5b77a70feb89066b14573f754fb3875750cc532f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-02 06:31:20.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"3a4f1005f35b6e3cb2fd4b86688af06993c5c5f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 08:08:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/350e5ee032264513a26405d3d549ca09 : SUCCESS in 38m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f6883d1af17447280e798dc025fc0ce : SUCCESS in 39m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3df8e9c93b6d861a2da202769d632a625f0ff96b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 08:44:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/197dc13bf8f9473a8032c72a9f71b2de : SUCCESS in 4m 19s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e727c5e748864fc7954261980ce05ee9 : SUCCESS in 50m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/88dc32a814c6449a9ba13e71d6cc2fce : SUCCESS in 1h 22m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/74086528c6bf4b94a10daf1b34a6842d : SUCCESS in 1h 28m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cba1bacb0a474899a4a0cd41d4c5c3cf : SUCCESS in 1h 45m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66903751a1dc4917a3c1a119a8cf86f7 : SUCCESS in 47m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a30b998338c144dfa37463ee5446d1be : SUCCESS in 34m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8013552204e9461c8cfcdf9ba374b27a : SUCCESS in 15m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70cd43833bc941c8a5e562d98b934eee : SUCCESS in 27m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/643f267fcc3e4ce8acc9235aaac332f4 : SUCCESS in 27m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f45b001f58e94c8f9250561133c20fda : SUCCESS in 36m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d39910f933459c932eaf222fc7a218 : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2031234cb84c4f27a5b8b1f954afdd56 : FAILURE in 45m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcf15201ac314c5cb13831f1382e40f7 : SUCCESS in 1h 11m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/38d904717ad14fecb7842591ce4b43bf : SUCCESS in 1h 41m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/95be628e401f4e7aa119c26e6430d51f : SUCCESS in 1h 05m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/484b914a527940859d845957ca70da25 : SUCCESS in 58m 55s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1510ab9ffa334d67a3193a067b79df2f : SUCCESS in 1h 44m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0cca3a7b23d44dacb10c832ef8c922c0 : SUCCESS in 1h 30m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/39a7376fd7dd4d1184c4d671376ac9f9 : SUCCESS in 1h 17m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcab8abad6bc48edb76407aee840757e : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":6},{"id":"fecb7972bd00c42641844e2417169fb510ca8ecd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-02 10:46:17.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"28d503fc48f6add7595e41f323040aea43249443","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-02 12:24:08.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"9e62818ad9f0ddb820784b2e209c8145315b6629","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:13:53.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2d18307b53944909c4c9b25c3a80436 : SUCCESS in 42m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c2d2b5d1219443c9df9845db8ec7421 : SUCCESS in 42m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a77a98d741ebe5d0648e46ff755d905886114cf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 14:09:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3ada8a3071f741c5b20fe6a863d4dd4b : SUCCESS in 4m 39s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/212dd7cee7a543b29f876cdba3587127 : SUCCESS in 47m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/64606aa642d448f89bb4a0e95eb52e7e : SUCCESS in 1h 15m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4ccf1828e5954ee590e08a116ebceea0 : SUCCESS in 1h 13m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e83bc02f2b8d45fd8c9461cf286682f3 : SUCCESS in 1h 24m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1244d993b3e46e28c4908643257d801 : SUCCESS in 25m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25bb3bddb9534fb2ae7c643121dfa67e : SUCCESS in 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4ae3a14062c4f3b8c23433dccc283ce : SUCCESS in 14m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79d5305b51e44c959457370de271b331 : SUCCESS in 30m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db3716009f3245cba728b633284e8574 : SUCCESS in 26m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eb489b2f2f543f3bd72b6335711ef72 : TIMED_OUT in 40m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1aee62cf834032ab2247283f55c4c4 : SUCCESS in 11m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3890d11ea6f42198a5c6cce5371f4b1 : SUCCESS in 48m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/67995ca6de744b52be28f89f6bb0bda2 : SUCCESS in 1h 17m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0d974cbe395c4b20a6da0c70facfa9b1 : SUCCESS in 1h 23m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b6d81df758e4d4cab335d728799f0a0 : SUCCESS in 1h 04m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/910034d19e574696bafe419f0a2e0410 : SUCCESS in 1h 03m 29s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7aee2790dbdb4f0da8efc68786772e3e : SUCCESS in 1h 20m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c2eb015630f0405698a6269c7c8497b0 : SUCCESS in 1h 29m 39s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f7732a57c944cbd91567ad9fd8f1796 : SUCCESS in 1h 43m 10s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/052c3242831e43e1971205b0a961a657 : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":6},{"id":"a87e841abc553d2363d4cd1b048eaaecaec501fe","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 09:07:57.000000000","message":"Patch Set 6:\n\n(4 comments)\n\nWhat about release note? Will You add it in separate patch on top of all others?","accounts_in_message":[],"_revision_number":6},{"id":"a7fc24ed56eb1f1875fd8a82a05f399dbc1507ce","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-05 09:19:35.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ef5a44b76484d1e00e052652011f619d446a1b06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-05 10:08:25.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"ea2d76f75547e281f19414fa446acb3ed4d7dc8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 10:52:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c8248c5e27a2316530c0fe5333130b912e8007ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 12:23:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d6ff53fd0d69448b9be6054c4503b03b : SUCCESS in 3m 33s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/34475ec8e49c44739190b058c4b8a724 : SUCCESS in 43m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/84e349d0ed884647b20d49e1c0df3e22 : SUCCESS in 1h 42m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5a762b2193084f9c901ac1ca004bf855 : SUCCESS in 2h 03m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/50e2796228ac4aa3beba9aa2c3fffc49 : SUCCESS in 1h 50m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/597a7b362c2c41de99579dc547ba3026 : SUCCESS in 45m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/babc93c8aac64957bd13888dfded4a79 : SUCCESS in 32m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c19d2d42394449bda53374f1815479a9 : FAILURE in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95c453d01e33488a915f8c55793fd8fc : SUCCESS in 30m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e50235be3254f1eabd59ceb82e6a8fc : SUCCESS in 37m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/565433e83b9441d49af534ca40940382 : SUCCESS in 32m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01bf41f468c42f9bccc03f73102aeb1 : SUCCESS in 11m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3f4ec582e6743949ae776d5b420302b : SUCCESS in 44m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e1f1e7cb3f9459b84ad76ca33540442 : SUCCESS in 1h 21m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5d39a71bb35640fa83e935c6f78ad86a : SUCCESS in 1h 31m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c9a2907e46545ecaa15e64485ea0a99 : SUCCESS in 1h 23m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e10aab8aa54c45a49b06aa649a19e87a : SUCCESS in 55m 40s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b84a3f75e7894e8fa59302399e8366c2 : SUCCESS in 1h 35m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/00030bc56cb649ef847cd72263645e8b : FAILURE in 1h 11m 45s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/62ab1cdba9824171a6e70b41447b8560 : SUCCESS in 1h 40m 20s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a5992c394404e7bbd60e6f85b133f9f : SUCCESS in 57m 53s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/agent/l2/extensions/dhcp/test_ipv6.py","accounts_in_message":[],"_revision_number":7},{"id":"85ab85260b730788a781a777210413bda2efeaa9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-05 19:09:41.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"342ed6f4c43a3fcb13f723418c28ac50c0d3b7a0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-08 11:26:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"74668d01a7a94480b1d7eb56d9f39abffdc79766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-09 02:02:53.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"13f2ab5eb07ace7ccbec62bb303430035c467acf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 03:11:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ba612481734410a9f5720356c184c3f : SUCCESS in 48m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8a79549911640799debe948e103c4c4 : SUCCESS in 55m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d29c0389eb2328c31f7c6990c2495f75d37e2a39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 03:49:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e45e9ddc9573494ea1306279858239a3 : SUCCESS in 2m 32s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3f493343990b417c8da2f17672f0a743 : SUCCESS in 43m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0dea6da7b7dd48c39e52631bab1122c6 : SUCCESS in 1h 32m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/da83233a607d4de8933e01aa6379f739 : SUCCESS in 1h 32m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cd3f81a29cdd45a7a2316fd1cf64c0bb : SUCCESS in 1h 29m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31a13429563046648aa393abaef1d924 : SUCCESS in 42m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a3b6249c5db44e1ac5f71a290c0e6c5 : SUCCESS in 42m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43c22defcb584bf8afa1440859d7902a : SUCCESS in 14m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d13a6cbba388484f8c4c1042cf2b3b32 : SUCCESS in 31m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e2a1013f12a4692b1665d2e7ab25714 : SUCCESS in 40m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fab909222c604a3f83397c4d8c57b7d8 : SUCCESS in 33m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb308cf0ef244a7aba47e79768a0702 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d96a23670fb34853ae47787c31d53bf2 : SUCCESS in 52m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea6719a1d1d640d98fe18e2893547d54 : SUCCESS in 1h 25m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2ad10f163088439ea6b0995f07ef2f70 : SUCCESS in 1h 30m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69743a1fd941497eae171cf75fddcc1f : SUCCESS in 1h 14m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a3a15abcf11e4d52bb9f61eb0c8f8505 : SUCCESS in 1h 14m 02s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/adeb850c2b3f448f886089a6dcdf2483 : SUCCESS in 1h 22m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ecc8cfb0bc7949d3afc642c384df29f9 : SUCCESS in 1h 45m 00s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f7eef2d10c264aec9008cdddd1ff04bb : SUCCESS in 1h 35m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04c5dfd2db0e4caa86b3e9ffad84bd35 : SUCCESS in 1h 14m 20s","accounts_in_message":[],"_revision_number":8},{"id":"90143ee2af39d1c8b4fc14a560d0342d990ec128","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-09 08:36:01.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"d9f43c0829c81769041d225b676529c29401cb83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-09 09:06:15.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"8fc0f4095442662a89752cca1c40e0e5edabe5e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 09:24:55.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"77d7c30ea8284f71b30c0c8d7cec631d5b98bd1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:21:17.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/66cad242296c4f0bb7c3d366d0abddd5 : SUCCESS in 2m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/567cbc1a2f1f456aab94b11d5035b31a : SUCCESS in 55m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2df12acc3431453e9e1ff7cfe049d5ad : SUCCESS in 1h 28m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f9542208453f44c597183ed05d45b949 : SUCCESS in 1h 23m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6da2afd82c2c4553ac20546e03b81954 : SUCCESS in 1h 21m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91a819e5db9c4af88d8c912abae9c49d : SUCCESS in 38m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbae038b6c1549278647af2276cc4176 : SUCCESS in 34m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dd9316338744e4b817ef70ce44248fd : SUCCESS in 15m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3abd266ffa5c42edace1259043e92943 : SUCCESS in 32m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b88e04d301243f79c910471cc7dc734 : SUCCESS in 33m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8325c56fc23a4449a9fc1f37be63dc91 : SUCCESS in 32m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c2f12de01c44623bd172c3da161ecfe : SUCCESS in 10m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/64bf92926ba64b8fa4d8ec8921e70579 : SUCCESS in 53m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd0a4b8239c946bfb06343ec270751c6 : SUCCESS in 1h 18m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fcd2523e13874ad5b14becb7cc96d294 : SUCCESS in 1h 36m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5ed45e40103488fa13909fffac52ef3 : SUCCESS in 1h 35m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/96202a36ee654a21af278e4cfcff8d5f : SUCCESS in 1h 15m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/73b9093b202b4afa8764d9af90093d7c : SUCCESS in 1h 18m 50s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/142de3e8723f4748a33a731a31da44ba : SUCCESS in 1h 41m 15s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fe0e5503afb445aebb8c312d0be06abc : SUCCESS in 2h 06m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ace4c93f570148e98356e9358dacd08d : SUCCESS in 1h 11m 14s","accounts_in_message":[],"_revision_number":10},{"id":"162ae75cd5106fa5655c1871266d5f30b96e0a7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-10 03:03:31.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"fd9bd5256814f40abb127d571c9ec9b31297acd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 04:02:06.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cff74e72bdb4052890e1266a1f52216 : SUCCESS in 49m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3889a21d07754f178bf258ffab2bce33 : SUCCESS in 41m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4595c393eca7dae50ce9c9477be7d18ba3f83b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 04:50:14.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f8f3393949b1409bba5f9d8ea7bda8be : SUCCESS in 3m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bb889270f9794f2e9ca30aa4cc1fa748 : SUCCESS in 47m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f0a038de7a934aa6abf6c54a5e1e4839 : SUCCESS in 1h 19m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6f123298ef584b85868e476e73c94ce2 : SUCCESS in 1h 16m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/71db5336a08742d7aaf54af4423551ac : FAILURE in 1h 44m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a64d40093582450ab2f4703b45e5faf9 : SUCCESS in 31m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5adee38a37a5498082a767645f7573ff : SUCCESS in 22m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7f310906b9240ff8d6d16f12dbdc00a : SUCCESS in 11m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2200a8b780474799ae018111f900d7b6 : SUCCESS in 23m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bff0d937a1747208a0234b70b9e99b4 : SUCCESS in 28m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1df5d847f3b64d86980b5a998c1bf398 : SUCCESS in 27m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d12b25d6ab93487db772be1c2e23648b : SUCCESS in 9m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/515f481fef524b0f8b588f7d3f2f33cf : SUCCESS in 51m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e6e0e22731184ba083a2259aea36c05f : SUCCESS in 1h 27m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/79c939d2e1a946a2aaf9526ee22b429e : SUCCESS in 1h 21m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2b885920e4243a58d0bb0c5ea7c1764 : SUCCESS in 52m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/21409b59155641caaec04c0af81176ac : SUCCESS in 1h 00m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c92222024eed4d2a8c76ecdec5f71500 : SUCCESS in 1h 14m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/83c31305e1aa4c598437ab577abbb041 : SUCCESS in 1h 23m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be35a18fbea04d75844192708754690d : SUCCESS in 1h 29m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb8cf2679a664dffa2b416be5e7f30d2 : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":11},{"id":"fb795f6e85b53719d4fab40ec2721bb641a1d981","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-10 05:29:35.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"1aa8f72c0618448d809696abf278526877a5e58e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 06:26:56.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b18616bddbe34607a86f31ce39f7334c : SUCCESS in 46m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b6a6c54689349c4bf64d8f8c9d4d6f5 : SUCCESS in 52m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3b5260a48d555377208e4d3dbdb57dd91a86ec32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 08:26:47.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b7fac4e433e477986ef409b6e7d0c4e : SUCCESS in 2m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8785841b0f47438b960357eacaa2f4b4 : SUCCESS in 47m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8a48c6c740894f47a3280bf3da30a620 : TIMED_OUT in 2h 53m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/99bed46670bf48cd88008f3b52c055d0 : SUCCESS in 1h 23m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0789fa86baac4d55894c233e20c35b10 : SUCCESS in 1h 29m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69b5a729601047428b9d1c9efb952d05 : SUCCESS in 48m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fbaa58746084052943292d5488b4cad : SUCCESS in 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d1955065bcb46db8e9e87008ab14c7d : SUCCESS in 17m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70fe9019602443ba90f9878964d358eb : SUCCESS in 46m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db48a23a69334f62ba69d5639de6eb77 : SUCCESS in 24m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea1e3d7db4c34bcbb5c96cfb85d8bff8 : SUCCESS in 28m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af258f07c1d040898a3261d5c73981a5 : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1cc9d18741fd4fd8bb46d34463e8af41 : SUCCESS in 46m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/50ec12201cae4c35b3e53d98707134c3 : SUCCESS in 1h 24m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/208984779393444da8a9e77dcff87356 : SUCCESS in 1h 36m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ad976735099482dabc82a567f75f556 : SUCCESS in 58m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2df0e83701f641d0b05a2bcbbe7bc050 : FAILURE in 1h 01m 54s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3ac16341021a4c0192a7eebf0f3ded26 : SUCCESS in 1h 39m 38s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dcfa6fd504d2465d8f96003a30e3034b : SUCCESS in 1h 27m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16157153756842f9a250a859509b0c3f : SUCCESS in 1h 42m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43ca0b05321445c9861549078e78ad77 : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":11},{"id":"dcc78f3ba9f276d23d09d20ecc481e69963828b1","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-10 14:25:54.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"f8618ddb82afa4ff484abd1e904ebeb440e613e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:32:42.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b9e9e208135475fa265212c640a298a : SUCCESS in 53m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f1771dc4a954fb6940680a94dc18a01 : SUCCESS in 52m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5f28c5def985def336f4c119060f208a071d636f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 20:02:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/202540e5d151406599fd5812c29eb2d3 : SUCCESS in 3m 03s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b15b4027b1b440be9405f3aa671edc4b : SUCCESS in 55m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/05e495d17aa44008869d2ef82551eb6b : SUCCESS in 1h 51m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f898d311841e43cba87d80166b8d2501 : SUCCESS in 1h 24m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/647d3615997b41158dff93fe383b5cfa : SUCCESS in 1h 18m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b16bfdabb8d4f4fbe6e92aaf01cd54b : SUCCESS in 43m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55da4aaff3134f409dc5231f2c54defd : SUCCESS in 58m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39888c674b3548b1a17ffa1286b46c02 : SUCCESS in 12m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d55e51335a449148d82af3ca55203ab : SUCCESS in 31m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64371117168c44a8ba1128c1b3d9615c : SUCCESS in 41m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e0b14b245d542858569c50a4d6e4b90 : SUCCESS in 40m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1808bd952ee944038c74bd00b62c2de6 : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/70fb6eabdf384db192483a6a96bb8be7 : SUCCESS in 59m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2936029ff25b40cda937c0751657bf76 : SUCCESS in 1h 25m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0aa6d798324549fb9e46aadd609a91d0 : SUCCESS in 1h 21m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/09725fffd5b741a5817e937782ae4a3c : SUCCESS in 1h 07m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a12933b29a724a7fb070d83c87b1ac77 : SUCCESS in 1h 06m 58s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/161771a7919e427cb1ba911d6de968fb : SUCCESS in 1h 57m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/deebfdfd72874046bfc48f0889fc29ed : FAILURE in 1h 11m 33s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1ac34ff22ab445489366f34b31ccc41 : SUCCESS in 1h 46m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cec612d948154eb5aab617ed7009c11a : SUCCESS in 1h 33m 33s","accounts_in_message":[],"_revision_number":11},{"id":"3548aa8b9823995ed9e3230b49ea300e06246a7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-11 09:32:43.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"9b7e43c899fd51d95a36f0846a0e3deb8d837991","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 10:29:19.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7e68e7febc64beaa5fd31ae4649749f : SUCCESS in 43m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d3b2e1dffe2453690db22d488faba5a : SUCCESS in 47m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"36751403d77e5b47d49485bd9967326510492748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 11:58:42.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5070132975a04085ab1bb946b6b499b2 : SUCCESS in 3m 20s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e771894500df430d8fd334684b5a8e5e : SUCCESS in 54m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/763661f99de5419c80cbc02b13129769 : SUCCESS in 1h 37m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1661e64b0f724a28b107598a3d89e4a3 : SUCCESS in 1h 17m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/731421a098ec4956931e006e8a30d187 : SUCCESS in 1h 44m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8b55dd881c3411182863f22283465ac : SUCCESS in 26m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fe0fb1a084d472eae4a08229b754c09 : SUCCESS in 29m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6364b354e3c24d02982a700276b3d004 : SUCCESS in 13m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/691129390f0041f3b90a1890351a8ce1 : SUCCESS in 31m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe3dc3b40e1f40789da2d6ae23e683c7 : SUCCESS in 33m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d7b399198924c3cb8208be940313e8d : SUCCESS in 41m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3556e5a9f38413eb6bf2691357d93b9 : SUCCESS in 8m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d9080afd3234087a05f612cf0a81e3c : FAILURE in 50m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/835ead1b73d744508214367df1219dcb : SUCCESS in 1h 16m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/289ba8dc221148938201e9125f342205 : SUCCESS in 1h 02m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4822a2b3e1644352992987672de02d74 : SUCCESS in 53m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/89f616dcb44e47e79916b6c031941592 : SUCCESS in 56m 52s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9150dc88c2754843913d420d4e399335 : SUCCESS in 1h 16m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/feb05c3d47b04cf8b300673cf02fbcac : SUCCESS in 1h 40m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad80324b6d254ad8a1c70572adb7ab43 : SUCCESS in 1h 41m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9b29cb42b2a47bdb098c5f4943c7807 : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":12},{"id":"ec7e1d5a67566cfc6e4ecb48f8e4655a53e05b04","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 20:01:16.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"976afbca30b09b3a55f6371ac29e24caf559c09f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 22:09:57.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"863592f595d883eb33e9727aa613ad1bc46d1e28","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-12 00:47:53.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"5841edfec443c5452e2eb3a6189dbeca129b5e44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 01:34:03.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d9cb20096504382b9483a27fd1e391b : SUCCESS in 44m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30a7dad402514bb0a880c02e778bc7b2 : SUCCESS in 39m 53s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a66f28ef5d78df8528c2a830b28be068902a5ea6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 02:33:50.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"15ac092ab05b712393674fa680ff21989445b72c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 02:50:43.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1a2279105ae948c0b4b248cf3a7174c3 : SUCCESS in 2m 48s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3e8b01abadc04970b2047065a10734cd : SUCCESS in 39m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a85055ed30a147e5a98362fb9dfb9b57 : SUCCESS in 1h 17m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f80dc8eb5d864d0dafa983e08155b28a : SUCCESS in 1h 43m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8c2d54428ccc4fbe84a81441aa14b05d : SUCCESS in 57m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2ce068c6d2e45caa9dd7f96f0064341 : SUCCESS in 43m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaf7789c27b74d32be8db8173feae9db : SUCCESS in 37m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a490374c0a4a5d9238827172159ca1 : SUCCESS in 13m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9806bf1df330450eaaba6533137b3953 : SUCCESS in 35m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38caea0d83ae45029a693a8b87240286 : TIMED_OUT in 1h 00m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b28e73d450804c269bcba5780830f3b9 : SUCCESS in 35m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/546ef721fbe2411189e275c0e6eb304c : SUCCESS in 8m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b2b466bcf3e4348b49333d7159e5ce0 : SUCCESS in 51m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/33ad17c6686e46edba215989bf009a17 : SUCCESS in 1h 31m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a45b2b85ec184637ad37c279712998f9 : SUCCESS in 1h 30m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/59f9b367560545418b6a55e52333410b : SUCCESS in 1h 04m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fe7f8353dd144690a68f6653aed9c9f3 : SUCCESS in 1h 07m 21s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b9d557dd76cd420cb94140a1ba1527eb : SUCCESS in 1h 30m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a938e5eb34284c22b4238c48ec349a7c : SUCCESS in 1h 45m 38s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/db7bbbbc36b946c9a3ecb3e76f8fabfb : SUCCESS in 1h 55m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a86a137b71dc430a853b645f656f47a3 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":12},{"id":"6dac76e66ecc1278701f94a87ac8c8b7b276bad9","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-12 06:25:35.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"e1b55a413a7270bf46a924d1a9235279ea70c7d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 07:19:05.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44a1870ab4eb47c79088d8da7bc89c93 : SUCCESS in 47m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5d0051943bf4e25a10d2e07dd582877 : SUCCESS in 45m 58s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d4ac2eec44e4a8e5d6cd24d2c4f1953dc88ac2d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 08:13:44.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b765615292c04a599ba3769a007ab695 : SUCCESS in 3m 44s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a48f34f89fff408fb523f101a155abf1 : SUCCESS in 52m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/816d4c40057d40799bf4237b8b679e93 : SUCCESS in 1h 30m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/efcf62d44b2645fd8dac2c055170d7e7 : FAILURE in 1h 20m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/62acef76666f438c8b2b7d5184287bb3 : SUCCESS in 1h 20m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/540c0fe745a4495599bbe7a6da12013b : SUCCESS in 44m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b447cedd614407193b4b7b1eb1cfa05 : SUCCESS in 28m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce452aa2004473d9afb4f1f6bfd8f76 : SUCCESS in 12m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb6303f279664b4bbd5bc74c19032393 : SUCCESS in 33m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9fea6dab11e4262be328357c43848f9 : SUCCESS in 30m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2148322285104f068b7122a018a5cf56 : SUCCESS in 35m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44142217a81430a8aea4f526ab1e7a4 : SUCCESS in 11m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/64778fdf11e14199a4d5da3348ef032d : SUCCESS in 52m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ab2a676fa5e4110a708b965071f8603 : SUCCESS in 1h 19m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fd70a89005dc401aab543783b228d46c : SUCCESS in 1h 28m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbb3326409d64bf68aed35df16e21dfc : SUCCESS in 1h 01m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7fed4e2ff70149779b3761dded54e7e5 : SUCCESS in 1h 03m 13s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ae4c16d881df4a6db4d658a6f40c9c78 : SUCCESS in 1h 14m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/44c12f7f846741c0b08df5d431e35abd : FAILURE in 1h 13m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/60441497badc4370ab1dbd664d197827 : SUCCESS in 1h 21m 11s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d2be75abb1e45648783ac322719e06e : SUCCESS in 45m 18s","accounts_in_message":[],"_revision_number":12},{"id":"553319a15bfedb1b28a825ddd6d1074f2a62f0fb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 12:39:10.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"d270cc22bf5b432afcbe1622da8bf5d1c5a83a10","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 23:14:13.000000000","message":"Patch Set 12:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"efe37728573b4a9c2eaacbc62eb131e7ed57ee53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 02:38:58.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"146b71169f273931ebed0ae76daeb23b161379fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 03:32:37.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46df46f0eb9b4075b8cbb9b328e5a4ef : SUCCESS in 45m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50ffb8300dce4a1ab3693e040c641f99 : SUCCESS in 46m 51s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"da2e3d1697d89fbfca831f40adac9a6341b7652b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 04:21:28.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6450fdda7d5945ae8a0469eb54559cb1 : SUCCESS in 4m 22s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ed4b638daf1d4a899fdeb09f69654814 : SUCCESS in 51m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1b8a13030eed4d2bb6c5f302f7ba0e1a : SUCCESS in 1h 33m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dafabd87c52c4b3f989cfa447cb3c598 : SUCCESS in 1h 27m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/37f2451fd319451fa0d8a60e5aa29f6b : SUCCESS in 1h 23m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46276661420c4ecaa64c4b5386dc295b : SUCCESS in 34m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcde908dac754776b21ab1c3f9b903b9 : SUCCESS in 32m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e95d84ca35a47a997a63c0a1823822a : FAILURE in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a08ff3e509644959fa8c8d591dd9232 : SUCCESS in 36m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86d3ba1be52f4d5db73abf3f73150f50 : SUCCESS in 34m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d5b2bf81f104e98926be6bc572161af : SUCCESS in 32m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b6661eb17a54fa1b90929619470aab6 : SUCCESS in 9m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2404e4248d744b89a13126683c2337c : SUCCESS in 49m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb09868cbbdf49748124688ed0b7516a : SUCCESS in 1h 24m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a38dde2ab9cf419c9f89601604f85a7a : SUCCESS in 1h 24m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4423cd46a269438a91cf59787200004d : SUCCESS in 1h 21m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/63f6dcfa7265402d8ec245a312c0de99 : SUCCESS in 1h 03m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fd0cf3de13ca4793a27d5b458c0ab8db : SUCCESS in 1h 17m 15s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a3ac8a117991463fa308903a6831b90e : SUCCESS in 1h 40m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b90d1102b2c4a2b9d21afd9f7927c04 : SUCCESS in 1h 36m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5129d39482394d49a13f5da4ba449e41 : SUCCESS in 1h 07m 08s\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/dhcp/ipv4.py","accounts_in_message":[],"_revision_number":13},{"id":"e0d4d88a8c3abfe06fdc6c88c05138b4b7938c97","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 05:34:05.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"32e746deb827962abd1e4d4fc00072f8ed851e8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 05:34:33.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"cf8f6ede28d9651ee0114912cdaefbcf1d170fb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 06:45:11.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80e7c48a0624411ba844acb6c615ae50 : TIMED_OUT in 1h 01m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2aefdfc8b014c7d9836c12712be3ad0 : TIMED_OUT in 1h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e4982b63ea59fa533b9f91671c0b18b091171fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 08:16:46.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6834e5cdd33a4d6ba41c9f800811d960 : SUCCESS in 3m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ba3eaa9a91e8445ba00494f58c6ee667 : SUCCESS in 1h 26m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/074c537a0f1946d68df3bf328ea4d065 : SUCCESS in 2h 37m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1f42753b18c1427bb09b4500ff33092e : SUCCESS in 2h 10m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d71a6ba1769f4a4291b1e4376c320b71 : SUCCESS in 1h 12m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0989b373156a41c681f0875ba6ec0df2 : SUCCESS in 49m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7270be807ba648f998aa98e20510a8c1 : SUCCESS in 37m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38cb72b23d864c8cad88baf5c3897487 : FAILURE in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4659c66dc764ca7a65cabb612725efa : SUCCESS in 49m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de569c8c3f984561a006e861e6b6db41 : SUCCESS in 33m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f1523f0252e42a0a8f8e87a3a9d7e36 : SUCCESS in 40m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85dc47aa094e4d77a5f5e5fa939c547b : SUCCESS in 8m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d40a424eb02043eea992f6a7149329f5 : SUCCESS in 45m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a5eef8007b264726af26f2681cb4c2ed : SUCCESS in 1h 19m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7ccc477f7f9b4621b2a5bca793158284 : FAILURE in 1h 24m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4a7a46d949a40b2abf119b56ea34eeb : SUCCESS in 1h 04m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/207217b4f59e42b398fcb9802412a912 : SUCCESS in 58m 56s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4023381bb1da43c6b234bdd3172fd9ba : SUCCESS in 1h 14m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c29e83e27ca84f5a96be791dfd8fdf10 : SUCCESS in 1h 25m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/55aa49977a564b1681d9c0d2f037e434 : SUCCESS in 1h 33m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/311c07d768b946a8b4c8832328fcc979 : SUCCESS in 1h 01m 32s\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/dhcp/ipv6.py","accounts_in_message":[],"_revision_number":14},{"id":"0e2ac15270aa479564d60934d499c876fccb6e2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 08:21:41.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"b2ea17b3bab207280dd2579de513f2f5d3a14455","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 08:35:15.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4e1fd7f0dd5182cacd75b1475fa211ea391ec2b0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 09:33:13.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"06fecfbd7b6b884a49b61ae5fb50fec1525b3404","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 09:37:41.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"92ceac535f9d3bd5f947c6afadd394b82aa24065","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 10:46:02.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a14bb8458fac4fc9b86d19f395d16f12 : FAILURE in 45m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ecde1d2e88984e54b79b85f05044bace : FAILURE in 57m 03s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"000fc3413ce57897cdb7a35cc457a3afe23749c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:34:31.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4602141771e74921831e5695e3be27c5 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b080484664134c569486a508e0eab138 : SUCCESS in 56m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3326f83989af497e98805c2e5b59cccd : SUCCESS in 1h 34m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5a2606a59f954ccb98dedefc5950555f : SUCCESS in 1h 43m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/71d2b860b2ca4efab16444f078042ff8 : SUCCESS in 1h 34m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a84ff2ac0f54b5f89ff4491d782319c : FAILURE in 24m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f761d7e6b4d34b03b165bf33befe86e7 : FAILURE in 36m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7832aaf162cd42e2b60fe20c0cbc0b13 : SUCCESS in 12m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d73237562ba34283bcbc7b69fbb04c6e : FAILURE in 33m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75d9c94f9ea24352a0fc745af49d4b28 : FAILURE in 34m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cd6e99856a94977bbb63e379c24d4ad : FAILURE in 34m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e91eededf94e69b51df0107efd21eb : SUCCESS in 12m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/773846053dd24c3380490892c7bcdfcb : SUCCESS in 49m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c93afca8c8294f4a83d87bbd628fe8ca : SUCCESS in 1h 14m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/81c78ca2e48143a09c792492da7f0f18 : SUCCESS in 1h 30m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/85c65bae20df4ef2b53ba9b7254f76ba : SUCCESS in 52m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/19b1fc693b4e457eb770c46da202a2aa : SUCCESS in 1h 08m 32s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/727dfcd966574a00a98c3db2998f3004 : SUCCESS in 1h 20m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/059066a7a36642abae28e3883d22763e : SUCCESS in 1h 26m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8e5db1384574c88a3437f489e8e2de0 : SUCCESS in 1h 34m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/691442fb34f140e9870c296a369158b2 : SUCCESS in 1h 10m 38s","accounts_in_message":[],"_revision_number":16},{"id":"b091f334f21fb0b99dd91addbe096931278bb3b6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-15 11:42:00.000000000","message":"Patch Set 16: Code-Review-2\n\nWe are now passed feature freeze deadline for Wallaby. Let\u0027s wait until master will be opened for Xena release to continue with that.","accounts_in_message":[],"_revision_number":16},{"id":"57861c7616dce4e066fd0135b5d2722e526180d1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 13:39:48.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"1aa202a8645ce5f1683bae43c519e9280b23a1ff","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 16:20:02.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"ecfebfce62dd01bb9fb711b98202c5accd280626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-30 09:26:52.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"a711984271bd5b28f869a08da0d91739ac9f34e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 10:23:52.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/191b5ec898564880a2e06a2094b5e58f : FAILURE in 46m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2808b7762a2c4a39b7eade76e5187ae6 : FAILURE in 45m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"358e0d0eb873645da1c8b7ca8da33a6b3b1e57b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 11:20:07.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1fa07a1994374674a8c304a315a5ff2c : SUCCESS in 2m 47s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6dfd2dc0f70c4b72ad6499cc90e81d7f : SUCCESS in 45m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f6e6cab3d3b24e2da35a428a6cb39c5c : SUCCESS in 1h 49m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c8ed4e411c0e41d489846adba78f4f23 : SUCCESS in 1h 48m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f705ff0fde95442792b382de00037d4e : SUCCESS in 1h 22m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a23b07b6ef634e74bd27aa5c13c86076 : FAILURE in 33m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e8dddb4554d490a9eca513c1988bf8a : FAILURE in 25m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba88f7d2c3a43c3baf9ba37268ae9d7 : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f07a949cbeaf4047ada95f718018a8fa : FAILURE in 23m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f1dd9a7508f410fbd8ba3d46d56f497 : FAILURE in 22m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ce005a8d4474548a3866dc61b462a0b : FAILURE in 24m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3cf3e10acd74806bb1161f2391e79d2 : SUCCESS in 10m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c7ccecc6cf244b9abcea6bfce0b083ea : SUCCESS in 47m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/980fe1fd9b574b0f82bdaa178459b9ca : SUCCESS in 1h 21m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/00dbd8da980a416cb122cdd5b8021b8c : SUCCESS in 1h 20m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ff343e384994092bd6ee48225e6f361 : SUCCESS in 1h 00m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8d7aa2f365734c598f0f3365dcdb2f53 : SUCCESS in 1h 19m 20s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/199ff7dd2d7746d299445a934b816f48 : SUCCESS in 1h 29m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8a097ea7804b40bd91e30f8c672a7d5b : SUCCESS in 1h 38m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d55c0712fc324d188a4c048e5d5f0fc9 : SUCCESS in 1h 31m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f0f7a2021884ae5889ce56814095760 : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":17},{"id":"7ddcb35c615d84e4f896e45b052fd5a2921deb22","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-30 16:53:03.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"1edaf6fc053f4c82d41810422a6796b798643a00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-31 02:32:38.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"d6640e14193b36acb857ccc6d8db786cdf85a759","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-31 02:40:24.000000000","message":"Patch Set 18:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":18},{"id":"f6ea80b9080f755fcaa36a4b1db78008b9bec0c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 03:20:35.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e69dc14107514455b2e4d2180e1b7d14 : SUCCESS in 36m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/709dca7723a940efa621780fe3e57730 : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"2f94056403e8ea561d2dfcb917fa46387bcff272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 04:11:31.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f138ca90e4ce44d082df0cfcf106d608 : SUCCESS in 3m 30s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0624c3fe20644cc9880b099c781bc903 : SUCCESS in 43m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/96b115aee9e149e1bf115663b8abf8b3 : SUCCESS in 1h 35m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/14ed4dd60573407986a559d052450aeb : SUCCESS in 1h 28m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/30d3d0c55da24bf197f61c8a931dccfa : SUCCESS in 1h 28m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a739c20b1ddd4045821ab49413d96c88 : SUCCESS in 39m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b52f170ec16d4abf969d7e363ed3c19d : SUCCESS in 34m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2c2c8c5eab84552bbaac3d2cf39f871 : SUCCESS in 12m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f84d6e596805407c8c5fbfbd1606d46d : SUCCESS in 25m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/033267b8f5384720bb9f24b988d2cd8f : TIMED_OUT in 1h 01m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbd1fd7f40be4ea59e3be3479421c74a : SUCCESS in 33m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd804b2b4b344a9cafcd7883f15a1179 : SUCCESS in 9m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d32eb3fb42e94098b254bec60520581e : SUCCESS in 1h 28m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9bba2ff084f74f128ae04e6ab54ebfee : SUCCESS in 1h 25m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/17eb50d61b5a43ca9e08f600c7a3eea8 : SUCCESS in 1h 19m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/edd0f0624e1a4412a4c71fa69164f651 : SUCCESS in 1h 06m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/57dc0c26eb35488fb2b54f2d9466e758 : SUCCESS in 1h 10m 36s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ee639aa7f5d9452f8562fbc67500faa5 : SUCCESS in 1h 09m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fdda65b995664938b0082ab088c7ba3f : FAILURE in 1h 13m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/973276a916dc4b13b898d9147a6752ec : SUCCESS in 1h 34m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1573ad07b3934c52832d996ad27c01b5 : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":18},{"id":"9e570881b05e082857bfee3a036414be64792d87","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-31 05:37:26.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/67/776567/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"746e32283961390bc2caeb2cd6aa1f2ccd36d6b3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-31 06:18:25.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"d5b9158ee8aa6d0c249820b0d0e94b6b7614362c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:12:49.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25cdf3f6d1b54ce8b5ec939e89fb99d7 : SUCCESS in 41m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bebfa22475d14d0497efcd141c0ad0d8 : SUCCESS in 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"669b397fa685508d5033a410efe5ae8e4da8472b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:07:44.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/22a8292b076f4b1891fe1ff75688c222 : SUCCESS in 3m 07s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c8639f97d87e4316a9bf35d854bd5d9e : SUCCESS in 40m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fa0c0956ae7e42fdb3a0ddfae0241e2f : SUCCESS in 1h 57m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a6f9f7c0b4c7459ab8d69646a0e67da6 : SUCCESS in 1h 38m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1de63b031080424e9df90c94a37f9dbd : SUCCESS in 1h 12m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9475ddbcb23e4213aad061e51f5babad : SUCCESS in 35m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64a051a07e8244a780ca9ec71e96217b : SUCCESS in 37m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a16815307dc459c9265019b246cee2d : SUCCESS in 15m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41a27a034cc145f8972fe840a855e625 : SUCCESS in 33m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e088ddb7606446669b016c4ce5555c5c : SUCCESS in 28m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdd95eebe5954d65b0db77d1e4e150cc : SUCCESS in 27m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8688f83610b944fc90140cb12c7de036 : SUCCESS in 12m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/383a856fd94e42dcb7612a3a4a6c8ba3 : SUCCESS in 53m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/03ef2387a1244619a1c1c6b139838249 : SUCCESS in 1h 31m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c2397017b26e40da839913385e943aef : SUCCESS in 1h 38m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6c34d80c2874872ad8cdcc2acd1c564 : SUCCESS in 1h 04m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4bf315e91e9b4509b15698f5baaf3483 : SUCCESS in 1h 06m 20s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a27e4fc9c3754f0ea69defa40981bd8f : SUCCESS in 1h 19m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e6be490785cc444fab1f4c900303a7f6 : SUCCESS in 1h 30m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8263cc0669d43a394c38474e8646085 : SUCCESS in 1h 34m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b13fbfffe1a84967baca01625eadd013 : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":18},{"id":"acd2286d139336e6b7cb81076c20ce80d1a3b689","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-07 14:45:34.000000000","message":"Patch Set 18: -Code-Review","accounts_in_message":[],"_revision_number":18},{"id":"7835980c9ec9b92fd93e32c75f6ac7f38ea98645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-14 02:06:17.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"6604f7d1826a3ccd726f3418c6ff26a6b2c596d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 02:59:57.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b82645527e7d448eb87553bbc1e4fb3f : SUCCESS in 50m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/092faf7a88b949e9a2f43f0d1537a1b3 : SUCCESS in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"629fd260372f46e2afa4e0be23dd44793cfed0d4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-14 04:15:19.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/19 : SUCCESS in 2h 07m 08s","accounts_in_message":[],"_revision_number":19},{"id":"cb024255ea072ddf7aa0e056ddb25af368ed1eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:20:32.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dee9628f6dac40d9b18cb606ef7d8019 : SUCCESS in 2m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b26fd18178d0448eae17493de7e05808 : FAILURE in 1h 06m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6ff6a13824624cb08607b9e4e6427a50 : SUCCESS in 2h 11m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f790f866983349f6831281cc3ecdcc15 : SUCCESS in 1h 55m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4d93521fe9f6494c979a33a094eb16b4 : SUCCESS in 1h 26m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5c2cd201b0be4c9fbb11587aa85da4ad : SUCCESS in 2h 01m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/957de35ee2fd47398ae604666c666472 : SUCCESS in 38m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1073952322c14015aeda880581cc66e9 : SUCCESS in 35m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3999303a0a44be9a3dc980473f9b3b : SUCCESS in 12m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/004d28f7eacc4fdd861bb39027de086e : SUCCESS in 48m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d79079e3d13d4de88a62fc44cc602514 : SUCCESS in 29m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65d232594b2d429f918081c57faa4931 : TIMED_OUT in 40m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ccfc771c95c4d85952cd86581f23809 : SUCCESS in 8m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e260aa194764626beb1e032452c0d1a : SUCCESS in 54m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/edfc1bc98d274e92b9d97108a2729a6e : SUCCESS in 1h 22m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9beaf55c71894bedbff00ecfef53d595 : FAILURE in 1h 27m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d724c46823e746cd803d868b1833e86e : SUCCESS in 56m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/61b7a7a418894a20b5cb85fe807d0df1 : SUCCESS in 1h 19m 56s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3034717e01c143ca9dd857530cee95be : SUCCESS in 1h 20m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c5728b0b169546edbb63e8e73b254e05 : FAILURE in 26m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1f8717d5da864fb09175ccbaca0923cb : SUCCESS in 1h 26m 41s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38cafb6cb07b4665a1ca99d7de7e40a4 : SUCCESS in 1h 08m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/41aec35da5544aeeaa5f037e9182e471 : SUCCESS in 1h 08m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2d2d78806f784022b8bef8b0df1b2e69 : SUCCESS in 33m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/408744c3113c4aa1a121762095e65b51 : SUCCESS in 1h 16m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/56e135100b9547218c27b7fa53c3503a : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":19},{"id":"fc248dbbe904960797305eb16533978d442a5d14","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-14 06:25:26.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"2b3be8be5ca256c94e0562c7afbe04917ac40e46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 07:21:45.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdaa6cad4cdc40aba9446fbbfff976df : SUCCESS in 51m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92294048448341bb859edc1886e9b65b : SUCCESS in 51m 58s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"237b39675c76098b3dc11ecac69aafefa02332ec","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-14 08:17:41.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776567/19 : FAILURE in 1h 44m 27s","accounts_in_message":[],"_revision_number":19},{"id":"c43fa49e712e619f44d323d52d9faabe558ef4dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 09:19:51.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3fc5a4e3062948bda46d3f421e3b597a : SUCCESS in 4m 19s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/16ab684eca9d442a843beb9f98efbd67 : SUCCESS in 49m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2f84d0c7a4fc480e894f8f0ee79207d7 : SUCCESS in 1h 53m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3b11381398774c04b5915399820692eb : SUCCESS in 1h 44m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2fde22ec3cad40278f06fdda70c65584 : SUCCESS in 1h 18m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2067b3affc404d55af4af964affb2c12 : SUCCESS in 2h 23m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c46f083720eb4ec3b7a82f174f7b2a8f : SUCCESS in 44m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eddeddf34e2c44e1934491a484125f6f : SUCCESS in 32m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c1aa4d62a2452797410ac21f93d18d : SUCCESS in 14m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dd6220736ee4196a84d650d27a27245 : SUCCESS in 42m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5756af994eea4491b6c3f3079fef93bf : SUCCESS in 43m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57cbb6e533d8496086a520882e766620 : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b4b763e3e14c2392f48a75a7db482c : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d11107a95fc47fa9e0948fd0f45d993 : SUCCESS in 50m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab3641e397c4469db1b3195b325e48bd : SUCCESS in 1h 19m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c13f5f644f2a4dc5be7a4a3ffbf668cc : SUCCESS in 1h 26m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f99f002991c0442f96db2bbc8105b87c : SUCCESS in 51m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0cf3a3d8e5b54e7cbd0a2d224498ce2f : SUCCESS in 1h 03m 36s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cd4cbf625476498dadf8898ad14f8b24 : SUCCESS in 1h 19m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b16e21eacf5644da8842894039f48774 : FAILURE in 1h 11m 19s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9747c21c38a49629a071fe3d75a9345 : SUCCESS in 1h 16m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24f91df37b4345f7b612e1c50860b794 : SUCCESS in 1h 04m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c08b1e31a8c04fb19e840ddae1707408 : SUCCESS in 1h 08m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7b92049fc76a42c79fd05f05733814d1 : SUCCESS in 38m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7b4cb3ff20ba415092da4fecd3edf923 : SUCCESS in 1h 22m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a0562eadfe50482293a9e95f6e2aa515 : SUCCESS in 1h 24m 51s","accounts_in_message":[],"_revision_number":19},{"id":"070454daec1bc83630e530fbc05e3960d18d4bd5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 08:58:57.000000000","message":"Patch Set 19:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/67/776567/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"2c16d186ea55d504d9a5a2fe1171072d195f9545","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-19 11:14:41.000000000","message":"Patch Set 19: Code-Review+2\n\n(3 comments)\n\nI wonder if we can maybe switch one of our neutron-tempest-plugin scenario jobs to use that dhcp extension instead of dhcp agent to cover it in scenario tests also.","accounts_in_message":[],"_revision_number":19},{"id":"b5eb05573c8aac39d9f893dbc8b6ee0445645097","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-19 11:20:32.000000000","message":"Patch Set 19: Review-Priority+1","accounts_in_message":[],"_revision_number":19},{"id":"491b0409260c8041563c5bfe25d2be6fa6267f59","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-20 03:20:05.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"7ee2fd5486f3f0090a318b6d6a9087a854c4194c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-20 03:31:03.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"0e9890e32e1c67db8aa5b9ca9c1eeeeb5eea290e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 04:18:31.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/735f852bb215417e91aaaa8893192120 : SUCCESS in 45m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/476891122daf4c479f24195da4eeceee : SUCCESS in 44m 48s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"1f2e2156aba0ea203be8735508cac382a83915ac","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-20 05:19:59.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/20 : SUCCESS in 1h 47m 04s","accounts_in_message":[],"_revision_number":20},{"id":"5fd07d53fdc3c2b33e6e08bc197e301d0d1138e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 05:49:35.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc574953854e494191ccf57826b2d13b : SUCCESS in 2m 57s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd698afcc92344d59c0fd6a49fc3bb16 : SUCCESS in 39m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/50927bc5f1fa4de9ab1fb494bb5e37ed : SUCCESS in 1h 32m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/922c682c351a47f08a107448327ac37d : SUCCESS in 1h 44m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/59216e36cc284002b23f7c6b6ce9a5ed : SUCCESS in 1h 36m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8347f2bf1a0d44ca8e3c59879da37dee : SUCCESS in 1h 55m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c58fbb56e38843b7a912f9eb7c1484f4 : SUCCESS in 32m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f54eb7c7a5de4748a8106edb5786d822 : SUCCESS in 58m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05ba48ce77324400b13facf57e346489 : SUCCESS in 13m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b7de82b48f142d1b5523e61b7a8c1ef : SUCCESS in 30m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1d6263b0ff44b2b8e632e25f3380060 : SUCCESS in 47m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8eecba7e58ac4b72bb78f0d9d4797b31 : SUCCESS in 23m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/279e1854aa5f4bf2a9170d8b1a35916a : FAILURE in 5m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/41593aa55f4544a1b2dd19e574287f99 : TIMED_OUT in 2h 15m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fe48430ac68495e95506c2e091dad71 : SUCCESS in 1h 11m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c4c078c61fca4faab543afacd36fab69 : POST_FAILURE in 1h 31m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/70ff49ae74fd4f75bdef3299c96e0757 : SUCCESS in 1h 16m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/daff58f9eb82497283c56e23ca98c931 : SUCCESS in 1h 34m 21s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2785e0ef9a3d4243b92a3feec6c48e72 : POST_FAILURE in 1h 45m 30s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/636120311ba7460ca19a3a85971bf231 : SUCCESS in 1h 12m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fadc383edfee4f9b8842fe66f9c9743a : SUCCESS in 1h 47m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad0305793bf44088a25f3a40ba9b1fc6 : SUCCESS in 1h 20m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1fa1e9a4414b4481ba035723bf22c93f : SUCCESS in 1h 30m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0bdb2e4936024fbeb235ad67f7702951 : SUCCESS in 36m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef9053ad958349a2a6abef33f38ed491 : SUCCESS in 1h 22m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/39fdcf3a24c14d5d9ef96084951cc3d5 : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":20},{"id":"48d300429719759ee2bbe75fa7ba1e7829749cce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-21 00:53:13.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"97e0f52c4ab33ca9e09c724f6d8de69d9611843b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 01:44:46.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f051ee02a424a619223094a3b3010b4 : SUCCESS in 34m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f353d3eac604df5a68715aa24909f61 : SUCCESS in 48m 53s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"745bf6f783821f494dbfad62ae149d1b13f36167","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-21 02:31:40.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/21 : SUCCESS in 1h 36m 16s","accounts_in_message":[],"_revision_number":21},{"id":"9f32f6e0e65676b93fe8934acace24beec923a4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 03:12:06.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5eb97057029447f98e213feaf597b8ae : SUCCESS in 2m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fa98145426e94d3e93014350f6a21b54 : SUCCESS in 40m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6b5db54625554bc3828ca7db5d027180 : SUCCESS in 2h 08m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/64b2c57524fa43199ea5a8e2b84bdce9 : SUCCESS in 1h 44m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6505bba987194e34b563da7a9d645546 : SUCCESS in 1h 00m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1493a11788774ff4839b7e82c394d83f : SUCCESS in 1h 37m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69912c7e2c954e5ca1249a0a70dd6bb7 : SUCCESS in 32m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/898ee7501b2b4b068c84ca7cc120e97c : SUCCESS in 21m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eff47da0551441758375376676553ed9 : SUCCESS in 12m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02312f5449974ea69e06df3d4aea19e9 : SUCCESS in 19m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23d44c837551421da311b4e43bfba7a9 : SUCCESS in 22m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10ac55a6c2b74f52bc473fc641d491b5 : SUCCESS in 28m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac8ebe56cc7a405ea2ee3597d0a783b3 : SUCCESS in 10m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b12b72a908f140c98f9b3928fd0587a5 : SUCCESS in 47m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e4f826d8c1a48949ae136c714366f48 : SUCCESS in 1h 20m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/15d9ec589a994b668a3712d0c88c6b7b : SUCCESS in 1h 17m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e8b09cfc896428889b469d5c4d44254 : SUCCESS in 57m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e79612b3a22245e7b7c42238cc671361 : SUCCESS in 1h 00m 18s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/30ccd4ccbecc47078f1edde4a2dd4c55 : SUCCESS in 1h 11m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c660071e3e3c4b75901936c97b2e24ba : FAILURE in 1h 24m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9175a3701d3043e6b0bbd75404877b3c : SUCCESS in 1h 31m 11s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbaab5847152425da787e23924b0797e : SUCCESS in 57m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f2d55a7d4fcc487c81581a927dbb1f9c : SUCCESS in 1h 08m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5292a29525164877b7f41a5dc00e0e39 : SUCCESS in 31m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f72da81ff3a94f9ba81ecf5fdf49fe5c : SUCCESS in 1h 20m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6942981eef1a4407afdffa2cb59c9060 : FAILURE in 1h 21m 30s","accounts_in_message":[],"_revision_number":21},{"id":"0d351bcdbc37e80e2a3f14c9523ce84daca924d0","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-21 05:28:42.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"32bbb08b404efc5326d3460c61658e667bd188f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 06:09:22.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b70cfd7a1134d14a66e32628dfc50ad : SUCCESS in 35m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56a10c29771b42899b4b8db3de86727b : SUCCESS in 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"9f35a2a59c9489c04ce34d1c298e2f96ad5b13a4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-21 07:21:21.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/21 : SUCCESS in 1h 51m 49s","accounts_in_message":[],"_revision_number":21},{"id":"b9a58b847d9a403cb74019a07e55fd1b3c85d848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 07:52:05.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a2fe713be41c43b1bd36720b53922035 : SUCCESS in 2m 55s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ad4df5e4813349b29bb413d33a3bde55 : SUCCESS in 51m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8e92a76d5e83460fab01e67661e8ef58 : SUCCESS in 1h 46m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a8da3ac7a6cb4b388de2bd9957c4fac9 : SUCCESS in 2h 19m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/96c08d360fd14b0d9e3a118d5a069512 : FAILURE in 1h 24m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f9657c4d52414fc2ba134c9e9895c0cc : SUCCESS in 1h 33m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6c4fe1d819f48b6a316673603868160 : SUCCESS in 37m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/710841bf3b6344cdab7c1ad70ed7b09a : SUCCESS in 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9434ee3783d54829bff3b5ab609a9a42 : SUCCESS in 9m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba0b9085e8454e038c476bfd904d503c : SUCCESS in 31m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de453acaa03b4ac79d5e6551cf8f4eb3 : SUCCESS in 24m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1881980461e0406d9cc71b49682397ac : SUCCESS in 24m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ae3b030f1d4e4281538e4a695e7c84 : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e6b958017834474bb0e61e7c196538f : SUCCESS in 52m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e691e8e15fc479d9305ac443cb2a915 : TIMED_OUT in 2h 14m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e28fcc1fbd71465991ad998d50189fdd : SUCCESS in 1h 27m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee767c102f2a4cd7a598622a3c1a6876 : SUCCESS in 57m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e573a0af3e2f428c9e7eb5641e05b05e : SUCCESS in 1h 28m 21s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0abfafab98f344979291fd896005b28a : SUCCESS in 1h 10m 32s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4592808395ae456ab1df629eed381cd8 : SUCCESS in 1h 22m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ffa0769f4ca40fa98a87e8ef90d304c : SUCCESS in 1h 31m 55s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed27d22376cd4eb087c2f72e277445fd : SUCCESS in 1h 02m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e7d21dac21194827905cb5adabbfca80 : SUCCESS in 1h 05m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/09672366bdfa4360b10f62d0e64c6d27 : SUCCESS in 35m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/398a3a7d5bc24b5f8884085601db7b61 : SUCCESS in 1h 24m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d98079ebd81f4c68b465d190b005f2b7 : SUCCESS in 1h 24m 37s","accounts_in_message":[],"_revision_number":21},{"id":"252962a6fc9b0758f298b5714c5f8530dbaf8f2b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-21 08:27:08.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"4ea777ed88e7b5a6cf5b44f416f09946f4a64c36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 09:12:25.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/85997165df674d9280879d4d5044bf91 : SUCCESS in 38m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e239917d5be4e7dab06903f1e59e6b9 : SUCCESS in 38m 41s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f71a27c0e4cdb612d1d73035e7a21c91205e7b49","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-21 10:14:23.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/21 : SUCCESS in 1h 45m 13s","accounts_in_message":[],"_revision_number":21},{"id":"a1e1720755d9c946720818af8b66de5ee6c53361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 11:08:33.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d96ee45ff70b41679199b34cd73a072f : SUCCESS in 4m 49s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fd80e78189d941e3899393d63a3adfbd : SUCCESS in 40m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8fc300cfd0bc41f8a9002bdbaba4129c : SUCCESS in 2h 06m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ab56929076884d4a83a03659e43499b1 : SUCCESS in 1h 41m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/485b2024bf9c44b682d4f45a6de3f38d : SUCCESS in 1h 25m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d5a171cb972a4c69881b957ca6d040c0 : SUCCESS in 1h 43m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d06f610b839c4ade8b01bd5d7632533a : SUCCESS in 41m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e82cf91f6de5408b92f52f91e92bae2c : SUCCESS in 36m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90cc5d6371a94700b97174b8ee337933 : SUCCESS in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be8d627fecb0419887e7b802e9e9d982 : SUCCESS in 25m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d51812b5a03b4365934e23e1ba166501 : SUCCESS in 27m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cec499f69ff4599aec19d6804c5ab9d : SUCCESS in 29m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9baaf7d2cda4525bb7aa296f0ed5793 : SUCCESS in 11m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba659e191610427e8d3b7bd4274a2ff0 : SUCCESS in 52m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/47083681d8644447bf33dd7a54c4ccd0 : SUCCESS in 1h 21m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7e7827447ac14d058b4770d60323b7d7 : POST_FAILURE in 1h 03m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ba020fd3a8241fc9914fdff90250c2d : SUCCESS in 53m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1ff58bd8deaa4123bb8390c185007c3c : SUCCESS in 1h 11m 22s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0f8da9d297d2479a8d401218b07d9e9a : SUCCESS in 1h 16m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/835dba80ae284f7fab8034122e72fff9 : SUCCESS in 1h 32m 33s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4f112c9199c42d39534918c6a4567f0 : SUCCESS in 1h 32m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9211ab496b7444b5894dd9bc36cb7fcb : FAILURE in 1h 04m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/113e7eeb493e4a9eab2496bc3fc6d050 : SUCCESS in 1h 07m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/16019286f1f94781b8f6a5f05cc2a517 : SUCCESS in 33m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d723cc986b95492080e5e8450aa57015 : SUCCESS in 1h 16m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/979bac0411bb422d909df9835eaf66c7 : SUCCESS in 2h 12m 09s","accounts_in_message":[],"_revision_number":21},{"id":"1bca49e1594fe208580c3fb9aa2055688086d912","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-24 08:29:58.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"821e41cf8f472261c053886c75f45c706c5c7676","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-24 09:14:39.000000000","message":"Patch Set 21: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":21},{"id":"b0d91eeac969120b6f07deddab3924e2254794a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 09:24:44.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4269c388f2946ee8f87ceae9ff52f2f : SUCCESS in 43m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/919f272d46d24a61881fdc2484f4911f : SUCCESS in 53m 02s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"218cec7caf0f4d89fefa099a36cd2cb51a000f30","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-24 10:37:57.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/21 : SUCCESS in 2h 07m 00s","accounts_in_message":[],"_revision_number":21},{"id":"2f4c6286ea807b49191e8ce8d2cbbbe1fbabae7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 11:20:55.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ab3f62d5028e4dba9e5423cf53c04ee5 : SUCCESS in 2m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/de3c42d39d8e4f1188297a8f5733dc02 : FAILURE in 33m 11s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4e602bbd60514911aed03b6326af1b30 : TIMED_OUT in 2h 49m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/44489634a07b4fb2b9ef0d54c131082d : SUCCESS in 1h 46m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f271b9dbf15f40539f44f7c2512a70c7 : SUCCESS in 1h 36m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3eb39649da5c42b1a5817b05f4c4cc6b : FAILURE in 2h 01m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/762276ecbb1f4637b3339dc9a1f082e9 : SUCCESS in 25m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f0215fe5a174f49948272d5ff58a08d : SUCCESS in 33m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f4b658732c4ea4b54a1b921eaed33f : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c9ddb16fda44dfabf7167b9863064ac : SUCCESS in 31m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dda8b383c6a4750a35feae3081f1527 : SUCCESS in 23m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12f7e94195df4e3aa5770b3b5a1856b2 : TIMED_OUT in 41m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f67e20922a99493abb4d377c66432d9b : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa94f30d00ef462d9a74758eb6e9463f : SUCCESS in 1h 00m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/681eef50d3634a53822b667457481cd9 : FAILURE in 1h 16m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b77a3b2209454948ac646c990159ec06 : SUCCESS in 1h 12m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ac748d59cb84052acefb786a4f58938 : FAILURE in 1h 59m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a999e596142a4c02aa22ca472ff8437e : SUCCESS in 1h 07m 49s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4f36c6ae2ead4c8886f1c1dbefb3f0e7 : SUCCESS in 1h 15m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ffe9ed046cb54bbea55d7f7ac2766c74 : SUCCESS in 1h 35m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/377eff817e454433916c5f17f43e1ed9 : SUCCESS in 1h 28m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1076a62adbb842e69d9bf431b36c2edc : SUCCESS in 54m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/750dd42fc0fe494bb9023440fd5b05dc : SUCCESS in 59m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1e5578725771458b80003d4f375825ee : SUCCESS in 29m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d54b2d6b370145b6909e276927a694fe : SUCCESS in 1h 15m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/027c6a2ab75a475f9ae93920d9ce3321 : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":21},{"id":"599f438e6f277979d730bd78b4a09a12c3a1877c","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-24 11:34:11.000000000","message":"Patch Set 21:\n\n(5 comments)","accounts_in_message":[],"_revision_number":21},{"id":"59300cbed303e43a663c4142df9b10174b2c0451","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-24 12:00:23.000000000","message":"Patch Set 21:\n\n(4 comments)","accounts_in_message":[],"_revision_number":21},{"id":"729c2e983392c5bb9fcbf6fef6e8ae5aec734d16","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-25 03:28:11.000000000","message":"Patch Set 21:\n\n(6 comments)","accounts_in_message":[],"_revision_number":21},{"id":"cdfe0be1e9d8066cbd48acdb11357d7fbb7f362d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-25 03:30:34.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"28e52ef3f1c5eec150ca0fffa7209862607820db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 04:22:03.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8144efce9254542824c17856350659f : FAILURE in 49m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/333bceeb3b1b4a2cae76e0f71725a651 : FAILURE in 49m 24s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"e3ed1a24e53bf1ea47d150231e5ec22df3d87ffa","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-25 05:08:21.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/22 : SUCCESS in 1h 36m 05s","accounts_in_message":[],"_revision_number":22},{"id":"cffe962b2ef7e26337b0337246221d363b820a78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 05:28:35.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/01a96a36adf341f192c5c2392ae77620 : SUCCESS in 2m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ebda2add2bb54e999a481d55d0d327fb : FAILURE in 32m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/208b8ad9cb04483e96789703301fe4a3 : SUCCESS in 1h 49m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2d8b2321c7544e8eb97be06e994340ba : SUCCESS in 1h 22m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b35d5934aa84f04b81c03cf50c525cf : SUCCESS in 1h 20m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c9531c6d427345d1b1eadd1718ffe04d : SUCCESS in 1h 56m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcfd4868eb034c7dbbd3572dedc7c77b : FAILURE in 32m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce60832d369e48689333e55fc3c3c596 : FAILURE in 25m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cffbe0598124797a50ec9e1a1a125d3 : SUCCESS in 10m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f934f0d728544f7ba9aff0de8ec8aae : FAILURE in 26m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf66d764f7674d4780e4e789c8346225 : FAILURE in 22m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e73477245a7b4c45a708b22eb5e4c006 : FAILURE in 30m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/612b2b5409e5425fa46731bd9f793734 : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d0344698ee340978a87cba40950c54e : FAILURE in 44m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/741eba8cc05b45ed94c63f6e6d23eb3d : SUCCESS in 1h 18m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4b6bc4c1f56d4fd4848056c5e65e17f1 : FAILURE in 1h 22m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ade49e2be1354dd4ac594083f70b71db : SUCCESS in 56m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6e681e2d569d4cb6b862d232464541c7 : SUCCESS in 59m 27s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/364985d8cc5f4c0abea1b69c0f689bcf : SUCCESS in 1h 18m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/68bff96d5a67449c8fa7e083e29675f3 : FAILURE in 19m 01s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b240442becb24f43b77b6a16a2de15da : FAILURE in 1h 31m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7f195e25a9e4e838e653e2a85016939 : SUCCESS in 59m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d52b3ab12bc14b78b7e35a08923b4bbb : SUCCESS in 52m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/07a64d4f839b423b8ede78e7f0eb40f1 : SUCCESS in 28m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/06c0881ee4024471961b8714037044b5 : SUCCESS in 1h 40m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/53cc4a77949e46f6978d198204918c7e : SUCCESS in 1h 36m 06s","accounts_in_message":[],"_revision_number":22},{"id":"46ab78ad816864ccc04802b54dc8fa2d24100bf8","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-25 07:35:36.000000000","message":"Patch Set 22: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"c3b47b4f92dd2a645e756423602625d7fc2cac2c","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-25 08:09:50.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"d9f953bab2b2beda28f27311605bfcfc062ac277","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-25 08:12:32.000000000","message":"Patch Set 22: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"847e87642ae13b51b834896338d6f9ec81914993","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-25 08:27:15.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"079be9670f6917fcc5ae3ece71c37679503a299f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-25 14:35:45.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"a47f88229f4ec98b1770e7ca359330071dd0baf4","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-26 01:04:17.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"4dce861aff5adf82a39994d440b40a09907fbff5","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-26 08:10:21.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"fd0c32195f106e08d7de492688c12b911e5ec44d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-26 08:19:32.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"59ae64d000c160f427277b90cad3784bfd1d721b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-26 08:20:43.000000000","message":"Patch Set 22:\n\nThe concensus was that if we need this new cfg option let\u0027s make it True default and document it to be clear for deployers/admins\n\nThanks for the efforts","accounts_in_message":[],"_revision_number":22},{"id":"6f75af7681427d3f2534b6d0b8748311856bef5b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-26 10:17:21.000000000","message":"Patch Set 22:\n\n\u003e Patch Set 22:\n\u003e \n\u003e The concensus was that if we need this new cfg option let\u0027s make it True default and document it to be clear for deployers/admins\n\u003e \n\u003e Thanks for the efforts\n\nACK, will change that.","accounts_in_message":[],"_revision_number":22},{"id":"0810a93e3f685a843d25b236c3978e3e7abc2fe4","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-26 11:11:31.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"d1d196f15e0019435dbf9df924ab266417483219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-26 11:12:30.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"83caa08fd3d9bff8b08faf96fb42f00471d0d9e9","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-26 11:15:58.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"fd557ec0573fbea2b9caf9b775fd4c932212aeff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 12:06:48.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28cb74e2a8fa4c449ffb19d836836c4a : FAILURE in 47m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad54de479e9649df9cd04f37cfcb1b64 : FAILURE in 47m 02s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"468d05b063fbc16e0120f724e0e12b154ed1cafd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-26 12:27:09.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"2c0520279a7e53168bc8c6e7f28b6706b252a5aa","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-26 12:42:01.000000000","message":"Patch Set 24: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":24},{"id":"4092ba48c0abd770a8ecc8e259d45bb9dfed7a93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 13:15:34.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93cf90b9cbaa40439e27f4bded6094c6 : SUCCESS in 44m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95fb0f1def3947ee9a0b4461d7d1f6e5 : SUCCESS in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"e77304af22b865070dc293f1d674a75df48fdc9b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-26 14:25:32.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/24 : SUCCESS in 1h 56m 23s","accounts_in_message":[],"_revision_number":24},{"id":"4aec06eec52a2104a0d26822570aaeaf6bf29bf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 14:38:10.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/45fbbd05551c495f8bc6ae528dd26bc1 : SUCCESS in 6m 16s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/640e78428be044bca799ed10bf5d15e5 : FAILURE in 55m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d6ad034554a542808b1d15d7d0a22492 : FAILURE in 18m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/88e9ab7a0cbe427f87531f3c48ec4a43 : FAILURE in 19m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98ca78ae7ab04fd59edb094f27141280 : FAILURE in 46m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/89e7f4c8e3694891860517a9fb149fef : SUCCESS in 2h 09m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a48f65a7779b4638a195979ae67c6885 : POST_FAILURE in 40m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29cf3cccb8f9495b869a1948ad09843b : POST_FAILURE in 40m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b185a0e2fadb4b3298c1b7a8baa5247c : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6b6fcaf54f04db3972845dc0d808cf6 : SUCCESS in 28m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39d64ea772ce4953937359f779ebca4c : SUCCESS in 34m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67bcaef84241442387d070e04445dfe8 : SUCCESS in 34m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5956d603f0ac42e68f0323112aef572b : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa2a3b98acfc40a380a7ad1ee94996dd : SUCCESS in 8m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e64ef995dd143c29bde5a931e4e20da : SUCCESS in 40m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aed61b40cb2b420a989494800ba58495 : FAILURE in 1h 05m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/792198f93b4c47f3b2bbc3164f3d7bac : FAILURE in 1h 46m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9299d9d106c44f58bb217ce356111e5 : FAILURE in 30m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dbaceb0d3c444e2380bf0e420b4b1c01 : SUCCESS in 57m 48s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f10551439ab048bea06227a0c7a77717 : SUCCESS in 1h 05m 51s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6bc58940c7ce423b99829ed4791c79c9 : POST_FAILURE in 27m 00s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/439859bee4a44aafaf03a9a959aa3da7 : SUCCESS in 1h 24m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fd4e0097757444bbfb6a9c907cfbc2f : SUCCESS in 1h 07m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d256f42b8f5c46e8b8951cc64a6c6657 : SUCCESS in 1h 06m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/35b13b67c2b941e392e0593626f7e6da : SUCCESS in 30m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d6d45c5ce36646fdaaede27a13b73749 : SUCCESS in 1h 19m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dbe072ff03884c1e813897a5c6bb6461 : FAILURE in 38m 43s","accounts_in_message":[],"_revision_number":24},{"id":"fbd4d567ee237e81d23209ee97f1a6792d1b7588","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-26 14:59:19.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"b67b82895f6882aea23f3b4a08b6263c3de1ba6c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-02 04:04:38.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/67/776567/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"9f3092f3514b1066fe05e84595a6045e7d05d7ea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-05 18:39:19.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/67/776567/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"132cfb4195dd3c4b902fa0976ea085fabb3c9377","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-05 22:50:25.000000000","message":"Patch Set 24: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/67/776567/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"dabd742aca9beeb234da984e28bc6b96d98e24ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-07 03:40:41.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"dc81e0d7fc324f2fb58acc5bf443d542136f14e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 04:31:39.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da11cd31cbe8427d8d1f3858ae78c684 : SUCCESS in 35m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bde29cc62515489e8646991c1a0ee9c7 : SUCCESS in 48m 28s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"dfa147dde31ae8a54a2b620201292520bbbb3a61","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-07 05:40:15.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/25 : SUCCESS in 1h 57m 33s","accounts_in_message":[],"_revision_number":25},{"id":"299b42ef316f86fa60948614d4a74646017c672e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 05:49:12.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/23d6e362950b41dfb86606e778fcf5a7 : SUCCESS in 3m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6959383c8a9e478eaa8242b6f7c5e0d5 : SUCCESS in 52m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a10eba5d559f4e1fa076faa8243fb359 : FAILURE in 1h 39m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fbfba490d40948088170068a27359254 : SUCCESS in 2h 02m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e5945e027734864b531715e8fc62ed7 : SUCCESS in 1h 05m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/20c0b5220ff1443ebc3cff9608ec914a : SUCCESS in 1h 36m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e4adf5c4ca24db9a471da70799e07b1 : SUCCESS in 40m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3708d580a61441b895a36dd85fec86a3 : SUCCESS in 34m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e6687a6be447418a98607155f6b481 : SUCCESS in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42a3c119a88042a5a44dd5801e76315d : SUCCESS in 23m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dec18debcf364199bc1caa4f79e48c70 : SUCCESS in 23m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dd77556d03c4689971720171a34a41e : SUCCESS in 27m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5959379ba36470d8b2dbef414989642 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96c5b8fb53a342e9bd2cb2cb11dba12c : SUCCESS in 7m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f79bb8cd7d29497a835c686ddb59b422 : SUCCESS in 44m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3fab4db30e6d43f9ba6b8a76e8ecb268 : FAILURE in 1h 08m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b0ecd353827c4a95a988440dd3430243 : SUCCESS in 1h 13m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35934fd8274f4539a31e613d5a8471b6 : SUCCESS in 53m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a6e918e31722429781707fbeacad83a2 : SUCCESS in 59m 14s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/03ff8eb1861c429fab0c92df07f0c9ec : SUCCESS in 1h 12m 57s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/987df382943942489cc91a92738c77f3 : FAILURE in 22m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bdf8ce2508c64f1fa778f1c05801da5a : SUCCESS in 1h 24m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b7d9f746199483dac44f27b5f7b403e : SUCCESS in 1h 10m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5a4d51b4972642248f78bad0dd6c59f5 : SUCCESS in 1h 09m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ee8b9e0bf3194700b26db4046cea090b : SUCCESS in 28m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1370a7c4fcd04f3881afbfa69e799791 : SUCCESS in 1h 21m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6f79475eae2e4d24acd3f7251d1ea581 : SUCCESS in 1h 07m 59s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"981d9949ba44c439bb288904de37998b296a83c0","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-07 07:57:40.000000000","message":"Patch Set 25: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"914546a42a49ff2d89ee78612d0a58f62b73d3da","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-07 08:08:02.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"edf4564e6d8d8f7575f8bf2e24c939743266cb1d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-07 08:35:15.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"d887db8b14288cdcd9edcf5944a10a54e5f3d758","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 06:33:39.000000000","message":"Patch Set 25:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"8ae0b6b89b6193711a36c06f764fc6975ca6d7b6","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-10 01:14:01.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"8bfbf1de77abb2a99e25d6f4611bfb8590b7f1e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-10 01:27:34.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"634230cc2f64127a7e1c63509f2abfc93e3c2fb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 02:11:52.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a51225c75dbf477faa7e63aa7f0f15a3 : SUCCESS in 36m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0f68b800239424bb02af7fa2f6d53da : SUCCESS in 36m 07s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"eae5a505cfd0a552b6a7ffb014cec5d4b34c059b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-10 03:15:54.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/26 : SUCCESS in 1h 45m 33s","accounts_in_message":[],"_revision_number":26},{"id":"73871527f6fc818a9220db75833cd22e661c8d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 04:21:32.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dd87703a70bf40aa8fb2545dcde3983a : SUCCESS in 3m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0a9b981f8a35415aab23342eb907feb9 : SUCCESS in 1h 21m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/17e81a02aff54125aa0f98f4b3a19847 : TIMED_OUT in 2h 52m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c9d7d72a46874a12b430de2eb9d28f2e : SUCCESS in 2h 35m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/63c4dfe6a87248a88e3395af49f8180b : SUCCESS in 1h 25m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/71ebd530c87442829821ad1797e1c02b : SUCCESS in 1h 52m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35443559eb644e929228a5665ae18a50 : SUCCESS in 37m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/754911628cbc4db5a49e65f27162c71d : SUCCESS in 27m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e19b29f7334610893db1bae3834413 : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/048a619d6dc04f8f809686f6abf1d7b6 : SUCCESS in 24m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd870b42892c4f88987f326b6b66c03d : SUCCESS in 28m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c66ab4bc93c04645b792c6f33ca2407b : SUCCESS in 26m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/000b66d60291484f9eab919544ff991b : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce00c5027f1c424abb83eda692bb0b63 : SUCCESS in 11m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7adc010cb7f04418b8d0db2c1c1aa7a1 : SUCCESS in 46m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc4c287447404f9ba4e41e216d603803 : SUCCESS in 1h 18m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6acd0111139943fe976ea32dc909a13e : SUCCESS in 1h 11m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/db2a58d544b94760a4dbf2492b9ae739 : SUCCESS in 56m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3bf33e6d5ba74d52ada225efa630b175 : SUCCESS in 57m 39s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/83552e0ec76d47af8316c984f2e60004 : SUCCESS in 1h 13m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f32ec8285ca94228ac7519084202d948 : FAILURE in 1h 04m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d5cd8d4fd7cd4c3980e5032f0f95d3e5 : SUCCESS in 1h 30m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e084272c86b945a2a6abec7b415ee1a7 : SUCCESS in 1h 19m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/45d63d95a9294b28a3ba6e188fb719d1 : SUCCESS in 51m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/aa65d44a99674212b3e6f1d86634aceb : SUCCESS in 25m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ff82d6324be44e1ea9a3179d3b40ed2f : SUCCESS in 1h 23m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3ec4bbf9d431410ca49c2829bba1c85d : FAILURE in 1h 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"d4261e2f9b7e9cc0b6f455ade74720fed0066bd1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-10 07:16:17.000000000","message":"Patch Set 26: Code-Review-1\n\n(1 comment)\n\nSorry, one more thing. I should\u0027ve notice it on previous iteration","accounts_in_message":[],"_revision_number":26},{"id":"f37f2605566e845878bc26cc55c7e7f575fb3eff","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-10 08:03:27.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"d01afe8818af53a9430451b714722c44fe5dc608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-10 08:04:15.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"22f3549fffd9aef0493dbf1c484f4ea36eaa5001","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 08:53:37.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0386f5f9a2494d3fb79415d7057383d1 : SUCCESS in 47m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0a25152898a45c99cfe89eba4e53130 : SUCCESS in 45m 38s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"4e60aab1357662b7daaec1fdfbd24e12ceb60e4a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-10 09:42:53.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/27 : SUCCESS in 1h 36m 44s","accounts_in_message":[],"_revision_number":27},{"id":"047e1568b09e2473491b05b36e2d88a24127ea00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 10:15:59.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/60dfec6dbfeb405fb0f2c5ecd465d59e : SUCCESS in 3m 47s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3049440c830447dc8877162de38eff6f : SUCCESS in 35m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/58c4be9e2ed546e8906bf6f3ba5e92a8 : SUCCESS in 1h 23m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8c296037554443a4998a7ff6e92de279 : SUCCESS in 1h 16m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e1735fc03054f008f0f83314a2cc670 : SUCCESS in 1h 50m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1cbcb6b5cd8948c2af114e5f774abf9d : SUCCESS in 2h 07m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85edecae579b4e7080bea7fbda95863c : SUCCESS in 51m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c77a2a59d37349e5ab63275ed4cdb1e2 : SUCCESS in 32m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44e7b54ad2d347b8ad16356c6819b6db : SUCCESS in 15m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e32d2d6a00d54e9498004e334fbda267 : SUCCESS in 31m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26601e386d1b44819be13db9270df30f : SUCCESS in 27m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5b8b93901004f1eaf6430e838042398 : SUCCESS in 27m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523471a5ee9c43378e8bb15172a8037c : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/106e0b24d887408ea0720412a1d6edb0 : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/10594d0fc81a42fa85eb4af7d6c2a7ee : SUCCESS in 52m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/599b8f0bb2584f3b88130e33be2e0b7c : SUCCESS in 1h 20m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4faed92dbafc447d915b6fecaae96076 : SUCCESS in 1h 37m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcdb2fec9d1c4add83398f9fdc6ab2e0 : SUCCESS in 1h 02m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/adb01b219648412381476fc2792ae770 : SUCCESS in 1h 05m 22s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dafff441632f4cb7abfc1d243e28cae7 : SUCCESS in 1h 16m 36s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3c144bf136a64ff787753bfd7af828ad : FAILURE in 1h 10m 24s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbe4ece638a84b4a9a39a97bc99b9dc2 : SUCCESS in 1h 32m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cfb8712e9164265ab3a306a3e24d5a2 : SUCCESS in 1h 01m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c028be2443264946b6110876cf15464f : SUCCESS in 1h 02m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d9fd7a044cbe409db7c99cb33a02dc10 : SUCCESS in 34m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c6e085f74e394e98bf053cc6c725e160 : SUCCESS in 1h 18m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/25a1bd3cec2343338afb42fcb2e77dcf : SUCCESS in 1h 15m 07s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"d43eafe3dc308db9d3db738a841baf3d1b075316","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-15 14:59:02.000000000","message":"Patch Set 27: Code-Review+2","accounts_in_message":[],"_revision_number":27},{"id":"e7bf8a47307fd18760b2f7494c3e9930ff9150a8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-17 07:20:22.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"6f569f9743089f88c2337cce83343e6a309c193b","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-06-18 03:56:35.000000000","message":"Patch Set 27: Code-Review+1","accounts_in_message":[],"_revision_number":27},{"id":"e4d0418b218a38babba06ad259c9685ad561ff65","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-18 12:02:09.000000000","message":"Patch Set 27:\n\n(7 comments)\n\njust some nits and questions","accounts_in_message":[],"_revision_number":27},{"id":"09a3758597fb1087c1313e85f0f0d9dbe69010d3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-18 13:11:13.000000000","message":"Patch Set 27: Code-Review-1\n\nwell, -1 because of https://review.opendev.org/c/openstack/neutron/+/776567/27/neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py#b285","accounts_in_message":[],"_revision_number":27},{"id":"aafeee41e143f82cfe303522d4922ec0cdb357bc","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-18 13:26:22.000000000","message":"Patch Set 27: -Code-Review","accounts_in_message":[],"_revision_number":27},{"id":"7980871d17c6cbba777fe3c80d78575ac68d579c","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-18 14:41:22.000000000","message":"Patch Set 27:\n\n(8 comments)","accounts_in_message":[],"_revision_number":27},{"id":"04e41564605eb5041e2f68f4a5adbb2ebb9973f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-21 06:51:36.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"cf2395caf076af6148b218c51acf408dc952aba6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 07:38:49.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e51a5ab012043798faca38b1d7aceae : SUCCESS in 44m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c5a17f2fd694c7da4868a154da23dc2 : SUCCESS in 44m 28s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"c96ec9175794cc8807d49c1d432795502e5a528d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-21 08:25:37.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/28 : SUCCESS in 1h 32m 38s","accounts_in_message":[],"_revision_number":28},{"id":"2178155bb4effb98dd69ef132d37db57804ab9c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 08:57:03.000000000","message":"Patch Set 28: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d7b044e4b58344e4b7904791b9efea21 : SUCCESS in 3m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cc4b6e22c4d646b08a0038eea55cab68 : SUCCESS in 41m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9cd006459bd44fb8b8c9ce6bf94d1a2b : SUCCESS in 1h 53m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b8b2e6bed4e24fc38e1643eead24e266 : SUCCESS in 2h 03m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/90c3518ea3584ab28eb3092591949ea9 : SUCCESS in 1h 39m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/aafb95d6adce476ea533b122de87f882 : SUCCESS in 1h 42m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0559b9642e454a3f92feda2f8fae4cf8 : SUCCESS in 32m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05d1f7a017c64646b42b36d95f24f5aa : SUCCESS in 33m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9145643ee441a7b4feb650d7605591 : FAILURE in 9m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac5748ffd9042dd9af18c8c3327f2ce : SUCCESS in 22m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f50607cbfe8044678d69eb5ea1f97647 : SUCCESS in 23m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/522a9592bf034f13808aaf213d41d1ba : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd32ebc0c2c643a9823fe4ed8fb0b85e : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67ee7b5781df4068aa2f0e899d8a2849 : SUCCESS in 11m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f96040a6637a4cb99686ad01c586293f : FAILURE in 48m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/38dbd1f12a694f688f3f2889b7450094 : SUCCESS in 1h 24m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/032802f79f2d4728a565df6cbb5c2279 : SUCCESS in 1h 18m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e3c968d4f154c0cb21c87cd94893488 : SUCCESS in 1h 04m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/14d16c3ef87c4d15bf55c513f55f5a78 : SUCCESS in 1h 04m 30s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dfd78a5c4e4b489783a5ae7ba0c0fffc : SUCCESS in 1h 22m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9ffa744443cf47b8bbcd37eb0c6b6f5c : SUCCESS in 1h 23m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67555e47e28b4435ba7efcfb25605cad : SUCCESS in 1h 31m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab3e84e63f0c4237ba971043e2b3cd13 : SUCCESS in 54m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f53e9af70a88490389115c0941265808 : SUCCESS in 1h 00m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d48889c7a9fa4e41a7ad65f44d36a18d : SUCCESS in 35m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5ae84616cedc4ca292a91ed99d0103ef : SUCCESS in 1h 22m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c2253a1a56ff4fc48581c9dea911b3cd : SUCCESS in 1h 16m 03s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"8a337bf215ef6d7ced5ba78888aea6d3339454df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-21 10:30:09.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"951a20788fbef90c9b91212877a54be4eaccb59e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 11:34:07.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb671c47ce1f4349873d0c54d3e9a3c9 : SUCCESS in 42m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfcc8baabe274869a2a1a5209af245d5 : SUCCESS in 41m 12s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"e3bba7f14143074f94a6347be5c8eee4d721f637","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-21 12:26:36.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/29 : SUCCESS in 1h 55m 14s","accounts_in_message":[],"_revision_number":29},{"id":"b841f8933bbee9b5c7a7bc51d950cb6141447309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 13:39:24.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/76d6b91c027b4db9b16dd1beacc66385 : SUCCESS in 3m 16s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e5b79da6fc214fd59856a48b5bc9c70a : SUCCESS in 41m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6e654b20ad1d428eadb210b816be0829 : SUCCESS in 2h 14m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9a21316ad7bb4378a76fa1598d5239c6 : SUCCESS in 2h 23m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd2de4bbfe4c4411921cf4c7b1e71261 : SUCCESS in 1h 37m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a43e7b1ca41a40808729f6c7759b8482 : SUCCESS in 1h 45m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29b44c5bc07a4d24b60a1fef592d8894 : SUCCESS in 34m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8300416d6d64a98aa263f312cd4f0e1 : SUCCESS in 40m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f1035b7cd8a42c48932000f28ed873d : SUCCESS in 15m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b9886ba62764343bfbcc4a5d3cff910 : SUCCESS in 27m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42777152a36d4aa680ccf33434546d65 : SUCCESS in 23m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d568f9ed93af49ca8215f0bbc622d131 : SUCCESS in 21m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e760f8f836417ba64ffae67732e467 : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b45b4033ee94147816a20a0452e45eb : SUCCESS in 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/602962ee4f2f45698fa4ac98a7ce1db1 : FAILURE in 46m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae00dc9813fd4aadad3c457bed041748 : SUCCESS in 1h 23m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9bcf7de0aa1d440698de65f0005189f0 : SUCCESS in 1h 07m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/29a9d98792f2455ca77a2df00bedc24b : SUCCESS in 1h 01m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0a76d6d0321046dabdebbfad83a58acb : SUCCESS in 1h 02m 08s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0e738ee7836c4fe086c016cb15276b3e : SUCCESS in 1h 31m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/067f117ef231425ea312c124f27d8560 : FAILURE in 1h 36m 37s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e337863e65644bbe8e4b1e23d5abad08 : SUCCESS in 1h 31m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/feef1329b5dc44b0a1e17d679df3079d : SUCCESS in 58m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b67caa3590814c5da252d63697cce01e : SUCCESS in 1h 11m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bbfab0d2ea2549b6b17ea674b8f71f35 : SUCCESS in 28m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5f03358f17374ce4a9fc3d3879c7bd8b : SUCCESS in 1h 08m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/40524a3051a74ed2bb9236cd69e0979f : SUCCESS in 1h 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"7881e55ba3f55f4901cb49d19ab92ebd110afd5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-22 01:40:35.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.","accounts_in_message":[],"_revision_number":30},{"id":"7c2783cff77995a2f61629bca8e913014a8c5541","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 02:31:43.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25ce8757a75348a0904caf461f9bd6d4 : SUCCESS in 43m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad711b85ae8e4fabaa9f0888fb56b03e : SUCCESS in 43m 41s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"35f411b63c5d847ccebe8dc93c6711da7c073a6f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 03:38:51.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/30 : SUCCESS in 1h 57m 05s","accounts_in_message":[],"_revision_number":30},{"id":"33d9f9759e42f10468b22221639e100d9e653505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 03:51:52.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cc5d7430d81a4b799ec336c17d69a463 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/34e1119330bf47ac82496c566e06894d : SUCCESS in 41m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3d1b85500ec1480b80ea73597e24499e : SUCCESS in 2h 05m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c2bc2260119045c4a15f18f54cf478be : SUCCESS in 1h 48m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a02c767646e4456b87286115856e8d9 : SUCCESS in 1h 08m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/45c601ce1e084dca8e16d67363ad5447 : SUCCESS in 1h 31m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bc7b7b2e9994ecfaea153b7619ee450 : SUCCESS in 27m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e06ea05e223b4679b777c04a6d8c42bf : SUCCESS in 39m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb8d6fa6301b49488927afd76c89099a : SUCCESS in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a4d4deb06e24d99826ba2163a2a5ea4 : SUCCESS in 21m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd18bb78e04545d1a2e31ac1f98c73f2 : SUCCESS in 25m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e55c823e29664d4e836cbe06b3537a4f : SUCCESS in 20m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5836bb277c8b417aa081db3d10ea99ec : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/904f2453282b40dd915b597a23f89f31 : SUCCESS in 13m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7f1808470ee476f818e95241b11de70 : SUCCESS in 59m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4bee72d28614f96b5bb84f19b955b7f : SUCCESS in 1h 29m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/afa5e57b3b3341aa97d626d7e58ce496 : SUCCESS in 1h 06m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbbfe888ab2f4262ac884058462bbe5a : SUCCESS in 55m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c20dea35639f4f74938f5cbf01853e48 : SUCCESS in 1h 27m 37s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cdc7d952658640ed87259e41fd4c417f : SUCCESS in 1h 03m 51s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c1a35a06e7b74add873a0ba871eb5df3 : FAILURE in 1h 09m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f7b60e2da36a4d0aa5ef189278a75e03 : SUCCESS in 1h 21m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/784450d8c8e44a1aa5b8a73994efcf27 : SUCCESS in 1h 17m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2228ba5fd32141b3a83e9ded67dba853 : SUCCESS in 1h 00m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0da676f2e6604ec591c9dd1112c592c3 : SUCCESS in 27m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b910f2ced0d430d933c38dbade64161 : SUCCESS in 1h 08m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f9d408a43289498581bb6a8a545eb877 : SUCCESS in 1h 08m 18s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"ca7212e2fd12dd0c4590dad8e2620414f3f2fd1b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-22 14:17:12.000000000","message":"Patch Set 30: Code-Review+2","accounts_in_message":[],"_revision_number":30},{"id":"b42cafc76fe3f3cb7a0ed4f0e76eca2362b3f5fe","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-23 08:44:31.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"b5c33e741edb3e82bcd9da0fc25cfddce961ff22","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-23 09:46:53.000000000","message":"Patch Set 30: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"74cf647f09185d4cc68cafb2485c2f1289cd44eb","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-24 00:43:01.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"a39f4e04a5dbfc3c9f899e1aa03b65bd4b2632a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-24 00:43:45.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"2ec63e525bbe00542c1e898c320d2c6e06efbc89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 01:18:16.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f9506315da44c16bd731956930d20e7 : SUCCESS in 28m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fcdc26b0ef12488fae54dbc579a671e4 : SUCCESS in 30m 03s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"3bdd7a3b9afcc3fb24f15e38fa00e817b92e06e1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-24 02:29:53.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/31 : SUCCESS in 1h 44m 52s","accounts_in_message":[],"_revision_number":31},{"id":"517a342da543766af8e925db6451e001a0dc4fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 03:01:16.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/93aaba1812d34b959dba1479acf85b58 : SUCCESS in 3m 43s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/660cb2c4eed740ab840497256d57d379 : SUCCESS in 36m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/de2c90b140ff487195b99ff1e40722d0 : SUCCESS in 1h 50m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/53ebbbf76bc6412d93691b8c1cf7a5f3 : SUCCESS in 2h 00m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8e395f1f545f4e4ba9be76d2231cf830 : SUCCESS in 1h 12m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e76d24463e8147cfbb3221ffd559084e : SUCCESS in 2h 11m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0da0831b5d347b18abe37b2fc0e1c00 : SUCCESS in 31m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ffbb2d37b8543afbc6ccd104d2a5ae0 : SUCCESS in 40m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/323c84d6ec684793ad3cfa597897d906 : SUCCESS in 12m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/798dd8a4ab604fc1a535eb98c56642d2 : SUCCESS in 26m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a86471a03cd4064b231447a1beaca33 : SUCCESS in 22m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c2bbe5504314cc580bf93e4bc5706db : SUCCESS in 22m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64116e61c2342d182fed18f87c325c3 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74ed3f8d8d9d468dab159e4cd201eb66 : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/393566d2988b4695bbf3110037e006a7 : FAILURE in 42m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff9386104b1c4a75ace01f82e3b4998d : SUCCESS in 1h 19m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/607a11a207154202a51e8f41d0271398 : SUCCESS in 59m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8eb1f9c9ac644e85bbf1a749824f2820 : SUCCESS in 56m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8f0677c2854444739c397f0c56e7641d : SUCCESS in 1h 09m 18s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e16cae3e31614fd29c62791a6c1e7102 : SUCCESS in 59m 38s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/306ccb56a79a42a79370dad1bd5e2faf : SUCCESS in 1h 12m 26s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67a35310b74a4159b3aaffd7141d2ab4 : SUCCESS in 1h 18m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92ee56f8321d4d589f3161a4bf91d8fb : SUCCESS in 1h 03m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/71dddd719a46489781ab4f1cd4f80cf3 : SUCCESS in 52m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ff8b939c021d439d9ce16ddb63c5a8d8 : SUCCESS in 28m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/286ae8d3ebd04aaeb6234cd631ab3d67 : SUCCESS in 1h 03m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2c48d1b4571a4dbfb5658a3198be1efa : SUCCESS in 1h 07m 04s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"5c338d7ce6ab4f2fb4270b8463356fa723fb8b94","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-24 07:04:03.000000000","message":"Patch Set 31:\n\nrecheck","accounts_in_message":[],"_revision_number":31},{"id":"12ec644cf3b9037d7290620ed2f35a3ea0d05bd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 07:48:05.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c67ddbd69d04486299494d0f9c34219e : SUCCESS in 41m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29a16cfba2d342cd8df1973a93320d25 : SUCCESS in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"59c51af5680e9805b06fea32ac52b409d53463a1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-24 07:56:48.000000000","message":"Patch Set 31: Code-Review+2","accounts_in_message":[],"_revision_number":31},{"id":"f70b2cc5d2af70658de4707bc78a1dafee0ea7bd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-24 09:34:40.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/31 : SUCCESS in 1h 31m 32s","accounts_in_message":[],"_revision_number":31},{"id":"a190ac40474a998462bd1276cdba381157a498c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 09:41:16.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e6d6bafec2b444f7846d8bbf7b45c37e : SUCCESS in 3m 35s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/90c73dd0ec2948858448533a0d8ed59c : SUCCESS in 43m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/634c7f1c5042460dbca7cebd509366b6 : SUCCESS in 2h 24m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5489a1765d5145edae75a5a6004ee8e0 : SUCCESS in 1h 40m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fb1aca9871bf4398a512f5a1aa0f7c08 : SUCCESS in 1h 49m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8242f4377e05456293c7d56dae9b5cc9 : SUCCESS in 2h 30m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/041d6a721eb24926b4df45f7dc0522d0 : SUCCESS in 34m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/791d0a7b3316434d99f70f4dd96806e7 : SUCCESS in 39m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e547d3e3ac1d4d6fb200bcfc3aea4862 : SUCCESS in 14m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a1d2cb0f1d841d29a164103075edfc1 : SUCCESS in 31m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da538223081e49afabb00670938404d6 : SUCCESS in 23m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e790d0b1281f452ebfdf17f5dba87eb0 : SUCCESS in 37m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd90cbcf5b1c40479815a27f32584571 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/449f274dd4c04fcb9a2fdd5764d77819 : SUCCESS in 15m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0da1e3761434c438ae381a7327443ba : SUCCESS in 52m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b525b8393589441d85830573206655ec : SUCCESS in 1h 37m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/49b4c6fcafc04c35bc1fd37063467432 : SUCCESS in 1h 05m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b3c61bfb4c446519861dc9f7000b281 : SUCCESS in 1h 01m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0985c0a940a6474f9dce6d4616d2d5d8 : SUCCESS in 1h 08m 22s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d64a88d7c81c4e90b22a6796d2fc06d8 : SUCCESS in 1h 16m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a02d3e4d7a90402697fc3d85bb4f1a6d : FAILURE in 1h 02m 54s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/372949c11c6a4348b4b6e89ea10a9a7b : SUCCESS in 1h 29m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bba1c343da9f4045aa0b9ecdf93be8f8 : SUCCESS in 1h 00m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3c226d8bce394ef39b491ed67cbbe63d : SUCCESS in 1h 08m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f4d657764ab14dd0a3844b6000a6e6b4 : SUCCESS in 32m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d0ad67a31c649f0ae6a46a2c7fba5ca : SUCCESS in 1h 12m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e17ef4d3eaa145a882b1bdea7ccad9a3 : SUCCESS in 1h 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"2839f08bf9f8fa23b1885512b950a4ad1a48caaa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-24 13:33:45.000000000","message":"Patch Set 31: Code-Review+2\n\nLooks good, waiting for another last reviewer","accounts_in_message":[],"_revision_number":31},{"id":"4c7049752bffe353f50a7d11576097916c3b322e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-24 14:11:18.000000000","message":"Patch Set 31: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":31},{"id":"508862ca221e94db84411bdf74ed7241d4356e2f","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-26 09:40:05.000000000","message":"Patch Set 31:\n\nrecheck","accounts_in_message":[],"_revision_number":31},{"id":"486adf0d204a79f43d4bda0395058f3245e3667b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 10:17:33.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/749b8a02688d4676904e76bbf0ecd9ea : SUCCESS in 31m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8520e2afdc0444cea5f0bb375c72136e : SUCCESS in 32m 19s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"9afc2c5311968c36d92de3bf8610d6836557245a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 10:18:02.000000000","message":"Patch Set 31: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":31},{"id":"01c931253139c8aa411d5c0b1c76b0d1af210f4b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-26 11:28:28.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776567/31 : SUCCESS in 1h 47m 26s","accounts_in_message":[],"_revision_number":31},{"id":"0985f5d3f8afe1ba48aee52ea50bbd1a1df03722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 11:42:22.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d954246896d4436b801f39051b271173 : SUCCESS in 5m 08s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8fea4fa488f04982b9d41ce6a46bd98e : SUCCESS in 38m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/79c3e8016e4f4b728cfe8c06cc45fb0c : SUCCESS in 2h 00m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/41e4f86858c34d56b6c0028fe43dec60 : SUCCESS in 1h 47m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/213ddaa255d1471082bac05620ffab42 : FAILURE in 1h 06m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/93b734c2b4d84408a521c3f7a09c1931 : SUCCESS in 1h 40m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0721de8d4e884e3380eecfe02194472a : SUCCESS in 27m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ac97718b36a42e79c258422c81c3230 : SUCCESS in 31m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/046a9afc2ed24fea9d0ffe9ceb5f48c2 : SUCCESS in 33m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d442a1dee347edaa3b0a3629ae8f36 : SUCCESS in 15m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8d7d07608354fb99571d7819dc322f9 : SUCCESS in 24m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e097182522ee4908825759d1f4096ca9 : SUCCESS in 23m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00195db4ed114b7eb84d36c81ede59f9 : SUCCESS in 23m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfd02f26fa3f4578a105b4391e258388 : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21726aa640a641ab81bc713dfc9530fa : SUCCESS in 12m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9257c75a0f6641ef9d95698d6ac8c7e1 : SUCCESS in 44m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5bce5b14724438185d300e590bffdbc : SUCCESS in 1h 23m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f543553320bf47c992765c2c6a82923d : SUCCESS in 1h 06m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7bdd20ffbb745738cf8d87364e115bd : SUCCESS in 1h 00m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/68574813377e47b3ba468604ee24e611 : SUCCESS in 1h 01m 34s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0f2a0f6fc37d4a35a5184963471d9b00 : SUCCESS in 1h 08m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e0455559c27340949646a60776ab8b94 : SUCCESS in 1h 16m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dc5a32dddbd24d7fbc7efc78731ee901 : SUCCESS in 1h 21m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/398c51f2a71f49f0930aaa260c75333e : SUCCESS in 48m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/43f7e38760cd4e02ae292997720b9958 : SUCCESS in 1h 03m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7ca1421d7a084c08a420631645681bf2 : SUCCESS in 35m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/29d67e74aff846bdbdfd54cc2d3f48be : SUCCESS in 1h 05m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ab4d1627b5f9491aa82c7b228ab657d2 : SUCCESS in 1h 18m 53s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"dc71d128f4ba57b511ad242ae3cf48b5c14ca7dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 12:12:43.000000000","message":"Patch Set 31: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/87482b8455164e31b711eda0b58702d9 : SUCCESS in 4m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6aa71ed938f04d7a82254a75f2a6a9f3 : SUCCESS in 47m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a4e19c9b43544ab6b79d76e76f5158ad : SUCCESS in 1h 35m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c7efd8a74ec049c189d9a4a6d68130dc : SUCCESS in 1h 30m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1b135fc31c6a4a77bbbfd086c3199418 : SUCCESS in 1h 43m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e42adb0ed51b403ba8a3b32a7b8ac114 : SUCCESS in 1h 21m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4863b0ad876499a92a6a6df426191f6 : SUCCESS in 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d2fb0ff2bf40aa85f02c55885eda9a : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad83bcd704734e14be2f454ed3429125 : SUCCESS in 25m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d69a8c6b301d40f3b08345247d99ea79 : SUCCESS in 23m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e795ff086a4233ba86d2164e021986 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/551752bab3e341129d41e66d39f50c8f : SUCCESS in 12m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f94e56c345da4b629419d2422cae0719 : SUCCESS in 41m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/680d28b0c22742e494959074cd11d30c : SUCCESS in 1h 20m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b7ed6b2cb9ee44d4a4fe84dd97502d8f : SUCCESS in 1h 12m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6717d35a65cd4441a3eb851a52121457 : SUCCESS in 53m 22s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e70b0c8288f409d8fb051ab77930a94 : SUCCESS in 1h 25m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a0102eb3b774738abda71983202d742 : SUCCESS in 53m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0f5b0b87b0f543a4b16348ea646f49af : SUCCESS in 50m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/97be1cc5734345abbd8fb1e211532c7e : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":31},{"id":"d402c090e3234db61ebfd73a86dde09af962dc23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 12:12:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":31},{"id":"c183f41d55768a756b03b4fc5b3dc7de39fc45ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 12:14:19.000000000","message":"Patch Set 31:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f69ca98769ac4e05bee457b25d4edefd : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b85948d8c01c4232950396d998806a58 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":31}],"current_revision_number":31,"current_revision":"56e8498a4dac9e4db6185cc5d97612d824c6e8d5","revisions":{"3160584d7402615bf2d8a0aabc70d94a67df1a1f":{"kind":"REWORK","_number":1,"created":"2021-02-19 01:54:59.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/1"}}},"commit":{"parents":[{"commit":"7bdd3be8a02b75efb050e480325151f4dd79e1cd","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bdd3be8a02b75efb050e480325151f4dd79e1cd"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-19 01:53:59.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3160584d7402615bf2d8a0aabc70d94a67df1a1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3160584d7402615bf2d8a0aabc70d94a67df1a1f"}]},"branch":"refs/heads/master"},"9546269a71c76f7480906b419d318ac15828a77d":{"kind":"REWORK","_number":2,"created":"2021-02-19 04:55:32.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/2"}}},"commit":{"parents":[{"commit":"7bdd3be8a02b75efb050e480325151f4dd79e1cd","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bdd3be8a02b75efb050e480325151f4dd79e1cd"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-19 04:53:34.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9546269a71c76f7480906b419d318ac15828a77d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9546269a71c76f7480906b419d318ac15828a77d"}]},"branch":"refs/heads/master"},"06afa93bfc3ded85f04e7ec04212ab8cef7dfba8":{"kind":"REWORK","_number":3,"created":"2021-02-23 01:43:56.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/3"}}},"commit":{"parents":[{"commit":"a6d2c3df7b67eba502b03dfc3cd5b8c73537e729","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6d2c3df7b67eba502b03dfc3cd5b8c73537e729"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-23 01:31:46.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06afa93bfc3ded85f04e7ec04212ab8cef7dfba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06afa93bfc3ded85f04e7ec04212ab8cef7dfba8"}]},"branch":"refs/heads/master"},"ccf752eb73ab2a3d8208608c530e8b9b396dec1a":{"kind":"REWORK","_number":4,"created":"2021-02-23 12:08:17.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/4"}}},"commit":{"parents":[{"commit":"a6d2c3df7b67eba502b03dfc3cd5b8c73537e729","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6d2c3df7b67eba502b03dfc3cd5b8c73537e729"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-23 12:04:39.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccf752eb73ab2a3d8208608c530e8b9b396dec1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccf752eb73ab2a3d8208608c530e8b9b396dec1a"}]},"branch":"refs/heads/master"},"15f8ca2b2dc002d99efb545a454fd317744d6879":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-25 03:11:49.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/5"}}},"commit":{"parents":[{"commit":"d020a1756e1730d2f0766b4a3468c4705189b0c7","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d020a1756e1730d2f0766b4a3468c4705189b0c7"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-25 01:58:33.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15f8ca2b2dc002d99efb545a454fd317744d6879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15f8ca2b2dc002d99efb545a454fd317744d6879"}]},"branch":"refs/heads/master"},"e097303a8babb766ef3d3382e58a1357815ad1e0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-03-02 06:31:20.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/6"}}},"commit":{"parents":[{"commit":"6cf80301654a5fcd97b011333972709b65ab6bb9","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6cf80301654a5fcd97b011333972709b65ab6bb9"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-02 06:06:39.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e097303a8babb766ef3d3382e58a1357815ad1e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e097303a8babb766ef3d3382e58a1357815ad1e0"}]},"branch":"refs/heads/master"},"1fd18767033a9e00c34fe1cc4c7b32512539e463":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-03-05 10:08:25.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/7"}}},"commit":{"parents":[{"commit":"072290c38396b0f3b2edc68cee17426b71617021","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/072290c38396b0f3b2edc68cee17426b71617021"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-05 10:05:24.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fd18767033a9e00c34fe1cc4c7b32512539e463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fd18767033a9e00c34fe1cc4c7b32512539e463"}]},"branch":"refs/heads/master"},"84ab2de031dc23a04935c1a15eceeb27d650734a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-03-09 02:02:53.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/8"}}},"commit":{"parents":[{"commit":"202f3c59b5ab0eee86f5e244d9fdaaa6d78a7d25","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/202f3c59b5ab0eee86f5e244d9fdaaa6d78a7d25"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-09 01:55:34.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84ab2de031dc23a04935c1a15eceeb27d650734a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84ab2de031dc23a04935c1a15eceeb27d650734a"}]},"branch":"refs/heads/master"},"24ca373023fb9291544775e95b5ea4d495495b14":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-03-09 08:36:01.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/9"}}},"commit":{"parents":[{"commit":"6172463749133eb7af17e3e8be02cf419077181f","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6172463749133eb7af17e3e8be02cf419077181f"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-09 08:34:46.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24ca373023fb9291544775e95b5ea4d495495b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24ca373023fb9291544775e95b5ea4d495495b14"}]},"branch":"refs/heads/master"},"22650b0f2139bc4772d3fd45c91fa7afcd53e2d6":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-03-09 09:06:15.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/10"}}},"commit":{"parents":[{"commit":"25eedad4df5ccf0c13976a60572b583214850f51","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25eedad4df5ccf0c13976a60572b583214850f51"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-09 09:05:19.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22650b0f2139bc4772d3fd45c91fa7afcd53e2d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22650b0f2139bc4772d3fd45c91fa7afcd53e2d6"}]},"branch":"refs/heads/master"},"257218227f57e1fb9efb6c4f2eaba295dd770b4e":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-03-10 03:03:31.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/11"}}},"commit":{"parents":[{"commit":"0381522b628c08484a21c753c4e39574d28d74c5","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0381522b628c08484a21c753c4e39574d28d74c5"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-10 02:45:59.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/257218227f57e1fb9efb6c4f2eaba295dd770b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/257218227f57e1fb9efb6c4f2eaba295dd770b4e"}]},"branch":"refs/heads/master"},"851103e72edf52faf5c241ea758648173d297152":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-03-11 09:32:43.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/12"}}},"commit":{"parents":[{"commit":"be4811fed6aba83cf13c8f156e113e6d8db4adef","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be4811fed6aba83cf13c8f156e113e6d8db4adef"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-11 09:29:29.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/851103e72edf52faf5c241ea758648173d297152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/851103e72edf52faf5c241ea758648173d297152"}]},"branch":"refs/heads/master"},"cec58ae294166e31943cd32500ba12842c395d9a":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-03-15 02:38:58.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/13"}}},"commit":{"parents":[{"commit":"775c4377cfaef572e467bc2f6f6aa6ecdf36192b","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/775c4377cfaef572e467bc2f6f6aa6ecdf36192b"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-15 02:36:54.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cec58ae294166e31943cd32500ba12842c395d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cec58ae294166e31943cd32500ba12842c395d9a"}]},"branch":"refs/heads/master"},"3a4c4e382e9322c3d7288f78af889416745bdf3a":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-03-15 05:34:33.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/14"}}},"commit":{"parents":[{"commit":"49abf8c5d8e137fd87bab28d36ae600a1bd23938","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49abf8c5d8e137fd87bab28d36ae600a1bd23938"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-15 05:32:36.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a4c4e382e9322c3d7288f78af889416745bdf3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a4c4e382e9322c3d7288f78af889416745bdf3a"}]},"branch":"refs/heads/master"},"d8c520feab02fa4ac79ef08f3f3de700f0b47b68":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-03-15 08:21:41.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/15"}}},"commit":{"parents":[{"commit":"b6281a4f7db63f119e43680212c6e519d7be24db","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6281a4f7db63f119e43680212c6e519d7be24db"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-15 08:19:25.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8c520feab02fa4ac79ef08f3f3de700f0b47b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8c520feab02fa4ac79ef08f3f3de700f0b47b68"}]},"branch":"refs/heads/master"},"686b54cfcf489f2daf205690b48b27aa60ac04e1":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-03-15 09:37:41.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/16"}}},"commit":{"parents":[{"commit":"cf1ec008211e010cfc5acd64318cd3ea93766af5","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf1ec008211e010cfc5acd64318cd3ea93766af5"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-15 09:32:19.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/686b54cfcf489f2daf205690b48b27aa60ac04e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/686b54cfcf489f2daf205690b48b27aa60ac04e1"}]},"branch":"refs/heads/master"},"507d55c2615367e83c6d22c35d0b0d37c5029484":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2021-03-30 09:26:52.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/17"}}},"commit":{"parents":[{"commit":"e9ac35c3b777975d6af22bfe1962eb798f44a6e0","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9ac35c3b777975d6af22bfe1962eb798f44a6e0"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-30 09:25:59.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/507d55c2615367e83c6d22c35d0b0d37c5029484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/507d55c2615367e83c6d22c35d0b0d37c5029484"}]},"branch":"refs/heads/master"},"7d64918cbf7d34f152b4e0b1f85d0cca3135894d":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2021-03-31 02:32:38.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/18"}}},"commit":{"parents":[{"commit":"aa173924ad57d4ed6ed0af55a847075005097034","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa173924ad57d4ed6ed0af55a847075005097034"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-31 02:31:41.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d64918cbf7d34f152b4e0b1f85d0cca3135894d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d64918cbf7d34f152b4e0b1f85d0cca3135894d"}]},"branch":"refs/heads/master"},"270e3ac410a4dc0abee311d27bda83c2c8589e62":{"kind":"REWORK","_number":19,"created":"2021-05-14 02:06:17.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/19"}}},"commit":{"parents":[{"commit":"2d48e6f1270e6f88b3f1a994d2bb438f0c276b36","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d48e6f1270e6f88b3f1a994d2bb438f0c276b36"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-14 02:01:13.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/270e3ac410a4dc0abee311d27bda83c2c8589e62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/270e3ac410a4dc0abee311d27bda83c2c8589e62"}]},"branch":"refs/heads/master"},"b7ec07c3d600e7603919e3654ccb45412b1bb6c9":{"kind":"REWORK","_number":20,"created":"2021-05-20 03:31:03.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/20"}}},"commit":{"parents":[{"commit":"357c79297c2cecc1428014f1bf77bd96004b8774","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/357c79297c2cecc1428014f1bf77bd96004b8774"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-20 03:22:27.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7ec07c3d600e7603919e3654ccb45412b1bb6c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7ec07c3d600e7603919e3654ccb45412b1bb6c9"}]},"branch":"refs/heads/master"},"4c5ebe7ad8d3749c154c79577d87c050c4cc25b7":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2021-05-21 00:53:13.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/21"}}},"commit":{"parents":[{"commit":"5441e9d77b3a4692fd5fc71d3c57e92fc844432a","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5441e9d77b3a4692fd5fc71d3c57e92fc844432a"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-21 00:52:00.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c5ebe7ad8d3749c154c79577d87c050c4cc25b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c5ebe7ad8d3749c154c79577d87c050c4cc25b7"}]},"branch":"refs/heads/master"},"b1e59a58667892011cb06c2790f4b8bfa662e329":{"kind":"REWORK","_number":22,"created":"2021-05-25 03:30:34.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/22"}}},"commit":{"parents":[{"commit":"7faa0ea285d2238750a6220ac0bf6470014f512c","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7faa0ea285d2238750a6220ac0bf6470014f512c"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-25 03:26:59.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1e59a58667892011cb06c2790f4b8bfa662e329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1e59a58667892011cb06c2790f4b8bfa662e329"}]},"branch":"refs/heads/master"},"23aa0c530e03f6271a31a6c80f1c7c7554c51f86":{"kind":"REWORK","_number":23,"created":"2021-05-26 11:12:30.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/23"}}},"commit":{"parents":[{"commit":"847a57eefd44beb2d7bf14960ffd0d22dc7f3faf","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/847a57eefd44beb2d7bf14960ffd0d22dc7f3faf"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-26 11:10:14.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23aa0c530e03f6271a31a6c80f1c7c7554c51f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23aa0c530e03f6271a31a6c80f1c7c7554c51f86"}]},"branch":"refs/heads/master"},"ae3a94d318cc2e0abb7b2052afbc677c62757438":{"kind":"REWORK","_number":24,"created":"2021-05-26 12:27:09.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/24"}}},"commit":{"parents":[{"commit":"62ce5012ca8e2e9a93fbe0ed9c7d20223714252e","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62ce5012ca8e2e9a93fbe0ed9c7d20223714252e"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-26 12:24:51.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae3a94d318cc2e0abb7b2052afbc677c62757438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae3a94d318cc2e0abb7b2052afbc677c62757438"}]},"branch":"refs/heads/master"},"403c69542943c1b884dbc12dd86cccc3a6f610b2":{"kind":"REWORK","_number":25,"created":"2021-06-07 03:40:41.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/25"}}},"commit":{"parents":[{"commit":"deea7731a80b0bf7c83c5c2bd08a890a3512f2e3","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/deea7731a80b0bf7c83c5c2bd08a890a3512f2e3"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-07 03:25:54.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/403c69542943c1b884dbc12dd86cccc3a6f610b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/403c69542943c1b884dbc12dd86cccc3a6f610b2"}]},"branch":"refs/heads/master"},"d7b15a5eaee7403eabd6c3beb23c37899df8e03e":{"kind":"REWORK","_number":26,"created":"2021-06-10 01:27:34.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/26"}}},"commit":{"parents":[{"commit":"a292bea168941acfce5c2b058c2f178e8ff5c1ae","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a292bea168941acfce5c2b058c2f178e8ff5c1ae"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-10 01:15:25.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7b15a5eaee7403eabd6c3beb23c37899df8e03e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7b15a5eaee7403eabd6c3beb23c37899df8e03e"}]},"branch":"refs/heads/master"},"3a0185a4a8bb13a086f4cc9003de2577024494cc":{"kind":"REWORK","_number":27,"created":"2021-06-10 08:04:15.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/27"}}},"commit":{"parents":[{"commit":"a292bea168941acfce5c2b058c2f178e8ff5c1ae","subject":"Add the DHCPReponder for IPv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a292bea168941acfce5c2b058c2f178e8ff5c1ae"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-10 08:00:54.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a0185a4a8bb13a086f4cc9003de2577024494cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a0185a4a8bb13a086f4cc9003de2577024494cc"}]},"branch":"refs/heads/master"},"4b057dbd3888f3384972da044a9c09326e7676ab":{"kind":"REWORK","_number":28,"created":"2021-06-21 06:51:36.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/28"}}},"commit":{"parents":[{"commit":"b08310df9f9061cfdcbc1d4428739db6de3e0e7f","subject":"Merge \"Add localport const and refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b08310df9f9061cfdcbc1d4428739db6de3e0e7f"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-21 06:43:52.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b057dbd3888f3384972da044a9c09326e7676ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b057dbd3888f3384972da044a9c09326e7676ab"}]},"branch":"refs/heads/master"},"b5efcd4869c756c027fe32e7ddec167e1d130782":{"kind":"REWORK","_number":29,"created":"2021-06-21 10:30:09.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/29"}}},"commit":{"parents":[{"commit":"b08310df9f9061cfdcbc1d4428739db6de3e0e7f","subject":"Merge \"Add localport const and refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b08310df9f9061cfdcbc1d4428739db6de3e0e7f"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-21 10:05:06.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5efcd4869c756c027fe32e7ddec167e1d130782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5efcd4869c756c027fe32e7ddec167e1d130782"}]},"branch":"refs/heads/master"},"ebd5307f84740038af2c809f9f0953786b72e39e":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2021-06-22 01:40:35.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/30"}}},"commit":{"parents":[{"commit":"2e403a3bfbf297d0955cf9e2f9ff21a4809337bb","subject":"Merge \"Report ExternalDNSOverQuota exception for recordset quota error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e403a3bfbf297d0955cf9e2f9ff21a4809337bb"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-22 01:39:20.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebd5307f84740038af2c809f9f0953786b72e39e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebd5307f84740038af2c809f9f0953786b72e39e"}]},"branch":"refs/heads/master"},"56e8498a4dac9e4db6185cc5d97612d824c6e8d5":{"kind":"REWORK","_number":31,"created":"2021-06-24 00:43:45.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/67/776567/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/776567/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/776567/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/776567/31"}}},"commit":{"parents":[{"commit":"ad2bc847ab6768ed1f0a6ac8415c331040cded60","subject":"Merge \"Bump neutron-lib to 2.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad2bc847ab6768ed1f0a6ac8415c331040cded60"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-06-28 02:55:01.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-24 00:38:38.000000000","tz":480},"subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","message":"Add agent extension \u0027dhcp\u0027 for ovs agent\n\nAdd a new ovs agent extension to support distributed DHCP for\nVMs in compute nodes directly. For large scale deployment, this\ncan be used to reduce the number of neutron agents. Large scale\ncloud can benefit from it.\n\nFrom the perspective of virtual machine, this will reduce the\nprobability of DHCP request failure. The VMs will get a higher\nlevel availability for DHCP R/R, no single point of failure\npermanently. If one host goes down, VMs in other hosts will not\nbe influnced by it.\n\nFor the perspective of network performance, after using this\nextension, the DHCP broadcasting packages will be limited\nto the host locally.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: Id8a4c501daad7c2185e6d69441182666ef987e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56e8498a4dac9e4db6185cc5d97612d824c6e8d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56e8498a4dac9e4db6185cc5d97612d824c6e8d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
