)]}'
{"id":"openstack%2Fneutron~776568","triplet_id":"openstack%2Fneutron~master~I40d6464953fbc4d4ca2c05a6051eba43cd05fedd","project":"openstack/neutron","branch":"master","topic":"bp/distributed-dhcp-for-ml2-ovs","hashtags":[],"change_id":"I40d6464953fbc4d4ca2c05a6051eba43cd05fedd","subject":"Add fullstack test case for OVS DHCP extension","status":"MERGED","created":"2021-02-19 01:54:59.000000000","updated":"2021-06-29 08:52:41.000000000","submitted":"2021-06-29 08:36:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"776568-bp/distributed-dhcp-for-ml2-ovs","meta_rev_id":"3fab16731c7e7a72f86767bb2a95c0fb5f42ccf3","_number":776568,"virtual_id_number":776568,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-29 08:36:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-28 19:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-29 06:22:09.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-29 06:22:09.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 02:17:28.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 03:03:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 03:52:48.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-19 04:03:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-20 10:49:21.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 09:16:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-18 11:20:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-06-29 06:22:09.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"867a94477cfef1e4d06b9fb6e32aa173c6934de1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 01:54:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca37e9baa56e91c6cd26e49b545f33355504dd26","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 02:17:28.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"fc69fdd9a67a6a35c818d2260e0e95fdb7dcc25d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 03:03:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01506dd23fa8400fb3cb7d377387d08b : FAILURE in 53m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa0d6c59997447259e0735f98f2d576c : SUCCESS in 55m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ab839127ab6f25480b7b95a6720942ca6da2359","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 03:52:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/1 : SUCCESS in 1h 50m 10s","accounts_in_message":[],"_revision_number":1},{"id":"fb2b003d552908d991767000ed1565a55dc802bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 04:03:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eaf92c1cf54e451fa03a4d774a5d7e78 : SUCCESS in 1h 00m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7e9d52c86bb647c1ba36746aeff868e3 : FAILURE in 2h 01m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e63265666c7498d86bb24b074e08800 : SUCCESS in 1h 49m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d7bbd005a1f64bf7b430bd2a0d067ec0 : SUCCESS in 1h 26m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/60c323c3d6704ab89b3289ab2ab700f4 : SUCCESS in 1h 51m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57948a0be8734f988a2c00a553544057 : SUCCESS in 43m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/716d5fd979194eb1aa290fcf06ab753f : FAILURE in 29m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86b5c8fec01949d29df0ff961a1bab7a : SUCCESS in 13m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e42a37e016704ee1b9e1e97589e7240a : SUCCESS in 32m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fdcaa11b1d24312b7b753c337ca3f86 : SUCCESS in 30m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0817ca88a3014902b26e9073ba0fd713 : FAILURE in 33m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939fe2291cec4dbe83ceb3892b4e6c03 : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/61244ca0032545488360f6c337168370 : FAILURE in 44m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ff93e52d92f4fc8b2ef026197d74c2d : FAILURE in 1h 05m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/17789c2925364e48a960ad97d6f44ceb : SUCCESS in 1h 30m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8b5967867ba4a9e9f8a07792d53a85d : SUCCESS in 55m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/59b4bc789dfb4dce9fcaf61f7520e836 : SUCCESS in 1h 04m 58s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0ff23d896f4c45e1b8b668fd1734720b : SUCCESS in 1h 23m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/54f8df9a1ba34c05a178453dfe203b40 : SUCCESS in 1h 37m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/473549feacce486f9ccc2dd764417524 : SUCCESS in 1h 39m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09f854f85a7e41e6a3075ddd7afae9d3 : SUCCESS in 1h 00m 33s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/45353c07321841b29ac0a3e726392a9f : SUCCESS in 1h 07m 59s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/55f9123784b64a35973eb64d3c0285d6 : SUCCESS in 1h 11m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1ce919b5a612448c85068616b63d06ea : SUCCESS in 38m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9945040b62a147a7bb573a1140c126ec : SUCCESS in 1h 37m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5c606f9d49b94036bee2465cf292206f : SUCCESS in 1h 28m 03s","accounts_in_message":[],"_revision_number":1},{"id":"30f47bd0dda4df828885453effe15de7f6367b32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 04:55:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"385b1e8c2b17ca99a8365adcdd39c4d843f74cfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 05:53:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a1936828db7444fb5ff57352c5054b8 : SUCCESS in 48m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad34cd0f7f704f7b846ba37d62f7edb7 : SUCCESS in 47m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b777431f2ac7b4ef2a01244be413e7e0e70de7aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 06:48:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e5a46078988046a3999f52c5b4c1a6ef : SUCCESS in 57m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cddbc4d5abac44d581304b36985a7b2a : SUCCESS in 1h 20m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ad3875b749a6475cb0aaea86d25a0a8c : SUCCESS in 1h 43m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7be88b0e33ff4774ade1b70c9310485e : SUCCESS in 54m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d1ce9eb137194e53a74d1e399516cdf0 : SUCCESS in 1h 47m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b0db89fa9f5451fac1b181dd982a3c5 : SUCCESS in 32m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c14ac3da49f486c808d4838ebe3b785 : SUCCESS in 37m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4722c7889f244ce6b1e32ec0b1e57c7d : SUCCESS in 12m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/913dba90a7f7460d95af569f50cf6ffc : SUCCESS in 33m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfbf98249fa640d38019decfb0141d14 : FAILURE in 25m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d8b1de1cdf341d8a75093f4de482aad : SUCCESS in 30m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dadb29e159fc4dfcb5c1b6d59b11d32e : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9dea847b37a74ad696c16d1163198f35 : SUCCESS in 55m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/31d35ed89a874b7b9d491b885f2a334e : FAILURE in 59m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/27996571675e4d4fa19e53e4369ea475 : SUCCESS in 1h 45m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a35c1140da24244b4b9d2d8fd5f34b7 : SUCCESS in 1h 01m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cbbf2dcddaf047f8a76e85c480514d7d : SUCCESS in 55m 27s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3da9eff0af8c450bbce444d941e3b76e : SUCCESS in 1h 24m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/98601b54c3134497a98870a3193c9562 : SUCCESS in 1h 29m 13s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7cd3b97aabab495995ca2827ff682d1b : SUCCESS in 1h 41m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d4aecfb04cb4a8382c8abc5d8d9646c : SUCCESS in 58m 30s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8452595dac2a45a091b2626ba0ea33d2 : SUCCESS in 1h 02m 27s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8a82a7f0a693453bb8f7083e5f1a0b81 : SUCCESS in 1h 15m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/96cdfc90e560439ebb47646fe4b1a9f0 : SUCCESS in 34m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c1e631dee75a4975aec9bc5a1044bc2d : SUCCESS in 1h 36m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9294cd4e9f2e49bfbb4a42b2c21490b8 : SUCCESS in 1h 27m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ff8fd29c4120533cc0cb0e9ecfb8afed01b5ac3c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 07:00:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/2 : SUCCESS in 2h 01m 31s","accounts_in_message":[],"_revision_number":2},{"id":"88b6555a4533b80009ae9d0aff7f7bf1b894bb80","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-19 16:00:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cc96085c288702b20ae6a6ff7604fcdaf1535b12","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 16:04:01.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776568/2 : FAILURE in 42s","accounts_in_message":[],"_revision_number":2},{"id":"545db835c07bf0314a9af6567838ce1720b1fe7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 19:01:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42c87e1f07828162434d0e33c6dae6ca4d4d5101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 20:49:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/29b2fb753190464a8aed10deaf8226fb : SUCCESS in 47m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/357871de2e264dc8a44057d1d618a871 : SUCCESS in 1h 46m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d620f1d6d1bd4237bf5a5c229164520d : SUCCESS in 1h 30m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d848ab1fcdab42a9b55072e48fb91b20 : SUCCESS in 1h 33m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9d5add316c494358b4501c9595cbe860 : SUCCESS in 1h 42m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7bcee096fa64806bf72651e241ba0fe : SUCCESS in 39m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96341b43d59a474f971199dfa21dfc8e : FAILURE in 39m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d4cb4af84634907ae5dbfc484bc5c1d : SUCCESS in 14m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81bcc57f7a57411a930c9a9f08557329 : SUCCESS in 28m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60c3cd2c8f2043d6886c9660c219ce10 : SUCCESS in 28m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d54ec44e5074f3089141473bda03870 : SUCCESS in 38m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf550d6b46d2428ab970b61362fe4439 : SUCCESS in 9m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8eac09a364c7412a8b0061bd55f05b89 : SUCCESS in 46m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab48d5741d884ff3a524d4a6efbd3248 : FAILURE in 1h 02m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b69c0de18c7f45c393444e453974bc45 : SUCCESS in 1h 40m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e76bcd7b3a2a45e281a6b4ca7524a0f6 : SUCCESS in 1h 06m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3d4f0e7d73d6449d8423741eb243b32f : SUCCESS in 1h 07m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f1fc083e983f4d469f5624a64b1a9401 : SUCCESS in 1h 28m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6059f345f4ec4fe1a66887a526ac02b3 : SUCCESS in 1h 25m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dfaf773ff41342889c40bf2d81d55b97 : SUCCESS in 1h 36m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1686dc4f39b34a2f9ed223c26f4fe385 : SUCCESS in 1h 00m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/de5abe75c0984f42ba8fb4c9e9b19821 : SUCCESS in 1h 12m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/edab82b5870f4f1d892159e68c2d83f4 : SUCCESS in 34m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e61093634b824297a9fca1e4f10af8b4 : TIMED_OUT in 2h 13m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5f05d66570724d8a82dc125deaa8b6fe : SUCCESS in 1h 25m 19s","accounts_in_message":[],"_revision_number":2},{"id":"ed627baa0c6e74aaac61a9f92526c34dcccd72a9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-20 10:49:21.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/68/776568/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"970208e5073ed83ad0ce7a29df982a7fdb413109","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-20 13:29:45.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/68/776568/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"cbe9878efaedf57cae2226de6a807fdefd1728ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-23 01:43:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"482f39acd4d6d6e307951908ad8969aa5a69707e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-23 02:39:07.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776568/3 : FAILURE in 48m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f9c20fe0f7949ab5137d08815e2e74869c79b3a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 02:57:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e9d9dae6b124b6094a9e905cb0145ce : TIMED_OUT in 1h 00m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75f724dfe19d4aa4b4e4017b5a5a68a6 : TIMED_OUT in 1h 00m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6c1e2ea2473d7775c2733164a011985c0aa72a05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 03:36:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ed305127005d429081285b2136fce13b : SUCCESS in 45m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6adf1868c9d240c3ae090f5f75d59bcc : SUCCESS in 1h 48m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c247edff2b2f4c04897259e99de7b57c : SUCCESS in 1h 28m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/841fe1df88c04808aa48f3b308e946f5 : SUCCESS in 1h 18m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/76368509a1834a7fb679eab30ee242b7 : SUCCESS in 1h 34m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e3e5acd234947429534ace5af24b74d : FAILURE in 39m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b26806aed24e45c29cbfb9715f2acd92 : SUCCESS in 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0047ea20645346b0bf1036f99fd58a78 : SUCCESS in 13m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24bd43583d02486a865419971693f0a6 : FAILURE in 41m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dce9208ba4604bdc91da400336c641b0 : FAILURE in 39m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f7559dcf53a4e1dabfac4a9d3465d4e : SUCCESS in 35m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e3fed6460fc4776a5f371f41e2126c8 : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2beafaa4ba34fc59ac17aaea7d095d9 : SUCCESS in 53m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2537dbb0c07343a19c2409483a258e81 : FAILURE in 56m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1f90508dd7a84e92ba1216d2e201aab8 : SUCCESS in 1h 26m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0834a67d1481495c827998be65f67bd2 : SUCCESS in 59m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/28b2d0e96ad74651a1bc44d69cf51e81 : SUCCESS in 1h 01m 25s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b63568c2b4b14bc884c68c7ccd8455bd : SUCCESS in 1h 05m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/da8c9f582588455d9bb0fe46ccc19a50 : SUCCESS in 1h 32m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ac87b8ccdfd41a3a2f6882ad6826574 : SUCCESS in 1h 29m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43d1fb27a223402c913db10b5da989ef : SUCCESS in 56m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/471dbda4f3434aa8bcd3c10da3474118 : SUCCESS in 1h 03m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ec7239ef049f48d1be19f198641a7cef : SUCCESS in 27m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a6d1f83c1554f848dec3fa7fdd59b54 : SUCCESS in 1h 23m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e7d72cb8653c448383caccb671cc3d87 : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":3},{"id":"9c34751be41c1b5140d6c9097b9791e0d6aa3bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-23 12:08:17.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"460427168200e829b3191fb37cce3eb9acb490cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 12:55:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/035098eed5524333873bc43dd3eb6791 : FAILURE in 35m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46fe29466c6b4e5ca141f7d8ba0342c0 : SUCCESS in 40m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"19ec4c6cca30e8e8a1909ff805cbe804c3731659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 16:33:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9e22095d45774b5695c41da21a97e086 : SUCCESS in 49m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b8cd39c5148c435482c4e2b6d87cb62b : SUCCESS in 1h 20m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/97d8a4a632044d368557fc99bb427858 : SUCCESS in 1h 22m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e3310db2cb92413c8d72fe9bbc3b282c : SUCCESS in 1h 24m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/33c61d1f580e4205bcc3bd51e0b82e9d : SUCCESS in 2h 02m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98826e8851344830afa042bf2e1876bb : FAILURE in 46m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82a25504747e41eead3a96f112bb4137 : FAILURE in 28m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9c65701679149f4ac3cc5b1b968aa57 : SUCCESS in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9da18b4374434c7393609c702c0efb0c : SUCCESS in 22m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ec1dc0523c14bffbd48f09e25de30f7 : SUCCESS in 26m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbdd14e3c4864885aa6f8fb6ea8ccb55 : SUCCESS in 28m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3426afe78bc64f0f9618034b46de1099 : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fce06c1b43847bc97301acd2f90a212 : SUCCESS in 46m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b532a36115d04eaaa379c2459a0bdf98 : FAILURE in 58m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2d67ebacaed54a53841c1d327da1ea0c : SUCCESS in 1h 30m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/32431ab21f7148daa4e27d050f66c00d : SUCCESS in 54m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/613a2c9076e84188b79b4e47b259dccd : SUCCESS in 1h 30m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b6feee766a3f47abaa3acd373bfd5b50 : SUCCESS in 1h 37m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ccf0f15de4eb448e9e50022f2b7a3938 : FAILURE in 1h 23m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f287fd8a665440d3b043f455b35288ae : SUCCESS in 1h 40m 21s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecdc8de76fe64c71ada0f83fd22cd7ef : SUCCESS in 1h 10m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a41db8857e59434ba68d09bbf234168a : SUCCESS in 1h 20m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d5178de7c1b741db9d7e454ade59adaf : SUCCESS in 33m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2f80cb4e193476cb13ae99d9a8289f4 : SUCCESS in 1h 29m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b60d965b9ac34b45b3a61dd0aaa4cc09 : SUCCESS in 1h 27m 04s","accounts_in_message":[],"_revision_number":4},{"id":"2e52f0a44ff283dcead7d470670312f281b01177","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-23 22:43:02.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"5484b61a46743f920cd44de6d37e0e1475ab3dca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 13:26:02.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/68/776568/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"e7875787bb7ad1fc34342ab4ecbc37e1d21aee8d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 16:08:59.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/68/776568/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"5318421dd5e8afc0ae8c04ed2b6f346656514b66","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-25 01:06:20.000000000","message":"Patch Set 4:\n\nRun this new fullstack case locally, it was passed.","accounts_in_message":[],"_revision_number":4},{"id":"b9430f6b6ef663752a046e3031dc8dadaa8169d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-25 03:11:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"23c2af27ced91c06bc16f50cae9a2a0b40898a45","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 03:33:50.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"b142f61de42e72af42dc044d6bc36b42254bdee1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 04:06:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/599aea5a6f4144eb85cc0886a865a974 : SUCCESS in 43m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a128e40674904b2086c24ecc798180f7 : SUCCESS in 42m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d8c308e65ebe104798afdc396bb878d48ab49c21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 05:57:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c4447aafd01b467480b17b282982c1af : SUCCESS in 58m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2d6779eee0704f1eba6ca3b9bf788d24 : SUCCESS in 1h 40m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/52f195519ac041f597fbdaf1f676b97f : SUCCESS in 1h 33m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08f8bb568cc5420c8e9ac0cb70793d26 : SUCCESS in 1h 48m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b23df6adfdd04e438dbcfb3a3849ae7a : SUCCESS in 1h 59m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b5e947417ec427aa5e70f0b8c685f10 : SUCCESS in 43m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97cfcd4487f0479c83a213565f05197c : SUCCESS in 30m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/799bab0b790841f181c8d8d9f592f99e : SUCCESS in 13m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/543d4a12162e429ea4940477176db17f : SUCCESS in 30m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da34fa37bfb1483baa704976f546bf4c : SUCCESS in 26m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/174f94709fe3435d996688dd16aa9dd7 : SUCCESS in 29m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58bf9ed8b7df4bcfb933dec5ac0a81ff : SUCCESS in 9m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/324c04fbb3d14f6f9b90f1f4dfd1bb39 : SUCCESS in 53m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/616c4972913148d58890149d385f976e : FAILURE in 1h 06m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/04a320cf6107478e8db82d808c6db41e : FAILURE in 1h 53m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2410dd89d2b04b8a82417a9aae4cb359 : FAILURE in 49m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/709f0853547149719eb5b6d527d14d28 : SUCCESS in 56m 21s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3b638347ddba40bb86e217067b8b9c1c : SUCCESS in 1h 27m 10s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/09600c3e7573444f95ca82de80d0d096 : FAILURE in 1h 10m 42s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd7ec298241743489e58875ce19be3d0 : SUCCESS in 1h 53m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf4c14adeffb4d7190c2f68d5fe57575 : SUCCESS in 1h 07m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e55ec3ea9c434aaebac13dbea537fef4 : SUCCESS in 1h 22m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/45323b56ad27411a8d021a3bf7c813c9 : SUCCESS in 30m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/53d0564124cb483eb94b769a79b6d8c9 : SUCCESS in 1h 31m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/77d2bf85203a40b1a06cf740c1213062 : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":5},{"id":"98d3bc11ce8b6361ec533738d253af63e898a23d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-25 07:33:33.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/68/776568/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"ce70d795ead7f0a9bd9b1fe7d4eb087b48e30cb4","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-25 08:48:57.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"73f0ad25d0be3268eef84b00a29fc35b47983ba4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 09:40:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b8797adbc0c489e96560eedec805d8e : SUCCESS in 44m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5813571f98214920a3b48b65e6da0da8 : SUCCESS in 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1ebd5c35e10a468a06d595e78858ef8097d61206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 13:01:11.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fc18dc6dbbdc40f187243ef2fc675ab3 : SUCCESS in 49m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3dfa79391d72444b9acf2d9ada04f120 : SUCCESS in 1h 24m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/decf1e4d173d4d7ca746fc3d06c29680 : SUCCESS in 1h 41m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f24c40d3956b4ef2a041409cf2073ef9 : SUCCESS in 1h 28m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6def2bb4f77f4920969e38ef9f4b7cd0 : SUCCESS in 1h 47m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10279e5e91ca422cafd5114fd6338f09 : SUCCESS in 25m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5401c4e46cb14f34b6ad356ff06ca300 : SUCCESS in 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efec56860b92478cb0b6537930954c92 : SUCCESS in 11m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44bed0bed9954c16be14ac44a87c1803 : SUCCESS in 19m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf885b45a8d346a1953e559ba91e5325 : SUCCESS in 28m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e51ac56bc93e481facd8745debb22a62 : SUCCESS in 32m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2d93a5eebe43f99c47fbdc2e5cd1c1 : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/17d43a046ef145eaadad31a48bc3c491 : SUCCESS in 48m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b6d98b4d93b46e9879b08d7d1052fe6 : FAILURE in 1h 04m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/99163d64cf7f4deab892a9b4a653dcd6 : SUCCESS in 1h 32m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c957f05d5ac24f4abcd73e23b7e45522 : SUCCESS in 1h 00m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/704c03af44dc4deba73fb8fac9b7648a : SUCCESS in 59m 40s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f4fb939419c445bb962ddcb38bdbf6d4 : SUCCESS in 1h 21m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/85c9b9964e5745f2830928225caa2a63 : FAILURE in 1h 21m 08s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d2d88212f654c2b984a8abe15e2acbc : SUCCESS in 1h 31m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed0c16dcd6194ecb92ddec35bb57653d : SUCCESS in 1h 04m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9927eeed1c36423089fe7356f2dd0e3d : SUCCESS in 1h 26m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3930696e146849d6a0ccc6bbf7f71f2d : SUCCESS in 38m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/975dc05bbd7045b5a913b430c3f09447 : SUCCESS in 1h 44m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/40750c31091d426dad920f88a8630741 : SUCCESS in 1h 38m 14s","accounts_in_message":[],"_revision_number":5},{"id":"18e7fedc2747f0f15f82f07a34cd305c3714831a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-02 06:31:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2030961687f1d00df37ee3df204468a08f14843c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 08:29:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8679a494d1074390b605c801f162c7ad : SUCCESS in 38m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/771b565554f84829aa3241b44f5e63a5 : SUCCESS in 46m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d4225523e4c129ce1104b151d92dea101352ac33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 09:46:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/42884ba067f74e099ed411a4eb071d3d : SUCCESS in 53m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bc40647666ac4148b711e188b2733d60 : SUCCESS in 1h 28m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a54e0bf652b945788b965ecf174ea663 : SUCCESS in 1h 38m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ea601d05d52047aea8b4a41679291de5 : SUCCESS in 1h 38m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ca4ac170811f4764a9267c73a71ea54f : SUCCESS in 2h 44m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d691aeeb828440d93dbcace85fd1f59 : SUCCESS in 49m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20740ef75c25493688ce9696b274f579 : SUCCESS in 40m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc3dd60e7d9842beb628126b175e306d : SUCCESS in 23m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af846256ae984a1ca0743d5544f4afce : SUCCESS in 31m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e174b215166045f6a0abe4ff70e581ed : SUCCESS in 27m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90addbd578774496a378718a10d736b0 : SUCCESS in 25m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74603214cc694c30be4ac4c1374c089b : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/79f1c1ccf7ee4b0fa547b2ec671c5418 : SUCCESS in 48m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eed20e770b72490ab46515393ddecf0c : SUCCESS in 1h 15m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f2ec30f777ef49769b79a8549b59019b : SUCCESS in 1h 29m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/14a3088cb38d4dec848e5e7a6f68a936 : SUCCESS in 1h 00m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ddf1d1f73e0b4ecdae4698ff4cbf1dcd : SUCCESS in 1h 03m 10s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6f20b9513ca94200bcaed6c7d9921931 : SUCCESS in 1h 21m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d434e1ef17e04830ae2d1f4bd1dcc2fe : SUCCESS in 1h 27m 14s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4fb6782258d4920aadbd120064fa09f : SUCCESS in 1h 33m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/545e7a88152e448496b68aca3c83d0cf : SUCCESS in 1h 11m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8696c27c1a6c482fa96cce7fb481032b : SUCCESS in 1h 09m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b969596434b443ce91758ad2657d5625 : SUCCESS in 35m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0750b2de337f4b55ab3262441c566bb5 : SUCCESS in 1h 24m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/87144d4dfa434a29a9fc901f3aa7130a : SUCCESS in 1h 24m 19s","accounts_in_message":[],"_revision_number":6},{"id":"66f81ea46078a9d71d585bc25f8175a6aa217302","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-02 11:03:56.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"9db6f2c623e4f31766655e18eec1d579da9015cf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 09:16:18.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"00ddffcd924bacf981156e4ae0d9d7951cdb297b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-05 10:08:25.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"843307c39e2ec54f75fa89e10f75fbf0837feaa5","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-05 10:14:09.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fb7ad93206453cc37013d7f67e1a3573c1f41564","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 10:52:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bcd9bd597e94c248328885d7ff4221af63f1df2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 12:30:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/03cc640ee89a44059ce5af24228085cb : SUCCESS in 45m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d2ef8aac13094a12823b14a4705b34f3 : SUCCESS in 1h 39m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fde923df58094e788654814ca138ed64 : SUCCESS in 1h 28m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ee269bac0dd74e978d9c4e4768e08f08 : SUCCESS in 1h 23m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bac1ab495eee47a3bc9ecdf4f023b133 : SUCCESS in 1h 34m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8068908edf043ce9236f3dcdf161cf3 : SUCCESS in 40m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90c361b167004f408b261039aac61403 : SUCCESS in 33m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9956e377e3714b5ea3d1ea726ce2e74f : FAILURE in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a955502426b4644a5b40a62a6bc8d08 : SUCCESS in 43m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2a320981a44436ca962d3b1f9883aad : SUCCESS in 28m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/979bb3c92c63460ba1d0083951aaa688 : SUCCESS in 32m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e40c84e806f4a319b3c0c3d0c967719 : SUCCESS in 12m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ff7aeb9379143d1ad58070f5d5acf9b : SUCCESS in 56m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fe5a7a545884d85a0f6d28475dafeca : FAILURE in 1h 16m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dd7785d623484126ac1f5a4156a2c92c : SUCCESS in 1h 27m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9ddb2108a2f4a649018de0ea1464781 : SUCCESS in 1h 16m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0e607b22cbc9457fa898428402ce6f95 : SUCCESS in 1h 04m 27s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/395c49e1409044c58b4d1bad088114dd : SUCCESS in 1h 23m 56s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ac566f4e3fdc47519933296207f789bc : FAILURE in 1h 15m 08s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3574a42f201a48618c48a5a9514048bb : SUCCESS in 2h 04m 10s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16db598149ae4f6286914ef20a4ba7de : SUCCESS in 1h 14m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b2513d52685d45adb63c1539075340a8 : SUCCESS in 1h 29m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c4ff62c94c9d45bcbbf4283644782984 : SUCCESS in 36m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/59c1ecd4d01b484faf13a239c894effc : SUCCESS in 1h 32m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4f2ba84236ad491785867847ed32e0af : SUCCESS in 1h 34m 21s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/agent/l2/extensions/dhcp/test_ipv6.py","accounts_in_message":[],"_revision_number":7},{"id":"0635ae4f2d30522da1194334597c66f979e8a7e3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-05 19:16:16.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"c1ec9d796d622fc47de9aa4d93563bc92420d819","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-09 00:45:09.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"24657b1279521738a50b8af8884e51333f20fe8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 01:32:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4534e173337a4252b3c800dfedab9118 : SUCCESS in 43m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d5bf84ed83642e1aae6952957eecdcb : SUCCESS in 41m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8be4291d99d6b6b067c02143a8479ebcc07eb205","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-09 02:02:53.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f82cefe161121a3f4ef0673303b58637ad1a4f45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 03:04:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c74ae98f5fe456badc13e078a48b863 : SUCCESS in 42m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6df3df7a090a4f408c935e8b49ec6c79 : SUCCESS in 52m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"921da5b28fcf9702e1ef6a739ecafe0ffb7f295c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 04:22:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/481e3603699c4a498d748f7193a87eec : SUCCESS in 54m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1e8459c3e537483e93a51b50e25f71c3 : SUCCESS in 1h 40m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8af0f352bace4a5ba1d3add513d59959 : SUCCESS in 1h 37m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05151bc52a034c219d8dfbdf3230a2f7 : SUCCESS in 1h 08m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/27a367282c8b413da7b73db7fd59ad7a : SUCCESS in 1h 41m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f950003fb8ab4a09b9409199e8888caf : SUCCESS in 34m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86c0b4b3b2834dcb9d9d12269c2c9398 : SUCCESS in 39m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/625289d1abb843af96beb874d60486ff : SUCCESS in 13m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27c8f86576e54b3f97bb33c0798fc7b2 : SUCCESS in 30m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e4b45620b2640ba99d3826255788282 : SUCCESS in 39m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4508b679fc144bd9090d250cc94d06e : SUCCESS in 34m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255f0f77c7cf4f9a9399a51a59d354d1 : SUCCESS in 11m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bafd2ce66aae4c87ab989eda79233c94 : SUCCESS in 44m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/232b627e3a844dd395e3a167aececac9 : TIMED_OUT in 2h 14m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e755359b0ced4a99810bb9cebe31befa : SUCCESS in 1h 35m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e28ac8131e64bfeb25c4c6be38decfb : SUCCESS in 55m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/17636cc42fdd4bd9b37a90d73e99221a : SUCCESS in 1h 09m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1b861182f1234c89bd16076979178c55 : SUCCESS in 1h 50m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a399f7f53130494fb7b025dedd03b3cc : FAILURE in 2h 02m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eab8401d6fc846289cf9f953133dbc72 : SUCCESS in 1h 30m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afc072cd83c4455fb257bad8683df064 : SUCCESS in 1h 06m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3fd962a94b9347aeb48ee1d5c2837986 : SUCCESS in 1h 03m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/12938ceb365d40efb8649d009407690a : SUCCESS in 42m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/01b5ccca695049b08baa0e0d7ece26ba : SUCCESS in 1h 21m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d5128d671b1b4b688c1fa65e1c47c045 : SUCCESS in 1h 24m 22s","accounts_in_message":[],"_revision_number":8},{"id":"bc0bface724f1e86032d74414e06766fa2e1acc9","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-09 05:52:17.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"8ecea04a292ef505326d5aa3fcb63f2145311577","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 06:40:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9677763ea152472bac11f817ec2bda99 : SUCCESS in 42m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82068887d3594e73a6dbec17eb7ef8cd : SUCCESS in 42m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0db93be25ba9198d20386cbfbbc2f36890b0b264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 07:49:19.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/06928c0dd0bc41bf995f3d575ac614df : SUCCESS in 50m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ab5e9b1310b14599addcf5e4412b30fb : SUCCESS in 1h 22m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/68a2cd0297dd47ed9cbdeba7e120a313 : SUCCESS in 1h 05m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e7ccb721fd1d4692b6535888dbd5eaf4 : SUCCESS in 1h 14m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/71533a13dcc54c8aafc2ed2ba67f416a : SUCCESS in 1h 51m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecb7e39ca0044337ad496bc7a69d5e6e : SUCCESS in 27m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30390bba53f8490885eec390259197d4 : SUCCESS in 23m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ca43ef02654a2899bf4a2c18c77e52 : SUCCESS in 10m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd393b1e6d664a39aa040aeaec5b7425 : SUCCESS in 28m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0e512a0896445deaacd9a38c9d0438c : SUCCESS in 27m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b0d96e9beae4661a7d2abdc15692a75 : SUCCESS in 27m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a8ce84154747f29c93723bb94d8134 : SUCCESS in 9m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8c2a27c29e94ab3b2ff2642e59caf57 : SUCCESS in 49m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f93f599a6a5644fd96d0af6d5b7682dc : FAILURE in 1h 23m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/55f33b3e9fbd4707bfa9d5185bae4d9e : SUCCESS in 1h 03m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1b389868ba14aeeb1457a4c2f31b6de : SUCCESS in 53m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a3cad322bdaf46a39b6401b14f512ec0 : SUCCESS in 1h 00m 29s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0877f0565a4b409a9a659a24ee09dc6d : SUCCESS in 1h 32m 58s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2437fd1bc1c446d6b42ad31e88f998ea : SUCCESS in 1h 10m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ea6c0c4e5b65422c894698f085467a43 : SUCCESS in 1h 51m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6537e01f7412461fac933429e014d48d : SUCCESS in 1h 19m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7ba324657a344e3d88cda4789a68ec9f : SUCCESS in 1h 22m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/10ff68e828044e7d87969fa765973b22 : SUCCESS in 31m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/71a3bef6fcbf446687938740640ef420 : SUCCESS in 1h 45m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f2aca6bc573e46c1b95edebf31b0d52d : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":8},{"id":"72d895266b209772131ecce32dcfe54681c9524f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-09 08:36:01.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"c6ebfb6dcb2589e5aa90d2403320e71de1d7ceee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-09 09:06:15.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"c04b0e6126cd6a001f736fb32219efae36e979c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 10:06:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8754f90a965469aa07e2940e2c2b402 : SUCCESS in 49m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9bb3450b4df276345b9a72311576fc815014b066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:14:55.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aa04a31283874bf3b0d07764b673fee7 : SUCCESS in 48m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/baf1122e69b14574a27558e19b310d79 : SUCCESS in 1h 37m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7811e802cf9b4a3ba326ae1df91785f8 : SUCCESS in 1h 30m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f7adc4b302ed404f958120ff76f6d65c : SUCCESS in 1h 12m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1802ae1cffb4433c901acd2bd1d85043 : SUCCESS in 1h 31m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6df03690a8fd4a8aae997cea1116baf7 : SUCCESS in 31m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24a9c772192441c4aa796c84cd2073eb : SUCCESS in 28m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25bf7baac9aa42f981b84d8ad30320f0 : SUCCESS in 13m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c16b255b7fc9456f96dd2c953bd80992 : SUCCESS in 38m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e32762e699944915a65b7c01180a827f : SUCCESS in 44m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3ef29527e554c359215da0e02a39767 : SUCCESS in 33m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716e3b36ef054bc78827174bfda8f774 : SUCCESS in 9m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/66a092939d6a423583db5c74702dc308 : SUCCESS in 1h 00m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b770e338cfdf4b8c98d9e8c94285fe49 : SUCCESS in 1h 43m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4d1336e9e56e4f4d9768944ced677c1e : SUCCESS in 1h 36m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a8d37b29cd847b3998831b8fe8e2368 : SUCCESS in 1h 01m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/75e381454b2241f0831e1f6de98f92a3 : SUCCESS in 1h 11m 05s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1c126f28f8714be386c38294f056658c : SUCCESS in 1h 12m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ec959d6009cb45a38f42ea676a07b294 : SUCCESS in 1h 59m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c30923e959d24a43bbeb1374800a8973 : SUCCESS in 1h 47m 54s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/016e255d66d54945aad7afeaef1f1cc4 : SUCCESS in 1h 01m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f3d6445f6e7a481a913c0851812f223b : SUCCESS in 1h 09m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/00fafd06920f48a6be906001dc347b77 : SUCCESS in 37m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2a29665c31ab404b8c5f0167b3e331a8 : SUCCESS in 1h 42m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b6874da64fd8453fba023737109c5623 : SUCCESS in 1h 27m 04s","accounts_in_message":[],"_revision_number":10},{"id":"7ee513c98653e6d79a6a5b265a9010f6dd7eba4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-10 03:03:31.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"8c3fcb66bf73152b035e1433fd7838e82b8e2ce5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 04:03:06.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67a4b7e14a4a46c6b52b56f98bdd4b88 : SUCCESS in 42m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6545cbe9542847e084a055d3b353d62a : SUCCESS in 51m 22s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ae0d5b8e4b5220cfe5f5e8baea252dfc3a88cb2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 05:21:08.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b74f7a3a397446168891cf0331c90fbd : SUCCESS in 1h 05m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fbb68d40b4c5416aa31ea929a6fec879 : SUCCESS in 1h 40m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f4d6981128fd49749cae5ab347a485e1 : SUCCESS in 1h 43m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/94dd4539740b4212bbef91e8f64a91e3 : SUCCESS in 1h 41m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1da2a2908300423c96f96f9e5ab444ee : SUCCESS in 1h 52m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9944a015a5a4885883bb69c0c4d9588 : SUCCESS in 42m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa17f5f0c07c41b4b1bc6bdd20197780 : SUCCESS in 48m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/600fdb378a2846fdbce017f37be43d89 : SUCCESS in 12m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d193e4bf3e743dfa4f82915ad5f962f : SUCCESS in 49m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3df6bb8a5fff4f7baba24e920045aa71 : SUCCESS in 34m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3faefe0d82ee4305b196a35e232a3b88 : TIMED_OUT in 41m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ab16d4504248bc8ce22bdfe8af6414 : SUCCESS in 10m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5251e12dfdfd4b7185a1feb523a8cdc0 : SUCCESS in 1h 05m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d45332502b64494a9097e2f0e142e48 : TIMED_OUT in 2h 14m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ead0b04e310f44139f455e2eadf2580b : SUCCESS in 1h 19m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/536eafd964764256b1f57b4f49830648 : SUCCESS in 54m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/77438ad0a0ce4e9288ec8e8a260c3403 : SUCCESS in 58m 55s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5e2ade24107548ae9946c8ea26662a1a : SUCCESS in 1h 16m 44s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/399a5631ea4044388b3c1099ad892bcd : FAILURE in 1h 04m 58s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c41030e40be94b28ad6a46da1315f211 : SUCCESS in 1h 31m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/daa7b65ee3b040e5bd2f5fa8cf882bac : SUCCESS in 1h 08m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cd7321e2fb2c43acbb9bfb5c5953d7d7 : SUCCESS in 1h 09m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7fdf46b5988842bea5e19925ba0ed51d : SUCCESS in 38m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a15fc3f2d19d4344bc72ecc6803934c0 : SUCCESS in 2h 00m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3d8a5b7e4268474998897ad6b200ea9f : SUCCESS in 1h 19m 07s","accounts_in_message":[],"_revision_number":11},{"id":"61047844ac6d7423482e1c74712cd039eda46a59","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-10 05:29:15.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"4632370c1691ea7384e3e107cdd4c0647e9daef6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 06:21:15.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ecdc3875a4d42168b3cc56502dc46cc : SUCCESS in 49m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76c234ad94944a0e882fdcbdb0c7ecfa : SUCCESS in 47m 25s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2aa1fd1b2ff68b54eb71cab52e34a196a2292f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 07:55:52.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e8f9d0057248445f8d1153df9405874b : SUCCESS in 44m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c7284e44131d49a28f6ec1a05ae7d068 : SUCCESS in 1h 50m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f9b1206f673949bfbfa7179c04be3c29 : SUCCESS in 2h 23m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/18d0200d2f194dda915e398b4329b2c0 : SUCCESS in 1h 15m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1fbcda8659044689abd81e9157c238ff : SUCCESS in 2h 10m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89b698576302432590ac27a21cb4db22 : SUCCESS in 51m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6589d5fcedcc443ca3da7e17b7419535 : SUCCESS in 47m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f952befb9b1e4e1a98c6f1d443a80aab : SUCCESS in 12m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/078e262978e74a29b7e501253707c666 : SUCCESS in 54m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c2f2b92d64347c381c37c40d11aaa27 : SUCCESS in 47m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa4f252b5e2d480aa9240fe7a7b02576 : SUCCESS in 38m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a787f7c1b2c14f08b8c3e1842a46d037 : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/09868428baab4cd2b72e0ce65d9f0cc5 : TIMED_OUT in 2h 17m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/89bfdb7280064666bd9eced734a49092 : SUCCESS in 1h 39m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/480afebbea714afa965f4f5d8332e278 : SUCCESS in 1h 38m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/78aa0212942a462180df906e256e9e25 : SUCCESS in 1h 06m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/445c56553b8c4e9c86d3fcdac73bdb2e : SUCCESS in 1h 02m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/07306f9c63be4c598b6e1e562317d38d : SUCCESS in 1h 28m 31s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/23a6ecaf8b564bff97a9c056958713ca : FAILURE in 1h 19m 41s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a03ff3ba08e84b37a6dd86283b4f09c6 : SUCCESS in 1h 46m 43s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b0d8cf2bcb1436d981c305a35454f27 : SUCCESS in 1h 00m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/76352fe2d6024171b0598b6574969bd1 : SUCCESS in 1h 12m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2f0d7684f99440fc81f580b20b549a85 : SUCCESS in 34m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4d60f5cae4964acb9817e86d99202544 : SUCCESS in 1h 27m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9cfa222ab2184bcfaa45bd13b41aaaf6 : SUCCESS in 1h 22m 05s","accounts_in_message":[],"_revision_number":11},{"id":"cc960e849346a90d6205f298834d03ca9b7fd445","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-10 14:26:35.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"a6bdeafc8162f0023e81d4c86dbb333665c23367","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:45:26.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c450f5ddc06d4e54b7ccf7454ccefd5f : SUCCESS in 1h 00m 03s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ce3900e96a66763f2b50c8e21643b8fd60dd58a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 20:16:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/729037bed9f54f25ae4686f4c25bc9ab : SUCCESS in 46m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ff2a237a4def4fca954c5a08cb3706f1 : SUCCESS in 1h 45m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/88cb6d58cd2d4a35a27ddc2f6036a4a1 : SUCCESS in 1h 27m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/92d7b8fe60e14f68b7df55e56e4833cf : SUCCESS in 1h 55m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/490470f4aac0454f833d60dff0725098 : SUCCESS in 2h 10m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/080e482540fe49eaa350f20145972277 : SUCCESS in 38m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1945dbca077b41f9b13a04dadfb8e183 : SUCCESS in 34m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f64e7632444d579251da1f74e5bfc3 : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b01d840026634d94bf95c6635ab63f3e : SUCCESS in 49m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd2e43525a23485bb1009b58038eeef3 : SUCCESS in 37m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbdef4e1433f4f7787897a923405ee9c : TIMED_OUT in 42m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a52e577ccdc4f258fb10ef1ca1c2bbf : SUCCESS in 13m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d417b67323a740698ab24ba05c7bc4c2 : SUCCESS in 56m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2625b97557da4122a567650e160401b6 : SUCCESS in 1h 26m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/53509a4c7dd849ddbd09f7245f35f9bd : SUCCESS in 1h 45m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d337be82055140629d52f42d694de39c : SUCCESS in 1h 08m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8cf01a05baa94bcbaa44718c3184244c : SUCCESS in 1h 12m 22s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/49a25541b38245fab44bda96603bf312 : SUCCESS in 1h 20m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/77e4088dee9546c6823fd350e181bde5 : SUCCESS in 1h 37m 50s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c369405235034fc1aa28a1c2e734dfb0 : SUCCESS in 1h 33m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f801fec9025b4086aea75a25433bd4d5 : SUCCESS in 1h 13m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d2a32ebc3c21488f928b1e138d55058b : SUCCESS in 1h 23m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6e18d6f58e8641db9a0013e866e44d37 : SUCCESS in 38m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c0fd7e4f1a64e318b13887cdf490f2e : SUCCESS in 1h 50m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/11ee13c13bde48cf8ee2a3a56c1445d7 : SUCCESS in 1h 33m 44s","accounts_in_message":[],"_revision_number":11},{"id":"17d310670149bb34db4b5156040a8677247565d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-11 09:32:43.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"e94370f215053578fda9074d86199311f19cd920","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 10:49:36.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1e8248420f74702b88bfdac0a546451 : SUCCESS in 44m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e74d76ba6307449c814319373937a286 : TIMED_OUT in 1h 02m 00s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"27586ea1e47bc19c1c588c27da3c5942d8bf81d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 12:01:36.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a60f3ff844c14c7eb4d67898ecc10b37 : SUCCESS in 1h 08m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/81803e65d04e4822820d74a64d7b8db6 : SUCCESS in 1h 20m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/be5ddf294def4a0b8f8eed2a8e721636 : SUCCESS in 1h 30m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/97a28eb5ffd24a9bb35f65815fafd55a : SUCCESS in 1h 31m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0eb9ae5dd1db43d293d32446b148d92e : SUCCESS in 1h 41m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95f54c3388704683a31d1bfba156d88c : SUCCESS in 50m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28f5f2f6d5f34aa9b1f61e3b98dfb21a : SUCCESS in 38m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8135118054d243c59c726848a4d0f9c4 : SUCCESS in 15m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae02dbb14f994d7a9bfadbb0430dee72 : SUCCESS in 46m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cb300f6a0904c0ab09db1e03cd35784 : SUCCESS in 1h 00m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d845a0c0393d4cff8479d296fa4000fd : SUCCESS in 36m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7792db3355134845a0bc35be50ec05a4 : SUCCESS in 11m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e813696a63d044e89dc1ad83e48202b9 : SUCCESS in 57m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fdaa6df11b74dc2b935070e46b429c4 : SUCCESS in 1h 16m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5c6dde582bd04c97b5927befe4f70c81 : SUCCESS in 1h 26m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a49d838b1c5492397e8a0aa026ddc7f : SUCCESS in 45m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4ddff9f136ab4816bd2253e906ef158c : SUCCESS in 1h 10m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/312c81326c7540b185dae80f4332372a : SUCCESS in 1h 23m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/de32e396d81d4b18848c956530a060e0 : FAILURE in 1h 09m 20s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fef8d05d9b4a486782f2ce7bdb92e143 : SUCCESS in 1h 36m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c8f2f5580d74a4592583c687957dbca : SUCCESS in 1h 04m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ef55b14830774ee2a070303a9a4e26c2 : SUCCESS in 1h 12m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/038cb763058640c9a1f7a55c45025c70 : SUCCESS in 44m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8896f489e36d459d88f3811d2ba1c5d1 : SUCCESS in 1h 37m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/96970eb6bebd4600b39bbb2969d46f95 : SUCCESS in 1h 39m 23s","accounts_in_message":[],"_revision_number":12},{"id":"e9e8f2502ea7d08a047e12d9b9fd16e3770c0be4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 20:50:48.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"b33a225f63787ae8e228be250a6480050e4522c5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 22:20:26.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"45083cc51ec8ccc0487a7f2a3024f5abc85bc5b3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 03:48:04.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"1d4d5b8b4dfec587534cf33822f1b62bcdd7421d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 12:59:43.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"f1955f114dd4313c84005dfb23af13bd267e5551","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-13 02:41:44.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"08f3103df238f3c0f5f7e6ebfb2d2a10f9a6c15e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 02:38:58.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"756892517c3db05419a4e52fc80323d5f0b50460","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 03:48:51.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b74f1bcb1d974537910c3d3c9737b412 : SUCCESS in 57m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d87b90c391ee473e9c2fede8c1f62115 : SUCCESS in 56m 58s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c3213a16b947e993042af33fa7c327cfe6dccbff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 04:44:11.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8ab49ed80f6c4b3d9292afeb426da8b4 : SUCCESS in 49m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/949959b955d24e609462ad49c24a6fe3 : SUCCESS in 1h 47m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7b84111719c044119e76b633d3ffaa7c : SUCCESS in 1h 43m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/149490a79bcb40758e306163532ad4ab : SUCCESS in 1h 23m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7a68bede2c014251ba0a45161270db09 : SUCCESS in 2h 03m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6cf46a7717b44849154ad43a55f4213 : SUCCESS in 43m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2d8ba28c7e642068f9740434b320c81 : SUCCESS in 32m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b8faaf6fe0241558fa51f8096ec9e55 : FAILURE in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6f6c036791d44f1bc0851d7b5621654 : SUCCESS in 37m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/588ff6c5b7964919b6d04664d2cb4711 : SUCCESS in 37m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5bd85912eda4350a87147a18f4912fa : SUCCESS in 30m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe1776fcee54c8c9a3bf687c9e8ae0e : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8ed44ddf7184cc4962610f5d8c9a7a8 : SUCCESS in 47m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7262145c6bc7417a9ef52e302d036cad : SUCCESS in 1h 25m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bac90d1079eb49128a27235e100c32e8 : FAILURE in 1h 32m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2d89f8ef2de4ad5b494444c8c2c5994 : SUCCESS in 59m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c81e69621ef040708a095493d0e22ec2 : FAILURE in 50m 18s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9554695ab26d402cba299b6ae2bc8b00 : SUCCESS in 1h 22m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1b4abd21a8374952b7a53d873d7d2e11 : SUCCESS in 1h 29m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a67ce8061dca44149650ce1f66fba8a3 : SUCCESS in 1h 39m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b23cbdc091984b3eac62cbdbebbcfec3 : SUCCESS in 1h 11m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/14423d7fcea14bc0b31fcd93f3f63ad6 : SUCCESS in 1h 07m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/408024eab0394400b886af887733bb3e : SUCCESS in 34m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/de29a3aeab804dc19b4c321f1132deec : SUCCESS in 1h 37m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/df9fe8e6e09e420e8d0270f55eb60f29 : SUCCESS in 1h 34m 42s\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/dhcp/ipv4.py","accounts_in_message":[],"_revision_number":13},{"id":"888c40dc1da197ac414f7f52b1c2a2004ca3ed65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 05:34:33.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"76afc1974c6bb911a0c1645ab83029b306da4c09","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 05:47:20.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"d501bae8bf3cf7dbf69d84170db3824a08e7c4c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 06:50:54.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9614aef301394bc89ea6d0140654f351 : TIMED_OUT in 1h 01m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9a227ac48d24e5c8aed4c1f25376c41 : TIMED_OUT in 1h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b647949d2b29f046a7d2802f25f68c3eb374e527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 07:26:15.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1e6b221d965f4ac399b4706d14947697 : SUCCESS in 36m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2897a03546784c64b040fe3664a51540 : SUCCESS in 1h 33m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0f9f1cbe3e364b17ad456d15395ff419 : SUCCESS in 1h 47m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/23cbb05838cf4516aac9811a4419ea75 : SUCCESS in 1h 24m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/44f27564be104c05aee55117f67a8ed3 : SUCCESS in 1h 42m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74fdd575720f469fa7936f7542c4535d : SUCCESS in 37m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e576534ffd24311a7ac42b40a0bc311 : SUCCESS in 32m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042fadd1d4c44cdc8f0526fd54bb73ce : FAILURE in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7363f8c19f8e45b892047868030814c4 : SUCCESS in 38m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b6df701f74440afb75e5cdd19838bf9 : SUCCESS in 32m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85f3f9ccc03e4b858173c9722fe90bcc : TIMED_OUT in 43m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161d99097c524a8aa4914826cc0c64d5 : SUCCESS in 11m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8534b7f11a5244088400ccb79d9c7ca5 : SUCCESS in 57m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddda74a6d0614573853f4a196561b463 : SUCCESS in 1h 24m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dcc7ce7dab72463595b35b516db863ae : SUCCESS in 1h 34m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a51da896451045b8a780d9bc0712c64f : SUCCESS in 1h 06m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5fa4879d7c8742ba8c22b70a50fdd9bb : SUCCESS in 1h 11m 00s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/90a2dfa60ec54703a90e60c55b555821 : SUCCESS in 1h 26m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ef1b4156f36d404882379e90cbe6ac23 : SUCCESS in 1h 33m 23s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/51872e7f0b4848d09ca8e2ba7f813a3e : SUCCESS in 1h 39m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25dde285ae2747f9b25551f455377470 : SUCCESS in 1h 11m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bed156ce3704423ea1758e322bec749a : SUCCESS in 1h 14m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7099c95b9d354390808f49a2aa354c17 : SUCCESS in 34m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7b7919c82b9b4958957cc48ee6981a9d : SUCCESS in 1h 42m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4958d80fc7924906b67052cd4490a851 : SUCCESS in 1h 41m 51s\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/dhcp/ipv6.py","accounts_in_message":[],"_revision_number":14},{"id":"5ba8fb5efa0fd3c7d2deb13043f3b6d3ed147c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 08:21:41.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"6f1182c3441ff905f03172294ba59a0025ff64bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 08:35:20.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ab45dd1e745675a6daec0f1da00615f2e75b7f57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-15 09:37:41.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"818f0782c0c8aa6fc2e0190fbc8a1b271afae69d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 10:03:26.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"d65c0414eba1a2b49d4b6cc94f2fb7b6a4d0267f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 10:45:58.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72450aeee4274f29ab1911fedc02b3ee : FAILURE in 56m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebe6a72529a148ab822f7218c6a8d10c : FAILURE in 59m 27s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"7915beeca9ba00857986604c0b52fa156a28e6fb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-15 11:42:22.000000000","message":"Patch Set 16: Code-Review-2\n\nWe are now passed feature freeze deadline for Wallaby. Let\u0027s wait until master will be opened for Xena release to continue with that.","accounts_in_message":[],"_revision_number":16},{"id":"ea3ec24f5f410a72c6d1e704e28dcec906f89dbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:46:20.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0154ba59665d475cb9886a4fb00bbf50 : SUCCESS in 53m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7c7b63c0390341b4b41061a20691dea3 : SUCCESS in 1h 44m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a88c2865592f43f3b6b6f902907db760 : SUCCESS in 1h 16m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f689e19998c84623af6892536eea5f5b : SUCCESS in 1h 32m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b1aa4ee792b3406699e63d1d0e96a463 : SUCCESS in 1h 51m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d98e0cb1bd484cf897832ae711599ece : FAILURE in 37m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c4413542c1d4f7e9508e9204aee1092 : FAILURE in 33m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/108aa3033fc74ee88da2d9264292e1f8 : SUCCESS in 11m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4163ec2c26234096aa4fe81f0d7bdae8 : FAILURE in 45m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5de11d7cfefe439c9243908379ae3d8f : FAILURE in 38m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68426b910b48444ab45bbfb7dc29ca9f : FAILURE in 32m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4441647240ce417c9bbaba27d5368ebb : SUCCESS in 10m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/562d6e394a034109a0ae58288e96a025 : SUCCESS in 50m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/43ec633f804642a5bd7c3cd344660e5d : SUCCESS in 1h 28m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6b82bfc4162b405887d11a20e203801a : SUCCESS in 1h 28m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/64abd0144a5840dfa31747ebf0fdc910 : SUCCESS in 1h 03m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/10c3454c2fe04fcabdd667a096f3d31a : SUCCESS in 1h 09m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ebceb91ed89749d4bb4f3f9bf468c073 : SUCCESS in 1h 33m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c7994d0adc34bcaaf824f0edc628a37 : SUCCESS in 1h 27m 22s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ef84021627744e59afdb234ec1c87e7 : SUCCESS in 1h 39m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc4ba20bff11439da87510239443e439 : SUCCESS in 1h 06m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/58032a990bd64b328243880b66a2b102 : SUCCESS in 1h 20m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8a908edfc64a4bd3b7871865ab05d82b : SUCCESS in 34m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1a8d1f6569447c1ad8e1015e520eccf : SUCCESS in 1h 24m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0525756a160f417b9a8f7c54ee50bbf4 : SUCCESS in 1h 26m 12s","accounts_in_message":[],"_revision_number":16},{"id":"d9bfe0baf516c7299dee0c0ce18dffed58b798f7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 14:25:37.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"601801a7de82ca35abf5c771ed56558ce5a0fb85","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-15 17:04:40.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"fe5417c85191b673420226f7ad09ae2b2e91fe6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-30 09:26:52.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"265474ff1cd5293ab02956d4ee5b4543c75bae02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 10:23:46.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/145f6b89eff6408ca295d55be69a0d74 : FAILURE in 44m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/912e5c1107eb415592bbb4a3a258fd18 : FAILURE in 46m 06s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0ee77df582a43b1f8b07fa9299477c87bd231332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 11:54:27.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bc4fccad48584e1f9819b42f11a66947 : SUCCESS in 38m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dd49d504f9204daab75bd741b3c6c129 : SUCCESS in 2h 01m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/59f29d805ae94eb88ddf42b806faab06 : SUCCESS in 1h 38m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/787d196515664adc86d10814cf389f59 : SUCCESS in 1h 33m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/48179924ba614e7b99816ecba5e35ead : SUCCESS in 2h 21m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/213604937df1490b9d6b64122a8a672f : FAILURE in 32m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/329d2b61205940e8a92b5f7d5cbc08f4 : FAILURE in 37m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3af41e9c81940a0b453b10a132cc9b0 : SUCCESS in 13m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72243140f36e4ac1b7282d15b573f422 : FAILURE in 27m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c762fe1b3c44f05933e9776d78e08df : FAILURE in 27m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9052433cb3844230914ac2c39110bd18 : FAILURE in 28m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da982d8ddf3498ea6a48173dbcfd2bf : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3842fc69451347a080e677d126703fdb : SUCCESS in 54m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a476b6496c34d92b205af59c4288fb3 : SUCCESS in 1h 26m 01s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c50639f3752846648adc068c3a040b07 : SUCCESS in 1h 27m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a00e4d73514488ea9778687eecefd3b : SUCCESS in 1h 11m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e3258b28f8c04ae6b4059a2e4945ef70 : SUCCESS in 1h 08m 03s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/86e2964b0862451f84ce9181977bb2be : SUCCESS in 1h 30m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ae552b52bbfe464fa57c913901695d6b : SUCCESS in 1h 29m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e589170e08744724a2e0f1299b849468 : SUCCESS in 1h 37m 19s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e014b1881bce457dbaae3c430aff3237 : SUCCESS in 1h 03m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/13f8441d5ea34004b3e3fa93eba8a600 : SUCCESS in 1h 12m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/04b540b0bff740509cf9a1004f4f97f4 : SUCCESS in 39m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8fc85f9e769b48cf95d9cb3f7b843382 : SUCCESS in 1h 28m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e92dd0a9567446d4a08c6ad57ba45136 : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":17},{"id":"d7daccfaf1d0796635bd38943bbbcf7749e248c1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-30 17:40:53.000000000","message":"Patch Set 17:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"300e3b0c6da16869d35dcba65df817859856bbaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-31 02:32:38.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"a44948806b8f406853ec213d5cb7dc0f7c38707b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-31 02:48:30.000000000","message":"Patch Set 18:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":18},{"id":"7d434e1da286995c23864ae25523b1661096f1ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 03:20:45.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1eef1db0c1c64ef1904aa3e2cad94f64 : SUCCESS in 40m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/969fa83ba98c44dea9c2ac58984a9aba : SUCCESS in 35m 50s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"c6217420d286f2d2eb446adb4f1be10f1489c794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 05:08:32.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a30f4cfc9e034af8be3ca057ca0453d7 : SUCCESS in 43m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9aa5e09ecdad4974bf8351b2f6ad2083 : SUCCESS in 1h 40m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d6e3284682724796923ac0f6eb217859 : SUCCESS in 1h 22m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e9e4a5757904381864db2d354f37311 : SUCCESS in 1h 25m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ef5b1de888e8488e90b08932ee71af34 : SUCCESS in 2h 28m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b934d7b318a647c5aca073f56d494270 : SUCCESS in 37m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5660aafc6eeb4622baeabb9a4dc84add : SUCCESS in 39m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904d28b514a6450abdd531005b21a719 : SUCCESS in 13m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b446f65aa3a84a078a5014fb5c87f2c6 : SUCCESS in 23m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79f603e229648c38dde4a173819e8fa : SUCCESS in 30m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd013334ed354a04a59bbd8269eb58a4 : SUCCESS in 30m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f84009fcd74c15a9f583be7fcb9591 : SUCCESS in 12m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da11f80e937a4aab9ed885a501a54d64 : SUCCESS in 54m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/05fe575bd4f14c74a7833597293c6bd1 : SUCCESS in 1h 15m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/029c92144e2645b592e96fe970c998ff : SUCCESS in 1h 30m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec94b28f77bb49a9b51910eba52de72f : SUCCESS in 1h 03m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2a0000a9ac644fc99ead64968a4cc2c3 : FAILURE in 44m 14s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/07b919d749e9409c93e5fe942972a137 : SUCCESS in 1h 23m 05s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/02b1bdbed5744465ac14bb440d264b60 : FAILURE in 1h 21m 26s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d93cbebb892b4a27a8cf3740468813a9 : SUCCESS in 1h 42m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4635b24d207d468da0f9b9ea4b521861 : SUCCESS in 1h 02m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/baaaf784c6e44708b8ca8f3af14fa352 : SUCCESS in 1h 14m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/38a8629a48c64c7e97a8a64cdf2e6a95 : SUCCESS in 31m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a29b88054eb044b99d330782e4822ac3 : SUCCESS in 1h 27m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b635cc7a24774037a6261b071439e2a0 : SUCCESS in 1h 27m 51s","accounts_in_message":[],"_revision_number":18},{"id":"f26a82f756a2a1c004c71627b033b7122fb1b5da","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-31 06:58:43.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/68/776568/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"c675dd09d8e77c56022e56baa9c00de007955f08","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-07 14:45:37.000000000","message":"Patch Set 18: -Code-Review","accounts_in_message":[],"_revision_number":18},{"id":"992517a8d48d91b262d22cb7d628e85b7e3a086f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-14 02:06:17.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"f71ce93af8d95669ccbab19c5fd5d52e640ff61e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 02:53:30.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9815e4908b946758abda3b56c0dfe86 : SUCCESS in 38m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f6ae707ef0641a88c09b5cc87ad5812 : SUCCESS in 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"a327e972a8b41c0481e4a3bf44c6b2159bfa6fd3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-14 03:44:15.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/19 : SUCCESS in 1h 35m 30s","accounts_in_message":[],"_revision_number":19},{"id":"2ed85e91a41d9293058c83e58075121dae242f8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:07:15.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0279629279c748f299fc11186d93167d : FAILURE in 48m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/befb7b8db4fe43d8bdebf0e92fac1a95 : SUCCESS in 1h 58m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/01b73bbdd8d341ce88018f1a9d5f9498 : SUCCESS in 1h 45m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f3ddca08c12445aca05b5fe958e865e4 : SUCCESS in 1h 27m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/56e5165b4e3a42b784de75360c6813ac : SUCCESS in 1h 48m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e946022db6c40c7ab8dfb2a24c31b6c : SUCCESS in 39m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1819164c0ab4b37935a16ebdd09defd : SUCCESS in 32m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e68065f6774f468cd3d073daec3b4b : SUCCESS in 15m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8eee6d3ca97742ac81015943bd322f5e : SUCCESS in 31m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27077f01528145e78ab9023f5ee644a4 : SUCCESS in 25m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f08abcbca60467fa5d9d577ed275597 : SUCCESS in 29m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91666964c3f045ddb45a2e445e04724e : SUCCESS in 9m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b308867f24a496099a5f6199d82ad85 : SUCCESS in 45m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f397b8b6e8764efcba203230d4e7f30d : SUCCESS in 1h 09m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fb157b130334435eb05431abce19c479 : FAILURE in 1h 33m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/483a1640eea546a4bb89d8b3590635b5 : SUCCESS in 56m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a7fd2b3908ef4240baa4b1da1ba40bd9 : SUCCESS in 59m 47s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b47fec1613ee4d478806c79ea2e9cef6 : SUCCESS in 1h 16m 10s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6f0020d42ba2404cb8e07df24f3bbdca : FAILURE in 23m 21s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/702259a2b6ed4d4ba29fa87fecb7eca2 : SUCCESS in 1h 31m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f9a07ed99bf473b80def33deb543785 : SUCCESS in 1h 04m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f5af87bdfa374c8bbc6fd1a463d67eab : SUCCESS in 1h 08m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7197be31093c4abd9d336eacf6acc5af : SUCCESS in 35m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd39fa7b35054d99ac953f38724fa2bd : SUCCESS in 1h 54m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2a1f43e21f194beca6e1ac312a812921 : SUCCESS in 1h 20m 36s","accounts_in_message":[],"_revision_number":19},{"id":"81e6e1321d936572d411639c4a21f26d10734b45","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-14 06:25:49.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"fdbb1d28fee5179292945b38c914fdb26ce885c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 07:36:29.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9058a76222a479b96c9c65f186fb91f : SUCCESS in 49m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d783c38fe88a4092a2a93aa1275d1e29 : SUCCESS in 53m 42s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"d0fb2308f5d876efc3b6f10d8bb3efa651b5ec5f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-14 08:32:38.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/19 : SUCCESS in 1h 58m 43s","accounts_in_message":[],"_revision_number":19},{"id":"f2f11001b62463013387fdca1711e2b415995ada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 08:55:43.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/257605e8b65f407eb415d451d312032e : SUCCESS in 35m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a318fcc9815b40d5a6bcc66a91e7f69e : SUCCESS in 1h 40m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/732a32592c9442e392fa336dfd63cc15 : SUCCESS in 1h 50m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eb98bf6685cd4d76b3c9a246c98d6e8a : SUCCESS in 1h 18m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b4614da4457a4584b863b2882a103c95 : SUCCESS in 1h 59m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6964c74ff6684fe08e54e71d55671773 : SUCCESS in 33m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19664fabc2134ab08c4411b866ecac23 : SUCCESS in 28m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793b690c2b2740a08c009f5500a69f96 : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbb3daf9dbb54e9eb2b5e7585f18a970 : SUCCESS in 27m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9d22a3b86b848c7afe2d8f785fcb0cb : SUCCESS in 25m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04aa841d2c0c47f19c953f50421e2d15 : SUCCESS in 26m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505d99aa13e541fd8ffb78de83cd8aa8 : SUCCESS in 12m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/88e4c6fc67e448e98a4ac290f1e17736 : SUCCESS in 46m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8de0431809654a738133a201038ad070 : SUCCESS in 1h 21m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/975a0340eb0d4edfb0b8a307ebf161a6 : SUCCESS in 57m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e83c5c3a3dda49faae550653fcaacc1d : SUCCESS in 1h 00m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5a9a2a5a1ffb4d7aba7ba28bac5134dd : FAILURE in 46m 45s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/89443c9cb00349718a4d0513e97a0a03 : SUCCESS in 1h 09m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5a02b4dd078456f99531f4fb36588a0 : FAILURE in 1h 16m 33s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/748320694f5f4275a820f1c48dfdc40e : SUCCESS in 1h 28m 04s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f64eaaccf954462897839685ccb2aa2b : SUCCESS in 1h 01m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/37a7c656c53b4797996d70d2b603d397 : SUCCESS in 1h 09m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7c2014d0c20442f8a34f86ede24277aa : SUCCESS in 35m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/addd339c2ef3462ca61d95a5a73d66a2 : SUCCESS in 1h 19m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0d418359da4e4952b58e271fd56d296e : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":19},{"id":"a6a1e6613d4d9552cfa22293f4b5c042e954f6cb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 08:59:42.000000000","message":"Patch Set 19:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/68/776568/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"9996d8c1ed3bd40da78b8cd5866e71e9a414fe33","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-19 11:20:20.000000000","message":"Patch Set 19: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"295851711038ddfa113cad8d3d8354c3fed9c281","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-19 11:20:40.000000000","message":"Patch Set 19: Review-Priority+1","accounts_in_message":[],"_revision_number":19},{"id":"0c3473b7f3e713ffef2e213927f0f30976815cba","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-20 01:40:38.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"f782b26b4d777925cf45d26a02a81f54dd599033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-20 03:31:03.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"ea56996c1569e1ae5601d3d5549805d715b908b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 04:19:18.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2781b5b789ac4bfdbf795876701ab4c8 : SUCCESS in 40m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70dcd0cf1dec4bb4bfe2c73cae7ee8c5 : SUCCESS in 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"54bc261a963c5887542a180c7fbfc474f5a621de","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-20 05:09:40.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/20 : SUCCESS in 1h 36m 08s","accounts_in_message":[],"_revision_number":20},{"id":"91fb5197353a0027458fec371dbeec1b2f9791bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 05:50:46.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/21331d0ab831453b988ab9e4751f68ce : SUCCESS in 54m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9b8599815e024b28bc14b33c06e71ed9 : SUCCESS in 1h 30m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/91d41259d8194e2fa39be36fc8922da8 : SUCCESS in 2h 06m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2776774264f542e786c065dd34bca35c : POST_FAILURE in 1h 42m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0a889ced4fbc4e989be2f3527fd40676 : SUCCESS in 1h 23m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4d503e1cc3442a38fe001fec9067840 : SUCCESS in 38m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/534b48ee51c747e887df54d9899c8909 : SUCCESS in 36m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69944a4de437478aaaf53a10e80f2c89 : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb8c561c7f384e8dafc69a59f7058cc6 : SUCCESS in 32m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3443497b7a264083b951ba3bfe9160e5 : SUCCESS in 26m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1a9285b0b9c4157a98fee022b8a8a69 : SUCCESS in 30m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f3764bc08f1411898eebf1b34b08aca : FAILURE in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da897b00568c4c9f8c3c90b1b6c2844c : TIMED_OUT in 2h 17m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/babdfd12f4724aab8b3897429355d095 : SUCCESS in 1h 13m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9b550ccfda664bc0bb0163fb8ef0aec3 : POST_FAILURE in 1h 44m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5be5387c4cdb4a54b28801943a43e36c : SUCCESS in 1h 02m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5964df76bb0d4d2398332dc3d2bb4588 : SUCCESS in 1h 08m 47s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ad7d32de9a574f59b60cf7d876732340 : SUCCESS in 1h 21m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c97d58976bd4a42af4726d68f5d6d92 : FAILURE in 57m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb31601439b94ce0ac0cda71be9957a5 : SUCCESS in 1h 24m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/764433288dee4787a0faeb6d0e18c6e5 : SUCCESS in 56m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c1d4ba65ed484cb9b17fcccb40777593 : SUCCESS in 1h 31m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7a4c378ef7364c6597875b08e1adcf7f : SUCCESS in 28m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bebab1cb278e4e01914755ecc0afa900 : SUCCESS in 1h 29m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b78343bea12c489fa873d5556906f148 : SUCCESS in 1h 26m 03s","accounts_in_message":[],"_revision_number":20},{"id":"e63928d875d1d67ee6e82d68626fc9bb7f0fb6ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-21 00:53:13.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"b27c4a4ed46f4dc502f48ab49f1ac2a3060018d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 01:33:21.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f1d415abdc24ee1b947c88f13d430f7 : SUCCESS in 33m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9cb1cc184794daaa4d2ddb2bd732d86 : SUCCESS in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"3e10cbd9e33ad53da29254abcc308ea7f3161a53","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-21 02:45:10.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/21 : SUCCESS in 1h 48m 59s","accounts_in_message":[],"_revision_number":21},{"id":"44312f5a72b494cf6ec631d9bcb5506e3ad39eba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 02:54:28.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/484b1911de0d45fbbafb47bab2190f53 : SUCCESS in 38m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fee2a6dc48274cdabfc9f9d7f5882f81 : SUCCESS in 1h 39m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a0e635b768fc471cb20deb715a70b7e1 : SUCCESS in 1h 36m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a5f9b79a31dd4de583689753bf60bdc7 : SUCCESS in 59m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dad5365f529e4733aedf44875cc1c2c8 : SUCCESS in 1h 47m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/659cebeabe0e4a24b247f38788350bd7 : SUCCESS in 35m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10f341f2828d4837905e359dc2eb3edb : SUCCESS in 31m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1840618a6ae44af9be1715767b4d215a : SUCCESS in 12m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7748d8ea3e8247008ec755efa6283015 : SUCCESS in 30m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d7930f78abe4e2eb93e50512c6de482 : SUCCESS in 22m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd4bb66c710f41fc83cdfb5061e6f0ee : SUCCESS in 28m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e3a30737b446fe968e40724a76a3ee : SUCCESS in 10m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d15e2f69be554199b96cb668bdcc7229 : SUCCESS in 43m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/88d1ef57cf1a46d88590c98ceffc81d0 : SUCCESS in 1h 10m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/82f6e0638f044b8d9e66fdb5d9911b09 : SUCCESS in 1h 16m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6bca13fb07f14d67a8a3494708e8e8c8 : SUCCESS in 52m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/24ee39abd5c74a49a5bbd886d79b0842 : SUCCESS in 1h 01m 16s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/95e5ddfb3b2646b3bd15bbaaddb61b29 : SUCCESS in 1h 09m 10s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/35024027ae3e44b488f140f20a6c3369 : SUCCESS in 1h 13m 59s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3a7c0169d1c462ba032cfef07503af6 : SUCCESS in 1h 22m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22a4d352bbea4914a2910908345bf689 : SUCCESS in 42m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4d56fb5df99e48e9b25512b89f085eba : SUCCESS in 50m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/28740b28d6384a738f679e85f5348704 : SUCCESS in 27m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/20adfe85fdf445b6b12dba96337e3cec : SUCCESS in 1h 00m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d5d4c464ac9e4a019b609d94db432d96 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":21},{"id":"6209b8233c56ff5ffacfbe67b2f5441300d7e318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-25 03:30:34.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"f5af028c4e5edb5547673d1a8743a8cd5697b446","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 04:22:11.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88b61de43d1b464282d0e73b1e6d3a1e : FAILURE in 49m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df356e0e0e8243b298d4648ebfde86ad : FAILURE in 40m 53s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"b2c9c73155bf6edfc715729f536da3526ec06304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 05:24:26.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dba901cf71a948b5adcd9448d286cacb : FAILURE in 48m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a4f5f49e9f4446cc9716ba2faab1136a : SUCCESS in 1h 25m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/26fd3b715a6a4db8a888d032b6badfe3 : SUCCESS in 1h 26m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/60d0442ee8f34bcabb29ab577ecb32f8 : SUCCESS in 1h 06m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/358a452e25924008b2ccd672433e7e8d : SUCCESS in 1h 52m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac14ff5d7c70429fb809d2464e40d789 : FAILURE in 42m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3935507966434446bc2841353d5de476 : FAILURE in 32m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5625c5709ea482cbaa21e5ffd5cf4ca : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b7f0fddac3a4916833be769307f7fec : FAILURE in 33m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4251ea51642b4a64836827cd543135eb : FAILURE in 32m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffc3919ea23c4879bccfec65149a5ae9 : FAILURE in 25m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2beade8315746c59289a151b1a47aef : SUCCESS in 8m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e170f1d3358a42bead62cc6231207964 : FAILURE in 45m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/192aaed29e4d4002983d570f8e7be1d0 : SUCCESS in 1h 17m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/900b173ab3474fd79b69e33a095a0870 : FAILURE in 1h 24m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9520e5366284392a40ec1cefba4e96c : SUCCESS in 58m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/202e985a3b674cbda691d28615ef6c38 : SUCCESS in 59m 58s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fec1d6e89c7e497c926f365785acdcbe : SUCCESS in 1h 13m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4b1b4b23832b480db9c5a698d2a83332 : FAILURE in 17m 19s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a68f218c2224ca8979a33a069c89087 : SUCCESS in 1h 33m 15s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f1ee0dc7fa743e3b15d27686475d01b : SUCCESS in 59m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c3b6e058dbae49b6823675f01a190d38 : SUCCESS in 59m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ccb03a89605d44558d924c41c9e9dfa0 : SUCCESS in 21m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7b4506875c71465fb043cf0eae1cffbc : SUCCESS in 1h 07m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8b2e00cae4ff440aa206bfd5ebef964b : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":22},{"id":"80855741fd58879c227c869413f736314952b486","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-25 05:25:22.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/22 : SUCCESS in 1h 52m 32s","accounts_in_message":[],"_revision_number":22},{"id":"46636468aeecf6cb8baa66fc2b2614052d017ab6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-26 11:12:30.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"6ca0edce0ac57c6e0360e4a48d49da1a4412790e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 11:53:15.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/40afce956fcf474e99790b9f20df3efd : FAILURE in 35m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ceea65d8d82a4f15a5eb181887585d38 : FAILURE in 33m 48s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"28792b7ccd978c42239b7466a762694ee6cc9718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-26 12:27:09.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"9af96e5d43f57026789a32ef8e6f3f18df477095","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 13:12:39.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29c07cbd39da4588ab9f1a727638c108 : SUCCESS in 35m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e9f5c300e9849779d6dc9f7ba387995 : POST_FAILURE in 40m 40s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"3e57510cc7af4b18ddd96861c39b2bdeadb6da9d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-26 14:09:02.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/24 : SUCCESS in 1h 39m 17s","accounts_in_message":[],"_revision_number":24},{"id":"df78fd0cdac15d0d5c31e24753648d63c41711d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 14:46:26.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dbf9ee6eab6343a687dfc970147eb8c0 : POST_FAILURE in 45m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8c2c2d4811c2425298bf7b32d62cfcdc : FAILURE in 2h 03m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/50805d21bfde44ea91fcae32bdae60ba : SUCCESS in 1h 33m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/723e68aae92b48b1892746c606112583 : SUCCESS in 1h 13m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5904efd049a54b1ca1e772a5ecf3baa8 : SUCCESS in 1h 36m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17f6bb80e91643408bbff66c7da09b9a : SUCCESS in 31m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4274f2ec8804e7fa388118edb56a205 : SUCCESS in 39m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5f7e9a2b22b4ea09e651d032739e3a2 : SUCCESS in 11m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e1f8cce64d040589ed052d157cec4a4 : SUCCESS in 29m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53b9c975f97e42949464bc909631eaa2 : POST_FAILURE in 36m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e187ea805c0149338dd9aad031a9b579 : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/294a10e041934b95b253b440b613cbae : SUCCESS in 7m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd6c1b30040d44d3acddaf47b41b52d6 : SUCCESS in 39m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c7f04b150cdc49379db863a72527b924 : TIMED_OUT in 2h 16m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a4c155ef6dac4d9582a46b5945999295 : FAILURE in 1h 43m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8273324a3e774e0e93db5abd3d1d2fcb : SUCCESS in 49m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1d29efc1b7354fb79c13aa8e452d7989 : SUCCESS in 1h 08m 24s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9bfbc4e28f394150a2ea37731e6e8b67 : SUCCESS in 1h 27m 14s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ce543622e821449d881404beeab41379 : FAILURE in 29m 43s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a668e074ee664ef9a7bdc30d4aece361 : FAILURE in 31m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3e8eb868363480aaf9e061ab686cf03 : SUCCESS in 1h 11m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/61da7b7211be40e79b6d876f67ba686a : SUCCESS in 1h 02m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/36c4717e56574220b8ddb5966043a0a0 : POST_FAILURE in 39m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/559b10bc0c374b01ad8c2beac709f630 : SUCCESS in 1h 19m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7454dd4983dd4e31bae518c268caa4c3 : SUCCESS in 1h 27m 04s","accounts_in_message":[],"_revision_number":24},{"id":"f21c9df2b89b676542e6402e0fd3cf534d337c3f","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-26 14:57:50.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"9ba5400d2417231246718e71243b5ec3ee79a145","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 15:39:52.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfe517c1ab554b16a89149234d419e02 : SUCCESS in 36m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/734ce40a90834363aee1f27994a7246c : SUCCESS in 36m 45s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"8836d8d3e590318043fa70991672c534b5a47645","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-26 16:38:29.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/24 : SUCCESS in 1h 39m 38s","accounts_in_message":[],"_revision_number":24},{"id":"3afdfcf8e4e5588a52193a08cc4d2fad8c63b979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 17:11:03.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/60d474a1501a48aeb8b9825997866f16 : FAILURE in 40m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d79618fcfe594c678622e600aec31cb3 : SUCCESS in 2h 12m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0ab5c49b26024640aed55a976e2540b6 : SUCCESS in 1h 39m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5d16bc9da35c4a91809aff9a667aadf9 : SUCCESS in 1h 17m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0cad518a6b754f669a514f58113146e5 : SUCCESS in 1h 52m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/403561c90f864128a89719c594ad23e5 : SUCCESS in 36m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75a0b3c17b074bfc84e7152770cd0710 : SUCCESS in 29m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d31c0c007442b18da63b2b462152c1 : SUCCESS in 12m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b3bbeea763a4bcd884b9b1552c1201b : SUCCESS in 28m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c852004548546dbbff771b57984f4d7 : SUCCESS in 23m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c803cd48da240ddbf46ed9faff29eb2 : SUCCESS in 23m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7794f0788854d1fa20cfd7668ea9ca2 : SUCCESS in 10m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b0b545c0d6a48ad9ada4b3539ca3860 : FAILURE in 42m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eba60477992842898fd52e7882d61c4e : SUCCESS in 1h 48m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bc77c4cfa49c4771a45462d11c436f80 : FAILURE in 1h 29m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ddfcdb8d561b4f72b2eee7a25ac1293d : SUCCESS in 1h 04m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/632660644e684e0f99710dd8031986e9 : SUCCESS in 55m 37s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/157bbc9ab61f4d92bea6b4ae5710e99e : SUCCESS in 1h 30m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/31530aba3851445c83329fe1082ce587 : FAILURE in 24m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/36e93e12646a427f960dcefc07103410 : SUCCESS in 1h 38m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/026c0f8bea884074901f59b4ec539133 : SUCCESS in 1h 04m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c788f2d1ad074ed790b9d12525c9e23d : SUCCESS in 1h 09m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5fd3ae5d6ad24b54bb8eb204c76d59a0 : SUCCESS in 30m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bdd9b9e5ec534ba296c8cee6f6c7ddfb : SUCCESS in 1h 24m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3a925bb39a71459e9ec381d95d6d48c5 : SUCCESS in 1h 21m 49s","accounts_in_message":[],"_revision_number":24},{"id":"ba90c6e4d78ff680be0cb8676269bd0629395599","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-02 05:36:00.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/68/776568/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"98479cc6bb43497b076a6f018d8599f8d5cde1ce","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-05 19:56:35.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/68/776568/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"29c88ebb4eb43429ae62a36e21ab2b3ee9dfcdb7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-06 00:19:04.000000000","message":"Patch Set 24: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/68/776568/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"83e44da3caf113d6b24b521e26d755ccda04f684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-07 03:40:41.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"133d3f4df2aa1b3fc0c2e7e668f1e57c00de8f3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 04:30:07.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/875df1d13e764ba780720eba007dd84b : SUCCESS in 46m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/906a8f1c147840128f6a16fecd38030a : SUCCESS in 46m 57s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"6705449f6c3b7affce04e736c06e2f8e5a540b9c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-07 05:19:14.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/25 : SUCCESS in 1h 35m 56s","accounts_in_message":[],"_revision_number":25},{"id":"3cc9aa6e250e7567bc9c251242d7132e873b789d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 05:53:22.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/13078527582a42829c6b2fbc81f5f2d3 : SUCCESS in 48m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/256aaf863bbe4f5c96b76b2085ca3d49 : SUCCESS in 2h 01m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3107ebcd33144eafa1541d0effce9cb3 : SUCCESS in 2h 08m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1fb005b3b9b0409ca19f574fbad6e031 : SUCCESS in 1h 10m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a5610fc536ca4a54827bb928737f0e3e : SUCCESS in 1h 32m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eff7da9bec8b4c1ca7e27a29e048cb4c : SUCCESS in 44m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23d23d81bdff4ccc858126867a55133d : SUCCESS in 39m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e81bfbf759cf4098bc07c3a8533ae469 : SUCCESS in 10m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d950237fc04c4540b19b41b354e3934a : SUCCESS in 32m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/627d1f39df0540e690ac64d18c42ea66 : SUCCESS in 34m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6d20fab9a944e8fa3e42e46ea0bbfa1 : SUCCESS in 27m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/370a4f53947b4423bfb727c011cbab31 : SUCCESS in 9m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aeb29275ed4448a582a05207aae777e5 : SUCCESS in 50m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6aca477312f54f43a4fac3196cbcddce : FAILURE in 1h 10m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0aa4d80f626749ad971cfeff97d22cc6 : SUCCESS in 1h 25m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/627c9a7ae32c4ff3b5ea6185743dc062 : SUCCESS in 1h 05m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/33e364ff38bb418da152ba9363ae669a : SUCCESS in 47m 31s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ebf13d6d008d4cc58e53da4b45cdff95 : SUCCESS in 1h 22m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/71a282058aed46e39b3f3c15aeda9347 : FAILURE in 29m 13s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c15be75c60e54ac4a948131a62536047 : SUCCESS in 1h 31m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9440bec95164e06ad5da1cae88d2720 : SUCCESS in 1h 19m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0778d3a22cf3483ba62e9dd9eabafefb : SUCCESS in 1h 10m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2c576c93fffa43a4afe7902d233862cf : SUCCESS in 35m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b6ba12fcae1444109baa9d021d658299 : SUCCESS in 1h 14m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3ceb0014f6764cd58e3064caf92256fa : SUCCESS in 1h 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"9664721bec09be68889c403d2c7d668d5a64de48","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 06:35:58.000000000","message":"Patch Set 25:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"6ab38316d3b9b2f8b7b9bfb14dcba1d3ca95468a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-10 01:27:34.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"12249b6860c7f80e3703f898b776d813cd37a74e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 02:12:59.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c10c1c64e542469db38708dedcdc332a : SUCCESS in 37m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9244e56b889547e9a3e4f5d4b4c612d6 : SUCCESS in 43m 17s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"2033a6140a620382606af859863eadfab46e981b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 03:21:27.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/13bdd701ba2742829df37fb34db7879a : SUCCESS in 40m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2536fe757e5c41ccbe78cb96fafcc4d1 : SUCCESS in 1h 50m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b8700eba890d44d880a5c326d3745a21 : SUCCESS in 1h 40m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/88f05398884447949767301982b347bd : SUCCESS in 1h 40m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/438794bb03bf4a699b18c63e428c4fa5 : SUCCESS in 1h 35m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87780ee7f96d4e709957718eadb4ef3c : SUCCESS in 34m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df17a0db6b2d4b9d8f81133417959523 : SUCCESS in 25m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21eb42a7bfb943ad84bf64944690f898 : SUCCESS in 16m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0071dbb771f54f9a8614c928592d1258 : SUCCESS in 31m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e5b9cc07a3b4458ab014b5356c5fc2a : SUCCESS in 28m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df98c1bbc4df4f6996030e47d21069fe : SUCCESS in 31m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65d24aa01784df08c7ab1aeb1725d92 : SUCCESS in 8m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/90382819eeb24db3962c7763ba280e78 : SUCCESS in 1h 01m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e0f515d603342a9a8e9754215879f3f : SUCCESS in 1h 23m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0fbb7797c2cb4095bd43471ef8fb817c : SUCCESS in 1h 12m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4b24f4cab414b4c83784116a4884233 : SUCCESS in 58m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/17ae351d9fa649fba2f9830b13db1b6e : SUCCESS in 1h 25m 32s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0ca23e58923041038bdfad695fb5d57b : SUCCESS in 1h 15m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b32f4580d9934ba5968fa606a1aa531e : FAILURE in 59m 49s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0276e6c945604aebbb7326d11085424d : SUCCESS in 1h 26m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59493add0a144b6daa2b1184c42e09b2 : SUCCESS in 1h 00m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e72a8054f15f44b4a7ba2b2bfbe4bece : SUCCESS in 52m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/148202b6cbbc4202bd77459e04c4ff28 : SUCCESS in 30m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d4b6fae82fbc4acd92c6f779ac68b3aa : SUCCESS in 1h 26m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d1bfcc91b51643c7a67819313359ec8c : SUCCESS in 1h 07m 31s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"bf4d588b7add5a13f248bf98c6aef9296447dff6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-10 03:56:29.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/26 : SUCCESS in 2h 25m 06s","accounts_in_message":[],"_revision_number":26},{"id":"e6237cc4ec8b211ba2771b2001637b742754c7d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-10 08:04:15.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"7a71c426407a4071a7db08649c203b3088a43479","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 08:59:35.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ff8317f7d874a0697e3ad2c126f21f8 : SUCCESS in 53m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97b54b7eb4e04939946922b3d5c314e5 : SUCCESS in 52m 30s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"610a6a0ffaf276861e829e415b849e5be072a5e4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-10 09:53:16.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/27 : SUCCESS in 1h 46m 06s","accounts_in_message":[],"_revision_number":27},{"id":"3b24adc8af7c9bab46a8747594ed9da569befc30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 10:09:28.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/135073f3ad3d42efaf07e24f52b3fe34 : SUCCESS in 43m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6abea7d166e24e3d80fecf5d4a908f5f : SUCCESS in 1h 40m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/540b49e16f034f0c8072d1830ad345a8 : SUCCESS in 1h 47m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/464bd4dcb898420f95fbe2568cd1509c : SUCCESS in 1h 22m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8fd99af573a14ecfac687c3a97d1d1ea : SUCCESS in 2h 02m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be157dd051eb4f4a9e3ca8172172e99a : SUCCESS in 34m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99f3515a54ec4dd883387c44d138c5c6 : SUCCESS in 25m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2342aabadf1b412e985d7a6ca3b775c5 : SUCCESS in 12m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43e6d3706f304c54b5d9a7bf1103d639 : SUCCESS in 25m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd0c9cddc8414c6ba2b2a5083ac36883 : SUCCESS in 24m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8d3cf3461a64687b4b1f6a298dda98a : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9468f6a7608e414187a782192ef99743 : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6dc0af4d4b3046d390fe062286d18886 : SUCCESS in 49m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/71934ef7576c48529d5f171891ce5963 : SUCCESS in 1h 29m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6853e0e188614d539c73c624f2801089 : SUCCESS in 1h 25m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c733a679b0ff47d2b3db29763dd776d2 : SUCCESS in 58m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/81d9101bf3df4412b60918e6ed06110b : SUCCESS in 1h 01m 25s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f1c5aecf4fdc44419d87c38e75347c9a : SUCCESS in 1h 10m 32s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b2ace981a2ce49b1abd8a87cb8b16a47 : SUCCESS in 1h 16m 01s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4c3a35441c394063bdcfb67031fda8e9 : SUCCESS in 1h 33m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d5b240256954eb49120c9ec53e313e1 : SUCCESS in 58m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0945dbb4d24947558fbf77cf177f8dfa : SUCCESS in 57m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d6305ad384cc4c80b3b1344ad2e8fe17 : SUCCESS in 28m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1f271be46617474887adff4e7f7bfea1 : SUCCESS in 1h 18m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7ef67663abde4730ba5cd93eeaa33f7b : FAILURE in 1h 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"0eb3f22ecfe49cc3a0a607c4b2db3d6ea20c23d8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-18 11:20:08.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"c829990b0de69775fcbe1786d426fce43eda389a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-21 06:51:36.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"378c024929480a49cbe8b08c954ac7389925b325","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 07:38:03.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df9a2f5d505b47beb8d19c25a8499e13 : SUCCESS in 43m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc1a7c1c9e4645eeac8e97d654c228df : SUCCESS in 29m 46s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"555db643b43498c8bd7fb3b9c2b38e0d339fdd12","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-21 08:27:46.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/28 : SUCCESS in 1h 34m 17s","accounts_in_message":[],"_revision_number":28},{"id":"f30c9dfb28f14d4bd692bfc0ac793974f39ab3ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 09:03:41.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/64bb1fd5117a43588c8f0e7df1add1a7 : SUCCESS in 46m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6de5bb8ede54402eaf774fdac72a8a40 : SUCCESS in 2h 05m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9e16a6c5435a437095651291baf6c2fc : SUCCESS in 1h 53m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ffba0958a5c4f9c847d92fc4685ee80 : SUCCESS in 1h 33m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/89014e21ebdd4866a42435f2e683ed93 : SUCCESS in 1h 40m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7a44e835aef481a88c6515edee2e68c : SUCCESS in 34m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e58c63fcd4a24d0790c91c87b5abf9d6 : SUCCESS in 41m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06615817bda4b3fa385b39eeb4795ca : FAILURE in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e470439d30d4fa0b7b2a83a2d798f0d : SUCCESS in 28m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a45859e5334cb9b71613a6095b781b : SUCCESS in 34m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fad4712cce26490ca6c59d4846060f1a : SUCCESS in 30m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d96841d15d4a70b045aaa74f504bcb : SUCCESS in 9m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9a9b36037894ae7915a6377036f4c67 : FAILURE in 47m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/14fd21e41b7d4fb3bb9311a54e4c10ec : SUCCESS in 1h 28m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f47f8d16b81645d8af79e235e59b7117 : SUCCESS in 1h 09m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/87288c0074634d26833132b1aa5cdc87 : SUCCESS in 59m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/23e57fad5d6f4c9fadc48f2046ca7602 : SUCCESS in 1h 06m 23s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2b501f19b9ed4fea9f7c838dd9d8b014 : SUCCESS in 1h 07m 14s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/41b2acf288bc43e9af3bb98e6981ce43 : FAILURE in 1h 13m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4b31e834305f4deda9ad666bb5368d10 : SUCCESS in 1h 33m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62bf478719c444b9a8c6e9cca9302088 : SUCCESS in 58m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d9578fef10e34c209a8c79713d7542f7 : SUCCESS in 55m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ac99785c5b5b4dcc8c9b3726265728fe : SUCCESS in 31m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74e0699df8474deca943dcf9748ec405 : SUCCESS in 1h 09m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/abed63cf7ced46dfa6002ed85f8e87aa : SUCCESS in 1h 22m 46s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py","accounts_in_message":[],"_revision_number":28},{"id":"468f0ed8fbca68e6526cad58cbd6612136d59ed9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-21 10:30:09.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"cfa952c9b7a83fdc07a8b6ff2fa27f392acdaeff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 11:27:41.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc936d1d5b754cbb93ed6253d53f1973 : SUCCESS in 50m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/031330ff1ade437f957733fc607ccc1e : SUCCESS in 39m 21s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"92f4455475dd52bf4aa664eff062f8e7dd577629","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-21 12:24:00.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/29 : SUCCESS in 1h 51m 48s","accounts_in_message":[],"_revision_number":29},{"id":"bde51fe72779484413fbcd8581e07b55bbeca471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 13:41:03.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4d279f108cd04463a05f77f465b99cb2 : SUCCESS in 1h 03m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6d30b8c116e74bfba149819f74f03453 : SUCCESS in 1h 50m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a9e8303e74f243679c2d7dd75edd5060 : SUCCESS in 1h 39m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/60ac01f6ba3947019626f71cdcd997c7 : SUCCESS in 1h 26m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7734a158019949719800837de93767c6 : SUCCESS in 2h 22m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d95d2c5bc184ce3bf9b21e06098bc03 : SUCCESS in 46m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ba71a0be8db474787fb713f2510a743 : SUCCESS in 40m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e134cfe49f674e38823a05e54d3acc95 : SUCCESS in 12m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c70f0af5033f4f3aa05654ca1ba0e9c7 : SUCCESS in 36m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f51b7c7b0344624aa7956c3c7a9e86b : SUCCESS in 22m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42dd96707a2b4ec4b69899e7b7a2cd39 : SUCCESS in 24m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b759e74b01bf404f840dcbff5237c72e : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/29a755eb30ce478d845aa24927419fa5 : FAILURE in 42m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/48d16e856d0040f691c18089ce7111f7 : SUCCESS in 1h 17m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/84b99f63cb304ec4979e1177aec40414 : SUCCESS in 1h 03m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bfd0468c03eb4bd7ac272e03a1f2b61f : SUCCESS in 53m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/476792d7ce2a4306a36f8c938c60efa8 : SUCCESS in 1h 03m 41s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6759138c56034ac78720a07063dfcbc2 : SUCCESS in 1h 13m 10s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/681a28ee379044d289a59a80e4f4895b : FAILURE in 1h 14m 33s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cacd71ce7f6c4f88842f4619a62d706a : SUCCESS in 1h 53m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a560ab3137f47e4b2ee778e9d0adb8f : SUCCESS in 49m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a0f446c069874461821102d45011ddf9 : SUCCESS in 1h 00m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fe5c82cc75214dc49c5e71d0b43f62b3 : SUCCESS in 26m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c414ac3549b4c879033b5fce789e91c : SUCCESS in 1h 12m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a36fb9752a2d4e289a1a9a1e55f5bb0e : SUCCESS in 1h 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"e8b3490e7b3c67e05db74776f7cf39dc8a2187bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-22 01:40:35.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.","accounts_in_message":[],"_revision_number":30},{"id":"f18b7412a53fea8ab06a3c36f5af2ee79f62046d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 02:28:38.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/acb458b0a07c4d859465cc77254d9712 : SUCCESS in 41m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f068d321adca43cf9e77769cd949d12d : SUCCESS in 41m 07s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"7969c372928e9c5cd244e14e4d421ad9aa4cc87b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 03:16:23.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/30 : SUCCESS in 1h 33m 52s","accounts_in_message":[],"_revision_number":30},{"id":"a0a22ff8583ac282ddf5d538833d6f4e85cac36b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 03:34:49.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/db656a924c8a4851b6b7e85af5856f19 : SUCCESS in 40m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e43e47b231664288914fab987a86af39 : SUCCESS in 1h 36m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/691d4f2f85e044aab9cd5cc0f09aee25 : SUCCESS in 1h 36m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2a208cccb0e141d0aed0ba8316df87d2 : SUCCESS in 1h 43m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/33ee0dcda5c84baba7aa629983d95377 : SUCCESS in 1h 52m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a9a45ab666841fcad995275062d1e76 : SUCCESS in 27m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba6c46b7d04f4420b910e3e03acc837d : SUCCESS in 37m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfcb4573f93e4d72a7f4eb494dd909da : SUCCESS in 12m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3c76ab9760e4849bd40c4319cc66f3a : SUCCESS in 21m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de2a08b72bfc475cb6c7197653ee8550 : SUCCESS in 30m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0be0b26a75e7400d816ebf8f66b6b891 : SUCCESS in 24m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a665459c85e34f53821e00e3e197cbc5 : SUCCESS in 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e76260ed71b499e928f003c7a4b6725 : SUCCESS in 43m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/38f5baafa38e4a25819fb68715f93380 : SUCCESS in 1h 26m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0f53657bf1d64e4ca3b3fd1439dd2115 : SUCCESS in 1h 06m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8490812bccb2436e8dcb5424bdaf367f : SUCCESS in 59m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7b6407725c694ebfa35eb5a0743c932f : SUCCESS in 1h 01m 56s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3d35f2d29093470db294e09d61c2b75d : SUCCESS in 1h 06m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a79f1ef5d808454c8f5f00a2a5efc135 : SUCCESS in 1h 08m 32s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35d96afb7ad241b2a475ae75d8a0f29a : SUCCESS in 1h 20m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d7343a1cda7452c847ac840c16b5ebd : SUCCESS in 49m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dc3543cae806420a874f0e5581b7bdd4 : SUCCESS in 51m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ca209716cf1e48caac91316f7e9ad932 : SUCCESS in 31m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e1cdcb7861574f699571dfb601673377 : SUCCESS in 1h 15m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/83636a0f31bf4578b6d4f1c5f9a47dae : SUCCESS in 1h 08m 08s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"93669d8b5fc2db771bae2721b79288fd870b60f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-24 00:43:45.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"96bd372537ff83c7f92073a660f351d98fb41b3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 01:24:59.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7386da6da1b54c56b53e5a820891c3f9 : SUCCESS in 39m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fcbffdfa9d14bac884c1cffc7c4addc : SUCCESS in 39m 15s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"e28b47ed9725d8f1c5b465d9082b256443123335","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-24 02:33:26.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/31 : SUCCESS in 1h 47m 59s","accounts_in_message":[],"_revision_number":31},{"id":"750bbff37285c9f669cdd72c0b56e8f671616c16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 03:19:24.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9057343411c745c183fbf6d6c7688df2 : SUCCESS in 34m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f3ed5f6a41674c51b66e1c60d52b9be9 : SUCCESS in 2h 27m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/62d9fe3d712149feb91e7871e5259207 : SUCCESS in 1h 31m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5248bcd2b8994f44ab0aa4b767b06027 : SUCCESS in 1h 26m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9660d8483d3c43eea4ba53f7d53146bb : SUCCESS in 2h 03m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca660ac4f3204713b28155427b05446f : SUCCESS in 30m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f389526415274ea6b75d5f77840b2490 : SUCCESS in 35m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7605bc8f1dae4599867e1fce1146fa84 : SUCCESS in 11m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e5e1c88f371463db3f73d9cca5437d2 : SUCCESS in 25m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abfecaafbbaa4c73b5747370fc7ac4cd : SUCCESS in 23m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14eeb9d5c27f405f8ccc16b0014b7a0c : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9434ce4f31840759f0cb85737048ac9 : SUCCESS in 9m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/59f09ee3a9ba4f18899119da128fd4f9 : SUCCESS in 45m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fcaa316beebe4d27ad8fda8b015cbcfa : SUCCESS in 1h 17m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/876aaef9d60d44bd8b26a84d9d8e0927 : SUCCESS in 1h 01m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa6e038d561d44f3b69858d638980a79 : SUCCESS in 1h 06m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/29a447bedbf8478795c27f42a1736179 : SUCCESS in 1h 05m 24s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c9471a6e85a548679c16d6dd5cbe6c17 : SUCCESS in 1h 03m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3f093f8e6b4d45a885ed93e8381b47a3 : FAILURE in 1h 10m 08s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/49f798d9f5bc40c3a5a8f1e6ceb98772 : SUCCESS in 1h 34m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e0beb57693840a69251854799965055 : SUCCESS in 50m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d1230f5cfe144d76ae8d7ed934f0407c : SUCCESS in 52m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e3bbd8692c05402da236079623ce8ee5 : SUCCESS in 27m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/854cb0afdd564e2cb70865d22e4154cd : SUCCESS in 1h 06m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8dea56b2439a49a4be21c70e0d2e8845 : SUCCESS in 1h 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"56f6f3a909d15f004862c3a8609bd1c7bb0895f4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-24 14:33:32.000000000","message":"Patch Set 31: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":31},{"id":"08e44448161d12e0912415afe922214ab6271e1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-28 09:30:40.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"a3070fd4f57e8e3a3d220b2dab5027c7cbb5f9ef","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-28 10:12:21.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"20d8e281025b29a6fd26bc770ff98de206094d84","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-28 10:12:35.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"49bb3210e6f3349467572881d59c37e1266099c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 10:24:37.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db5c7f1282bd4cfe8cd1bb3e7c366587 : SUCCESS in 49m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35156bc6f547444289cd541ca71c1c1b : SUCCESS in 51m 33s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"e246faa06407fc5233ce82e962c2efedb413bd51","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-28 11:19:53.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776568/32 : SUCCESS in 1h 48m 12s","accounts_in_message":[],"_revision_number":32},{"id":"1c16e574550c6c736b7011532450f0109672e5f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:57:44.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1eafd9c7a30b4beeb877a64ab4cd03df : SUCCESS in 44m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4196471a8ea04e5ba81193ac1f5be970 : SUCCESS in 2h 22m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/79726b57278f49898c2ff2f999461804 : SUCCESS in 2h 00m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/838f1919a0474f8e9661f4e0866a72f3 : SUCCESS in 1h 30m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e5fa4c4c16ff474796c13a7084111203 : SUCCESS in 2h 16m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c1c9d2ac3e94a06a7da98a16a36a2d9 : SUCCESS in 36m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23127f3f03b743709ce2780c21562cbd : SUCCESS in 35m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc35b42c9c044395846c474cda69ce87 : SUCCESS in 40m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a25057ca524bf4bc564236892775dc : SUCCESS in 16m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3a6d34da05447048af40d06e7db75af : SUCCESS in 27m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6788e007b43d4e5abd65a9119f31d35d : SUCCESS in 26m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e58c0e83c2634592aafa397179b6dfa6 : SUCCESS in 25m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/404dd894b1294ea7bc52896fcd78ec4d : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2abac650f1be48ac871ab1c57af2e399 : SUCCESS in 1h 03m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f67d8be72a3640bcb846d37f1b5718fd : SUCCESS in 1h 21m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9753690834534399a82a54753e3fd482 : SUCCESS in 1h 11m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/600a8bdd62294b1a9f02af110ee0480a : SUCCESS in 1h 23m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/55fcf71123f94087a40e97d322495272 : SUCCESS in 1h 09m 00s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5a55f8ffc21c46ac9db10682c072175c : SUCCESS in 1h 29m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/86c8f1ce9d3942bab856325655d63a7f : SUCCESS in 1h 34m 04s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e723105e3fdc4cf09f5a96d8e95ef7c7 : SUCCESS in 1h 48m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/333f7cc298d9493a8c35cded0b954cbb : SUCCESS in 1h 04m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7c30b059187d420b8ae16dd91b265106 : SUCCESS in 1h 17m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d206bcd74e6d454183321645f0975bc1 : SUCCESS in 35m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b87221e86d44698a6a3300d1d79f89a : SUCCESS in 1h 30m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6d5e79ab387b4007a61cc2364e11d358 : SUCCESS in 1h 15m 15s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"f1f0782813d5e32614c613dff0b6048e43df4062","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-28 19:57:32.000000000","message":"Patch Set 32: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"91d7b230e478048aad2a5e18abfd89f06f87b45e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-29 01:07:09.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"10ef724365e5657367efd4ccfc192826980977a8","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-29 06:22:09.000000000","message":"Patch Set 32: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":32},{"id":"e11f7acae94a7b591f6a02cc8a756064a01b7f5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 06:22:39.000000000","message":"Patch Set 32: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":32},{"id":"93bdb0d9bf47d2216a7bc7cfc93d953f90c97135","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 08:36:23.000000000","message":"Patch Set 32: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e2bc868490a94faeb7dc6474c26b2120 : SUCCESS in 47m 55s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/95390ecef9404d8a9f6bd0b5c7205b8f : SUCCESS in 1h 55m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/283649ae34d44327b67fceee71b22341 : SUCCESS in 2h 07m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/22c3603b6be64f15b93c912c01fdef16 : SUCCESS in 1h 39m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d4da9f74b1e04f7899df13d3bd14914d : SUCCESS in 1h 52m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85a0394538c64e8e853607704893c4a2 : SUCCESS in 41m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af975dda92d04045a070b11076f2b04f : SUCCESS in 16m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6ba5ac843ed489a9f8d103cede6040f : SUCCESS in 27m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/207c84a816e04e3486c18908d6b242b3 : SUCCESS in 27m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2042f5d90345298d2bbcba94461a7d : SUCCESS in 13m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0441f1b38a2842e5b7501f9c2629fae1 : SUCCESS in 51m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9525f441a55d47e6ad83277a68210dec : SUCCESS in 1h 27m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/998db2b48aa64543bb5ca5e44ba38c13 : SUCCESS in 1h 18m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/52c04e5f72e1487fad8151f1cbead0f1 : SUCCESS in 1h 02m 17s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/c577d0dc9e284cefa8db3ab631567431 : SUCCESS in 1h 24m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c0d12bbd34c42329962c97e21dcc0fb : SUCCESS in 56m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a5be071e737843aeb6d362d69aee1e91 : SUCCESS in 55m 51s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6698e905768041bebcd8e7cf884f63b2 : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":32},{"id":"ac87c4f2a27162aacc63119c525131ff261585d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 08:36:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":32},{"id":"3fab16731c7e7a72f86767bb2a95c0fb5f42ccf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 08:52:41.000000000","message":"Patch Set 32:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b3370daa32a4430a37c2cf26f8fd65a : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":32}],"current_revision_number":32,"current_revision":"4b22eea4be54015c6b517257013597edbe725806","revisions":{"8e6be1606a4d108f7e3d9e0fb69704e55430d94b":{"kind":"REWORK","_number":1,"created":"2021-02-19 01:54:59.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/1"}}},"commit":{"parents":[{"commit":"3160584d7402615bf2d8a0aabc70d94a67df1a1f","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3160584d7402615bf2d8a0aabc70d94a67df1a1f"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-19 01:53:59.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e6be1606a4d108f7e3d9e0fb69704e55430d94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e6be1606a4d108f7e3d9e0fb69704e55430d94b"}]},"branch":"refs/heads/master"},"0c7e933a931f8089b1ea2ccfde2d09a4250cb687":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-19 04:55:32.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/2"}}},"commit":{"parents":[{"commit":"9546269a71c76f7480906b419d318ac15828a77d","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9546269a71c76f7480906b419d318ac15828a77d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-19 04:53:34.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c7e933a931f8089b1ea2ccfde2d09a4250cb687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c7e933a931f8089b1ea2ccfde2d09a4250cb687"}]},"branch":"refs/heads/master"},"b2e8dc5318ec23e2b4695e4e3b509d72de2c9a4e":{"kind":"REWORK","_number":3,"created":"2021-02-23 01:43:56.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/3"}}},"commit":{"parents":[{"commit":"06afa93bfc3ded85f04e7ec04212ab8cef7dfba8","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06afa93bfc3ded85f04e7ec04212ab8cef7dfba8"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-23 01:41:57.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2e8dc5318ec23e2b4695e4e3b509d72de2c9a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2e8dc5318ec23e2b4695e4e3b509d72de2c9a4e"}]},"branch":"refs/heads/master"},"c2462dfd09a0e358aacd56e9fa383a1b1d1134ad":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-23 12:08:17.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/4"}}},"commit":{"parents":[{"commit":"ccf752eb73ab2a3d8208608c530e8b9b396dec1a","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccf752eb73ab2a3d8208608c530e8b9b396dec1a"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-23 12:04:50.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2462dfd09a0e358aacd56e9fa383a1b1d1134ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2462dfd09a0e358aacd56e9fa383a1b1d1134ad"}]},"branch":"refs/heads/master"},"e000c954951dfe089f83ef0d89682489b5385c22":{"kind":"REWORK","_number":5,"created":"2021-02-25 03:11:49.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/5"}}},"commit":{"parents":[{"commit":"15f8ca2b2dc002d99efb545a454fd317744d6879","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15f8ca2b2dc002d99efb545a454fd317744d6879"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-02-25 02:00:51.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e000c954951dfe089f83ef0d89682489b5385c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e000c954951dfe089f83ef0d89682489b5385c22"}]},"branch":"refs/heads/master"},"fe6fa208189114eff0643127a9f7835a8ce636ae":{"kind":"REWORK","_number":6,"created":"2021-03-02 06:31:20.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/6"}}},"commit":{"parents":[{"commit":"e097303a8babb766ef3d3382e58a1357815ad1e0","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e097303a8babb766ef3d3382e58a1357815ad1e0"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-02 06:07:36.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe6fa208189114eff0643127a9f7835a8ce636ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe6fa208189114eff0643127a9f7835a8ce636ae"}]},"branch":"refs/heads/master"},"b856f4a6b0b1500bce068232e2432215fdc79a24":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-03-05 10:08:25.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/7"}}},"commit":{"parents":[{"commit":"1fd18767033a9e00c34fe1cc4c7b32512539e463","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fd18767033a9e00c34fe1cc4c7b32512539e463"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-05 10:05:24.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b856f4a6b0b1500bce068232e2432215fdc79a24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b856f4a6b0b1500bce068232e2432215fdc79a24"}]},"branch":"refs/heads/master"},"a1acdbea7a7a51b0f391eea1ff0b7bf37e7e4220":{"kind":"REWORK","_number":8,"created":"2021-03-09 02:02:53.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/8"}}},"commit":{"parents":[{"commit":"84ab2de031dc23a04935c1a15eceeb27d650734a","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84ab2de031dc23a04935c1a15eceeb27d650734a"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-09 01:57:32.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1acdbea7a7a51b0f391eea1ff0b7bf37e7e4220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1acdbea7a7a51b0f391eea1ff0b7bf37e7e4220"}]},"branch":"refs/heads/master"},"16e6f3d64a68462e43bf8830f4de28a10d056e3a":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-03-09 08:36:01.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/9"}}},"commit":{"parents":[{"commit":"24ca373023fb9291544775e95b5ea4d495495b14","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24ca373023fb9291544775e95b5ea4d495495b14"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-09 08:34:46.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16e6f3d64a68462e43bf8830f4de28a10d056e3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16e6f3d64a68462e43bf8830f4de28a10d056e3a"}]},"branch":"refs/heads/master"},"f13db0ac9b5100cdb24f698fcc1a2b3d43e0c84c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-03-09 09:06:15.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/10"}}},"commit":{"parents":[{"commit":"22650b0f2139bc4772d3fd45c91fa7afcd53e2d6","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22650b0f2139bc4772d3fd45c91fa7afcd53e2d6"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-09 09:05:19.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f13db0ac9b5100cdb24f698fcc1a2b3d43e0c84c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f13db0ac9b5100cdb24f698fcc1a2b3d43e0c84c"}]},"branch":"refs/heads/master"},"d483e2c5445faa70e0d78ebe69da68003dd98be7":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-03-10 03:03:31.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/11"}}},"commit":{"parents":[{"commit":"257218227f57e1fb9efb6c4f2eaba295dd770b4e","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/257218227f57e1fb9efb6c4f2eaba295dd770b4e"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-10 02:45:59.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d483e2c5445faa70e0d78ebe69da68003dd98be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d483e2c5445faa70e0d78ebe69da68003dd98be7"}]},"branch":"refs/heads/master"},"d36285705dac70b8de7bd30868b7e46bfa6a6c7f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-03-11 09:32:43.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/12"}}},"commit":{"parents":[{"commit":"851103e72edf52faf5c241ea758648173d297152","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/851103e72edf52faf5c241ea758648173d297152"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-11 09:29:30.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d36285705dac70b8de7bd30868b7e46bfa6a6c7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d36285705dac70b8de7bd30868b7e46bfa6a6c7f"}]},"branch":"refs/heads/master"},"0a81bf842e4e80d368536311319ac2f5e50b11d8":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-03-15 02:38:58.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/13"}}},"commit":{"parents":[{"commit":"cec58ae294166e31943cd32500ba12842c395d9a","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cec58ae294166e31943cd32500ba12842c395d9a"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-15 02:37:01.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a81bf842e4e80d368536311319ac2f5e50b11d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a81bf842e4e80d368536311319ac2f5e50b11d8"}]},"branch":"refs/heads/master"},"95cd3b489823cd18fa1b151c7950519de6bbee3d":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-03-15 05:34:33.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/14"}}},"commit":{"parents":[{"commit":"3a4c4e382e9322c3d7288f78af889416745bdf3a","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a4c4e382e9322c3d7288f78af889416745bdf3a"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-15 05:32:36.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95cd3b489823cd18fa1b151c7950519de6bbee3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95cd3b489823cd18fa1b151c7950519de6bbee3d"}]},"branch":"refs/heads/master"},"2eba27c356f6a16bb75dd060d03924929e540e40":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-03-15 08:21:41.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/15"}}},"commit":{"parents":[{"commit":"d8c520feab02fa4ac79ef08f3f3de700f0b47b68","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8c520feab02fa4ac79ef08f3f3de700f0b47b68"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-15 08:19:25.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2eba27c356f6a16bb75dd060d03924929e540e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2eba27c356f6a16bb75dd060d03924929e540e40"}]},"branch":"refs/heads/master"},"1cf6334dcee9a1042fe6a41960b49d3d21704fd0":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-03-15 09:37:41.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/16"}}},"commit":{"parents":[{"commit":"686b54cfcf489f2daf205690b48b27aa60ac04e1","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/686b54cfcf489f2daf205690b48b27aa60ac04e1"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-15 09:32:19.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1cf6334dcee9a1042fe6a41960b49d3d21704fd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1cf6334dcee9a1042fe6a41960b49d3d21704fd0"}]},"branch":"refs/heads/master"},"bcbf5a007e09855f9df851ff2f4a290b4026a253":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2021-03-30 09:26:52.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/17"}}},"commit":{"parents":[{"commit":"507d55c2615367e83c6d22c35d0b0d37c5029484","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/507d55c2615367e83c6d22c35d0b0d37c5029484"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-30 09:25:59.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcbf5a007e09855f9df851ff2f4a290b4026a253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcbf5a007e09855f9df851ff2f4a290b4026a253"}]},"branch":"refs/heads/master"},"55e4ba4b38a7e3296b0717c6c59e33ff822e3a4b":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2021-03-31 02:32:38.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/18"}}},"commit":{"parents":[{"commit":"7d64918cbf7d34f152b4e0b1f85d0cca3135894d","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d64918cbf7d34f152b4e0b1f85d0cca3135894d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-03-31 02:31:41.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55e4ba4b38a7e3296b0717c6c59e33ff822e3a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55e4ba4b38a7e3296b0717c6c59e33ff822e3a4b"}]},"branch":"refs/heads/master"},"5cebda8d95f822ae00c09e3fc389fedc77315379":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2021-05-14 02:06:17.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/19"}}},"commit":{"parents":[{"commit":"270e3ac410a4dc0abee311d27bda83c2c8589e62","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/270e3ac410a4dc0abee311d27bda83c2c8589e62"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-14 02:01:16.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cebda8d95f822ae00c09e3fc389fedc77315379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cebda8d95f822ae00c09e3fc389fedc77315379"}]},"branch":"refs/heads/master"},"2bd87672afbad6b4893ac84cf617627fa3fb8ce0":{"kind":"REWORK","_number":20,"created":"2021-05-20 03:31:03.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/20"}}},"commit":{"parents":[{"commit":"b7ec07c3d600e7603919e3654ccb45412b1bb6c9","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7ec07c3d600e7603919e3654ccb45412b1bb6c9"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-20 03:22:27.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bd87672afbad6b4893ac84cf617627fa3fb8ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bd87672afbad6b4893ac84cf617627fa3fb8ce0"}]},"branch":"refs/heads/master"},"5d07903f75a08df272687fa550c3134dd35af2f5":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2021-05-21 00:53:13.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/21"}}},"commit":{"parents":[{"commit":"4c5ebe7ad8d3749c154c79577d87c050c4cc25b7","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c5ebe7ad8d3749c154c79577d87c050c4cc25b7"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-21 00:52:00.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d07903f75a08df272687fa550c3134dd35af2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d07903f75a08df272687fa550c3134dd35af2f5"}]},"branch":"refs/heads/master"},"6629c24624b8f927578898c8c06d7b7bdabe09ed":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2021-05-25 03:30:34.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/22"}}},"commit":{"parents":[{"commit":"b1e59a58667892011cb06c2790f4b8bfa662e329","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1e59a58667892011cb06c2790f4b8bfa662e329"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-25 03:29:40.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6629c24624b8f927578898c8c06d7b7bdabe09ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6629c24624b8f927578898c8c06d7b7bdabe09ed"}]},"branch":"refs/heads/master"},"140f959106a905c3a3a8eb3eb20e3d2252278167":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2021-05-26 11:12:30.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/23"}}},"commit":{"parents":[{"commit":"23aa0c530e03f6271a31a6c80f1c7c7554c51f86","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23aa0c530e03f6271a31a6c80f1c7c7554c51f86"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-26 11:10:14.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/140f959106a905c3a3a8eb3eb20e3d2252278167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/140f959106a905c3a3a8eb3eb20e3d2252278167"}]},"branch":"refs/heads/master"},"8e6a7d9b4e2dfdf82fd4dd0df947342bdf64f61a":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2021-05-26 12:27:09.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/24"}}},"commit":{"parents":[{"commit":"ae3a94d318cc2e0abb7b2052afbc677c62757438","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae3a94d318cc2e0abb7b2052afbc677c62757438"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-05-26 12:25:01.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e6a7d9b4e2dfdf82fd4dd0df947342bdf64f61a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e6a7d9b4e2dfdf82fd4dd0df947342bdf64f61a"}]},"branch":"refs/heads/master"},"0bf530c21b7955abec01a2dac86525bb6c0fbc7c":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2021-06-07 03:40:41.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/25"}}},"commit":{"parents":[{"commit":"403c69542943c1b884dbc12dd86cccc3a6f610b2","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/403c69542943c1b884dbc12dd86cccc3a6f610b2"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-07 03:25:54.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bf530c21b7955abec01a2dac86525bb6c0fbc7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bf530c21b7955abec01a2dac86525bb6c0fbc7c"}]},"branch":"refs/heads/master"},"ca27acef6c287e0477b2afb914f82887b9172b32":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2021-06-10 01:27:34.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/26"}}},"commit":{"parents":[{"commit":"d7b15a5eaee7403eabd6c3beb23c37899df8e03e","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7b15a5eaee7403eabd6c3beb23c37899df8e03e"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-10 01:15:30.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca27acef6c287e0477b2afb914f82887b9172b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca27acef6c287e0477b2afb914f82887b9172b32"}]},"branch":"refs/heads/master"},"b8efc120d28c2083b3305c45bdb93374e1caca01":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2021-06-10 08:04:15.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/27"}}},"commit":{"parents":[{"commit":"3a0185a4a8bb13a086f4cc9003de2577024494cc","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a0185a4a8bb13a086f4cc9003de2577024494cc"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-10 08:03:04.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8efc120d28c2083b3305c45bdb93374e1caca01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8efc120d28c2083b3305c45bdb93374e1caca01"}]},"branch":"refs/heads/master"},"732ce36e3011bc619db351117c9ea499a452b0aa":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2021-06-21 06:51:36.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/28"}}},"commit":{"parents":[{"commit":"4b057dbd3888f3384972da044a9c09326e7676ab","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b057dbd3888f3384972da044a9c09326e7676ab"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-21 06:43:55.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/732ce36e3011bc619db351117c9ea499a452b0aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/732ce36e3011bc619db351117c9ea499a452b0aa"}]},"branch":"refs/heads/master"},"49ad498f51e66310e1bd39aceecc8d14449ee537":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2021-06-21 10:30:09.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/29"}}},"commit":{"parents":[{"commit":"b5efcd4869c756c027fe32e7ddec167e1d130782","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5efcd4869c756c027fe32e7ddec167e1d130782"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-21 10:05:12.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49ad498f51e66310e1bd39aceecc8d14449ee537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49ad498f51e66310e1bd39aceecc8d14449ee537"}]},"branch":"refs/heads/master"},"28d38d0f7f8de5212d0b7600fdf9501f4f53183a":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2021-06-22 01:40:35.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/30"}}},"commit":{"parents":[{"commit":"ebd5307f84740038af2c809f9f0953786b72e39e","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebd5307f84740038af2c809f9f0953786b72e39e"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-22 01:39:20.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28d38d0f7f8de5212d0b7600fdf9501f4f53183a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28d38d0f7f8de5212d0b7600fdf9501f4f53183a"}]},"branch":"refs/heads/master"},"f4b7281f0e237aa087d6ff5ca47f8ab1424a86f9":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2021-06-24 00:43:45.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/31"}}},"commit":{"parents":[{"commit":"56e8498a4dac9e4db6185cc5d97612d824c6e8d5","subject":"Add agent extension \u0027dhcp\u0027 for ovs agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56e8498a4dac9e4db6185cc5d97612d824c6e8d5"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-24 00:42:30.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f4b7281f0e237aa087d6ff5ca47f8ab1424a86f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f4b7281f0e237aa087d6ff5ca47f8ab1424a86f9"}]},"branch":"refs/heads/master"},"4b22eea4be54015c6b517257013597edbe725806":{"kind":"REWORK","_number":32,"created":"2021-06-28 09:30:40.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/776568/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/776568/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/776568/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/776568/32"}}},"commit":{"parents":[{"commit":"3127bd1d575594dc3882bb5deb52f0f9816744fa","subject":"Merge \"[ML2] Change way how list of supported API extensions is made\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3127bd1d575594dc3882bb5deb52f0f9816744fa"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-01-27 01:49:56.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2021-06-28 09:27:47.000000000","tz":480},"subject":"Add fullstack test case for OVS DHCP extension","message":"Add fullstack test case for OVS DHCP extension\n\nThis patch will create two fake VMs to verify the DHCP R/R\nfor its NIC. Ping 4/6 will be run after the interface request\nand config the IP address via DHCPv4/6 by using dhclient.\n\nFor fullstack test fake environment, setting enable_traditional_dhcp\nto False means enable DHCP extension for OVS agent automatically.\n\nPartially-Implements: bp/distributed-dhcp-for-ml2-ovs\nCloses-Bug: #1900934\nChange-Id: I40d6464953fbc4d4ca2c05a6051eba43cd05fedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b22eea4be54015c6b517257013597edbe725806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b22eea4be54015c6b517257013597edbe725806"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
