)]}'
{"id":"openstack%2Fneutron~776643","triplet_id":"openstack%2Fneutron~master~I1a5205303ebede533ffa6b3b85d4ec2f4b112426","project":"openstack/neutron","branch":"master","topic":"get_alloc_routerports","hashtags":[],"change_id":"I1a5205303ebede533ffa6b3b85d4ec2f4b112426","subject":"Implement \"IPAllocation\" router ports allocated retrieval","status":"MERGED","created":"2021-02-19 13:59:28.000000000","updated":"2021-03-01 10:36:13.000000000","submitted":"2021-03-01 10:33:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776643-get_alloc_routerports","meta_rev_id":"ed17dd75385657edd6bc73da8fb6878e3adea75b","_number":776643,"virtual_id_number":776643,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-02-25 12:54:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 10:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-01 08:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-01 08:52:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-03-01 08:39:27.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2021-03-01 08:39:27.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 15:17:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 15:50:22.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-19 16:19:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 23:39:52.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-23 09:45:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-23 19:38:21.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-01 08:39:27.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-01 08:52:56.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"c2f3384f3094a6b472cb7bbed4a6ec9bb5803b29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-19 13:59:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de44284ddd4d4291309e15475f003a6b6edeefd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:17:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05d5d94d68b94f69871ea620553567f2 : SUCCESS in 44m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2dc10e611cb415c8cdf7f586694c462 : SUCCESS in 47m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"394703e2c41100d8514dffa57fe6d7246a4eab7e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 15:50:22.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776643/1 : FAILURE in 1h 50m 07s","accounts_in_message":[],"_revision_number":1},{"id":"fbce8ea76b9337ba5dc53a02b2ca657315ce6235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 16:19:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d5561deae07747759bdc69c2d7f926ae : SUCCESS in 56m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7d59d0e91c58431992b7a737f6aac287 : SUCCESS in 1h 41m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8219c04d156a4376bd1ea766ae4a620e : SUCCESS in 1h 24m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fadefef7d2a145a5815cd061fece956e : SUCCESS in 1h 29m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/67a5fb68db154dfe86d27b239a0e61cb : SUCCESS in 2h 06m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/123100b80465409eb704a252ef41fea7 : SUCCESS in 41m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41dd2742c9bd4c448d3da7535a6e6d82 : SUCCESS in 28m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0cad6211d594ab1a93c5becc5e4ce32 : SUCCESS in 13m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a40d6e4b1a094c4e933c36059c91952e : SUCCESS in 31m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/602edde8bbd843fa84b360e7801917b4 : SUCCESS in 20m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34efd58178fb4ebba0d1780a5ab999ab : SUCCESS in 32m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de21145e2e4420093e1da59e919c249 : SUCCESS in 9m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eec04ea9047049d2b5dcea2368069c46 : SUCCESS in 47m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/47542b90c5fd4fb39ff57faaaa7ddae1 : SUCCESS in 1h 08m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0946fc4390d04e8c9717cbbc71a270b4 : SUCCESS in 1h 51m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0880af9288d5479990e20b202392412a : SUCCESS in 58m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a987559bd0dc41ac9db6ed78131cee62 : SUCCESS in 1h 02m 12s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/37a11279be1e4128b6c800c0d1c5bd0e : FAILURE in 1h 08m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7ba0384508424a399aeb819ff525bbf2 : FAILURE in 1h 15m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a787d814b43f420b80524818102907ab : SUCCESS in 1h 43m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/628b4c40d73941b0b4c766abbbee69ab : FAILURE in 1h 15m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a02565c74dfd42078bd473f3f83c5783 : FAILURE in 1h 02m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e767531993804e6797b8a8ef5ae5ad2d : SUCCESS in 46m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f27696d6c1a4f3f9fb4215fbfd5ac86 : SUCCESS in 1h 39m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9a96638063a24dc196556e51de2cb0dc : SUCCESS in 1h 32m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ddd577d1668d3960f1cf71479268ef129538d89d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 23:39:52.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/43/776643/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/43/776643/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"7f2b9568c6e03fd803de6ffb4949679dc39b64e8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-21 10:22:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bfb20407300027eec8f2ec3599dfd5755543bc3f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-21 10:54:02.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776643/1 : FAILURE in 31m 10s","accounts_in_message":[],"_revision_number":1},{"id":"baa812b6c569d79ee2b8c23ae3fbaa8e377200ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 11:12:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63e870e575cf4abda333cfedb8b403ea : SUCCESS in 42m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8bee7d7c720c4d33b5c0eb117a6d4618 : SUCCESS in 45m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3aefdfe5374bcb76a493df225ccb97594921a691","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-21 11:32:31.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/43/776643/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/43/776643/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ad2b1eafb6f9d3e770f99d57f748f04404ccc3fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 12:02:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b53ad0b9d1354eb7a4e5a998aebf074e : SUCCESS in 50m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fc3ee0a8b47149bda67c41ff99fbaebf : SUCCESS in 1h 36m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/91eef7e782c8451b926b411c195f1cbc : SUCCESS in 1h 06m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f34574748f594ea38b4596bf9b38a727 : SUCCESS in 1h 21m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b3ee9718c2004eac91017fa26a48ec39 : SUCCESS in 1h 39m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/763eca9b3fa44b4a95e29fb54d5ae74b : SUCCESS in 36m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aefb199e2c724b858801ed72a0882fb3 : SUCCESS in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/074dff3520e649a1894a5fbf6f6cf40b : SUCCESS in 11m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07f87fcaee5e4ca286776515efface9c : SUCCESS in 20m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38feac0972b049cbbb6171d80db8913f : SUCCESS in 26m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c3f5e07c58a447ba30070203640301d : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ccfdb8574be486983da5797e9a2e151 : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad644126a97a48b587f2c086fcaa69f4 : FAILURE in 1h 05m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a93938aa46424e7ba91304031805eddb : SUCCESS in 55m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7532ccff8f264133a76a3f44e784458d : SUCCESS in 1h 22m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3696a2edba0148f2b55b41486476c650 : SUCCESS in 57m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6883870d1ca34212886ef1274dcc06d9 : SUCCESS in 54m 19s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/86048e53bbbc48dc802b3fe1cf77c721 : FAILURE in 58m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/334c503a59d345b3a831201f6eb5f284 : FAILURE in 1h 13m 58s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6e10488a9229421884e38f8fd1bdb0fd : SUCCESS in 1h 19m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9872fe48c2b14d3ab42de6797f98249f : FAILURE in 48m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b79264ec5fa34f8597bb4074f0ee6691 : FAILURE in 1h 00m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cb20aded62fb4a0db9f1fe8e2087eb92 : SUCCESS in 33m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ac2ee92c27394a9db14a6355e6876e88 : SUCCESS in 1h 07m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/55e911abaa2e41379358d7e4352301c1 : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c3575ddc2026af1669124a06c3b1d1d6143b618f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-22 08:25:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89db9f26684657435b1170f3f605deb233bcde02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 09:07:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb9b6e8d13f54aee92e42b2449fb6ad8 : SUCCESS in 39m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87c3d52e1aff4d39930f6c4543ad8fbc : SUCCESS in 37m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f96cac7c43b514b9fcd7f97b4e78f15b39faeab","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-22 09:13:00.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776643/1 : FAILURE in 47m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7086bbea13c82e3bbb1da231d24c6f8518837f20","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-22 10:12:44.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/43/776643/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/43/776643/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e9e59c35fd4e4dd97812edb155b01ba6b2fdc40a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:30:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0a3004a807174b67bcc5f4134dfb4270 : SUCCESS in 42m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/de76cd34d2b84911be6897cab3a913e4 : SUCCESS in 1h 21m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c3318a2b9b504567a484ea9d8df3e3a2 : SUCCESS in 1h 16m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e5d34d04d4a46f6b075022506bb615b : SUCCESS in 1h 59m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0a51afce0cd74529ae30c688738f66cb : SUCCESS in 1h 51m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5da7d39a7c8474fbf6a5915ced5983c : SUCCESS in 37m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb6ca6bcbade4d11a8c469aa75b120a0 : SUCCESS in 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ffd25f4f0b04c27ad6b4f0c17f97e78 : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/190b21408dd447869d6ddf26ffb0ec2f : SUCCESS in 30m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8864cab2739438bb96e23064931a3e5 : SUCCESS in 28m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/955ba4e5eaa54623a1fad887a0d2fe5e : SUCCESS in 31m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64123892293544da8fc915f893e636ab : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a47ca0e5757f49d8abccf5220ba1cc2b : SUCCESS in 49m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3119a44d2bef42a1bb59a60ad822f183 : SUCCESS in 1h 16m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0c223a32deee4b9ab9c59cd3305c516b : SUCCESS in 1h 35m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/724ccdfc6da44ae6a8bba3f90c13f6f4 : SUCCESS in 1h 05m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/979e55d2b17a47e191f9a158285d245b : SUCCESS in 1h 05m 23s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/44ea8deb1c52407d85fe7d49b3752980 : FAILURE in 1h 11m 49s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/046fc639d56041f4bc9f45e40581342c : FAILURE in 1h 19m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/48eec4ce489d4385aa9c036feca22b4d : SUCCESS in 1h 36m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c1f6a4c6cf141a582c316f557781e4e : FAILURE in 53m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/874b937aadbf44c38dfdce319c08b026 : FAILURE in 1h 03m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/60886276630a4e6e83d97c51e3f51aa9 : SUCCESS in 34m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5eadcb055573487f9e251ac2dcdeb919 : SUCCESS in 1h 31m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b0d6e197f4f54fc399a9cf2979e2f0d0 : SUCCESS in 1h 31m 31s","accounts_in_message":[],"_revision_number":1},{"id":"dee4524fc7ce191669ff975d0e63d1a219101de5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-23 09:45:10.000000000","message":"Patch Set 1: Code-Review-1\n\nI didn\u0027t check whole patch yet but it seems for me that the same 2 tests are failing all the time. So it may be related to the change itself. Please check that before rechecking next time :)","accounts_in_message":[],"_revision_number":1},{"id":"864dccdcc5eb5ea8687c796f5b754be45d91a2cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-23 11:30:42.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/43/776643/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/43/776643/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e4eabceebc9cec0c70352dcdfa1cbf2727bf45ab","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-23 17:44:14.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e I didn\u0027t check whole patch yet but it seems for me that the same 2 tests are failing all the time. So it may be related to the change itself. Please check that before rechecking next time :)\n\nYeah, I saw it, the same test. I don\u0027t know why this is happening but I\u0027ll investigate it tomorrow. This patch has very low priority, as you can see is just an optimization, but always welcome.","accounts_in_message":[],"_revision_number":1},{"id":"a2f9abc736ab2d0787e4a7bebb6339c6e8067a69","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-23 19:38:21.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/43/776643/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ba3b62596c1fffdd1bd7234bb7338f9ba339725e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-23 20:51:56.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/43/776643/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/43/776643/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9d568b0a2e8d741cecaf1ccda7dea8bc260d46d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-25 10:41:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bcf94e6146e949568e8f84b825748499c1dc0863","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-25 10:46:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"57ed12c2acc9c2c16492726fffa251e8291f2dbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 11:34:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc93bb66584b494da9709c0b686cc40a : SUCCESS in 38m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72eaa366169f44bca31e12626f67564b : SUCCESS in 38m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4e72b05b425cb127cc0bf0eafc388a7c9f9f2fc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-25 12:16:56.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/43/776643/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"9aba3b20ad679305477218e20437fff92c881d60","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-25 12:54:10.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/43/776643/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"73f888aea9539e690a3c7153dce3ba72bd78b2ea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 16:37:57.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/43/776643/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/43/776643/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"397489ba6a37889fc5b5ca4948350936d5b2ab77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 17:44:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6291d259518c49a8bec53257c8d9ef6f : SUCCESS in 51m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/706eff182edf42389cde2b29ac55c137 : SUCCESS in 1h 31m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/433cab50b2794b98814481543c5bdf80 : SUCCESS in 1h 43m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9b4bd90489664726bf6d8998d1b5571d : FAILURE in 1h 10m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a3282238cef64bea89cf411dab323c26 : SUCCESS in 2h 10m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa9087e4fc8843e4b1549ed482b489ea : SUCCESS in 40m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b01c4ab90073416e85bcc063b906ae52 : SUCCESS in 33m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ce67780e32a40f398550ebbd56727fd : SUCCESS in 15m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ccdd7f6fc01475e9ff48b94e0ab7fdb : SUCCESS in 33m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67ef19f9844546c0b498140cfc4caedd : SUCCESS in 30m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2819bbfe3a884d6e80eeb9f1bb1f91ed : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be3b106588545b7a3300cedf9e6f9f0 : SUCCESS in 12m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d55f728198d4f8e8e95d251511695b7 : SUCCESS in 49m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/be48faa398ea465ba62abe2d651f20fe : SUCCESS in 1h 11m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/381de18854af4418972de37e660172cc : SUCCESS in 2h 02m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0936072eb77442b0a97e88566840076c : SUCCESS in 1h 00m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7c27f6a11584de69d50458ab945e6e9 : SUCCESS in 1h 07m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/58892e5fb513487da0e070f88019b366 : SUCCESS in 1h 19m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1943242f5e034e0b954751612eb6236f : SUCCESS in 1h 33m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f4f171b06ebe490aae9508c305d599a3 : SUCCESS in 1h 55m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28fa68369956465eab6a4664135f3eee : SUCCESS in 1h 10m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c2ae5564dbdf49d9ba66eb1903fa24ad : SUCCESS in 1h 25m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cf732c8c3be74819b9ff7d7d80b80477 : SUCCESS in 36m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f37ac9c18755448d9ee5ad10fc07d70d : SUCCESS in 1h 33m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3fcaaeaf172149a7a8206aec2818a760 : SUCCESS in 1h 26m 38s","accounts_in_message":[],"_revision_number":3},{"id":"b427a22d8fb9425b185dddbc901c310ee18708d2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-26 18:48:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fe1e6431a6e4e8105ed653da163ddaa5e2d8fab3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 19:36:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f857277168340de8cee069b4cdf544b : SUCCESS in 35m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6282992b3a2f4ddc8583c51de8af3be4 : SUCCESS in 41m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c7641ad76e4fb5bc9359011bd5df8cd0dc4c85e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 20:41:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1a38f23040ff4a0e807e09180c804295 : SUCCESS in 47m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f3560b597a654110a530ee2dac7bbe16 : SUCCESS in 1h 34m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6ba50ad4d4b74104b67280756980b092 : SUCCESS in 1h 23m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9fcc88dd96bf4733934d7c77b69fbc09 : SUCCESS in 1h 18m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9ad50eee85e4490e963395644ba07454 : SUCCESS in 1h 51m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7141b874dc674c9b9482bf94304a8619 : SUCCESS in 42m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63b515b2abee48c88cc4663e902eeaeb : SUCCESS in 46m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0aff61d87714608ac4a827ba5c40de0 : SUCCESS in 15m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eec310889d91438693a4fadb45a0c505 : SUCCESS in 30m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1721cbb674c1416a89720547a3a9b75e : SUCCESS in 27m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c15f85c240b4c1ea03acdb90e58f2e1 : SUCCESS in 34m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b5358d10284701ae73a800d74fa654 : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a13eb703a2c41a2a67c4f216c5eee1c : SUCCESS in 46m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca4490e617154208b04b5c4f8870fa9e : SUCCESS in 1h 11m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/058a1b0912ea4cec852cd111e19c8f4d : SUCCESS in 1h 32m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd717544ddc14fa2820602d56490e8d2 : SUCCESS in 1h 00m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/715899739995441781a0b53c556b01e7 : SUCCESS in 1h 08m 56s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2d0c3607a95940b7af1ddee95744d47d : SUCCESS in 1h 13m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7c8b9f07f2964af79972b63516ab7e63 : SUCCESS in 1h 33m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1132dfef6b3e4b01a9a4e3bf9622fd00 : SUCCESS in 1h 35m 21s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c531f93f590649e0ab797264f3eec4cb : SUCCESS in 1h 02m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6242235c9b534c7cb09b8228c8918a47 : SUCCESS in 1h 04m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6eab35af63d74a3a8d7575fb53e3edb5 : SUCCESS in 27m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/da883e304929461cad0335e126d61dc7 : SUCCESS in 1h 07m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d1bb07133b2a483f8bbf2ec6dde19037 : SUCCESS in 1h 16m 26s","accounts_in_message":[],"_revision_number":3},{"id":"e7519854cb4d93691cc77b6ffe198a035f8918f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-26 20:51:50.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/43/776643/3/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/43/776643/3/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"1bece38609e7ce6583320574377cbdc97b89e962","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-01 08:00:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2a389b527607508c8c335395981f76ef0884624e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-01 08:39:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"01053822cd565856ff968ec2f6dc7b47611f357e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 08:40:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c579469c58cd4b3f53edd9dd5c79ca586eedf6a0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-01 08:52:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b47514ba6863b630508e0063f9151148f2e44282","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:33:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b8c47da45cd84585bfb42b01cc22b260 : SUCCESS in 48m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/49994b32a6694416a13eef4a5af8e3ec : SUCCESS in 1h 34m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b36323f2189d4db396931edb87377bde : SUCCESS in 1h 46m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d334a948953943aaa192d234d1cdb5fc : SUCCESS in 1h 44m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e2052037f6d24ac8bf36fcb2084b47df : SUCCESS in 1h 34m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f23e3b9aca964c249a7ae784e6102edd : SUCCESS in 31m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf5e8111c275498c947ef9f4d00746ff : SUCCESS in 13m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6f4ad03d48845de945e004abe9d068e : SUCCESS in 32m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/661090215c7a487b8cfcceac26115f1b : SUCCESS in 30m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288fe64bb2c346e4ab0404177d30d0e8 : SUCCESS in 9m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6636df1820c243d68107d6e72a85c153 : SUCCESS in 54m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2feb04a39a5474a962a822f04eadff9 : SUCCESS in 1h 08m 40s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6de0f15d978b4be283096d7e01dde1a2 : SUCCESS in 1h 16m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4c2518731f84f2eb036622f56e537cf : SUCCESS in 1h 14m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c9b0816207bc49929952660c2335fb11 : SUCCESS in 1h 04m 04s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c394ee62250465495bf1e5149185c42 : SUCCESS in 1h 33m 57s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33807b77e7814181846139e95ed16702 : SUCCESS in 1h 00m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d29fb6a0e5b6475cafa4756ea98d91dc : SUCCESS in 1h 08m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/05ccbe1b43804889bfbe89710d4bf6d9 : SUCCESS in 33m 11s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a12b6844550e44c8a0be90d6f5d9247e : SUCCESS in 1h 22m 57s","accounts_in_message":[],"_revision_number":3},{"id":"f2ffe8fb7d30751bdac13717faeae01eda0972de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:33:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ed17dd75385657edd6bc73da8fb6878e3adea75b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:36:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fded112efa748a9ba21e18397ea1c50 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b7956d80ca481864502ba16a7c3a5f47891e11c4","revisions":{"ea62ef30d5a5d8254219806c283acac236c3d444":{"kind":"REWORK","_number":1,"created":"2021-02-19 13:59:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/43/776643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/776643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/776643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/776643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/776643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/776643/1"}}},"commit":{"parents":[{"commit":"fa5725f06c67ea51b417f4a82c34c4eb14d7a7b5","subject":"Merge \"Make neutron-tempest-dvr-ha-multinode-full to be 2 nodes job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa5725f06c67ea51b417f4a82c34c4eb14d7a7b5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 13:52:22.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 13:59:13.000000000","tz":0},"subject":"Implement \"IPAllocation\" router ports allocated retrieval","message":"Implement \"IPAllocation\" router ports allocated retrieval\n\nImplemented a method, in \"IPAllocation\" OVO class, to retrieve\nthe allocated router ports.\n\nThe new query used dispenses with one of the inner join queries\nrequired in the former query, making it faster.\n\nThis method allows to filter by gateway IP address and can return the\nwhole query list or just the first register found.\n\nTrivial-Fix\n\nChange-Id: I1a5205303ebede533ffa6b3b85d4ec2f4b112426\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea62ef30d5a5d8254219806c283acac236c3d444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea62ef30d5a5d8254219806c283acac236c3d444"}]},"branch":"refs/heads/master"},"e0091c282973b9c8cc2dc5fb120920e64f6da6eb":{"kind":"REWORK","_number":2,"created":"2021-02-25 10:41:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/43/776643/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/776643/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/776643/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/776643/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/776643/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/776643/2"}}},"commit":{"parents":[{"commit":"fa5725f06c67ea51b417f4a82c34c4eb14d7a7b5","subject":"Merge \"Make neutron-tempest-dvr-ha-multinode-full to be 2 nodes job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa5725f06c67ea51b417f4a82c34c4eb14d7a7b5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 13:52:22.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-25 10:40:16.000000000","tz":0},"subject":"[WIP] Implement \"IPAllocation\" router ports allocated retrieval","message":"[WIP] Implement \"IPAllocation\" router ports allocated retrieval\n\nImplemented a method, in \"IPAllocation\" OVO class, to retrieve\nthe allocated router ports.\n\nThe new query used dispenses with one of the inner join queries\nrequired in the former query, making it faster.\n\nThis method allows to filter by gateway IP address and can return the\nwhole query list or just the first register found.\n\nNOTE: I think I found an error. The gateway IP used to check the\nallocated router ports was the currently used one, instead of the\nnew one provided, that is the IP address to be tested.\nNow testing in the CI...\n\nCloses-Bug: #TBD\n\nChange-Id: I1a5205303ebede533ffa6b3b85d4ec2f4b112426\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0091c282973b9c8cc2dc5fb120920e64f6da6eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0091c282973b9c8cc2dc5fb120920e64f6da6eb"}]},"branch":"refs/heads/master"},"b7956d80ca481864502ba16a7c3a5f47891e11c4":{"kind":"REWORK","_number":3,"created":"2021-02-25 10:46:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/43/776643/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/776643/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/776643/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/776643/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/776643/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/776643/3"}}},"commit":{"parents":[{"commit":"fa5725f06c67ea51b417f4a82c34c4eb14d7a7b5","subject":"Merge \"Make neutron-tempest-dvr-ha-multinode-full to be 2 nodes job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa5725f06c67ea51b417f4a82c34c4eb14d7a7b5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 13:52:22.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-25 10:46:12.000000000","tz":0},"subject":"Implement \"IPAllocation\" router ports allocated retrieval","message":"Implement \"IPAllocation\" router ports allocated retrieval\n\nImplemented a method, in \"IPAllocation\" OVO class, to retrieve\nthe allocated router ports.\n\nThe new query used dispenses with one of the inner join queries\nrequired in the former query, making it faster.\n\nThis method allows to filter by gateway IP address and can return the\nwhole query list or just the first register found.\n\nTrivial-Fix\n\nChange-Id: I1a5205303ebede533ffa6b3b85d4ec2f4b112426\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7956d80ca481864502ba16a7c3a5f47891e11c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7956d80ca481864502ba16a7c3a5f47891e11c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
