)]}'
{"id":"openstack%2Fneutron~776701","triplet_id":"openstack%2Fneutron~master~I8ba38b8ace8852009fba8712aafa9f88c2b93ccb","project":"openstack/neutron","branch":"master","topic":"bug/1578989","hashtags":[],"change_id":"I8ba38b8ace8852009fba8712aafa9f88c2b93ccb","subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","status":"MERGED","created":"2021-02-19 19:39:14.000000000","updated":"2021-07-20 21:25:54.000000000","submitted":"2021-07-20 16:42:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"776701-bug/1578989","meta_rev_id":"a1d6ce94b198c2a4b8b47e952268b76be882eca4","_number":776701,"virtual_id_number":776701,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-07-20 12:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 16:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-20 14:18:30.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2021-07-20 21:25:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-20 14:18:30.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 19:41:12.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-19 20:23:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 22:52:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 01:38:41.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-24 02:46:43.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 18:21:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-25 18:21:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-29 10:02:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-31 08:29:30.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-04-19 11:56:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-06-14 19:43:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2021-07-09 09:52:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REMOVED"},{"updated":"2021-07-13 09:41:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"2de06db3dc92d31ba365b112d8b283e16c09b135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-19 19:39:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6002086fb73a24ecfdddad86e1ea1bb68d14a50b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 19:41:12.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776701/1 : FAILURE in 44s","accounts_in_message":[],"_revision_number":1},{"id":"601034f2c36d94fceb9ec4869405bb7b61256dab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 20:23:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df831767ebf743198aa4d7357945467c : FAILURE in 37m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05df8ec98ed34b7b9d20aa21af6a1dc4 : FAILURE in 38m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8460becb00e73dd6914ac8ec579cf68ddf01edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 22:52:51.000000000","message":"Patch Set 1: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3ab427266dee4eae84aee28798f65d7c : SUCCESS in 52m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6a4f6e251bb74d3ca7731592f2e995fb : FAILURE in 36m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92e2726788454861925fdee2c0abbee8 : FAILURE in 21m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ce990ffa7c74aca85c61feb3b4007f5 : FAILURE in 29m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9498cb56ca074170bc5c729c85f23882 : FAILURE in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f5fd11f3f92402faf6ca2bc4b5cd69f : FAILURE in 30m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97420e417f534329a9d6b98f8926ff7f : FAILURE in 26m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b56c2656c3c44c67b1133ecbacc84b0c : FAILURE in 28m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdefea755e744205bcc6fd4d4941ff80 : SUCCESS in 12m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/509c2af895624209877fa8e453cee5c8 : FAILURE in 47m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c99743007ba4c27839b27c40faeaafb : TIMED_OUT in 2h 16m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a6c556644cac409a8deabf31f1d8930a : SUCCESS in 1h 30m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cef89c0a4171498b8233074d1e9a8c5f : SUCCESS in 1h 14m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/16dfebbe954149abb2f8f318c58eeb70 : SUCCESS in 1h 07m 17s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/659c599884fb41d2917e0a392ce19c3e : SUCCESS in 1h 26m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/98403b0f6f994b038dc0751302ef0ac8 : SUCCESS in 1h 26m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cf1bbba0dac44b78bed2687a835f59ea : SUCCESS in 1h 27m 42s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cd9fc8d7bb4467d838e53bb36db00ad : SUCCESS in 1h 05m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7c1e55469e964f5aa96e8d2dafb5e5ac : FAILURE in 19m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/44e2185ce68e4f96993860812bedf343 : FAILURE in 21m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/84052abddc754e309a24c8b03a5d8886 : FAILURE in 26m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d11d0b6d1f79423883780db6604285eb : FAILURE in 31m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f60eddcdcd1d059104fdb7144c6ce509f5b3cf76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-23 17:44:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bd667eb78a533f3f5bdd563fc13292602da43822","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 18:22:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/718b6bb9aa874f74b54122f4d62249f4 : FAILURE in 35m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5007e72d19d4964bad76ece27a1df23 : FAILURE in 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cbd3fcf09f9abf4b2990b0fd9a176f0b04248460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 00:08:37.000000000","message":"Patch Set 2: Verified-1\n\n(23 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/81a76496325b4a4a95cfe308e1584a97 : SUCCESS in 50m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9b4ed22aaf7c4d118950ed51a58b945d : SUCCESS in 1h 36m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/37c8659cb56f44f0959a33bcf9ce0b92 : SUCCESS in 2h 21m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20b16d2aab874f0382372a9ce1f449c8 : SUCCESS in 1h 18m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3131f1bde84c40e7aca746f12f941d5c : FAILURE in 29m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f43f8dc616bc407d82d8d9aab2320e17 : FAILURE in 34m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3258f8a2c0e94ac7804a04fd929a454d : FAILURE in 42m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b06c528a5b4c456eb19fb63d5f7c00fb : FAILURE in 11m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ee789bfda8d4f07ba492e5255e0854f : FAILURE in 27m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d52c9a71d554a32bc385c69297c5b3a : FAILURE in 29m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc0919bb434c4b3f9d569fded09e8f39 : FAILURE in 32m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee79bb506283476eaf88e05a8f12286b : SUCCESS in 10m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/941b5c02728a4797842fe2673dde828e : FAILURE in 44m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cea91e078cc420592688882d33e89a9 : SUCCESS in 1h 08m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1a8e93471751436695cf109902c204c1 : SUCCESS in 2h 03m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb7da57b94c34f94a573f33118ae42b4 : SUCCESS in 1h 04m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/34a95470acab45dcabbe2b5b2b972e3f : SUCCESS in 1h 14m 20s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8ff49d9626794298ad437b0ee2ec22fb : SUCCESS in 1h 19m 32s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0512596ff70e449182a3c5aded50a812 : SUCCESS in 1h 45m 58s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa6e5384e0b142e68632c4f4f343bb6c : SUCCESS in 1h 38m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f276c32fa5bc45188d545cdefafcb931 : SUCCESS in 1h 25m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4fab731f38c5478dbb24277de54c94c3 : FAILURE in 25m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fe956165a2f74df6814de186a31a728f : FAILURE in 21m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2f9c7c52f9a84a3e925c6273c8e1078f : FAILURE in 20m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/824175f99fb94cf08a4177689ecca888 : FAILURE in 26m 58s","accounts_in_message":[],"_revision_number":2},{"id":"257cf733d86eaad828baf4d4ce6345dc2c816ac4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-24 01:38:41.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"087a6bc5d3db640de04d96c11cc5efaafabd10d8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 02:46:43.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/01/776701/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a9423bd2a76f14f72ce67bae70b912c854edf3f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-24 17:53:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9a81d14795def97c50f30f4558d8cf34524f8df4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 18:42:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3acb1e0502be439fb4c8f612772a7626 : FAILURE in 41m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ec462ff06f041a6b6d867866b25b153 : FAILURE in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9aafd7eb9a595aaa872d988e5a24652a94a13738","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-24 19:09:49.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/01/776701/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"1dcfca58b445b626fe2b4a359a4139b43c4d80ae","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-24 21:51:54.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"28d69110e175567408d7860de870cc4f4baa9f14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 01:37:13.000000000","message":"Patch Set 3: Verified-1\n\n(15 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6aca59d5754e47a1800eb0b06d30adae : SUCCESS in 55m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/38c94a13eac7491ba11461804c814cfc : SUCCESS in 1h 41m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6066318d9cd2484799f9da2321fd7902 : SUCCESS in 1h 19m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c03eff69e11f4e20937d3f11d495b016 : SUCCESS in 1h 43m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/86939ddbb32f41ce8c9431b8a4b96302 : FAILURE in 26m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03267c134aac448ba25a36f8d86a7558 : FAILURE in 36m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e32b4c639cfd469ca6df734bbc696862 : FAILURE in 30m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e52e8e00c3a41f4b2a5e64c7276ed38 : FAILURE in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61b8b3e5d98f4be7a07f7590faea95d1 : FAILURE in 33m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56d1fb83b62a4a8785b659c771eb3420 : FAILURE in 31m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d58a1334c994d33a3c60a8e5073a2c1 : FAILURE in 35m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42064c17eff345669a718e670419dbf5 : SUCCESS in 10m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e6fb5d98b304d819465f0ee7ab0addf : FAILURE in 52m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b2860cc67f34bf5be3f786825e185ee : SUCCESS in 1h 13m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9afd67a0facc4f2f97528515c8c6a4df : SUCCESS in 1h 33m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/731597cbcb3746d48f8302853c73d866 : SUCCESS in 53m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/01aff99776094252b1ac37f9ed831bd1 : SUCCESS in 1h 04m 26s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d263b5a6bca84709852dad54a7616d44 : SUCCESS in 1h 18m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d4a43811a204461691b2ba80b0ea8f7f : SUCCESS in 1h 32m 20s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b170509a0a54425a94b41ee845b8826 : SUCCESS in 1h 32m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7106bebb201e43aea34daa957fa8f71e : SUCCESS in 59m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/79bc300fcfca46d88a143497fec1b7c2 : FAILURE in 24m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0c09e77b30344e3882f98b4f7ae795a6 : FAILURE in 25m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/449e39dfdfd4406888ffaf094886a977 : FAILURE in 26m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8cc71cb13e3b46ca911350b697b1b1da : FAILURE in 26m 28s","accounts_in_message":[],"_revision_number":3},{"id":"c8106beaaaeea70a21f2b85c8eedee8f8884ad18","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-25 02:03:04.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/01/776701/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"8a7fc420b90a10ec862e5331533f03ac1d1388d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-02 11:49:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7b5e21a68feabcdc1bb8c621bc100e1f0818f58a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 12:39:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39ccf912e3754c709a72defa274e3a2b : FAILURE in 42m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/856fac5220074f37b4311311ac800bbe : FAILURE in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2e78224a6600c345623428d4dfe052da05e3ec7a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-02 13:34:07.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"a77ebcdf0d91da8f086da3c9025a6c758fb3bf05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 14:11:25.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e45e1c9011fa4e099df83e18903f7a6e : SUCCESS in 48m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1284569dbc39457f9ba2287d445588ed : SUCCESS in 1h 59m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f65d3c64ffbf45e2b7fd79eabbce4e5f : SUCCESS in 1h 48m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1dfe046bddca476bb12ea8716e0e1602 : SUCCESS in 1h 30m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0727ac90223845baa352cd9d30e34dc1 : FAILURE in 27m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3d188ff64b5432a8bfcae729169b8d9 : FAILURE in 43m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dc29d87b5cb40fe93199f1e46f14e0a : FAILURE in 25m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01459c9ad31145a895d1a7e16de11123 : FAILURE in 12m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee2088e4fa5f482dbfce8dd891bbd0f1 : FAILURE in 33m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ef26af9d48c43fabe6d2b04d8222d60 : FAILURE in 31m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8340470f86c54317a4c617c43e303325 : FAILURE in 27m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70d4df4a4f784852b956ca4d8c3edbfa : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d28b6dcb4b8d458fae37eb938ddcd010 : FAILURE in 56m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/86ae5cfa8acd4970b4cc3004550a73e3 : TIMED_OUT in 2h 15m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5b6e9ce6471449f5b3567f8067938844 : SUCCESS in 1h 31m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19f9c3b2926e45e0880611d92e685393 : SUCCESS in 1h 00m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5155a505cd3c4ff1af2bd61b9683f96c : SUCCESS in 1h 00m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d369c1fbbfe14624aacb50016e111acf : SUCCESS in 1h 24m 35s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/914ad119ad3a4c338cbfbec08ca24a6d : FAILURE in 1h 28m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8d21f3162aa24fcba5a6d5a5496f4ba9 : SUCCESS in 1h 34m 45s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ab782d5023047409460ee0b0e3874db : SUCCESS in 1h 06m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2906a721491143c08c20a78ba084e0db : FAILURE in 30m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2b366e60289c4e3c8348c7814ce61a94 : FAILURE in 26m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d80881d4bf084e59b5a1248f570306e6 : FAILURE in 30m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7ffad47b1ac248e3af87044381db8d6f : FAILURE in 30m 58s\n\nWarning:\n  Comments left for invalid file neutron/objects/port/extensions/port_numa_affinity_policy.py","accounts_in_message":[],"_revision_number":4},{"id":"dbc42ad574893b9f3d8e8b4ac76022fe78dffa29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-03 10:59:31.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e5a5a86ab7ab66bd68c971d99ed48d71d00a148b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 11:47:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99fccd669be74bd88203d3dbcfe42e21 : FAILURE in 40m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/126622cd35254ae18b25f425783c3ff6 : FAILURE in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"70111289162822ad967eb8d31ef7f447eff5d38b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-03 12:32:31.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"201d01c1339f0647c0c9746e0030bde7f66d816d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 14:08:20.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/64ce276a97504f9cb755bd51c4d6a9d7 : SUCCESS in 1h 09m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6a5a37797c2e42f4a1dd18906c2ae3a8 : SUCCESS in 1h 52m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8b5dd18befe14126a0cd8ebc484dfb2c : SUCCESS in 1h 37m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/18eeb5fb9e0944ad9947ec6b3642979d : SUCCESS in 1h 18m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bb084cf1b0774c8ebdd9948bd6abe5bb : FAILURE in 27m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fa0987780c641d9837f9c052d2bb4d3 : FAILURE in 30m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/149bc6ed92264f2aba7bcf2784090f3a : FAILURE in 30m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf95a872fe044bab16f2cb1f96a5963 : FAILURE in 18m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b86415c405e64d5f977d16372775b045 : FAILURE in 32m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec683d05221c42efbd83b68f3d0d9593 : FAILURE in 43m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ea3126805cb4ec1b5946c47efb6545d : FAILURE in 35m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e450dfe90e584556969c260e2c37e2f9 : SUCCESS in 12m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5c309899c1f4feb9c9e2860827d3eb1 : FAILURE in 42m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/122eb06a044d423c8c654023b8080a3c : SUCCESS in 1h 24m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/26ca23f45cd648f283923e7f2de23902 : SUCCESS in 1h 32m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c75a0d4bf1747cb821783b3134a7a52 : SUCCESS in 1h 02m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a75dce96450e4e6b93bd22b8d1d1d1a8 : SUCCESS in 1h 16m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a438c9a4ca1647dc997e154da73ed7b5 : SUCCESS in 1h 20m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/24b2fc707bec44ab9f1f8edefc9da9e6 : SUCCESS in 2h 13m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e58cdc4508a14ac5a9211666f0072916 : SUCCESS in 1h 52m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/500a821ab94d4d96b2faa71e9ca7bce6 : SUCCESS in 1h 06m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/79bf22cbe08f4dccb375f85f0333c308 : FAILURE in 18m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7b7b00bbb9bc43788bf4b98ed68ac73f : FAILURE in 24m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b38f9c1c59514044bd4e0d267bf0503f : FAILURE in 24m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4c5371e39a0243458c7cf1fbc5b7b46d : FAILURE in 32m 14s\n\nWarning:\n  Comments left for invalid file neutron/objects/port/extensions/port_numa_affinity_policy.py","accounts_in_message":[],"_revision_number":5},{"id":"977ef851b8165309b495d24066e2a69f885d9553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 14:09:00.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/64ce276a97504f9cb755bd51c4d6a9d7 : SUCCESS in 1h 09m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6a5a37797c2e42f4a1dd18906c2ae3a8 : SUCCESS in 1h 52m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8b5dd18befe14126a0cd8ebc484dfb2c : SUCCESS in 1h 37m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/18eeb5fb9e0944ad9947ec6b3642979d : SUCCESS in 1h 18m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bb084cf1b0774c8ebdd9948bd6abe5bb : FAILURE in 27m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fa0987780c641d9837f9c052d2bb4d3 : FAILURE in 30m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/149bc6ed92264f2aba7bcf2784090f3a : FAILURE in 30m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf95a872fe044bab16f2cb1f96a5963 : FAILURE in 18m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b86415c405e64d5f977d16372775b045 : FAILURE in 32m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec683d05221c42efbd83b68f3d0d9593 : FAILURE in 43m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ea3126805cb4ec1b5946c47efb6545d : FAILURE in 35m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e450dfe90e584556969c260e2c37e2f9 : SUCCESS in 12m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5c309899c1f4feb9c9e2860827d3eb1 : FAILURE in 42m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/122eb06a044d423c8c654023b8080a3c : SUCCESS in 1h 24m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/26ca23f45cd648f283923e7f2de23902 : SUCCESS in 1h 32m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c75a0d4bf1747cb821783b3134a7a52 : SUCCESS in 1h 02m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a75dce96450e4e6b93bd22b8d1d1d1a8 : SUCCESS in 1h 16m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a438c9a4ca1647dc997e154da73ed7b5 : SUCCESS in 1h 20m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/24b2fc707bec44ab9f1f8edefc9da9e6 : SUCCESS in 2h 13m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e58cdc4508a14ac5a9211666f0072916 : SUCCESS in 1h 52m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/500a821ab94d4d96b2faa71e9ca7bce6 : SUCCESS in 1h 06m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/79bf22cbe08f4dccb375f85f0333c308 : FAILURE in 18m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7b7b00bbb9bc43788bf4b98ed68ac73f : FAILURE in 24m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b38f9c1c59514044bd4e0d267bf0503f : FAILURE in 24m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4c5371e39a0243458c7cf1fbc5b7b46d : FAILURE in 32m 14s\n\nWarning:\n  Comments left for invalid file neutron/objects/port/extensions/port_numa_affinity_policy.py","accounts_in_message":[],"_revision_number":5},{"id":"35c42464085403fee1ce6c8c10ce55db9b13e10a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-03 14:09:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"00f7e80c6eeeb5d136e9b14c01fe356b715e5738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 14:09:50.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/64ce276a97504f9cb755bd51c4d6a9d7 : SUCCESS in 1h 09m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6a5a37797c2e42f4a1dd18906c2ae3a8 : SUCCESS in 1h 52m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8b5dd18befe14126a0cd8ebc484dfb2c : SUCCESS in 1h 37m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/18eeb5fb9e0944ad9947ec6b3642979d : SUCCESS in 1h 18m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bb084cf1b0774c8ebdd9948bd6abe5bb : FAILURE in 27m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fa0987780c641d9837f9c052d2bb4d3 : FAILURE in 30m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/149bc6ed92264f2aba7bcf2784090f3a : FAILURE in 30m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf95a872fe044bab16f2cb1f96a5963 : FAILURE in 18m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b86415c405e64d5f977d16372775b045 : FAILURE in 32m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec683d05221c42efbd83b68f3d0d9593 : FAILURE in 43m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ea3126805cb4ec1b5946c47efb6545d : FAILURE in 35m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e450dfe90e584556969c260e2c37e2f9 : SUCCESS in 12m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5c309899c1f4feb9c9e2860827d3eb1 : FAILURE in 42m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/122eb06a044d423c8c654023b8080a3c : SUCCESS in 1h 24m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/26ca23f45cd648f283923e7f2de23902 : SUCCESS in 1h 32m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c75a0d4bf1747cb821783b3134a7a52 : SUCCESS in 1h 02m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a75dce96450e4e6b93bd22b8d1d1d1a8 : SUCCESS in 1h 16m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a438c9a4ca1647dc997e154da73ed7b5 : SUCCESS in 1h 20m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/24b2fc707bec44ab9f1f8edefc9da9e6 : SUCCESS in 2h 13m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e58cdc4508a14ac5a9211666f0072916 : SUCCESS in 1h 52m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/500a821ab94d4d96b2faa71e9ca7bce6 : SUCCESS in 1h 06m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/79bf22cbe08f4dccb375f85f0333c308 : FAILURE in 18m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7b7b00bbb9bc43788bf4b98ed68ac73f : FAILURE in 24m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b38f9c1c59514044bd4e0d267bf0503f : FAILURE in 24m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4c5371e39a0243458c7cf1fbc5b7b46d : FAILURE in 32m 14s\n\nWarning:\n  Comments left for invalid file neutron/objects/port/extensions/port_numa_affinity_policy.py","accounts_in_message":[],"_revision_number":5},{"id":"d5120c862c2a0dc2de0b2dda993d48e9f9895024","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:13:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d237a4898c8440398cccaee7fa1ddff : SUCCESS in 40m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c33c1e81a4dd4ae9b066727b78ab064e : SUCCESS in 41m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6b42c31f8a0f0ebaecb99d168ea3077894fd29a9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-03 15:41:32.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"094299bbd209d7a1c23ddaf7838757ea8a06ce5f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-03 17:35:11.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"55fce23e4dd411e06c075b31dcdc0dc4175f139c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 18:06:58.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/04044442404a489ea83c29bb431dd9ec : SUCCESS in 1h 00m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/eb012bfedd724511b1281a15f87a9cd3 : SUCCESS in 1h 20m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/64d97c896b984815a7c516043204ed69 : SUCCESS in 1h 33m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/591684a1b1b34a8d9307593e8afd9eff : SUCCESS in 1h 38m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b4dc995820a94d108a87092a072f4ba4 : SUCCESS in 1h 56m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74e352f347c34986b01ab6996557bc2d : SUCCESS in 39m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11967c0441ae44a1a1d829e5a537ba5f : SUCCESS in 20m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f65052954b46c296c9b9d2add5a920 : FAILURE in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f87d12ca30134adfb4f7af5923d625f5 : SUCCESS in 39m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/769e0792318e4bc380cd8ba782e1e82e : SUCCESS in 34m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/357c0fd9f503408abad46d94dc71a76f : SUCCESS in 26m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c3c1e19edfb4183b4a39752dfd06503 : SUCCESS in 13m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/252b3beea22e4ea284bc3c1e7d63d375 : FAILURE in 50m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c736836a76a4d8aa58ec5c850f51db0 : SUCCESS in 1h 24m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/49b147941eb44e35aba7275f1ead068b : SUCCESS in 1h 39m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cdb8129ec08c49e783537f94b38a82bf : SUCCESS in 1h 24m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/36093d1e6b2747bdb2338382a54db477 : SUCCESS in 1h 07m 31s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f5a1229fd75345f3a2b6fa3a873e97a8 : SUCCESS in 1h 26m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/58a72b40a9e14f9280247c3df133f1ed : SUCCESS in 1h 38m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e26b35459484b36821cef3c93352f76 : SUCCESS in 1h 27m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b95e758f7ad440839c858b13295a0fea : SUCCESS in 1h 07m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e9d6df34cfd240079635d6875cc6685b : SUCCESS in 1h 11m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a76a086e2f41454d87fb9557bb111c6d : SUCCESS in 37m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a3e7e1bbbc3c47e38f449cec7b0e235a : SUCCESS in 1h 25m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6b5ffb9346ca44d992e2cd01aec3d5d9 : FAILURE in 33m 52s","accounts_in_message":[],"_revision_number":6},{"id":"aff275c907514044b54874518769b680f895d341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-03 18:23:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"52327d6b1e70ce01e901e7de6ea6625bdc25e927","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 19:07:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3cf658725180440198aed414d05db347 : SUCCESS in 36m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1da3145c1eb645898afa4113721b451d : SUCCESS in 37m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"11d58c23f7cacfe3b9cb528c87bd6a638be85d90","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-03 19:40:40.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"537b5121c24e9daebc48a1fa43bd7d913086354b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-03 19:47:32.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"ad6cb68f87d3159ee1bb1ee0130ed064e6ddeedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:08:13.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/353528513ad04e38a10fb8d00df78ab4 : SUCCESS in 56m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e6d7d63c98db4a82835134d4084dc5e4 : SUCCESS in 1h 25m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cb095f0219204787bf1750b518276bca : SUCCESS in 1h 28m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9150a26f75fc4b1098159ca10b5e110e : SUCCESS in 2h 12m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3b65d389bd534652951ffa6d91268d1f : SUCCESS in 1h 55m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a498fa83be54bd0aea84ae2c1c71024 : SUCCESS in 33m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5be8384a7ef7411493b91a81d6df29fc : SUCCESS in 30m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6ec8006b234882a077eea1429e8658 : FAILURE in 12m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e45c80f60ab64160930d6f9d7ce243a3 : SUCCESS in 33m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20aeeeeab4804d32a377ffd30a65df5f : SUCCESS in 39m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6f6f9b0b5d3465086768321b2b4b659 : SUCCESS in 32m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e90f69fe8ece479d9bbb93102027f577 : SUCCESS in 9m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/59502cd94ccc487fb33937f83b91a796 : FAILURE in 53m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/537a97d0746d46fab2361b99b13110bd : SUCCESS in 1h 27m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a64c3c61abf2423c84c096aaf0e7a5bd : FAILURE in 1h 54m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d171ce8bf42f43cc8f7599a8b0e322b1 : SUCCESS in 1h 08m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7ee9f9798204c779da8679e6d1ffd55 : SUCCESS in 1h 13m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9f8600a6ce60491494db839ef57a0aea : SUCCESS in 1h 36m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c3748574271241619eaee25814a3ce36 : FAILURE in 1h 20m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a38d29fc6bf44fe083476ba0e3a4eb23 : SUCCESS in 1h 47m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82890a277b144c32ad383157de0edcb0 : SUCCESS in 1h 05m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2d48389e81904ba3ac9435b43054b260 : SUCCESS in 1h 15m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/751bfffe59f14e69855b1709446b01e1 : SUCCESS in 27m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd741a3f218f4efb93e43546653cd6fc : SUCCESS in 1h 57m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a83104b94da4427cb5d9196118faf98b : FAILURE in 35m 17s","accounts_in_message":[],"_revision_number":7},{"id":"7f483fd0539dc2fc49386d20611915cecf0e4037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-04 09:33:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"69eb598b403a58a5887b4f1728c4eae598772eb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:25:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/268c305641324d64967cb986db1894f0 : SUCCESS in 47m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4552a099c9fe4dd6a2b676ac5af3eed3 : SUCCESS in 46m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"641318294f446fecb3b6ad41378bd899e95b7447","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-04 11:06:28.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"4f28e3e315de90952ca664aeec3491ef2ff2235f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 12:22:54.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ce8ac3b9227745c5aa2d163134037fe5 : SUCCESS in 39m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/caea824632594c388fc707f616a0e80e : SUCCESS in 1h 40m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cb50743b7fde44c6bc3cf6f716b7ca66 : SUCCESS in 2h 26m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f2d4b9a63fe84a839412714ad3cacb0a : SUCCESS in 1h 40m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7358836e3a604cb38db77a6b80caf786 : SUCCESS in 2h 43m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1191bdaa7768454e9529782602f632d0 : SUCCESS in 24m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2679b8dee1054c0bb4234ee1a45cd41b : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db3497af7584a37a6b12bf4624c7491 : FAILURE in 10m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4377cfcc46a45c2a2cb39d24d82a1d3 : SUCCESS in 34m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f73c082ecc141f1b7acd8a1786c023e : SUCCESS in 36m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8cc15e8fd8f4bff8136b70b6d3603f8 : SUCCESS in 21m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0903fcbebf63431591848db57b3e56a8 : SUCCESS in 8m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/42eb31dbd74f4f52a325438ce876bb7f : SUCCESS in 3m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b72620e3a35649eebd99f599a0a9dc1b : SUCCESS in 53m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74a845488fb14af99a4d8b2556666eeb : SUCCESS in 1h 43m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fcd1285946dc496b88ea3929ef2d5646 : SUCCESS in 1h 03m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e7882d8237b4c10851d19b7b812f08c : SUCCESS in 58m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3673eb76f4c9464c8e9387719d786711 : SUCCESS in 47m 55s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b30413ea59b34f44a0a7b28971bd94c1 : SUCCESS in 1h 09m 05s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/345f38ed83c2499482bcac22025cffe5 : SUCCESS in 1h 55m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fe6cc829b8eb474bbc9c12646f526a52 : SUCCESS in 1h 39m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbfabbb3ba964316900b7d458c696d7b : SUCCESS in 1h 15m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/767e422596f748c8bfd9b996addaf5e5 : FAILURE in 1h 24m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9aa3db5c66f04ba98c3adbef964fe7ff : SUCCESS in 35m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b22488c7af7543b283cdc5f6584db2b4 : SUCCESS in 1h 05m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/08d816fac66d4dde959594c7a11d7d82 : FAILURE in 31m 06s\n\nWarning:\n  Comments left for invalid file neutron/conf/db/l3_agentschedulers_db.py","accounts_in_message":[],"_revision_number":8},{"id":"8079b396db12794c38f53db297b95be9ecf64914","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-04 14:24:00.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f01c350de3fdaf7ab3b9d5c05bd22be3ace66077","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 15:16:04.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83f701f27b724b39b3e58efe1761e482 : SUCCESS in 42m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f0870a7fe42454e997bfcfd069a0ffd : SUCCESS in 41m 30s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d6d62e614822bf223b142bb71289c3fd99563bc5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-04 15:57:26.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"e2250ef33111c472bf74d987bcc30eb10fdfce2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 17:09:52.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e91cc6403e5b418989b98e683464edab : SUCCESS in 56m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e7c98ea712cd4d4fa59742173f9bf87c : SUCCESS in 2h 04m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/801c7060c5d84b3e955c10d6adc4b802 : SUCCESS in 1h 24m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dc7e37b3ebf345d2920ed785da605f15 : SUCCESS in 1h 27m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/27da9d431858450190cc064846682c66 : SUCCESS in 1h 45m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75c7678354ab4a889b54a56d5fcfe1ef : SUCCESS in 33m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f146f648df84e75a03cd980e2832541 : SUCCESS in 29m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9dcaa797264010a404f536f1c8934a : FAILURE in 13m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/740bffc4811343218821441481f86509 : SUCCESS in 36m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8919c7242117415db742a7cf62b52527 : SUCCESS in 42m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb9a2c01ff554900afc82d43d9608d15 : SUCCESS in 36m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31fd6cbf6b7f487886f92f33884b1bc1 : SUCCESS in 10m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2383aefbe3145cc87fc07fec3bda1cc : SUCCESS in 3m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/50d999b5078f43caa3bae3b7feec6dfd : FAILURE in 53m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/92d3b7c0bf164021bf20bce5bf612e9b : SUCCESS in 1h 28m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f10bb465ac344c8bbab3977eacdfc065 : SUCCESS in 1h 26m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b51d701cc6f747dfaa8eedc1f6978569 : SUCCESS in 1h 01m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/93cb1345008946ceb7d28cb31def816f : SUCCESS in 1h 02m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c47322c2c7db46d1b883fd479e91e46b : SUCCESS in 1h 31m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/67b0d967c42a4cbbb1e8a324215ae030 : FAILURE in 1h 31m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/684a64c53d514a82897390ce16d8f221 : SUCCESS in 1h 50m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/887afc353dd7435bb72c0bc8db07ee49 : SUCCESS in 1h 21m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2de315df1b0544e6a244772a12727d56 : SUCCESS in 1h 08m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0b1ed6fbff934838b39384db539cde5d : SUCCESS in 42m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b83b2c23aef49a6b15f74a1bf54872d : SUCCESS in 1h 45m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b20e853037704ba58778910f8f4d9630 : SUCCESS in 1h 27m 24s\n\nWarning:\n  Comments left for invalid file neutron/conf/db/l3_agentschedulers_db.py","accounts_in_message":[],"_revision_number":9},{"id":"18beef27ae9ff8ae529197e228caf357006655c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-05 12:41:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"92822335f98264ab532efc906c80307075a1c558","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:33:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75c858d8c659424cb7fafed04c205c4a : SUCCESS in 39m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3f7a5b1a17848d5ac927ba1c7b46e1b : SUCCESS in 44m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"29c34888da17827a116574e52a7c7a544ecb8a86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:50:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5aee0e7a7fbc42be87d59ed2fbb8fc1f : SUCCESS in 48m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/70a6dab8787c429fafc1e486d5f1833b : SUCCESS in 1h 24m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/00fb1f7ab8b3460aae899b0e168b948d : SUCCESS in 1h 53m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2498a2e92f5f4504a22de6111d84d7cf : SUCCESS in 1h 35m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a6556b64f4e64875b3f78f6464c09cf7 : SUCCESS in 2h 02m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc9706b7dbd646359cf54341ea3b87fe : SUCCESS in 41m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1b2bec4f31c4a4e82df3ba2dc558c7f : SUCCESS in 26m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/925dba3135234a86b694084109b71416 : FAILURE in 11m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b4bb0357605418ba256c18236b43ffd : SUCCESS in 33m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0be10760c9bb4008ae437616eb5d5e69 : SUCCESS in 31m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08a1c120b6e241fdb6a2e2008ba8be06 : SUCCESS in 30m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73d9b049af854a1b9f32beb141b0dac4 : SUCCESS in 11m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/baaa2632808e41d5bf8b316d53757d1d : SUCCESS in 4m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/480d348f69f4434b9f97b4cb15edd00e : SUCCESS in 47m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ff6aa5f7bbc4a68aa98e109d9873308 : SUCCESS in 1h 19m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9a042ba391024ef58e268d576e75c2a2 : SUCCESS in 1h 31m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de0550f24c4c4fda980c95d2011fe337 : SUCCESS in 1h 01m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9837ddccd29f4b948635a9d38b23ba70 : SUCCESS in 1h 00m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0d9f1c2de9d3437ea50252f288ae90ad : SUCCESS in 1h 19m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7adcb44bd4784a79bd0e263541e37dcf : SUCCESS in 1h 33m 01s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1b42c79b3304891a49bb36b1a945440 : SUCCESS in 1h 42m 23s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02d6ceeea9094c8ab52973b79813011c : SUCCESS in 1h 04m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cff808eac0d844b5a0a57e01f1703fee : SUCCESS in 1h 13m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c7505902cfc147fa8264e82de90d382b : SUCCESS in 36m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5e6dc7b516e4b68927168f5051ae7f0 : SUCCESS in 1h 31m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/56c6737574944fe8a6d58945493381b4 : SUCCESS in 1h 24m 15s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file neutron/agent/ovn/metadata/__init__.py has only 0 lines\u0027\n  Comments left for invalid file neutron/agent/ovn/metadata/__init__.py","accounts_in_message":[],"_revision_number":10},{"id":"f7c5fe30152c3c15c0167cff9a6b016610baba4c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-05 18:34:18.000000000","message":"Patch Set 10:\n\nI don\u0027t understand last pep8 error...\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"da6fc726224f0f94584a3ee5563988e1bc960a76","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-06 07:17:35.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"8747b09ba106cd626d650e865dd9d4a4645c1fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-06 07:17:49.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"35be8c40b09a6e98c52bdf1323dd0b4e0d0781b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 08:23:55.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50cbbec19be8448896a77e80eb07766c : SUCCESS in 54m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81ee01ab81544484ac76d1caeb4bef0f : SUCCESS in 53m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e70fdde104a4f7b20a919326500c7d058977fdd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 10:02:03.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2862142befc540449241adf128aa1880 : SUCCESS in 49m 51s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/11acace767ea4ae680467ee801d4e217 : SUCCESS in 1h 48m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a784316710854449828010bb6e276046 : SUCCESS in 1h 43m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21fda40cc0124e83a5249f6ba0a16004 : SUCCESS in 1h 35m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/68998a97926242e397130fc032ff3f1a : SUCCESS in 2h 39m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebef8d6e10f641bf8c92a3860995e99e : SUCCESS in 39m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/833ecdeeb1fc4800ad6af4ebfd2c4388 : SUCCESS in 32m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/679cef632fad48299fc7550bc52d681c : FAILURE in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f511a05d64314263bced5724ab05d029 : SUCCESS in 40m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e011a7a99dc4820b7e7f8aaf87b26ae : SUCCESS in 31m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d323609e0574ede9f666ce3cef6488e : SUCCESS in 31m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c6c08fab64438e8d60de8c5d154ce4 : SUCCESS in 10m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc913f4dcba743ee8ce084212712947b : SUCCESS in 4m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1cce87f4dbb54ddd9544cf3c5a130aa7 : SUCCESS in 51m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b01c255513442f8968ece93a331e91c : SUCCESS in 1h 22m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/db937cc25b0f4cc099377892452f8476 : SUCCESS in 1h 35m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/699c8705dde04e698336ab0c963de10a : SUCCESS in 1h 07m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9bd1df8690124ed4a561b9dbd7f9c48a : SUCCESS in 1h 00m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/34bccc08f1ec40949e6250d1f12ea4b2 : SUCCESS in 1h 16m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fbaaeb414aeb4b1a897d4cdb4d862dbe : SUCCESS in 1h 25m 39s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e31704a0668a42e4b5fd07f9964a7e10 : SUCCESS in 1h 36m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a488f47ef504b30ba6534f8050f5075 : SUCCESS in 1h 11m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8cf1e1bb49e0459ca23bc2464ed6755e : SUCCESS in 1h 06m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/df725f4e4c28473cbc4263f327532b42 : SUCCESS in 41m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1800f6f74ba44978815ca4ac8119adf1 : SUCCESS in 1h 31m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d8fba31008874fdc817f578e24b9f8d3 : SUCCESS in 1h 22m 43s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file neutron/agent/ovn/metadata/__init__.py has only 0 lines\u0027\n  Comments left for invalid file neutron/agent/ovn/metadata/__init__.py","accounts_in_message":[],"_revision_number":11},{"id":"0db52e14ef86abaee4bcd123e789939f895aa0de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-07 18:20:32.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"f86a8cda9b8d2f15661db5c1020beb2a6efc0d10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 19:21:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2d2cac550874c528f14dcb4799205de : TIMED_OUT in 1h 00m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2afc324879574f289074001c2ded6b53 : SUCCESS in 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"906486967b3e951776cfffcba853840502c4b000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 20:32:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f9d40c83670a46e0a23c4f0f75524671 : SUCCESS in 46m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/39a7c5b40d9c4fc19890ce0878d14285 : SUCCESS in 1h 28m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dacf034c5c984439a76fea3c76a45a13 : SUCCESS in 1h 03m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ea196e77c80c4045aa8b4515f4ba1862 : SUCCESS in 1h 44m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fdd993853724497e86908b386e2e85d3 : SUCCESS in 2h 06m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff121ae0ba7247fca6ef39f2d8342134 : SUCCESS in 42m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f65aa21a7dbc4e6c8c0dc78172d8ccfd : SUCCESS in 36m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee207cae75d4b0db0a8d5aba07231b2 : FAILURE in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb611d6ab6414a0fa4c724d057cfff22 : SUCCESS in 39m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05c514153481452598d038ef72ff84e9 : SUCCESS in 39m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77b1b2c440ef485d9b57dd2e087d79ea : SUCCESS in 33m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1f02839c3e41fabe4d71612e289d96 : SUCCESS in 9m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f8b46b12f7ee47a0ac86ea586517e485 : SUCCESS in 3m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6922e27df5674635885b48faab251267 : SUCCESS in 51m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c28b2b617444d47ac0d2475189d0898 : SUCCESS in 1h 20m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3da8ce7eb37847e1a2a5bd8f59680637 : SUCCESS in 1h 47m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d384a15b10f54d8d9ed3872e3e690c96 : SUCCESS in 58m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/db821437251340f0b356a9b2960a1ae7 : POST_FAILURE in 47m 08s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b1ac7b177f934d398d65bd9135da98da : SUCCESS in 1h 19m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b4410ceaf80243e78c6a58523357dc8d : SUCCESS in 1h 29m 14s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b52e1db4e4a64041b5e7fc65f50bf173 : SUCCESS in 1h 49m 42s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f02670178a041f888ccd89557899689 : SUCCESS in 1h 06m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/079a676caf7749b58d227bbbb9bdd0cf : SUCCESS in 1h 07m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/def876e0b9d1452c8be4350792172696 : SUCCESS in 39m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e0c8581ded36484aa7aa9872014726b0 : SUCCESS in 1h 33m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/52e5f41bfb75419fa1e93b1130b6b8fe : SUCCESS in 1h 28m 49s\n\nWarning:\n  Comments left for invalid file neutron/quota/resource_registry.py","accounts_in_message":[],"_revision_number":12},{"id":"a12ef91b5887e6aafec31c06f30c732220fce127","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-07 22:43:24.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"4bd0a36d5d1d53195e55cabc425d8200af31f2c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 23:28:28.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad72d5021842472fa156d348f4776d5c : SUCCESS in 38m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0dddaad8f364df4988836d1d531bc66 : SUCCESS in 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"beb512f78a78ca9624343549d7f8f162a977f4f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 00:53:34.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9b1118df67c448f48cdb2c16d7d03ed9 : SUCCESS in 1h 06m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/79701e34744642db8bc41135d6dc6bc3 : SUCCESS in 2h 07m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/633a4e9680ec4eeba1bad8fba4009fce : SUCCESS in 1h 43m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9293e096a1ca4f4bbfcbfe2d06055f44 : SUCCESS in 1h 52m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9b8f4a943a564a909a22ee0a759063cc : SUCCESS in 2h 02m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45c7d61349cf41dcae35f23231f8d3a4 : SUCCESS in 47m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96b5b271c8934e21a3fe5de787e1c245 : SUCCESS in 40m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc07824a0274c2695cce641d90066d3 : FAILURE in 10m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eddfd15d27384d30b37aab68e359acaf : SUCCESS in 39m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/877571e99a34475a81e92105106c6a2e : SUCCESS in 41m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/413a65d8915245a695e4c0e5a475e9ee : SUCCESS in 40m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e2d50448a7a4b37a5ccffd878efd3a9 : SUCCESS in 10m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b7be493f01c9472db0f57421ba35dfa8 : SUCCESS in 3m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca405b01c4bd4389a28e04a4a4c61e6c : SUCCESS in 56m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f3348b7c3d641fc8b7331c102933fdf : SUCCESS in 1h 20m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/290df40c0c884d14a40c1e4b2e0b17c1 : SUCCESS in 1h 58m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e66370e997f4ad1b49091d1d14a778b : SUCCESS in 55m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/60600edf5851475696370baf502d6383 : FAILURE in 1h 13m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4033887a5dc843dcab397558421c9888 : SUCCESS in 1h 46m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d2eed4dc9b924e9481b8881b9bd26006 : FAILURE in 1h 12m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71ce50589e4840c69ee8de180069a179 : SUCCESS in 1h 46m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f5ce9c87c5945af83afe1c84eb448c8 : SUCCESS in 1h 04m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2e8cb5ecae88468289c1d6f3bcdb7640 : SUCCESS in 1h 04m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/301fab1fd2c043f98e9ad2a702a105d4 : SUCCESS in 34m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f0c1192c4922486a9463cd9fc154277f : SUCCESS in 2h 06m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fc7f07ade2b04348aaebe5017991a3bb : SUCCESS in 1h 24m 38s\n\nWarning:\n  Comments left for invalid file neutron/quota/resource_registry.py","accounts_in_message":[],"_revision_number":12},{"id":"6de7a1eb78c59274d56b4cd1102935f228cc7c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-08 14:50:22.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"09f785ade0d86bf43b9ed32ae63bd6690f2fc339","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:06:24.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64360cccdb764431bf89ad1f62e8096b : FAILURE in 8m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/baf152bf2e6043399f45c17c520ed989 : FAILURE in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"481dd4deb742969725fb5e0211c24db190781625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 17:45:44.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d0245cd996b24bd7876f66bcc09bb090 : SUCCESS in 46m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b4392a41163a4d3889c9dfb7ec681331 : SUCCESS in 1h 29m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aaaeb6883bb34fc7bd7eed370d0fdd81 : SUCCESS in 1h 27m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d92344b4690e4931ac1f4418994e9fed : SUCCESS in 1h 16m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9ebd0caf2bed49b9a3cc5726a1902954 : FAILURE in 28m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0585f0699cbc43caa1d9d10eefd66892 : FAILURE in 8m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a300f795e0b6481ab353e251f5edfc48 : FAILURE in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8639ea7dea7740ba90b9e3724e327cd0 : FAILURE in 14m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b166a977f5f4c6bba94e3bfa4bf7d4a : FAILURE in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3efe227bb2474ca1b16c4751da107817 : FAILURE in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc8c48b8083d4217be8009337d0208ee : FAILURE in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8920f6882c4859b440e58419585991 : SUCCESS in 10m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9bf1abaf73994fe79e10e699d4d4fa95 : SUCCESS in 6m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/94e2aa92aa6f4ae3b730294b69112772 : SUCCESS in 53m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a29a3906409f42d3a67fb31c1cda9b13 : SUCCESS in 1h 31m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8271778ada4140ff8eeb98859bc9a031 : SUCCESS in 1h 37m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/979427276bc04317ba9f01c7f6c3875e : SUCCESS in 1h 00m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5e87052c79f646d7940da5d33f31f900 : SUCCESS in 1h 13m 11s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f9a3f0490943424eaf198bb7f6f9a60e : SUCCESS in 1h 36m 57s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d91f5156b3cc4e8d9f0f18d1609e2f41 : FAILURE in 1h 41m 51s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/45ca949c171d40c286b258ffeade1b60 : SUCCESS in 1h 51m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3de7485c75d045c4abb891754429025d : SUCCESS in 1h 07m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/68fc7f8a6ea744e380fce3c5dad80251 : FAILURE in 32m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7ad47cd40d0a4a98a06f4ca25cfbb83e : FAILURE in 29m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/780f180608fa488b99f4ddece2b4b259 : FAILURE in 30m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/579cb9e663be40ff89a2bc53a1af8c0d : FAILURE in 34m 14s\n\nWarning:\n  Comments left for invalid file neutron/notifiers/batch_notifier.py","accounts_in_message":[],"_revision_number":13},{"id":"b074f7c5d8fce6ac9732960309d22556f5a9b32a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-09 11:06:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"906428db1bbf709ad8ec141acb335266c711ac9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:04:36.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/403ef44364d54798b113d855f9c00141 : SUCCESS in 52m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"74d18e6ff434138223f864b789c3fc129991c6b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 13:18:49.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3ff282a0e7474b09bc80ea93e08fb70c : SUCCESS in 40m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/51561f88489b4fb2ad5013bcd5ef3389 : SUCCESS in 1h 48m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d94f859d6eec42f6bd043162aefedabc : SUCCESS in 1h 25m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f38da343c38c404298ddeb87cc33143a : SUCCESS in 1h 26m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5ff3a317d9184cf8b1ba29a815f82bdc : SUCCESS in 1h 39m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1b56d0638ba4cbc92d27759acf28f13 : SUCCESS in 1h 07m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0513d3d2e9014be0a5bee219a3914598 : SUCCESS in 34m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/072357f46d354ef698dbd6ca63c3073e : SUCCESS in 13m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af74d9a67bc8478b93609f92eb3d06f4 : SUCCESS in 37m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb41e5d0486d45a4a9189a01b25d3e3f : SUCCESS in 34m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05afdb1842c747ebb3daea1c80ed725f : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536397041e6043f0bbe918d6318f0764 : SUCCESS in 10m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e112cf33d39454d92c34a0ed6412eff : SUCCESS in 5m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0051bbe1de043d0b8bbde71df9e8384 : SUCCESS in 47m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/69a013e4d6434270b2dc8402ffa53773 : SUCCESS in 1h 17m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/41e499819bce41a695c196bda1ee7f60 : SUCCESS in 1h 24m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7aaaaee5f62741bd90faf2d434c11986 : SUCCESS in 59m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/44c1d3abeeb043a6ab29393fe8737c42 : SUCCESS in 56m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fa1637b7046c46f4b12cb5c386dcb30c : SUCCESS in 1h 18m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3ae4aa2ff2a845208ac5f444dd2c0ec6 : FAILURE in 1h 23m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/847aff74572e4f559958edf117860e94 : SUCCESS in 1h 33m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dede353fe964f4db71ec9f46205dcbf : SUCCESS in 1h 40m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/88f239e7c9d543299c31a3bc03569fe7 : SUCCESS in 1h 42m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fa9f63f7b41240a7a8005e1596e4667d : SUCCESS in 30m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/96adc740762749129c93ea49402cf58e : SUCCESS in 1h 35m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bfc427381de546ebbad8dc6e08c54cd4 : SUCCESS in 1h 20m 31s","accounts_in_message":[],"_revision_number":14},{"id":"6e9188a335e1e9702213a4c630b619e620bbef59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-10 14:08:13.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"ca39d2424d8873e5722f804cdd4faa4f745316e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:04:16.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c94700dd5db4c048162624ccb8da71e : SUCCESS in 48m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fae0db642a2145ecaa88bda895763bbc : SUCCESS in 49m 11s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"56da423ce73f23189c8fc88daf02f14df714006a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-10 16:44:40.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":15},{"id":"568ff750419cdf076fc1ad2dfa7cd3ab2d217042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 19:59:27.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3f364fad46f945ddb5aa595969e0445a : SUCCESS in 1h 02m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b772afe004444fb3bdb66ed674cb775d : SUCCESS in 1h 41m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/517b198181ae48a4bca3eabb181a6ba1 : SUCCESS in 1h 19m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/63d876cc97fd45d596bf54f178966ea4 : FAILURE in 19m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/92d5e7243f17454ab74ea9ff5dce86cd : SUCCESS in 1h 42m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/142984991bfc44d59f5e17982e6732b4 : SUCCESS in 44m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/900aa7e48d5b46dbbfac9164a38c9f2c : SUCCESS in 38m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/129d4593265442e8bd043224d7fe36ca : SUCCESS in 14m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af82cbd445b748ab9a2d78c9a693ba34 : SUCCESS in 38m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d1f2c3742a741be9e01d6ae306c4b8d : SUCCESS in 44m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/825c898d4d5a4824b0473904f2bd29be : SUCCESS in 37m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812d31b34be949078ab2898bb754325e : SUCCESS in 10m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d10b5925bf0841498cdcf5e3437cfdc7 : SUCCESS in 4m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/46caa02768b3452eb66e32975239f39e : SUCCESS in 57m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9502f064d6b34101a8c924c7bd3be7c1 : SUCCESS in 1h 29m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/702a2c79890c465c8015b5f7ac228afa : SUCCESS in 1h 42m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d756007c5b1e4ac5885f1661f960991b : SUCCESS in 55m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fe757636f8144c38abb73119c2161bf8 : SUCCESS in 1h 12m 09s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c0803ce93b66449795a6bfe4534efb50 : SUCCESS in 1h 31m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8ff78a706e9141e78bdf64a2cf47a9ba : SUCCESS in 1h 53m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/529c6e02246f4791b23ebdc93c7799e3 : SUCCESS in 1h 49m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf704daa0b574216a8f56b8c9df89c96 : SUCCESS in 1h 11m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/041e37c8a2fb45529c054afc269d6cd1 : SUCCESS in 1h 18m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/05069362f4d44466b9ef15b844dd7341 : SUCCESS in 35m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/20f43e1bf07c42888cd18bb2262ac8c7 : SUCCESS in 2h 03m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1eae3fe8145b4a2e95ac73231c19cbc0 : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":15},{"id":"38da3a9ade768fa5439cf67555ca5c5d8f43e3dc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-10 20:28:57.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"02aeafb9b5cd3fbd3485fc1faf50cb66e6eb964a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 01:59:44.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"2b9986f0c14b6523a4562a864eec1716e2ac37cb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 07:11:06.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"c30d2a995cf8f1b0b9d09c6bd4801f18fb2d3e7d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-11 08:20:59.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"8a4ea2659c6d706c97e2e595fa78ddb3d098670a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 09:18:35.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b69609b965d450b89f7a6cab2ee2117 : SUCCESS in 49m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/251c5a7dab0746548068589fa7dad3e9 : SUCCESS in 48m 09s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c7d74afa014b4ede143dd55db3f75a46923e5b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 10:19:16.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0a89d98f42f342b784acf7272c2b359a : SUCCESS in 49m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/86d8ada7fa104824b73c7a8ea4925877 : SUCCESS in 1h 45m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d1ad9232c3a9496a9267d2394de0e1e4 : SUCCESS in 1h 24m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/57792dd2e74940b5bfb7c5d3eaeff619 : SUCCESS in 1h 36m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ae1b770b576741dea52dfb55b7662a3e : SUCCESS in 1h 55m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61222be6be034429b66d6b47d2e52344 : SUCCESS in 43m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29fa7252ab964029a29053353b8d9ff4 : SUCCESS in 29m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c5e857de6347168c3f6d15aa7bdbf6 : SUCCESS in 11m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dc2f5a0899a43df9c652e2f5b201ce9 : SUCCESS in 30m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31fccb3c73fe4467a8af6432ce7a4114 : SUCCESS in 32m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b83dbe3c8454aa9b7409195a5b31996 : SUCCESS in 32m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7eb5a64b56e4d6a9208e729ce1bce2c : SUCCESS in 10m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e2ea4066cffd43638a6731cb47c1dacf : SUCCESS in 3m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48de7bb4f0964b9e814fc2036e17cfad : SUCCESS in 47m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d913f90be70943ddb79b4fab6c6f021f : SUCCESS in 1h 15m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0e04b221e8ff4e1b8829ba497a44c455 : SUCCESS in 1h 45m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b359e6a3dc7485b9633e58c40590653 : SUCCESS in 54m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4b414518525e470e8009a9ca2d293d57 : SUCCESS in 1h 04m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/82ed6ce0be794287a6f25685b64bf76a : SUCCESS in 1h 26m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2511e9093a1e4cb48c8ca32a454d0460 : SUCCESS in 1h 35m 22s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed20fb4760264d06aa58977040883ede : SUCCESS in 1h 28m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1df4b80f42242db9ed93828e52e3ca2 : SUCCESS in 1h 00m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/20b4917666a145f59a6080651d6f5901 : SUCCESS in 1h 12m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ed87dc8daf014406aa3420f2f135aeb3 : SUCCESS in 31m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a75d1d87ec7e44d6a5b07147e0dedfae : SUCCESS in 1h 42m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c7fa959fddef496ab9346a35e15b54c7 : SUCCESS in 1h 23m 26s","accounts_in_message":[],"_revision_number":15},{"id":"cf7aa43c5e82689d27af7652b0bc3ee63d599acc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 12:33:04.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"fefbbb6ac37bd6d267f47f7cc436141a72537f58","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 05:19:15.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"1e8c9d0e2c72f13677813ddf7d168dd78e09ed5a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-12 15:32:12.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"fedef136d9e3533e4f7f3d12f0302564ebf4bcc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-20 15:39:56.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"cf27af3d2fc225ffab1c696ce12023490fc56bc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 16:30:21.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bda1fea33d4a4256ba9ae2c9bcd45f4b : SUCCESS in 44m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41d277d8b84b46bc85f257d8b6841eb6 : SUCCESS in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"291a0f45b0adf82e052be7a60ad37e15a3cab2cd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-20 17:14:23.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"84baf2fe089fb4fa88e3bdf507ce99123fded272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 17:25:23.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a2afcd673cb3497d9886d77e4242fbe8 : SUCCESS in 54m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6a30881892ad4511bff9d758daa5be56 : SUCCESS in 1h 34m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c1a9161bf2cb4fb6a5c1771faef1cc60 : SUCCESS in 1h 38m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5cb074c9f3c64f71990cbe4317549cec : SUCCESS in 1h 25m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d9a3fe137d9f4429985be9323656565d : FAILURE in 25m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4ec0020718b4341a2a599065f0adc2c : TIMED_OUT in 1h 21m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53219757455449e4bda69ceb093be9aa : SUCCESS in 37m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f417aafb33413282e0bd2c7bb66208 : SUCCESS in 12m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a8ebb6b5067428dac705c725ecf8aaa : SUCCESS in 41m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6e6de02118a483cb30996ff693a7137 : SUCCESS in 35m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba230d8283f94a4283e19123d07546f6 : SUCCESS in 26m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a25cbaf71f04bc79b4b1e9fbe95e2a1 : SUCCESS in 11m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a5284eaf68f49e1a4cbdc7602da5286 : SUCCESS in 59m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9cbfdc1b7644f3fa10889a06d8fbe8a : SUCCESS in 1h 25m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f23e2201667e4ff0b072e78fcd037277 : SUCCESS in 1h 29m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8faaf5b3e851482fb015d712044464ba : SUCCESS in 1h 07m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b8e5000125154111a327b65a882dbcd8 : SUCCESS in 1h 06m 01s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9aadb0b881244885a1b9bb0f580b40fe : SUCCESS in 1h 25m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3e9d5d5dfe5a40929809f210cf3eab33 : FAILURE in 1h 20m 48s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d33e32c448b749f8a3909c0f441c0a75 : SUCCESS in 1h 19m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6f8223e3abb4d8c8ca69fbf3bd31b4f : SUCCESS in 1h 02m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/69a9a11cde6841cebc7f04a4c6444267 : SUCCESS in 1h 10m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7beb1854ba2e426189a4c15000f716f0 : SUCCESS in 34m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ab37aae66684486794ea2fde24c929db : SUCCESS in 1h 34m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b514a6facd8b4f0a992c78fabcc36b76 : SUCCESS in 1h 22m 52s","accounts_in_message":[],"_revision_number":16},{"id":"cb0096a5292fcdc85378fc25c75885fc7337197a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-22 10:49:59.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"e3477b8425490c651f2cb8d8c4ff4a0994fcc9b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 12:09:20.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ef0c879564a462ea1e14d032ef689e86055c7995","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 13:08:09.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8865441174a14cbd8e041fe0b351ffb5 : SUCCESS in 52m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ab52caec31084cdb9314b9705b552ca5 : SUCCESS in 1h 11m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/172d21702d434b88b646c9155d5beab0 : SUCCESS in 1h 50m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1d5ef9089e9444e890f578b54d541bc8 : SUCCESS in 1h 41m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5d594ec432fa4c9daf2fcd7830a94bd8 : SUCCESS in 1h 59m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6ae396609034be4a6535bafbb394e08 : SUCCESS in 48m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c2c5e765db046c8bbbbb39c9d92f0e0 : SUCCESS in 40m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af0549420c234ab29a390c4a1c663cdf : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45a5304be05b48c99cc2d2e1632a6ebe : SUCCESS in 38m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0fe69a9f3cd45478bec423874d1fdcf : SUCCESS in 39m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42132b239861469a8db66a2242fa0811 : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d745e0fbec54d1fb18df784e129c36b : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f6a412e61df146f08fc23914d09dd05c : SUCCESS in 46m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d89abd179e349d0ba6ee6b7351742ac : SUCCESS in 1h 27m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/515bd776c44b40fe9101eb6fb2518ea4 : SUCCESS in 1h 25m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a093a8a295dc44619490242a569190b1 : SUCCESS in 1h 07m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/911a8068d5c34a348ab33721bd087dfb : SUCCESS in 1h 05m 17s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ccd66521ab594bb98381597eec61f106 : SUCCESS in 1h 21m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/70d028da1011492d9947fda63540694e : SUCCESS in 1h 39m 59s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa3c0b97c97543cc85243d5413e3a125 : SUCCESS in 1h 39m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7a12f762d4f467b8c11e65331432fe7 : SUCCESS in 1h 05m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/150785f7468345b9b612ef9b0a2949dd : SUCCESS in 1h 11m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ca2146773591497e8ad23b1b43f6f5b4 : SUCCESS in 35m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/251838eb7f0449648c0e9450efe3e7be : SUCCESS in 1h 26m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ac5819084ad149549244ce97001a9b0f : SUCCESS in 1h 21m 36s","accounts_in_message":[],"_revision_number":16},{"id":"3c5b9a83d48733b3b4e27af30fef36fc5706af79","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-29 10:02:28.000000000","message":"Patch Set 16: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"5613b55126b473534b0d215677e10513785627fd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-29 15:37:55.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"7c3bc188f20e2c4b213e957edbc3fc3c58410e85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-29 15:38:26.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"91d274964f4e0b0fa8cfbbf11774a1c197d3ba43","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-29 18:08:47.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/01/776701/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"ac0e5008326b60a3a2a82f7b17435b3eb94ea06e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-29 19:24:47.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":17},{"id":"18684c388b2097742d465051e2b05a1014b710f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:31:57.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/16ed6ed0f1bc45b5a561da282b5af7d3 : SUCCESS in 54m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c4963055a1fd460eb6c40c90bd75b6af : SUCCESS in 1h 38m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d911d5b1241b48baa48602c8e4e21270 : SUCCESS in 1h 48m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9cb900de155647249f9b3be90779117f : SUCCESS in 1h 28m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4ae32450706546e6aad41851f9fcff56 : SUCCESS in 2h 23m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f48b32bada594f8daf7fc7708a908b3e : SUCCESS in 46m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86c627512210427bb19427fd640d0eb9 : SUCCESS in 34m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7237e05009d0432c84d0248b01bb6a84 : SUCCESS in 17m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f2bf7c468ce47da95123d3c1275960a : SUCCESS in 32m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffeed19fe2164ae7a7021f08fbd0eaff : SUCCESS in 27m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5600066c8b7045878a073c80265261ce : TIMED_OUT in 41m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52090383973e4de385712abc8cdb893d : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdad129f457d4d8e87d7802aa24d2026 : SUCCESS in 52m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/27786e3d2d05419e89c9dae034c08070 : SUCCESS in 1h 23m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b531aa33ba2545c7812334108ae6e69b : SUCCESS in 1h 24m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/49d0cdd15cf84d28b4f54ff389583c36 : SUCCESS in 1h 04m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e1c1f742b7bd4c679c7adaf9132f6e82 : SUCCESS in 1h 07m 13s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8406db09ca40428192dcaad0a4176f8f : SUCCESS in 1h 10m 32s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/568c75a9437c4178894e9db4acb40924 : SUCCESS in 1h 17m 48s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6cdebde963444170b5fa0c9bfd26033c : SUCCESS in 1h 50m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a32e923b5390429491f3b5eef47f2ce2 : SUCCESS in 1h 16m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7053c9ca5ec24a6aa10f7b17bd230c2a : SUCCESS in 1h 06m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/45264ea5ae1248619abb09e8bb9ba61f : SUCCESS in 33m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7e75777982e644d98fdeff925eb61df0 : SUCCESS in 1h 26m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/182f41add0984cbe9e2c2bbfd78b1759 : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":17},{"id":"89e12f3848652313e121e47e4137f7adbb9a4a1f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-30 12:51:46.000000000","message":"Patch Set 17: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"f76b208012540685458364bdc2457e67a88d43c1","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-03-31 08:29:30.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"a1313dde96907ad1247e5615f1d8cd3407afb2b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-15 11:16:16.000000000","message":"Patch Set 17:\n\nHi folks, can you take a look at this patch? Thank you in advance!","accounts_in_message":[],"_revision_number":17},{"id":"e63cca698275be18bdb03e27c1332fe64ec8fc04","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-22 15:24:20.000000000","message":"Patch Set 17:\n\nHi folks, do you have some minutes to review this patch?\n\nThanks a lot!","accounts_in_message":[],"_revision_number":17},{"id":"2aef25caf5764bd6d7bcfaac5287eded31d5f2e3","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-04-23 10:52:54.000000000","message":"Patch Set 17: Code-Review-1\n\n(4 comments)\n\nThanks Rodolfo,\n\nSorry for the delay in reviewing it, I think we need some clarification regarding the format of the values in the ovn-cms-options as we are parsing it in other places which may not be compatible with the value assumed in this code.","accounts_in_message":[],"_revision_number":17},{"id":"9b99f116961964a1f53aa54abb2854c41cc07ef0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-23 14:11:28.000000000","message":"Patch Set 17: Workflow-1","accounts_in_message":[],"_revision_number":17},{"id":"e84bebaa104f5f9cd5807e5f4233e4f9def76820","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-17 15:57:51.000000000","message":"Patch Set 17:\n\n(4 comments)\n\nThanks for your feedback and sorry for","accounts_in_message":[],"_revision_number":17},{"id":"f33a93b0df609a1e36e89ae9d1c4caebda9d5f10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-17 16:09:25.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"8fe5d38e0880e307e8d7e102fd00726cd11f86ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 16:55:57.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c93e4f556ca54924baea383262fbcc76 : SUCCESS in 43m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/237a6ef4176049ee8c7330d62f21f2ec : SUCCESS in 40m 18s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"de9112778a4f2591f7989815c28051a09e26a598","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-17 17:45:09.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/18 : SUCCESS in 1h 34m 39s","accounts_in_message":[],"_revision_number":18},{"id":"d4091b22d905e4473b2d554c8cfc6e2f41bb7223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 18:48:39.000000000","message":"Patch Set 18: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/37804538e6e541be98cea5b9a8fa60a8 : SUCCESS in 41m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b7e1ec9b8b414b0099ea6224758c0299 : SUCCESS in 2h 34m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f352aee1efa24b14812e9d11464e1274 : SUCCESS in 1h 55m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ab628ac706d46c59c75074c854fb6fb : SUCCESS in 1h 47m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/39be9e8d6e274291ad1bdce78d3d040d : SUCCESS in 2h 21m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffebaa2ccd21456a8610e208e9e2a137 : SUCCESS in 40m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cf2e2e076a14c3d86db49fba8db2ac2 : SUCCESS in 38m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd04b6682d04bdeba2298944236b1de : FAILURE in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e318e05584464c48ae4fb08cb6653941 : SUCCESS in 38m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2584140112074a5184f4c7e95f09893e : SUCCESS in 32m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b922e8a3bdc240dd9bffebf855e0be0e : SUCCESS in 34m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a56d9ff2043347ada83857c78461335b : SUCCESS in 8m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbc6a2d4ffb64dbb8e0ae0e38a50b8ef : SUCCESS in 58m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e0b147ac92249d6be5f68966be84522 : SUCCESS in 1h 14m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/854c15e88d644b03b5ba5f19a870d364 : SUCCESS in 1h 28m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5fd4b57fca864d1ca64fba81cea254ef : SUCCESS in 1h 06m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5bf3c353b7bd404ea430503159790986 : SUCCESS in 1h 09m 13s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8f83022ab6ce4b5ab41e996fc04cd406 : SUCCESS in 1h 01m 05s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7c384f8a0e50433193700b8679713687 : SUCCESS in 1h 22m 13s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/59671e41b053434a8e543b6fcefad926 : SUCCESS in 1h 17m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55a1fbc6c21d46d98c4de1baad6ddd61 : SUCCESS in 53m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a6a12e7eb1034d33b610aaf66f2b6222 : SUCCESS in 57m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/72fe162909694c2e980f5b16951aa8af : SUCCESS in 34m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9f77fff0dffd41cbb3a2930a861a683d : SUCCESS in 1h 14m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7e1615ff700e49209e3c59294397e629 : FAILURE in 29m 32s","accounts_in_message":[],"_revision_number":18},{"id":"f9a6d8848c0327f659ed6c7e05a3293b769ae86e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 08:40:42.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"7401b1d2ae0899e46eb93f8f4c3335a22fa9efb2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 09:24:25.000000000","message":"Patch Set 18:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/01/776701/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"eb3ed54613d7bf057d1791eec916657679e82e2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 09:26:09.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9948887108414835b8287488c10cf615 : SUCCESS in 43m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4cf3968caf14fa38b5f896bc07b981e : SUCCESS in 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"1b3600b364c649ab024daa1411fc409912a8a2a8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 09:29:59.000000000","message":"Patch Set 19:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/01/776701/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"e73e0fe878ffd00e7764f4f6ccb0a21e6c8b87f2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-18 10:16:57.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/19 : SUCCESS in 1h 35m 20s","accounts_in_message":[],"_revision_number":19},{"id":"f1c854eaeb64f4f71b75657abd67d5912a2ea8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 10:46:06.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aac6df5636bc4e28bac1b374964e1973 : SUCCESS in 40m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7e8a21612655446fba4ffc6ce79d5e19 : SUCCESS in 1h 27m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a0e9e005610f47fabb14e31f4ada158e : SUCCESS in 1h 32m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/147b30e1523a4514bfa4f0a2d8e2d163 : SUCCESS in 1h 09m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1d45c6779f854c6ca53be1036ab4f522 : SUCCESS in 2h 03m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa6d3ee9bd084a3cb0c2ca79ab167427 : SUCCESS in 25m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7179716991d84f35a30caea84d510e9b : SUCCESS in 26m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f8641bb2cc04e938ba96daf0e45701c : SUCCESS in 10m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21141d523c594c6993ea3654b0995ab2 : SUCCESS in 26m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3767f9772981433480ab3d9b1b92edde : SUCCESS in 18m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f42cd4c227e24e17b97dc187012072cf : SUCCESS in 24m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/589e0f1efdc64f45a27f57c056b99c1f : SUCCESS in 8m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3920d7481d8d4b0c8dbd400771204731 : RETRY_LIMIT in 5m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa00802dcfcd499486957b83dc9f3e97 : RETRY_LIMIT in 4m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/773fb1dcd11747299f3651cd02667d3b : SUCCESS in 1h 22m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa16c95c45d84f958e79183b98dd7d6d : SUCCESS in 50m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/076e6021c14f4607834b6bdf64c2fbbb : SUCCESS in 57m 29s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e0a3f9f2f1e44208bb4d61912d484ef6 : SUCCESS in 1h 01m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7ffdd6be0d5c45958b09d9cd138333d5 : FAILURE in 1h 22m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54c11479299b4f96995f8e5bcf450333 : FAILURE in 1h 34m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c85556123f1c4cf5b0dd814c1477aba8 : SUCCESS in 44m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2508dd3d66394639a4ff7264f67122cb : SUCCESS in 52m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3bc9f8dfd1eb4793994b112e9618b383 : SUCCESS in 26m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9654c478018d49f0ac33e72da6dd731c : SUCCESS in 1h 00m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c369711df7774de6854db24a32ccbadc : FAILURE in 25m 36s","accounts_in_message":[],"_revision_number":19},{"id":"cc9d6d833e85924dc9a8aba541c4f7cb08e094cf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 11:12:05.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"0eb9c49159d805f33743be337b32383582446c08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 11:52:07.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b56758bbbe3f4e5f866c9205fcdc3d79 : SUCCESS in 35m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59a8c591966e4ca695fc9108092e0901 : SUCCESS in 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"7a385f8d2ec86dafcd6df73354bb6acb1d916f0a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-18 12:46:36.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/19 : SUCCESS in 1h 33m 52s","accounts_in_message":[],"_revision_number":19},{"id":"8ed90fce3349a8551e1b98f01444afebcbf9090f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 14:24:53.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b6778ad79cd04cd194b695f8a73f913b : SUCCESS in 46m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0ba664817eb844ef9e2b58ff642f6b90 : SUCCESS in 1h 37m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fe351519a1584a15b6e9ba3b3b69c552 : SUCCESS in 1h 48m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ba61d47ce59b49a5895851f5ab85862a : SUCCESS in 1h 16m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a2492cb5890e48dc8913ed35b9749595 : SUCCESS in 2h 15m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c98a7eebe114f98985e6d4a99c66fb3 : SUCCESS in 41m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93aba13159b04163b49779c0167cfc5c : SUCCESS in 31m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52ea838c68564d1d86d445f023ce16c1 : SUCCESS in 12m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5540a1de4e2a4b749cd3adb07190f88f : SUCCESS in 24m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b73adaae8184564b5c47b06a95ad62a : SUCCESS in 21m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e78393a7859c4b0b9b8e6f7b27abca43 : SUCCESS in 32m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e2f24b1efec4992b180d49c22674731 : SUCCESS in 11m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/52a754cd21ee418fa3b194e3330ba918 : RETRY_LIMIT in 4m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/67a9974306d84cc1a47014ab2f9dff25 : RETRY_LIMIT in 7m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7d207c0d6eaf44068086ab34690c796d : SUCCESS in 1h 25m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee853e2a9bb446758d062cfe72a37495 : SUCCESS in 51m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2d20029237ce47d9b02ab0d8901f6b26 : SUCCESS in 55m 55s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1ed78d73c7224158896b760b433b6748 : SUCCESS in 1h 02m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dc1f4800c1204fda8cdb9da047397bba : SUCCESS in 1h 19m 42s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d3906311b5341e1b8ffd273587c6ca5 : SUCCESS in 1h 39m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1390ca1ee1cd4d729ab7fc353119240b : SUCCESS in 48m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f5548e37a535459a9a55e21703b06c92 : SUCCESS in 1h 03m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7069aff524564a52822f7f69740051af : SUCCESS in 25m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6414fb427fb942e6abd91ba66dfaee46 : SUCCESS in 1h 22m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/abc7a9309bad4b97960bfa1886b25d1f : FAILURE in 23m 21s","accounts_in_message":[],"_revision_number":19},{"id":"bf731a75d86fb71706d3bdf6529ac313e1663fd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-24 08:36:11.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"953d94c7ae239dabf7b005f5cd6a903e76706b22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 09:31:06.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/84700f17a9694c9fa8865bf93c60f725 : SUCCESS in 53m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a07ceb176be84a80aaba76a580fc69ca : SUCCESS in 39m 43s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"dc81f450f27b16db436510c3577bf9578264b886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 10:39:14.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ee1f73f98094405c9a385fad4cca0df2 : SUCCESS in 45m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d11846123669409abe77b606b102e5fb : SUCCESS in 2h 01m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/56ea90a01642474a802abb9f56c6ee97 : SUCCESS in 1h 47m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2926e987eebb43e9b5fd3de06aa8f5cc : SUCCESS in 1h 14m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0270f536788744d097fc3f0ea770bfc9 : SUCCESS in 1h 38m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f6d9d97bc4645f8aa2dd4342bf000c6 : SUCCESS in 34m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1741a489a83e4add99b70b879fab528f : SUCCESS in 27m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/147d5eedac544b17a32df4febd452b78 : SUCCESS in 12m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55f3dfce3bf94acf86952bd3085a4221 : SUCCESS in 27m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4a1108e7f7c43118c4dcf0a435089dc : SUCCESS in 24m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcd3e27b730b4fa8bf3460c050daeb8a : SUCCESS in 23m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a1c099c718481f92226e93eb528967 : SUCCESS in 9m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9e62051098c4902bb382e2ade1b40a3 : SUCCESS in 45m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7850f56804284b2ea3e8003d169953e7 : SUCCESS in 1h 23m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/520288c193144808912c2adf674c5666 : SUCCESS in 1h 12m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c9342b7e4874bb194e1e52c004e945d : SUCCESS in 57m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/921754e877df4c909b8ee02dabf657bb : SUCCESS in 59m 58s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d5c7aaedfe8b422b8a4084db7eee03e5 : SUCCESS in 1h 11m 10s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/923badcee6ce471c837dd59814eafde6 : FAILURE in 1h 05m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c70a5065e1c748c98b3dea446fd692c5 : SUCCESS in 1h 30m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d5d9f6f70fe4501bee12b270e5e80e9 : SUCCESS in 52m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e10f1e960fb84f808dc5e485ee798ed9 : SUCCESS in 1h 01m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dc20d622276e45a3a45bdc3cb8524937 : SUCCESS in 30m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2a62356da2cb42f7a82f3ddcba85a168 : SUCCESS in 1h 21m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/386aad47487e4be6bc99ce46e9d5304d : FAILURE in 30m 40s","accounts_in_message":[],"_revision_number":20},{"id":"70e21cf4e30401452d184290c8781b2d748ae621","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-24 10:46:33.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/20 : SUCCESS in 2h 09m 09s","accounts_in_message":[],"_revision_number":20},{"id":"ebf71a4e3069237624c7085688e1894116877029","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-24 10:53:08.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"11e64e86fb3f71229ed9c39814878afc9bb80ad2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 11:34:31.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7629877341ea4333aac9cb98b7515594 : SUCCESS in 35m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca1b3fd857c9441c8ef6412125eef295 : SUCCESS in 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"907db433964528db3907f6576ff0f6f9bb1afa76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 13:03:41.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0946262a348645839c3480cbeafdaaff : SUCCESS in 50m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/87866d7dc62a4ae4aa8ce885cf75a81b : SUCCESS in 2h 03m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dbdd56a185c4418c95bd4f585dfda0ad : SUCCESS in 1h 54m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/58642521f9ed494b9775064bbe1fecdf : SUCCESS in 1h 32m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a678262a300f458d85ede017bc773db6 : SUCCESS in 2h 04m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/527ffea861f344da8d9bd016ef570d46 : SUCCESS in 42m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47d7234497a645cba559fdd767e84365 : SUCCESS in 32m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b04e35f491444c5db9990eb12c243330 : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3ce9562debb406fb0ebf0d583051c89 : SUCCESS in 26m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e63c678789694cd7a33aa98689521b78 : SUCCESS in 30m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db8c68ea06864cfb9e24a558ea302e22 : TIMED_OUT in 41m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47eab7786b464ed2a49d754dd333960c : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bafc51e315fc4374bb064e376a191a3d : SUCCESS in 43m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b8ca06781c945889c3fd4b77562f942 : SUCCESS in 1h 31m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/464366cc3a4d471489aa48d18d362e2c : SUCCESS in 1h 04m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5710570e4e0a4f5bb4dcdfeba455a8e9 : SUCCESS in 1h 09m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/77d2129cbc7d47b5b210e463b489ff61 : SUCCESS in 54m 40s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2880e1edb45d49fb83436ebf9a915334 : SUCCESS in 1h 24m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ea1f42b887fe46b985729ee4f7f2f0f2 : SUCCESS in 1h 34m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2fe9b7c6e5df414c90faca31b4bdc775 : SUCCESS in 1h 27m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efbc099b585948c18d6a1bf77e41d48b : SUCCESS in 54m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4fe2b0cec9b046ab9a63a7dc03b10f94 : SUCCESS in 59m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d5a0b15dcf25414395cbc025051a1200 : SUCCESS in 28m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8565093c324347e5a441a46656cbe6f4 : SUCCESS in 1h 21m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dbf3a88d5c5c44328feffa393a1b5a6e : FAILURE in 19m 52s","accounts_in_message":[],"_revision_number":20},{"id":"55caf1224f023e26ca45aa0b8e754563aa0c0dba","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-24 13:21:11.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/20 : SUCCESS in 2h 26m 29s","accounts_in_message":[],"_revision_number":20},{"id":"7732c3a8a7803365c3cda01ba7fd663ce738873e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-28 19:45:28.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/01/776701/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"156c1aefcd55e6576385f499a3008fc6fb746b10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-03 15:55:38.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"061f248295f73ad411a79d0cdb671b82662a5895","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 17:07:25.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32d05582c9884e3a88e4ffa8dc11b6a9 : SUCCESS in 55m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ec2288228e44ea19ac59ad893e7dc5b : SUCCESS in 59m 13s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"74d225bf95ec8fecd429d34e1a5131dab882d5eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 18:53:09.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/73e60c45830f4f2e925885781fb0c970 : SUCCESS in 41m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/719d235b14ac47249e74b2ec87e6e7df : SUCCESS in 1h 58m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/87675d4895c147e5a28343363b94f3c2 : SUCCESS in 1h 52m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7b0a121f3e8b401e9bb5e75113187026 : SUCCESS in 1h 36m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/34ee3da99f86494e93703966de2fdeb0 : SUCCESS in 2h 22m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b5e0af642134ba19e78de7147ad3217 : SUCCESS in 32m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6bd39cdce354c73b5851edcc660dd65 : SUCCESS in 32m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e58daf8d6eb499f85251e318fa16a7c : FAILURE in 16m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7798faf226ae48479d59b57520be2be3 : SUCCESS in 25m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/785b56ded9074df0a06b4dfe61ba21a7 : SUCCESS in 24m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e40531596faf4032ac664fbd3b69592c : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eebe680394934436acbe73ab5cf3ff73 : SUCCESS in 8m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6b149cf99484968a83e7be0151dc9c5 : SUCCESS in 45m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5b60b15eca04c8f8bc9e835e484d4dd : FAILURE in 1h 12m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8a1e3d69edf3471b8e0dcdcda03f7485 : SUCCESS in 1h 19m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/16b8f0e0473541b7829f594300e1a119 : SUCCESS in 1h 06m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f89c242d95334b20bb26bccb38eb1255 : SUCCESS in 1h 14m 11s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6a42f2e0d030444ba0a326ca359ddc37 : FAILURE in 25m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3b1e53eb5af84560bbc98c3816bddad0 : FAILURE in 30m 21s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e075a3d17fcd42efb97c47b7a713246f : FAILURE in 26m 10s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1288541e3c5546caa0826dd03804ac4e : SUCCESS in 1h 03m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/05e82c43bb804407beaaae58b73abae6 : SUCCESS in 1h 05m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/07b59649141045819464cabf8b37008c : SUCCESS in 34m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/837be02f2af546789e33e8b2cd652a97 : SUCCESS in 1h 31m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f3be0e441c524235834a3e8b09de435e : FAILURE in 30m 05s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":21},{"id":"dac5facb6c8309d4f14bdbdb7f43d7055d67a4f7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-03 19:17:53.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/21 : SUCCESS in 2h 04m 58s","accounts_in_message":[],"_revision_number":21},{"id":"0581625c1ccb09e93c629b4a238c97dfd5e7a232","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-04 23:08:19.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"ed00b456bee7513e2105e2c3f078a70f834e48ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:10:19.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b50eb831b4b240f6bfa462da5fd3f459 : SUCCESS in 43m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20c234ed389d4ad89f3be0c227e720d5 : SUCCESS in 28m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cdbabb06e4843caad72ad7d1f1e5691 : SUCCESS in 10m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ba89cdd57824daba1566e0769564de9 : SUCCESS in 29m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bcad04e0da849d9b18f1adba57b5734 : SUCCESS in 24m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e54cb1fc82324b86a10b41d87df1f9c7 : SUCCESS in 24m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c4e5b93c640430297364e2e32370fda : SUCCESS in 7m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/22fd67ccca0449eda43ccd4cdb884489 : SUCCESS in 57m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4deb0b94b2914ca0945aac45ffc29666 : SUCCESS in 59m 25s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"15694c8af2f99e100f33891411c9723f36b63657","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:33:00.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa795c22244b42a6a2399535e655f5a0 : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff89a05ae77741c8b7928b1fe8e76fc0 : SUCCESS in 40m 52s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"de11330a5ba7d4c51a78cfac349c3f8c5f6eba69","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-05 02:46:09.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/21 : SUCCESS in 1h 50m 44s","accounts_in_message":[],"_revision_number":21},{"id":"d95555c4dd6f8925d4383d60eb8c05ed3e26920b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:22:37.000000000","message":"Patch Set 21:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"c41cad390e6da1e5ca75141ca8645e098fc0c4c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-11 09:16:42.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"006232888d8d84f286cdfa69c3a8b7e7949d4b6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 10:02:32.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53717962abb743cfa4685a0318167c39 : SUCCESS in 35m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78d75977bce74ce7a0eedf48fa9a0263 : SUCCESS in 43m 14s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"8dd4b376cc7f372678c7e132defd0f10817aad6c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-11 11:03:06.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/22 : SUCCESS in 1h 44m 42s","accounts_in_message":[],"_revision_number":22},{"id":"a665ef0afa3711505e7a720c4ae7ca1266a7d0c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 11:40:48.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/70df8581ede446f2bd0c4030714c5079 : SUCCESS in 41m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e3c6e7c3135d430eb9b47d1775f72654 : SUCCESS in 2h 17m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/47986d10315345d6a4847cfae31890c6 : SUCCESS in 1h 57m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e3076e2ff97d49e28b72499d73dec4c2 : SUCCESS in 1h 26m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0d69bdfa92394ab9a833c90f4b966fa3 : SUCCESS in 1h 29m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0285544d66e74a30bfd379ae15bd984c : SUCCESS in 35m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0439a629c2fe4260bc9ef81fc3a58067 : SUCCESS in 31m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2189b5f9fab74a09a85b093cec419546 : SUCCESS in 14m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc5e89c822204d768e5863813fa8629b : SUCCESS in 29m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14fcca984a764c0c91b3572b65749dd2 : SUCCESS in 26m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a2dcd87f586446fa0f0b9af6b464e87 : SUCCESS in 23m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/637d0ef427794ecd8b58bf63314ebe98 : SUCCESS in 9m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f07d8fb09f0048779d0e964344248f00 : SUCCESS in 57m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f628c02c1e674009affaeea08872cc34 : SUCCESS in 1h 24m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5ee767cb3ce74742a4f5a85a7f74094d : SUCCESS in 1h 22m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/943bdc914d6b42128c66756900514ae8 : SUCCESS in 1h 05m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f643f4f5cfe743f1bfcd9061a8350da8 : SUCCESS in 1h 05m 42s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bf1421a9e20b4b119479a5197422bcad : SUCCESS in 1h 13m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/821bd8fc35fa43fca4a9bd11bc4e54ac : SUCCESS in 1h 22m 32s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ab965762b304fc686c1a1d9764263e0 : SUCCESS in 1h 30m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ee40bb7bf344fce85823815c8a699fd : SUCCESS in 59m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4ae2b154465d4081915511126f3bfea5 : SUCCESS in 1h 00m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8e5cb93572e04027b77c74d95169e3ff : SUCCESS in 28m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/75f46800924d437faf94ea7aea3995b1 : SUCCESS in 1h 30m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bf6c5a267f104c3480569cadfbf960d3 : FAILURE in 30m 02s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"7d56d561dd2c4f292cb22cdea3f2ec8ae250ac8a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-22 09:27:43.000000000","message":"Patch Set 22:\n\nping fellow reviewers, if you have some time to check this patch","accounts_in_message":[],"_revision_number":22},{"id":"6fc99600fff07e320ab6c3b0a7ec416e7fff249e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-23 11:34:54.000000000","message":"Patch Set 22:\n\nping fellow reviewers","accounts_in_message":[],"_revision_number":22},{"id":"e98f43b7f5271cec721a0a4b3d0891d39f53cbf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-24 08:16:42.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"809b0e9d77028b3a78c46e7e75166d66a57a4b61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 09:00:01.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0fb2b1d9422430c94c4a768fbd3d6c4 : SUCCESS in 35m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e33c824b2df342148f9dce8e80792df1 : SUCCESS in 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"92023403f097e349a8c01be6f8833e8a8f3e9f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 10:17:24.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/606d88c609c94bf797aa4afb0e267f78 : SUCCESS in 41m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c38977315bd44235aafe7d0e4541d7ba : SUCCESS in 1h 34m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1751676a94c74bdaa32f33a4f6e23a0a : SUCCESS in 1h 50m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36f2796c21cb4a1fb6647b4f38adc5f2 : SUCCESS in 1h 27m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0ae329bf1e0d49ae8bede7d19b52693a : SUCCESS in 1h 54m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85ebcbae652b4b03adea585de9e9643d : SUCCESS in 31m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30d20e015a7544b988202828c37f1306 : SUCCESS in 36m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/574224cce0fa423aad52d129e57baeae : SUCCESS in 13m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83c44063389f4deb8af36f24bd867b56 : SUCCESS in 23m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff2d6be05be74f9b8a2218a73ea239f5 : SUCCESS in 22m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a32a8202844457e92bd772dc7b81c8e : SUCCESS in 28m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/545c733f62f04fc5b01ee4337b020e2c : SUCCESS in 11m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/87a57883ceb34836b300f30f838494c6 : FAILURE in 52m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/394011a4801f46038f18ec9d00dceee3 : SUCCESS in 1h 28m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/94f441e1dc1e4d0ab3c25d3ebdd9b58b : SUCCESS in 1h 10m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c350e976b464396a2e7bb4d7898262c : SUCCESS in 1h 02m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b67fc7dcf3f94ba38bbb84baf9e157b8 : SUCCESS in 1h 06m 39s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8ffb7bc72aee4c5f8e1867078b69e8bf : SUCCESS in 1h 15m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9084ea380f904a75acd14b155790a0b3 : FAILURE in 1h 04m 55s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/69821a2eb5ea4225ad591456054b4946 : SUCCESS in 1h 25m 46s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd9303070bdf443090738bfb9a32a919 : SUCCESS in 1h 04m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f72a0bd4719d4744ae537ccb7c292f6a : SUCCESS in 1h 05m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/64d334f0477e445b80fb384980e37286 : SUCCESS in 31m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/215ebef6b4d5478ead60b299a497dbd2 : SUCCESS in 1h 09m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/95f5c73b053a4d0986ffd3cda008f7ba : FAILURE in 28m 12s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"93b8e4e8797c120368cff3a5f3e40c7bd0420b50","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-24 10:25:04.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"5c04334f4332d2af21422fafa0831e35228890cc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-24 10:35:51.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/23 : SUCCESS in 1h 56m 50s","accounts_in_message":[],"_revision_number":23},{"id":"c7836e258ed96461813f64b9fd641762ffd08e44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 11:07:55.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a673e281d075457f8a99a6bdfe7529b2 : SUCCESS in 39m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1566550227874635802a4c3b22d2a251 : SUCCESS in 38m 19s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"ce1ae3bf123de5a34d4f22cd2af43efe3112e998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 12:50:08.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c2480fbaaafc475d971a431471ee579f : SUCCESS in 40m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/68a91334b1f1421691d13518890d5a8a : SUCCESS in 2h 02m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c274a09c731c4a8d93e34760dae6601f : SUCCESS in 1h 45m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bbaa7f4bfe4b4a5c837a65b34c5fdbab : SUCCESS in 1h 32m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c523e18d2ad7488aa1d0b465e378da00 : SUCCESS in 2h 15m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f387beed588e46c7bc2bef6652f8368a : SUCCESS in 39m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/677972efe30f4ed4ba8195cf8c9d227c : SUCCESS in 38m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b50ee893af4478e93f99a10610040ea : SUCCESS in 39m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/762b0458dbc9400d9c56d61f4b36e310 : SUCCESS in 15m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d57296fbc43b4ac78ef7da8f5d369c2c : SUCCESS in 24m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7516d58aa5041e0b9d97d73da52527d : SUCCESS in 20m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f426274e97174833b7f18586423c4682 : SUCCESS in 21m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc94e7941f8b42a69458ea079bddafc8 : SUCCESS in 10m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/07d9d26901004cc6ab17db871060dd27 : SUCCESS in 54m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3c131429d3b48f7b8bd78419318c8e9 : SUCCESS in 1h 29m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/aaf5232410ee473d9976b5ffc734b4d8 : SUCCESS in 1h 08m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/36def8f6af494aa4aff65afd84bd7dba : SUCCESS in 1h 02m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9c837e8cda224c8f85a9d9fa556b838e : SUCCESS in 1h 04m 27s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/489d2eedc4c04cdbbaaffd4ff1e9fa46 : SUCCESS in 1h 18m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/721d0c030a644feaa8ca3ef44fc37821 : FAILURE in 1h 01m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/10ff3dcb4ef34f1ca3c5bd4bc2735ee9 : SUCCESS in 1h 47m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dba4638d25849cb81333a409437fa5b : SUCCESS in 57m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1072cb9f934b4e45bfc6f2d60e2ec9b2 : SUCCESS in 58m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/744f9be9b68e4d71827a61f58841c794 : SUCCESS in 29m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa89ac0d302944cc853552f6bc14bab5 : SUCCESS in 1h 08m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/113550d36bb44818a11f0f305695fab8 : FAILURE in 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"2fff68d894a7294799f8024fefafb75c9d174cf9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-28 17:20:11.000000000","message":"Patch Set 23:\n\nhi folks, do you have some mins for this patch?","accounts_in_message":[],"_revision_number":23},{"id":"3e1a5fe7fdffe60fe750d296524dc620a4b47b6e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-05 09:38:48.000000000","message":"Patch Set 23: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"92a8ef904f48692675d3d63f149c939ab2a100b2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-09 09:52:37.000000000","message":"Patch Set 23:\n\n(1 comment)\n\nping fellow reviewers","accounts_in_message":[],"_revision_number":23},{"id":"280f91a2ea99ba3c5f4192511198835d74434d00","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-09 09:52:39.000000000","message":"Removed reviewer Mellanox CI.","accounts_in_message":[],"_revision_number":23},{"id":"2fb8d6c30c63530c923188119bab49c8f93bd479","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-13 09:41:20.000000000","message":"Patch Set 23:\n\nping fellow reviewers","accounts_in_message":[],"_revision_number":23},{"id":"652bae27175ac0af1b59869c15e3e1aa7940b081","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-15 16:14:33.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"80a4e92ceb829172925562bd5fa7cfedf427ebda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 17:06:19.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ffdccff995e4ffcacab4c8cc6f97278 : SUCCESS in 42m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc1b54d4adde4f95aac0755e0410f63a : SUCCESS in 42m 47s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"152899ee4496b4877259ad1edfbdca2e69c3bb97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 18:30:08.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6f961a66b5b9422bbaf30638e3607f95 : SUCCESS in 40m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/efdc18b2ed93422496db4c42cfa94ada : SUCCESS in 2h 12m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e21db18330d84e24ae26729b5f52fe91 : SUCCESS in 1h 56m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d375217eaf8b422ab150e02e5f70c2ca : SUCCESS in 1h 23m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3bbe510fb90d454d95978f798e0b812f : SUCCESS in 1h 48m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/53129e45e96240dbb8214fa14f6e8395 : SUCCESS in 25m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6db9b847d359441bba9af8c46e9e2aa0 : SUCCESS in 32m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e0aaee97ac14066b5f813f8483155c9 : SUCCESS in 36m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9af674497e60405f825dd03c99abc09d : SUCCESS in 15m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ecb47c8180d4d8a8eafd0928b9773a9 : SUCCESS in 23m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3f3e3ed351b4e29915c7a95da280684 : SUCCESS in 26m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9705b45230846568d2b9067ba872206 : SUCCESS in 25m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b348578421408cb2d61bb05296768e : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e39b4b2eebf64a1082af39d8e0fda4ce : SUCCESS in 48m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/233c925d4c3240d9813c91eb80754d46 : SUCCESS in 1h 24m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ee75aee372464ddab828b9df41ab227a : SUCCESS in 1h 07m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebda69c3bacb445697e08648bec21e98 : SUCCESS in 1h 00m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0aabdb1e3d1b4886a946a1015317a799 : SUCCESS in 1h 00m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b880a0bec4954086b35e1f1441f6de71 : SUCCESS in 1h 07m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/82eb3aa07aef4ec38a550b001dfde31c : SUCCESS in 1h 16m 15s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/0bb923baa4e14cc885f5bb8e17d1009c : SUCCESS in 1h 27m 00s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79589f047db44c61aca4fbe4eff14c59 : SUCCESS in 53m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ee99a7f4e4ff429ba9b0ceabb57d5aaa : SUCCESS in 59m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/44ee22bfc68c4c1b8433c6562d378958 : SUCCESS in 32m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2c75c21175954bdda27a0d0ad8e9de0b : SUCCESS in 1h 08m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/75d8256866884741b7ee304bb77f68e7 : FAILURE in 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"925d6ba25ae8e14a85be974be70a8c448c55a4f0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-15 19:48:31.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/24 : SUCCESS in 1h 45m 16s","accounts_in_message":[],"_revision_number":24},{"id":"fddb8602f793229e275db56233e007900ab88579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-07-19 15:47:57.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"491525c1101c528be392f3a676a403abc6d58218","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 16:37:14.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a6e939b799e483b9fc709e53a723a8c : SUCCESS in 41m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6781659aba2241b288e7a14eeb1bbd2f : SUCCESS in 40m 57s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"2b06092f58611087a03ba3a670edf1583f3b2526","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-07-19 17:19:57.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776701/25 : SUCCESS in 1h 30m 54s","accounts_in_message":[],"_revision_number":25},{"id":"8d25a75beb0a242896b265835c619a5b78971728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 18:08:02.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0b2a8c1e83d3484fb96817280f93a3bb : SUCCESS in 46m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2dc312f335e343ca86833cee78c23d01 : SUCCESS in 2h 19m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b8c8cb19adbf4b64b3d4b165e1c64c5a : SUCCESS in 1h 31m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/92f41c745a774d7aa1ad31adf7de1fe8 : SUCCESS in 1h 26m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/33372c4e08394d34a6569297ae230f5c : SUCCESS in 1h 39m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0540a54076694a59b7753bad5119e51f : SUCCESS in 32m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f48c43371b2b43e6832322c5fdac651c : SUCCESS in 30m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ecad71f22dc414aaf2318b389c90beb : SUCCESS in 36m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc85989de9e34a1aaf40040a43cab7e5 : SUCCESS in 13m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9e9741083144d7c9fef245aeea4562d : SUCCESS in 22m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e62019c2e4b04065a9921648c7c81b6d : SUCCESS in 23m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1405d244fc2b4c589e49078c8a08cd3f : SUCCESS in 24m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b765da9767145c6aac4e278c7416047 : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e912e24eacc945378b2d6ad573facc27 : SUCCESS in 43m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f44ba17ebb034fd9bee20bb8a4071b95 : SUCCESS in 1h 26m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f8a11690d1b349fd92879fa6dbb9f899 : SUCCESS in 1h 07m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/26b25dd38c5745e6962f2107b33d1ea1 : SUCCESS in 57m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3c371a40fc044bff92cf648c8a0dbd02 : SUCCESS in 1h 06m 48s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e78aa995054d4c9ca8e58a81af01cd7f : SUCCESS in 52m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9a68896c81cd4ef5b6effe527a64f7be : SUCCESS in 1h 17m 50s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/64843599583d4fdd999b72e298e35133 : SUCCESS in 1h 26m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/710a9fbf940c40b8ac8d03d3b3bfcc35 : SUCCESS in 1h 06m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c61c10b8b4584754bb81651ab8169234 : SUCCESS in 59m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/12c48d4c2eb448e8815d84a7077567ab : SUCCESS in 33m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2342296ba7641be984f31f4d959409f : SUCCESS in 1h 15m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5344d9bdec9346738499114c10a81aff : FAILURE in 30m 53s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"8000f0e9f0aa950746feedbbdcbc68fadda66e0e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-19 21:25:09.000000000","message":"Patch Set 24:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"8fc629938d93342f0855a397f9c5a6265c1a5096","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-07-20 12:27:06.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2021/Arista_Tempest_Testrefs/changes/01/776701/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"9970ab64eb167a8d3cb9ffca7020d0091b0cb09a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-07-20 14:18:30.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1\n\n(3 comments)\n\nThanks Rodolfo,\n\nI\u0027ve added a few nits but I think they could be done in a follow-up patch. Let me know what you think.","accounts_in_message":[],"_revision_number":25},{"id":"772e465648f7857ee3661400cad065c1fe4faabb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 14:19:08.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"eb08f8f2bc93625d6d0619a8efa92eacc56c8003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 16:42:48.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d823d65d1fc54044b06b50b743dde9a1 : SUCCESS in 43m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4c2c178e8af14ea79a27fac7d2227f90 : SUCCESS in 1h 57m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/955974d6c71b41488850c4544a753bac : SUCCESS in 2h 10m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8d0800b7cbdf410281ae4b36b9365fab : SUCCESS in 1h 39m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/88319212165e4cb4b0c7288cc779385c : SUCCESS in 1h 22m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e54a06f3497a4e7e9bb3809780cf69eb : SUCCESS in 39m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3cab229a2343efb16bb46648cffc85 : SUCCESS in 15m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed329e838dda4fbb8211dadd6450da78 : SUCCESS in 26m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83b8b7dc7cbb466387ee42ad5e3b8123 : SUCCESS in 23m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d0e95d9c6b4ce0851e195b5883ff34 : SUCCESS in 9m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/88390c39ec6c435fa91040e5ae47d638 : SUCCESS in 48m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/12c0ca93412b4260b081ee8043716a09 : SUCCESS in 1h 24m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8738ed08be2c47569ef19e3f5decaf32 : SUCCESS in 1h 08m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/02e68af61cd342ee979df7a81b6c90eb : SUCCESS in 55m 58s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/2f2b26364f0f4e88a8f86c50b8d6f58a : SUCCESS in 1h 34m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ca90eda7ed7411aaed67e45097f7445 : SUCCESS in 52m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/846c235ac8924dafbca08ba21afc09bf : SUCCESS in 58m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d34a6095a51847718779e1342b2710ce : SUCCESS in 28m 38s","accounts_in_message":[],"_revision_number":25},{"id":"555075525d1fce365d8adf61dcb2f6120c2af373","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 16:42:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":25},{"id":"fb80341d75aaf4d8d387e828c94963a794a068ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 16:44:31.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143851fa062d46079311a65a0b140010 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":25},{"id":"cf6d1c5fd573f91c151ebbff2de272c1ad2ba5f5","tag":"autogenerated:gerrit:revert","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-07-20 21:25:54.000000000","message":"Created a revert of this change as Id8cdd9c69e4fef2d9c335447b498958add8b7816","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"df5cb28737453e2dfee17e177160fa6d7a59b7e0","revisions":{"3fdba9155d40cda9c30ba6780492c92db51f310a":{"kind":"REWORK","_number":1,"created":"2021-02-19 19:39:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/1"}}},"commit":{"parents":[{"commit":"f416060d41de0705b631eb3164c984543aecff99","subject":"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f416060d41de0705b631eb3164c984543aecff99"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:32:04.000000000","tz":0},"subject":"[WIP][OVN] Add a SB Chassis event to track changes in BW config","message":"[WIP][OVN] Add a SB Chassis event to track changes in BW config\n\nAdded a SB Chassis event to track any change in the resource provider\nbandwidths or the inventory defaults.\n\nTODO: complete the OVN client Placement API extension\nTODO: finish the chassis event implementation\nTODO: use the placement API to communicate this info\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3fdba9155d40cda9c30ba6780492c92db51f310a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3fdba9155d40cda9c30ba6780492c92db51f310a"}]},"branch":"refs/heads/master"},"e9edc7ebee3baa142e1066beaa16023f9c164d34":{"kind":"REWORK","_number":2,"created":"2021-02-23 17:44:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/2"}}},"commit":{"parents":[{"commit":"f416060d41de0705b631eb3164c984543aecff99","subject":"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f416060d41de0705b631eb3164c984543aecff99"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-23 17:44:29.000000000","tz":0},"subject":"[WIP][OVN] Add a SB Chassis event to track changes in BW config","message":"[WIP][OVN] Add a SB Chassis event to track changes in BW config\n\nAdded a SB Chassis event to track any change in the resource provider\nbandwidths or the inventory defaults.\n\nTODO: complete the OVN client Placement API extension\nTODO: finish the chassis event implementation\nTODO: use the placement API to communicate this info\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9edc7ebee3baa142e1066beaa16023f9c164d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9edc7ebee3baa142e1066beaa16023f9c164d34"}]},"branch":"refs/heads/master"},"393705b858e70476346c85358fca60e0c0480705":{"kind":"REWORK","_number":3,"created":"2021-02-24 17:53:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/3"}}},"commit":{"parents":[{"commit":"f416060d41de0705b631eb3164c984543aecff99","subject":"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f416060d41de0705b631eb3164c984543aecff99"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-24 17:53:30.000000000","tz":0},"subject":"[WIP][OVN] Add a SB Chassis event to track changes in BW config","message":"[WIP][OVN] Add a SB Chassis event to track changes in BW config\n\nAdded a SB Chassis event to track any change in the resource provider\nbandwidths or the inventory defaults.\n\nTODO: complete the OVN client Placement API extension\nTODO: finish the chassis event implementation\nTODO: use the placement API to communicate this info\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/393705b858e70476346c85358fca60e0c0480705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/393705b858e70476346c85358fca60e0c0480705"}]},"branch":"refs/heads/master"},"79074d0cb38d016f2d3b3b4e48713d2f2f2b180c":{"kind":"REWORK","_number":4,"created":"2021-03-02 11:49:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/4"}}},"commit":{"parents":[{"commit":"f416060d41de0705b631eb3164c984543aecff99","subject":"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f416060d41de0705b631eb3164c984543aecff99"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-02 11:38:24.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79074d0cb38d016f2d3b3b4e48713d2f2f2b180c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79074d0cb38d016f2d3b3b4e48713d2f2f2b180c"}]},"branch":"refs/heads/master"},"2981ffcaf5526f1cdce74fabe6aafb825b9e61c9":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-03-03 10:59:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/5"}}},"commit":{"parents":[{"commit":"f416060d41de0705b631eb3164c984543aecff99","subject":"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f416060d41de0705b631eb3164c984543aecff99"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-03 10:58:23.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/776656\n--\u003e bump neutron-lib to 2.10.0\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2981ffcaf5526f1cdce74fabe6aafb825b9e61c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2981ffcaf5526f1cdce74fabe6aafb825b9e61c9"}]},"branch":"refs/heads/master"},"39cc862420a9a86c0f4646a03c12c91d7e232430":{"kind":"REWORK","_number":6,"created":"2021-03-03 14:09:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/6"}}},"commit":{"parents":[{"commit":"f416060d41de0705b631eb3164c984543aecff99","subject":"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f416060d41de0705b631eb3164c984543aecff99"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-03 14:07:59.000000000","tz":0},"subject":"[WIP][OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[WIP][OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/776656\n--\u003e bump neutron-lib to 2.10.0\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39cc862420a9a86c0f4646a03c12c91d7e232430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39cc862420a9a86c0f4646a03c12c91d7e232430"}]},"branch":"refs/heads/master"},"c6ec6495dc15b3b6c642864349f3939f4b4195fb":{"kind":"REWORK","_number":7,"created":"2021-03-03 18:23:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/7"}}},"commit":{"parents":[{"commit":"f416060d41de0705b631eb3164c984543aecff99","subject":"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f416060d41de0705b631eb3164c984543aecff99"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-03 18:22:52.000000000","tz":0},"subject":"[WIP][OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[WIP][OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/776656\n--\u003e bump neutron-lib to 2.10.0\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6ec6495dc15b3b6c642864349f3939f4b4195fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6ec6495dc15b3b6c642864349f3939f4b4195fb"}]},"branch":"refs/heads/master"},"3f5a8993a20cb613c7770d31cb45552ab27d11ad":{"kind":"REWORK","_number":8,"created":"2021-03-04 09:33:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/8"}}},"commit":{"parents":[{"commit":"f416060d41de0705b631eb3164c984543aecff99","subject":"[OVN] Set OVN mech driver \"resource_provider_uuid5_namespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f416060d41de0705b631eb3164c984543aecff99"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-04 09:33:00.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f5a8993a20cb613c7770d31cb45552ab27d11ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f5a8993a20cb613c7770d31cb45552ab27d11ad"}]},"branch":"refs/heads/master"},"e9a96087374ec2e2f543d9034614bd944fddc9b1":{"kind":"REWORK","_number":9,"created":"2021-03-04 14:24:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/9"}}},"commit":{"parents":[{"commit":"891a715e4f513ef6d42f048be2b86645aa0e0443","subject":"Merge \"Trivial: check if ipv6 is available for IptablesManager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/891a715e4f513ef6d42f048be2b86645aa0e0443"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-04 14:23:42.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9a96087374ec2e2f543d9034614bd944fddc9b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9a96087374ec2e2f543d9034614bd944fddc9b1"}]},"branch":"refs/heads/master"},"199416d737313290030e119e8bd91c3832672eb5":{"kind":"REWORK","_number":10,"created":"2021-03-05 12:41:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/10"}}},"commit":{"parents":[{"commit":"891a715e4f513ef6d42f048be2b86645aa0e0443","subject":"Merge \"Trivial: check if ipv6 is available for IptablesManager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/891a715e4f513ef6d42f048be2b86645aa0e0443"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-05 12:41:16.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/199416d737313290030e119e8bd91c3832672eb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/199416d737313290030e119e8bd91c3832672eb5"}]},"branch":"refs/heads/master"},"501a0e5565c008d8cf65387b2121ee5615d5e858":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-03-06 07:17:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/11"}}},"commit":{"parents":[{"commit":"5187397da1976ba600f3d222e3714866083c58d7","subject":"Merge \"Implement secure RBAC for the logging API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5187397da1976ba600f3d222e3714866083c58d7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-03-06 07:17:49.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/501a0e5565c008d8cf65387b2121ee5615d5e858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/501a0e5565c008d8cf65387b2121ee5615d5e858"}]},"branch":"refs/heads/master","description":"Rebase"},"6497d75f266a5bde526fc62faca660dca76456f1":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-03-07 18:20:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/12"}}},"commit":{"parents":[{"commit":"2328cd7c8476a723ce388a08856642c5117c2f71","subject":"Merge \"Remove try block from privileged ip_lib module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2328cd7c8476a723ce388a08856642c5117c2f71"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-03-07 18:20:32.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6497d75f266a5bde526fc62faca660dca76456f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6497d75f266a5bde526fc62faca660dca76456f1"}]},"branch":"refs/heads/master","description":"Rebase"},"acc48c8836027cc163038ea0a1f1d3815a89be6f":{"kind":"REWORK","_number":13,"created":"2021-03-08 14:50:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/13"}}},"commit":{"parents":[{"commit":"2328cd7c8476a723ce388a08856642c5117c2f71","subject":"Merge \"Remove try block from privileged ip_lib module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2328cd7c8476a723ce388a08856642c5117c2f71"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-08 14:39:04.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/acc48c8836027cc163038ea0a1f1d3815a89be6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/acc48c8836027cc163038ea0a1f1d3815a89be6f"}]},"branch":"refs/heads/master"},"6cf0409e2aa624e31bc067a2c9cff04a8fdcddc8":{"kind":"REWORK","_number":14,"created":"2021-03-09 11:06:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/14"}}},"commit":{"parents":[{"commit":"2328cd7c8476a723ce388a08856642c5117c2f71","subject":"Merge \"Remove try block from privileged ip_lib module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2328cd7c8476a723ce388a08856642c5117c2f71"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-09 11:06:07.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6cf0409e2aa624e31bc067a2c9cff04a8fdcddc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6cf0409e2aa624e31bc067a2c9cff04a8fdcddc8"}]},"branch":"refs/heads/master"},"d1286b5d3e5141868f42dcfc09365b67c0d6c43b":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-03-10 14:08:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/15"}}},"commit":{"parents":[{"commit":"57af40eb3194e29ddb2de4fa71870c2d2ca2cf95","subject":"Merge \"[SR-IOV] Add support for ACCELERATOR_DIRECT VNIC type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57af40eb3194e29ddb2de4fa71870c2d2ca2cf95"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-10 13:48:27.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1286b5d3e5141868f42dcfc09365b67c0d6c43b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1286b5d3e5141868f42dcfc09365b67c0d6c43b"}]},"branch":"refs/heads/master"},"c7d1febdab2f2f7e07380f30394e6e65b652d240":{"kind":"REWORK","_number":16,"created":"2021-03-20 15:39:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/16"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-18 21:22:15.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7d1febdab2f2f7e07380f30394e6e65b652d240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7d1febdab2f2f7e07380f30394e6e65b652d240"}]},"branch":"refs/heads/master"},"575ad589dfb7167fe32e445560969e28409f06f0":{"kind":"REWORK","_number":17,"created":"2021-03-29 15:38:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/17"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-29 15:38:07.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/575ad589dfb7167fe32e445560969e28409f06f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/575ad589dfb7167fe32e445560969e28409f06f0"}]},"branch":"refs/heads/master"},"905dea6af88cf73ab26827fc753ce5114866b366":{"kind":"REWORK","_number":18,"created":"2021-05-17 16:09:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/18"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-17 16:08:48.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/905dea6af88cf73ab26827fc753ce5114866b366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/905dea6af88cf73ab26827fc753ce5114866b366"}]},"branch":"refs/heads/master"},"5648d0fb5c7003b69efc953cfba0de25e2c32ba6":{"kind":"REWORK","_number":19,"created":"2021-05-18 08:40:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/19"}}},"commit":{"parents":[{"commit":"174052499951e1ab9c5290aa72bef246db53d5ba","subject":"Merge \"Check for existence instead of fetching the whole net object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/174052499951e1ab9c5290aa72bef246db53d5ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-18 08:40:23.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5648d0fb5c7003b69efc953cfba0de25e2c32ba6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5648d0fb5c7003b69efc953cfba0de25e2c32ba6"}]},"branch":"refs/heads/master"},"86007473cb7ee10d86d4decb855c1302b0dced71":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2021-05-24 08:36:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/20"}}},"commit":{"parents":[{"commit":"b24d2fd7690a01d56e4b09588e52ff7f9481e8c5","subject":"Merge \"Deprecate and remove \"get_agents_db\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b24d2fd7690a01d56e4b09588e52ff7f9481e8c5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-05-24 08:36:11.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86007473cb7ee10d86d4decb855c1302b0dced71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86007473cb7ee10d86d4decb855c1302b0dced71"}]},"branch":"refs/heads/master","description":"Rebase"},"1e3bbf89f936291ca7160a76197465aac887d4a0":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2021-06-03 15:55:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/21"}}},"commit":{"parents":[{"commit":"7e98d189270c408b4a5f5b69e9e479a458407b22","subject":"Merge \"Switch neutron-rally-task job to ML2/OVS and skip \"test_models_sync\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e98d189270c408b4a5f5b69e9e479a458407b22"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-06-03 15:55:38.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e3bbf89f936291ca7160a76197465aac887d4a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e3bbf89f936291ca7160a76197465aac887d4a0"}]},"branch":"refs/heads/master","description":"Rebase"},"b06dda0fb2f568548f8d2d6ac9fb8fc664738bc1":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2021-06-11 09:16:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/22"}}},"commit":{"parents":[{"commit":"389584c389f7cb7a13e75587dae558658e1ccb1a","subject":"Merge \"[Doc] Remove one of the known DVR HA limitations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/389584c389f7cb7a13e75587dae558658e1ccb1a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-11 09:15:10.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b06dda0fb2f568548f8d2d6ac9fb8fc664738bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b06dda0fb2f568548f8d2d6ac9fb8fc664738bc1"}]},"branch":"refs/heads/master"},"53e4043b3e0b0e61dba80b728ef0bc4f468970e0":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2021-06-24 08:16:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/23"}}},"commit":{"parents":[{"commit":"2692953017e70271c117ba0f52bcd99ca57459a6","subject":"Merge \"Remove FIP agent\u0027s gw port when L3 agent is deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2692953017e70271c117ba0f52bcd99ca57459a6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-06-24 08:14:02.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53e4043b3e0b0e61dba80b728ef0bc4f468970e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53e4043b3e0b0e61dba80b728ef0bc4f468970e0"}]},"branch":"refs/heads/master"},"80af554b81d037b279289708221536d295d006af":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2021-07-15 16:14:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/24"}}},"commit":{"parents":[{"commit":"474f5d700b6652426f93186890702c1608fa0fdc","subject":"Merge \"[OVN] Fix ML2/OVN + Neutron DHCP agent use case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/474f5d700b6652426f93186890702c1608fa0fdc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-15 16:12:17.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80af554b81d037b279289708221536d295d006af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80af554b81d037b279289708221536d295d006af"}]},"branch":"refs/heads/master"},"df5cb28737453e2dfee17e177160fa6d7a59b7e0":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2021-07-19 15:47:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/01/776701/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/776701/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/776701/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/776701/25"}}},"commit":{"parents":[{"commit":"7be20080903d9bb9d3c0f1b40bfb5311e2d4d08e","subject":"Merge \"[L3] Use processing queue for network update events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7be20080903d9bb9d3c0f1b40bfb5311e2d4d08e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-19 17:27:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-07-19 15:25:57.000000000","tz":0},"subject":"[OVN][Placement] Add a SB Chassis event to track changes in BW config","message":"[OVN][Placement] Add a SB Chassis event to track changes in BW config\n\nAdded a new OVN Client extension: OVNClientPlacementExtension. This\nextension is in charge of handling the bandwidth information stored\nin the OVN database, in the \"Chassis\" registers on the\n\"ovn-cms-options\" dictionary.\n\nThree new keys are created to store the resource provider information\nneeded to parameterize the network backend bandwidth information,\nfollowing the implementation done in OVS and SR-IOV:\n- resource_provider_bandwidths\n- resource_provider_inventory_defaults\n- resource_provider_hypervisors\n\nWhen the OVN Client is started, the Placement extension will check if\nthe \"placement\" extension is loaded. It will also create an event to\ncheck any configuration change done in any \"Chassis\" register.\n\nThe Placement extension will read the initial configuration stored\nin the OVN database and will populate it to Placement API, creating\nthe needed resource providers, traits and inventories.\n\nNOTE: This patch belongs to a series of patches to implement\nminimum bandwidth scheduling blueprint in OVN backend. The next\npatch will make OVN backend scheduling aware using the information\nstored in Placement API and the port information passed by Nova when\na VM is created.\n\nPartial-Bug: #1578989\n\nChange-Id: I8ba38b8ace8852009fba8712aafa9f88c2b93ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df5cb28737453e2dfee17e177160fa6d7a59b7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df5cb28737453e2dfee17e177160fa6d7a59b7e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
