)]}'
{"id":"openstack%2Fneutron~776829","triplet_id":"openstack%2Fneutron~master~Ifce864ab1c75c2a165db3d8c295002ef3f944277","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"Ifce864ab1c75c2a165db3d8c295002ef3f944277","subject":"[ovn][metadata] Remove local binding for the metadata port","status":"ABANDONED","created":"2021-02-21 10:37:28.000000000","updated":"2021-02-22 12:59:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f0f0ebf0339ce92e556def86816274617efbcd1","_number":776829,"virtual_id_number":776829,"owner":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-21 13:42:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-02-22 12:54:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":-1,"date":"2021-02-22 12:36:05.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-21 10:40:19.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"state":"REVIEWER"},{"updated":"2021-02-21 11:15:10.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-21 11:21:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-21 12:39:29.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-21 13:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe22c539680d3680a603542b9b0fb7e99f066997","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-02-21 10:37:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5decbcb1eae6116ef88e88cd50eff603775d7d94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-02-21 10:38:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"041ef1d915c0fcfd95277ead1d62f6bfe2e404cf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-21 11:15:10.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776829/2 : FAILURE in 35m 54s","accounts_in_message":[],"_revision_number":2},{"id":"dc47bdf839cb3635e114a2b7403a6d5e7cb75a95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 11:21:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e995c540ddc4765a672704df8ac3385 : SUCCESS in 36m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1378b76d99e54f59b1f3673a91d159ac : SUCCESS in 38m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"96520e53aacd196d6c42a1cc528ae28354305f98","author":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"date":"2021-02-21 11:41:32.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":2},{"id":"d5d7706aee74b529785569586f5f419e229ccd95","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-21 12:39:29.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/29/776829/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/29/776829/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"1ed9b6a25b4bedbc791035b803836b304344e337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 13:42:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6d23a74552d647f09b39471ab56737ec : TIMED_OUT in 2h 49m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/070cd6e057fc418088f80830d5bb612a : SUCCESS in 45m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/539148bf8cf24c53bd193055b1d954f5 : SUCCESS in 40m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/358f0fdbb00148d9b12e98f1d30584c0 : SUCCESS in 13m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92c169d053a94e1ba3d871aa35b83a2f : SUCCESS in 40m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/963d552458914cd39130549fff20f2d6 : SUCCESS in 31m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bca6fee3b32d4dbfb2a29e8f841f0c4f : SUCCESS in 38m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2730dc6e9885477f8ea2bd0ea4945ac4 : SUCCESS in 8m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/24c1dd00c9ca42cda7d9eb7afa12fa89 : SUCCESS in 46m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c59c669928a9449b9b9667ef4d406979 : TIMED_OUT in 2h 13m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/01c7c06e31dd477d956c6258e345b61f : SUCCESS in 1h 22m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/209dac1521c541d9a4a55b07170d4951 : SUCCESS in 1h 08m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a092a4f0ea9347d097843718b74053bd : SUCCESS in 1h 08m 31s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e44956d884c848248fdf96b704dcbdf7 : SUCCESS in 1h 13m 14s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/539dac1f801246388d75fad624b18175 : SUCCESS in 1h 28m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5bf2b4c1bfe45febbe21c183ac29f99 : SUCCESS in 1h 25m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ef9d2ae999f4aa3afdbce8be971cbd4 : SUCCESS in 1h 14m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c3cb983d6c704ab1a7aa52bb2b2caab7 : FAILURE in 1h 13m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/69b799a9b99840b3b51f2c62e0056c0a : FAILURE in 57m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4c74d3fb7dbb4d2999611f0b095258f9 : SUCCESS in 1h 42m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/84423a89cbde460ca49cc14c82d5fa7e : TIMED_OUT in 3h 02m 39s","accounts_in_message":[],"_revision_number":2},{"id":"f782f1a8ec4a3fe842ccddb3ed6810a8fcfcf73c","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-02-22 12:36:05.000000000","message":"Patch Set 2: Workflow-1\n\nChecking again since I believe iface-id is needed after all 😊\n\"2021-02-21 11:50:34.731086 | controller | curl: (7) Failed to connect to 169.254.169.254 port 80: Connection timed out\"","accounts_in_message":[],"_revision_number":2},{"id":"4825f1c56fc912c21efd5b5ca6d51a2fda96e3aa","author":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"date":"2021-02-22 12:54:14.000000000","message":"Patch Set 2: -Code-Review\n\nI think we do need iface-id for localports too.","accounts_in_message":[],"_revision_number":2},{"id":"7f0f0ebf0339ce92e556def86816274617efbcd1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-02-22 12:59:39.000000000","message":"Abandoned\n\nConfirmed with the core OVN team, iface-id is needed :)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef5f4763dac1e391f9b38622114c46273962f043","revisions":{"3e788b582f79a92da60760829035ff99977ecf0d":{"kind":"REWORK","_number":1,"created":"2021-02-21 10:37:28.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/29/776829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/776829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/776829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/776829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/776829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/776829/1"}}},"commit":{"parents":[{"commit":"90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13","subject":"Merge \"Bump neutron-lib to 2.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-02-18 13:26:27.000000000","tz":60},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-02-18 13:26:27.000000000","tz":60},"subject":"[ovn][metadata] Remove local binding for the metadata port","message":"[ovn][metadata] Remove local binding for the metadata port\n\nRight now, the metadata agent sets the external_ids:iface-id\non the localport and that creates a local binding in OVN.\n\nWhile this is ok, it is not needed and hence this patch is\nremoving it for simplicity and avoid creating local bindings\nin OVN.\n\nSigned-off-by: Daniel Alvarez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ifce864ab1c75c2a165db3d8c295002ef3f944277\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e788b582f79a92da60760829035ff99977ecf0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e788b582f79a92da60760829035ff99977ecf0d"}]},"branch":"refs/heads/master"},"ef5f4763dac1e391f9b38622114c46273962f043":{"kind":"REWORK","_number":2,"created":"2021-02-21 10:38:32.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/29/776829/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/776829/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/776829/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/776829/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/776829/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/776829/2"}}},"commit":{"parents":[{"commit":"90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13","subject":"Merge \"Bump neutron-lib to 2.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-02-18 13:26:27.000000000","tz":60},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2021-02-21 10:38:18.000000000","tz":60},"subject":"[ovn][metadata] Remove local binding for the metadata port","message":"[ovn][metadata] Remove local binding for the metadata port\n\nRight now, the metadata agent sets the external_ids:iface-id\non the localport and that creates a local binding in OVN.\n\nWhile this is ok, it is not needed and hence this patch is\nremoving it for simplicity and avoid creating local bindings\nin OVN.\n\nSigned-off-by: Daniel Alvarez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ifce864ab1c75c2a165db3d8c295002ef3f944277\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef5f4763dac1e391f9b38622114c46273962f043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef5f4763dac1e391f9b38622114c46273962f043"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
