)]}'
{"id":"openstack%2Fneutron~776876","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Ic35c2f9bceacec8eeba67a2b1ea0cd0b0ffc72fe","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1894843-stable/victoria","hashtags":[],"change_id":"Ic35c2f9bceacec8eeba67a2b1ea0cd0b0ffc72fe","subject":"Fix deletion of rfp interfaces when router is re-enabled","status":"MERGED","created":"2021-02-22 02:01:10.000000000","updated":"2021-02-23 00:53:27.000000000","submitted":"2021-02-23 00:51:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776876-bug/1894843-stable/victoria","cherry_pick_of_change":775372,"cherry_pick_of_patch_set":3,"meta_rev_id":"5414c2f2b2b8a8e91dee22c9934ff22e18ab2763","_number":776876,"virtual_id_number":776876,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 00:51:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-02-22 09:14:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-02-22 15:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-02-22 15:29:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 03:52:59.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-22 05:25:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 09:14:16.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-22 15:29:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"1a1ad51a965871bf77761ec2b163477fe9725c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-02-22 02:01:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0c07737541bc212b9a6108e0340e50a998b0d8f8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-22 03:52:59.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776876/1 : FAILURE in 1h 50m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4d2772b067a1923ab97fb997eb6822f5cd8d8a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 05:25:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/1bfff38448064e06a981f8fa6b987b5a : SUCCESS in 54m 12s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/e03e0c339ab5448298b25507e6ab0f7b : SUCCESS in 1h 32m 53s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/5cdc210151f34db4a9498ba63328e3ed : SUCCESS in 1h 18m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/80b7352350254856a6bdd163d295225f : SUCCESS in 1h 27m 33s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/d3812a51a8b64059b898531fcb269337 : SUCCESS in 1h 24m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1917c4e6c6d34130be2c40c674d29feb : SUCCESS in 47m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f10997e0284148628d43270c60dac196 : SUCCESS in 19m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1666bb2c9f94b329eb190174a031a9c : SUCCESS in 28m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fa7550af7e345bc8a9dbe1cf8d82d7c : SUCCESS in 28m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f50bbd5de4a4d53ade5d5cc4ec7ebc3 : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/17522ebf841d4d75aac3a9cb17850883 : SUCCESS in 1h 17m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6b44396e7fe491daa308891d9ea02aa : SUCCESS in 1h 07m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fd177f5a42e2416cae29ed89327aba63 : SUCCESS in 1h 41m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1257c3fbd0d404ca4a6fbb43dfc71ab : SUCCESS in 1h 04m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/68b91ec7dc49407a8fe7774d00a1a742 : SUCCESS in 1h 16m 39s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/13629e26fc514e7caf7cec226a8c6bc2 : SUCCESS in 1h 11m 59s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f6b137889f514af985c5c287d3342aff : SUCCESS in 1h 14m 52s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/91214c781c1441f9ab190d61ccfae733 : SUCCESS in 1h 16m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/90e0ae99c6bf44a5af9bdce524862b2e : SUCCESS in 1h 41m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0034a8c7719a4514ba72c9fa5cc56c9b : FAILURE in 1h 19m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e2a773b335774f5ea10d41a92f94e77a : SUCCESS in 1h 18m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2e5e1dbb61b7415c9db4df66b5cd3219 : SUCCESS in 1h 25m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/99121f84f3ae4cc09795e618e6dbb1b3 : SUCCESS in 2h 04m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7d8b7c2693f4cc6ae011bffc2dabdb9 : SUCCESS in 1h 00m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/53d9f92d1e474f5ca883c596135c7923 : SUCCESS in 1h 19m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2b146251eca040a9b150b01c2907342c : FAILURE in 33m 20s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3c910a595f114ad59eacdd5550e1f327 : SUCCESS in 1h 27m 11s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/840eff4a521b49c695b6b165594ca4cb : SUCCESS in 3h 21m 59s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5c595b8029044e5a8ca7c9413ab51bb4 : SUCCESS in 1h 20m 18s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/95e08b86ef1b4d06a767af06ab0d45cb : SUCCESS in 2h 04m 15s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4d7f0c3528e145d9b1fd6669846a3067 : SUCCESS in 2h 08m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/98fbfd44426f409093e86bdbe09a9610 : FAILURE in 33m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"071c0b1c5b75bcd58827a65f9189bdbe21e6d6ae","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-22 09:14:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6365776b553215d1b8924b21a8d38031a1f146d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-22 15:29:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e1e2c62110537ba70e91de44d6a02ef270401e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 15:30:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3b041d317a4502ce2d30f5b7e04d1bd175d7eeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 17:56:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/75f6e44ce01749b398340af8e6d23a94 : SUCCESS in 51m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85681aff8cda4ca7bb85826ebdc51982 : SUCCESS in 14m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e11428bc261c4d27a396402534dff4d4 : SUCCESS in 27m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fea3ae12468647f880c842f91ff15cc8 : TIMED_OUT in 41m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd867f3003a8487d815e4ba949052362 : SUCCESS in 8m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/744b708d450a4f4abd46da6875729708 : SUCCESS in 1h 19m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f22103fbb09a44e7ae2e507f267daae1 : SUCCESS in 1h 13m 28s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/920fe2d24a2740f1bcacf0679dcfe4a0 : SUCCESS in 1h 13m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d333528e427c40b7b7a7826512372a9a : SUCCESS in 1h 06m 32s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2613eecb9a9b4266b96773220d04c212 : SUCCESS in 1h 17m 20s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a699a61a80cb4654b53a6c6c280be75c : SUCCESS in 1h 24m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/843c528003c94bffaab5b605c198f0e3 : SUCCESS in 1h 02m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9adc3d349326470aa7ed7fbe68c568f2 : SUCCESS in 1h 23m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f5f4d165e834e8ca5d7cd7779e971b7 : SUCCESS in 2h 19m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7926c63860e947949a4c1db01e42f788 : SUCCESS in 1h 04m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f43e2127757e4cb9a800101c8f179698 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"22ae43a7160ca19565e9d8884129c10502320ee1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-22 18:01:47.000000000","message":"Patch Set 1:\n\nrecheck TIMED_OUT","accounts_in_message":[],"_revision_number":1},{"id":"cd45fe7e64e65ba090fba759f2fe9bc0d944b8b0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-22 19:47:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776876/1 : SUCCESS in 1h 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5e1a6b4ec67cb9ee25908b4c2040cebc421e29c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 22:52:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/4ca78039d7a0408fb5b27ad32a58abef : SUCCESS in 39m 46s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/673a425032a14f58857e35a27a07662a : SUCCESS in 1h 07m 51s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/516f35f1c2d44a7fbacd5f6ab2967f04 : SUCCESS in 1h 32m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/bd37990c7440487c90ce1d8c8b0cc069 : SUCCESS in 1h 27m 23s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/8e7d7f3fc968454e890b5257c1964904 : SUCCESS in 2h 27m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99c2bce9b8724046aa5c3c730d1a7d1a : SUCCESS in 47m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a5856b8902d4793a8686c1d83ae286a : SUCCESS in 12m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fc3459f61a841b2b9fcfcf19edbec2e : SUCCESS in 30m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23ea44d6c2a54ffdb5ac87bf9f9eafe7 : SUCCESS in 26m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88878c607150473388faad44c17addf1 : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccfed2556bc247f09dc401dc266ee0cb : SUCCESS in 1h 12m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0a6af579e7b49beb7627f08face3cda : SUCCESS in 1h 25m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4d1dc755e2de471899f575d86ea93fe2 : SUCCESS in 1h 36m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e049d80d41a4b4dae1e123a605aa0ad : SUCCESS in 1h 14m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0e88cce1ddc243a2976867b66913e7c5 : SUCCESS in 1h 10m 14s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/7691714a9e024d60a2a0b02a46a6bb0d : SUCCESS in 1h 31m 23s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/26a4043e3da54b749e03d3fcc4f12bf2 : SUCCESS in 1h 17m 29s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/afa579cf4ab24b9a8e52926cb7c812f0 : SUCCESS in 1h 26m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c8c1c1cea3ff4007b5c8843540dff35f : SUCCESS in 1h 41m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2a4308635e474524b342568904554e17 : FAILURE in 1h 12m 55s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e0cec17fdbc9410080bc1a9f778c4b8c : SUCCESS in 1h 13m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3ca3e3e0937f4734b5b6c3699210b0d4 : SUCCESS in 1h 13m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cac621e6a6894522a238b2062582bfb9 : SUCCESS in 2h 13m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bccd2003132d46cf8c8c7f5d5d767409 : SUCCESS in 59m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4fd9c112e4b34e9ab7d666875844dd96 : SUCCESS in 1h 14m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9bb395c94c4c4770949b041f2fe178e3 : FAILURE in 38m 03s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4ea71b7c39cf41acb67db12ac1b07f39 : SUCCESS in 1h 17m 26s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1d3cc4dd9d5741ec91bc7263d9bd7286 : SUCCESS in 3h 23m 13s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/54d03b09ac0946b19ca2a34128d6af5e : SUCCESS in 1h 05m 41s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d6f64b033ea84dafbf44b1fc94088f3c : SUCCESS in 1h 20m 06s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ef9a34f045b740548d0da15e7e8b0251 : SUCCESS in 1h 59m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/50370deb55ba4a84b52a28584d861723 : TIMED_OUT in 3h 02m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e65ea89f85246a4cf1271cfa0e4e2664239c9b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 22:52:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa6f7a0e1649bd299c781c162ff4b6329876f264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 00:51:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/22f6165d28b8414db64f2400577c8d82 : SUCCESS in 40m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6001bac447a4450957afe77269107bd : SUCCESS in 14m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/372f48f718994454b69387e59ee8a983 : SUCCESS in 24m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b119b2d636c46038c03a3e3b7fe8ccf : SUCCESS in 28m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d241add2546d485d8df817ea15964f35 : SUCCESS in 8m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1f1ba624b7b1489884ebf61efeffa0c0 : SUCCESS in 1h 09m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/268b4908cf744c4492253fd84fd5629e : SUCCESS in 1h 07m 55s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2bd39d60fea5439a8b217d2989a937a3 : SUCCESS in 1h 09m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/17ec6ea98a6544958489f7de9065013f : SUCCESS in 1h 20m 20s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d8fad7cd2f804ce3b5e65f895ea7deca : SUCCESS in 1h 08m 09s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8a6b0bd181d5492eab87f4e2bbdaca7c : SUCCESS in 1h 19m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f71f32baee24fc7a20f2cabc9d9a4d5 : SUCCESS in 1h 04m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/efda39af180d40bdbb27957ce5993ac7 : SUCCESS in 58m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6d14a1ba7ed54e2a92667f9a21e20557 : SUCCESS in 1h 56m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7983081d2b8b4059b1601a9436f1f8e9 : SUCCESS in 55m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/47fd417f0dcb4837bc6fa741696657df : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":1},{"id":"393fe59a7e03fd6ae8959c654cef358fdd32a460","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 00:51:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5414c2f2b2b8a8e91dee22c9934ff22e18ab2763","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 00:53:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2739c254457a43b5b7c27d325d0b8130 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7322ed2aefb4d18465449c405868310355b5312","revisions":{"f7322ed2aefb4d18465449c405868310355b5312":{"kind":"REWORK","_number":1,"created":"2021-02-22 02:01:10.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/76/776876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/776876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/776876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/776876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/776876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/776876/1"}}},"commit":{"parents":[{"commit":"830746417481e8666ac03a3912cc1438281617e4","subject":"Merge \"Fix update of trunk subports during live migration\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/830746417481e8666ac03a3912cc1438281617e4"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-02-12 12:14:38.000000000","tz":330},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2021-02-22 02:01:10.000000000","tz":0},"subject":"Fix deletion of rfp interfaces when router is re-enabled","message":"Fix deletion of rfp interfaces when router is re-enabled\n\n1. When dvr router is disabled and enabled back, the rfp\ninterfaces are deleted on the nodes without snat-* namespace.\nThis is due to creation of snat namespace during router\ninitialization stage on all the nodes. At later stages, since\nthe gw_port_host is not bounded on this node, external gateway\nis removed which triggers removal of rfp interfaces and snat\nnamespace.\nCreate snat namespace only on the nodes where gw_port_host is\nbounded.\n\n2. In case of DVR SNAT, when the l3 agent is rescheduled to\nanother node, the rfp interfaces on qrouter-* namespace are\nremoved. Instead of calling external_gateway_removed() which\nfurther deletes the rfp interfaces, the qg-, sg- interfaces\nneed to be unplugged and snat namespace need to be deleted.\n\nCloses-Bug: #1894843\nChange-Id: Ic35c2f9bceacec8eeba67a2b1ea0cd0b0ffc72fe\n(cherry picked from commit 393d484e07e53496ff068d14047c97d28febd03a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7322ed2aefb4d18465449c405868310355b5312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7322ed2aefb4d18465449c405868310355b5312"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
