)]}'
{"id":"openstack%2Fneutron~776877","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ic35c2f9bceacec8eeba67a2b1ea0cd0b0ffc72fe","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1894843-stable/ussuri","hashtags":[],"change_id":"Ic35c2f9bceacec8eeba67a2b1ea0cd0b0ffc72fe","subject":"Fix deletion of rfp interfaces when router is re-enabled","status":"MERGED","created":"2021-02-22 02:01:41.000000000","updated":"2021-02-23 19:50:04.000000000","submitted":"2021-02-23 19:35:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776877-bug/1894843-stable/ussuri","cherry_pick_of_change":775372,"cherry_pick_of_patch_set":3,"meta_rev_id":"b7a7c0830d24ceb9f18f592d46f244e0d5d4bd8a","_number":776877,"virtual_id_number":776877,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 19:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-02-22 09:14:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-02-22 15:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-02-22 15:30:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 02:03:46.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-02-22 05:18:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 09:14:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-22 15:30:19.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"564691e77f5151b90fac6c434877be98e6ed75ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-02-22 02:01:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7fa0b8ddd88e1cc2c11acf40d607ba6cd98dfd91","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-22 02:03:46.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/776877/1 : FAILURE in 1m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8021e08d21a511c4b6bbd51766ee5318a2d12803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 05:18:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/47762926cdec45aea75016f489a5fb67 : SUCCESS in 53m 20s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/a2bc60e6dcb648a48ec943ad8d3b53af : SUCCESS in 1h 22m 38s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/942fa9b4032547af9981c7accc681309 : SUCCESS in 1h 22m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/aa5680ba12024893b501951d1cd5b3e0 : SUCCESS in 1h 35m 13s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/e29f95a9a5e843fe81c012fde19d8352 : SUCCESS in 1h 21m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c31bda0bfe734d54ba65b91f378c983e : SUCCESS in 32m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32f42e371dba464cbdd8793a8c815168 : SUCCESS in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/810b1a6452804bd2b54c52f39e59394a : SUCCESS in 40m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe0b4710f390423289d792ba852142ea : SUCCESS in 28m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0ca5dd1f7df453085d7199e2c91e547 : SUCCESS in 38m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70fa56847c4b4f81a21be7b6c63a3bc7 : SUCCESS in 8m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/61603530bdea46daa3768e6648e00a42 : SUCCESS in 54m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6e07a03aa15d4751986fa66ec15ed4fa : SUCCESS in 1h 12m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/13a8fa432c174761be4e0d993e563885 : SUCCESS in 1h 15m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/efcec50fc7a145f8ac1942ece005362b : SUCCESS in 1h 52m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/30c68ac1c1f648c79ee19927fc9d887d : SUCCESS in 1h 17m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/24d3007098794798b733319443213ef0 : SUCCESS in 1h 30m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b977d7aa2da245eba2086caa588c38fd : SUCCESS in 1h 18m 34s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/511ed495105340559da5433007a649d3 : SUCCESS in 1h 09m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4b6b89fe6b064a258fe02c4e6373f1f6 : SUCCESS in 1h 14m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5ea0b6f8f01c48fdb3594dbc2c8fa4b0 : SUCCESS in 1h 36m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/65639a996fce48e18006f22ce1fc20d2 : FAILURE in 1h 21m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a488ebb8318c4a4597f43de87d5f485b : SUCCESS in 1h 15m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7d319ae68a2644d5ae582730e25e4d64 : SUCCESS in 42m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/90ef231e7e324b0fa80d6d0f28deaf91 : SUCCESS in 2h 01m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7640d4b838654b57b1dd1cb7828ae2eb : SUCCESS in 1h 01m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d265c82c07684d1095be641d3d957e58 : SUCCESS in 1h 20m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7323ac877abc487e90a414b790849f1c : SUCCESS in 34m 43s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4840dd2e9d4e4d7b89aa3a06f263d3d8 : SUCCESS in 1h 25m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4803d1060fe94ed0b836a84a4a34690c : FAILURE in 56m 21s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b1b382bb23204961aaf3baa523ab3112 : SUCCESS in 3h 14m 55s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fe800f8053a04a3a969e88bb73fed981 : SUCCESS in 1h 20m 45s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4da2165ed92745f19b20d9c9d5045495 : SUCCESS in 1h 38m 57s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3c9ce730ef61424b8573a1836d27490d : SUCCESS in 2h 06m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7be017c3c46b4c37992793f751bb570e : FAILURE in 1h 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b43d597c214b5da1882c07768c367b0fd3f9a0f1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-22 09:14:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"05dde692ab4b32f3119b83e45adddcbc48efcb9e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-22 15:30:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7e70c621dd9028aff1f0f4364aa1a9b18553bd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 15:30:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9607a25e8a86e4e27c31497912e53a232deabd19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 18:16:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/7470316b53b546029968dafad4f1df78 : FAILURE in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e00b396dc4641ee9181003fba7be6bc : SUCCESS in 10m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6333fa2dca2458a99bd53bb2b8f660d : SUCCESS in 26m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bb485ee2d344ec9bc36c02cd898dcce : SUCCESS in 32m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f94c4b6582654d23a31d5e41761447d5 : SUCCESS in 11m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c57deba89124da4bb7751de32d65c95 : SUCCESS in 58m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/442d9a9e5c6d4c7c981e6534d8c23f74 : SUCCESS in 1h 11m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c5eb7484433c4581bf5f74d8e322d374 : SUCCESS in 1h 15m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0c6a36568b0e421b955b94b1a933cec4 : SUCCESS in 1h 14m 06s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5a6fe01c7b44b30b243652f73608514 : SUCCESS in 1h 27m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b183e7c374fd45e2b65484adcd4c08fc : SUCCESS in 2h 00m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/035e4ae303824e37bbd2adef1e2daf77 : SUCCESS in 1h 34m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/13382c82de404e1394924c2653a5df21 : SUCCESS in 1h 49m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a3b2efb3bb04930b01a26d4b43eb01e : SUCCESS in 50m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9496664c24274ac98a0bb331b2ed8640 : SUCCESS in 1h 08m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a9a43488b7df8174b3e99b5b54506e2b20bae535","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-22 19:49:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b3994e768596819ac63c7e88e8d653ef3d517db","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-22 21:47:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/776877/1 : SUCCESS in 1h 57m 38s","accounts_in_message":[],"_revision_number":1},{"id":"37e6a5345120f39b06a74814a877258aea30ea4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 23:45:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/49e0da7b2fbe4858b72854ae34c54c40 : SUCCESS in 53m 22s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/05a54812f14846ed9f73112228a50af4 : SUCCESS in 1h 23m 19s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/7d48f07f52fa43f084826b42a9ce94a7 : SUCCESS in 1h 02m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/fe3deb4944fb4f49a82bc2fa6d837d0f : SUCCESS in 1h 21m 36s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/65291e3e7d9f4654a0296730dde107be : SUCCESS in 1h 32m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61efd47c4ba84fe48f2531cbb2874cde : SUCCESS in 38m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3093521b41c84612aed7da4a7dd6aafb : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0026ba40548941a1b43f408d62140679 : SUCCESS in 27m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9aadf03a5c8e43f688202d53c3540db9 : SUCCESS in 36m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92babcd631794b1aa316e4b57aff958d : SUCCESS in 17m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e3850240744dc89b032d02a39620c0 : SUCCESS in 9m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/57e8138eba124dbdb179e4e342bb0ac6 : SUCCESS in 56m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/66e03e7d52294364a926bbfe4c8769f2 : SUCCESS in 1h 11m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c9d3714ff6346d5b1d7d998885dd7d9 : SUCCESS in 1h 06m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/745f52b63a794e8b823d56a670ef4804 : SUCCESS in 2h 29m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8804be2bb78244c193aa7b27c2733048 : SUCCESS in 1h 14m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b4d36094bbc140fb89416b5c4dd2f13c : SUCCESS in 1h 18m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8cb0db5f05fb4e8f8e8b0efe260e9cd9 : SUCCESS in 1h 18m 28s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b65299466484a7f8a2d42640fb1dada : SUCCESS in 1h 13m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3bce5312107c4f7e87512721d139ad5a : SUCCESS in 1h 09m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fb90684dc9ec4c46870b9a1b38eb28f6 : SUCCESS in 1h 38m 02s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/74cf9775791c4cc398524e05405f7fa5 : FAILURE in 1h 49m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/915cd2b4da2948c4b121f7736e74952d : SUCCESS in 1h 14m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9e3e5a5ae6df4cb18c4633268852c99d : SUCCESS in 1h 05m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a686cfe6b54b4593a95d4bd940cebadd : SUCCESS in 1h 52m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cd7f396a3a44e4aa91c761ba7785ded : SUCCESS in 45m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fc17f28b4c644f809ac8be7ae74a7d51 : SUCCESS in 1h 14m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ab580aa1f49e43969d40e9313946fc0a : SUCCESS in 39m 04s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fa8c25f499fe493a80d65d83a6113e5a : SUCCESS in 1h 22m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6827beaf7c8647d7bbcfc1eb403b782d : SUCCESS in 52m 48s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9b915b42cf5045ac95a2eb7cfdfb4d8d : SUCCESS in 3h 34m 16s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cc9523481ec0466c817a44da0690c8a8 : SUCCESS in 1h 14m 09s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/35505d0c06c54bebb0d846fca65253b1 : SUCCESS in 1h 59m 17s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ef4ad1bb9e874297995dac138e17b3f4 : SUCCESS in 2h 04m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fda2862be2344d8c9ac51d308832da9f : FAILURE in 1h 47m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e9cbc81ae2ffe3602e7a8bc7fdef755f40618e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 23:45:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d541b1160e442325fd50c2434284b6a5917b2b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:35:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/13220ca3485d46d8b002bccbd3f80c1f : SUCCESS in 40m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f3fb3a5b95487da10c95bb83f8b93f : SUCCESS in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfd02069c7c34034b3827871ca2cff88 : SUCCESS in 23m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7af77ecb2b89436498eedb9e6bbadaef : SUCCESS in 25m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5abdc4c16efb4c1890cdcc933aeda3da : SUCCESS in 8m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7f8deaf0dad442a1b933643f396b5808 : SUCCESS in 48m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/44f86f9ec0414bdb84397344ac7916d2 : SUCCESS in 1h 07m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fa878c30400a4872ae43899b501c12b2 : SUCCESS in 1h 05m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b97577fe4abd48f79985251e2a82bf2e : SUCCESS in 1h 14m 21s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a19a01a1362e4821b6018fe1f5fc230c : SUCCESS in 1h 01m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/43f5d1edaeff4009b432a80b66f3e40a : SUCCESS in 55m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e49cf66cfe7b4251bc5a02bd7f77455a : SUCCESS in 1h 00m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/84346751cf5a4dbda0978323070ebb68 : SUCCESS in 1h 32m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82fa234784074d8bbd91b8a28993684b : SUCCESS in 43m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8ca5d8c64fe94d889a00b824fd3da865 : SUCCESS in 1h 09m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e6542b2982946d805b52a217ce6c903ff4941675","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:35:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b7a7c0830d24ceb9f18f592d46f244e0d5d4bd8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:50:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ee0cdc678da4e3e98759d50a2c35ef1 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73e1672d6fc8ccabd87c97773d51b9efede9ba55","revisions":{"73e1672d6fc8ccabd87c97773d51b9efede9ba55":{"kind":"REWORK","_number":1,"created":"2021-02-22 02:01:41.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/77/776877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/776877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/776877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/776877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/776877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/776877/1"}}},"commit":{"parents":[{"commit":"2e87ff7c1bdb07a0890017ad89f282ad3882db22","subject":"Merge \"[OVN] ovn-metadata-agent: Retry registering Chassis at startup\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e87ff7c1bdb07a0890017ad89f282ad3882db22"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-02-12 12:14:38.000000000","tz":330},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2021-02-22 02:01:41.000000000","tz":0},"subject":"Fix deletion of rfp interfaces when router is re-enabled","message":"Fix deletion of rfp interfaces when router is re-enabled\n\n1. When dvr router is disabled and enabled back, the rfp\ninterfaces are deleted on the nodes without snat-* namespace.\nThis is due to creation of snat namespace during router\ninitialization stage on all the nodes. At later stages, since\nthe gw_port_host is not bounded on this node, external gateway\nis removed which triggers removal of rfp interfaces and snat\nnamespace.\nCreate snat namespace only on the nodes where gw_port_host is\nbounded.\n\n2. In case of DVR SNAT, when the l3 agent is rescheduled to\nanother node, the rfp interfaces on qrouter-* namespace are\nremoved. Instead of calling external_gateway_removed() which\nfurther deletes the rfp interfaces, the qg-, sg- interfaces\nneed to be unplugged and snat namespace need to be deleted.\n\nCloses-Bug: #1894843\nChange-Id: Ic35c2f9bceacec8eeba67a2b1ea0cd0b0ffc72fe\n(cherry picked from commit 393d484e07e53496ff068d14047c97d28febd03a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73e1672d6fc8ccabd87c97773d51b9efede9ba55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73e1672d6fc8ccabd87c97773d51b9efede9ba55"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
