)]}'
{"id":"openstack%2Fneutron~776895","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I84718b8ac4d8d6e39426e1c4485c2d7fe28185dd","project":"openstack/neutron","branch":"stable/train","topic":"bug/1915530-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"I84718b8ac4d8d6e39426e1c4485c2d7fe28185dd","subject":"[OVS FW] Clean conntrack entries with mark \u003d\u003d CT_MARK_INVALID","status":"MERGED","created":"2021-02-23 13:41:58.000000000","updated":"2021-03-10 18:24:22.000000000","submitted":"2021-03-10 18:19:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776895-bug/1915530-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":776894,"cherry_pick_of_patch_set":1,"meta_rev_id":"66528032946cb987ccc266aa28598065498da23f","_number":776895,"virtual_id_number":776895,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 18:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-10 10:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-03-03 09:15:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-03-10 10:00:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-23 17:54:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 22:59:07.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-03 09:15:14.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-10 10:00:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"1822fadce4a7a6461a391db811f14c01781d05ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-23 13:41:58.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"631311b64bf22025279e1a90abc31138ca0c8441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 17:54:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/ea7f374c3ca84567b8059930e9d7b8c7 : SUCCESS in 1h 50m 10s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/27b767891f5d4e4cb175da4656f818e6 : SUCCESS in 1h 45m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/8ac00b38096b46388bf4ab724f7c848d : SUCCESS in 1h 23m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ce19236d4104c21a04eb6537a72075f : SUCCESS in 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/918f3b754e634e91b5ce7906d6c582ab : SUCCESS in 11m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa155739b4984595934685d2c45f9474 : SUCCESS in 30m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4620971eeb89467d8939d44985922771 : SUCCESS in 24m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb8e1d6516a84bac8c243d46f30a29f5 : SUCCESS in 23m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7427398a9e4bff9321b07c4b899668 : SUCCESS in 12m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/3acfd9e4510745268a21abe94ea2c0a1 : SUCCESS in 1h 04m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1442502757844c6c9c36665a8f52aeda : SUCCESS in 1h 15m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b16c62979856469a8e354eb68a2f7013 : SUCCESS in 40m 54s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/eab25298aa6c4259a03856f832f6e7e5 : SUCCESS in 44m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4d8dd3236142420da85adf57e05d15be : SUCCESS in 1h 15m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6cc785a422f8422fb0ac21117ef29ec2 : FAILURE in 2h 10m 01s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ec660c9fad0e408abe91feb23322177e : SUCCESS in 1h 20m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/99a043e805754f83bc60bbb5e1f2107d : SUCCESS in 1h 07m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a6163ca86f64c7f828ea136f148beed : SUCCESS in 1h 03m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fd2f427449c243d1a12ecfe340fb8365 : SUCCESS in 1h 08m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bcd9edf6a54d4ca990d02d42538acec8 : SUCCESS in 2h 00m 22s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fea4fb7d53d3484d9c57925f9e12dad2 : FAILURE in 2h 22m 54s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28f70ca0edbe44fb849041bb434aae33 : SUCCESS in 53m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/52a009325d474ec79e029895e68e11ba : SUCCESS in 53m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8dac99c6ad444293b58e20b06c11e562 : SUCCESS in 1h 06m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ceecb9528474803bdafe32bcc300b7c : FAILURE in 1h 51m 10s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d89a3ae0516149d68e62ce6570f71e8e : SUCCESS in 2h 03m 43s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/b697671f051a46ddba56b4b47cc79965 : FAILURE in 18m 10s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a3dd30f3052446dcad6e2bf6f39b24ed : SUCCESS in 48m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4b2139af7574663bd3d2895652a831e : SUCCESS in 48m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5c10b471ad904871a614368daa9ffe24 : SUCCESS in 1h 44m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a23e1105394a49b0933178c9e050ddcc : SUCCESS in 38m 35s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/793008dfccdc4e16971ccbd374e8b396 : SUCCESS in 1h 21m 34s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef785fd5e2fe44079a849622fac415d3 : SUCCESS in 1h 19m 44s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/cb81e7fdb9d848b199838721bc0a22d1 : FAILURE in 22m 15s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54cb83291d21494e29c1f325d4b6aaaa8f089567","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-23 22:59:07.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"94b06d33fce78abba0bc56023b4f8d8d174d322d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-25 09:12:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fd6820a24c2b60fd81e47d20fe47a46890cea294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 17:19:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/4a402b1e14054a889d7ee8f7342ef887 : SUCCESS in 1h 48m 45s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/3dcb7428fd6e48b79bc7b0327ceea1f1 : SUCCESS in 2h 11m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/d2a8867031f04f348d1d60c8b8b90961 : SUCCESS in 1h 22m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2860767f4d64467a820eda85d56df44 : SUCCESS in 28m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c846cd3e06497fa70a2d06c82cfdeb : SUCCESS in 12m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc578cfad3044658aaddb15466ad862f : SUCCESS in 22m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ced2f80142a48f39374e518519f151e : SUCCESS in 31m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bfd0425f4c14213aabfc951138f6f44 : SUCCESS in 25m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7077dd7ae44691bea0db5b0c74baad : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/47c4584cf8834bd6b70a0208c13beab1 : SUCCESS in 1h 04m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e852e742b2544e5ea1e68ad0d8b732fc : SUCCESS in 1h 10m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0ad9a6a692394480b481744c4c03212e : FAILURE in 52m 46s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b7bad45a00bc4fc686c32d32fdbce3b7 : SUCCESS in 48m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/dbde98e69b364b4694cc11311936964c : FAILURE in 1h 18m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/838698c77613458f8b399986f2f52ad7 : SUCCESS in 1h 59m 36s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8f7ea3db929e4036b1c1feceb435c549 : SUCCESS in 1h 19m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1999c542183744c0b537731aa60b7253 : SUCCESS in 1h 12m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/18d5cda59bb54793bbc2206a7f4cfe50 : SUCCESS in 1h 11m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9af304e0ff674cf38f6c57c4bb8cd08d : SUCCESS in 1h 10m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/48544b573da84e8083c48d1b379f0b52 : SUCCESS in 1h 44m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4fcde3ebe22442abbdcaa42a3a9d1590 : FAILURE in 1h 20m 28s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cf7404621add4703a51442a96d7ccefc : SUCCESS in 54m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a49acbd3601d472686a255376615f255 : SUCCESS in 50m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/32b117fa88c44b51bb3b1d2f4a759793 : SUCCESS in 1h 24m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4cfbceb07cf94630980549f0c7579915 : SUCCESS in 2h 11m 28s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4fe268505a224f99af0521f25e4af545 : SUCCESS in 1h 50m 58s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/ccaf6dd192c541a2b81deec22178f715 : FAILURE in 18m 50s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7d78730a83414e02bb935c1ab7cb92a6 : SUCCESS in 1h 06m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3782ab2a0f8d446fa2e97d41fee3a56f : SUCCESS in 59m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5bfe715c6a8e41238013cb0cddd17ced : SUCCESS in 1h 24m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/437701f5fc8b408cb548f52196384db2 : SUCCESS in 52m 36s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a1d74b2a2344b509c3cee31b6a096e8 : SUCCESS in 1h 13m 17s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fb61913586b4568a65cae009812f4e3 : SUCCESS in 1h 21m 48s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/19c225c6c26741d8a6800c4857a7e59d : SUCCESS in 2h 51m 47s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e91f5b5b196b4ca28a904f819500c2eb : FAILURE in 1h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0f0cc65084b3ef426de2dc966b0557435c71230","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-25 20:51:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"915394ca690dee9fc43599f8da311653f03e34c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:35:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/006748dc70a9456cbca96809329fbfe7 : SUCCESS in 1h 24m 32s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/ebc1b9a614c4415ab6b6b7295fd8900d : SUCCESS in 1h 28m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/a01136eae8a04275aa65114cbb2bfb95 : SUCCESS in 1h 42m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d35da0af7ee945ad8a50aa3513f638a1 : SUCCESS in 35m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2998d77efbf04a6da5d4e0a57a6c007a : SUCCESS in 11m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fb1a109849d4702b8715f2e204c3963 : SUCCESS in 21m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cf4ed2760894e45af60f81cb752ed49 : SUCCESS in 26m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69aacc92578c4bea880a07e78b39f59d : SUCCESS in 25m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d18f9a2317fd4ef998514ea7ff1638f7 : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/fc47784ed7514b4c99410c485b9059a7 : SUCCESS in 1h 08m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/84f08536dd87424d99a2633092b30b64 : SUCCESS in 1h 08m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5ed08ae283be418f8223b959a19123f9 : SUCCESS in 46m 51s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/33fae0ca807346c1813307183dfe6c7c : SUCCESS in 38m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6daeaa221d394e048b2976e926cb48d0 : SUCCESS in 1h 18m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/641f48da0f1d4e059933b77c1dfdc404 : SUCCESS in 1h 57m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/87a1ace716004ddcb2084ea2e61114b7 : SUCCESS in 1h 19m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/28de611245674532ba9dbb3d49185b28 : SUCCESS in 1h 03m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/204a4428c65f4a4a9afc173e0536f073 : SUCCESS in 1h 07m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5151eaacb91841ed86f11aad97f9b8f9 : SUCCESS in 1h 10m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c30c3bb71c194daeb815f5eb0a5dc34b : SUCCESS in 2h 01m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4a9993fda223441e907be4b8c34f89ea : FAILURE in 1h 43m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/41bd8120c60e4c3f81ff31c55b48c602 : SUCCESS in 1h 17m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5a7c5ec043a0417882b3f66d6ccddac1 : SUCCESS in 1h 12m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0b43b45f2c7c409ea98e928e52f9adb9 : SUCCESS in 1h 15m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/926085fe203446b8aee6a9c08512e644 : SUCCESS in 2h 09m 04s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0a0a5fb0769946b1b171886032359da9 : SUCCESS in 2h 33m 07s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/19f8a5d08e2c49cbb1a7301b0f30cd07 : FAILURE in 18m 35s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/93a7204291c84b5e8d9c1614752a175e : SUCCESS in 59m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb96e489007e4910b675a69be0112ac9 : SUCCESS in 1h 15m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ab603948dc8c4b10a067a26438f31acc : SUCCESS in 1h 27m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb277e4a876645a7b54057abc7ee09b4 : SUCCESS in 44m 54s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/56b241855f19442aae8e04d586237953 : FAILURE in 1h 13m 03s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/083282912a824ffe96b41b7e2e924866 : SUCCESS in 1h 11m 18s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/067d7b9d477146ee9e2b3da4a0805feb : SUCCESS in 2h 43m 48s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e66a6065fdec43e7b837b4c3b2f32ba1 : FAILURE in 1h 06m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f558424d56e0e92a617bf9909058b01348f42094","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-03 09:15:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9382cbfe0f411a58f6bfe6214153e3007613b619","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-10 10:00:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"efcb17fcd6de32494f0263a353ed6cee5dbd728e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 10:01:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7668fffccaa7a4c13012b81836690d5a779bf1ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 18:19:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe93dbfaf9744d28bdae82ba55217f0 : SUCCESS in 12m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07c19b77315c478d8b81190f9eff0cd6 : SUCCESS in 21m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/864bfd6a29db4a7f9bbb020e6056d8fd : SUCCESS in 21m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ebce66a83e74e7ab9b9821f52499127 : SUCCESS in 22m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e576827ce0154618980fb1ee6f2dcb95 : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/6a41a61fb49c47628c9298e1d8b53f30 : SUCCESS in 59m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6a7ad2738c9343738aaae2f85c43e1d8 : SUCCESS in 1h 08m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7811e47cc13b43449cf48a4b00b57fb4 : SUCCESS in 41m 53s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/228c139192f54c019d4906457c6b71f0 : SUCCESS in 42m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f383ec79c9d047fe977e64aeca61ebdc : SUCCESS in 1h 14m 30s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e1f893ed06344b0b9fee16a06b621d15 : SUCCESS in 1h 13m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/70736c6db3544a33a315cf1d92a360db : SUCCESS in 1h 11m 08s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7cb19589727048f69089ce5716caeb09 : SUCCESS in 1h 06m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3baf33041b714b29981770d4e7e38b00 : SUCCESS in 1h 09m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9584ac082da4102bb1cc107481167b8 : SUCCESS in 1h 06m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/da3199862ccb4f78812b7edf9a3ae33f : SUCCESS in 1h 01m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f56594c7fa6243608f68d4316d0fb4b3 : SUCCESS in 1h 32m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9217b723866742c589b99286419b8a87 : SUCCESS in 1h 00m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55c3858335d34f7cb5db44837649775d : SUCCESS in 45m 23s","accounts_in_message":[],"_revision_number":1},{"id":"db976915a9b61922a02bc9af1326d371058a32eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 18:19:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"66528032946cb987ccc266aa28598065498da23f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 18:24:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7cb5ad1c0c4d988b55f4765ea1e4c5 : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6da1543f492bf50260cbcf99075b87a7b9e34281","revisions":{"6da1543f492bf50260cbcf99075b87a7b9e34281":{"kind":"REWORK","_number":1,"created":"2021-02-23 13:41:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/776895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/776895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/776895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/776895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/776895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/776895/1"}}},"commit":{"parents":[{"commit":"526bb04096e02ed79bc6df3391c887e7d003c2c9","subject":"Merge \"Fix update of trunk subports during live migration\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/526bb04096e02ed79bc6df3391c887e7d003c2c9"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-16 10:19:44.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-23 13:41:58.000000000","tz":0},"subject":"[OVS FW] Clean conntrack entries with mark \u003d\u003d CT_MARK_INVALID","message":"[OVS FW] Clean conntrack entries with mark \u003d\u003d CT_MARK_INVALID\n\nIn case when security group is removed from port, connections\nestablished to/from port which were hitting some of the removed SG\nrules will be marked with CT_MARK_INVALID in the conntrack table.\nDue to that such existing connections will be stopped.\nBut then if SG will be added again to the port, those conntrack entries\nmarked as invalid will not be removed and those hang connections will\nnot be revived.\nTo fix that, this patch adds deletion of the conntrack entries marked\nwith CT_MARK_INVALID and related to the port of which SG are\nupdated/added.\n\nCloses-Bug: #1915530\nChange-Id: I84718b8ac4d8d6e39426e1c4485c2d7fe28185dd\n(cherry picked from commit 1e220d7136965acbfe50215d982cac0bd3dcf9c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6da1543f492bf50260cbcf99075b87a7b9e34281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6da1543f492bf50260cbcf99075b87a7b9e34281"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
