)]}'
{"id":"openstack%2Fneutron~777535","triplet_id":"openstack%2Fneutron~master~If96384030ef6ee8e92f684b839ab0ef7d2381d88","project":"openstack/neutron","branch":"master","topic":"bug/1911132-router-notfound","hashtags":[],"change_id":"If96384030ef6ee8e92f684b839ab0ef7d2381d88","subject":"[OVN] Fix RowNotFound: Cannot find Logical_Router","status":"MERGED","created":"2021-02-25 10:07:03.000000000","updated":"2021-03-23 15:51:40.000000000","submitted":"2021-03-23 15:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"777535-bug/1911132-router-notfound","meta_rev_id":"d53b632c9d60b247e0a7e189184665f3de13d7d6","_number":777535,"virtual_id_number":777535,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-02-25 11:30:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-23 15:46:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-22 23:45:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-03-23 13:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-03-23 13:46:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 10:53:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-25 11:30:13.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-25 14:05:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-02-25 14:11:31.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-25 17:13:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-27 11:11:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-22 23:45:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-23 13:46:21.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"b751ad1994bf9d8ff40e49e1b40c56ee0c3026bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-02-25 10:07:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f87cfb9a6a74b4db7dca3cce4369962c4c07323b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 10:53:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa3e375dd8084752a6179e2585a5065a : SUCCESS in 41m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d11e9f603bf2433c8c0eb9320ec230e6 : SUCCESS in 40m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8378a9d4aa1a41d7ecba0d3f429d029ebd480963","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-25 11:30:13.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/35/777535/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"2b89c2ef802f2fecf9ca34f78b230130ca5c5f29","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-25 14:05:18.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"29e618307a94a8a8ad3a02a2e9aae25916ed06e2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 14:11:31.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/35/777535/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/35/777535/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"bb4513842ab85929ed86fefeb1709fdad2f8cec5","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-02-25 15:29:50.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"85301bc21a22431b540aa83d0344a46d632fdce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 17:13:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/696cabeb163449a6a93237e66d752978 : SUCCESS in 50m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/08f470cc2a5243c9994ffacc1ed166f1 : SUCCESS in 2h 04m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/918d991fe9e44afd899e04dc398a407e : SUCCESS in 44m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b02d3c87332483482d17eb7f6da8660 : SUCCESS in 41m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/422e6f65a28349488b3ba1011ded64e5 : SUCCESS in 12m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c34dcf42bb314ac0b77f291e94e9d2dc : SUCCESS in 28m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64cf7b4783514df78a4f44835e9f2898 : SUCCESS in 31m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a822c827d966409f91785345a17fc313 : SUCCESS in 27m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92cdb11648f04125a8cf1ed18dfbe5fe : SUCCESS in 11m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d4a238fc87645c1b9dab0e51233f972 : FAILURE in 55m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8f65cda357b74a47a0f103d7acc73b8e : SUCCESS in 1h 17m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/90cfd8eb24e7426b9097c031fcc5c19f : SUCCESS in 39m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e37874612eba423395fd63eb5c8a91ea : SUCCESS in 1h 32m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f15f5acd6a1d41679f8352fc29227b55 : SUCCESS in 1h 26m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f1d68904181275af07b59f8357efa27d90823b90","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-15 15:16:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d99ecd31f33018c8f3343f5677905bb734399a7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 16:29:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a3d45f91061481cb926aefd903771f8 : SUCCESS in 41m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0eeca0749b074d1898122a3053e831d7 : TIMED_OUT in 1h 02m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ad28a69b9b8d03c49b19b743d0ab88595de810e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-15 16:35:32.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/35/777535/1/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/35/777535/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cd5d80bdbbcac7aec4353ea676c166503cd54ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 17:05:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1577a32f04934157ad5c75d2369d6317 : SUCCESS in 58m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c129348164424a0fb48f886dc6147f9f : SUCCESS in 1h 37m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ccb29ee1bd3477b98fddc4013644195 : SUCCESS in 56m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54d04369d7be489c8bdb70ff8b4f0f0a : SUCCESS in 40m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58cb57f80734b20b282f09cadce6d3e : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f37c2edbd3b4bfc9924fbf7a670e48a : SUCCESS in 40m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84f58d30c9c9431b81226665dfe1e63b : SUCCESS in 41m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dffd6e50014f41179beb7998a7e808b2 : SUCCESS in 29m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a52ebecfd6d4fe8801237c45dc806ad : SUCCESS in 12m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb88e2f8990a4d58aa39a5e75fee0f9c : FAILURE in 59m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f96199eeb8e54443abe0eaf651e8c85a : SUCCESS in 1h 10m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a1098a2dbd3748c29cb627b9455aa2a3 : SUCCESS in 38m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/610c328a2a1343fc853cd1212cf83c76 : SUCCESS in 1h 38m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5e7264ff6689481593b2e0d1f0cbfd2d : FAILURE in 29m 46s","accounts_in_message":[],"_revision_number":1},{"id":"940d5290d942ccb592156594f520d6705e424340","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-22 08:55:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"73c3176def9abdd249beb97affee8d13841d7680","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-22 10:07:28.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/35/777535/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/35/777535/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e037e7ea27fbf95171615ba94f680cbb2a9751ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:10:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8e07b2903d641b198f3d9ecaac31c2c : SUCCESS in 44m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b2282862d9d456eb0ac5ca911ca503f : TIMED_OUT in 1h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6df6de66733c5bc460a68d7d9e903d2082b093cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:36:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1b27c5be0a364833b5723fe05a9d2941 : SUCCESS in 43m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/31e30c3d2c854d0fae4bc6a27c7e9ccf : FAILURE in 22m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4a9d3b80b73445e9ce7cd4cda250984 : SUCCESS in 37m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/607006cb7cff4964bf0708a8df614352 : SUCCESS in 38m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a46ec1f3487644c78dca23287a23dfea : SUCCESS in 13m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65bc834f2cb54465945ec3927dac4d6d : SUCCESS in 41m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d882c1d7e94f43f7a48bd5eca0235538 : SUCCESS in 44m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61bdf01c23814b90ae8c68ca390b0622 : SUCCESS in 31m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0533e6c64df4e0fbf2dd2fb06be97ac : SUCCESS in 10m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0f96b4a35f4443f861f18be7159c372 : SUCCESS in 47m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9322bb0f2dbe40ae8e21ac6c3706040e : SUCCESS in 1h 19m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/74fbfbbb985f4c42943e8d39ef92f90a : SUCCESS in 27m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/38a871a7f8c440fb969eddfce6e2741a : SUCCESS in 1h 29m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0576dae0c8e14c43a14c08e4bcf6e3b5 : SUCCESS in 1h 24m 01s","accounts_in_message":[],"_revision_number":1},{"id":"de361c9d57bea55695cf37f6b681bb14813ec017","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-22 10:46:46.000000000","message":"Patch Set 1:\n\nrecheck\n\nunrelated error:\n\n2021-03-22 09:21:48.293976 | controller | tee: /etc/openvswitch/system-id.conf: No such file or directory\n2021-03-22 09:21:48.294050 | controller | 8a325ddd-4421-4f26-b9f9-b120bddfab94","accounts_in_message":[],"_revision_number":1},{"id":"fcb098a78635a1dfd53c911ddb3114649ccb523b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 11:50:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4f16733f68d48be966242b10efcec9d : TIMED_OUT in 1h 02m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a13e6749ee6664c7c7065fe22428d5e4cc7fc186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 12:32:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2efcdf48b59e4094bbbb2c401bc5e06d : SUCCESS in 43m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/260d7b595ccc4690a1e6fa059a8a3393 : SUCCESS in 1h 37m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77dfced3655d4c8cb4f7d033fca168dc : SUCCESS in 42m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1523513c04f14655ad712f7c7d9614dd : SUCCESS in 39m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50558d143e0d49c194513278c382b400 : SUCCESS in 17m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3069ae11f99641f5a5dd8cbb849032a6 : SUCCESS in 44m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef92fdf19c254d06b2ba08f3b70fe9db : SUCCESS in 27m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3259a033467d4a208af34b9d84dac5a9 : SUCCESS in 30m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb805ee8c5d4493a6e53baca950fa72 : SUCCESS in 11m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e204544a220b4235803269e151414294 : SUCCESS in 57m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c58b2341c54f4c83840f6669808f17f4 : SUCCESS in 1h 17m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8adc106c6db64dbcba845d55ad31a850 : SUCCESS in 34m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b5d9f152c96b4e0c8a5e873749c7e2fa : SUCCESS in 1h 37m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/aa8822c91fa64562a1eda70547df1ce1 : SUCCESS in 1h 31m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b7a7bec1388495309da2eb9c4f42df85d7a94378","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-22 13:43:21.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/35/777535/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/35/777535/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"58a192f1c62ef4cc73c18d179e8f1564672ccb44","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-22 23:45:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d0b1e9835ab426a691cf36e8df2763910b268eb","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-03-23 13:46:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e5a8d6d050b4278a60564d599f97b937f62c9f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:47:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76a23357f6880f055d979fe5eb715cf61f14d4d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 15:46:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2e14f5b645524e4d9dde85bdcbb2fbeb : SUCCESS in 54m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8e932979d21445848c339991f0ecdfc0 : SUCCESS in 1h 51m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52d00bcb69e7413db63cc3362afe37e0 : SUCCESS in 51m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69bf188447294d1d9a1cd57d081233bb : SUCCESS in 22m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d220db82c20a4eae95754a31a5194eae : SUCCESS in 23m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16e1271f7e884146be86310810083114 : SUCCESS in 22m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f7d41c1425c421ea5fbc9e6233dd5b5 : SUCCESS in 9m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3d9507c61f94e4f9fb059c9a1fceada : SUCCESS in 47m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d12867d9f34f49e0b485553ae2534d84 : SUCCESS in 1h 36m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a7bc15ca2b0c447c9b94d56763494d8f : SUCCESS in 33m 32s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d1266554c83845b598187bf33131c230 : SUCCESS in 1h 54m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e5402558cb7b004b60bd554ead92ab53f5540362","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 15:46:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d53b632c9d60b247e0a7e189184665f3de13d7d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 15:51:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41f8d39180f4108aae0bd50ef06cde1 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0f6d6189a6f8d17b411e82337a97ec7cddd999a","revisions":{"f0f6d6189a6f8d17b411e82337a97ec7cddd999a":{"kind":"REWORK","_number":1,"created":"2021-02-25 10:07:03.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/35/777535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/777535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/777535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/777535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/777535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/777535/1"}}},"commit":{"parents":[{"commit":"872650e4639cbc50b92f38c0fd4f07e5cec7e873","subject":"Merge \"Don\u0027t update AgentCache when Chassis_Private.chassis \u003d\u003d []\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/872650e4639cbc50b92f38c0fd4f07e5cec7e873"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-02-25 10:03:11.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-02-25 10:03:11.000000000","tz":0},"subject":"[OVN] Fix RowNotFound: Cannot find Logical_Router","message":"[OVN] Fix RowNotFound: Cannot find Logical_Router\n\nThe execute() method in ovsdbapp has the log_error parameter set to True\nby default; this was causing the RowNotFound error to be logged as part\nof get_lrouter() even tho that exception was being handled correct (it\u0027s\npart of the logic of that method to return None if the lrouter does not\nexist).\n\nThis patch sets log_error to False in the execute() call. Note that,\ncheck_error is still True, so any exception other than RowNotFound would\nstill bubble up from that method because it\u0027s not handled, that means\nthat we won\u0027t miss any other exception with log_error being False.\n\nChange-Id: If96384030ef6ee8e92f684b839ab0ef7d2381d88\nCloses-Bug: #1911132\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f0f6d6189a6f8d17b411e82337a97ec7cddd999a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f0f6d6189a6f8d17b411e82337a97ec7cddd999a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
