)]}'
{"id":"openstack%2Fneutron~777616","triplet_id":"openstack%2Fneutron~master~I74ddf5556f7cc49c70b5609b8bbac70928782a4a","project":"openstack/neutron","branch":"master","topic":"bug/1916761","hashtags":[],"change_id":"I74ddf5556f7cc49c70b5609b8bbac70928782a4a","subject":"Delete router arp entry when port deleted","status":"ABANDONED","created":"2021-02-25 16:12:00.000000000","updated":"2021-03-04 12:49:06.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"d9a57ee0850b2c7edab36011e81da0a9df2d516e","_number":777616,"virtual_id_number":777616,"owner":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-02-25 19:22:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-25 23:03:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":-1,"date":"2021-02-26 06:15:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 17:00:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-25 17:44:12.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2021-02-25 17:58:57.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-25 22:14:56.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-25 23:03:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 06:16:49.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-26 06:16:49.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-04 10:53:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"9059b9d34a9efc1faa66a1c9c26b3278d1e69299","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-02-25 16:12:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eff039d2e52059bd29970330aedd1625bcc33c74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 17:00:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c9992af1140425aa96cae65082215c3 : SUCCESS in 40m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8edf42fa2934d27916cf54a383c4bbb : SUCCESS in 41m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1f4d639fec322287544fdf556930395849f912a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-02-25 17:34:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"473e25ff68f1a7d7d94281935d287f67e39c724c","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-02-25 17:41:30.000000000","message":"Patch Set 2:\n\nStill WIP since I need to filter routers that are connected to the same subnet as the port. Other than that it works as expected, deleting port results in notification to all routers to cleanup arp entries for that port and I confirmed that arp entry disappears.","accounts_in_message":[],"_revision_number":2},{"id":"982c8523783b6f9337e495c1c01db80a61977095","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-25 17:58:57.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/16/777616/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"284c61baafee8933e35c66a52665f3bad4054f78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 18:28:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/454f901082c54f0098b012600059f9b1 : SUCCESS in 46m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b417ab8ba9945b9974fdd5116cd9f67 : SUCCESS in 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7f2edcebc23c196bd1212cf7eabc4949d1c82efb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-25 19:22:00.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/16/777616/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"6c5d7c593541cb8cf1da0350cdd133f55fd83e6c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 22:14:56.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/16/777616/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/16/777616/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"cdec0e34f274ceec05db94ad1c23ca836f58be69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 23:03:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d5146d8fde734a43a54da699b56dd35f : SUCCESS in 1h 11m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ec74c7c082824370a11664a75bb81065 : SUCCESS in 1h 45m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/448ff7558a104d5d82029a2bf8859e2d : SUCCESS in 1h 53m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/49f9be2dd90f420a9b77296b27f188f1 : SUCCESS in 1h 41m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4a9d6366ffb548deaf4397afea6856b1 : SUCCESS in 2h 09m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b2842f203e54badaf97c0681d7398a3 : SUCCESS in 23m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a07f4aa4f4f47d2bbf34bf3656185e0 : SUCCESS in 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa8621dafcd249509c482c0c95fd76d7 : SUCCESS in 11m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5484f3d2d5bf4321a45fe565eafc425f : SUCCESS in 38m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed2880028280420ab9bd640dff0c14a3 : SUCCESS in 18m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/841fe8ebf1cd41988b1657b95ca015f9 : TIMED_OUT in 41m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06afabf3673148409ea90936067b595b : SUCCESS in 11m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d702fec4084e44658c076af6a7304ee2 : TIMED_OUT in 2h 15m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7d9b3c34a2a4ec89c388a1680574d7f : SUCCESS in 1h 19m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ba24db111fac46d79cabd7b3f132b9ae : SUCCESS in 1h 36m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2af82d64bd7543938bd230505d53cf29 : FAILURE in 1h 01m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8a075bab323f438a920c9ec415493bac : SUCCESS in 1h 13m 52s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6ab7452cb3384fabafa3579dce439963 : SUCCESS in 1h 30m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/63febfecfb314ea9b50940dd053e0a8e : SUCCESS in 1h 46m 20s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28a31bde99db43e89f5b3001404011a1 : SUCCESS in 1h 44m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98dd1a53d9e44112a53cd6dad0408d21 : SUCCESS in 1h 18m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8bb903442c9743d3849cb5b9399c8aab : SUCCESS in 1h 09m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7fe97c31791d480abe3eb52ff60ece46 : SUCCESS in 32m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7e9990e0805e44ffbe5c9892add675f6 : SUCCESS in 1h 50m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/09d098ed7c8a4898bb5dbade4c00787f : SUCCESS in 1h 23m 35s","accounts_in_message":[],"_revision_number":2},{"id":"731dced3d0bc2babd5a5b514fb819b9193a1a070","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-26 06:15:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nMaybe revert that patch can be the right approach.","accounts_in_message":[],"_revision_number":2},{"id":"dacd0f96eeff3cb527a2b54ff2cbc092404a647c","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-02-26 08:56:54.000000000","message":"Patch Set 2:\n\nLiu, so the problem here is that the arp responder isn\u0027t in the path taken by packets impacted by these arpo entries i.e. floating ip that get DNATed in the qrouter ns. As I explained in the launchpad bug, it is fip packets specifically that are impacted since once they are DNATed the dst mac address of the NATed ip gets set based on whats in the arp cache in the qrouter ns.","accounts_in_message":[],"_revision_number":2},{"id":"42d01ba4f79f99598c79c0a7a0fed89728d67bc4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-04 11:27:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f83fc2cd9af7c960246a8069899c192e2858376","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-03-04 12:09:58.000000000","message":"Abandoned\n\nOriginal problem patch being reverted at https://review.opendev.org/c/openstack/neutron/+/777903","accounts_in_message":[],"_revision_number":2},{"id":"d9a57ee0850b2c7edab36011e81da0a9df2d516e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-04 12:49:06.000000000","message":"Patch Set 2:\n\n@Rodolfo Alonso, you are right, I\u0027ve created a revert for https://review.opendev.org/c/openstack/neutron/+/653883","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"511bde9d80d330d50a3c22d56108b2b5c760fa74","revisions":{"739de839a9217cbec6c3a44ae8ac435b92c83c9d":{"kind":"REWORK","_number":1,"created":"2021-02-25 16:12:00.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/16/777616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/777616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/777616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/777616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/777616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/777616/1"}}},"commit":{"parents":[{"commit":"5be68205b14a1d6d8ab5764eca2ab228c63af518","subject":"Merge \"Use normalized_cidr of the SG in the backen drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5be68205b14a1d6d8ab5764eca2ab228c63af518"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-02-25 16:09:32.000000000","tz":0},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-02-25 16:09:32.000000000","tz":0},"subject":"Delete router arp entry when port deleted","message":"Delete router arp entry when port deleted\n\nCurrently when a port is deleted it is possible for\nPERMANENT arp entries to be left being which\nconflicts with future ports with the same ip address\nbut different mac. This patch ensures that when a\nport is deleted, all routers are notified to delete\nany associated arp entries that may be left behind.\n\nChange-Id: I74ddf5556f7cc49c70b5609b8bbac70928782a4a\nCloses-Bug: #1916761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/739de839a9217cbec6c3a44ae8ac435b92c83c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/739de839a9217cbec6c3a44ae8ac435b92c83c9d"}]},"branch":"refs/heads/master"},"511bde9d80d330d50a3c22d56108b2b5c760fa74":{"kind":"REWORK","_number":2,"created":"2021-02-25 17:34:22.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/16/777616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/777616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/777616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/777616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/777616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/777616/2"}}},"commit":{"parents":[{"commit":"5be68205b14a1d6d8ab5764eca2ab228c63af518","subject":"Merge \"Use normalized_cidr of the SG in the backen drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5be68205b14a1d6d8ab5764eca2ab228c63af518"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-02-25 16:09:32.000000000","tz":0},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-02-25 17:31:12.000000000","tz":0},"subject":"Delete router arp entry when port deleted","message":"Delete router arp entry when port deleted\n\nCurrently when a port is deleted it is possible for\nPERMANENT arp entries to be left being which\nconflicts with future ports with the same ip address\nbut different mac. This patch ensures that when a\nport is deleted, all routers are notified to delete\nany associated arp entries that may be left behind.\n\nChange-Id: I74ddf5556f7cc49c70b5609b8bbac70928782a4a\nCloses-Bug: #1916761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/511bde9d80d330d50a3c22d56108b2b5c760fa74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/511bde9d80d330d50a3c22d56108b2b5c760fa74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
