)]}'
{"id":"openstack%2Fneutron~777780","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~I2cc58c30cf844ee0ecf0611ecdec430086464790","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1916022-stable/victoria","hashtags":[],"change_id":"I2cc58c30cf844ee0ecf0611ecdec430086464790","subject":"Remove update_initial_state() method from the HA router","status":"MERGED","created":"2021-02-27 18:14:21.000000000","updated":"2021-03-03 20:39:55.000000000","submitted":"2021-03-03 20:37:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777780-bug/1916022-stable/victoria","cherry_pick_of_change":776423,"cherry_pick_of_patch_set":6,"meta_rev_id":"125f30c487e387c3acbc5fe42f50939a37cc1582","_number":777780,"virtual_id_number":777780,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-03 20:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-03-03 13:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-03-01 17:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-03-03 13:58:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-27 21:29:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-01 17:08:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-03 13:58:06.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"f5c082a23e162290727ca4996f77fa733d12a108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-27 18:14:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b405eccc00c3f46dbd353bfd2c72fb11c4062c90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 21:29:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/79d08fc13d7347bc9bd2ddb97b796bba : SUCCESS in 1h 36m 05s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/61c75e0f65c04692b523e0a4037bf1b8 : SUCCESS in 1h 47m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/3012b220a7b14426b17beadc2940a9e7 : SUCCESS in 1h 49m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d27b92bd05644f4850955f58bfae0ff : SUCCESS in 46m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3678354c574dda86a35373515b2160 : SUCCESS in 11m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f86b01301534679bc390bb795e4c1c0 : SUCCESS in 24m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/970bf60083974dd5befcb5c69ab08e1e : SUCCESS in 24m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1d896459a1a432e9a38c2dd4f216e53 : SUCCESS in 8m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48326ad8a5a646639bb88e77bc0ca0c5 : SUCCESS in 1h 06m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0219275853b4279a08763d4a7c1104f : SUCCESS in 1h 28m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/44d1349b4b9c400db3a677f23c010e78 : SUCCESS in 1h 34m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/44585e5beb044872a9123fc382725b4f : SUCCESS in 1h 12m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f4398fc5fa874c05afa738281933b6f0 : SUCCESS in 1h 20m 08s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/a06e1174881a4afc9e11c299c67539ae : SUCCESS in 1h 26m 20s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/23d0a8e9e4b647efb90271b1c7437ddc : SUCCESS in 1h 03m 00s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d08bc359f4947148a0232d9b60febe1 : SUCCESS in 1h 17m 47s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e45f2e26ab3e420cb6242f6fcd4a97ba : SUCCESS in 1h 44m 47s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e416a857cfb14c94943a17e7a9ec8d83 : FAILURE in 1h 35m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b61c5c356bbe4fd297cf94bb73ae5206 : SUCCESS in 1h 26m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3c08b1b610654adab93f2740e90be80b : SUCCESS in 1h 14m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd5c8410836242c68914f14df866f470 : SUCCESS in 2h 00m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2795989a0842471c89ee888ac7945c26 : SUCCESS in 1h 00m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ccacf77012294989af87c70da6917238 : SUCCESS in 1h 13m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c883d20b8f3d4acca0549e2e1f73550f : FAILURE in 37m 13s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2fc5d9d35a69484a8baf7b74eed760b7 : SUCCESS in 1h 19m 05s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/855e913e352b4302af143ae1056d02fb : SUCCESS in 3h 13m 20s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c3528195784947bc8398bfd27cad60cf : SUCCESS in 1h 05m 03s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/25636a05a996445595d6d91cb98112f0 : SUCCESS in 1h 50m 33s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e5565bfadd7c4dd7bb1ce2018ffc4c65 : SUCCESS in 1h 44m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f227c7d108194a0cace135159ecd417b : FAILURE in 36m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db50d87b9c3d6ade899348efe2372b3b5283d7db","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-01 17:08:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e20bfeab4f5d5c9c00f24a24a4e4805a3694b5e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-03 13:58:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f8427587c03e2d8e29272f27d34bb27270bb047","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 14:02:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e090293431522c3e8853039a88ea47d0f37bf0ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 20:37:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f4fa07434940e58b06eadf905c6486 : SUCCESS in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aee3f4f45a024aeda19022fe50886c48 : SUCCESS in 27m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a77dcf886cbd4060a4bb0bedb9d420f2 : SUCCESS in 31m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3200776a76b84d6db4265dde59380857 : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/18225265cf0a42f981d18a09c454304a : SUCCESS in 1h 09m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce6cc48ed511413eb443537386af166f : SUCCESS in 1h 18m 00s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/581c2df08d6d458abd304475b56acc83 : SUCCESS in 1h 27m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c8882b45f5514eef9f319809dc5402d9 : SUCCESS in 1h 19m 39s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ae1b74cc28e4a0797a31ab3ae94d00a : SUCCESS in 1h 21m 24s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4506a25ed23842778dd986a9be7934ae : SUCCESS in 1h 39m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/81cd25c8a8b647bf976c4b3daec8b907 : SUCCESS in 1h 18m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8bf358ce0f9b41b597872a0048f82eda : SUCCESS in 1h 13m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2183eec3207045a6803f5f235640d19a : SUCCESS in 2h 24m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f012b0980cf0473dbe253b6dade03091 : SUCCESS in 1h 10m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b29f8564264749769144ac1a5960dc4e : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":1},{"id":"84ccffb61f56c2576b9fe5b1d932bf2f257b3857","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 20:37:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"125f30c487e387c3acbc5fe42f50939a37cc1582","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 20:39:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a675d280faf04e598df8ad64a403d969 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2ad3aa5d13292c3189005d107e26685367affe5","revisions":{"b2ad3aa5d13292c3189005d107e26685367affe5":{"kind":"REWORK","_number":1,"created":"2021-02-27 18:14:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/80/777780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/777780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/777780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/777780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/777780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/777780/1"}}},"commit":{"parents":[{"commit":"16c51f3a6486d4402ce175df608c5eb4f9d8aae0","subject":"Merge \"Don\u0027t configure dnsmasq entries for \"network\" ports\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16c51f3a6486d4402ce175df608c5eb4f9d8aae0"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-18 11:48:13.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-27 18:14:21.000000000","tz":0},"subject":"Remove update_initial_state() method from the HA router","message":"Remove update_initial_state() method from the HA router\n\nThis method was intended to check state of the HA router on the\nnode and update it in the neutron server.\nPatch [1] added check of the initial status to the\nneutron_keepalived_state_change_monitor process.\nIt also could cause some race conditions and event which is setting\ncorrect state of the router will be not processed thus router may endup\nwith two nodes with \"primary\" state in the Neutron\u0027s DB.\n\nNeutron_keepalived_state_change_monitor was notifying agent about\nrouter\u0027s initial state only if this state was \u0027primary\u0027.\nNow it will notify agent always to let agent set router\u0027s state as\n\u0027backup\u0027 if needed (that was previously done by this removed\nupdate_initial_state() method).\n\n[1] https://review.opendev.org/c/openstack/neutron/+/642295\n\nChange-Id: I2cc58c30cf844ee0ecf0611ecdec430086464790\nCloses-Bug: #1916022\n(cherry picked from commit 0d8ae1576724b196580c9af3782fd3c9e9072bef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2ad3aa5d13292c3189005d107e26685367affe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2ad3aa5d13292c3189005d107e26685367affe5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
