)]}'
{"id":"openstack%2Fneutron~777782","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I2cc58c30cf844ee0ecf0611ecdec430086464790","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1916022-stable/rocky-stable/queens","hashtags":[],"change_id":"I2cc58c30cf844ee0ecf0611ecdec430086464790","subject":"Remove update_initial_state() method from the HA router","status":"ABANDONED","created":"2021-02-27 18:24:35.000000000","updated":"2021-03-11 15:09:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":777869,"cherry_pick_of_patch_set":2,"meta_rev_id":"b1bc69a085fc8364b301a4b98cccde45c8a17553","_number":777782,"virtual_id_number":777782,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-05 21:55:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-27 20:26:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 17:37:13.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-11 09:56:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"}],"messages":[{"id":"49ee00c025791e26cd0c9c66207a5382b166f450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-27 18:24:35.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"a8b5c660aef4da59beb8ed0bd94a3cbff1d8a978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 20:26:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/95f9547f25e0455aa5303f6523eddbff : FAILURE in 51m 52s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/035e6cc83f334425896972829355e724 : SUCCESS in 37m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e570232968354075a97ac0593619c988 : SUCCESS in 8m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6130d21c9f4b41da8960c430d7d28bb9 : SUCCESS in 21m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a3cd2c43ca384c0b924d995c15a17e11 : SUCCESS in 25m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/395417bdab294846a582cd8ec23bf2f9 : SUCCESS in 6m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/42c8ae15cd8a42778eaba308d6daf7a0 : SUCCESS in 1h 38m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2f58a00b65b840e3a94294b0b4cd22c4 : SUCCESS in 1h 05m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5481cc3810a426fa86648222350886d : SUCCESS in 51m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/43d2184d92ea49119e25a9d6d251805f : FAILURE in 36m 39s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/31f8f2885f5845eeb056bf35eefc0574 : SUCCESS in 1h 55m 30s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/eced207676654d2d833f5a116b9a14eb : SUCCESS in 1h 44m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6017c572635f4e1eba1334722628a79e : SUCCESS in 1h 09m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4870d633e4cc4d4985a7d5e15514f0b3 : SUCCESS in 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"eb6b50d114381cb997b8e3120969479a25a3399a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-28 06:49:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a3c9a15eba578bc2dc51f68e27adb4f63069a172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 08:34:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/65c42a5cfd8540ab8a8dbfa5fe320669 : SUCCESS in 56m 31s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/ca4ca4554db34389926b5233da303213 : SUCCESS in 53m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cfe7263887f47b7a86f2b3acf1ce6f9 : SUCCESS in 13m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3690022180e6417a90e8bee3b1c03141 : SUCCESS in 24m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/14b200ae8ef046a5a552e86663eef9ff : SUCCESS in 26m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9cb6b6a2d3404afd84092e01126c8a9f : SUCCESS in 9m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c8bed9bf321a4665af9db23eefb229eb : SUCCESS in 1h 33m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/14bd92a4744b4a16a3e5f603cb35597a : SUCCESS in 53m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee842637f6d140428a8d83d7227f3ee9 : SUCCESS in 1h 21m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/97737d15a1d04f7cb450881b98969fc8 : FAILURE in 52m 41s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/3b574716c7a849e8ae9795cf3679f157 : SUCCESS in 1h 36m 13s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6d3b9248ef3e48cf96a03c6c00381403 : SUCCESS in 1h 41m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e54546b923e64fd58b29480157cb8f38 : SUCCESS in 1h 26m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f840606aa6264147bd768d5024718c5d : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":1},{"id":"93f3507f582f237f516cc501255a79cf8485add1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 15:41:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7221a740819720631d4c1386581399e90362df89","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 17:37:13.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"197a7dd46878427019651e66df9d5166c68f969c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:55:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/715cb0f885be4c72867064c09c73ed24 : SUCCESS in 44m 23s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/a94e35c6252842cf98f1d01e96a0eae2 : SUCCESS in 57m 48s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/3e9ae953250048369a15761272391fec : SUCCESS in 58m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e351c1862bc5430e849c1293f21b1e2e : SUCCESS in 12m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74066d519340460f804be35fbda6a613 : SUCCESS in 14m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8298e17f8c6145a9acb12afa0f1d5e3b : SUCCESS in 35m 51s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/06a3dacfaf6f4d9285f04f3871ffc361 : SUCCESS in 6m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e4a05247e52a4be18b34e0eb6bc62c61 : SUCCESS in 1h 18m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/32f16c1a567c42c8a10f35dfa994007b : SUCCESS in 49m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c929d8bb79643cfbe25ffc49857f5b5 : SUCCESS in 1h 40m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7a7200c11c1247399b1934105a0db606 : FAILURE in 44m 45s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/adaea222d915417b8fcee87b43aee730 : SUCCESS in 1h 29m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/213dcfec579f42e887eb2398dd556b09 : SUCCESS in 1h 22m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2a65186d59174454b484db003c77ac45 : SUCCESS in 1h 16m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b64020d0ab674ebc874fefbee4e17e0a : SUCCESS in 23m 21s","accounts_in_message":[],"_revision_number":2},{"id":"6c28e5b3c7a2b130ae7488d5f2d59b1ebf1d6873","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-09 08:23:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3acbdfc12856d8a41203100ef7fdc663cb717251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 11:37:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/17152171d6ac4ba288a97e53c96ca87d : SUCCESS in 48m 45s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/90c359426dab45d9888304f84b2e7513 : SUCCESS in 1h 06m 47s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/3555dcd84bdb406a8245d19e3c4ed01a : SUCCESS in 40m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a842eac979864704a07e0d9e978c0a9f : SUCCESS in 10m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/051f4bed04ad4e7eb451db7a680d1b7c : SUCCESS in 39m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b2edaa20c2224e0b85c092639a0a7dcf : FAILURE in 16m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ae8a9adfc916463c9df4eea537cd4623 : SUCCESS in 6m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/22e03b144e5f4233a1dd88aca1a89489 : SUCCESS in 1h 15m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3b52b1d1a1404fceaabe395655ffc412 : SUCCESS in 49m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb50963b946343e68b76fdb3fc9773b4 : SUCCESS in 1h 06m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/138f83152a724a07959c3813ef460561 : FAILURE in 39m 22s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/a65214e6920e430aa3bdc170449aea9e : SUCCESS in 1h 43m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7a7d11722a804ea7a0762d6084578e10 : SUCCESS in 1h 20m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2ac6d5baeed74e8eb42640971a2b1015 : SUCCESS in 1h 02m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3ee49aa8f004f13b5c53afcba518648 : SUCCESS in 25m 47s","accounts_in_message":[],"_revision_number":2},{"id":"df1beffa13ed027bc9f46da520637e97863e83e4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-11 09:56:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"19f1578feac3f602e83e9842e28365c26ba46d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 12:49:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/259e2a783e844042af4d86082b775abf : SUCCESS in 52m 26s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/ed77c11680a3403cbbf8fe294a37aae7 : SUCCESS in 55m 39s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/d738aad32be740a3a1c6f38cfda90893 : SUCCESS in 53m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/431c6ca1b50c4cd98be2c4a04c547873 : SUCCESS in 11m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2488612656124fefa2f8d6f989607caf : SUCCESS in 29m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cd9814c4c0974252b8a945e0ec7535b0 : SUCCESS in 30m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/03c9f62fdabc4d5c8232fb6839da2076 : SUCCESS in 5m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4af1efd319ca44c1ba83be82c7a2c2fb : SUCCESS in 1h 15m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/115c3ddfa6f54fdfb2c37f407040cd53 : SUCCESS in 51m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2528cd1dcb924b2594c36d4685965bff : SUCCESS in 1h 16m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/346ad3882c8e4344b0d3585e6962f67c : FAILURE in 37m 55s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/22abe26e1b6d4e6fbabfa16257ea522e : SUCCESS in 1h 25m 23s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cb647cf5e61a475783719a38ec01d686 : SUCCESS in 1h 15m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/033173ed659547cd979ba7a403dbaea5 : SUCCESS in 1h 15m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db506386cee94db3adc82085e2aa5bda : SUCCESS in 24m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b1bc69a085fc8364b301a4b98cccde45c8a17553","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-11 15:09:38.000000000","message":"Abandoned\n\nThere is too many problems with functional job and that patch on this branch. And fixing of them may be pretty complicated in legacy zuul job.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c5e3316df08e988641451d8d6423cf15786eec85","revisions":{"12a3a706e47cce049213ca350e2a6cb8dca6cb86":{"kind":"REWORK","_number":1,"created":"2021-02-27 18:24:35.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/82/777782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/777782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/777782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/777782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/777782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/777782/1"}}},"commit":{"parents":[{"commit":"2e52d3f554240b74ef528f1d6bdf173f9563393a","subject":"Fix losses of ovs flows when ovs is restarted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e52d3f554240b74ef528f1d6bdf173f9563393a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-18 11:48:13.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-27 18:24:35.000000000","tz":0},"subject":"Remove update_initial_state() method from the HA router","message":"Remove update_initial_state() method from the HA router\n\nThis method was intended to check state of the HA router on the\nnode and update it in the neutron server.\nPatch [1] added check of the initial status to the\nneutron_keepalived_state_change_monitor process.\nIt also could cause some race conditions and event which is setting\ncorrect state of the router will be not processed thus router may endup\nwith two nodes with \"primary\" state in the Neutron\u0027s DB.\n\nNeutron_keepalived_state_change_monitor was notifying agent about\nrouter\u0027s initial state only if this state was \u0027primary\u0027.\nNow it will notify agent always to let agent set router\u0027s state as\n\u0027backup\u0027 if needed (that was previously done by this removed\nupdate_initial_state() method).\n\n[1] https://review.opendev.org/c/openstack/neutron/+/642295\n\nConflicts:\n    neutron/agent/l3/agent.py\n    neutron/agent/l3/ha_router.py\n    neutron/agent/l3/keepalived_state_change.py\n    neutron/tests/unit/agent/l3/test_dvr_local_router.py\n\nChange-Id: I2cc58c30cf844ee0ecf0611ecdec430086464790\nCloses-Bug: #1916022\n(cherry picked from commit 0d8ae1576724b196580c9af3782fd3c9e9072bef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12a3a706e47cce049213ca350e2a6cb8dca6cb86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12a3a706e47cce049213ca350e2a6cb8dca6cb86"}]},"branch":"refs/heads/stable/queens"},"c5e3316df08e988641451d8d6423cf15786eec85":{"kind":"REWORK","_number":2,"created":"2021-03-05 15:41:54.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/82/777782/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/777782/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/777782/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/777782/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/777782/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/777782/2"}}},"commit":{"parents":[{"commit":"a10a0cc3c4473b5062f53a8143233bab38935e79","subject":"Fix wrong packet_type set for IPv6 GRE tunnels in OVS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a10a0cc3c4473b5062f53a8143233bab38935e79"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-18 11:48:13.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-05 15:41:54.000000000","tz":0},"subject":"Remove update_initial_state() method from the HA router","message":"Remove update_initial_state() method from the HA router\n\nThis method was intended to check state of the HA router on the\nnode and update it in the neutron server.\nPatch [1] added check of the initial status to the\nneutron_keepalived_state_change_monitor process.\nIt also could cause some race conditions and event which is setting\ncorrect state of the router will be not processed thus router may endup\nwith two nodes with \"primary\" state in the Neutron\u0027s DB.\n\nNeutron_keepalived_state_change_monitor was notifying agent about\nrouter\u0027s initial state only if this state was \u0027primary\u0027.\nNow it will notify agent always to let agent set router\u0027s state as\n\u0027backup\u0027 if needed (that was previously done by this removed\nupdate_initial_state() method).\n\n[1] https://review.opendev.org/c/openstack/neutron/+/642295\n\nConflicts:\n    neutron/agent/l3/agent.py\n    neutron/agent/l3/ha_router.py\n    neutron/agent/l3/keepalived_state_change.py\n    neutron/tests/functional/agent/l3/test_keepalived_state_change.py\n    neutron/tests/unit/agent/l3/test_dvr_local_router.py\n\nChange-Id: I2cc58c30cf844ee0ecf0611ecdec430086464790\nCloses-Bug: #1916022\n(cherry picked from commit 0d8ae1576724b196580c9af3782fd3c9e9072bef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5e3316df08e988641451d8d6423cf15786eec85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5e3316df08e988641451d8d6423cf15786eec85"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
