)]}'
{"id":"openstack%2Fneutron~777977","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I97ada62484023b9833ed12afd68eb4c8d337fd1f","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1896734","hashtags":[],"change_id":"I97ada62484023b9833ed12afd68eb4c8d337fd1f","subject":"Improve \"get_devices_with_ip\" performance","status":"MERGED","created":"2021-03-01 10:32:15.000000000","updated":"2021-03-04 00:49:31.000000000","submitted":"2021-03-04 00:46:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777977-bug/1896734","meta_rev_id":"bd4353a75f61987dfd4f9cb2e0144bde1f568bc8","_number":777977,"virtual_id_number":777977,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 00:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-03-03 09:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-03-02 19:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-03-02 07:27:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-03-03 09:28:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-01 14:44:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 07:27:34.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-02 19:40:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-03 09:28:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"dc283b82d2be25c2c12dae2c6c7d7daef31d31e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-01 10:32:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8df0a80f4ff97cfe270caac56d9590dfd8d656b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:44:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/023e6db65fc94cfd98f3cfc0ecd45524 : SUCCESS in 1h 05m 21s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/ff4523ff85704222a3d6f427aaf8bc73 : SUCCESS in 2h 15m 06s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/3a571a327d4b4f66b54edc7df907a88c : SUCCESS in 1h 09m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/1423131706634d468d94832b637f9b43 : SUCCESS in 1h 26m 42s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/26fa6bf28ef64985a7f1a113e4937740 : SUCCESS in 1h 29m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d2a9781a69542a39e2dfb1b4b8e3374 : SUCCESS in 34m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae115d8d39594b6ab14d048fc206d757 : SUCCESS in 14m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1cc35d943d3457792e99e6b450188b3 : SUCCESS in 23m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5f16f09be91402c91c4aa4386141498 : SUCCESS in 32m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2287fe9d728476986438663831c32ae : SUCCESS in 31m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/567b1fcb01074e0d8099a5f4fe48c94b : SUCCESS in 9m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ed748ec5117443fc880ed5b251c98a06 : SUCCESS in 56m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0469b762b5a7404a8e1312200a202426 : SUCCESS in 1h 10m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f592219fcb94d86982330b7d41a32a4 : SUCCESS in 1h 16m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/351ce77844fe40268e1326975546bf0f : SUCCESS in 1h 52m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2810e288703a488f9ccf272b2a942190 : SUCCESS in 1h 25m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e9dc6f8164b048efa80c991a8632c1e1 : SUCCESS in 1h 31m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c88d5a4d61a948369f6b5267626180e7 : SUCCESS in 55m 55s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8baadf8f652849c8a0f1cffea01311de : SUCCESS in 1h 09m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8500a63f57f0459a8f6974ddcef6ff38 : SUCCESS in 1h 17m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9ea683c0a69c42a592f8c64f44a38a0d : SUCCESS in 1h 45m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a44406c67b304c34be77600581459bf0 : FAILURE in 2h 04m 53s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a60d08f1a7a342f2a06f03af05099975 : SUCCESS in 1h 13m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4419958fc7ad4a52b2ea17056c67bc43 : SUCCESS in 45m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/49941d3074ed4174a8c2cb242205797e : SUCCESS in 2h 03m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c642389bc731453ca8e48fb51fd3f2e0 : SUCCESS in 55m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c0927ea98d8f4023928792cfcd4b68e4 : SUCCESS in 1h 23m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0023b3a32175494a9fea3d90da5f7463 : SUCCESS in 37m 51s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0b1651bc13544cda8c01c244929a13ea : SUCCESS in 1h 15m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6609650baa234be1ae11d2b2b4aa11f4 : SUCCESS in 52m 48s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/df34d52559534e9e8b9c4578db1cf8d7 : SUCCESS in 3h 26m 35s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1a3907fb343b499d92e281c0d449ccf1 : SUCCESS in 1h 17m 29s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/450fa462d8194d42997a63f8be17742a : SUCCESS in 1h 57m 31s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/26dc12a9f9a44c60978ae7c77795dca0 : SUCCESS in 2h 06m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9e033394111d491ea790241bc412ffeb : FAILURE in 1h 54m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5c1eab9c5b9108d02be5f37b2be542e9bd866f4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-02 07:27:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7001a8223444a3804d96a0655c67203853cec848","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-02 19:40:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"446f8f17d9a2854fa9f62cbe2cd9e8a8f6b264b4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-03 09:28:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b930c452a61001ba3d36c843d4a77c1b7e28e1c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 09:29:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"618ba07c8583fe4aca7bfdf05dc12a8ac0d79248","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 12:53:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/a7f016193a2d40fb958d729874118873 : SUCCESS in 1h 03m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09bd910be3484282a01dbfb0e9b2f064 : SUCCESS in 13m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e30e00946f18468ba98f271bc9517c52 : SUCCESS in 25m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80723b8a903743c794464784f5f60323 : SUCCESS in 27m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aba10f74381f4f8fb788768584788c13 : SUCCESS in 12m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0b136522bed44dfbbdf70e4e5ce680dc : FAILURE in 57m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b15729d674d5466c9d0202448f93d104 : SUCCESS in 1h 21m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f706ab0d57614d10b16d6e45b9758bd0 : SUCCESS in 1h 06m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a3225e1adabf4f5797f0a81fe1d4bd0a : SUCCESS in 1h 08m 00s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d9721bd42df24102b750b3644fb4ac38 : SUCCESS in 1h 33m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3899fcff2c64798865efca320235c0c : SUCCESS in 1h 20m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9767e6507e9840c2baf48eb3355f3e6b : SUCCESS in 1h 41m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86d13daffe004fa6bbe71f76aa4294ff : SUCCESS in 2h 15m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78925207b58148618278a718a7c58056 : SUCCESS in 56m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0657d725d76c493cbc7c728784ca447a : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5f695ed3d25ed258b246cdbeaf65a9ff41c4a5d9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-03 13:55:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"598cbb1c91f68d4fe290e27c948f12146c1e4447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 17:55:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/be875a659fb74a6eac3b800f80a1dc3b : SUCCESS in 48m 19s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/754dfc3a2b104059985e5fb711d2f9a0 : SUCCESS in 1h 31m 41s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/298c21de11884b43b2a294c6c1e84c70 : SUCCESS in 1h 35m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/8b4da9ec56294733b83a266652111323 : SUCCESS in 1h 43m 01s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/189711a9dba6450e8bce86492aec9b6b : SUCCESS in 1h 29m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3670477391ba4a26997deca19feaf8f6 : SUCCESS in 35m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84e6ec0ef15e40e2873021068b22952f : SUCCESS in 16m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f50515df944c431e9544e2acda33886c : SUCCESS in 24m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce4b21acccdd4ac281cc33ca21dee355 : SUCCESS in 25m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee208cb6c10b47ed9c5739b4cbee21fd : SUCCESS in 28m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3969efc48b134ee5adb1c360552af5af : SUCCESS in 10m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/da7f878153cb464a9818690511d859ac : SUCCESS in 52m 14s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cd012b01c1a24fca99bf5eeaa86e4501 : SUCCESS in 1h 14m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fde1ffcdcc342b99f4fac0607e6d095 : SUCCESS in 1h 15m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b452bf8bc84f4d26a5335e00a5479bbf : SUCCESS in 2h 06m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f9b3bd1b3dc4fe283f939b486d7fcc7 : SUCCESS in 1h 41m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fd0ee049fe124202bb795478fe85e41a : SUCCESS in 1h 28m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2491814c32fd43c8b8b556aaab20129f : SUCCESS in 1h 28m 02s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/065b7de7e41b4265a7499a317a499287 : SUCCESS in 1h 24m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/acec446db5a64a479b9ffd71c130a846 : SUCCESS in 1h 20m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/63096c76b40046e18daa7a3d03c26540 : SUCCESS in 1h 41m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4152a90e54994d39a33fefc7a319c123 : FAILURE in 1h 55m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cb69ac6b647c4eefb8a6d5388ec7c9c4 : SUCCESS in 1h 31m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d65a08e9c62f45fba746ac8e81530c54 : SUCCESS in 1h 10m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/02970036d6474149aca041eac70ef087 : SUCCESS in 2h 20m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de7a6a70071d4d91a82f07bf3e1fb722 : SUCCESS in 1h 00m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e2aa5c3a0dab4720977209ea1e1df6f1 : SUCCESS in 1h 14m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7f27685eb6344ba891357e325e3e0ce8 : SUCCESS in 41m 42s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a01deb257ac44c4c89fdd1c6eef4aac6 : SUCCESS in 1h 39m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1987d6b7ca964ab09a709ec8de49fac1 : SUCCESS in 54m 49s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/21a6e77c9f1b43f5ae65bae45cd7f4a4 : SUCCESS in 3h 24m 00s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/66e4347698644e48a383288b1ebc9dc3 : SUCCESS in 1h 24m 36s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b3c4fe89cb74a8f832110f6da6364d0 : SUCCESS in 1h 42m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0c7c2807d89a48f9a7505a2f4d39aff2 : SUCCESS in 1h 58m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d4a7b6995c5b4f0498c08c8877a45b1f : FAILURE in 1h 51m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f2f22495c8715f90aeb718b89f5eb9754185898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 20:02:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c88da7ad840217c334ccc2c7270b2ea65f3f437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:46:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/c8c36b062a22453c9f5df3f6c940f780 : SUCCESS in 47m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00919db222cc4ff6b84456699befe389 : SUCCESS in 20m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d07d56b7e9a49b9aa55e04764b05fce : SUCCESS in 28m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a57e051297af428b83e6808252f8c5ae : SUCCESS in 31m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06dd7f1c59974a0785a87c2be9ae4220 : SUCCESS in 10m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/27afbddfa6b849e3990432a6f118b337 : SUCCESS in 1h 00m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fa0b47f79e474129a9278a20052d4780 : SUCCESS in 1h 17m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a4e15647d1ae4a3089818a9b70e0fced : SUCCESS in 57m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d635ff4fa13d42b989da1f4456b02934 : SUCCESS in 53m 25s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d0578970591e412e8c4351ef8cacd201 : SUCCESS in 1h 13m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0625cf0ae624ae684fe0af8950bf8dd : SUCCESS in 1h 16m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dcd84c3816aa43358bde14f6d1032bac : SUCCESS in 1h 19m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6521b6ec1c640b4baae65e9b7ea1385 : SUCCESS in 1h 59m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76d647d8e66e4878b5ead98243e60e2a : SUCCESS in 55m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/31e25c89fb1a4328a18fd01ec3008d8d : SUCCESS in 1h 26m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9ee1e6350709fe7b31186ded52457ee6aeb63c3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:46:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bd4353a75f61987dfd4f9cb2e0144bde1f568bc8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:49:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e18ae824de741999f760b76078042dc : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f4ef314ce2b1a4c54a4216ef0dc8068c51e5981","revisions":{"2f4ef314ce2b1a4c54a4216ef0dc8068c51e5981":{"kind":"REWORK","_number":1,"created":"2021-03-01 10:32:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/77/777977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/777977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/777977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/777977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/777977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/777977/1"}}},"commit":{"parents":[{"commit":"83fbb1db5ae573b6be755af2a474dc9b694e5534","subject":"Merge \"Avoid race condition when processing RowEvents\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83fbb1db5ae573b6be755af2a474dc9b694e5534"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-24 14:46:10.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-03-01 10:31:58.000000000","tz":0},"subject":"Improve \"get_devices_with_ip\" performance","message":"Improve \"get_devices_with_ip\" performance\n\nAs reported in LP#1896734, there is a limit in the size of information\nthat can be transmitted in one single message between an application\nand the privsep daemon. The read socket buffer is limited in size;\na message exceeding this size will generate an exception.\n\nIn order to limit the amount of information to be sent, this patch\nimproves the performance of \"get_devices_with_ip\". In the previous\nimplementation, the whole list of network devices from a namespace\nwas retrieved. In some environments, the list of devices could be\nso big that the list returned by \"privileged.get_link_devices\" can\nexceed the read buffer size (as reported in the LP bug when the\nOVS agent tries to retrieve the list of IP addresses in the system).\n\nNow the function calls \"privileged.get_ip_addresses\", that returns\na much smaller list. This patch is also reducing the number of system\ncalls to just one; the previous implementation was retrieving first\nthe devices link information list (that method was returning a much\nbigger blob) and then, per device, retrieving the IP address\ninformation.\n\nChange-Id: I97ada62484023b9833ed12afd68eb4c8d337fd1f\nRelated-Bug: #1896734\n(cherry picked from commit a6cdf273c06a3bba713d7b8d0ef3a9b71b0ffc57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f4ef314ce2b1a4c54a4216ef0dc8068c51e5981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f4ef314ce2b1a4c54a4216ef0dc8068c51e5981"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
