)]}'
{"id":"openstack%2Fneutron~778080","triplet_id":"openstack%2Fneutron~master~Ie91a6a50fbda806f0c6ccafa9d3b38083715df65","project":"openstack/neutron","branch":"master","topic":"bug/1917370","hashtags":[],"change_id":"Ie91a6a50fbda806f0c6ccafa9d3b38083715df65","subject":"[Functional] Mock classes from the ovn maintenance module","status":"MERGED","created":"2021-03-01 22:04:03.000000000","updated":"2021-03-03 01:04:40.000000000","submitted":"2021-03-03 01:02:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778080-bug/1917370","meta_rev_id":"c8a972c55fff7cce36f478664b4d571ba00312e4","_number":778080,"virtual_id_number":778080,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-02 00:16:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-03 01:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-02 15:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-02 15:05:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-03-02 17:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-02 11:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-03-02 17:00:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-01 22:04:44.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-01 23:27:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-01 23:30:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 00:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 11:14:49.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-03-02 15:05:36.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-02 15:48:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-02 17:00:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"bd5084a1b14575eb5c0c33c28d03b2d7c429e253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-01 22:04:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e4cd4c831a83a06656d168deec7f18961739eb0","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-01 22:04:44.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"8a3d64ba1d1a6c1888bb40c83d9351c7790c42dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-01 22:37:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"380fe0412774159429c99dcf3596ccf1dab2c8a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 23:27:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d809e27f109d4739a7d6a88468c43c2d : SUCCESS in 35m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98eec7bdd78b4361bcadc3cf9f2e17f9 : SUCCESS in 40m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"628187e42cdd5c86a0002a819045737c453916d7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-01 23:30:53.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/80/778080/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"803cd278216d5394c21ef7d4ba7bab959df4b18b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-02 00:11:29.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"1e6c07fa39241ac637aeea6505199230c7782f63","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-02 00:16:25.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/80/778080/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"c9c0a37bb6526400ea12068d4f20f829e0f51c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 00:28:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/67575e8971914e60b6bfeedf0f88018a : SUCCESS in 48m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1e03404857cd4a9bbd3be201a8e9ebfa : SUCCESS in 1h 36m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bec56a7812824cc2a1b5bfee7302a12d : SUCCESS in 1h 48m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12e96a7ccbbd453597ca70f62487c225 : SUCCESS in 1h 31m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0f76773962c942a7b9d9699b69dc24d3 : SUCCESS in 1h 40m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a4323f30b494867b424012d395773fb : SUCCESS in 39m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d206e7a5b3c74a33860b5c669a744ce7 : SUCCESS in 29m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c0a67de34249be9be5917616f8a14e : SUCCESS in 14m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe47542610744faf8a39a33c1813f319 : SUCCESS in 29m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4fb71bd8d954ee49e0760459e322f4b : SUCCESS in 33m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/786cca9859f7491e9079b8bb3676e5b6 : SUCCESS in 34m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed36f6711114556b2d64765f28bc98b : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f8024e2248448f8a1c0c0dfa4c648dc : SUCCESS in 52m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea9df92e63d84c949981d54427b6b559 : FAILURE in 41m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0d877a93ae234c99bf4d384781732186 : SUCCESS in 1h 46m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3aca3cb9c2fa4325a6bfde89f141747a : SUCCESS in 1h 02m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6cf8611091cb4fb6999670e2a1d0c41c : SUCCESS in 1h 06m 11s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/483c5b35cd154c028e3bfd473d675bb6 : SUCCESS in 1h 23m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/efd8285ac3b4494289bde7a57b63c4a5 : SUCCESS in 1h 32m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89fae5a4b96a4512a5b75782f59a2a07 : SUCCESS in 1h 41m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27932b6d184345ae92c3b3defb775364 : SUCCESS in 59m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b256d693fb514cb1b2470cc9968526f9 : SUCCESS in 1h 07m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f06ef7ee14c647e5805e876a50c7a70b : SUCCESS in 35m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f102cbd80ed4e96ba64304f3249bd3d : SUCCESS in 1h 41m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4e766a14886d4ecdab47582c3b48fcbc : SUCCESS in 1h 25m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f293af2a9b27e3b69740dc5b2db36073a3f9f9b0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 07:50:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3fe388b257fd9629c41d4f0b826ccc0e5f2c1f88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 09:02:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4410cc7cb5d4ee8ac89aa980fbf1bbc : SUCCESS in 40m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33ccf2bb52054039b80f0834d67c84cd : SUCCESS in 39m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c808d597dec28088c9f70b853f8dc76adf2a8728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:06:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/02323322eef64088999ce8a5ffb39b21 : SUCCESS in 47m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c0df22d4fe5d4fd68635b14fe8dd5cdb : SUCCESS in 1h 22m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c5f8c1790782465c89218e0e11548cdf : SUCCESS in 1h 17m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e3cbb86644b6436497e17f87f8372cdf : SUCCESS in 1h 26m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/62014dbbe72f42b7bdbb23e04b76f4c3 : SUCCESS in 1h 35m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81188a79112140649f379095c751dae5 : SUCCESS in 24m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb5e509c668646fd93e4bcb9376f020c : SUCCESS in 25m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5043a53d1dc4673b974f468591c2baa : SUCCESS in 15m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d79945a313f4685afa0fcb1f256522a : SUCCESS in 25m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98e64ed2888145989c279763f9512b2a : SUCCESS in 20m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9fd3e7265344a499e436299fd3441a7 : SUCCESS in 25m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb36fbdcaa8e485f8ec6a4c29985a703 : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/229e27a6d3f9487fa9e6817b960b9714 : SUCCESS in 46m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ce6c500d9124a229dec06eca4647449 : TIMED_OUT in 2h 14m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bb7af641227b49a29307847d97437a82 : SUCCESS in 1h 35m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e306aab18144f088514293b76b42117 : SUCCESS in 1h 01m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/922593cecc904eb5b0ac0dd2974cbd14 : SUCCESS in 1h 05m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/de32e99b273f4644802f8a20e14c5d76 : SUCCESS in 1h 16m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bbbc7cc755e84df9b00316e30c259d9a : SUCCESS in 1h 30m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f140e89c8d84a7f9c1493786dee4d09 : SUCCESS in 1h 37m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab53572c3a83420db2e4ede9e66092fa : SUCCESS in 59m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/76fbf7c7705a4d2dabf92d2ed868b838 : SUCCESS in 1h 33m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9b15cd5b254f45948656d7abf0fb51b8 : SUCCESS in 33m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a955613b56c474ba302685142a68747 : SUCCESS in 1h 59m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/dfcda5c3a34447e5a9754d51b2437615 : SUCCESS in 1h 22m 10s","accounts_in_message":[],"_revision_number":2},{"id":"91106522e62deca6bb043447d9ee4330da0d0612","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 10:59:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fa9bde12f3752f7752c72495238c5d4c4911e8cc","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-03-02 11:14:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"77bc46ea17eaea44413a869807190e98e65ff904","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 11:55:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aade2053d1434907bf829393ac6a0cd9 : SUCCESS in 37m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/768f26b8addf482fbfe1ddd52799e237 : SUCCESS in 46m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aad67d3e9e829460a3c8779afa8af868d4e2d1ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:42:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/10c6c916e935448a9769c160167f15c2 : SUCCESS in 48m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d2ff8c69ef134731a157d8a59c42535e : SUCCESS in 1h 43m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7a72c8b98c29469ca4c2d755290554dd : SUCCESS in 1h 50m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f2a8df70c4a64d80b86f6baa2d7c230e : SUCCESS in 1h 41m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3d5b1a8d6ac546dcb482112f22f5dcc7 : SUCCESS in 1h 45m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0422dd7590d6485c930459da0db3df3b : SUCCESS in 39m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8523407ed5b94cc1933ed53e3ad00814 : SUCCESS in 32m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936e74e026924f70a8a5c5071676b2c2 : SUCCESS in 13m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66d362e7cc51432c82c6c31ef692356a : SUCCESS in 29m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f77d9e93ba44c5fab9d6c2672bab758 : SUCCESS in 32m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00134cb1e67a436393de089d9246ba5c : SUCCESS in 33m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de247d9169f46678ff59a6253f201cb : SUCCESS in 9m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/875857e7174540ba8b824ee269677cbc : SUCCESS in 49m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c90b06415e974342b93f370c9343c4c3 : SUCCESS in 1h 14m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c530e621be8f4a9bb6fa7e6f0e0cd973 : FAILURE in 1h 40m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/afde111c52e5400a915a05e7727e4c05 : SUCCESS in 1h 20m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/04893723e0f047af8a31241e6d09aab9 : SUCCESS in 1h 03m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a2e3fa6f8e734f778d86ec004f3f81cf : SUCCESS in 1h 23m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f38ee9f08b334a9090198a6b9e0f312b : FAILURE in 1h 13m 10s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72fe390f3dc14fc59c2fb36d9e9a9578 : SUCCESS in 1h 49m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34364e06a3e9421a8d932c59d364f1b6 : SUCCESS in 57m 20s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4856375f7c604aaf9f97a658ebecce2f : SUCCESS in 1h 04m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/93b2572fca6d408fbb5daa0be4d09550 : SUCCESS in 33m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bbaef5abfa1b43e6a5c9fe424e2b7906 : SUCCESS in 1h 41m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a66175887c454c37bcd415b01511eafb : SUCCESS in 1h 30m 40s","accounts_in_message":[],"_revision_number":2},{"id":"78a9c1623146bc0c7f053394c08ad3a282609df2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-02 15:05:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a5c08e556b7ef642d1f44684cee0ab97ad38a1ab","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-02 15:48:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eca8f6c3e3482fd946543d70a4e68b9e55651656","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-02 17:00:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e6c5f4eac49305909ac9b887bd963b6571ab4ae","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-02 17:01:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"96c66eb245e5991cd8de4802b50453c56d00515c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 17:53:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8088257bdeb14b48b51fc414f322e79f : SUCCESS in 41m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f036f233f5fc40b9a60b73abc5364959 : SUCCESS in 46m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f542641a48f553d7f48274ebd38a6ea41fae029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 20:08:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3da13c681a5f4f26988aa9c822567e03 : SUCCESS in 51m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8ee08347d6d94c16829c07479da3d489 : SUCCESS in 1h 30m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e5f0e97f29fa4c979c1d3f63bdef8c9d : SUCCESS in 1h 31m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0d42ebfd54d14c47b8e2bf5c1b7e9df6 : SUCCESS in 1h 15m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8131367677994874ba176d58f7dce606 : SUCCESS in 1h 41m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bddcda6954cc4869a520a5ec81c0abbd : SUCCESS in 38m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3815d1d7cbf4d80aa6af91a308ecce5 : SUCCESS in 32m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f59cb16078542b4a5bc7c4d111e1f5c : SUCCESS in 14m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bef3e4d61c8b488587af00f0ee10a80d : SUCCESS in 33m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ebee34ae0f04039ba168ffd7fe9809d : SUCCESS in 28m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae979d5a3a2240d9a66c3bb0d86daca6 : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d494b9e939b743afb50b30d793968106 : SUCCESS in 13m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e2e13d44c3c49899fc88d9eb6e91b57 : TIMED_OUT in 2h 14m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4f140770b73452e969f21a9c33aad4c : TIMED_OUT in 2h 17m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1b68eb3f250040628013dd66b13c1f0f : SUCCESS in 1h 32m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/47c751cda15c4e5fae4e73ef2d26d3c8 : SUCCESS in 1h 15m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eb6d8e58d56d47cf8693d5fff81c44eb : SUCCESS in 1h 03m 46s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9a475649a2f74d7b970941e9d907b527 : SUCCESS in 1h 14m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ecd4bb7481e544eda3e26a54572d8e52 : FAILURE in 1h 17m 28s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40744cf7e23e4fbf98c399bec1773edd : SUCCESS in 1h 46m 54s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e72d0d5234b74fc88c7bd564a801faca : SUCCESS in 58m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3ce19f85326b4b8ab36137b9b9c1bacc : SUCCESS in 1h 03m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/72012382fd65422b96247fb1715f35b5 : SUCCESS in 37m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0c845edf58844624888198a23cb5c2b6 : SUCCESS in 1h 30m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4d6c3357e2074069ad92e0fea7e8e7e0 : SUCCESS in 1h 25m 19s","accounts_in_message":[],"_revision_number":2},{"id":"0eea3fdc310bbf9eee95e412a8169aa4c44743a3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 21:13:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7e8b76e886448e3155dbf1114831f32e461c73f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:58:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ebca1bbd148f42498f0421b8a5be4659 : SUCCESS in 35m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17ac2c38efba4bc08fe6ba503b18456e : SUCCESS in 36m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"390a8cb09dd1e69fea49322e2e4097d86bfe40c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 23:12:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1cecb492e1b9447984efecb678c5d557 : SUCCESS in 1h 03m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6b8b75fad4c24946987ce73c6c4bda8c : SUCCESS in 1h 42m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6becf26306c64832832a2f893c9d3106 : SUCCESS in 1h 46m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c14b153c7a374d0ba9bac4a16ebeb40c : SUCCESS in 1h 11m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/77c06a12698c4daf9168f54914b6b0a7 : SUCCESS in 1h 58m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8fbefa6d2a14f6895df3069d3e01e12 : SUCCESS in 22m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cc1bd1a70254989bf18d6f4303cccc0 : SUCCESS in 48m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcbaf9cd246e48a0907666edb5a5e478 : SUCCESS in 10m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0855db3b21d4f638b41721dba5fcc19 : SUCCESS in 29m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79c22a68dc3c4618a8167c26a7f06ac8 : SUCCESS in 32m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49c4240b1a884b6bac2d000c6d063176 : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d06323597c2407782883b6fcc596529 : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da181155186140fc8b2bef88ef43ecb9 : SUCCESS in 47m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a16661ad79ab4af89b57c343aec17c6f : SUCCESS in 1h 09m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7ff4f7d7cd3648d88fb27f0b82e3ff6f : SUCCESS in 1h 44m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/76b2d4e182b54999823510af40687038 : SUCCESS in 59m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ac9a8b16408246e0997bafbfef033220 : SUCCESS in 1h 16m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/03493144f17646db8cc3ea7b09b94a14 : SUCCESS in 1h 30m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9f25545bec1f46ee9d4f61d0bc4a84e2 : SUCCESS in 1h 19m 02s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e10c3fb9eb144581a6ecb1b125f4fa97 : SUCCESS in 1h 16m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfa0dd6e16b04bb6a986551f50daeb63 : SUCCESS in 1h 30m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fb60e844e8d4425eb06c1cadc715faec : SUCCESS in 1h 00m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/402299a95a1d4e0c9c39eb0c9841d37b : SUCCESS in 27m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f17e78da4504909b0bfb158c152cb39 : SUCCESS in 1h 37m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/780977eb76be464796814b1252d6db18 : SUCCESS in 1h 08m 11s","accounts_in_message":[],"_revision_number":2},{"id":"434ba44269dc1ebc18a8f6f1dcb7c7b1d166a2e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 23:13:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0bd54f55a4b5b7f949d11643a24e192e404806c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 01:02:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/35d8925a5ee64addbccddd23f0f737e7 : SUCCESS in 45m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/86f95184af2d4462af9d31304b676e3b : SUCCESS in 1h 36m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8408d0604284427fbdaa6cc1906e6126 : SUCCESS in 1h 14m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/68ef95f403084e6298304aad349c9000 : SUCCESS in 1h 47m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ed09ee7aa4a640ae9dec455ebcfc767d : SUCCESS in 1h 18m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba5b11eaa4f343e7b2704ed6d8b9efd4 : SUCCESS in 25m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25445631ba194c24ae5de0ef4097c1b2 : SUCCESS in 10m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40b59ddc26c14e69aa9d32ffbc3fb2c8 : SUCCESS in 26m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/543594d4b82e48718434bc294009d0a3 : SUCCESS in 23m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3357293eafee43e29f6318784c2a9a1b : SUCCESS in 9m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/523384d46cc14d8a9d5513a6b4f51d6c : SUCCESS in 47m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3b1272498c94a36a3331cb69c340e10 : SUCCESS in 1h 26m 27s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f0da140e3c92407bb13510c2b6cf2fc8 : SUCCESS in 1h 19m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/61694c69c931482ea13346f1f2c93fe9 : SUCCESS in 57m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1645c16ae1047d881b9d39fa7dd863c : SUCCESS in 1h 01m 12s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3857b32bdcb841fc9d571aa44daa8a32 : SUCCESS in 1h 38m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24da122716f145e780745d927c8e87e9 : SUCCESS in 1h 00m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b741754c80234aef952d700f8198a587 : SUCCESS in 1h 08m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5619cc4f474e49b39e2bcf50e745c2fc : SUCCESS in 30m 26s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a9287f096bf24560bdbd6031927a7701 : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":2},{"id":"77474ff7910efd4c2fee9759dc631fee8132d147","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 01:02:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c8a972c55fff7cce36f478664b4d571ba00312e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 01:04:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27bc29b1127d47b790760760142f691f : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"952c84779571a27777348aea80532b471e8ad01a","revisions":{"53c6de6090e70236b25a21c91f8da34c63a970c3":{"kind":"REWORK","_number":1,"created":"2021-03-01 22:04:03.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/80/778080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/778080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/778080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/778080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/778080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/778080/1"}}},"commit":{"parents":[{"commit":"77ee0847f5fe886ae6a1c4a9b9949d3590567e0d","subject":"Merge \"[FT] Reset \"AgentCache\" singleton in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77ee0847f5fe886ae6a1c4a9b9949d3590567e0d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-01 21:49:58.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-01 21:49:58.000000000","tz":60},"subject":"[Functional] Mock classes from the ovn maintenance module","message":"[Functional] Mock classes from the ovn maintenance module\n\nClasses MaintenanceThread, DBInconsistenciesPeriodics and\nHashRingHealthCheckPeriodics are now mocked in the TestOVNFunctionalBase\nclass if maintenance_worker is not needed.\n\nCloses-Bug: #1917370\nChange-Id: Ie91a6a50fbda806f0c6ccafa9d3b38083715df65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53c6de6090e70236b25a21c91f8da34c63a970c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53c6de6090e70236b25a21c91f8da34c63a970c3"}]},"branch":"refs/heads/master"},"952c84779571a27777348aea80532b471e8ad01a":{"kind":"REWORK","_number":2,"created":"2021-03-01 22:37:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/80/778080/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/778080/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/778080/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/778080/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/778080/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/778080/2"}}},"commit":{"parents":[{"commit":"77ee0847f5fe886ae6a1c4a9b9949d3590567e0d","subject":"Merge \"[FT] Reset \"AgentCache\" singleton in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77ee0847f5fe886ae6a1c4a9b9949d3590567e0d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-01 21:49:58.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-01 22:31:09.000000000","tz":60},"subject":"[Functional] Mock classes from the ovn maintenance module","message":"[Functional] Mock classes from the ovn maintenance module\n\nClasses MaintenanceThread and HashRingHealthCheckPeriodics\nare now mocked in the TestOVNFunctionalBase.\n\nCloses-Bug: #1917370\nChange-Id: Ie91a6a50fbda806f0c6ccafa9d3b38083715df65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/952c84779571a27777348aea80532b471e8ad01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/952c84779571a27777348aea80532b471e8ad01a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
