)]}'
{"id":"openstack%2Fneutron~778149","triplet_id":"openstack%2Fneutron~master~Iec949922be0779a00bf405c2f1d5646c588d323c","project":"openstack/neutron","branch":"master","topic":"improve-neutron-ci","hashtags":[],"change_id":"Iec949922be0779a00bf405c2f1d5646c588d323c","subject":"Reduce number of the workers in fullstack gate job","status":"MERGED","created":"2021-03-02 08:50:28.000000000","updated":"2021-03-02 21:46:10.000000000","submitted":"2021-03-02 21:43:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778149-improve-neutron-ci","meta_rev_id":"65e434a2a5b3b428b35baa72d9ccbe2636acaa1a","_number":778149,"virtual_id_number":778149,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 21:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-02 15:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-02 15:29:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-03-02 19:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-03-02 19:35:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 09:02:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-02 09:37:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-02 10:05:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 15:29:12.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-02 19:35:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"415e8cb3803603435773a5f83f45451883032e86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 08:50:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5ea01e89ec6640275050e55764f358903df334f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 09:37:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11d214b502f64c4c980cd921679b150c : SUCCESS in 41m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34fa44f442d14c0daef06bf641e1fc30 : SUCCESS in 37m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06efae29a5c0f05ef880e35321a5a8031f1f1f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:05:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/561cfe6f37be4cb694380b5485ef32a4 : SUCCESS in 26m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/026f213c3af34a03a8ef538fa7e90d8b : SUCCESS in 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dee43fdc98b548ada041e796ef1c1e53 : SUCCESS in 10m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a89baeb2eb7d4db9a4d7b2513e267b1f : SUCCESS in 18m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/099d4b0d319d445aa85b928d3ed7122b : SUCCESS in 18m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8556a9975b64fd9b5bd71c0575f5777 : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73013a76cb4f4a6e88398f8d99bfd4b7 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/422b0f1b8eeb4757b29329a23bb2ddf5 : SUCCESS in 7m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/348fc57d71cb481489d308666aa15a38 : SUCCESS in 43m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c5f9be3b077403a86187e71d704592c : SUCCESS in 1h 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"53ef03c2cf3bde812c7c8d055b0b713309745c4d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-02 15:29:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"234bd896ab5898c6f046c2631a6c42d0611557d0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-02 15:29:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"10dde8f9e52e1797ca43f7205b48dd80504977a5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-02 19:35:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da6028894c0c77b6fd84f8d7176ff46a0598e102","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 19:36:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"239a843f55bf6195d6398b3c3c5fc38f60ccec4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:43:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3654b87aa09b49a19d118e643a151e2d : SUCCESS in 33m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd85396a7e9b4977b55021bb6d52e126 : SUCCESS in 13m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7aef97ca5384d0caf84c5169b0a62f9 : SUCCESS in 25m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d9282ae2ebf4734818974aa93eb6884 : SUCCESS in 17m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c457398eaff148098862db5e52e8dca1 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7e649e270fe44ce8c3d777b7eea8a86 : SUCCESS in 12m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/97c840af184f4832aed1d59f592a94ee : SUCCESS in 58m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/023f1be5f83046ecaa2264eb0e50dbbe : SUCCESS in 1h 35m 11s","accounts_in_message":[],"_revision_number":1},{"id":"007324a0369daded01ac214c59b5f20d4771b037","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:43:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"65e434a2a5b3b428b35baa72d9ccbe2636acaa1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:46:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/324f23bb432c456281f9eea334267b4b : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bab6bf378ddc4fce8dcc9b02471bd207 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9542b6a0164cbe0c0fd32b15d0d5db652d47a2f","revisions":{"d9542b6a0164cbe0c0fd32b15d0d5db652d47a2f":{"kind":"REWORK","_number":1,"created":"2021-03-02 08:50:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/49/778149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/778149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/778149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/778149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/778149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/778149/1"}}},"commit":{"parents":[{"commit":"77ee0847f5fe886ae6a1c4a9b9949d3590567e0d","subject":"Merge \"[FT] Reset \"AgentCache\" singleton in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77ee0847f5fe886ae6a1c4a9b9949d3590567e0d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-02 08:47:08.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-02 08:47:08.000000000","tz":60},"subject":"Reduce number of the workers in fullstack gate job","message":"Reduce number of the workers in fullstack gate job\n\nFullstack job is using different tox env in the gate\n(dsvm-fullstack-gate). To reduce memory usage in that job, some\ntime ago, in patch [1] we reduced number of test workers from 4\nto 3. But it was done for \"dsvm-fullstack\" tox env, not for\n\"dsvm-fullstack-gate\" thus it had no effect on our CI runs.\nNow it is changed also in the CI job\u0027s tox env.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/764907\n\nChange-Id: Iec949922be0779a00bf405c2f1d5646c588d323c\nRelated-Bug: #1906366\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9542b6a0164cbe0c0fd32b15d0d5db652d47a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9542b6a0164cbe0c0fd32b15d0d5db652d47a2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
