)]}'
{"id":"openstack%2Fneutron~778178","triplet_id":"openstack%2Fneutron~master~Idb743bbfd5454b8d245c4dd4b8c3d7021e90d238","project":"openstack/neutron","branch":"master","topic":"bug/1917448","hashtags":[],"change_id":"Idb743bbfd5454b8d245c4dd4b8c3d7021e90d238","subject":"Fix wrong packet_type set for IPv6 GRE tunnels in OVS","status":"MERGED","created":"2021-03-02 12:05:32.000000000","updated":"2021-03-03 10:17:47.000000000","submitted":"2021-03-03 10:15:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778178-bug/1917448","meta_rev_id":"723c68feb031df2a1496d2bcbaaef89687e7450d","_number":778178,"virtual_id_number":778178,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-03-02 14:02:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-03 10:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-02 14:01:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-03-02 13:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-03-03 07:53:19.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-03-03 07:53:19.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 12:59:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-02 13:59:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-02 14:01:06.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-02 14:02:55.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 14:30:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 07:51:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-03-03 07:51:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-03 07:51:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"480dc011bfe3a8aa7909e4296b97a961f1945fe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 12:05:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6540b3941c1f9c43229db7417e4b3a89e028749","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 12:59:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43bfd4d7a7fe46499787c4b9e0a2f996 : SUCCESS in 47m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e2b13fd148b46d1b5b5419486482489 : SUCCESS in 39m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0863d36a59816a591b78b0ae7ddc3171e26f76ba","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-02 13:59:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1659ae0e2c99ec404e45e41193f87c4321cd330c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-02 14:01:06.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good per https://github.com/openvswitch/ovs/blob/master/Documentation/faq/configuration.rst (Q: Does Open vSwitch support IPv6 GRE?)","accounts_in_message":[],"_revision_number":1},{"id":"4206f751e24f1ffec5bb6b62d96e8cb94375b10f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-02 14:02:55.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/78/778178/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a92aa1474afe092ab995b0b923723f87a4e0b37f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 14:30:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c622f8f156af4ffdbd2b9c239ef24a60 : SUCCESS in 47m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7f53320fd55b4ec9b750f5a9b24d88cb : SUCCESS in 1h 04m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/39709c3af44d4da0afd5613e00e55a32 : FAILURE in 1h 57m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4f7e6506ee18461abf90abe7b94c1d5d : SUCCESS in 1h 45m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/921cf5fb70e14673a4d552866f6f3b9e : SUCCESS in 2h 17m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6948a1ed8714850b43fade185b7ac92 : SUCCESS in 37m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38914d8ce43d43bfbdf82cf82e6b2aa5 : SUCCESS in 31m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60dcbde39bf42239fc632dbc39239aa : SUCCESS in 13m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea1cbcf33c6844dc968811b532f0588f : SUCCESS in 29m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f8ad4a2d6b0408d81a894b21a0abb68 : SUCCESS in 28m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73aa7cc955ab42e0b72b7980f323ed2d : SUCCESS in 32m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ca3ba091654432955c78bfc3248630 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa7cdbd5faf2437fb4b89a5d628bc1b9 : SUCCESS in 1h 03m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/635d51076e9542778392b445b5115d20 : SUCCESS in 1h 08m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1486eb6744b6430287a23e070e228eba : SUCCESS in 1h 34m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ce98d7cb94f4ebc8cca4e363e9237ba : SUCCESS in 1h 01m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c37446408e794ffeb34efafc3226886c : SUCCESS in 1h 05m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e4f6285a3c8e45908e176e0a7d8e11b8 : SUCCESS in 1h 16m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4546b92664bb429594d62d5e7e69a0f2 : SUCCESS in 1h 20m 19s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a381a4e009794f22bd15c2b605473c75 : SUCCESS in 2h 02m 45s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcd0e51302d448a8bed881fc86b0043a : SUCCESS in 1h 02m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9aeb729c38784cb0b621382bf2ac6b94 : SUCCESS in 1h 07m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2a967ba31a314038b54f1e57c8bb5e32 : SUCCESS in 32m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c20f1cd36744c409a1d742a592d875e : SUCCESS in 1h 23m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a2adb1b0873f41d5b70ab28781dc82cd : SUCCESS in 1h 23m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e0b6ac1cf44f0d83bf4ff2f8203e7e821b1bfeb6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 14:48:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d6ffa675ad606037bd0e357e2c71d9df847386bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:04:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8354fd1febae46c692e9821ac7f968b4 : SUCCESS in 42m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0b3cbbf052d4befac96f57b863751ba : SUCCESS in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59c238b91e0794627ffc03e57b998f1b116f2e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 17:18:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fada401e92d14544ba524042f1b2dfce : SUCCESS in 45m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/40fb8d05eb7148a2a698839839945231 : SUCCESS in 1h 35m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ce0a745a4c24433ab323461c7f1b97f7 : SUCCESS in 1h 32m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b21c0104d16f4be2bd97aa005e43a8d4 : SUCCESS in 1h 34m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/67c00b4b092e4df6a2de0c60f2cc14cb : SUCCESS in 2h 23m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56c24ccff0ab4c35bacd9ba45ed2cbef : SUCCESS in 43m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82cc1d1ab13a496a94f2a38f082ec6c5 : SUCCESS in 39m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b19b774df384fd6a57ab15219706fde : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64c356b9c965444391a93631d104d653 : SUCCESS in 40m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb9ff60595a84de78d302ccab7cee25e : SUCCESS in 28m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9641e30306ca41d1bdca26fe9005820c : SUCCESS in 31m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/727cf078eed7471b9e16809601588192 : SUCCESS in 12m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/31b924714be34ad689378f60e5c4bf3f : SUCCESS in 49m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f0671a674b2409fae87a4d769ee1486 : SUCCESS in 1h 19m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ecc8b99953d74072a9b9ea4cf038150f : SUCCESS in 1h 45m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b36b2d744abd4e8b885892be06ba7d76 : SUCCESS in 1h 03m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4e7a0fd2ff4e4107821482aaad9348a7 : SUCCESS in 1h 05m 45s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7805536dde63490c8cd9ff651da26364 : SUCCESS in 1h 36m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/26a0b376fea747c591ba6b9a8c0c374c : FAILURE in 1h 35m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb46479cd9994383a618fa38e8a770c9 : SUCCESS in 1h 38m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/554a4aecae424744a6315818751b157a : SUCCESS in 1h 04m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2152b4908ae04d6eacfc59e59b1e5b2e : SUCCESS in 1h 18m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9a79d86a7b6c44a49a5561a243b162eb : SUCCESS in 41m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f27ff7be07c43f598aa2b1f158852a6 : SUCCESS in 2h 02m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1444bfdefd244acfaddca7923299ff31 : SUCCESS in 1h 24m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5098d92195178fd2caf90c4f14029b4e77f4fdfa","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-03-03 07:53:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cec19eb9da974580d0bbb2daa3b982a68b2bda4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 07:53:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19df56ab99dc01d8699985eecd30bc345ad8585b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 10:15:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cc5dc9e5a7b44ca8b72543579102173a : SUCCESS in 1h 04m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1238e2b5ca4a4c5a95fc2506442cc954 : SUCCESS in 1h 56m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0127c7400db6446dbb620851c2916fb0 : SUCCESS in 1h 18m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/04d26d51cfc5445fba847221bb53a73a : SUCCESS in 1h 47m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e5ab85b37c804f4693590796489dd72c : SUCCESS in 1h 12m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f16baacf8603466abff250bfe20e324b : SUCCESS in 34m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c330a78f3900470990816dced52be2fb : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf559fbbc5d34070a64b1334c79d6b77 : SUCCESS in 31m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3abff45977a2413d9c0d6db1b9e840a2 : SUCCESS in 23m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/604ed83348354942bd25eac3efbff0b5 : SUCCESS in 11m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fce0154f85814c9d9b2af2cb82a966a7 : SUCCESS in 47m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7c2cf2470bd412aa604f9aabcf3b9bc : SUCCESS in 1h 24m 48s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/815b47ca04a845d3acbbafd712afeb30 : SUCCESS in 1h 19m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c59fb0bdeb3349c9a36a434e77cacad4 : SUCCESS in 1h 01m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b574b918b86b433c81817ed8b2b2b2e2 : SUCCESS in 59m 08s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b176c2f50679455f93b2da75c59c29a9 : SUCCESS in 1h 41m 31s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d6a1fba0b6d427484b1f43bd34b260c : SUCCESS in 1h 01m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/46a0dc84203d4fcda102a4c3a8968165 : SUCCESS in 1h 11m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/be9ceefbf3c249a4886e9a679b3e2564 : SUCCESS in 33m 50s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9142644b5aa346d99fcb4e048eac9e31 : SUCCESS in 1h 21m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a4a12644d5bdaa9064981d5abe4c4bd6f19608bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 10:15:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"723c68feb031df2a1496d2bcbaaef89687e7450d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 10:17:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84e42278916d4573acd0b225573e99ae : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1dbe389392bbfadebe8c12b06171cf04a24d32ff","revisions":{"1dbe389392bbfadebe8c12b06171cf04a24d32ff":{"kind":"REWORK","_number":1,"created":"2021-03-02 12:05:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/78/778178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/778178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/778178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/778178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/778178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/778178/1"}}},"commit":{"parents":[{"commit":"77ee0847f5fe886ae6a1c4a9b9949d3590567e0d","subject":"Merge \"[FT] Reset \"AgentCache\" singleton in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77ee0847f5fe886ae6a1c4a9b9949d3590567e0d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-02 12:01:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-02 12:01:39.000000000","tz":60},"subject":"Fix wrong packet_type set for IPv6 GRE tunnels in OVS","message":"Fix wrong packet_type set for IPv6 GRE tunnels in OVS\n\nIn patch [1] it was by mistake set to \"legacy\" but it should be\n\"legacy_l2\".\nThis patch fixes that mistake.\n\nCloses-bug: #1917448\n\n[1] https://review.opendev.org/c/openstack/neutron/+/763204\n\nChange-Id: Idb743bbfd5454b8d245c4dd4b8c3d7021e90d238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1dbe389392bbfadebe8c12b06171cf04a24d32ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1dbe389392bbfadebe8c12b06171cf04a24d32ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
