)]}'
{"id":"openstack%2Fneutron~778402","triplet_id":"openstack%2Fneutron~stable%2Frocky~Idb743bbfd5454b8d245c4dd4b8c3d7021e90d238","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1917448-stable/rocky","hashtags":[],"change_id":"Idb743bbfd5454b8d245c4dd4b8c3d7021e90d238","subject":"Fix wrong packet_type set for IPv6 GRE tunnels in OVS","status":"MERGED","created":"2021-03-03 10:22:59.000000000","updated":"2021-03-03 20:00:03.000000000","submitted":"2021-03-03 20:00:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778402-bug/1917448-stable/rocky","meta_rev_id":"1c7cf9dccb4e1142d49ce3a471b3d28edff3fbc0","_number":778402,"virtual_id_number":778402,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-03 20:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-03 10:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-03-03 13:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-03 13:48:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 10:30:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-03 13:11:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 13:48:24.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"c00122daf0f2684c5ed43619601436884b003a50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-03 10:22:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a9dec9de0fdc76144148ebde85da530bc6dd78e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-03 10:30:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a5c36465b4cade569b53fa29c4e99943073a76f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 13:11:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/cb5b4950c1f249b4ae3a179f08c1a05e : SUCCESS in 47m 08s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/9e375b22676c4a2cbb87251e244562a2 : SUCCESS in 1h 09m 50s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/7090f38eee134834851ca939cc4ea7b3 : SUCCESS in 1h 10m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/e6ab4c7ba2634019b3a11b000cb4154b : SUCCESS in 1h 06m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c522cac090704c5c84ff992bc10d53e9 : SUCCESS in 12m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5768cac8d49f41d59153a6292fae29fe : SUCCESS in 30m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/489640a1485f44579dc5728ed77f429b : SUCCESS in 34m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a4c8a027b90d4b128ec0c81b5a4c2168 : SUCCESS in 10m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/619b956177bf4df48e5a6d8b0a51cb16 : SUCCESS in 1h 39m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/645d27e529984e49bca0ecf288ab20a9 : SUCCESS in 46m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6d26f785ac6452aa8b210d08db98559 : SUCCESS in 1h 24m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/39060f11d0b5499b86e9221ed25e1687 : SUCCESS in 40m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1f4d4a2c9ee3409285d162c1d3599836 : SUCCESS in 1h 17m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ab1a4670baec46f78d28025a6835e34c : SUCCESS in 1h 29m 01s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cc72ab7b0d094efc89b12f4d475dc3b6 : SUCCESS in 1h 20m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/758de53a490648b199ab3895cc12fe26 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d73bec0737f1dae39bcfb64f8d1af2a33a5e14e5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-03 13:48:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60b4691b4291f917603043f2cd798ce1e3d4749a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 13:56:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"241d8e70c755682f73238b3a909dec1ff33f4b78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 20:00:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/ca873fefd53d425aa937d34148e0a91d : SUCCESS in 46m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7653dd5de194c3494e47f7c7f9af60e : SUCCESS in 11m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b5c089487fd4fd3a9c8422ffe0b81f5 : SUCCESS in 24m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f41bfa2490ee400196bbd1c7799e577a : SUCCESS in 23m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/82fc20ef49de49b39d2fb95062b2592b : SUCCESS in 8m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/445809fc7e8141d0b8604ae14a65d61c : SUCCESS in 1h 18m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3b649403e5674f2b8c50caf1aad1ad35 : SUCCESS in 55m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6ff09f346e046e58106868500ed10fe : SUCCESS in 1h 42m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/293b6fa2f8044ca3b0c0f4170b9a77f3 : SUCCESS in 36m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0f757de6f8544cc5ad46f974a6bec721 : SUCCESS in 1h 50m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4a86d4f2f3cd4a049f639a8750d86aef : SUCCESS in 1h 47m 01s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3949db50e37c4090b3d8949ce73bf9bd : SUCCESS in 1h 14m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07a5d59c3c08472f8a642e42be168adc : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1c7cf9dccb4e1142d49ce3a471b3d28edff3fbc0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 20:00:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbfefd1c098ee5a08337848ffc605300d330e383","revisions":{"dbfefd1c098ee5a08337848ffc605300d330e383":{"kind":"REWORK","_number":1,"created":"2021-03-03 10:22:59.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/02/778402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/778402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/778402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/778402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/778402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/778402/1"}}},"commit":{"parents":[{"commit":"36fc4708be6994893d31e8c6bbbb58ec8b7723bc","subject":"Fix losses of ovs flows when ovs is restarted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36fc4708be6994893d31e8c6bbbb58ec8b7723bc"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-02 12:01:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-03 10:22:08.000000000","tz":60},"subject":"Fix wrong packet_type set for IPv6 GRE tunnels in OVS","message":"Fix wrong packet_type set for IPv6 GRE tunnels in OVS\n\nIn patch [1] it was by mistake set to \"legacy\" but it should be\n\"legacy_l2\".\nThis patch fixes that mistake.\n\nCloses-bug: #1917448\n\n[1] https://review.opendev.org/c/openstack/neutron/+/763204\n\nConflicts:\n    neutron/tests/functional/agent/common/test_ovs_lib.py\n\nChange-Id: Idb743bbfd5454b8d245c4dd4b8c3d7021e90d238\n(cherry picked from commit 1dbe389392bbfadebe8c12b06171cf04a24d32ff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbfefd1c098ee5a08337848ffc605300d330e383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbfefd1c098ee5a08337848ffc605300d330e383"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
