)]}'
{"id":"openstack%2Fneutron~778452","triplet_id":"openstack%2Fneutron~stable%2Ftrain~Ic32413f88dd0d2a0479d692b6d824503f7819462","project":"openstack/neutron","branch":"stable/train","topic":"bug/1917610","hashtags":[],"change_id":"Ic32413f88dd0d2a0479d692b6d824503f7819462","subject":"DNM, Just a test of the tempest fix","status":"ABANDONED","created":"2021-03-03 12:37:12.000000000","updated":"2021-03-05 12:58:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3037c6228ad6d6b653fe3e9e4100e13857fb6c4f","_number":778452,"virtual_id_number":778452,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-03 15:31:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 15:31:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72fcbc1a8d8d1bebf11939ba02f747eccb35083b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-03 12:37:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc449f7c1c0f116d4da5ae15c188bf355e34ff82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:31:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c1f6d5ee4bb54f40a33cbe56ad2ef4d4 : SUCCESS in 4m 10s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/47457bae598144bb8bd94e48e6d562e7 : SUCCESS in 46m 57s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/220530bcb17948759aab45580ea71af6 : SUCCESS in 1h 20m 35s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/a2e6e44878d34b9dae0f5f64e741e2b5 : SUCCESS in 1h 26m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/c67301003be54235ae72c1997e73da6c : SUCCESS in 1h 27m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f62a54e31437496499ea3b2f275f98c0 : SUCCESS in 30m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e2897907b6f46a4a37f5d614670a173 : SUCCESS in 12m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b018c21c5fbb4459be29daf4c1bf5375 : SUCCESS in 24m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7a1711f8f5a4f99a10405b98795f7d0 : SUCCESS in 23m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc64e50557af41d282292b9da855292b : TIMED_OUT in 42m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487bde3d63ac4f96a27af017aa6e8195 : SUCCESS in 10m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ea6eb48139a403eb877c6bcdf67272c : SUCCESS in 4m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/05980caea04f44d59f889b8db9a1a0bb : SUCCESS in 57m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/72232e7f893d4622bfdc148073c54329 : SUCCESS in 1h 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a6c0726abed48fbbe6b10a2a9266b6c : SUCCESS in 10m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cdff0c9564384b02856083621299c303 : SUCCESS in 45m 01s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/adc314d47dd5442bb4745a48ce60d3ed : SUCCESS in 43m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/512a358d0d954019943e4770ed8d5eab : SUCCESS in 1h 08m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/edbbcf495cfb478e85e6facb0d44ef8f : SUCCESS in 1h 24m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/763d415fb7b54303ae5e38a242c5d2c0 : SUCCESS in 1h 15m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/84a2df1b207e434f9fcfda4ec85f3598 : SUCCESS in 1h 11m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b658dae7dbfc49f2a8cd3ac10229f78b : SUCCESS in 1h 05m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f264a8c3c74c4cf1986d1e36b606dd73 : SUCCESS in 1h 03m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/20fe3134361b454ebf98eabf72277f84 : SUCCESS in 1h 47m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/950960cfb1af4f8983bcd8ef3c8eb4de : SUCCESS in 1h 50m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c20b4212a1a1493b81060cf572e95fbe : SUCCESS in 1h 07m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8dcdfd8b3e854f31b85106df8af69555 : SUCCESS in 40m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/da78a34624ee46d8b36c1a8b44e46a67 : SUCCESS in 1h 22m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2524d2287cdb4e5396d9e09e3430d95f : SUCCESS in 2h 15m 58s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/93cedeb04d484e1aa4aab78260f28b95 : SUCCESS in 2h 07m 44s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/b4480426ff7847d3bd632fea8cbe36a6 : FAILURE in 18m 45s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/38d400103f2a448a8aca201dfdf0e54d : SUCCESS in 1h 09m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bdb98938e384b48a83bae11b3d741af : SUCCESS in 1h 00m 02s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6947e8c647df4b53857d6d74316f8d05 : SUCCESS in 1h 57m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5419cb53a8054731a05d01d1eecc4e18 : FAILURE in 44m 28s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/433f3c188adf4f478271e98ebc35342f : SUCCESS in 1h 12m 11s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8718915e67334134858d4735963d5fdb : SUCCESS in 1h 16m 45s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/0234793acbb24d7f9e6b8450b558d4c8 : SUCCESS in 2h 38m 15s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0a5b007df6024025a534fee9b543dc39 : FAILURE in 1h 11m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3037c6228ad6d6b653fe3e9e4100e13857fb6c4f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 12:58:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c00017719efb85785656e27f72d0288cbc0494f0","revisions":{"c00017719efb85785656e27f72d0288cbc0494f0":{"kind":"REWORK","_number":1,"created":"2021-03-03 12:37:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/778452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/778452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/778452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/778452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/778452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/778452/1"}}},"commit":{"parents":[{"commit":"4381f792c8bd8e5d309ed50288dd4a15a3628e0f","subject":"Merge \"Don\u0027t try to create default SG when security groups are disabled\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4381f792c8bd8e5d309ed50288dd4a15a3628e0f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-03 12:36:03.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-03 12:36:03.000000000","tz":60},"subject":"DNM, Just a test of the tempest fix","message":"DNM, Just a test of the tempest fix\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/778451\n\nChange-Id: Ic32413f88dd0d2a0479d692b6d824503f7819462\nRelated-bug: #1917610\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c00017719efb85785656e27f72d0288cbc0494f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c00017719efb85785656e27f72d0288cbc0494f0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
