)]}'
{"id":"openstack%2Fneutron~778595","triplet_id":"openstack%2Fneutron~master~I0970acf02f1e8c51bf4f838ce8c3b4986d9d8f3c","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I0970acf02f1e8c51bf4f838ce8c3b4986d9d8f3c","subject":"[DNM] Please ignore: forcing jobs to use specific ovs idl changes","status":"ABANDONED","created":"2021-03-04 00:00:06.000000000","updated":"2021-03-30 17:45:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dd56c3f73e5b5eed8d69a6ab6afda3090ae0d9e9","_number":778595,"virtual_id_number":778595,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-06 00:25:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":-1,"date":"2021-03-05 16:50:28.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 00:00:42.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-04 00:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-04 02:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ddf3f18db04c44d1e7478c9e2bae4e9247e53546","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-03-04 00:00:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c37a9bff062bb171cb1d1877b44458470d735b2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-04 00:00:42.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"8e0cf32767cf8e8c1be26c5952e3f203055db4b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:48:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b34d553b22d347b39da6701b2259ca43 : SUCCESS in 42m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5b49d6374924a579e4264cb7bd1fb39 : SUCCESS in 42m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de78a0b628565cd9f74682ad2fa0876b220d0c8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:12:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/493db8daaf534e1a95b63c1fdf8b2a2f : SUCCESS in 56m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0b1ea379929d4a86b67f62701d3df068 : SUCCESS in 1h 52m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c2d946b545a04f00a9ace04a573dd314 : SUCCESS in 1h 48m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6203c3634464c239e2b44dfddc2bccd : SUCCESS in 1h 43m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b216e91e047241968c79615bfcf03996 : SUCCESS in 2h 07m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ae0d8fb3d3a41348401045692d0c87f : SUCCESS in 43m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/936c58eed499424bb854787149da3dc3 : SUCCESS in 35m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f6565ff7fc6444b9725449b051871da : SUCCESS in 14m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79b7bc0be16f4e559ed3ef14267f729c : SUCCESS in 37m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b688e954a81149f89ebc6657cc3194f9 : SUCCESS in 43m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bfa8ce01d674286b30267587f58babe : SUCCESS in 32m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26666a457b94b09af481a1a6e651428 : SUCCESS in 9m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb2ad04a50074efe96b384f348923930 : FAILURE in 50m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e3a8bfb97504685ab04c27f5d8984d2 : SUCCESS in 1h 22m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5f1b36bf4d85453e995819e245497611 : SUCCESS in 1h 40m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/022ad93742384c8e92194e770644c6b1 : SUCCESS in 1h 07m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d088fe1c7a594b749659a6012504373a : SUCCESS in 1h 10m 32s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/475160bc3e614b20ad20636f28609378 : SUCCESS in 1h 25m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a878718993cc469589e50a838f282aa4 : FAILURE in 1h 12m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/13cfb6c92451486cadff70a288434040 : SUCCESS in 1h 44m 42s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00c5b47603524b9c9671915a7e025d35 : SUCCESS in 1h 08m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/acba6fee280f479688f6098f02693633 : SUCCESS in 1h 17m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/561dad5a70a141b7a04d51189e6dc190 : SUCCESS in 37m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c18e79be6d6e4c53b3855ecdc5c41e6d : SUCCESS in 1h 31m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/221a5fae98694c2280cf02510201af03 : SUCCESS in 1h 29m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e5fc0d748f5eba09346cef324657e0a0ac5653da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-03-05 10:25:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ace7b2f37c806838b6c94eaab7be737ce5dd5902","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-03-05 10:26:50.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"069ff1041451662bbcffd9204e62904e51e5c34d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 10:52:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"40169acbc6122aabcdd21a82eb824371a3ab5b36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:10:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/27a99e20675d4fe0a16cba84441278a3 : SUCCESS in 49m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8a6007646f6d44aab42a97309a12babf : SUCCESS in 1h 45m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e61dd8bd5864ef89e553877ab1b379b : SUCCESS in 1h 35m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/33569162ec504bc4abe2177451950ca0 : SUCCESS in 1h 25m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/489d28b86e334860b74c5570f23f39c2 : SUCCESS in 2h 18m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a328a16b9c1a4ae587eb66d4ad0685a3 : SUCCESS in 39m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6e387e789ac4b9da8f2f5b3a07d6d53 : SUCCESS in 32m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b386e428a6cb49cdbe4dc935de088051 : SUCCESS in 11m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3e30cb16b6a487685214e7c04b17d8e : SUCCESS in 36m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbf54515b4e64866a7f8869d2d998784 : SUCCESS in 33m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2d6028366904a9baa0bbc35a69f3fc6 : SUCCESS in 25m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891324a50cf24facb10b2e47275aa06c : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c6759255b4a49adbfedb64cda9f2993 : FAILURE in 59m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad736e74f7164840962e93fb48138edf : SUCCESS in 1h 26m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b2ff76d209174d10933f8161ebd88942 : SUCCESS in 2h 28m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9bd167fe870b4f06ba929e935b4bada5 : SUCCESS in 57m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6aa0f244ffff42e5833400219dea467c : SUCCESS in 1h 05m 05s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/64f929c9a6f74d9495b30f8d54f70ad0 : SUCCESS in 1h 25m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2306f68b4fd14e45935ef44c2987f20c : FAILURE in 1h 12m 42s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/263ce31e6f324d51b2904b0695837e93 : SUCCESS in 1h 39m 21s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35f4e914030a47e39ea38f9c5cadb8dc : SUCCESS in 1h 01m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/11134ae1fe634a3d94741c1793abb7f9 : SUCCESS in 1h 17m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f2e56ffb72334415af85b2d555b1d330 : SUCCESS in 37m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8a220a5b619f415b9513663a5b1372f0 : SUCCESS in 1h 38m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/44cbf2ff93fc4e0f98c687009f502f96 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8a18301162a24f3663a62fc117b99a6b4df6a4d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-03-05 16:49:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee95b61e576f5d27151ea2c6d55e5c31d21ed2eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-03-05 16:49:59.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"1d1211d4ba60df4afdd2dfb83e3c0801c022469e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-03-05 16:50:28.000000000","message":"Patch Set 4: Workflow-1\n\nrecheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"dae3f997eefde52964c6d8317fba5fd528ea49e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 17:37:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77bf56c39e8e4511a0e2f31afcc55012 : SUCCESS in 40m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bae4f636f4f841b6b8d7adbe58e79fe8 : SUCCESS in 41m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"86bc19ee555d3839b155071730790e4449f50745","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 18:32:32.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"860f0bef0e6f8c684e23c67245465bd9d4cc1f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 00:25:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b1020410b46c4e339bef1ba57699300b : SUCCESS in 53m 11s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8b0caca8de364e2a959274ecde48845c : SUCCESS in 1h 40m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d3f419bae541442b8635205bd6a853b9 : SUCCESS in 1h 30m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/26b0005877994ca480c8f6b5636f610d : SUCCESS in 1h 23m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8a4bb651745a41d78aa341285769ceb0 : SUCCESS in 1h 51m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e4dcf34a94b44418b3fa0df2c675b15 : SUCCESS in 32m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b2debc842654ad798ff2af6d361cd30 : SUCCESS in 30m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0493ad7d1e3a4500b263863d32b6d895 : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/934ee79edb9146248c0a92c3889ad578 : SUCCESS in 29m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd0a8a0159dc4c568b476174f5bd188f : SUCCESS in 30m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c1ce88a27a54445a518c4b7a7a51657 : SUCCESS in 29m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f644ebcc4e41d588860df175d4acff : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/105353dc8d0c418cabce142c01a8f769 : SUCCESS in 51m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3585fb08eb364d4abebfcf418eed7392 : SUCCESS in 1h 29m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a8cd0d6b88af40439c6c605a0dd15b6c : SUCCESS in 1h 28m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69e45556d5944ad28b6752bfda41fe1d : SUCCESS in 59m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/06423590fad9413a9eea9388bda1c796 : SUCCESS in 57m 00s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5b4bbeddb16245609ee9af91980090c1 : SUCCESS in 1h 19m 50s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f8f843464bce4f1c82ad5b75865ef386 : FAILURE in 1h 14m 49s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/38b96cc3bb264e6d817565c661b90cf2 : SUCCESS in 1h 37m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2e17c7ed4c743e48498c01dd2d34953 : SUCCESS in 1h 07m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2949a42fb8fc4ee1a53e841d1135ad04 : SUCCESS in 1h 13m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/79f897c65c5d4f3b873464035b686177 : SUCCESS in 42m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ca2ba9e351914be78452f495406b61f8 : SUCCESS in 2h 01m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2ece9b48de534e99ac7115bc0b54b4e4 : SUCCESS in 1h 24m 00s","accounts_in_message":[],"_revision_number":4},{"id":"dd56c3f73e5b5eed8d69a6ab6afda3090ae0d9e9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-03-30 17:45:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1f0fe59fb2b98a0ecdba49fd5e09d97639ed8ce5","revisions":{"3806581c25fae309a3b7978be3d0f36ce9eca549":{"kind":"REWORK","_number":1,"created":"2021-03-04 00:00:06.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/95/778595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/778595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/778595/1"}}},"commit":{"parents":[{"commit":"6571ff2286b9c4124e89785fb47423e4872dfcec","subject":"Merge \"Ignore python warnings in the fullstack job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6571ff2286b9c4124e89785fb47423e4872dfcec"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-03-03 23:59:39.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-03-03 23:59:39.000000000","tz":-300},"subject":"[DNM] Please ignore: forcing jobs to use specific ovs idl changes","message":"[DNM] Please ignore: forcing jobs to use specific ovs idl changes\n\nChange-Id: I0970acf02f1e8c51bf4f838ce8c3b4986d9d8f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3806581c25fae309a3b7978be3d0f36ce9eca549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3806581c25fae309a3b7978be3d0f36ce9eca549"}]},"branch":"refs/heads/master"},"e109b8ae3c919fd8e229e68e5aed3ae4aa5955bb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-05 10:25:59.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/95/778595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/778595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/778595/2"}}},"commit":{"parents":[{"commit":"cb64e3a19fdddb3eac593114a482c9dd69be68d5","subject":"Merge \"Add new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb64e3a19fdddb3eac593114a482c9dd69be68d5"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-03-03 23:59:39.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2021-03-05 10:25:59.000000000","tz":0},"subject":"[DNM] Please ignore: forcing jobs to use specific ovs idl changes","message":"[DNM] Please ignore: forcing jobs to use specific ovs idl changes\n\nChange-Id: I0970acf02f1e8c51bf4f838ce8c3b4986d9d8f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e109b8ae3c919fd8e229e68e5aed3ae4aa5955bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e109b8ae3c919fd8e229e68e5aed3ae4aa5955bb"}]},"branch":"refs/heads/master","description":"Rebase"},"f13005fb1c466dd117020393fe6346db86d5a966":{"kind":"REWORK","_number":3,"created":"2021-03-05 16:49:01.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/95/778595/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/778595/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/778595/3"}}},"commit":{"parents":[{"commit":"cb64e3a19fdddb3eac593114a482c9dd69be68d5","subject":"Merge \"Add new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb64e3a19fdddb3eac593114a482c9dd69be68d5"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-03-03 23:59:39.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-03-05 16:48:56.000000000","tz":-300},"subject":"[DNM] Please ignore: forcing jobs to use specific ovs idl changes","message":"[DNM] Please ignore: forcing jobs to use specific ovs idl changes\n\nChange-Id: I0970acf02f1e8c51bf4f838ce8c3b4986d9d8f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f13005fb1c466dd117020393fe6346db86d5a966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f13005fb1c466dd117020393fe6346db86d5a966"}]},"branch":"refs/heads/master"},"1f0fe59fb2b98a0ecdba49fd5e09d97639ed8ce5":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-05 16:49:59.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/95/778595/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/778595/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/778595/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/778595/4"}}},"commit":{"parents":[{"commit":"cb64e3a19fdddb3eac593114a482c9dd69be68d5","subject":"Merge \"Add new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb64e3a19fdddb3eac593114a482c9dd69be68d5"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-03-03 23:59:39.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2021-03-05 16:49:30.000000000","tz":-300},"subject":"[DNM] Please ignore: forcing jobs to use specific ovs idl changes","message":"[DNM] Please ignore: forcing jobs to use specific ovs idl changes\n\nref: https://review.opendev.org/c/openstack/neutron/+/763403\n\nTrigger with:\nrecheck experimental\n\nChange-Id: I0970acf02f1e8c51bf4f838ce8c3b4986d9d8f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f0fe59fb2b98a0ecdba49fd5e09d97639ed8ce5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f0fe59fb2b98a0ecdba49fd5e09d97639ed8ce5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
