)]}'
{"id":"openstack%2Fneutron~778684","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I080381cb8a88a27130ac21c0e27fc915e38bdedd","project":"openstack/neutron","branch":"stable/train","topic":"improve-neutron-ci-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"I080381cb8a88a27130ac21c0e27fc915e38bdedd","subject":"Ignore python warnings in the fullstack job","status":"MERGED","created":"2021-03-04 13:34:58.000000000","updated":"2021-03-06 02:10:30.000000000","submitted":"2021-03-06 02:08:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778684-improve-neutron-ci-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":778683,"cherry_pick_of_patch_set":1,"meta_rev_id":"612847ce8e72acf86bee9adcb571d81330bdb00f","_number":778684,"virtual_id_number":778684,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-06 02:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-04 15:11:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-03-05 15:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-03-05 07:20:31.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-05 15:48:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 15:11:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-04 15:12:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 07:20:31.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-05 15:48:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"3f95b07c00a2d354741424f1af80b9b25d88ec65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-04 13:34:58.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"e8dc8d22c8c1687da44f29b1a4e9e048ef5d66e9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-04 15:11:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"92ce4dd8f33f9e8909724eaf760a2dc684f45a07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 15:12:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb0d5a406fda4ac789bf6ec3c94fb3ab : SUCCESS in 28m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/581d6f14b10140f1b7f79a97754d347f : SUCCESS in 12m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40a292d6171d443bae344b3d4d59a96f : SUCCESS in 23m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9de26ec983ff44a890ab4c198dd7cd9f : SUCCESS in 26m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee8fe5f609db4b4dbbdedd2488f4cb5e : SUCCESS in 26m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf156704ea64ef880484c0112e603ca : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/de3e40aff1f049618c814658cd9462d7 : SUCCESS in 1h 09m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e7a503aea0714ca1a9c8a92b81628248 : SUCCESS in 1h 18m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75239c49dc9f4899b12b55b9586ede30 : SUCCESS in 11m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b49ed50aba914d59a1ad52a40433c62f : SUCCESS in 47m 36s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b55fc93eeb3b4670a402aea2487f621c : SUCCESS in 48m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1062b6ccc85440b0bc6d5a5456dc9171 : SUCCESS in 1h 13m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dd6c30084c5f4edd8ee714500df77f26 : SUCCESS in 1h 30m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/58c7380f3dc64577b28808e045110340 : SUCCESS in 43m 16s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/effac8fb95434a99b850de74d0e297de : SUCCESS in 1h 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6631d26c1ec960ec1d1b8a052cec35dc35da1c6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-05 07:20:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6742bb7d43848fe305bbe2ad105360737d10e420","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-05 15:48:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f4da96eb680cfe8f919b46222d020b944ac3134","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:01:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a625ca36af6b355c18dace9121a8dddd52da230","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 02:08:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb4667dac6f24ddaa4b22054581a6a5c : SUCCESS in 13m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fa9da78812d4d28a940c917cb7a4944 : SUCCESS in 21m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea5c596482d54d6d804196597ce2da7d : SUCCESS in 27m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c22b3f5bd9b34593bdd849d7482516de : SUCCESS in 31m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a07d0be04142078d2c32e7025a6c59 : SUCCESS in 11m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/1361989ce651469795469072cb272df9 : SUCCESS in 1h 10m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9647e890974744b7af06363ec286bc0a : SUCCESS in 1h 14m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e2e410de7f8497eb5ca93ab8d946a6c : SUCCESS in 10m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9e92235fb4154e59bc55587a25ebac29 : SUCCESS in 47m 05s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/e1cff43ce8b14d5e86077ccb83fee38b : SUCCESS in 43m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f0b7baafea164669a2446582852068fb : SUCCESS in 1h 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5f8c21d3803227c4dff0564bd51be60ec72472f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 02:08:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"612847ce8e72acf86bee9adcb571d81330bdb00f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 02:10:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/542619cbd55b472093ffb11e02775cd9 : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d46da1caffd84219b4c63befe1aa0705 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0165b1beb1f6861971887aa688dc7d4652ddd41","revisions":{"a0165b1beb1f6861971887aa688dc7d4652ddd41":{"kind":"REWORK","_number":1,"created":"2021-03-04 13:34:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/84/778684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/778684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/778684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/778684/1"}}},"commit":{"parents":[{"commit":"4381f792c8bd8e5d309ed50288dd4a15a3628e0f","subject":"Merge \"Don\u0027t try to create default SG when security groups are disabled\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4381f792c8bd8e5d309ed50288dd4a15a3628e0f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-02 14:34:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-04 13:34:58.000000000","tz":0},"subject":"Ignore python warnings in the fullstack job","message":"Ignore python warnings in the fullstack job\n\nIf there is too much output produced to the stdout/stderr during the\ntests, stestr can hangs and job will then time out.\nTo avoid that in fullstack job, this patch is doing the same what\nwas already done for functional job as well - set PYTHONWARNINGS\nvariable to \"ignore\" to ignore all warnings and not send them to stdout.\nAll info really needed to debug tests is still in the tests\u0027 logs in\nspecific directories.\n\nChange-Id: I080381cb8a88a27130ac21c0e27fc915e38bdedd\n(cherry picked from commit 14955f93f6841fc65b7d761f15cf2227f31042d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0165b1beb1f6861971887aa688dc7d4652ddd41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0165b1beb1f6861971887aa688dc7d4652ddd41"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
