)]}'
{"id":"openstack%2Fneutron~778688","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~I12406c1f654ba9ad8eb60ffc965293ecbd5b7be5","project":"openstack/neutron","branch":"stable/victoria","topic":"improve-neutron-ci-stable/victoria","hashtags":[],"change_id":"I12406c1f654ba9ad8eb60ffc965293ecbd5b7be5","subject":"Make neutron-tempest-dvr-ha-multinode-full to be 2 nodes job","status":"MERGED","created":"2021-03-04 13:41:31.000000000","updated":"2021-03-27 17:57:33.000000000","submitted":"2021-03-27 17:53:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778688-improve-neutron-ci-stable/victoria","cherry_pick_of_change":774070,"cherry_pick_of_patch_set":1,"meta_rev_id":"c7022aa879dbaf28f62003102c71e3d99d4e02dd","_number":778688,"virtual_id_number":778688,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-27 17:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-27 07:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-03-19 09:52:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-03-27 07:07:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 17:29:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 09:52:02.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-27 07:07:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"d5557375e70e023c0c95607c11eae289df453ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-04 13:41:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ad146b634dabd52de871a7debceef9a59fee2b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 17:29:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/398706f33be740b4bfce516beca2062f : SUCCESS in 47m 07s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/b43e56feed56478ea9527ca7c617ac6c : SUCCESS in 1h 35m 23s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/436e9e88639642238d1a9b5479ee2f46 : SUCCESS in 1h 46m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/5d79eb91d9d241b781dd897eec71141d : SUCCESS in 1h 26m 47s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/bf7cc0a756cd49c28ffa9733b9d691b0 : SUCCESS in 1h 44m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/752cad88fa9b43a1a16a5c574815f73c : SUCCESS in 31m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd03f59f00a4d2a90ddfac2dbcdaecc : SUCCESS in 16m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0a6424195d3433785eec174142dddf9 : SUCCESS in 27m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55af8634d31243a7a184f4980bf97f3e : SUCCESS in 26m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3066c7b3b3674305b8596a5d23f51c0c : SUCCESS in 9m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa7074f5431a46e086cecfaecac24c5a : SUCCESS in 1h 11m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/66adce81938d459ab8511fd580146129 : SUCCESS in 1h 12m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5a82f1a7ae1c441db297fc5ce07087c3 : SUCCESS in 1h 45m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5e16a9a7db0465890f4f35bb6654539 : SUCCESS in 1h 03m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/28b194ea9f91468da75881dcf20b7c11 : SUCCESS in 1h 06m 20s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/42d4dfad5876411e90c2b83f59221e0e : SUCCESS in 1h 19m 48s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1dbafc28f9d94aac9197173fb0e13834 : SUCCESS in 1h 15m 15s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/86b1e438bdaf4dd783221d237dd6812b : SUCCESS in 1h 09m 25s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4455efc2551a4e9888fe0b51dc612320 : SUCCESS in 1h 45m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ce61552c2f8f42bc85ecb7cf465c9ca9 : FAILURE in 1h 51m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b08f4e0d7c514ff28da0cc222b19cf94 : SUCCESS in 1h 12m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f8817751ff7a47f3844c68a49c6679d1 : SUCCESS in 1h 13m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8c6413edd45e4f39a63af0e8ee7387d0 : SUCCESS in 2h 23m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f887d30f2eec44419aafd1a44fdfd3d6 : SUCCESS in 57m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d3d77814f26f47cbb4348053ca468168 : SUCCESS in 1h 39m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c5b940a57758486fb979f0d05a0d1c83 : FAILURE in 39m 42s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5f7845954a05473896465eec73e41abe : SUCCESS in 1h 21m 27s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6d8045504d244724ac1f6742509787ed : SUCCESS in 3h 44m 12s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f82c69e25e664a29b93e0db79dcdd657 : SUCCESS in 1h 21m 35s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/08ab21e93aff4bdaabc74e1b783cf243 : SUCCESS in 1h 59m 46s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bb1a71ca3dad4d68a93ee6ce15fab83d : SUCCESS in 1h 58m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a2f1c9bc3f624aa0957a5d804ba77fc6 : FAILURE in 39m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d3937dfcd4b7dfac874c93bcf03e1705c6f5473","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-19 09:52:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"475938cc6e8e7f7d34c4465923513a02d0345d18","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-27 07:07:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ebd792523702d6a4f22f03b4646fe064b3b7e61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 07:08:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9194b918fae86663b6c5b8cd1654a517efb9c77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 14:13:16.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8f25eb148a862473319e4517f79fe21f42835c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 17:53:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/094c79211570425ead9fd9dee406dec8 : SUCCESS in 52m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d9a61f215a4933bf6f03f05fce483a : SUCCESS in 14m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1336ab07d5d400a8c67103ed78b05b4 : SUCCESS in 33m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b975ebadf38400a859959835566a568 : SUCCESS in 30m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a799b1ab31404039b5268c4dd5125a90 : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4516f49ff4b423a9bf5589fc529e5cc : SUCCESS in 1h 20m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4736d9a05b244493938135c9cd0c899a : SUCCESS in 1h 22m 33s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/12d0c0d564914b52ae6e6d0d5f0c84c6 : SUCCESS in 1h 26m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0cef5ac417ab4b4cbe332e8762c48558 : SUCCESS in 1h 07m 07s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/082650011c2240ad8c8b44e01e06d4db : SUCCESS in 1h 18m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b8001e9da07a40a9afded68253277428 : SUCCESS in 1h 30m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fd4487c5aeb4e42851dafbf08b15fce : SUCCESS in 1h 13m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/106a6dbe0830494da91603d7601109fe : SUCCESS in 1h 15m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/01f72071e64649048b20f18e039f14ac : SUCCESS in 2h 17m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1b6d335426d42f9a4d6c6f372c6141c : SUCCESS in 1h 06m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6c30c459eefe468b80a7172a36d47a0f : SUCCESS in 1h 28m 36s","accounts_in_message":[],"_revision_number":1},{"id":"96017501938511ff603cea4ece844a88bc330ebb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 17:53:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c7022aa879dbaf28f62003102c71e3d99d4e02dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 17:57:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49fb73b4f35b4110860da39aa4822f0e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e62628b2779a25215e0fc7335ffd56d5720b3e6c","revisions":{"e62628b2779a25215e0fc7335ffd56d5720b3e6c":{"kind":"REWORK","_number":1,"created":"2021-03-04 13:41:31.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/88/778688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/778688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/778688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/778688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/778688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/778688/1"}}},"commit":{"parents":[{"commit":"ec79932568056b17db1a5584cb507f09250b44c3","subject":"Merge \"[OVS FW] Allow egress ICMPv6 only for know addresses\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec79932568056b17db1a5584cb507f09250b44c3"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-04 11:21:03.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-04 13:41:31.000000000","tz":0},"subject":"Make neutron-tempest-dvr-ha-multinode-full to be 2 nodes job","message":"Make neutron-tempest-dvr-ha-multinode-full to be 2 nodes job\n\nIt was run on 3 nodes but I think there is no real need for that.\nRunning that on 2 nodes should be enough.\n\nChange-Id: I12406c1f654ba9ad8eb60ffc965293ecbd5b7be5\n(cherry picked from commit 1772e9543ef835eb24c3aacce5600593c6055d73)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e62628b2779a25215e0fc7335ffd56d5720b3e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e62628b2779a25215e0fc7335ffd56d5720b3e6c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
